)]}'
{"id":"openstack%2Ftripleo-ci~781721","triplet_id":"openstack%2Ftripleo-ci~master~Ib797bde92a0328e0c1baf3779c8a047729979e5a","project":"openstack/tripleo-ci","branch":"master","topic":"dockerfile-labels","hashtags":[],"change_id":"Ib797bde92a0328e0c1baf3779c8a047729979e5a","subject":"Add option to include labels in dockerfiles","status":"MERGED","created":"2021-03-19 13:44:42.000000000","updated":"2021-03-20 01:44:08.000000000","submitted":"2021-03-20 01:44:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"781721-dockerfile-labels","meta_rev_id":"d58746afef27598aafe3d53d8f86d8e2d91da566","_number":781721,"virtual_id_number":781721,"owner":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-20 01:44:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2021-03-19 15:01:47.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"all":[{"value":2,"date":"2021-03-19 18:33:49.000000000","_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":2,"date":"2021-03-19 18:32:50.000000000","_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"all":[{"value":1,"date":"2021-03-19 18:33:49.000000000","_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":16282,"name":"YaZug","email":"jschluet@redhat.com","username":"yazug"},{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}],"reviewers":{"REVIEWER":[{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":16282,"name":"YaZug","email":"jschluet@redhat.com","username":"yazug"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-19 13:47:22.000000000","updated_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"reviewer":{"_account_id":16282,"name":"YaZug","email":"jschluet@redhat.com","username":"yazug"},"state":"REVIEWER"},{"updated":"2021-03-19 13:47:22.000000000","updated_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"reviewer":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"state":"REVIEWER"},{"updated":"2021-03-19 13:47:22.000000000","updated_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"},{"updated":"2021-03-19 13:47:22.000000000","updated_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"reviewer":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"state":"REVIEWER"},{"updated":"2021-03-19 13:47:22.000000000","updated_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2021-03-19 13:47:22.000000000","updated_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2021-03-19 15:01:47.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-19 16:17:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a02303b3c01ac10be7b2eeb33114f458af944c9a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2021-03-19 13:44:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"de85f4b87e9178912afd3c689bd9f23afaa1393d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-03-19 15:01:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/e3d74f3a4af64e1aa650fe59bda5c482 : FAILURE in 1h 10m 37s (non-voting)\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/ffd9f089569946fb887aa7084aa3f702 : FAILURE in 1h 14m 56s","accounts_in_message":[],"_revision_number":1},{"id":"298c1c1be7b3c4fcaa4300fd409d2de542eab652","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 16:17:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-7-buildah-train https://zuul.opendev.org/t/openstack/build/c0442db83b014621b0812d6137b455c1 : SUCCESS in 1h 18m 47s\n- tripleo-build-containers-centos-7-buildah-stein https://zuul.opendev.org/t/openstack/build/7bf0667f9081484caa22e7cceddb9b00 : SUCCESS in 55m 12s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b74d3ea6f0ee4b12be9d57f9c673cccc : SUCCESS in 7m 34s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/25d068cceb18463a96345910cb0cbda3 : SUCCESS in 1h 11m 40s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/4712e47f02a34d4199558c4b722342b6 : SUCCESS in 1h 09m 29s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/1dcd3d3608084cb78804bcdfdd793b42 : SUCCESS in 46m 30s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/a90ceafe0c524f5faeb7a226e72e4f48 : SUCCESS in 2h 20m 06s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/464ad803bed9468e896edb6abb3228f9 : SUCCESS in 1h 19m 54s\n- tripleo-tox-molecule https://zuul.opendev.org/t/openstack/build/8a78db3cfe664d3d92884ee30fa544fa : SUCCESS in 8m 02s","accounts_in_message":[],"_revision_number":1},{"id":"c0e282b1d1a67dc52816fa4f82cab6e62014ad33","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2021-03-19 18:32:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"19821def245080cb677f49b3fa6e0e48f4587316","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2021-03-19 18:33:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6d5b07b22aa6f94e0db240f2217f5a4b5075c7be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 18:34:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"640fdcba04c43420346ae33b0a69c29992902d29","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-20 01:44:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-build-containers-centos-7-buildah-stein https://zuul.opendev.org/t/openstack/build/55e284ce54334ff6908c0eaf857865eb : SUCCESS in 1h 02m 30s\n- tripleo-build-containers-centos-7-buildah-train https://zuul.opendev.org/t/openstack/build/2852794cacf34603b8ea4af0dec8efd6 : SUCCESS in 55m 40s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/e4eb2db0314e49fca26d617a70bfbae7 : SUCCESS in 1h 23m 07s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/da567d8a60934a1fa7d50aa934fc127d : SUCCESS in 1h 26m 29s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/cc2053656572454cba9cbf0b7f95aae9 : SUCCESS in 1h 11m 50s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/de7aeeca9f14496ca69867abf20f565b : SUCCESS in 2h 28m 32s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/597f8b66571d4d64a5b8bdaeef1d7705 : SUCCESS in 1h 17m 14s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9de1a43895624cf7b62336bd529b195d : SUCCESS in 7m 09s\n- tripleo-tox-molecule https://zuul.opendev.org/t/openstack/build/884004e6568545d994a22fb2c546f841 : SUCCESS in 9m 49s","accounts_in_message":[],"_revision_number":1},{"id":"d58746afef27598aafe3d53d8f86d8e2d91da566","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-20 01:44:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"048dfa0a31beacf0450c89efc4da081bcba7f361","revisions":{"048dfa0a31beacf0450c89efc4da081bcba7f361":{"kind":"REWORK","_number":1,"created":"2021-03-19 13:44:42.000000000","uploader":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"ref":"refs/changes/21/781721/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-ci","ref":"refs/changes/21/781721/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/21/781721/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/21/781721/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/21/781721/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-ci refs/changes/21/781721/1"}}},"commit":{"parents":[{"commit":"cb091f095559e0ec8f8f322c00661d15e1080a00","subject":"Merge \"tempest is only removed in master/17\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/cb091f095559e0ec8f8f322c00661d15e1080a00"}]}],"author":{"name":"Ronelle Landy","email":"rlandy@redhat.com","date":"2021-03-19 13:43:17.000000000","tz":-240},"committer":{"name":"Ronelle Landy","email":"rlandy@redhat.com","date":"2021-03-19 13:43:17.000000000","tz":-240},"subject":"Add option to include labels in dockerfiles","message":"Add option to include labels in dockerfiles\n\nWhen building containers, add labels to the\ndockerfiles using the --label argument if\ndockerfile_labels is defined.\n\nChange-Id: Ib797bde92a0328e0c1baf3779c8a047729979e5a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/048dfa0a31beacf0450c89efc4da081bcba7f361"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/048dfa0a31beacf0450c89efc4da081bcba7f361"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
