)]}'
{"id":"openstack%2Ftripleo-ci~791908","triplet_id":"openstack%2Ftripleo-ci~master~I8d486945a88bacd9729eb5b9333c7f9d31a945ad","project":"openstack/tripleo-ci","branch":"master","topic":"tripleo-get-hash","hashtags":[],"change_id":"I8d486945a88bacd9729eb5b9333c7f9d31a945ad","subject":"Adds tripleo-repos in the tripleo-ci-base-required-projects","status":"MERGED","created":"2021-05-18 09:14:09.000000000","updated":"2021-06-30 21:56:06.000000000","submitted":"2021-06-30 21:56:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"791908-tripleo-get-hash","meta_rev_id":"f2cbcb200e605b0fc1987686a6509091ef8347d9","_number":791908,"virtual_id_number":791908,"owner":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-30 21:56:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-05-18 12:19:02.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"all":[{"value":0,"date":"2021-05-28 10:05:44.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":2,"date":"2021-05-20 20:18:33.000000000","_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":1,"date":"2021-06-03 15:00:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":2,"date":"2021-06-30 15:25:25.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":1,"date":"2021-06-30 15:25:25.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-18 12:19:02.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-18 12:54:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-20 20:18:33.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2021-06-03 15:00:03.000000000","updated_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"reviewer":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"state":"REVIEWER"},{"updated":"2021-06-30 15:25:25.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"2a0fa53b49b7975f742a48b6b9fc464dca0c9419","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-05-18 09:14:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a05c693b9b19ef27707f3b2af1182990f5aa2a76","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-05-18 12:19:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/e89be1bd0b104a739cf60c1746ac7b54 : SUCCESS in 2h 33m 21s (non-voting)\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/b41e330f68e24b3f9ed8760910834bce : SUCCESS in 2h 42m 38s","accounts_in_message":[],"_revision_number":1},{"id":"9c86f084aeab2a9a8a0ab05d3ce4d8add1d8e4d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 12:54:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-7-buildah-train https://zuul.opendev.org/t/openstack/build/771970bc5fda4b21956032adfcefdd7c : SUCCESS in 55m 38s\n- tripleo-build-containers-centos-7-buildah-stein https://zuul.opendev.org/t/openstack/build/039ef62628144890afa0e8c368639e20 : SUCCESS in 49m 59s\n- tripleo-buildimage-overcloud-full-centos-7-train https://zuul.opendev.org/t/openstack/build/0d3ff4a26f7a466fb7b49dd607e7c5f4 : SUCCESS in 41m 52s\n- tripleo-buildimage-ironic-python-agent-centos-7-train https://zuul.opendev.org/t/openstack/build/33d607fe8eb9423aa2b80ddcd0725aa8 : SUCCESS in 31m 42s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/297476c1effd4b6ca7b854fadd3cf433 : SUCCESS in 51m 31s\n- tripleo-buildimage-overcloud-full-centos-8-wallaby https://zuul.opendev.org/t/openstack/build/09f8b6b4fd294f1db7f36347b05a01f3 : SUCCESS in 49m 47s\n- tripleo-buildimage-overcloud-full-centos-8-victoria https://zuul.opendev.org/t/openstack/build/07738133670045b789124114229c20a5 : SUCCESS in 50m 45s\n- tripleo-buildimage-overcloud-full-centos-8-ussuri https://zuul.opendev.org/t/openstack/build/95c68751e6664eabbede7108ec50a0b4 : SUCCESS in 47m 43s\n- tripleo-buildimage-overcloud-full-centos-8-train https://zuul.opendev.org/t/openstack/build/8c9f0857ce7d4ecaa40718472dfde662 : SUCCESS in 56m 13s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/57857820782e4b6a99ff3e180ba4b80c : SUCCESS in 29m 50s\n- tripleo-buildimage-ironic-python-agent-centos-8-wallaby https://zuul.opendev.org/t/openstack/build/c9fa9584929148968e37f1f044c2ec18 : SUCCESS in 26m 33s\n- tripleo-buildimage-ironic-python-agent-centos-8-victoria https://zuul.opendev.org/t/openstack/build/0e22e897a37d41359b5c1c8e6ef7ec5f : SUCCESS in 29m 20s\n- tripleo-buildimage-ironic-python-agent-centos-8-ussuri https://zuul.opendev.org/t/openstack/build/0366d5f8001e4b2b92d81ce8d2ef9f56 : SUCCESS in 26m 44s\n- tripleo-buildimage-ironic-python-agent-centos-8-train https://zuul.opendev.org/t/openstack/build/9d3fa74e0f794d7e8fbc212d1934b71d : SUCCESS in 23m 03s\n- tripleo-buildimage-overcloud-hardened-full-centos-8 https://zuul.opendev.org/t/openstack/build/2229cfebf73e473b9d72dfaac9aa71bf : SUCCESS in 34m 29s\n- tripleo-buildimage-overcloud-hardened-full-centos-8-wallaby https://zuul.opendev.org/t/openstack/build/a41791e54f95403eb4bca9843805ad0e : SUCCESS in 30m 34s\n- tripleo-buildimage-overcloud-hardened-full-centos-8-victoria https://zuul.opendev.org/t/openstack/build/c6d7d753607c4d5f837f729219b79a62 : SUCCESS in 33m 38s\n- tripleo-buildimage-overcloud-hardened-full-centos-8-ussuri https://zuul.opendev.org/t/openstack/build/c4b03171812b45dabac2817cae934088 : SUCCESS in 25m 25s\n- tripleo-buildimage-overcloud-hardened-full-centos-8-train https://zuul.opendev.org/t/openstack/build/a2801fe688eb4b4f8b75649a45017caa : SUCCESS in 32m 10s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/edc7c65d8dba4b4e8c16c85900b18735 : SUCCESS in 6m 59s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/8a5e69115cc644f48129662518e51224 : SUCCESS in 3h 09m 31s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/85f8f0beebfa43fc9f1e7851f4f7fa0c : SUCCESS in 3h 28m 06s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/55930ff0bfbc4338a563321b0b69764a : SUCCESS in 3h 22m 43s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/0c5af2009e184d568386e570242ee3f9 : SUCCESS in 3h 19m 02s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/4549e88e6a594d3a9a0fc431b2a51095 : SUCCESS in 1h 48m 34s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/84968e79daab45f481148a99da789981 : SUCCESS in 2h 06m 00s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/df87c1adc54b4725a1fd3a1b58ba0d85 : SUCCESS in 2h 04m 25s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/04025b5338ff440c947abb6758e2ec84 : SUCCESS in 2h 09m 35s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/e77b3ce8f12540bc9ffd405b217022e2 : SUCCESS in 3h 18m 13s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/a8367b291be049df97776e57c2f3a1e1 : SUCCESS in 1h 56m 04s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/83c9ec50654c4ecfb81477aee41e4851 : SUCCESS in 1h 46m 50s\n- tripleo-ci-centos-8-containers-undercloud-minion https://zuul.opendev.org/t/openstack/build/bc03b6f51be34e5dbfe5cb7748526087 : SUCCESS in 1h 15m 13s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/8963d87a05fc43979aacf98154cd190e : SUCCESS in 1h 43m 43s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/fe99fa3d5c004d559e008dae6f583f3e : SUCCESS in 1h 41m 29s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/ed3d1ab89145417d8972f02fd8fda4ca : SUCCESS in 1h 22m 39s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/756e593522e4453f86ac14a870fe1e37 : SUCCESS in 1h 48m 12s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/7f88b9c1aba74aae99e6668faabdfcd2 : SUCCESS in 1h 10m 28s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/6526e78f837f49fcaf209601df0e6842 : FAILURE in 1h 57m 30s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/c7ceea9a357e480f84967db2ad5d76ef : SUCCESS in 1h 33m 33s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/11cf803cf97a4ea4a7e7ea753d7542e1 : SUCCESS in 1h 25m 03s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/8294f8d90e934850bc1dc066caa82583 : SUCCESS in 1h 39m 50s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/d40b985d96044f55857454eaa86eb17a : SUCCESS in 1h 14m 44s\n- tripleo-tox-molecule https://zuul.opendev.org/t/openstack/build/fb374e9279db4ecf8a71caeba6afe999 : SUCCESS in 11m 13s","accounts_in_message":[],"_revision_number":1},{"id":"ecb23cd8e21d4732271b9e0742d4530b3c116f6b","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2021-05-20 20:18:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e6fca4e28e7c562f165c77019d23ea099e31fad6","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-05-21 10:43:08.000000000","message":"Patch Set 1: Code-Review-1\n\nWIP thanks for voting Wes, not sure if we want this one or the sudo python setup.py install let\u0027s discuss?","accounts_in_message":[],"_revision_number":1},{"id":"1ceb0b66fba807911e8e9144ea28594d69311de0","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-05-28 10:05:44.000000000","message":"Patch Set 1: -Code-Review\n\nyes we do want that","accounts_in_message":[],"_revision_number":1},{"id":"63ce527b1baacf4c32a51db27e44e97ad7e6c670","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-05-28 10:15:54.000000000","message":"Topic set to tripleo-get-hash","accounts_in_message":[],"_revision_number":1},{"id":"84907996ab3ea404e7d6ac068c15300cad727902","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2021-06-03 15:00:03.000000000","message":"Patch Set 1: Code-Review+1\n\nAdding companion downstream job","accounts_in_message":[],"_revision_number":1},{"id":"a6d35324ff7d1ae80c18b4682e587d9af7bb1dcc","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-06-30 15:25:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ba1dae86164240b71008b73362bf68e13e80e6a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 15:25:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"126c5cac17d9b58ad6563fe73eb0104ce15d5bbd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 21:56:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-build-containers-centos-7-buildah-stein https://zuul.opendev.org/t/openstack/build/d4e117f7e6de4815ae890aaf1c6008de : SUCCESS in 1h 12m 12s\n- tripleo-build-containers-centos-7-buildah-train https://zuul.opendev.org/t/openstack/build/fed9f77aff894ad6a10bf00e5fe324fb : SUCCESS in 1h 13m 59s\n- tripleo-buildimage-overcloud-full-centos-7-train https://zuul.opendev.org/t/openstack/build/38b8d3ba345a44e8a05e8167105c1028 : SUCCESS in 43m 36s\n- tripleo-buildimage-ironic-python-agent-centos-7-train https://zuul.opendev.org/t/openstack/build/70c7c1dd0dce4cdc9d9a42b5c68e9042 : SUCCESS in 33m 23s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/8891e7c4bc4044e0996450f320bc3751 : SUCCESS in 1h 16m 31s\n- tripleo-buildimage-overcloud-full-centos-8-wallaby https://zuul.opendev.org/t/openstack/build/0b0311d610ae4b5d8c2b1f883d107639 : SUCCESS in 1h 11m 54s\n- tripleo-buildimage-overcloud-full-centos-8-victoria https://zuul.opendev.org/t/openstack/build/d35ba5fb583244e49107c6f11adfd7d9 : SUCCESS in 1h 13m 20s\n- tripleo-buildimage-overcloud-full-centos-8-ussuri https://zuul.opendev.org/t/openstack/build/20f7254ab0ab47f5a7e2d8c91b6820b6 : SUCCESS in 1h 16m 32s\n- tripleo-buildimage-overcloud-full-centos-8-train https://zuul.opendev.org/t/openstack/build/97024c300aa6434c9ac38a28f282cf84 : SUCCESS in 1h 25m 32s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/c28200c7ddf049dbb56b24b9cd4a8298 : SUCCESS in 44m 09s\n- tripleo-buildimage-ironic-python-agent-centos-8-wallaby https://zuul.opendev.org/t/openstack/build/40b32f5769944a08bde6ac106ed9389b : SUCCESS in 42m 20s\n- tripleo-buildimage-ironic-python-agent-centos-8-victoria https://zuul.opendev.org/t/openstack/build/391ed28b35904861bb7e2e9e74ba591a : SUCCESS in 40m 04s\n- tripleo-buildimage-ironic-python-agent-centos-8-ussuri https://zuul.opendev.org/t/openstack/build/cdd2a9fb8f164534ac6ab6de480e8d70 : SUCCESS in 42m 14s\n- tripleo-buildimage-ironic-python-agent-centos-8-train https://zuul.opendev.org/t/openstack/build/11259e216a8741bc880cd74192f02420 : SUCCESS in 42m 35s\n- tripleo-buildimage-overcloud-hardened-full-centos-8 https://zuul.opendev.org/t/openstack/build/ba7fc87174664b24b22985bcfdc06854 : SUCCESS in 49m 42s\n- tripleo-buildimage-overcloud-hardened-full-centos-8-wallaby https://zuul.opendev.org/t/openstack/build/7e229a1bba6e49eb9696c3908c9f96c6 : SUCCESS in 38m 45s\n- tripleo-buildimage-overcloud-hardened-full-centos-8-victoria https://zuul.opendev.org/t/openstack/build/3f35f147c6a8449391ac401b3e7adf4d : SUCCESS in 40m 10s\n- tripleo-buildimage-overcloud-hardened-full-centos-8-ussuri https://zuul.opendev.org/t/openstack/build/a03c5e69f8cb4ceb93b7e070b480d5d0 : SUCCESS in 36m 37s\n- tripleo-buildimage-overcloud-hardened-full-centos-8-train https://zuul.opendev.org/t/openstack/build/2263ee0e9f924c519c479fa259bec65b : SUCCESS in 37m 35s\n- tripleo-buildimage-overcloud-hardened-uefi-full-centos-8 https://zuul.opendev.org/t/openstack/build/453e3d9298a849128d2f4466f23b3487 : SUCCESS in 52m 36s\n- tripleo-buildimage-overcloud-hardened-uefi-full-centos-8-wallaby https://zuul.opendev.org/t/openstack/build/e8f68b0185a549fbb3086a3e594b3fdf : SUCCESS in 44m 08s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/4459c8c529f24474a90d99111903c9ec : SUCCESS in 3h 27m 00s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/1fbd170eee924920ab5f30bd6671c739 : SUCCESS in 3h 47m 03s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/f9ec08c4b0d342f5be39a64a3af70e48 : SUCCESS in 3h 49m 43s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/11ed1ab6987741bfb8eaf5ba6a63e610 : SUCCESS in 3h 41m 56s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/8945d6ad88a64dd885afce6cc1a48b26 : SUCCESS in 2h 02m 35s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/091fcbfd16a94beaa4d3327d23513ca6 : SUCCESS in 2h 41m 15s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/f4d8b81eb5ce43d6a96e0b0b01237418 : SUCCESS in 2h 15m 16s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/834f39abf2294d189f9ae55a2913db54 : SUCCESS in 2h 36m 47s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/cb3a78c5b2e640fb907e5299430a3662 : SUCCESS in 3h 00m 48s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/097416c86c14451cb55403a16816f736 : SUCCESS in 1h 54m 49s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/95532cfd15b04100b899577c9455bdf8 : SUCCESS in 1h 51m 59s\n- tripleo-ci-centos-8-containers-undercloud-minion https://zuul.opendev.org/t/openstack/build/0320ad3c2fd8488dbef8b696091f44ea : SUCCESS in 1h 34m 54s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/85064fe51ad842fbae822a6c68f54f73 : SUCCESS in 1h 39m 07s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/906049886e9643a4b7741935bed5ec04 : SUCCESS in 1h 52m 51s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/cddd32edba0a43c7a8d50d246044714d : SUCCESS in 54m 28s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/d31ffc00b92e4ced9a85eeda59cc1550 : SUCCESS in 1h 07m 16s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/4c8bb48475cf44a78beabc01d75e2b44 : SUCCESS in 56m 58s\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/d3b434bf6afb491d9220197b2e699274 : SUCCESS in 1h 06m 03s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/0b05a0117da941a2bdec5bf8ad78d0c1 : SUCCESS in 1h 35m 48s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/949220ca1b45451b9bd51c783f998f1f : SUCCESS in 53m 54s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/70a464c8aad14768b62af9c825ad625e : SUCCESS in 6m 01s\n- tripleo-tox-molecule https://zuul.opendev.org/t/openstack/build/c26b90e9d9c649dcac1762e58177b29f : SUCCESS in 13m 04s","accounts_in_message":[],"_revision_number":1},{"id":"f2cbcb200e605b0fc1987686a6509091ef8347d9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 21:56:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d0015e998580260bd127a8a2f9126fbca277d35e","revisions":{"d0015e998580260bd127a8a2f9126fbca277d35e":{"kind":"REWORK","_number":1,"created":"2021-05-18 09:14:09.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/08/791908/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-ci","ref":"refs/changes/08/791908/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/08/791908/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/08/791908/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/08/791908/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-ci refs/changes/08/791908/1"}}},"commit":{"parents":[{"commit":"a9db95204b23465972a365be5493521f490d8c93","subject":"Merge \"Revert \"Temp: containers-multinode-wallaby non-voting\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/a9db95204b23465972a365be5493521f490d8c93"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-05-18 09:12:08.000000000","tz":180},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-05-18 09:12:08.000000000","tz":180},"subject":"Adds tripleo-repos in the tripleo-ci-base-required-projects","message":"Adds tripleo-repos in the tripleo-ci-base-required-projects\n\nThis adds openstack/tripleo-repos to the zuul job definition\nrequired-projects. Used by [1]\n\n[1] https://review.opendev.org/c/openstack/tripleo-quickstart/+/791486\n\nChange-Id: I8d486945a88bacd9729eb5b9333c7f9d31a945ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/d0015e998580260bd127a8a2f9126fbca277d35e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/d0015e998580260bd127a8a2f9126fbca277d35e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
