)]}'
{"id":"openstack%2Ftripleo-ci~817455","triplet_id":"openstack%2Ftripleo-ci~master~I382ed5ff5d931325c786b43a3acd01084dc43b95","project":"openstack/tripleo-ci","branch":"master","topic":"cs9","attention_set":{},"removed_from_attention_set":{"8449":{"account":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"last_update":"2021-11-11 10:08:15.000000000","reason":"Marios Andreou replied on the change"},"12393":{"account":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"last_update":"2021-12-14 09:16:58.000000000","reason":"Change was abandoned"},"29775":{"account":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"last_update":"2021-12-14 09:16:58.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I382ed5ff5d931325c786b43a3acd01084dc43b95","subject":"Lay down EL9 compose repos before TQ starts","status":"ABANDONED","created":"2021-11-10 16:15:25.000000000","updated":"2021-12-14 09:16:58.000000000","total_comment_count":6,"unresolved_comment_count":3,"has_review_started":true,"meta_rev_id":"0093524ecfa6a8734463d66ea1ccc90ceaee7db0","_number":817455,"virtual_id_number":817455,"owner":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-11-11 11:09:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2021-11-11 11:09:43.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"all":[{"value":-1,"date":"2021-11-11 11:08:22.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"}],"REVIEWER":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-10 16:37:09.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"state":"REVIEWER"},{"updated":"2021-11-10 16:37:09.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2021-11-10 19:41:59.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-10 21:13:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-11 14:04:57.000000000","updated_by":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"reviewer":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"state":"CC"}],"messages":[{"id":"72a9096ba89d5c6a8b9641be2e77ce8d1e1dadb1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-11-10 16:15:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e31eeb175ef7c0c60b9397c390b78b5bdf655066","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-11-10 16:37:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0d99ff394c39e7047a7441aa9171595c7e18d5df","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-11-10 19:41:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/705260efc87b4df79dc29b79d4a98c54 : SUCCESS in 3h 01m 04s (non-voting)\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/6892143442a846beafa8aae8cc900561 : FAILURE in 45m 38s","accounts_in_message":[],"_revision_number":1},{"id":"134484f998bde5809c93c87a4c8e97430925cb07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-10 21:13:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3117c84928d84d718fc6ca565a5b63f6 : SUCCESS in 6m 56s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/20b1666945a548e695c1ef3cb586b98f : SUCCESS in 3h 53m 26s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/28d4ea6ca1644e8fb9706242963e633d : SUCCESS in 3h 55m 34s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/cf9b947eb7774347a6d04710edd77a43 : SUCCESS in 3h 29m 53s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/f9b03634e74f45e080efc42633165f19 : SUCCESS in 3h 02m 22s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/551d048f79e64969a2cb124a3383e9a3 : SUCCESS in 2h 08m 15s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/c8a99c5f3f4f480286d2c31de20a30ff : SUCCESS in 2h 03m 37s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/ed39b07b4f414fbd9bdf215a529a6c23 : SUCCESS in 1h 52m 49s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/d1ac60e166a44346855c1c940f5e6f80 : FAILURE in 2h 02m 16s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/4a94bd1e53a84878852d1c586587e854 : SUCCESS in 3h 48m 18s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/ffb1293ae8b44ed79831383703f4c25b : SUCCESS in 1h 40m 58s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/771f9345e4964a4688cd6de2b3c2fc2e : SUCCESS in 1h 35m 12s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/0a6c07ade3164fecb4ccba41f0a94a63 : SUCCESS in 1h 32m 57s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/4d1531f838e0461bb990a8b16c7d291c : SUCCESS in 51m 53s\n- tripleo-tox-molecule https://zuul.opendev.org/t/openstack/build/72393d7535874d60a03af271852b07e5 : SUCCESS in 18m 19s","accounts_in_message":[],"_revision_number":1},{"id":"e6a6b50804fb6c81ecaa540835946cb706011890","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-11-11 06:59:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b14042a8d0552fb9678fb9ee056caa6ede2880b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-11-11 09:13:19.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5aec9993aaba65ac1e8f74adb3c8f677f86c4af0","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-11-11 10:08:15.000000000","message":"Patch Set 3: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"4643f69c341b0db1e352e037814baa3cefda27b5","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-11-11 11:08:22.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"f53c87d5873a54d5e81a9f622acb6485bfc529eb","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-11-11 11:09:43.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"1cf4b2cf2cf42374cf2b96dcd5aa9818c165ed1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 11:09:45.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"0093524ecfa6a8734463d66ea1ccc90ceaee7db0","tag":"autogenerated:gerrit:abandon","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-12-14 09:16:58.000000000","message":"Abandoned\n\nhttps://review.opendev.org/q/topic:%22use_release_file%22+(status:open%20OR%20status:merged)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"41fb523c56dfb75ea39bba12a1edf4a209b4bd92","revisions":{"28f7280b21a79a5e8ad153051e73359ad442e19f":{"kind":"REWORK","_number":1,"created":"2021-11-10 16:15:25.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/55/817455/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-ci","ref":"refs/changes/55/817455/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/55/817455/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/55/817455/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/55/817455/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-ci refs/changes/55/817455/1"}}},"commit":{"parents":[{"commit":"ebeddec8da97a42b6801cbb57054dd6f6f8fb2eb","subject":"Wire up tripleo-repos to populate dlrn repos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/ebeddec8da97a42b6801cbb57054dd6f6f8fb2eb"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-11-10 16:14:48.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-11-10 16:14:48.000000000","tz":330},"subject":"[WIP] use tripleo-repos in CS9 main deployment","message":"[WIP] use tripleo-repos in CS9 main deployment\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I382ed5ff5d931325c786b43a3acd01084dc43b95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/28f7280b21a79a5e8ad153051e73359ad442e19f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/28f7280b21a79a5e8ad153051e73359ad442e19f"}]},"branch":"refs/heads/master"},"7f3a941d85dc1940a7a60640c903cbffd428b8c7":{"kind":"REWORK","_number":2,"created":"2021-11-11 06:59:41.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/55/817455/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-ci","ref":"refs/changes/55/817455/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/55/817455/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/55/817455/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/55/817455/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-ci refs/changes/55/817455/2"}}},"commit":{"parents":[{"commit":"ebeddec8da97a42b6801cbb57054dd6f6f8fb2eb","subject":"Wire up tripleo-repos to populate dlrn repos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/ebeddec8da97a42b6801cbb57054dd6f6f8fb2eb"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-11-10 16:14:48.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-11-11 06:59:15.000000000","tz":330},"subject":"[WIP] use tripleo-repos in CS9 main deployment","message":"[WIP] use tripleo-repos in CS9 main deployment\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I382ed5ff5d931325c786b43a3acd01084dc43b95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/7f3a941d85dc1940a7a60640c903cbffd428b8c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/7f3a941d85dc1940a7a60640c903cbffd428b8c7"}]},"branch":"refs/heads/master"},"41fb523c56dfb75ea39bba12a1edf4a209b4bd92":{"kind":"REWORK","_number":3,"created":"2021-11-11 09:13:19.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/55/817455/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-ci","ref":"refs/changes/55/817455/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/55/817455/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/55/817455/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/55/817455/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-ci refs/changes/55/817455/3"}}},"commit":{"parents":[{"commit":"ebeddec8da97a42b6801cbb57054dd6f6f8fb2eb","subject":"Wire up tripleo-repos to populate dlrn repos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/ebeddec8da97a42b6801cbb57054dd6f6f8fb2eb"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-11-10 16:14:48.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-11-11 09:03:31.000000000","tz":330},"subject":"Lay down EL9 compose repos before TQ starts","message":"Lay down EL9 compose repos before TQ starts\n\nIn order to test deployment on EL9, we need to prepare\nthe node.\n\nThe current nodepool image has centos.repo which pulls\npackages from CentOS mirror. But it is not always in good\nshape and it blocks the CI work.\n\nThis patch enables compose repos in the pre playbook.\nIt will help to move the deployment ahead.\n\nIt moves dlrn repos generation with in conditionals to avoid\nbreaking containers and images job.\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I382ed5ff5d931325c786b43a3acd01084dc43b95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/41fb523c56dfb75ea39bba12a1edf4a209b4bd92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/41fb523c56dfb75ea39bba12a1edf4a209b4bd92"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
