)]}'
{"id":"openstack%2Ftripleo-ci~826521","triplet_id":"openstack%2Ftripleo-ci~master~I303b32a3081c60739ad650ef5afc90e239a0d46e","project":"openstack/tripleo-ci","branch":"master","topic":"osp_trunk_https","attention_set":{},"removed_from_attention_set":{"8449":{"account":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"last_update":"2022-02-02 06:01:39.000000000","reason":"Change was submitted"},"9976":{"account":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"last_update":"2022-02-02 06:01:39.000000000","reason":"Change was submitted"},"10969":{"account":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"last_update":"2022-02-02 06:01:39.000000000","reason":"Change was submitted"},"29775":{"account":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"last_update":"2022-02-02 06:01:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I303b32a3081c60739ad650ef5afc90e239a0d46e","subject":"Adding vars for cert","status":"MERGED","created":"2022-01-26 20:09:27.000000000","updated":"2022-02-02 06:01:39.000000000","submitted":"2022-02-02 06:01:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":15,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"826521-osp_trunk_https","meta_rev_id":"24e52efe3a1113204a43d39a34120b14e0245aba","_number":826521,"virtual_id_number":826521,"owner":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":14611,"name":"dasm","display_name":"dasm","email":"dsmigiel@redhat.com","username":"dasm"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-02 06:01:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2022-02-01 18:41:16.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":2,"date":"2022-02-01 18:10:35.000000000","_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":1,"date":"2022-02-01 17:39:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14611,"name":"dasm","display_name":"dasm","email":"dsmigiel@redhat.com","username":"dasm"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":14611,"name":"dasm","display_name":"dasm","email":"dsmigiel@redhat.com","username":"dasm"},{"value":1,"date":"2022-02-01 18:26:50.000000000","_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"}],"reviewers":{"REVIEWER":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":14611,"name":"dasm","display_name":"dasm","email":"dsmigiel@redhat.com","username":"dasm"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-26 22:43:52.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-27 00:55:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-31 09:18:53.000000000","updated_by":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"reviewer":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"state":"REVIEWER"},{"updated":"2022-01-31 12:39:01.000000000","updated_by":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"reviewer":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"state":"REVIEWER"},{"updated":"2022-01-31 12:43:15.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2022-02-01 05:45:42.000000000","updated_by":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2022-02-01 17:39:38.000000000","updated_by":{"_account_id":14611,"name":"dasm","display_name":"dasm","email":"dsmigiel@redhat.com","username":"dasm"},"reviewer":{"_account_id":14611,"name":"dasm","display_name":"dasm","email":"dsmigiel@redhat.com","username":"dasm"},"state":"REVIEWER"}],"messages":[{"id":"5b190448552574fd470835e8aa6823472f00fbf5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2022-01-26 20:09:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"29b8f5d4c7ca35b1e3eb2d266beba4db68324c05","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2022-01-26 20:29:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a4fe2e33e06e1c003a96e5fd6393540e923deb3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2022-01-26 21:11:36.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"20acbab6955421238663adff622aa00cfee064d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2022-01-26 21:29:07.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a468a9cc38933d5587e01c3c8594a7e529012391","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2022-01-26 21:50:41.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"c5977bb94d1ba3095fc4daffe7867aa06af8be5a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-01-26 22:43:52.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/99eb453262374e2bb7273a6da2da1bd1 : FAILURE in 34m 10s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/437d7fcc84224f36be30f8feaa18511d : FAILURE in 51m 55s","accounts_in_message":[],"_revision_number":5},{"id":"16e397d2b757311190ab34f750b454e5c19450a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 00:55:09.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-7-buildah-train https://zuul.opendev.org/t/openstack/build/e91d337c3bbb44c58d62d98354315c35 : SUCCESS in 51m 21s\n- tripleo-build-containers-centos-9 https://zuul.opendev.org/t/openstack/build/c848cc0ae2cc44c69a6ff0d6ddb898af : SUCCESS in 37m 51s (non-voting)\n- tripleo-buildimage-overcloud-full-centos-7-train https://zuul.opendev.org/t/openstack/build/00e8df7c40f0437794201214ae835964 : SUCCESS in 42m 08s\n- tripleo-buildimage-ironic-python-agent-centos-7-train https://zuul.opendev.org/t/openstack/build/d10677161783496bb7c7bcd9d23531f2 : SUCCESS in 20m 46s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/0a36d14852bf4e98b51f4372300d8639 : SUCCESS in 50m 48s\n- tripleo-buildimage-overcloud-full-centos-8-wallaby https://zuul.opendev.org/t/openstack/build/dace8d4a33d948f4ad0e687490f70fc1 : SUCCESS in 56m 09s\n- tripleo-buildimage-overcloud-full-centos-8-victoria https://zuul.opendev.org/t/openstack/build/e49afefbc73d4c0fa8ef48d003b5ae5a : SUCCESS in 58m 28s\n- tripleo-buildimage-overcloud-full-centos-8-ussuri https://zuul.opendev.org/t/openstack/build/f62ec18bf0a541e39f1824c52f4dd73f : SUCCESS in 50m 47s\n- tripleo-buildimage-overcloud-full-centos-8-train https://zuul.opendev.org/t/openstack/build/a1d7d44721604ca5b2115b56e7cadd9b : SUCCESS in 46m 18s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/994b391ac2124abd96e99913f7ac9476 : SUCCESS in 31m 31s\n- tripleo-buildimage-ironic-python-agent-centos-8-wallaby https://zuul.opendev.org/t/openstack/build/23e25a022705421783370dac2429b65e : SUCCESS in 30m 14s\n- tripleo-buildimage-ironic-python-agent-centos-8-victoria https://zuul.opendev.org/t/openstack/build/b2d7637d1c7c4b2d8f342295dd3371fe : SUCCESS in 31m 25s\n- tripleo-buildimage-ironic-python-agent-centos-8-ussuri https://zuul.opendev.org/t/openstack/build/2c3c6c85a89b447fb10bbcd2f7476c5e : SUCCESS in 27m 09s\n- tripleo-buildimage-ironic-python-agent-centos-8-train https://zuul.opendev.org/t/openstack/build/01a655dbbc0a4c9d8ab50b1af05c027c : SUCCESS in 31m 59s\n- tripleo-buildimage-overcloud-hardened-uefi-full-centos-8 https://zuul.opendev.org/t/openstack/build/e6012845991a462780c353fa2eb89af4 : SUCCESS in 51m 47s\n- tripleo-buildimage-overcloud-hardened-uefi-full-centos-8-wallaby https://zuul.opendev.org/t/openstack/build/b53501763dbd4872b689e72fec3a25f2 : SUCCESS in 58m 17s\n- tripleo-buildimage-overcloud-full-centos-9 https://zuul.opendev.org/t/openstack/build/de6f10ec92e447e7ab8b3331a3227c59 : SUCCESS in 33m 49s\n- tripleo-buildimage-ironic-python-agent-centos-9 https://zuul.opendev.org/t/openstack/build/0e81fbdc47354c949b834daa071f0292 : SUCCESS in 40m 46s\n- tripleo-buildimage-overcloud-hardened-uefi-full-centos-9 https://zuul.opendev.org/t/openstack/build/549571b352a54a7e9b82113662c155dc : SUCCESS in 36m 33s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/008c8ee9edeb4ceebba47c6651d0f41f : SUCCESS in 7m 11s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/99937e7b681c4238aa90cf4d62f4deb3 : SUCCESS in 2h 18m 57s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/9eb2fa0c07cd419c8b45a9e272c66657 : SUCCESS in 2h 47m 43s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/42f0fb7dd6cd4683904720d6544f86b9 : SUCCESS in 1h 20m 59s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/633e7642c8624105b8bdebd44cdab24b : SUCCESS in 1h 14m 05s\n- tripleo-tox-molecule https://zuul.opendev.org/t/openstack/build/a4a04284bba94fc783e8ff5fcace8515 : SUCCESS in 18m 03s","accounts_in_message":[],"_revision_number":5},{"id":"296a41fde0f2c4b9a7d49cc8e39c11e8eca00ae0","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2022-01-31 09:18:53.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"1705f1f731f7c38e6ab60ea91966b270f7ac51b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2022-01-31 09:19:29.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"1851c18cfea330b44fe27bbf9bd2ce92c7a162bb","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-01-31 11:00:20.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/1467c31ef6d94c0f93a05db43064ef8b : FAILURE in 31m 11s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/789dda21991c486f9b6ab5dc7a59f097 : FAILURE in 1h 39m 08s","accounts_in_message":[],"_revision_number":6},{"id":"f8503bc0da1431c75cc212cb6def88664d245d6d","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2022-01-31 12:39:01.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"3783e93d3e24606a00c46102ecd1dd5fe9b87beb","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2022-01-31 12:43:15.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"004d64680378ea00681c10edd50aa22d2b239a36","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2022-01-31 12:59:33.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"698ead7fa31c653ad7b37e278be2a2229b1f37b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2022-01-31 13:32:42.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"7f497eedf0ecc253042fdf88a8fd22b776918f30","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2022-01-31 13:33:07.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"f20d145a776a4e95772572b018d4c37a11cd5fa0","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2022-01-31 14:45:58.000000000","message":"Patch Set 7: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"9174dfb9b4cb1b5f747a564dfe159d5b2e5cf384","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2022-01-31 14:46:19.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"e0ca4395bf1a48522eff19e25724ceb4bd931e9c","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2022-01-31 14:46:37.000000000","message":"Removed Code-Review+2 by Douglas Viroel \u003cviroel@gmail.com\u003e\n","accounts_in_message":[],"_revision_number":7},{"id":"bd1b56d3a1a170165b7080d6ca79cb04b0c98bbd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2022-01-31 14:52:56.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"e084ccfb98fe497d79d51069c0eab627defbf5f7","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2022-01-31 14:53:46.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"673c116d29f4f497d2a8b2ec2839bf97fed4c402","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2022-01-31 15:21:32.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"bb0c05c32f7a3fd976043211f59f09ad5592aa3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2022-01-31 15:44:11.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"812e50aff795a7cb2fee7619b5feca041b87fcdc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2022-01-31 17:06:59.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"9fe48448b4c4c6cd58ff9a9965d2710660f9deea","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-01-31 19:00:02.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/ad3bcbd4ad344d1fb1eeddf5a27c8fe3 : FAILURE in 36m 01s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/c6fae10e8e9d450fa1122be372b4d77d : FAILURE in 1h 51m 02s","accounts_in_message":[],"_revision_number":11},{"id":"291eaa68af013ed773037f62b5ce6360992329cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-31 20:56:46.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-build-containers-centos-7-buildah-train https://zuul.opendev.org/t/openstack/build/32ef3a1eda5648b499375b56f7bc0ff1 : SUCCESS in 49m 35s\n- tripleo-build-containers-centos-9 https://zuul.opendev.org/t/openstack/build/324527fad7214b9db38e9a62f37d80fa : SUCCESS in 41m 58s (non-voting)\n- tripleo-buildimage-overcloud-full-centos-7-train https://zuul.opendev.org/t/openstack/build/6b3704d8784246ea82e1c87f25c48244 : SUCCESS in 40m 29s\n- tripleo-buildimage-ironic-python-agent-centos-7-train https://zuul.opendev.org/t/openstack/build/916fe4fd0f0549569bb78c70fca3d703 : SUCCESS in 20m 38s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/07d201ac243949e2ad16dcf2dbc10dc7 : SUCCESS in 1h 12m 27s\n- tripleo-buildimage-overcloud-full-centos-8-wallaby https://zuul.opendev.org/t/openstack/build/30cd5961d1ee4fd0b306a6d7781b33bd : SUCCESS in 59m 05s\n- tripleo-buildimage-overcloud-full-centos-8-victoria https://zuul.opendev.org/t/openstack/build/6fa6656c5bd94ab1a6f7e239f9fabe9b : SUCCESS in 1h 04m 19s\n- tripleo-buildimage-overcloud-full-centos-8-ussuri https://zuul.opendev.org/t/openstack/build/f0cf597fb7f8433a82be8b2f023d5e52 : TIMED_OUT in 1h 34m 47s\n- tripleo-buildimage-overcloud-full-centos-8-train https://zuul.opendev.org/t/openstack/build/467dd88ae3454654a12f815a21c5c1d4 : SUCCESS in 1h 04m 35s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/3ce0661aa4834e2b8f6fde76059c8123 : SUCCESS in 33m 09s\n- tripleo-buildimage-ironic-python-agent-centos-8-wallaby https://zuul.opendev.org/t/openstack/build/774ac2224bd949fe9928702d182ddfff : SUCCESS in 54m 02s\n- tripleo-buildimage-ironic-python-agent-centos-8-victoria https://zuul.opendev.org/t/openstack/build/f787013e219242a5b8769e7a541f6db3 : SUCCESS in 32m 32s\n- tripleo-buildimage-ironic-python-agent-centos-8-ussuri https://zuul.opendev.org/t/openstack/build/d3eee11be8944148a209bd7b528efda7 : SUCCESS in 33m 10s\n- tripleo-buildimage-ironic-python-agent-centos-8-train https://zuul.opendev.org/t/openstack/build/b6d31d08a0fe4ce9b152ce06d77e6056 : SUCCESS in 44m 59s\n- tripleo-buildimage-overcloud-hardened-uefi-full-centos-8 https://zuul.opendev.org/t/openstack/build/07fc41c31d194ae685438ce58e5ef2f3 : SUCCESS in 45m 34s\n- tripleo-buildimage-overcloud-hardened-uefi-full-centos-8-wallaby https://zuul.opendev.org/t/openstack/build/3b4f3489709a4aa1b2d72216bd635957 : SUCCESS in 55m 44s\n- tripleo-buildimage-overcloud-full-centos-9 https://zuul.opendev.org/t/openstack/build/c97f64804fdf43f6b0307ac943b14b32 : SUCCESS in 40m 24s\n- tripleo-buildimage-ironic-python-agent-centos-9 https://zuul.opendev.org/t/openstack/build/5cb7bb1ee0e44530b8f5311920209afc : SUCCESS in 38m 07s\n- tripleo-buildimage-overcloud-hardened-uefi-full-centos-9 https://zuul.opendev.org/t/openstack/build/547f7e3642494085b86499e7b5074e0e : SUCCESS in 42m 49s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a723d5501195410e817a44e0f3085594 : FAILURE in 8m 07s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/b359f61d31fb4504b7db4a86e12afa43 : SUCCESS in 2h 27m 30s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/1ef564021e1e4020974280a8d6e9760b : SKIPPED\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/2fd51af7f3a74bc38fdef435d824a6fc : SKIPPED\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/574be8743d0a4f2f9e62ab3825ceff48 : SUCCESS in 1h 20m 46s\n- tripleo-tox-molecule https://zuul.opendev.org/t/openstack/build/d09ba64a2b654a7ba1580bdbdaa1b1f0 : FAILURE in 13m 49s","accounts_in_message":[],"_revision_number":11},{"id":"d68b4823aba9a87ef80021b4ce8d397347117f7c","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2022-02-01 00:11:27.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"c7f957cb8f7ae2a6325b9c5af413702709fdc82e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 01:49:11.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-build-containers-centos-7-buildah-train https://zuul.opendev.org/t/openstack/build/419855b111fd49e0bfabca9c403ad1cf : SUCCESS in 52m 17s\n- tripleo-build-containers-centos-9 https://zuul.opendev.org/t/openstack/build/ca111621b0774a27aa117496227534c2 : SUCCESS in 32m 52s (non-voting)\n- tripleo-buildimage-overcloud-full-centos-7-train https://zuul.opendev.org/t/openstack/build/2638570e2b5141b78d57af20063c9129 : SUCCESS in 1h 04m 42s\n- tripleo-buildimage-ironic-python-agent-centos-7-train https://zuul.opendev.org/t/openstack/build/5b696e9eda4144e0905d9a79a0826bb3 : SUCCESS in 35m 33s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/f6f890c7719f4807a624bfc240c3499e : SUCCESS in 1h 16m 54s\n- tripleo-buildimage-overcloud-full-centos-8-wallaby https://zuul.opendev.org/t/openstack/build/89c00cd5107f4e3f8914cbe7e3ca1bc4 : SUCCESS in 46m 30s\n- tripleo-buildimage-overcloud-full-centos-8-victoria https://zuul.opendev.org/t/openstack/build/e403f5c1e16242dda864b13eb9adcfa9 : SUCCESS in 57m 19s\n- tripleo-buildimage-overcloud-full-centos-8-ussuri https://zuul.opendev.org/t/openstack/build/bb73dd60641d44b0aefde8e1fc3f75d2 : SUCCESS in 1h 04m 32s\n- tripleo-buildimage-overcloud-full-centos-8-train https://zuul.opendev.org/t/openstack/build/78ba1195f6554755afb612cb1bfd7ab7 : SUCCESS in 1h 01m 56s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/9a5a2fea17144a9a8cd9a3602ca07a70 : SUCCESS in 36m 06s\n- tripleo-buildimage-ironic-python-agent-centos-8-wallaby https://zuul.opendev.org/t/openstack/build/243d3497b64f441597c94d9429a3806a : SUCCESS in 31m 53s\n- tripleo-buildimage-ironic-python-agent-centos-8-victoria https://zuul.opendev.org/t/openstack/build/357ea46ed709432a92336a82386640e5 : SUCCESS in 35m 55s\n- tripleo-buildimage-ironic-python-agent-centos-8-ussuri https://zuul.opendev.org/t/openstack/build/cf61b5be6ca74cc2901265f76a5f09a9 : SUCCESS in 25m 11s\n- tripleo-buildimage-ironic-python-agent-centos-8-train https://zuul.opendev.org/t/openstack/build/5dfd07a45f724ff1a200a042ea43d618 : SUCCESS in 34m 59s\n- tripleo-buildimage-overcloud-hardened-uefi-full-centos-8 https://zuul.opendev.org/t/openstack/build/a8b116909bda45dfade817787435252c : SUCCESS in 52m 30s\n- tripleo-buildimage-overcloud-hardened-uefi-full-centos-8-wallaby https://zuul.opendev.org/t/openstack/build/0889b6beb44c4b8ab150bec66e47f35d : SUCCESS in 47m 09s\n- tripleo-buildimage-overcloud-full-centos-9 https://zuul.opendev.org/t/openstack/build/f5df68ac439844e882bfffafe82c427b : SUCCESS in 46m 04s\n- tripleo-buildimage-ironic-python-agent-centos-9 https://zuul.opendev.org/t/openstack/build/d53335646da34fcf923667770285cf36 : SUCCESS in 27m 11s\n- tripleo-buildimage-overcloud-hardened-uefi-full-centos-9 https://zuul.opendev.org/t/openstack/build/b4df485fd3a345de8e4a2f07c58ca740 : SUCCESS in 37m 21s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fe967a1e04714e2480268eeb436f7f96 : FAILURE in 5m 49s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/047bd53190434c5797322d089f46a379 : SUCCESS in 1h 31m 49s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/8731f67372764b9da8e54594716aff2a : SKIPPED\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/0fd606c609bf48e98cd2206892290273 : SKIPPED\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/32a7fe06b14149399bd2eb60ca67ba70 : SUCCESS in 52m 13s\n- tripleo-tox-molecule https://zuul.opendev.org/t/openstack/build/e742e85ce98c477e9b54e6726532fb56 : SUCCESS in 15m 50s","accounts_in_message":[],"_revision_number":11},{"id":"95bdfc9a23543c272184972e3f9feafee122de8e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-02-01 01:53:49.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/33216e2691014666bc113d3ab98ec8ca : FAILURE in 34m 34s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/1e2891b0802744109697c3db1dd978e2 : FAILURE in 1h 36m 52s","accounts_in_message":[],"_revision_number":11},{"id":"2e1cd0baa5a46227e6f3761d85bb34cb8b285dec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2022-02-01 02:35:16.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"9a5b030d528e2e32daf65132bb8c98893df04087","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-02-01 04:13:19.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/3823c116449348319d314ef97d347ac7 : FAILURE in 30m 42s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/5de57108004147e4a0c1b1777c166007 : FAILURE in 1h 36m 35s","accounts_in_message":[],"_revision_number":12},{"id":"d163812696895fd6c2a90a79934d2c8a9a889d79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 06:01:49.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-7-buildah-train https://zuul.opendev.org/t/openstack/build/fa1856ecae454646be35952fda4434dc : SUCCESS in 1h 03m 20s\n- tripleo-build-containers-centos-9 https://zuul.opendev.org/t/openstack/build/f0e9c246761a4f3487495a5daca111b4 : SUCCESS in 1h 07m 41s (non-voting)\n- tripleo-buildimage-overcloud-full-centos-7-train https://zuul.opendev.org/t/openstack/build/91814c306c5243e4a7caca3f36d17eb4 : SUCCESS in 1h 05m 47s\n- tripleo-buildimage-ironic-python-agent-centos-7-train https://zuul.opendev.org/t/openstack/build/6362af430a9f46409c01b5efeb9e7831 : SUCCESS in 29m 35s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/8beaabefc4be40e3b4840182d53d8b1f : SUCCESS in 1h 15m 17s\n- tripleo-buildimage-overcloud-full-centos-8-wallaby https://zuul.opendev.org/t/openstack/build/6b1b837a532f4ff59f79ea901b260ca4 : SUCCESS in 48m 36s\n- tripleo-buildimage-overcloud-full-centos-8-victoria https://zuul.opendev.org/t/openstack/build/1df2305b068c4c50b9081de288fbaa42 : SUCCESS in 53m 25s\n- tripleo-buildimage-overcloud-full-centos-8-ussuri https://zuul.opendev.org/t/openstack/build/23673a89f73e41b6946eaa8673766533 : SUCCESS in 50m 16s\n- tripleo-buildimage-overcloud-full-centos-8-train https://zuul.opendev.org/t/openstack/build/a3e23f018ff94f38ab8f2d6ad09294df : SUCCESS in 46m 11s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/79c5f9ce00dc40fc95fe6c3d8a22e2ca : SUCCESS in 37m 12s\n- tripleo-buildimage-ironic-python-agent-centos-8-wallaby https://zuul.opendev.org/t/openstack/build/00445358105e4d82bf7c4ee3b024cedc : SUCCESS in 25m 21s\n- tripleo-buildimage-ironic-python-agent-centos-8-victoria https://zuul.opendev.org/t/openstack/build/319f037b3d0342dfbe2947989e56804f : SUCCESS in 29m 17s\n- tripleo-buildimage-ironic-python-agent-centos-8-ussuri https://zuul.opendev.org/t/openstack/build/c58bf751cb354aca976cb0cac64411ed : SUCCESS in 27m 30s\n- tripleo-buildimage-ironic-python-agent-centos-8-train https://zuul.opendev.org/t/openstack/build/121785e05f984a22a0c7c49c8a547363 : SUCCESS in 42m 52s\n- tripleo-buildimage-overcloud-hardened-uefi-full-centos-8 https://zuul.opendev.org/t/openstack/build/ca58198bb4a54ea7a17fb4ee3a8b8e86 : SUCCESS in 43m 22s\n- tripleo-buildimage-overcloud-hardened-uefi-full-centos-8-wallaby https://zuul.opendev.org/t/openstack/build/292fd5206f73498ab9dc9bd74e794789 : SUCCESS in 42m 33s\n- tripleo-buildimage-overcloud-full-centos-9 https://zuul.opendev.org/t/openstack/build/dc3bf0f9b803419a84973c4d171cea5c : SUCCESS in 45m 33s\n- tripleo-buildimage-ironic-python-agent-centos-9 https://zuul.opendev.org/t/openstack/build/db058d44e5934938a6dbfc364cc90d47 : SUCCESS in 40m 14s\n- tripleo-buildimage-overcloud-hardened-uefi-full-centos-9 https://zuul.opendev.org/t/openstack/build/ce807b6b037d4a81a891efe6ce37a29f : SUCCESS in 43m 04s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9e91855272354c04ba6b381eaedcce2b : SUCCESS in 6m 37s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/9234fc7f394744b38c89ab4e6ee9939f : SUCCESS in 1h 59m 14s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/82cb6617872641aa83c0eba8030fcd1a : SUCCESS in 2h 52m 35s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/4736425bf115499c97dd0e0b1273c070 : SUCCESS in 1h 28m 54s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/c2d6ba07301d4c1e84baafafef454d47 : SUCCESS in 1h 04m 22s\n- tripleo-tox-molecule https://zuul.opendev.org/t/openstack/build/61a753d58300401f9a4f282ef99126c4 : SUCCESS in 16m 26s","accounts_in_message":[],"_revision_number":12},{"id":"969a9d7619cd40cbe37f50c71efd118fa1d4c760","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2022-02-01 07:49:34.000000000","message":"Patch Set 12: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"90ed8ec79d8a298c41af1343d4f92b2e3bdc8bfe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2022-02-01 10:16:59.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"5e9184d10d7cc371ace7e938c727e93c5fb8c137","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2022-02-01 10:20:10.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"91c1245dbe2a2b854f33962f5c7b364186cc1209","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-02-01 12:22:02.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/4699fb7929044bf6a63a4f5cc9ed3938 : FAILURE in 36m 51s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/7a1ded7beefa4e4bb74f332116184fc0 : FAILURE in 2h 00m 27s","accounts_in_message":[],"_revision_number":14},{"id":"8889b4652c18aef2b2e95191c1dc82c2a905f033","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 12:23:15.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-build-containers-centos-7-buildah-train https://zuul.opendev.org/t/openstack/build/5ddeb1dc706348c5bef972471042a727 : SUCCESS in 41m 38s\n- tripleo-build-containers-centos-9 https://zuul.opendev.org/t/openstack/build/27d82c71aec44423968e31c3841689fc : SUCCESS in 36m 23s (non-voting)\n- tripleo-buildimage-overcloud-full-centos-7-train https://zuul.opendev.org/t/openstack/build/a8f8fe925af04f84929a922acfe308c2 : SUCCESS in 42m 09s\n- tripleo-buildimage-ironic-python-agent-centos-7-train https://zuul.opendev.org/t/openstack/build/f17819bb61554d6d9a5b3a900621d99d : SUCCESS in 20m 59s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/83b07270445e473eb32fccf4024940f5 : SUCCESS in 56m 10s\n- tripleo-buildimage-overcloud-full-centos-8-wallaby https://zuul.opendev.org/t/openstack/build/698609e607d94b9ca1511e35272f5c76 : SUCCESS in 1h 01m 15s\n- tripleo-buildimage-overcloud-full-centos-8-victoria https://zuul.opendev.org/t/openstack/build/d6c02fb9ef2c49759301d2eb975f6779 : FAILURE in 34m 58s\n- tripleo-buildimage-overcloud-full-centos-8-ussuri https://zuul.opendev.org/t/openstack/build/9f403d977b1a4638ba1397bb34f892eb : FAILURE in 31m 02s\n- tripleo-buildimage-overcloud-full-centos-8-train https://zuul.opendev.org/t/openstack/build/c9707e700ccd48d5bb9096a1a188047d : FAILURE in 33m 24s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/893faa151e7d408bb4b3247a1f3807d8 : SUCCESS in 31m 23s\n- tripleo-buildimage-ironic-python-agent-centos-8-wallaby https://zuul.opendev.org/t/openstack/build/e67d73516851431ca69dbd4d0e09866c : SUCCESS in 30m 42s\n- tripleo-buildimage-ironic-python-agent-centos-8-victoria https://zuul.opendev.org/t/openstack/build/6c54160d1b444711999a83445e5a1aaf : FAILURE in 28m 21s\n- tripleo-buildimage-ironic-python-agent-centos-8-ussuri https://zuul.opendev.org/t/openstack/build/67082a27222f451d8a2d6f3c68cc9010 : SUCCESS in 28m 00s\n- tripleo-buildimage-ironic-python-agent-centos-8-train https://zuul.opendev.org/t/openstack/build/51d1403022ec411484418b22e850702f : FAILURE in 31m 00s\n- tripleo-buildimage-overcloud-hardened-uefi-full-centos-8 https://zuul.opendev.org/t/openstack/build/b652b05615e94d00923b42302a83a4a2 : SUCCESS in 40m 08s\n- tripleo-buildimage-overcloud-hardened-uefi-full-centos-8-wallaby https://zuul.opendev.org/t/openstack/build/10775e5e6a174b94af514b3461e18eb3 : SUCCESS in 39m 26s\n- tripleo-buildimage-overcloud-full-centos-9 https://zuul.opendev.org/t/openstack/build/e8cc128c0f2142b787eb1f7382ad4cd7 : SUCCESS in 31m 12s\n- tripleo-buildimage-ironic-python-agent-centos-9 https://zuul.opendev.org/t/openstack/build/36c79380910a4638858308861e5ce41a : SUCCESS in 25m 43s\n- tripleo-buildimage-overcloud-hardened-uefi-full-centos-9 https://zuul.opendev.org/t/openstack/build/f5529b5b86f44d2eac2320ef9fb88d16 : SUCCESS in 31m 18s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e819c95ecada49f8bcf605528c23f7a9 : FAILURE in 5m 06s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/5c2ad3118c2d42a1b495d481f01d2078 : SUCCESS in 2h 00m 23s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/dbce505ea6084c08b56698bb2cf50c10 : SKIPPED\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/e4c3e1fa467d4571b3c25835f4274e6f : SKIPPED\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/d0952c3fee7a40b3a9fdf10d5c55acc2 : SUCCESS in 1h 07m 13s\n- tripleo-tox-molecule https://zuul.opendev.org/t/openstack/build/1be4b104e7414bb59db697b0274cadc7 : SUCCESS in 13m 40s","accounts_in_message":[],"_revision_number":14},{"id":"b8659360e1c4f7456d41ac3c8a7dd80726f98bc2","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2022-02-01 14:54:54.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"32e2fb11c7c42838a4f7c4c6f163eae497f02deb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2022-02-01 15:18:20.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"156ad4e15eb5160ffeb8610946bbf60542316be7","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2022-02-01 15:19:15.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"dbc0c4f9091f17068920f008a8c01bc4f2bb4e4f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2022-02-01 16:34:18.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"82f305490a53305bb17e1529013c6c4a73e35a64","author":{"_account_id":14611,"name":"dasm","display_name":"dasm","email":"dsmigiel@redhat.com","username":"dasm"},"date":"2022-02-01 17:39:38.000000000","message":"Patch Set 16: Code-Review+1","accounts_in_message":[],"_revision_number":16},{"id":"75d2e351579ae4aa118bd20a9641761be8a50cfb","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2022-02-01 18:10:35.000000000","message":"Patch Set 16: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"823799ede982180317b3a0be7392422d6ab6433e","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2022-02-01 18:26:50.000000000","message":"Patch Set 16: Workflow+1","accounts_in_message":[],"_revision_number":16},{"id":"1eebd430fb1c46f4fdb97f87ed77a9544c996395","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-02-01 18:41:16.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/1e0cba3c69a34145abe9300e99554961 : FAILURE in 32m 22s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/af8cc69d36564baf844dfbd96ca8cad7 : SUCCESS in 2h 05m 10s","accounts_in_message":[],"_revision_number":16},{"id":"2d86df1aaad937beede3dfdbd52e0f1ffbb48d62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 20:18:39.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-7-buildah-train https://zuul.opendev.org/t/openstack/build/d503a78c11f144908256ee442049ea60 : SUCCESS in 1h 04m 18s\n- tripleo-build-containers-centos-9 https://zuul.opendev.org/t/openstack/build/fc01de697e3b4e8897c5653e3ea93c24 : SUCCESS in 49m 22s (non-voting)\n- tripleo-buildimage-overcloud-full-centos-7-train https://zuul.opendev.org/t/openstack/build/1c63c179a4f94cf2b2f9000dacb3cc97 : SUCCESS in 55m 24s\n- tripleo-buildimage-ironic-python-agent-centos-7-train https://zuul.opendev.org/t/openstack/build/913a4d6902a84a4497549c9374a842ed : SUCCESS in 28m 15s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/3ef19916f2bd4abb92917438f307d406 : SUCCESS in 1h 13m 01s\n- tripleo-buildimage-overcloud-full-centos-8-wallaby https://zuul.opendev.org/t/openstack/build/451f95ff56a64b53abf765cc76d482a7 : SUCCESS in 1h 01m 29s\n- tripleo-buildimage-overcloud-full-centos-8-victoria https://zuul.opendev.org/t/openstack/build/d0dffc667e3846f18352895ad9df81ac : SUCCESS in 1h 25m 25s\n- tripleo-buildimage-overcloud-full-centos-8-ussuri https://zuul.opendev.org/t/openstack/build/d0fea3ef8539472e96b760efcfbd2c9d : SUCCESS in 1h 26m 46s\n- tripleo-buildimage-overcloud-full-centos-8-train https://zuul.opendev.org/t/openstack/build/ec28246f02b3432685113520f6b5e1db : SUCCESS in 54m 35s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/ba8caf7f3c3f4c5fac8f6890510e4cb6 : SUCCESS in 31m 04s\n- tripleo-buildimage-ironic-python-agent-centos-8-wallaby https://zuul.opendev.org/t/openstack/build/7c768d9940af4c378f3fbd05d71a80ee : SUCCESS in 39m 08s\n- tripleo-buildimage-ironic-python-agent-centos-8-victoria https://zuul.opendev.org/t/openstack/build/96ad5b739f534b7c9eafdd58a24bad1a : SUCCESS in 38m 20s\n- tripleo-buildimage-ironic-python-agent-centos-8-ussuri https://zuul.opendev.org/t/openstack/build/08dd452e64ca4c3d99ee2fa5a0a09cb0 : SUCCESS in 37m 18s\n- tripleo-buildimage-ironic-python-agent-centos-8-train https://zuul.opendev.org/t/openstack/build/d9b0b0c4bfd14854984f27f58dda9dc9 : SUCCESS in 42m 01s\n- tripleo-buildimage-overcloud-hardened-uefi-full-centos-8 https://zuul.opendev.org/t/openstack/build/708b0d9853ba470eb6333f33c596c811 : SUCCESS in 47m 30s\n- tripleo-buildimage-overcloud-hardened-uefi-full-centos-8-wallaby https://zuul.opendev.org/t/openstack/build/d31aabc0af7e41d3bc3401c56773f954 : SUCCESS in 37m 51s\n- tripleo-buildimage-overcloud-full-centos-9 https://zuul.opendev.org/t/openstack/build/171a820cc8b24c8c9760ed7c41777402 : SUCCESS in 32m 53s\n- tripleo-buildimage-ironic-python-agent-centos-9 https://zuul.opendev.org/t/openstack/build/4b9a910a292f453aacd4cc039a5eca46 : SUCCESS in 27m 22s\n- tripleo-buildimage-overcloud-hardened-uefi-full-centos-9 https://zuul.opendev.org/t/openstack/build/cb3d79ded54c4158991ba307a6204534 : SUCCESS in 32m 55s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/886af8b6b34e4b88a6278bee3d169ddc : SUCCESS in 5m 14s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/27e0d5e4906e4c44b322315a7da8c547 : SUCCESS in 1h 46m 07s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/ae3b14e95ef8491e9ea656eeb46da602 : SUCCESS in 3h 26m 27s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/f8a4212e4a3f45c883bd4a65fe13f7c0 : SUCCESS in 1h 37m 26s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/1e45c256c6e64b92a271fddbc5f41ccb : SUCCESS in 56m 03s\n- tripleo-tox-molecule https://zuul.opendev.org/t/openstack/build/fcab90f523014695b5eda222af668ea5 : SUCCESS in 15m 16s","accounts_in_message":[],"_revision_number":16},{"id":"6e5055e8f929f79fa2f26e71fc528dce023f954c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 20:19:56.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"67c6f8ff6f2bb0e6b91b4141b5c7f258faaae2cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 22:45:12.000000000","message":"Patch Set 16: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-build-containers-centos-7-buildah-train https://zuul.opendev.org/t/openstack/build/e4d85cacc3244ef2abff7b98f662c18c : SUCCESS in 57m 56s\n- tripleo-buildimage-overcloud-full-centos-7-train https://zuul.opendev.org/t/openstack/build/537736d3f2404d55843a87910526d440 : FAILURE in 20m 40s\n- tripleo-buildimage-ironic-python-agent-centos-7-train https://zuul.opendev.org/t/openstack/build/c77a9681cd154f58a971fa2e51eac8f2 : SUCCESS in 23m 30s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/2b185b3a1adc48f5b9f31c5bbb6e262e : SUCCESS in 1h 04m 53s\n- tripleo-buildimage-overcloud-full-centos-8-wallaby https://zuul.opendev.org/t/openstack/build/498eaa72241f41fd99df92f2fc808ca1 : SUCCESS in 1h 05m 04s\n- tripleo-buildimage-overcloud-full-centos-8-victoria https://zuul.opendev.org/t/openstack/build/48034f35ff044f23acd6d3ce204d78a4 : SUCCESS in 1h 01m 47s\n- tripleo-buildimage-overcloud-full-centos-8-ussuri https://zuul.opendev.org/t/openstack/build/d55b1735a83544969fbfdb5c8d958015 : SUCCESS in 1h 02m 53s\n- tripleo-buildimage-overcloud-full-centos-8-train https://zuul.opendev.org/t/openstack/build/33259969dfbc4ebe9bbd04fb77554a6b : SUCCESS in 57m 12s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/8d6e4719a78d44ea98e5f97a9fc1d844 : SUCCESS in 35m 28s\n- tripleo-buildimage-ironic-python-agent-centos-8-wallaby https://zuul.opendev.org/t/openstack/build/0b278ef196ee40f99bf448b3bf1ba524 : SUCCESS in 31m 04s\n- tripleo-buildimage-ironic-python-agent-centos-8-victoria https://zuul.opendev.org/t/openstack/build/3e9df96a7cd040238e27603bc393a06f : SUCCESS in 34m 36s\n- tripleo-buildimage-ironic-python-agent-centos-8-ussuri https://zuul.opendev.org/t/openstack/build/c6023ee145a14a27b92f5cce359406d9 : SUCCESS in 27m 54s\n- tripleo-buildimage-ironic-python-agent-centos-8-train https://zuul.opendev.org/t/openstack/build/57f928fd16dc40d0a3b03671d2ed01fc : SUCCESS in 34m 07s\n- tripleo-buildimage-overcloud-hardened-uefi-full-centos-8 https://zuul.opendev.org/t/openstack/build/69cc416373184e96aeb32e7e40bd303e : SUCCESS in 38m 38s\n- tripleo-buildimage-overcloud-hardened-uefi-full-centos-8-wallaby https://zuul.opendev.org/t/openstack/build/12ad8ef480f74b9c93953b8f8d788b64 : SUCCESS in 41m 29s\n- tripleo-buildimage-overcloud-full-centos-9 https://zuul.opendev.org/t/openstack/build/7f3d31101ecd494d92483cbe0a4fc929 : SUCCESS in 36m 59s\n- tripleo-buildimage-ironic-python-agent-centos-9 https://zuul.opendev.org/t/openstack/build/89835e98a4924f41974d7b962956f893 : SUCCESS in 29m 57s\n- tripleo-buildimage-overcloud-hardened-uefi-full-centos-9 https://zuul.opendev.org/t/openstack/build/3d6bd2142230409bb37f059ad6a62974 : SUCCESS in 36m 06s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/00f432e827fd40bfbc136f733508f891 : SUCCESS in 2h 14m 04s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/de7fba01da934b3188c0fa7d8b9ad291 : SUCCESS in 2h 24m 15s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/0b2454c4a6b74f5fbc9afb192e1fd89f : SUCCESS in 1h 12m 54s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/0346373156e3418bb573a792358ead78 : SUCCESS in 1h 10m 30s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fbbfd55a93b04090a7fcb028033869d9 : SUCCESS in 6m 39s\n- tripleo-tox-molecule https://zuul.opendev.org/t/openstack/build/6c63cf765a064e8a9bc60f41e4fe45a5 : SUCCESS in 15m 43s","accounts_in_message":[],"_revision_number":16},{"id":"0a841c569c6590d211a4533d7f9d28f913775991","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2022-02-01 22:57:35.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"bdbf01d3d60ec2c5eb68f7b0c6566ffdb3ac821d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-02-02 00:39:27.000000000","message":"Patch Set 16:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/776659cf160d40d3a202c284e36a0d12 : FAILURE in 37m 03s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/6749b8db6df140cca29d3f694c18606d : FAILURE in 1h 39m 33s","accounts_in_message":[],"_revision_number":16},{"id":"492461756fca7f0af5e18a85f8401e503f24b9d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 02:56:17.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-7-buildah-train https://zuul.opendev.org/t/openstack/build/1118d98e69784ee29d5043ef7784e334 : SUCCESS in 55m 55s\n- tripleo-build-containers-centos-9 https://zuul.opendev.org/t/openstack/build/97bfacc42d2846fcb9c02312f8d98da6 : SUCCESS in 49m 53s (non-voting)\n- tripleo-buildimage-overcloud-full-centos-7-train https://zuul.opendev.org/t/openstack/build/dae6b170c795476ba54015791d7d52d4 : SUCCESS in 40m 04s\n- tripleo-buildimage-ironic-python-agent-centos-7-train https://zuul.opendev.org/t/openstack/build/83dbbce286af4d4093891330f9548303 : SUCCESS in 21m 54s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/f7cb671a426e48798b545c6a7c127fc1 : SUCCESS in 58m 41s\n- tripleo-buildimage-overcloud-full-centos-8-wallaby https://zuul.opendev.org/t/openstack/build/1e087f60c1ea4c5cadfef050287a3a5d : SUCCESS in 59m 08s\n- tripleo-buildimage-overcloud-full-centos-8-victoria https://zuul.opendev.org/t/openstack/build/121db8687c2e481b907de0263be5b30b : SUCCESS in 56m 43s\n- tripleo-buildimage-overcloud-full-centos-8-ussuri https://zuul.opendev.org/t/openstack/build/552ee94bd39a4986aae0257758dd04b5 : SUCCESS in 54m 34s\n- tripleo-buildimage-overcloud-full-centos-8-train https://zuul.opendev.org/t/openstack/build/9bd3aa8e081e47239edcad39e5ad6212 : SUCCESS in 56m 20s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/eb8bd70226c5412d8004cc10c3e887e1 : SUCCESS in 29m 46s\n- tripleo-buildimage-ironic-python-agent-centos-8-wallaby https://zuul.opendev.org/t/openstack/build/f35ec4c6034c4ffc88d16c07902811d0 : SUCCESS in 28m 43s\n- tripleo-buildimage-ironic-python-agent-centos-8-victoria https://zuul.opendev.org/t/openstack/build/3b6d12e63deb4217b428ffd0fd869264 : SUCCESS in 29m 37s\n- tripleo-buildimage-ironic-python-agent-centos-8-ussuri https://zuul.opendev.org/t/openstack/build/32ca01e6e8c04fd5950542d88485fa0d : SUCCESS in 28m 47s\n- tripleo-buildimage-ironic-python-agent-centos-8-train https://zuul.opendev.org/t/openstack/build/ff0610f53be5427e8375790e745d6b65 : SUCCESS in 27m 37s\n- tripleo-buildimage-overcloud-hardened-uefi-full-centos-8 https://zuul.opendev.org/t/openstack/build/10a013cc94f34fc4a09feef2ca88e2b9 : SUCCESS in 48m 18s\n- tripleo-buildimage-overcloud-hardened-uefi-full-centos-8-wallaby https://zuul.opendev.org/t/openstack/build/efecfcfb8f2a43bf87d1ee6cb35631dc : SUCCESS in 40m 08s\n- tripleo-buildimage-overcloud-full-centos-9 https://zuul.opendev.org/t/openstack/build/de7121767f654726b498a067439da0f4 : SUCCESS in 50m 52s\n- tripleo-buildimage-ironic-python-agent-centos-9 https://zuul.opendev.org/t/openstack/build/deea2f5e173f4270873cde97653947ce : SUCCESS in 21m 51s\n- tripleo-buildimage-overcloud-hardened-uefi-full-centos-9 https://zuul.opendev.org/t/openstack/build/ca085bec160c4ca58cb1ca9944a3333f : SUCCESS in 45m 15s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3a38d1e526d74c5cb16def05824ade74 : SUCCESS in 5m 22s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/9cd2b545af0c4465b7d26181d01ba4ec : SUCCESS in 2h 23m 16s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/8af1833645604d218181eec1a5b81cf7 : SUCCESS in 3h 17m 10s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/d62185d0deb247129122713f5826a74c : SUCCESS in 1h 42m 26s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/806e15ef4bfe473b8c636037edd3c23b : SUCCESS in 1h 21m 37s\n- tripleo-tox-molecule https://zuul.opendev.org/t/openstack/build/bf271c89f20147a8ae062a22636fc9df : SUCCESS in 18m 20s","accounts_in_message":[],"_revision_number":16},{"id":"5783713980d8e6b9f3379cf3f48f7a8a7911a7ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 02:57:20.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"f6566633bc5127934c021b474120ec7af334be69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 06:01:39.000000000","message":"Patch Set 16: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-build-containers-centos-7-buildah-train https://zuul.opendev.org/t/openstack/build/3617cad1a9e24685b7cc9ef71f1d2cff : SUCCESS in 57m 37s\n- tripleo-buildimage-overcloud-full-centos-7-train https://zuul.opendev.org/t/openstack/build/13ee11d321c84f1791619e179438a22a : SUCCESS in 42m 50s\n- tripleo-buildimage-ironic-python-agent-centos-7-train https://zuul.opendev.org/t/openstack/build/2025d4ac525c4e1281646492adf6f38a : SUCCESS in 25m 24s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/2c4a59f4a7164873987946f420bc7d37 : SUCCESS in 1h 05m 00s\n- tripleo-buildimage-overcloud-full-centos-8-wallaby https://zuul.opendev.org/t/openstack/build/72bcf1f49eec473fb5ba9e8346c45a69 : SUCCESS in 1h 09m 39s\n- tripleo-buildimage-overcloud-full-centos-8-victoria https://zuul.opendev.org/t/openstack/build/8211fb4e82b34c7d866647c8829bd644 : SUCCESS in 1h 10m 19s\n- tripleo-buildimage-overcloud-full-centos-8-ussuri https://zuul.opendev.org/t/openstack/build/a098f7faa7a54191887b5728bf52d758 : SUCCESS in 1h 03m 33s\n- tripleo-buildimage-overcloud-full-centos-8-train https://zuul.opendev.org/t/openstack/build/8445206d1b6f4602ba35e8780e6be49d : SUCCESS in 1h 07m 11s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/e97210ef0de44643982a1acb0488e9c6 : SUCCESS in 34m 00s\n- tripleo-buildimage-ironic-python-agent-centos-8-wallaby https://zuul.opendev.org/t/openstack/build/2fad3410de5d4b389170dd2e0078d78d : SUCCESS in 34m 10s\n- tripleo-buildimage-ironic-python-agent-centos-8-victoria https://zuul.opendev.org/t/openstack/build/2a867de410564571bb4bf0bb6e75eae5 : SUCCESS in 34m 55s\n- tripleo-buildimage-ironic-python-agent-centos-8-ussuri https://zuul.opendev.org/t/openstack/build/d8f9de0137db4af7b5f607a5287c17bd : SUCCESS in 34m 16s\n- tripleo-buildimage-ironic-python-agent-centos-8-train https://zuul.opendev.org/t/openstack/build/c52465a9116f4748b0acb136f85f4f30 : SUCCESS in 33m 25s\n- tripleo-buildimage-overcloud-hardened-uefi-full-centos-8 https://zuul.opendev.org/t/openstack/build/7024dab746374355934dfa3a7c4ae048 : SUCCESS in 44m 16s\n- tripleo-buildimage-overcloud-hardened-uefi-full-centos-8-wallaby https://zuul.opendev.org/t/openstack/build/5294de99fc3a47e692483e4792415819 : SUCCESS in 51m 40s\n- tripleo-buildimage-overcloud-full-centos-9 https://zuul.opendev.org/t/openstack/build/6f34a802840e4a27bb76b448d94ed6ea : SUCCESS in 41m 01s\n- tripleo-buildimage-ironic-python-agent-centos-9 https://zuul.opendev.org/t/openstack/build/86a3033e36204f03945ba8992c77b4c5 : SUCCESS in 34m 50s\n- tripleo-buildimage-overcloud-hardened-uefi-full-centos-9 https://zuul.opendev.org/t/openstack/build/6b315b279399426a9dca11992f32b981 : SUCCESS in 39m 42s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/e64aec0705654a848df3a4b116d58e65 : SUCCESS in 2h 15m 27s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/d11208203c154adfa04312b2759714f5 : SUCCESS in 3h 00m 26s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/ca366a3bc0e944059c21c9e889fa1b1e : SUCCESS in 1h 33m 54s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/775197e425794996a2919803c7e1135a : SUCCESS in 1h 16m 58s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/47aef47bca514e4980b199a6d8db1956 : SUCCESS in 8m 30s\n- tripleo-tox-molecule https://zuul.opendev.org/t/openstack/build/1ea821f537a14a1bb6cf33137bc76687 : SUCCESS in 16m 32s","accounts_in_message":[],"_revision_number":16},{"id":"24e52efe3a1113204a43d39a34120b14e0245aba","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 06:01:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"846e0f0763e8f1b5e5cc6642d522da5d5b1175f5","revisions":{"682df1c5617a2e41483afabb0b9309f55de3b0ee":{"kind":"REWORK","_number":1,"created":"2022-01-26 20:09:27.000000000","uploader":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"ref":"refs/changes/21/826521/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-ci","ref":"refs/changes/21/826521/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/21/826521/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/21/826521/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/21/826521/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-ci refs/changes/21/826521/1"}}},"commit":{"parents":[{"commit":"3fb0d96f804fdb8bb043139d986c262dbbf0f922","subject":"Merge \"Set \"9-stream\" for CS9\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/3fb0d96f804fdb8bb043139d986c262dbbf0f922"}]}],"author":{"name":"Ronelle Landy","email":"rlandy@redhat.com","date":"2022-01-26 20:07:49.000000000","tz":-300},"committer":{"name":"Ronelle Landy","email":"rlandy@redhat.com","date":"2022-01-26 20:07:49.000000000","tz":-300},"subject":"DNM: Adding vars for cert","message":"DNM: Adding vars for cert\n\nIf a cert is needed to access resources,\nit needs to be uploaded to the image\nand mounted to the base container.\n\nThis is DNM as the real solution is\nto add a DIB element to do this work\non images.\n\nChange-Id: I303b32a3081c60739ad650ef5afc90e239a0d46e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/682df1c5617a2e41483afabb0b9309f55de3b0ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/682df1c5617a2e41483afabb0b9309f55de3b0ee"}]},"branch":"refs/heads/master"},"294802ddcaceb40d68e12419586fcf03cdbb12de":{"kind":"REWORK","_number":2,"created":"2022-01-26 20:29:31.000000000","uploader":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"ref":"refs/changes/21/826521/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-ci","ref":"refs/changes/21/826521/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/21/826521/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/21/826521/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/21/826521/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-ci refs/changes/21/826521/2"}}},"commit":{"parents":[{"commit":"3fb0d96f804fdb8bb043139d986c262dbbf0f922","subject":"Merge \"Set \"9-stream\" for CS9\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/3fb0d96f804fdb8bb043139d986c262dbbf0f922"}]}],"author":{"name":"Ronelle Landy","email":"rlandy@redhat.com","date":"2022-01-26 20:07:49.000000000","tz":-300},"committer":{"name":"Ronelle Landy","email":"rlandy@redhat.com","date":"2022-01-26 20:29:17.000000000","tz":-300},"subject":"DNM: Adding vars for cert","message":"DNM: Adding vars for cert\n\nIf a cert is needed to access resources,\nit needs to be uploaded to the image\nand mounted to the base container.\n\nThis is DNM as the real solution is\nto add a DIB element to do this work\non images.\n\nChange-Id: I303b32a3081c60739ad650ef5afc90e239a0d46e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/294802ddcaceb40d68e12419586fcf03cdbb12de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/294802ddcaceb40d68e12419586fcf03cdbb12de"}]},"branch":"refs/heads/master"},"ad14d4b103238b82d3a692ea9dfe87d0ed0b1c16":{"kind":"REWORK","_number":3,"created":"2022-01-26 21:11:36.000000000","uploader":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"ref":"refs/changes/21/826521/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-ci","ref":"refs/changes/21/826521/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/21/826521/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/21/826521/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/21/826521/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-ci refs/changes/21/826521/3"}}},"commit":{"parents":[{"commit":"3fb0d96f804fdb8bb043139d986c262dbbf0f922","subject":"Merge \"Set \"9-stream\" for CS9\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/3fb0d96f804fdb8bb043139d986c262dbbf0f922"}]}],"author":{"name":"Ronelle Landy","email":"rlandy@redhat.com","date":"2022-01-26 20:07:49.000000000","tz":-300},"committer":{"name":"Ronelle Landy","email":"rlandy@redhat.com","date":"2022-01-26 21:11:28.000000000","tz":-300},"subject":"DNM: Adding vars for cert","message":"DNM: Adding vars for cert\n\nIf a cert is needed to access resources,\nit needs to be uploaded to the image\nand mounted to the base container.\n\nThis is DNM as the real solution is\nto add a DIB element to do this work\non images.\n\nChange-Id: I303b32a3081c60739ad650ef5afc90e239a0d46e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/ad14d4b103238b82d3a692ea9dfe87d0ed0b1c16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/ad14d4b103238b82d3a692ea9dfe87d0ed0b1c16"}]},"branch":"refs/heads/master"},"bb8dbf5a3c72b34c89f1df7f2ea73cb675f725c0":{"kind":"REWORK","_number":4,"created":"2022-01-26 21:29:07.000000000","uploader":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"ref":"refs/changes/21/826521/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-ci","ref":"refs/changes/21/826521/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/21/826521/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/21/826521/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/21/826521/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-ci refs/changes/21/826521/4"}}},"commit":{"parents":[{"commit":"3fb0d96f804fdb8bb043139d986c262dbbf0f922","subject":"Merge \"Set \"9-stream\" for CS9\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/3fb0d96f804fdb8bb043139d986c262dbbf0f922"}]}],"author":{"name":"Ronelle Landy","email":"rlandy@redhat.com","date":"2022-01-26 20:07:49.000000000","tz":-300},"committer":{"name":"Ronelle Landy","email":"rlandy@redhat.com","date":"2022-01-26 21:29:01.000000000","tz":-300},"subject":"DNM: Adding vars for cert","message":"DNM: Adding vars for cert\n\nIf a cert is needed to access resources,\nit needs to be uploaded to the image\nand mounted to the base container.\n\nThis is DNM as the real solution is\nto add a DIB element to do this work\non images.\n\nChange-Id: I303b32a3081c60739ad650ef5afc90e239a0d46e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/bb8dbf5a3c72b34c89f1df7f2ea73cb675f725c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/bb8dbf5a3c72b34c89f1df7f2ea73cb675f725c0"}]},"branch":"refs/heads/master"},"b896bf296860f6c24e391c222c68bce12aa21206":{"kind":"REWORK","_number":5,"created":"2022-01-26 21:50:41.000000000","uploader":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"ref":"refs/changes/21/826521/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-ci","ref":"refs/changes/21/826521/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/21/826521/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/21/826521/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/21/826521/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-ci refs/changes/21/826521/5"}}},"commit":{"parents":[{"commit":"3fb0d96f804fdb8bb043139d986c262dbbf0f922","subject":"Merge \"Set \"9-stream\" for CS9\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/3fb0d96f804fdb8bb043139d986c262dbbf0f922"}]}],"author":{"name":"Ronelle Landy","email":"rlandy@redhat.com","date":"2022-01-26 20:07:49.000000000","tz":-300},"committer":{"name":"Ronelle Landy","email":"rlandy@redhat.com","date":"2022-01-26 21:50:35.000000000","tz":-300},"subject":"DNM: Adding vars for cert","message":"DNM: Adding vars for cert\n\nIf a cert is needed to access resources,\nit needs to be uploaded to the image\nand mounted to the base container.\n\nThis is DNM as the real solution is\nto add a DIB element to do this work\non images.\n\nChange-Id: I303b32a3081c60739ad650ef5afc90e239a0d46e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/b896bf296860f6c24e391c222c68bce12aa21206"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/b896bf296860f6c24e391c222c68bce12aa21206"}]},"branch":"refs/heads/master"},"915b32f1d340724f36cf4888791880e21e13517d":{"kind":"REWORK","_number":6,"created":"2022-01-31 09:19:29.000000000","uploader":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"ref":"refs/changes/21/826521/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-ci","ref":"refs/changes/21/826521/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/21/826521/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/21/826521/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/21/826521/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-ci refs/changes/21/826521/6"}}},"commit":{"parents":[{"commit":"3fb0d96f804fdb8bb043139d986c262dbbf0f922","subject":"Merge \"Set \"9-stream\" for CS9\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/3fb0d96f804fdb8bb043139d986c262dbbf0f922"}]}],"author":{"name":"Ronelle Landy","email":"rlandy@redhat.com","date":"2022-01-26 20:07:49.000000000","tz":-300},"committer":{"name":"Sandeep Yadav","email":"sandyada@redhat.com","date":"2022-01-31 09:19:24.000000000","tz":330},"subject":"Adding vars for cert","message":"Adding vars for cert\n\nIf a cert is needed to access resources, it needs to be uploaded to\nthe image and mounted to the base container.\n\nThis is needed in downstream where after enabling ssl for DLRN\n- container and image build is failing because necessary certs\nare not available.\n\nChange-Id: I303b32a3081c60739ad650ef5afc90e239a0d46e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/915b32f1d340724f36cf4888791880e21e13517d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/915b32f1d340724f36cf4888791880e21e13517d"}]},"branch":"refs/heads/master"},"41c31fec42a436e0d48619c933158db89a450c02":{"kind":"REWORK","_number":7,"created":"2022-01-31 13:32:42.000000000","uploader":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"ref":"refs/changes/21/826521/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-ci","ref":"refs/changes/21/826521/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/21/826521/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/21/826521/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/21/826521/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-ci refs/changes/21/826521/7"}}},"commit":{"parents":[{"commit":"3fb0d96f804fdb8bb043139d986c262dbbf0f922","subject":"Merge \"Set \"9-stream\" for CS9\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/3fb0d96f804fdb8bb043139d986c262dbbf0f922"}]}],"author":{"name":"Ronelle Landy","email":"rlandy@redhat.com","date":"2022-01-26 20:07:49.000000000","tz":-300},"committer":{"name":"Sandeep Yadav","email":"sandyada@redhat.com","date":"2022-01-31 13:14:46.000000000","tz":330},"subject":"Adding vars for cert","message":"Adding vars for cert\n\nIf a cert is needed to access resources, it needs to be uploaded to\nthe image and mounted to the base container.\n\nThis is needed in downstream where after enabling ssl for DLRN\n- container and image build is failing because necessary certs\nare not available.\n\nChange-Id: I303b32a3081c60739ad650ef5afc90e239a0d46e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/41c31fec42a436e0d48619c933158db89a450c02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/41c31fec42a436e0d48619c933158db89a450c02"}]},"branch":"refs/heads/master"},"e2b2bcf787aa953f3b060a92ff6d36339deee461":{"kind":"REWORK","_number":8,"created":"2022-01-31 14:52:56.000000000","uploader":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"ref":"refs/changes/21/826521/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-ci","ref":"refs/changes/21/826521/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/21/826521/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/21/826521/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/21/826521/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-ci refs/changes/21/826521/8"}}},"commit":{"parents":[{"commit":"3fb0d96f804fdb8bb043139d986c262dbbf0f922","subject":"Merge \"Set \"9-stream\" for CS9\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/3fb0d96f804fdb8bb043139d986c262dbbf0f922"}]}],"author":{"name":"Ronelle Landy","email":"rlandy@redhat.com","date":"2022-01-26 20:07:49.000000000","tz":-300},"committer":{"name":"Sandeep Yadav","email":"sandyada@redhat.com","date":"2022-01-31 14:52:27.000000000","tz":330},"subject":"Adding vars for cert","message":"Adding vars for cert\n\nIf a cert is needed to access resources, it needs to be uploaded to\nthe image and mounted to the base container.\n\nThis is needed in downstream where after enabling ssl for DLRN\n- container and image build is failing because necessary certs\nare not available.\n\nChange-Id: I303b32a3081c60739ad650ef5afc90e239a0d46e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/e2b2bcf787aa953f3b060a92ff6d36339deee461"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/e2b2bcf787aa953f3b060a92ff6d36339deee461"}]},"branch":"refs/heads/master"},"69aec9a7dba1ce0c3974b4c1135e94b14fdd49c3":{"kind":"REWORK","_number":9,"created":"2022-01-31 15:21:32.000000000","uploader":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"ref":"refs/changes/21/826521/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-ci","ref":"refs/changes/21/826521/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/21/826521/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/21/826521/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/21/826521/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-ci refs/changes/21/826521/9"}}},"commit":{"parents":[{"commit":"3fb0d96f804fdb8bb043139d986c262dbbf0f922","subject":"Merge \"Set \"9-stream\" for CS9\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/3fb0d96f804fdb8bb043139d986c262dbbf0f922"}]}],"author":{"name":"Ronelle Landy","email":"rlandy@redhat.com","date":"2022-01-26 20:07:49.000000000","tz":-300},"committer":{"name":"Sandeep Yadav","email":"sandyada@redhat.com","date":"2022-01-31 15:21:13.000000000","tz":330},"subject":"Adding vars for cert","message":"Adding vars for cert\n\nIf a cert is needed to access resources, it needs to be uploaded to\nthe image and mounted to the base container.\n\nThis is needed in downstream where after enabling ssl for DLRN\n- container and image build is failing because necessary certs\nare not available.\n\nChange-Id: I303b32a3081c60739ad650ef5afc90e239a0d46e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/69aec9a7dba1ce0c3974b4c1135e94b14fdd49c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/69aec9a7dba1ce0c3974b4c1135e94b14fdd49c3"}]},"branch":"refs/heads/master"},"16a2f7cabe3dded2f2abb9be2991d9e95f5b9576":{"kind":"REWORK","_number":10,"created":"2022-01-31 15:44:11.000000000","uploader":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"ref":"refs/changes/21/826521/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-ci","ref":"refs/changes/21/826521/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/21/826521/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/21/826521/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/21/826521/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-ci refs/changes/21/826521/10"}}},"commit":{"parents":[{"commit":"3fb0d96f804fdb8bb043139d986c262dbbf0f922","subject":"Merge \"Set \"9-stream\" for CS9\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/3fb0d96f804fdb8bb043139d986c262dbbf0f922"}]}],"author":{"name":"Ronelle Landy","email":"rlandy@redhat.com","date":"2022-01-26 20:07:49.000000000","tz":-300},"committer":{"name":"Sandeep Yadav","email":"sandyada@redhat.com","date":"2022-01-31 15:43:30.000000000","tz":330},"subject":"Adding vars for cert","message":"Adding vars for cert\n\nIf a cert is needed to access resources, it needs to be uploaded to\nthe image and mounted to the base container.\n\nThis is needed in downstream where after enabling ssl for DLRN\n- container and image build is failing because necessary certs\nare not available.\n\nChange-Id: I303b32a3081c60739ad650ef5afc90e239a0d46e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/16a2f7cabe3dded2f2abb9be2991d9e95f5b9576"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/16a2f7cabe3dded2f2abb9be2991d9e95f5b9576"}]},"branch":"refs/heads/master"},"b10bfd80307f0c5c5d27e4dced7b90aac0b36e56":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2022-01-31 17:06:59.000000000","uploader":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"ref":"refs/changes/21/826521/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-ci","ref":"refs/changes/21/826521/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/21/826521/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/21/826521/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/21/826521/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-ci refs/changes/21/826521/11"}}},"commit":{"parents":[{"commit":"051457b7cd20347240a871126d2943d20722b78c","subject":"Merge \"Ensure we get a fully updated venv\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/051457b7cd20347240a871126d2943d20722b78c"}]}],"author":{"name":"Ronelle Landy","email":"rlandy@redhat.com","date":"2022-01-26 20:07:49.000000000","tz":-300},"committer":{"name":"Ronelle Landy","email":"rlandy@redhat.com","date":"2022-01-31 17:06:59.000000000","tz":0},"subject":"Adding vars for cert","message":"Adding vars for cert\n\nIf a cert is needed to access resources, it needs to be uploaded to\nthe image and mounted to the base container.\n\nThis is needed in downstream where after enabling ssl for DLRN\n- container and image build is failing because necessary certs\nare not available.\n\nChange-Id: I303b32a3081c60739ad650ef5afc90e239a0d46e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/b10bfd80307f0c5c5d27e4dced7b90aac0b36e56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/b10bfd80307f0c5c5d27e4dced7b90aac0b36e56"}]},"branch":"refs/heads/master","description":"Rebase"},"cac255c9892494c246d0b2343b9a96942a9eec2d":{"kind":"REWORK","_number":12,"created":"2022-02-01 02:35:16.000000000","uploader":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"ref":"refs/changes/21/826521/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-ci","ref":"refs/changes/21/826521/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/21/826521/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/21/826521/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/21/826521/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-ci refs/changes/21/826521/12"}}},"commit":{"parents":[{"commit":"051457b7cd20347240a871126d2943d20722b78c","subject":"Merge \"Ensure we get a fully updated venv\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/051457b7cd20347240a871126d2943d20722b78c"}]}],"author":{"name":"Ronelle Landy","email":"rlandy@redhat.com","date":"2022-01-26 20:07:49.000000000","tz":-300},"committer":{"name":"Sandeep Yadav","email":"sandyada@redhat.com","date":"2022-02-01 02:35:11.000000000","tz":330},"subject":"Adding vars for cert","message":"Adding vars for cert\n\nIf a cert is needed to access resources, it needs to be uploaded to\nthe image and mounted to the base container.\n\nThis is needed in downstream where after enabling ssl for DLRN\n- container and image build is failing because necessary certs\nare not available.\n\nChange-Id: I303b32a3081c60739ad650ef5afc90e239a0d46e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/cac255c9892494c246d0b2343b9a96942a9eec2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/cac255c9892494c246d0b2343b9a96942a9eec2d"}]},"branch":"refs/heads/master"},"94e861a9a6aa25a9abbdfbecbe1e732693199bf5":{"kind":"REWORK","_number":13,"created":"2022-02-01 10:16:59.000000000","uploader":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"ref":"refs/changes/21/826521/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-ci","ref":"refs/changes/21/826521/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/21/826521/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/21/826521/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/21/826521/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-ci refs/changes/21/826521/13"}}},"commit":{"parents":[{"commit":"051457b7cd20347240a871126d2943d20722b78c","subject":"Merge \"Ensure we get a fully updated venv\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/051457b7cd20347240a871126d2943d20722b78c"}]}],"author":{"name":"Ronelle Landy","email":"rlandy@redhat.com","date":"2022-01-26 20:07:49.000000000","tz":-300},"committer":{"name":"Sandeep Yadav","email":"sandyada@redhat.com","date":"2022-02-01 10:16:30.000000000","tz":330},"subject":"Adding vars for cert","message":"Adding vars for cert\n\nIf a cert is needed to access resources, it needs to be uploaded to\nthe image and mounted to the base container.\n\nThis is needed in downstream where after enabling ssl for DLRN\n- container and image build is failing because necessary certs\nare not available.\n\nChange-Id: I303b32a3081c60739ad650ef5afc90e239a0d46e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/94e861a9a6aa25a9abbdfbecbe1e732693199bf5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/94e861a9a6aa25a9abbdfbecbe1e732693199bf5"}]},"branch":"refs/heads/master"},"be768a03a25d2a2a939e4476eadbb4ae27e67f33":{"kind":"REWORK","_number":14,"created":"2022-02-01 10:20:10.000000000","uploader":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"ref":"refs/changes/21/826521/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-ci","ref":"refs/changes/21/826521/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/21/826521/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/21/826521/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/21/826521/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-ci refs/changes/21/826521/14"}}},"commit":{"parents":[{"commit":"051457b7cd20347240a871126d2943d20722b78c","subject":"Merge \"Ensure we get a fully updated venv\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/051457b7cd20347240a871126d2943d20722b78c"}]}],"author":{"name":"Ronelle Landy","email":"rlandy@redhat.com","date":"2022-01-26 20:07:49.000000000","tz":-300},"committer":{"name":"Sandeep Yadav","email":"sandyada@redhat.com","date":"2022-02-01 10:19:52.000000000","tz":330},"subject":"Adding vars for cert","message":"Adding vars for cert\n\nIf a cert is needed to access resources, it needs to be uploaded to\nthe image and mounted to the base container.\n\nThis is needed in downstream where after enabling ssl for DLRN\n- container and image build is failing because necessary certs\nare not available.\n\nChange-Id: I303b32a3081c60739ad650ef5afc90e239a0d46e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/be768a03a25d2a2a939e4476eadbb4ae27e67f33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/be768a03a25d2a2a939e4476eadbb4ae27e67f33"}]},"branch":"refs/heads/master"},"c7d2dd9fe249f9d837d6ed58d15e739f3b6d7dc2":{"kind":"REWORK","_number":15,"created":"2022-02-01 15:18:20.000000000","uploader":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"ref":"refs/changes/21/826521/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-ci","ref":"refs/changes/21/826521/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/21/826521/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/21/826521/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/21/826521/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-ci refs/changes/21/826521/15"}}},"commit":{"parents":[{"commit":"051457b7cd20347240a871126d2943d20722b78c","subject":"Merge \"Ensure we get a fully updated venv\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/051457b7cd20347240a871126d2943d20722b78c"}]}],"author":{"name":"Ronelle Landy","email":"rlandy@redhat.com","date":"2022-01-26 20:07:49.000000000","tz":-300},"committer":{"name":"Sandeep Yadav","email":"sandyada@redhat.com","date":"2022-02-01 15:17:58.000000000","tz":330},"subject":"Adding vars for cert","message":"Adding vars for cert\n\nIf a cert is needed to access resources, it needs to be uploaded to\nthe image and mounted to the base container.\n\nThis is needed in downstream where after enabling ssl for DLRN\n- container and image build is failing because necessary certs\nare not available.\n\nDepends-on: https://review.opendev.org/c/openstack/tripleo-repos/+/827280\n\nChange-Id: I303b32a3081c60739ad650ef5afc90e239a0d46e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/c7d2dd9fe249f9d837d6ed58d15e739f3b6d7dc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/c7d2dd9fe249f9d837d6ed58d15e739f3b6d7dc2"}]},"branch":"refs/heads/master"},"846e0f0763e8f1b5e5cc6642d522da5d5b1175f5":{"kind":"REWORK","_number":16,"created":"2022-02-01 16:34:18.000000000","uploader":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"ref":"refs/changes/21/826521/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-ci","ref":"refs/changes/21/826521/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/21/826521/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/21/826521/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/21/826521/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-ci refs/changes/21/826521/16"}}},"commit":{"parents":[{"commit":"051457b7cd20347240a871126d2943d20722b78c","subject":"Merge \"Ensure we get a fully updated venv\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/051457b7cd20347240a871126d2943d20722b78c"}]}],"author":{"name":"Ronelle Landy","email":"rlandy@redhat.com","date":"2022-01-26 20:07:49.000000000","tz":-300},"committer":{"name":"Sandeep Yadav","email":"sandyada@redhat.com","date":"2022-02-01 16:33:44.000000000","tz":330},"subject":"Adding vars for cert","message":"Adding vars for cert\n\nIf a cert is needed to access resources, it needs to be uploaded to\nthe image and mounted to the base container.\n\nThis is needed in downstream where after enabling ssl for DLRN\n- container and image build is failing because necessary certs\nare not available.\n\nDepends-on: https://review.opendev.org/c/openstack/tripleo-repos/+/827280\n\nChange-Id: I303b32a3081c60739ad650ef5afc90e239a0d46e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/846e0f0763e8f1b5e5cc6642d522da5d5b1175f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/846e0f0763e8f1b5e5cc6642d522da5d5b1175f5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
