)]}'
{"id":"openstack%2Ftripleo-common~663441","triplet_id":"openstack%2Ftripleo-common~master~I46cc7c366531cf92214968a051d2a50b3671f71c","project":"openstack/tripleo-common","branch":"master","topic":"bp/reduce-deployment-resources","hashtags":[],"change_id":"I46cc7c366531cf92214968a051d2a50b3671f71c","subject":"Populate Ansible Host variables with AnsibleHostVarsMap","status":"MERGED","created":"2019-06-05 21:39:04.000000000","updated":"2019-06-18 23:46:53.000000000","submitted":"2019-06-18 23:46:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663441-1560901613918-eb0806c3","meta_rev_id":"fbb5db73debfa1a4db40e4acda1a33b3c332293f","_number":663441,"virtual_id_number":663441,"owner":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-06-18 23:46:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-06-17 14:33:04.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-06-18 20:42:13.000000000","_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":2,"date":"2019-06-18 20:41:34.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2019-06-18 20:42:13.000000000","_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-17 14:33:04.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-18 20:41:34.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-06-18 20:42:13.000000000","updated_by":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"reviewer":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"state":"REVIEWER"},{"updated":"2019-06-18 23:46:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"266b712608545dae6cbba358a70a5b4e6cf0f626","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-05 21:39:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"25273911601bf7ca69b36657af11dd9ee9c00e4b","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-05 21:51:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"54cbdb15b3ac3ca1b7b56aa821605bb7e5dd8807","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-05 22:11:03.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/41/663441/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/f184829/ : FAILURE in 17m 11s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/41/663441/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/53cbb22/ : FAILURE in 17m 54s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/41/663441/2/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/afe1ad0/ : FAILURE in 15m 17s","accounts_in_message":[],"_revision_number":2},{"id":"7290adc57671bfcbd7767d212ff017e9b801cfb6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 01:07:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/41/663441/2/check/openstack-tox-cover/4dc6467/ : FAILURE in 4m 53s\n- openstack-tox-pep8 http://logs.openstack.org/41/663441/2/check/openstack-tox-pep8/804ecf9/ : FAILURE in 5m 20s\n- openstack-tox-py27 http://logs.openstack.org/41/663441/2/check/openstack-tox-py27/1235d8d/ : FAILURE in 4m 59s\n- openstack-tox-py36 http://logs.openstack.org/41/663441/2/check/openstack-tox-py36/b26488b/ : FAILURE in 4m 11s\n- openstack-tox-py37 http://logs.openstack.org/41/663441/2/check/openstack-tox-py37/64113c8/ : FAILURE in 6m 02s\n- openstack-tox-docs http://logs.openstack.org/41/663441/2/check/openstack-tox-docs/c6ef9e9/html/ : SUCCESS in 4m 31s\n- tripleo-ci-centos-7-containers-multinode tripleo-ci-centos-7-containers-multinode : SKIPPED\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/41/663441/2/check/tripleo-ci-centos-7-undercloud-containers/7d12c31/ : FAILURE in 25m 14s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades : SKIPPED (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/41/663441/2/check/tripleo-ci-centos-7-standalone/fe391f6/ : FAILURE in 21m 38s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/41/663441/2/check/tripleo-ci-fedora-28-standalone/ca37f2e/ : FAILURE in 25m 02s\n- tripleo-ci-centos-7-standalone-upgrade tripleo-ci-centos-7-standalone-upgrade : SKIPPED (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/41/663441/2/check/tripleo-ci-centos-7-standalone-upgrade-stein/d00f518/ : SUCCESS in 1h 55m 00s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades tripleo-ci-centos-7-containerized-undercloud-upgrades : SKIPPED (non-voting)\n- openstack-tox-py35 http://logs.openstack.org/41/663441/2/check/openstack-tox-py35/653e040/ : FAILURE in 4m 37s\n- openstack-tox-lower-constraints http://logs.openstack.org/41/663441/2/check/openstack-tox-lower-constraints/2f25f0d/ : FAILURE in 7m 02s","accounts_in_message":[],"_revision_number":2},{"id":"d6682014838124bfcd27aa879a162973e8e91cbd","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-06 01:25:34.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d1e32581d34d7e4d7c14126558ab7a5d7f11acd7","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-06 01:45:07.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/41/663441/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/404748e/ : FAILURE in 18m 47s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/41/663441/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/1216c2c/ : FAILURE in 17m 06s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/41/663441/3/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/d3df641/ : FAILURE in 14m 35s","accounts_in_message":[],"_revision_number":3},{"id":"47687b3c156ba9e1e228d0f55a0a528f04eeeec0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 03:58:29.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/41/663441/3/check/openstack-tox-cover/f3a3c28/ : FAILURE in 5m 13s\n- openstack-tox-pep8 http://logs.openstack.org/41/663441/3/check/openstack-tox-pep8/f927069/ : SUCCESS in 4m 24s\n- openstack-tox-py27 http://logs.openstack.org/41/663441/3/check/openstack-tox-py27/da6a641/ : FAILURE in 4m 49s\n- openstack-tox-py36 http://logs.openstack.org/41/663441/3/check/openstack-tox-py36/a39ac9e/ : FAILURE in 4m 59s\n- openstack-tox-py37 http://logs.openstack.org/41/663441/3/check/openstack-tox-py37/4a0e04c/ : FAILURE in 4m 50s\n- openstack-tox-docs http://logs.openstack.org/41/663441/3/check/openstack-tox-docs/d1255e2/ : POST_FAILURE in 5m 21s\n- tripleo-ci-centos-7-containers-multinode tripleo-ci-centos-7-containers-multinode : SKIPPED\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/41/663441/3/check/tripleo-ci-centos-7-undercloud-containers/d8b8b91/ : FAILURE in 21m 50s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades : SKIPPED (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/41/663441/3/check/tripleo-ci-centos-7-standalone/afdd6db/ : FAILURE in 21m 00s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/41/663441/3/check/tripleo-ci-fedora-28-standalone/7518396/ : FAILURE in 40m 45s\n- tripleo-ci-centos-7-standalone-upgrade tripleo-ci-centos-7-standalone-upgrade : SKIPPED (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/41/663441/3/check/tripleo-ci-centos-7-standalone-upgrade-stein/828803c/ : SUCCESS in 2h 18m 23s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades tripleo-ci-centos-7-containerized-undercloud-upgrades : SKIPPED (non-voting)\n- openstack-tox-py35 http://logs.openstack.org/41/663441/3/check/openstack-tox-py35/8441e73/ : FAILURE in 4m 39s\n- openstack-tox-lower-constraints http://logs.openstack.org/41/663441/3/check/openstack-tox-lower-constraints/7bf8852/ : FAILURE in 6m 09s","accounts_in_message":[],"_revision_number":3},{"id":"839321477a631784404488896c6b80a2528a11ff","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-06 13:16:02.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"85c59877576fe486a4222ba17b65b710520d80ff","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-06 15:47:32.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/41/663441/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/2952ecd/ : SUCCESS in 2h 27m 00s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/41/663441/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/55e7093/ : SUCCESS in 2h 28m 23s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/41/663441/4/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/3f8d7cd/ : SUCCESS in 1h 17m 43s","accounts_in_message":[],"_revision_number":4},{"id":"13881755d370011b1511cf15f3b8b17e063af2e2","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-06 16:43:32.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"3a78f74d3b5990f2f15da79dee0ca0c415ab2b7f","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-06 16:45:13.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"bc0c1a086ba310b94d00ee6b7d9ddd146b7af457","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-06 16:52:47.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"ae98a9ded99836bb747e5a27d6db9e7639ba5466","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 16:53:27.000000000","message":"Patch Set 7: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":7},{"id":"33328450ae3bb8866f46250900d58fddf2c1b917","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-06 16:53:27.000000000","message":"Patch Set 7: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":7},{"id":"ad22607f6fe8324f0fffee981a1ca5e17ddfbeeb","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-06 16:53:30.000000000","message":"Patch Set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"917d4b4c2aae01bc0e52c758e4bb8b0705d175b5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-06 17:19:29.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/41/663441/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/1c71a96/ : FAILURE in 18m 22s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/41/663441/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/a50a779/ : FAILURE in 19m 49s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/41/663441/8/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/bc7ab5e/ : FAILURE in 13m 57s","accounts_in_message":[],"_revision_number":8},{"id":"5d72824d3b139f7f59ddb6795d184937a03d0773","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-06 19:36:27.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"1a8a18100f155260c5ddb4c3532d4a1383a04c5a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-06 19:54:49.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/41/663441/9/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/7c2fb94/ : FAILURE in 17m 38s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/41/663441/9/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/35f2393/ : FAILURE in 16m 26s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/41/663441/9/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/45c64d3/ : FAILURE in 13m 33s","accounts_in_message":[],"_revision_number":9},{"id":"c483f6667e5880226c10f16e79d4ab85472506df","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 02:32:12.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/41/663441/9/check/openstack-tox-cover/47f7bce/cover/ : SUCCESS in 5m 20s\n- openstack-tox-pep8 http://logs.openstack.org/41/663441/9/check/openstack-tox-pep8/ae953a4/ : SUCCESS in 6m 20s\n- openstack-tox-py27 http://logs.openstack.org/41/663441/9/check/openstack-tox-py27/ec781e1/ : SUCCESS in 4m 06s\n- openstack-tox-py36 http://logs.openstack.org/41/663441/9/check/openstack-tox-py36/37d1ff6/ : SUCCESS in 5m 13s\n- openstack-tox-py37 http://logs.openstack.org/41/663441/9/check/openstack-tox-py37/1f7b66b/ : SUCCESS in 5m 22s\n- openstack-tox-docs http://logs.openstack.org/41/663441/9/check/openstack-tox-docs/0812f65/html/ : SUCCESS in 8m 31s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/41/663441/9/check/tripleo-ci-centos-7-containers-multinode/dac0466/ : FAILURE in 27m 29s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/41/663441/9/check/tripleo-ci-centos-7-undercloud-containers/5ff0200/ : FAILURE in 24m 07s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/41/663441/9/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/a05adbf/ : FAILURE in 29m 09s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/41/663441/9/check/tripleo-ci-centos-7-standalone/03741b6/ : FAILURE in 22m 03s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/41/663441/9/check/tripleo-ci-fedora-28-standalone/1b28242/ : SUCCESS in 2h 01m 40s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/41/663441/9/check/tripleo-ci-centos-7-standalone-upgrade/c87fe49/ : FAILURE in 1h 23m 44s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/41/663441/9/check/tripleo-ci-centos-7-standalone-upgrade-stein/6537235/ : SUCCESS in 2h 04m 06s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/41/663441/9/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/f551df9/ : FAILURE in 1h 08m 58s (non-voting)\n- openstack-tox-py35 http://logs.openstack.org/41/663441/9/check/openstack-tox-py35/8d77cbe/ : SUCCESS in 4m 32s\n- openstack-tox-lower-constraints http://logs.openstack.org/41/663441/9/check/openstack-tox-lower-constraints/9bb139f/ : FAILURE in 5m 42s","accounts_in_message":[],"_revision_number":9},{"id":"0b07e5ead3768184d325b3e870e7b2a2ed91c389","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-07 13:32:42.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"88a8df75b79be189f915a12a56d9b1a9b46683c8","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-07 16:04:21.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/41/663441/10/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/f94cd53/ : SUCCESS in 2h 30m 05s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/41/663441/10/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/811dcbc/ : SUCCESS in 2h 30m 01s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/41/663441/10/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/ecba41f/ : SUCCESS in 1h 18m 04s","accounts_in_message":[],"_revision_number":10},{"id":"ad8f34c762613883ab20d35169a25fb4d9af935c","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-07 20:00:33.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.","accounts_in_message":[],"_revision_number":11},{"id":"74bc3f5b1c80ad2b36acb3ae3cce56e83e4eb999","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 20:03:20.000000000","message":"Patch Set 11: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":11},{"id":"c04a8d56a2d18d7e44a1aeaa619e12e2ff47a004","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-07 20:03:21.000000000","message":"Patch Set 11: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":11},{"id":"84b2c98f4a27c13f604094cf7837d2df36d314a6","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-10 23:43:45.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.","accounts_in_message":[],"_revision_number":12},{"id":"dd70867de70f22face561d4a04dd983ede2eb396","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-11 00:24:44.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/41/663441/12/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/b5f4d2c/ : FAILURE in 39m 16s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/41/663441/12/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/f21f1d9/ : FAILURE in 37m 55s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/41/663441/12/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/03af584/ : FAILURE in 33m 15s","accounts_in_message":[],"_revision_number":12},{"id":"c57e0c63c97e6f4ea6c4153a6ecd25aa485f050d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 01:50:52.000000000","message":"Patch Set 12: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":12},{"id":"279510e26dfcbd5679cb1d8d2d83eba4a574720b","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-11 17:02:22.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"ba718d9cf23f91bb45aa220e4e97a9bdca42fc7c","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-11 17:02:33.000000000","message":"Patch Set 13: Patch Set 12 was rebased","accounts_in_message":[],"_revision_number":13},{"id":"6dbf43666f9c61a9eca3ffbbecb19948ab909334","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 17:56:28.000000000","message":"Patch Set 13: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":13},{"id":"fb7ff0af8cc927a93e00afe010f73471ae501121","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-11 17:56:29.000000000","message":"Patch Set 13: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":13},{"id":"e8d011e41acb8065ac40ecce8622f02581d8d1c9","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-11 17:58:09.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"73ba6cc3058acd3450d08fff58f366d021e84a67","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 18:05:10.000000000","message":"Patch Set 14: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":14},{"id":"dde45fa7b95251e978cc2c90f1f1002cfcccca0a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-11 18:05:15.000000000","message":"Patch Set 14: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":14},{"id":"f6ecf8a9b8a207009f6e15059c32e799f022e044","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2019-06-11 18:33:56.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"78869a9adb92ea137c0a5ba4cf56270cb81152ed","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2019-06-11 20:49:34.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"f74fc3c9559ce026236fbbe8d9e8bc90fd3193f4","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-11 23:28:36.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/41/663441/16/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/7b7e773/ : SUCCESS in 2h 31m 54s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/41/663441/16/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/76baf8f/ : SUCCESS in 2h 36m 39s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/41/663441/16/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/2ec197f/ : SUCCESS in 1h 16m 11s","accounts_in_message":[],"_revision_number":16},{"id":"3f7206b280fcaa6ac09f5e4ba8fab6bd7d055d28","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 10:53:19.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/41/663441/16/check/openstack-tox-cover/9d2bd56/cover/ : SUCCESS in 6m 33s\n- openstack-tox-pep8 http://logs.openstack.org/41/663441/16/check/openstack-tox-pep8/dd00fe0/ : FAILURE in 5m 06s\n- openstack-tox-py27 http://logs.openstack.org/41/663441/16/check/openstack-tox-py27/b407c1d/ : SUCCESS in 4m 54s\n- openstack-tox-py36 http://logs.openstack.org/41/663441/16/check/openstack-tox-py36/c3a0b6e/ : SUCCESS in 5m 03s\n- openstack-tox-py37 http://logs.openstack.org/41/663441/16/check/openstack-tox-py37/7e2870b/ : SUCCESS in 5m 25s\n- openstack-tox-docs http://logs.openstack.org/41/663441/16/check/openstack-tox-docs/d2556f9/html/ : SUCCESS in 5m 03s\n- tripleo-ci-centos-7-containers-multinode tripleo-ci-centos-7-containers-multinode : SKIPPED\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/41/663441/16/check/tripleo-ci-centos-7-undercloud-containers/ce12237/ : SUCCESS in 1h 40m 59s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades : SKIPPED (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/41/663441/16/check/tripleo-ci-centos-7-standalone/484f776/ : SUCCESS in 2h 00m 16s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/41/663441/16/check/tripleo-ci-fedora-28-standalone/81bbb13/ : SUCCESS in 1h 44m 03s\n- tripleo-ci-centos-7-standalone-upgrade tripleo-ci-centos-7-standalone-upgrade : SKIPPED (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/41/663441/16/check/tripleo-ci-centos-7-standalone-upgrade-stein/8cf7d53/ : SUCCESS in 2h 06m 44s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades tripleo-ci-centos-7-containerized-undercloud-upgrades : SKIPPED (non-voting)\n- openstack-tox-py35 http://logs.openstack.org/41/663441/16/check/openstack-tox-py35/cfdf4ae/ : SUCCESS in 6m 35s\n- openstack-tox-lower-constraints http://logs.openstack.org/41/663441/16/check/openstack-tox-lower-constraints/8b81ce2/ : SUCCESS in 7m 39s","accounts_in_message":[],"_revision_number":16},{"id":"b32f0f5bfa4ba507b9764c1a07d2e229c43e26c4","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-12 14:59:47.000000000","message":"Patch Set 17: Patch Set 16 was rebased","accounts_in_message":[],"_revision_number":17},{"id":"503feaa19af9cda9ae6cedbeecbc2778f44c2791","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 15:00:21.000000000","message":"Patch Set 17: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":17},{"id":"fe115a6c7d5be7cb5b2e0a455a815cc937aa94e8","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-12 15:00:25.000000000","message":"Patch Set 17: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":17},{"id":"2c5d43aaed0d3763de28d878e1156f6b8bbb93d9","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-12 15:01:01.000000000","message":"Patch Set 17:\n\nrecheck","accounts_in_message":[],"_revision_number":17},{"id":"f473d5f8a409f0d4523f44ffd0d83449c98485fc","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-12 18:38:10.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/41/663441/17/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/916457f/ : SUCCESS in 2h 57m 55s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/41/663441/17/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/f63ae82/ : SUCCESS in 2h 51m 09s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/41/663441/17/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/467b529/ : SUCCESS in 1h 21m 06s","accounts_in_message":[],"_revision_number":17},{"id":"08a8868b7a8e42d9ef00fe6aa97d0deb19d6baf0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 04:38:39.000000000","message":"Patch Set 17:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/41/663441/17/check/openstack-tox-cover/0c16776/cover/ : SUCCESS in 3m 48s\n- openstack-tox-pep8 http://logs.openstack.org/41/663441/17/check/openstack-tox-pep8/b8c67be/ : FAILURE in 3m 34s\n- openstack-tox-py27 http://logs.openstack.org/41/663441/17/check/openstack-tox-py27/5e38067/ : SUCCESS in 3m 45s\n- openstack-tox-py36 http://logs.openstack.org/41/663441/17/check/openstack-tox-py36/6faf171/ : SUCCESS in 3m 11s\n- openstack-tox-py37 http://logs.openstack.org/41/663441/17/check/openstack-tox-py37/040b3e2/ : SUCCESS in 3m 29s\n- openstack-tox-docs http://logs.openstack.org/41/663441/17/check/openstack-tox-docs/467f863/html/ : SUCCESS in 3m 14s\n- tripleo-ci-centos-7-containers-multinode tripleo-ci-centos-7-containers-multinode : SKIPPED\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/41/663441/17/check/tripleo-ci-centos-7-undercloud-containers/58544af/ : SUCCESS in 1h 42m 44s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades : SKIPPED (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/41/663441/17/check/tripleo-ci-centos-7-standalone/2d0de56/ : SUCCESS in 2h 01m 00s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/41/663441/17/check/tripleo-ci-fedora-28-standalone/7c0ce48/ : SUCCESS in 2h 00m 52s\n- tripleo-ci-centos-7-standalone-upgrade tripleo-ci-centos-7-standalone-upgrade : SKIPPED (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/41/663441/17/check/tripleo-ci-centos-7-standalone-upgrade-stein/c5079ee/ : SUCCESS in 2h 16m 21s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades tripleo-ci-centos-7-containerized-undercloud-upgrades : SKIPPED (non-voting)\n- openstack-tox-py35 http://logs.openstack.org/41/663441/17/check/openstack-tox-py35/1f83a5c/ : SUCCESS in 3m 40s\n- openstack-tox-lower-constraints http://logs.openstack.org/41/663441/17/check/openstack-tox-lower-constraints/a446d37/ : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":17},{"id":"5c1a19b9e1762d10eee74e36874d05110aca0865","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-13 13:55:13.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"9d4d249c554da4c426f1f8317d9c2769935dbaca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 16:04:04.000000000","message":"Patch Set 18: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":18},{"id":"dce52e19ef75a8b98e03e8fe3bafc5256dd3b054","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-13 16:38:06.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/41/663441/18/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/908629f/ : SUCCESS in 2h 35m 17s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/41/663441/18/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/6e7313b/ : SUCCESS in 2h 40m 00s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/41/663441/18/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/f5e464e/ : SUCCESS in 1h 18m 19s","accounts_in_message":[],"_revision_number":18},{"id":"f4481b9d83b1feafdc46951d1036cda7fba01395","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-14 14:53:27.000000000","message":"Patch Set 19: Patch Set 18 was rebased","accounts_in_message":[],"_revision_number":19},{"id":"95c6065bb04131e9a0342b8433dc3243e5ae7009","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 14:53:43.000000000","message":"Patch Set 19: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":19},{"id":"ce2dd8af32f57d3482e8f8c4db55dae061a01b0c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-14 15:01:41.000000000","message":"Patch Set 19: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":19},{"id":"34e2a106dcca626f2066561f919c8e03b3d45e2f","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-17 12:05:53.000000000","message":"Patch Set 20: Patch Set 19 was rebased","accounts_in_message":[],"_revision_number":20},{"id":"5de22317470b55a090659ee16d2062cd829caa11","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 12:06:05.000000000","message":"Patch Set 20: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":20},{"id":"81f77d1439b69f2f91533c4c77ced6283f2c28bc","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-17 12:06:09.000000000","message":"Patch Set 20: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":20},{"id":"c29855b1fe39f3857c68eee7ee8c8a056501d9c6","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-17 12:08:06.000000000","message":"Patch Set 21: Commit message was updated.","accounts_in_message":[],"_revision_number":21},{"id":"002a82d336e1a32d0bb03a5a94c83d6dd1781e8c","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-17 12:13:38.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"f29f0679915ad5d9fd770dde9d597961de0c201c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-17 14:33:04.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/41/663441/22/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/b15da34/ : FAILURE in 1h 12m 39s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/41/663441/22/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/33e2dac/ : FAILURE in 1h 19m 10s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/41/663441/22/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/dc3de40/ : SUCCESS in 1h 42m 29s","accounts_in_message":[],"_revision_number":22},{"id":"a1cbc413d9bd2f0d5bddc8616d970326eab3165a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 18:29:33.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/41/663441/22/check/openstack-tox-cover/a51880f/cover/ : SUCCESS in 3m 58s\n- openstack-tox-pep8 http://logs.openstack.org/41/663441/22/check/openstack-tox-pep8/4686e92/ : SUCCESS in 4m 01s\n- openstack-tox-py27 http://logs.openstack.org/41/663441/22/check/openstack-tox-py27/a11cd66/ : SUCCESS in 3m 41s\n- openstack-tox-py36 http://logs.openstack.org/41/663441/22/check/openstack-tox-py36/e9c51be/ : SUCCESS in 3m 58s\n- openstack-tox-py37 http://logs.openstack.org/41/663441/22/check/openstack-tox-py37/1ec59f2/ : SUCCESS in 3m 47s\n- openstack-tox-docs http://logs.openstack.org/41/663441/22/check/openstack-tox-docs/9a25ae1/html/ : SUCCESS in 3m 41s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/41/663441/22/check/tripleo-ci-centos-7-containers-multinode/0917933/ : SUCCESS in 2h 21m 45s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/41/663441/22/check/tripleo-ci-centos-7-undercloud-containers/e8449fb/ : SUCCESS in 1h 34m 37s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/41/663441/22/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/6065e50/ : SUCCESS in 2h 36m 34s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/41/663441/22/check/tripleo-ci-centos-7-standalone/040959b/ : SUCCESS in 2h 01m 24s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/41/663441/22/check/tripleo-ci-fedora-28-standalone/654d276/ : SUCCESS in 1h 55m 11s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/41/663441/22/check/tripleo-ci-centos-7-standalone-upgrade/6b4939d/ : SUCCESS in 2h 09m 03s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/41/663441/22/check/tripleo-ci-centos-7-standalone-upgrade-stein/97ee4c7/ : SUCCESS in 2h 04m 02s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/41/663441/22/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/b495723/ : FAILURE in 1h 24m 17s (non-voting)\n- openstack-tox-py35 http://logs.openstack.org/41/663441/22/check/openstack-tox-py35/b7eb2ff/ : SUCCESS in 5m 30s\n- openstack-tox-lower-constraints http://logs.openstack.org/41/663441/22/check/openstack-tox-lower-constraints/ad9579d/ : SUCCESS in 6m 35s","accounts_in_message":[],"_revision_number":22},{"id":"01559f1ab50722a4aede58a228772b3707e5c3af","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-18 20:41:34.000000000","message":"Patch Set 22: Code-Review+2","accounts_in_message":[],"_revision_number":22},{"id":"b17bca3a7b2b95b3d632230f09d759cb8d6fe79f","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2019-06-18 20:42:13.000000000","message":"Patch Set 22: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":22},{"id":"52470d0f87d32409d36ffbaa136096e4bf1c274a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 20:42:32.000000000","message":"Patch Set 22: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":22},{"id":"1d6668e45b47a66dc848bf5ed8acc24178dd502d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 23:46:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":22},{"id":"255b5c7bef05a07669eb31445bed9c67dda95da6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 23:46:53.000000000","message":"Patch Set 22: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/41/663441/22/gate/openstack-tox-pep8/b0d5ea3/ : SUCCESS in 3m 36s\n- openstack-tox-py27 http://logs.openstack.org/41/663441/22/gate/openstack-tox-py27/405698b/ : SUCCESS in 4m 16s\n- openstack-tox-py36 http://logs.openstack.org/41/663441/22/gate/openstack-tox-py36/909db61/ : SUCCESS in 4m 06s\n- openstack-tox-py37 http://logs.openstack.org/41/663441/22/gate/openstack-tox-py37/22cda45/ : SUCCESS in 4m 19s\n- openstack-tox-docs http://logs.openstack.org/41/663441/22/gate/openstack-tox-docs/c0ba4a9/html/ : SUCCESS in 3m 56s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/41/663441/22/gate/tripleo-ci-centos-7-containers-multinode/b96f1ed/ : SUCCESS in 2h 06m 02s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/41/663441/22/gate/tripleo-ci-centos-7-standalone/eacb763/ : SUCCESS in 2h 01m 04s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/41/663441/22/gate/tripleo-ci-fedora-28-standalone/2437883/ : SUCCESS in 1h 49m 09s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/41/663441/22/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/e00cf81/ : SUCCESS in 2h 23m 40s\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/41/663441/22/gate/tripleo-ci-centos-7-standalone-upgrade-stein/804e818/ : SUCCESS in 2h 16m 56s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/41/663441/22/gate/tripleo-ci-centos-7-undercloud-containers/fcfd6d8/ : SUCCESS in 1h 31m 42s\n- openstack-tox-lower-constraints http://logs.openstack.org/41/663441/22/gate/openstack-tox-lower-constraints/afc0d44/ : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":22}],"current_revision_number":22,"current_revision":"0c2efab56302d11fdfdd496c5ee04b1456109744","revisions":{"3e33d374d9fab9b5107ed68a982cde1c556e9e7d":{"kind":"REWORK","_number":1,"created":"2019-06-05 21:39:04.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/41/663441/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/41/663441/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/1"}}},"commit":{"parents":[{"commit":"d46a9fb18e8ecff4896cb96f2c6f45de2c7e6c05","subject":"Merge \"Add tripleo-hieradata role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/d46a9fb18e8ecff4896cb96f2c6f45de2c7e6c05"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-05 21:38:27.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-05 21:38:27.000000000","tz":-240},"subject":"WIP - Populate Ansible Host variables with AnsibleHostVarsMap","message":"WIP - Populate Ansible Host variables with AnsibleHostVarsMap\n\nChange-Id: I46cc7c366531cf92214968a051d2a50b3671f71c\nDepends-On: I3880f0f72beb24ee0b0868dca48afce6328144d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/3e33d374d9fab9b5107ed68a982cde1c556e9e7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/3e33d374d9fab9b5107ed68a982cde1c556e9e7d"}]},"branch":"refs/heads/master"},"1f1be7048628d1e0089cc2416a565e52bc9ddda3":{"kind":"REWORK","_number":2,"created":"2019-06-05 21:51:37.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/41/663441/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/41/663441/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/2"}}},"commit":{"parents":[{"commit":"d46a9fb18e8ecff4896cb96f2c6f45de2c7e6c05","subject":"Merge \"Add tripleo-hieradata role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/d46a9fb18e8ecff4896cb96f2c6f45de2c7e6c05"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-05 21:38:27.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-05 21:51:32.000000000","tz":-240},"subject":"WIP - Populate Ansible Host variables with AnsibleHostVarsMap","message":"WIP - Populate Ansible Host variables with AnsibleHostVarsMap\n\nChange-Id: I46cc7c366531cf92214968a051d2a50b3671f71c\nDepends-On: I3880f0f72beb24ee0b0868dca48afce6328144d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/1f1be7048628d1e0089cc2416a565e52bc9ddda3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/1f1be7048628d1e0089cc2416a565e52bc9ddda3"}]},"branch":"refs/heads/master"},"ff5502667dd45a1b2f3fa0407119169b58da53bc":{"kind":"REWORK","_number":3,"created":"2019-06-06 01:25:34.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/41/663441/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/41/663441/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/3"}}},"commit":{"parents":[{"commit":"d46a9fb18e8ecff4896cb96f2c6f45de2c7e6c05","subject":"Merge \"Add tripleo-hieradata role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/d46a9fb18e8ecff4896cb96f2c6f45de2c7e6c05"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-05 21:38:27.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-06 01:25:24.000000000","tz":-240},"subject":"WIP - Populate Ansible Host variables with AnsibleHostVarsMap","message":"WIP - Populate Ansible Host variables with AnsibleHostVarsMap\n\nChange-Id: I46cc7c366531cf92214968a051d2a50b3671f71c\nDepends-On: I3880f0f72beb24ee0b0868dca48afce6328144d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/ff5502667dd45a1b2f3fa0407119169b58da53bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/ff5502667dd45a1b2f3fa0407119169b58da53bc"}]},"branch":"refs/heads/master"},"c2ee811c43356923a0f0f005023303e1eabc05bf":{"kind":"REWORK","_number":4,"created":"2019-06-06 13:16:02.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/41/663441/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/41/663441/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/4"}}},"commit":{"parents":[{"commit":"d46a9fb18e8ecff4896cb96f2c6f45de2c7e6c05","subject":"Merge \"Add tripleo-hieradata role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/d46a9fb18e8ecff4896cb96f2c6f45de2c7e6c05"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-05 21:38:27.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-06 13:15:48.000000000","tz":-240},"subject":"WIP - Populate Ansible Host variables with AnsibleHostVarsMap","message":"WIP - Populate Ansible Host variables with AnsibleHostVarsMap\n\nChange-Id: I46cc7c366531cf92214968a051d2a50b3671f71c\nDepends-On: I3880f0f72beb24ee0b0868dca48afce6328144d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/c2ee811c43356923a0f0f005023303e1eabc05bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/c2ee811c43356923a0f0f005023303e1eabc05bf"}]},"branch":"refs/heads/master"},"7fd9a191c7289aa4352d6909847f43ae8e6ea291":{"kind":"REWORK","_number":5,"created":"2019-06-06 16:43:32.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/41/663441/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/41/663441/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/5"}}},"commit":{"parents":[{"commit":"d46a9fb18e8ecff4896cb96f2c6f45de2c7e6c05","subject":"Merge \"Add tripleo-hieradata role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/d46a9fb18e8ecff4896cb96f2c6f45de2c7e6c05"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-05 21:38:27.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-06 16:41:27.000000000","tz":-240},"subject":"Populate Ansible Host variables with AnsibleHostVarsMap","message":"Populate Ansible Host variables with AnsibleHostVarsMap\n\nThe new interface in THT AnsibleHostVarsMap will be useful to lay down\nthe Ansible Host variables.\nThey are populated in THT but generated in config-download using the\nsame mechanism as the Group variables.\n\nChange-Id: I46cc7c366531cf92214968a051d2a50b3671f71c\nDepends-On: I3880f0f72beb24ee0b0868dca48afce6328144d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/7fd9a191c7289aa4352d6909847f43ae8e6ea291"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/7fd9a191c7289aa4352d6909847f43ae8e6ea291"}]},"branch":"refs/heads/master"},"551d6e0e24c14d7622a7c67f14c381db1ea64b81":{"kind":"REWORK","_number":6,"created":"2019-06-06 16:45:13.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/41/663441/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/41/663441/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/6"}}},"commit":{"parents":[{"commit":"d46a9fb18e8ecff4896cb96f2c6f45de2c7e6c05","subject":"Merge \"Add tripleo-hieradata role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/d46a9fb18e8ecff4896cb96f2c6f45de2c7e6c05"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-05 21:38:27.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-06 16:45:01.000000000","tz":-240},"subject":"Populate Ansible Host variables with AnsibleHostVarsMap","message":"Populate Ansible Host variables with AnsibleHostVarsMap\n\nThe new interface in THT AnsibleHostVarsMap will be useful to lay down\nthe Ansible Host variables.\nThey are populated in THT but generated in config-download using the\nsame mechanism as the Group variables.\n\nChange-Id: I46cc7c366531cf92214968a051d2a50b3671f71c\nDepends-On: I3880f0f72beb24ee0b0868dca48afce6328144d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/551d6e0e24c14d7622a7c67f14c381db1ea64b81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/551d6e0e24c14d7622a7c67f14c381db1ea64b81"}]},"branch":"refs/heads/master"},"2783d6451d56245d2b0c0db3b5a3d613bf11d2db":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-06-06 16:52:47.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/41/663441/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/41/663441/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/7"}}},"commit":{"parents":[{"commit":"6e448dcbed4517db9092fe256f1b41f70a7564fc","subject":"Merge \"Add the ability to compute osds number counting lvm devices\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/6e448dcbed4517db9092fe256f1b41f70a7564fc"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-05 21:38:27.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-06 16:52:47.000000000","tz":0},"subject":"Populate Ansible Host variables with AnsibleHostVarsMap","message":"Populate Ansible Host variables with AnsibleHostVarsMap\n\nThe new interface in THT AnsibleHostVarsMap will be useful to lay down\nthe Ansible Host variables.\nThey are populated in THT but generated in config-download using the\nsame mechanism as the Group variables.\n\nChange-Id: I46cc7c366531cf92214968a051d2a50b3671f71c\nDepends-On: I3880f0f72beb24ee0b0868dca48afce6328144d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/2783d6451d56245d2b0c0db3b5a3d613bf11d2db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/2783d6451d56245d2b0c0db3b5a3d613bf11d2db"}]},"branch":"refs/heads/master"},"a19ec7b4666966404ae3489604d3ad08f764b8c7":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2019-06-06 16:53:30.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/41/663441/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/41/663441/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/8"}}},"commit":{"parents":[{"commit":"6e448dcbed4517db9092fe256f1b41f70a7564fc","subject":"Merge \"Add the ability to compute osds number counting lvm devices\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/6e448dcbed4517db9092fe256f1b41f70a7564fc"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-05 21:38:27.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-06 16:53:28.000000000","tz":0},"subject":"Populate Ansible Host variables with AnsibleHostVarsMap","message":"Populate Ansible Host variables with AnsibleHostVarsMap\n\nThe new interface in THT AnsibleHostVarsMap will be useful to lay down\nthe Ansible Host variables.\nThey are populated in THT but generated in config-download using the\nsame mechanism as the Group variables.\n\nChange-Id: I46cc7c366531cf92214968a051d2a50b3671f71c\n\nDepends-On: I3880f0f72beb24ee0b0868dca48afce6328144d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/a19ec7b4666966404ae3489604d3ad08f764b8c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/a19ec7b4666966404ae3489604d3ad08f764b8c7"}]},"branch":"refs/heads/master"},"dfdd9673a8e9479f7bffec3ec03c743b7fb27d8d":{"kind":"REWORK","_number":9,"created":"2019-06-06 19:36:27.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/41/663441/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/41/663441/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/9"}}},"commit":{"parents":[{"commit":"6e448dcbed4517db9092fe256f1b41f70a7564fc","subject":"Merge \"Add the ability to compute osds number counting lvm devices\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/6e448dcbed4517db9092fe256f1b41f70a7564fc"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-05 21:38:27.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-06 19:36:17.000000000","tz":0},"subject":"Populate Ansible Host variables with AnsibleHostVarsMap","message":"Populate Ansible Host variables with AnsibleHostVarsMap\n\nThe new interface in THT AnsibleHostVarsMap will be useful to lay down\nthe Ansible Host variables.\nThey are populated in THT but generated in config-download using the\nsame mechanism as the Group variables.\n\nChange-Id: I46cc7c366531cf92214968a051d2a50b3671f71c\n\nDepends-On: I3880f0f72beb24ee0b0868dca48afce6328144d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/dfdd9673a8e9479f7bffec3ec03c743b7fb27d8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/dfdd9673a8e9479f7bffec3ec03c743b7fb27d8d"}]},"branch":"refs/heads/master"},"2ecee322bde97cd05c938da46be95e4236262e98":{"kind":"REWORK","_number":10,"created":"2019-06-07 13:32:42.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/41/663441/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/41/663441/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/10"}}},"commit":{"parents":[{"commit":"6e448dcbed4517db9092fe256f1b41f70a7564fc","subject":"Merge \"Add the ability to compute osds number counting lvm devices\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/6e448dcbed4517db9092fe256f1b41f70a7564fc"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-05 21:38:27.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-07 13:32:28.000000000","tz":-240},"subject":"Populate Ansible Host variables with AnsibleHostVarsMap","message":"Populate Ansible Host variables with AnsibleHostVarsMap\n\nThe new interface in THT AnsibleHostVarsMap will be useful to lay down\nthe Ansible Host variables.\nThey are populated in THT but generated in config-download using the\nsame mechanism as the Group variables.\n\nChange-Id: I46cc7c366531cf92214968a051d2a50b3671f71c\nDepends-On: I3880f0f72beb24ee0b0868dca48afce6328144d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/2ecee322bde97cd05c938da46be95e4236262e98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/2ecee322bde97cd05c938da46be95e4236262e98"}]},"branch":"refs/heads/master"},"10a93422ba10b33f4281c42f19aaef16802c5600":{"kind":"NO_CHANGE","_number":11,"created":"2019-06-07 20:00:33.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/41/663441/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/41/663441/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/11"}}},"commit":{"parents":[{"commit":"6e448dcbed4517db9092fe256f1b41f70a7564fc","subject":"Merge \"Add the ability to compute osds number counting lvm devices\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/6e448dcbed4517db9092fe256f1b41f70a7564fc"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-07 20:00:08.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-07 20:00:08.000000000","tz":-240},"subject":"Populate Ansible Host variables with AnsibleHostVarsMap","message":"Populate Ansible Host variables with AnsibleHostVarsMap\n\nThe new interface in THT AnsibleHostVarsMap will be useful to lay down\nthe Ansible Host variables.\nThey are populated in THT but generated in config-download using the\nsame mechanism as the Group variables.\n\nChange-Id: I46cc7c366531cf92214968a051d2a50b3671f71c\nDepends-On: I3880f0f72beb24ee0b0868dca48afce6328144d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/10a93422ba10b33f4281c42f19aaef16802c5600"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/10a93422ba10b33f4281c42f19aaef16802c5600"}]},"branch":"refs/heads/master"},"39bcebea14dabfb333fac82f4342ff5d2236f972":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2019-06-10 23:43:45.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/41/663441/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/41/663441/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/12"}}},"commit":{"parents":[{"commit":"48cbc438d952391222607129c98ed5b893c4d12d","subject":"Merge \"Remove ProcessTemplatesAction as base class\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/48cbc438d952391222607129c98ed5b893c4d12d"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-07 20:00:08.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-10 21:09:41.000000000","tz":-240},"subject":"Populate Ansible Host variables with AnsibleHostVarsMap","message":"Populate Ansible Host variables with AnsibleHostVarsMap\n\nThe new interface in THT AnsibleHostVarsMap will be useful to lay down\nthe Ansible Host variables.\nThey are populated in THT but generated in config-download using the\nsame mechanism as the Group variables.\n\nChange-Id: I46cc7c366531cf92214968a051d2a50b3671f71c\nDepends-On: I3880f0f72beb24ee0b0868dca48afce6328144d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/39bcebea14dabfb333fac82f4342ff5d2236f972"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/39bcebea14dabfb333fac82f4342ff5d2236f972"}]},"branch":"refs/heads/master"},"12c551660531c57eae3850a7bf0410d0a1998aac":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2019-06-11 17:02:33.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/41/663441/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/41/663441/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/13"}}},"commit":{"parents":[{"commit":"3fdc95b987fbe9c4081b0c41f52b9babdd1dbb98","subject":"Merge \"utils/config: generate param_config.json if param_config exists\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/3fdc95b987fbe9c4081b0c41f52b9babdd1dbb98"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-07 20:00:08.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-11 17:02:33.000000000","tz":0},"subject":"Populate Ansible Host variables with AnsibleHostVarsMap","message":"Populate Ansible Host variables with AnsibleHostVarsMap\n\nThe new interface in THT AnsibleHostVarsMap will be useful to lay down\nthe Ansible Host variables.\nThey are populated in THT but generated in config-download using the\nsame mechanism as the Group variables.\n\nChange-Id: I46cc7c366531cf92214968a051d2a50b3671f71c\nDepends-On: I3880f0f72beb24ee0b0868dca48afce6328144d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/12c551660531c57eae3850a7bf0410d0a1998aac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/12c551660531c57eae3850a7bf0410d0a1998aac"}]},"branch":"refs/heads/master"},"c64a32baec172c409e2acf6161063b76cffd9fff":{"kind":"REWORK","_number":14,"created":"2019-06-11 17:58:09.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/41/663441/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/41/663441/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/14"}}},"commit":{"parents":[{"commit":"71cfb353836bb4f19f134632c75405b61074c6bc","subject":"Render NetworkConfig into config-download dir","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/71cfb353836bb4f19f134632c75405b61074c6bc"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-07 20:00:08.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-11 17:57:55.000000000","tz":-240},"subject":"Populate Ansible Host variables with AnsibleHostVarsMap","message":"Populate Ansible Host variables with AnsibleHostVarsMap\n\nThe new interface in THT AnsibleHostVarsMap will be useful to lay down\nthe Ansible Host variables.\nThey are populated in THT but generated in config-download using the\nsame mechanism as the Group variables.\n\nChange-Id: I46cc7c366531cf92214968a051d2a50b3671f71c\nDepends-On: I3880f0f72beb24ee0b0868dca48afce6328144d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/c64a32baec172c409e2acf6161063b76cffd9fff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/c64a32baec172c409e2acf6161063b76cffd9fff"}]},"branch":"refs/heads/master"},"6057ab194dcbbc35fc6dfb1053a53ca3b8c8a7ff":{"kind":"REWORK","_number":15,"created":"2019-06-11 18:33:56.000000000","uploader":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"ref":"refs/changes/41/663441/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/41/663441/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/15"}}},"commit":{"parents":[{"commit":"d4b63550bfe08fbd10b0e6e6fb31adfb7c6bdbf3","subject":"Render NetworkConfig into config-download dir","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/d4b63550bfe08fbd10b0e6e6fb31adfb7c6bdbf3"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-07 20:00:08.000000000","tz":-240},"committer":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2019-06-11 18:33:48.000000000","tz":-240},"subject":"Populate Ansible Host variables with AnsibleHostVarsMap","message":"Populate Ansible Host variables with AnsibleHostVarsMap\n\nThe new interface in THT AnsibleHostVarsMap will be useful to lay down\nthe Ansible Host variables.\nThey are populated in THT but generated in config-download using the\nsame mechanism as the Group variables.\n\nChange-Id: I46cc7c366531cf92214968a051d2a50b3671f71c\nDepends-On: I3880f0f72beb24ee0b0868dca48afce6328144d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/6057ab194dcbbc35fc6dfb1053a53ca3b8c8a7ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/6057ab194dcbbc35fc6dfb1053a53ca3b8c8a7ff"}]},"branch":"refs/heads/master"},"24e73dd4b93ca704b80faf9fe0dd3b0b9430ed8c":{"kind":"REWORK","_number":16,"created":"2019-06-11 20:49:34.000000000","uploader":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"ref":"refs/changes/41/663441/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/41/663441/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/16"}}},"commit":{"parents":[{"commit":"d4b63550bfe08fbd10b0e6e6fb31adfb7c6bdbf3","subject":"Render NetworkConfig into config-download dir","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/d4b63550bfe08fbd10b0e6e6fb31adfb7c6bdbf3"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-07 20:00:08.000000000","tz":-240},"committer":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2019-06-11 20:49:27.000000000","tz":-240},"subject":"Populate Ansible Host variables with AnsibleHostVarsMap","message":"Populate Ansible Host variables with AnsibleHostVarsMap\n\nThe new interface in THT AnsibleHostVarsMap will be useful to lay down\nthe Ansible Host variables.\nThey are populated in THT but generated in config-download using the\nsame mechanism as the Group variables.\n\nChange-Id: I46cc7c366531cf92214968a051d2a50b3671f71c\nDepends-On: I3880f0f72beb24ee0b0868dca48afce6328144d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/24e73dd4b93ca704b80faf9fe0dd3b0b9430ed8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/24e73dd4b93ca704b80faf9fe0dd3b0b9430ed8c"}]},"branch":"refs/heads/master"},"666550d7d0b14d1c73f24689da7f15a4010026a0":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2019-06-12 14:59:47.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/41/663441/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/41/663441/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/17"}}},"commit":{"parents":[{"commit":"0ecdbd575f6c014347b13d4905cb7814ffb451eb","subject":"Render NetworkConfig into config-download dir","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/0ecdbd575f6c014347b13d4905cb7814ffb451eb"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-07 20:00:08.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-12 14:59:47.000000000","tz":0},"subject":"Populate Ansible Host variables with AnsibleHostVarsMap","message":"Populate Ansible Host variables with AnsibleHostVarsMap\n\nThe new interface in THT AnsibleHostVarsMap will be useful to lay down\nthe Ansible Host variables.\nThey are populated in THT but generated in config-download using the\nsame mechanism as the Group variables.\n\nChange-Id: I46cc7c366531cf92214968a051d2a50b3671f71c\nDepends-On: I3880f0f72beb24ee0b0868dca48afce6328144d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/666550d7d0b14d1c73f24689da7f15a4010026a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/666550d7d0b14d1c73f24689da7f15a4010026a0"}]},"branch":"refs/heads/master"},"083c7bafae86125435f1485ef896c9d554d2246a":{"kind":"REWORK","_number":18,"created":"2019-06-13 13:55:13.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/41/663441/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/41/663441/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/18"}}},"commit":{"parents":[{"commit":"0ecdbd575f6c014347b13d4905cb7814ffb451eb","subject":"Render NetworkConfig into config-download dir","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/0ecdbd575f6c014347b13d4905cb7814ffb451eb"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-07 20:00:08.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-13 13:55:01.000000000","tz":-240},"subject":"Populate Ansible Host variables with AnsibleHostVarsMap","message":"Populate Ansible Host variables with AnsibleHostVarsMap\n\nThe new interface in THT AnsibleHostVarsMap will be useful to lay down\nthe Ansible Host variables.\nThey are populated in THT but generated in config-download using the\nsame mechanism as the Group variables.\n\nChange-Id: I46cc7c366531cf92214968a051d2a50b3671f71c\nDepends-On: I3880f0f72beb24ee0b0868dca48afce6328144d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/083c7bafae86125435f1485ef896c9d554d2246a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/083c7bafae86125435f1485ef896c9d554d2246a"}]},"branch":"refs/heads/master"},"5ea24044ca55b713e442766105feb34c2653eadb":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2019-06-14 14:53:27.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/41/663441/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/41/663441/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/19"}}},"commit":{"parents":[{"commit":"85b7a4eea5c9a36f5a4d2bc4bef5836e2338ab8d","subject":"Render NetworkConfig into config-download dir","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/85b7a4eea5c9a36f5a4d2bc4bef5836e2338ab8d"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-07 20:00:08.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-14 14:53:27.000000000","tz":0},"subject":"Populate Ansible Host variables with AnsibleHostVarsMap","message":"Populate Ansible Host variables with AnsibleHostVarsMap\n\nThe new interface in THT AnsibleHostVarsMap will be useful to lay down\nthe Ansible Host variables.\nThey are populated in THT but generated in config-download using the\nsame mechanism as the Group variables.\n\nChange-Id: I46cc7c366531cf92214968a051d2a50b3671f71c\nDepends-On: I3880f0f72beb24ee0b0868dca48afce6328144d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/5ea24044ca55b713e442766105feb34c2653eadb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/5ea24044ca55b713e442766105feb34c2653eadb"}]},"branch":"refs/heads/master"},"3abc03c2145fe9653d83c2f2d34ad7a725802284":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2019-06-17 12:05:53.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/41/663441/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/41/663441/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/20"}}},"commit":{"parents":[{"commit":"a0a9713b6b3dc0ac8994505f58e9877681f02f45","subject":"Merge \"Render NetworkConfig into config-download dir\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/a0a9713b6b3dc0ac8994505f58e9877681f02f45"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-07 20:00:08.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-17 12:05:53.000000000","tz":0},"subject":"Populate Ansible Host variables with AnsibleHostVarsMap","message":"Populate Ansible Host variables with AnsibleHostVarsMap\n\nThe new interface in THT AnsibleHostVarsMap will be useful to lay down\nthe Ansible Host variables.\nThey are populated in THT but generated in config-download using the\nsame mechanism as the Group variables.\n\nChange-Id: I46cc7c366531cf92214968a051d2a50b3671f71c\nDepends-On: I3880f0f72beb24ee0b0868dca48afce6328144d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/3abc03c2145fe9653d83c2f2d34ad7a725802284"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/3abc03c2145fe9653d83c2f2d34ad7a725802284"}]},"branch":"refs/heads/master"},"5b304fe90418f4a01a9c84a7025068962a9163fc":{"kind":"NO_CODE_CHANGE","_number":21,"created":"2019-06-17 12:08:06.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/41/663441/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/41/663441/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/21"}}},"commit":{"parents":[{"commit":"a0a9713b6b3dc0ac8994505f58e9877681f02f45","subject":"Merge \"Render NetworkConfig into config-download dir\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/a0a9713b6b3dc0ac8994505f58e9877681f02f45"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-07 20:00:08.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-17 12:08:02.000000000","tz":0},"subject":"Populate Ansible Host variables with AnsibleHostVarsMap","message":"Populate Ansible Host variables with AnsibleHostVarsMap\n\nThe new interface in THT AnsibleHostVarsMap will be useful to lay down\nthe Ansible Host variables.\nThey are populated in THT but generated in config-download using the\nsame mechanism as the Group variables.\n\nChange-Id: I46cc7c366531cf92214968a051d2a50b3671f71c\n\nDepends-On: I3880f0f72beb24ee0b0868dca48afce6328144d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/5b304fe90418f4a01a9c84a7025068962a9163fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/5b304fe90418f4a01a9c84a7025068962a9163fc"}]},"branch":"refs/heads/master"},"0c2efab56302d11fdfdd496c5ee04b1456109744":{"kind":"REWORK","_number":22,"created":"2019-06-17 12:13:38.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/41/663441/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/41/663441/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/41/663441/22"}}},"commit":{"parents":[{"commit":"a0a9713b6b3dc0ac8994505f58e9877681f02f45","subject":"Merge \"Render NetworkConfig into config-download dir\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/a0a9713b6b3dc0ac8994505f58e9877681f02f45"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-07 20:00:08.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-17 12:13:30.000000000","tz":-240},"subject":"Populate Ansible Host variables with AnsibleHostVarsMap","message":"Populate Ansible Host variables with AnsibleHostVarsMap\n\nThe new interface in THT AnsibleHostVarsMap will be useful to lay down\nthe Ansible Host variables.\nThey are populated in THT but generated in config-download using the\nsame mechanism as the Group variables.\n\nChange-Id: I46cc7c366531cf92214968a051d2a50b3671f71c\nDepends-On: I3880f0f72beb24ee0b0868dca48afce6328144d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/0c2efab56302d11fdfdd496c5ee04b1456109744"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/0c2efab56302d11fdfdd496c5ee04b1456109744"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
