)]}'
{"id":"openstack%2Ftripleo-common~664053","triplet_id":"openstack%2Ftripleo-common~master~I83e5e4f8dd5bb9acd4e6b21bf86f729bfc5447d4","project":"openstack/tripleo-common","branch":"master","hashtags":[],"change_id":"I83e5e4f8dd5bb9acd4e6b21bf86f729bfc5447d4","subject":"Add bindep support to tox","status":"MERGED","created":"2019-06-07 20:03:34.000000000","updated":"2019-06-12 06:05:04.000000000","submitted":"2019-06-12 06:05:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664053-1560319504168-91b8215b","meta_rev_id":"ec32f2cd2c98db28bc37a490d734882dbf08b2fd","_number":664053,"virtual_id_number":664053,"owner":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2019-06-08 01:57:31.000000000","_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-06-12 06:05:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-08 04:20:19.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-06-12 00:56:12.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":2,"date":"2019-06-10 20:40:11.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-06-12 00:56:12.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-08 04:20:19.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-10 20:40:11.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-06-12 00:56:12.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-06-12 06:05:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"632cd93d25f034055865e96758b0484b750ffdb3","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-06-07 20:03:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"32c85a5f21057113304fd1f9f68edba8f9efc599","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-06-07 20:14:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d04656efec306916f06e5c9cedb27d997f78afad","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-07 22:39:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/53/664053/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/353559b/ : FAILURE in 1h 40m 13s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/53/664053/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/e5d70a8/ : SUCCESS in 2h 24m 04s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/53/664053/2/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/c97df64/ : SUCCESS in 1h 18m 51s","accounts_in_message":[],"_revision_number":2},{"id":"199e898b3677a8d6c4418b0ba2918b406c63acf8","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-06-08 01:57:31.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"0cbdea7d9d81139009d018ec02da923975b7805b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-08 04:20:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/53/664053/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/ec858dc/ : SUCCESS in 2h 15m 00s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/53/664053/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/c57bc75/ : SUCCESS in 2h 21m 12s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/53/664053/2/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/c42d0da/ : SUCCESS in 1h 08m 56s","accounts_in_message":[],"_revision_number":2},{"id":"0760b6255f5faafd6c4291eb488423da4b810736","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-08 10:06:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/53/664053/2/check/openstack-tox-cover/c40be71/cover/ : SUCCESS in 4m 10s\n- openstack-tox-pep8 http://logs.openstack.org/53/664053/2/check/openstack-tox-pep8/6604ee1/ : SUCCESS in 3m 10s\n- openstack-tox-py27 http://logs.openstack.org/53/664053/2/check/openstack-tox-py27/b125e56/ : SUCCESS in 3m 19s\n- openstack-tox-py36 http://logs.openstack.org/53/664053/2/check/openstack-tox-py36/ef9a22d/ : SUCCESS in 3m 35s\n- openstack-tox-py37 http://logs.openstack.org/53/664053/2/check/openstack-tox-py37/b893130/ : SUCCESS in 3m 37s\n- openstack-tox-docs http://logs.openstack.org/53/664053/2/check/openstack-tox-docs/692e029/html/ : SUCCESS in 3m 05s\n- build-openstack-releasenotes http://logs.openstack.org/53/664053/2/check/build-openstack-releasenotes/5dad0a0/html/ : SUCCESS in 3m 19s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/53/664053/2/check/tripleo-ci-centos-7-containers-multinode/89b76de/ : SUCCESS in 1h 57m 47s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/53/664053/2/check/tripleo-ci-centos-7-undercloud-containers/2178b2e/ : SUCCESS in 1h 39m 27s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/53/664053/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/a506fb1/ : SUCCESS in 2h 15m 34s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/53/664053/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/18a6c0f/ : SUCCESS in 2h 27m 06s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/53/664053/2/check/tripleo-ci-centos-7-standalone/73d5bfe/ : SUCCESS in 1h 52m 59s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/53/664053/2/check/tripleo-ci-fedora-28-standalone/ce2874c/ : SUCCESS in 1h 51m 56s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/53/664053/2/check/tripleo-ci-centos-7-standalone-upgrade/2631790/ : SUCCESS in 2h 03m 13s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/53/664053/2/check/tripleo-ci-centos-7-standalone-upgrade-stein/cde44b4/ : SUCCESS in 2h 02m 42s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/53/664053/2/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/63f02db/ : FAILURE in 1h 24m 17s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/53/664053/2/check/tripleo-build-containers-centos-7/514214c/ : SUCCESS in 51m 15s\n- tripleo-build-containers-centos-7-stein http://logs.openstack.org/53/664053/2/check/tripleo-build-containers-centos-7-stein/7c61c3b/ : SUCCESS in 44m 42s\n- tripleo-build-containers-centos-7-rocky http://logs.openstack.org/53/664053/2/check/tripleo-build-containers-centos-7-rocky/56ae61e/ : SUCCESS in 45m 03s\n- tripleo-build-containers-centos-7-buildah http://logs.openstack.org/53/664053/2/check/tripleo-build-containers-centos-7-buildah/1527813/ : SUCCESS in 55m 21s\n- tripleo-build-containers-centos-7-buildah-stein http://logs.openstack.org/53/664053/2/check/tripleo-build-containers-centos-7-buildah-stein/d44d8d1/ : SUCCESS in 1h 02m 37s\n- openstack-tox-py35 http://logs.openstack.org/53/664053/2/check/openstack-tox-py35/1262d29/ : SUCCESS in 3m 55s\n- openstack-tox-lower-constraints http://logs.openstack.org/53/664053/2/check/openstack-tox-lower-constraints/c1f1e7e/ : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":2},{"id":"f615a2fe53fc166008080eff37f26fd21f4c2ac7","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-10 20:40:11.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f7e9ca0e2f8fe133eb4d8c417822245397595ef2","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-12 00:56:12.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"61a27490cba2f722129ab6e26b0a24649fddb03c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 00:56:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"007c7ef49a102ec5b7d2df2090d3c9c2b4ad5c4b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 06:05:03.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/53/664053/2/gate/openstack-tox-pep8/47181c6/ : SUCCESS in 3m 19s\n- openstack-tox-py27 http://logs.openstack.org/53/664053/2/gate/openstack-tox-py27/b202ea6/ : SUCCESS in 3m 58s\n- openstack-tox-py36 http://logs.openstack.org/53/664053/2/gate/openstack-tox-py36/1c6bc9d/ : SUCCESS in 3m 56s\n- openstack-tox-py37 http://logs.openstack.org/53/664053/2/gate/openstack-tox-py37/2b5976b/ : SUCCESS in 3m 34s\n- openstack-tox-docs http://logs.openstack.org/53/664053/2/gate/openstack-tox-docs/128fe0b/html/ : SUCCESS in 4m 17s\n- build-openstack-releasenotes http://logs.openstack.org/53/664053/2/gate/build-openstack-releasenotes/dceca83/html/ : SUCCESS in 4m 09s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/53/664053/2/gate/tripleo-ci-centos-7-containers-multinode/479afa1/ : SUCCESS in 2h 29m 58s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/53/664053/2/gate/tripleo-ci-centos-7-standalone/da5f842/ : SUCCESS in 2h 10m 52s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/53/664053/2/gate/tripleo-ci-fedora-28-standalone/61d8d77/ : SUCCESS in 2h 10m 30s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/53/664053/2/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/f9c9bfd/ : SUCCESS in 2h 40m 22s\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/53/664053/2/gate/tripleo-ci-centos-7-standalone-upgrade-stein/4f3d8bf/ : SUCCESS in 2h 14m 07s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/53/664053/2/gate/tripleo-ci-centos-7-undercloud-containers/12c3155/ : SUCCESS in 1h 45m 56s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/53/664053/2/gate/tripleo-build-containers-centos-7/1ac17aa/ : SUCCESS in 46m 50s\n- tripleo-build-containers-centos-7-stein http://logs.openstack.org/53/664053/2/gate/tripleo-build-containers-centos-7-stein/5feaa66/ : SUCCESS in 48m 45s\n- tripleo-build-containers-centos-7-rocky http://logs.openstack.org/53/664053/2/gate/tripleo-build-containers-centos-7-rocky/8dcf0c8/ : SUCCESS in 43m 55s\n- tripleo-build-containers-centos-7-buildah http://logs.openstack.org/53/664053/2/gate/tripleo-build-containers-centos-7-buildah/bf86d40/ : SUCCESS in 53m 12s\n- tripleo-build-containers-centos-7-buildah-stein http://logs.openstack.org/53/664053/2/gate/tripleo-build-containers-centos-7-buildah-stein/54d75a0/ : SUCCESS in 1h 06m 39s\n- openstack-tox-lower-constraints http://logs.openstack.org/53/664053/2/gate/openstack-tox-lower-constraints/873990c/ : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":2},{"id":"ec32f2cd2c98db28bc37a490d734882dbf08b2fd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 06:05:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8862ea83858db56367f0fdea18faabc3c0841f90","revisions":{"4641e53fe1c5953962641a590531a25b223198c4":{"kind":"REWORK","_number":1,"created":"2019-06-07 20:03:34.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/53/664053/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/53/664053/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/53/664053/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/53/664053/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/53/664053/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/53/664053/1"}}},"commit":{"parents":[{"commit":"6e448dcbed4517db9092fe256f1b41f70a7564fc","subject":"Merge \"Add the ability to compute osds number counting lvm devices\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/6e448dcbed4517db9092fe256f1b41f70a7564fc"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2019-06-07 19:42:58.000000000","tz":-300},"committer":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2019-06-07 20:03:27.000000000","tz":-300},"subject":"Add bindep support to tox","message":"Add bindep support to tox\n\nThis change will allow a developer to run bindep via tox to install all of\nthe required system packages needed to run general tests.\n\n\u003e Integration `tox -e bindep`\n\nThis will also allow CI to using the bindep file to install basic required\npackages as needed.\n\nChange-Id: I83e5e4f8dd5bb9acd4e6b21bf86f729bfc5447d4\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/4641e53fe1c5953962641a590531a25b223198c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/4641e53fe1c5953962641a590531a25b223198c4"}]},"branch":"refs/heads/master"},"8862ea83858db56367f0fdea18faabc3c0841f90":{"kind":"REWORK","_number":2,"created":"2019-06-07 20:14:17.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/53/664053/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/53/664053/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/53/664053/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/53/664053/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/53/664053/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/53/664053/2"}}},"commit":{"parents":[{"commit":"6e448dcbed4517db9092fe256f1b41f70a7564fc","subject":"Merge \"Add the ability to compute osds number counting lvm devices\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/6e448dcbed4517db9092fe256f1b41f70a7564fc"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2019-06-07 19:42:58.000000000","tz":-300},"committer":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2019-06-07 20:14:01.000000000","tz":-300},"subject":"Add bindep support to tox","message":"Add bindep support to tox\n\nThis change will allow a developer to run bindep via tox to install all of\nthe required system packages needed to run general tests.\n\n\u003e Integration `tox -e bindep`\n\nThis will also allow CI to using the bindep file to install basic required\npackages as needed.\n\nChange-Id: I83e5e4f8dd5bb9acd4e6b21bf86f729bfc5447d4\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/8862ea83858db56367f0fdea18faabc3c0841f90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/8862ea83858db56367f0fdea18faabc3c0841f90"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
