)]}'
{"id":"openstack%2Ftripleo-common~668938","triplet_id":"openstack%2Ftripleo-common~master~I3c0580893aed5bda46d9104ffcee0f00cbabd9ee","project":"openstack/tripleo-common","branch":"master","topic":"stop-pre-upgrade","hashtags":[],"change_id":"I3c0580893aed5bda46d9104ffcee0f00cbabd9ee","subject":"Add playbook to stop containers","status":"ABANDONED","created":"2019-07-03 15:27:26.000000000","updated":"2019-07-08 12:38:45.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"13a739d3565af74f033b7000514ba23b9cbadf77","_number":668938,"virtual_id_number":668938,"owner":{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},{"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},{"value":1,"date":"2019-07-05 14:59:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-05 15:03:40.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":17823,"name":"Michele Baldessari","email":"michele@redhat.com","username":"mbaldessari"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"disliked":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":-1,"date":"2019-07-08 12:23:51.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},{"value":1,"date":"2019-07-05 12:01:33.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17823,"name":"Michele Baldessari","email":"michele@redhat.com","username":"mbaldessari"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},{"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":17823,"name":"Michele Baldessari","email":"michele@redhat.com","username":"mbaldessari"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},{"_account_id":17823,"name":"Michele Baldessari","email":"michele@redhat.com","username":"mbaldessari"},{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-03 15:27:54.000000000","updated_by":{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},"reviewer":{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},"state":"REVIEWER"},{"updated":"2019-07-03 15:29:12.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"REVIEWER"},{"updated":"2019-07-03 15:29:28.000000000","updated_by":{"_account_id":17823,"name":"Michele Baldessari","email":"michele@redhat.com","username":"mbaldessari"},"reviewer":{"_account_id":17823,"name":"Michele Baldessari","email":"michele@redhat.com","username":"mbaldessari"},"state":"REVIEWER"},{"updated":"2019-07-04 08:00:53.000000000","updated_by":{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},"reviewer":{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},"state":"REVIEWER"},{"updated":"2019-07-04 14:42:36.000000000","updated_by":{"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},"reviewer":{"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},"state":"REVIEWER"},{"updated":"2019-07-05 12:01:33.000000000","updated_by":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"reviewer":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"state":"REVIEWER"},{"updated":"2019-07-05 14:59:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-05 15:03:40.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-08 12:23:51.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"}],"messages":[{"id":"292c96cd2f3bb2cf2a67ea9f36f00732ef51a68c","author":{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},"date":"2019-07-03 15:27:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aced842bcf8f1b6efbcb36d4f3e91bc02e183c3b","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-07-03 15:29:12.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6d88c8e742d771fbaf4388dd4bf5e61ec7994fa6","author":{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},"date":"2019-07-03 15:29:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"21b42bd36b2aff226e853c15ea81fc2979123f05","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-03 18:18:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/38/668938/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/3bf7ef7/ : FAILURE in 1h 57m 55s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/38/668938/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/fc4a5df/ : SUCCESS in 2h 47m 00s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/38/668938/2/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/3beb420/ : SUCCESS in 1h 40m 11s","accounts_in_message":[],"_revision_number":2},{"id":"63fb854dd6df80a79f0ea9c9201a4d04d47ac955","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 19:24:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/38/668938/2/check/openstack-tox-cover/5c6a71a/cover/ : SUCCESS in 4m 12s\n- openstack-tox-pep8 http://logs.openstack.org/38/668938/2/check/openstack-tox-pep8/8d970b3/ : SUCCESS in 3m 12s\n- openstack-tox-py27 http://logs.openstack.org/38/668938/2/check/openstack-tox-py27/d55c749/ : SUCCESS in 3m 45s\n- openstack-tox-py36 http://logs.openstack.org/38/668938/2/check/openstack-tox-py36/8025b25/ : SUCCESS in 3m 30s\n- openstack-tox-py37 http://logs.openstack.org/38/668938/2/check/openstack-tox-py37/3403e53/ : SUCCESS in 3m 29s\n- tox-molecule http://logs.openstack.org/38/668938/2/check/tox-molecule/286f1ac/tox/reports.html : SUCCESS in 14m 55s\n- openstack-tox-docs http://logs.openstack.org/38/668938/2/check/openstack-tox-docs/2378aa1/html/ : SUCCESS in 3m 36s\n- tripleo-ci-centos-7-containers-multinode tripleo-ci-centos-7-containers-multinode : SKIPPED\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/38/668938/2/check/tripleo-ci-centos-7-undercloud-containers/e856268/ : SUCCESS in 1h 42m 52s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades : SKIPPED (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/38/668938/2/check/tripleo-ci-centos-7-standalone/07a02f8/ : SUCCESS in 2h 07m 48s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/38/668938/2/check/tripleo-ci-fedora-28-standalone/71043aa/ : FAILURE in 34m 06s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade tripleo-ci-centos-7-standalone-upgrade : SKIPPED (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/38/668938/2/check/tripleo-ci-centos-7-standalone-upgrade-stein/2676f12/ : SUCCESS in 2h 15m 58s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades tripleo-ci-centos-7-containerized-undercloud-upgrades : SKIPPED (non-voting)\n- openstack-tox-linters http://logs.openstack.org/38/668938/2/check/openstack-tox-linters/6829c88/ : FAILURE in 2m 04s\n- openstack-tox-py35 http://logs.openstack.org/38/668938/2/check/openstack-tox-py35/56229ad/ : SUCCESS in 3m 28s\n- openstack-tox-lower-constraints http://logs.openstack.org/38/668938/2/check/openstack-tox-lower-constraints/29d5d79/ : SUCCESS in 4m 50s","accounts_in_message":[],"_revision_number":2},{"id":"8e0b11881330ca903039f468f2c436a6f284b7a7","author":{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},"date":"2019-07-04 07:28:08.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b2feb5912183f03a8e36ade32716c88203478dc9","author":{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},"date":"2019-07-04 08:00:53.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"72af389e813b877eb84b78bd6009264a2cf79774","author":{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},"date":"2019-07-04 08:22:26.000000000","message":"Patch Set 3: Workflow-1\n\nHi! Thanks for the review I had a nit there and I needed to push another PS, Ill address the docker ps comment.","accounts_in_message":[],"_revision_number":3},{"id":"99e2fd6c51c4627f69fd1f528ee0b0b33315e261","author":{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},"date":"2019-07-04 09:41:32.000000000","message":"Topic set to stop-pre-upgrade","accounts_in_message":[],"_revision_number":3},{"id":"4b9b0c347545f5eb626ef22a982a7328557ba990","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 10:12:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/38/668938/3/check/openstack-tox-cover/6a8cded/cover/ : SUCCESS in 4m 05s\n- openstack-tox-pep8 http://logs.openstack.org/38/668938/3/check/openstack-tox-pep8/c8e625f/ : SUCCESS in 3m 26s\n- openstack-tox-py27 http://logs.openstack.org/38/668938/3/check/openstack-tox-py27/664f975/ : SUCCESS in 4m 01s\n- openstack-tox-py36 http://logs.openstack.org/38/668938/3/check/openstack-tox-py36/0ae81b6/ : SUCCESS in 3m 41s\n- openstack-tox-py37 http://logs.openstack.org/38/668938/3/check/openstack-tox-py37/dda6a10/ : SUCCESS in 3m 56s\n- tox-molecule http://logs.openstack.org/38/668938/3/check/tox-molecule/fb0fac2/tox/reports.html : SUCCESS in 15m 16s\n- openstack-tox-docs http://logs.openstack.org/38/668938/3/check/openstack-tox-docs/0aaa8ae/html/ : SUCCESS in 3m 49s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/38/668938/3/check/tripleo-ci-centos-7-containers-multinode/6b23481/ : SUCCESS in 2h 25m 14s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/38/668938/3/check/tripleo-ci-centos-7-undercloud-containers/0989b82/ : SUCCESS in 1h 33m 28s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/38/668938/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/6efad2b/ : SUCCESS in 2h 26m 08s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/38/668938/3/check/tripleo-ci-centos-7-standalone/81ab132/ : SUCCESS in 2h 09m 51s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/38/668938/3/check/tripleo-ci-fedora-28-standalone/ff0d4b5/ : FAILURE in 29m 58s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/38/668938/3/check/tripleo-ci-centos-7-standalone-upgrade/0287399/ : SUCCESS in 2h 30m 46s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/38/668938/3/check/tripleo-ci-centos-7-standalone-upgrade-stein/fe73ab8/ : SUCCESS in 2h 06m 53s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/38/668938/3/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/dd6e818/ : FAILURE in 2h 29m 39s (non-voting)\n- openstack-tox-linters http://logs.openstack.org/38/668938/3/check/openstack-tox-linters/41bc1ed/ : SUCCESS in 4m 25s\n- openstack-tox-py35 http://logs.openstack.org/38/668938/3/check/openstack-tox-py35/8beb0e2/ : SUCCESS in 3m 40s\n- openstack-tox-lower-constraints http://logs.openstack.org/38/668938/3/check/openstack-tox-lower-constraints/40e89d3/ : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":3},{"id":"26ab9a15fad6a597e908eda73e1abc5030d1543e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-04 10:26:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/38/668938/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/a8a69cd/ : SUCCESS in 2h 56m 21s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/38/668938/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/d32b93d/ : SUCCESS in 2h 44m 00s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/38/668938/3/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/c353144/ : SUCCESS in 1h 22m 58s","accounts_in_message":[],"_revision_number":3},{"id":"521f113fb9ea9121e5dac7054a0db8482c17dbb4","author":{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},"date":"2019-07-04 11:06:38.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"7fbd17a4472a757a60751cd8f85724e721b055ea","author":{"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},"date":"2019-07-04 13:00:38.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2438e6870fad6d83dd83e853d02567b9a7a4652c","author":{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},"date":"2019-07-04 13:31:04.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"119e665813701954c61c699b8aeb6d06bb3d9488","author":{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},"date":"2019-07-04 13:31:21.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"77a1713df4e57e240a53b704c009a07e5ed9dafb","author":{"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},"date":"2019-07-04 14:42:36.000000000","message":"Patch Set 5: Code-Review+1\n\nLGTM, will +2 after we have a bit more comprehensive testing with the t-h-t part","accounts_in_message":[],"_revision_number":5},{"id":"dfcdfaa696d7c55f86ed38a783b5f0deca673f0c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-04 17:06:25.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/38/668938/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/df23481/ : SUCCESS in 3h 32m 42s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/38/668938/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/1eb98b1/ : SUCCESS in 3h 33m 36s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/38/668938/5/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/8b6830c/ : SUCCESS in 1h 29m 13s","accounts_in_message":[],"_revision_number":5},{"id":"f48060f3f2fe0443f97432de1363fda7a259d24f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 18:00:29.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/38/668938/5/check/openstack-tox-cover/80881d9/cover/ : SUCCESS in 4m 10s\n- openstack-tox-pep8 http://logs.openstack.org/38/668938/5/check/openstack-tox-pep8/999e88e/ : SUCCESS in 2m 48s\n- openstack-tox-py27 http://logs.openstack.org/38/668938/5/check/openstack-tox-py27/500294f/ : SUCCESS in 3m 36s\n- openstack-tox-py36 http://logs.openstack.org/38/668938/5/check/openstack-tox-py36/169493b/ : SUCCESS in 3m 26s\n- openstack-tox-py37 http://logs.openstack.org/38/668938/5/check/openstack-tox-py37/018a5f7/ : SUCCESS in 5m 36s\n- tox-molecule http://logs.openstack.org/38/668938/5/check/tox-molecule/2f874e6/tox/reports.html : SUCCESS in 17m 30s\n- openstack-tox-docs http://logs.openstack.org/38/668938/5/check/openstack-tox-docs/176c3dd/html/ : SUCCESS in 3m 06s\n- tripleo-ci-centos-7-containers-multinode tripleo-ci-centos-7-containers-multinode : SKIPPED\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/38/668938/5/check/tripleo-ci-centos-7-undercloud-containers/8ecb854/ : SUCCESS in 1h 37m 26s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades : SKIPPED (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/38/668938/5/check/tripleo-ci-centos-7-standalone/d170efc/ : SUCCESS in 2h 00m 22s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/38/668938/5/check/tripleo-ci-fedora-28-standalone/12dab81/ : FAILURE in 29m 32s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade tripleo-ci-centos-7-standalone-upgrade : SKIPPED (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/38/668938/5/check/tripleo-ci-centos-7-standalone-upgrade-stein/b5ea99d/ : SUCCESS in 2h 02m 13s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades tripleo-ci-centos-7-containerized-undercloud-upgrades : SKIPPED (non-voting)\n- openstack-tox-linters http://logs.openstack.org/38/668938/5/check/openstack-tox-linters/b53b686/ : FAILURE in 4m 23s\n- openstack-tox-py35 http://logs.openstack.org/38/668938/5/check/openstack-tox-py35/b587dfb/ : SUCCESS in 3m 18s\n- openstack-tox-lower-constraints http://logs.openstack.org/38/668938/5/check/openstack-tox-lower-constraints/5d6bbfb/ : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":5},{"id":"2bfbf3f07558cf3a7bf6a0903cc6aab8b972ca79","author":{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},"date":"2019-07-05 07:56:22.000000000","message":"Patch Set 5:\n\nlinter fix","accounts_in_message":[],"_revision_number":5},{"id":"ed175f2ac57a97d38413b92f091d61b348f5048c","author":{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},"date":"2019-07-05 11:58:09.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"570ca1612ce0ffbf196967d720d2958cdecb024f","author":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"date":"2019-07-05 12:01:33.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"fefc680c95fd86faa22f6b0ca0fb633222b02054","author":{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},"date":"2019-07-05 12:17:23.000000000","message":"Patch Set 6: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit d1dffd383f6b4b6d2d4d42fed082e694e677fda8","accounts_in_message":[],"_revision_number":6},{"id":"b4df72c03d01e34ded71d3da2e3ec43725afae0a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 14:59:22.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/38/668938/6/check/openstack-tox-cover/4d3f800/cover/ : SUCCESS in 4m 02s\n- openstack-tox-pep8 http://logs.openstack.org/38/668938/6/check/openstack-tox-pep8/a7feaa2/ : SUCCESS in 3m 28s\n- openstack-tox-py27 http://logs.openstack.org/38/668938/6/check/openstack-tox-py27/4ef2bba/ : SUCCESS in 3m 18s\n- openstack-tox-py36 http://logs.openstack.org/38/668938/6/check/openstack-tox-py36/ee59a6c/ : SUCCESS in 4m 34s\n- openstack-tox-py37 http://logs.openstack.org/38/668938/6/check/openstack-tox-py37/77abff8/ : SUCCESS in 4m 21s\n- tox-molecule http://logs.openstack.org/38/668938/6/check/tox-molecule/4e619b0/tox/reports.html : SUCCESS in 19m 11s\n- openstack-tox-docs http://logs.openstack.org/38/668938/6/check/openstack-tox-docs/0a96cf2/html/ : SUCCESS in 3m 33s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/38/668938/6/check/tripleo-ci-centos-7-containers-multinode/65df447/ : SUCCESS in 2h 11m 39s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/38/668938/6/check/tripleo-ci-centos-7-undercloud-containers/4d5a516/ : SUCCESS in 1h 44m 53s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/38/668938/6/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/5ddf998/ : SUCCESS in 2h 27m 18s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/38/668938/6/check/tripleo-ci-centos-7-standalone/55b5e72/ : SUCCESS in 1h 56m 53s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/38/668938/6/check/tripleo-ci-fedora-28-standalone/c1c7ee0/ : SUCCESS in 1h 46m 36s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/38/668938/6/check/tripleo-ci-centos-7-standalone-upgrade/4cb498e/ : SUCCESS in 2h 28m 56s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/38/668938/6/check/tripleo-ci-centos-7-standalone-upgrade-stein/da69d65/ : SUCCESS in 2h 26m 09s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/38/668938/6/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/b784f79/ : FAILURE in 2h 20m 04s (non-voting)\n- openstack-tox-linters http://logs.openstack.org/38/668938/6/check/openstack-tox-linters/32554c5/ : SUCCESS in 3m 45s\n- openstack-tox-py35 http://logs.openstack.org/38/668938/6/check/openstack-tox-py35/d9d0e1f/ : SUCCESS in 3m 44s\n- openstack-tox-lower-constraints http://logs.openstack.org/38/668938/6/check/openstack-tox-lower-constraints/64148e6/ : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":6},{"id":"7256c35ff3ab99dda664052cba6277c90755620d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-05 15:03:40.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/38/668938/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/ae47e6e/ : SUCCESS in 2h 34m 59s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/38/668938/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/28a84e0/ : SUCCESS in 3h 04m 50s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/38/668938/6/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/172fdf2/ : SUCCESS in 1h 23m 18s","accounts_in_message":[],"_revision_number":6},{"id":"d6d66578236011a962f5c482cfdbeca1ff00ba50","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-07-08 12:23:51.000000000","message":"Patch Set 6: Code-Review-1\n\nplease add this role into tripleo-ansible repo where things are moving now.","accounts_in_message":[],"_revision_number":6},{"id":"13a739d3565af74f033b7000514ba23b9cbadf77","author":{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},"date":"2019-07-08 12:38:45.000000000","message":"Abandoned\n\nAbandoned in favor of https://review.opendev.org/#/c/669650/","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"8be70c7adb2eaba960ee222886382f06746761f5","revisions":{"56768de2d8496d384336ba305d90a62aaecce2f9":{"kind":"REWORK","_number":1,"created":"2019-07-03 15:27:26.000000000","uploader":{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},"ref":"refs/changes/38/668938/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/38/668938/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/38/668938/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/38/668938/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/38/668938/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/38/668938/1"}}},"commit":{"parents":[{"commit":"2ef56851589c7587719562503aad87ef02677f0b","subject":"Merge \"Handle empty NetworkConfig\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/2ef56851589c7587719562503aad87ef02677f0b"}]}],"author":{"name":"Carlos Camacho","email":"ccamacho@redhat.com","date":"2019-07-03 15:25:54.000000000","tz":120},"committer":{"name":"Carlos Camacho","email":"ccamacho@redhat.com","date":"2019-07-03 15:27:24.000000000","tz":120},"subject":"Add playbook to stop containers","message":"Add playbook to stop containers\n\nWe need to stop several containers on the upgrades workflow.\n\nThis submission allows to reduce the duplicated code in THT.\nIt also handles the case in which we have podman as the CLI\nbut the containers running with docker.\n\nChange-Id: I3c0580893aed5bda46d9104ffcee0f00cbabd9ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/56768de2d8496d384336ba305d90a62aaecce2f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/56768de2d8496d384336ba305d90a62aaecce2f9"}]},"branch":"refs/heads/master"},"facf82cafbb61711aac8e28ef064457f19937ef6":{"kind":"REWORK","_number":2,"created":"2019-07-03 15:29:34.000000000","uploader":{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},"ref":"refs/changes/38/668938/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/38/668938/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/38/668938/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/38/668938/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/38/668938/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/38/668938/2"}}},"commit":{"parents":[{"commit":"2ef56851589c7587719562503aad87ef02677f0b","subject":"Merge \"Handle empty NetworkConfig\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/2ef56851589c7587719562503aad87ef02677f0b"}]}],"author":{"name":"Carlos Camacho","email":"ccamacho@redhat.com","date":"2019-07-03 15:25:54.000000000","tz":120},"committer":{"name":"Carlos Camacho","email":"ccamacho@redhat.com","date":"2019-07-03 15:29:25.000000000","tz":120},"subject":"Add playbook to stop containers","message":"Add playbook to stop containers\n\nWe need to stop several containers on the upgrades workflow.\n\nThis submission allows to reduce the duplicated code in THT.\nIt also handles the case in which we have podman as the CLI\nbut the containers running with docker.\n\nChange-Id: I3c0580893aed5bda46d9104ffcee0f00cbabd9ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/facf82cafbb61711aac8e28ef064457f19937ef6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/facf82cafbb61711aac8e28ef064457f19937ef6"}]},"branch":"refs/heads/master"},"2d3ab49ffb0b8315815cfe4177668691d8748dc9":{"kind":"REWORK","_number":3,"created":"2019-07-04 07:28:08.000000000","uploader":{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},"ref":"refs/changes/38/668938/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/38/668938/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/38/668938/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/38/668938/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/38/668938/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/38/668938/3"}}},"commit":{"parents":[{"commit":"2ef56851589c7587719562503aad87ef02677f0b","subject":"Merge \"Handle empty NetworkConfig\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/2ef56851589c7587719562503aad87ef02677f0b"}]}],"author":{"name":"Carlos Camacho","email":"ccamacho@redhat.com","date":"2019-07-03 15:25:54.000000000","tz":120},"committer":{"name":"Carlos Camacho","email":"ccamacho@redhat.com","date":"2019-07-04 07:27:56.000000000","tz":120},"subject":"Add playbook to stop containers","message":"Add playbook to stop containers\n\nWe need to stop several containers on the upgrades workflow.\n\nThis submission allows to reduce the duplicated code in THT.\nIt also handles the case in which we have podman as the CLI\nbut the containers running with docker.\n\nChange-Id: I3c0580893aed5bda46d9104ffcee0f00cbabd9ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/2d3ab49ffb0b8315815cfe4177668691d8748dc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/2d3ab49ffb0b8315815cfe4177668691d8748dc9"}]},"branch":"refs/heads/master"},"7486b7249802d470f2f47a5f310028cd1ac31a87":{"kind":"REWORK","_number":4,"created":"2019-07-04 11:06:38.000000000","uploader":{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},"ref":"refs/changes/38/668938/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/38/668938/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/38/668938/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/38/668938/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/38/668938/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/38/668938/4"}}},"commit":{"parents":[{"commit":"2ef56851589c7587719562503aad87ef02677f0b","subject":"Merge \"Handle empty NetworkConfig\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/2ef56851589c7587719562503aad87ef02677f0b"}]}],"author":{"name":"Carlos Camacho","email":"ccamacho@redhat.com","date":"2019-07-03 15:25:54.000000000","tz":120},"committer":{"name":"Carlos Camacho","email":"ccamacho@redhat.com","date":"2019-07-04 11:06:26.000000000","tz":120},"subject":"Add playbook to stop containers","message":"Add playbook to stop containers\n\nWe need to stop several containers on the upgrades workflow.\n\nThis submission allows to reduce the duplicated code in THT.\nIt also handles the case in which we have podman as the CLI\nbut the containers running with docker.\n\nChange-Id: I3c0580893aed5bda46d9104ffcee0f00cbabd9ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/7486b7249802d470f2f47a5f310028cd1ac31a87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/7486b7249802d470f2f47a5f310028cd1ac31a87"}]},"branch":"refs/heads/master"},"6e3eda7f66c59f07bb2bb12bb3bd567a15f3eeaa":{"kind":"REWORK","_number":5,"created":"2019-07-04 13:31:04.000000000","uploader":{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},"ref":"refs/changes/38/668938/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/38/668938/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/38/668938/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/38/668938/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/38/668938/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/38/668938/5"}}},"commit":{"parents":[{"commit":"2ef56851589c7587719562503aad87ef02677f0b","subject":"Merge \"Handle empty NetworkConfig\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/2ef56851589c7587719562503aad87ef02677f0b"}]}],"author":{"name":"Carlos Camacho","email":"ccamacho@redhat.com","date":"2019-07-03 15:25:54.000000000","tz":120},"committer":{"name":"Carlos Camacho","email":"ccamacho@redhat.com","date":"2019-07-04 13:30:43.000000000","tz":120},"subject":"Add playbook to stop containers","message":"Add playbook to stop containers\n\nWe need to stop several containers on the upgrades workflow.\n\nThis submission allows to reduce the duplicated code in THT.\nIt also handles the case in which we have podman as the CLI\nbut the containers running with docker.\n\nChange-Id: I3c0580893aed5bda46d9104ffcee0f00cbabd9ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/6e3eda7f66c59f07bb2bb12bb3bd567a15f3eeaa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/6e3eda7f66c59f07bb2bb12bb3bd567a15f3eeaa"}]},"branch":"refs/heads/master"},"8be70c7adb2eaba960ee222886382f06746761f5":{"kind":"REWORK","_number":6,"created":"2019-07-05 11:58:09.000000000","uploader":{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},"ref":"refs/changes/38/668938/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/38/668938/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/38/668938/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/38/668938/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/38/668938/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/38/668938/6"}}},"commit":{"parents":[{"commit":"e1061bbecb71f80fa29dacc36678d61f781a4f5b","subject":"Merge \"Linting hardening via pre-commit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/e1061bbecb71f80fa29dacc36678d61f781a4f5b"}]}],"author":{"name":"Carlos Camacho","email":"ccamacho@redhat.com","date":"2019-07-03 15:25:54.000000000","tz":120},"committer":{"name":"Carlos Camacho","email":"ccamacho@redhat.com","date":"2019-07-05 07:56:14.000000000","tz":120},"subject":"Add playbook to stop containers","message":"Add playbook to stop containers\n\nWe need to stop several containers on the upgrades workflow.\n\nThis submission allows to reduce the duplicated code in THT.\nIt also handles the case in which we have podman as the CLI\nbut the containers running with docker.\n\nChange-Id: I3c0580893aed5bda46d9104ffcee0f00cbabd9ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/8be70c7adb2eaba960ee222886382f06746761f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/8be70c7adb2eaba960ee222886382f06746761f5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
