)]}'
{"id":"openstack%2Ftripleo-common~669197","triplet_id":"openstack%2Ftripleo-common~master~I242f7bb718b08724da21cafa86d3e8853d0ff453","project":"openstack/tripleo-common","branch":"master","topic":"bug/1835171","hashtags":[],"change_id":"I242f7bb718b08724da21cafa86d3e8853d0ff453","subject":"Use internal endpoints for configuration in octavia nodes","status":"ABANDONED","created":"2019-07-04 14:38:06.000000000","updated":"2019-07-15 11:58:26.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7755334f885727b0e949430b0accb00965122010","_number":669197,"virtual_id_number":669197,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"rejected":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":-2,"date":"2019-07-10 15:45:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-07-05 13:57:53.000000000","_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2019-07-08 10:51:27.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"recommended":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":1,"date":"2019-07-08 07:41:57.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":-1,"date":"2019-07-11 13:45:44.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":2,"date":"2019-07-10 12:24:08.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":2,"date":"2019-07-08 20:32:12.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2019-07-09 08:03:44.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-08 07:41:57.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-07-08 10:51:27.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-08 20:32:12.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-07-09 08:03:44.000000000","updated_by":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"reviewer":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"state":"REVIEWER"},{"updated":"2019-07-10 12:24:08.000000000","updated_by":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"reviewer":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"state":"REVIEWER"},{"updated":"2019-07-10 15:45:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-11 13:45:44.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"}],"messages":[{"id":"de1c7c2331c96abc8df0084917e9632f7fe997c5","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-07-04 14:38:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"14b3f7511466054c45dc4bdb13943af6d1ed1575","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-04 16:52:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/97/669197/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/f7119e3/ : RETRY_LIMIT in 4m 40s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 : NODE_FAILURE in 0s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/97/669197/1/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/ec5f462/ : SUCCESS in 2h 04m 55s","accounts_in_message":[],"_revision_number":1},{"id":"86d6c38c238eee872cc0e438a2042ccc52dc2734","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 19:30:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/97/669197/1/check/openstack-tox-cover/0f2f8d0/cover/ : SUCCESS in 3m 46s\n- openstack-tox-pep8 http://logs.openstack.org/97/669197/1/check/openstack-tox-pep8/def5182/ : SUCCESS in 3m 02s\n- openstack-tox-py27 http://logs.openstack.org/97/669197/1/check/openstack-tox-py27/bbdbc90/ : SUCCESS in 3m 16s\n- openstack-tox-py36 http://logs.openstack.org/97/669197/1/check/openstack-tox-py36/6660b92/ : SUCCESS in 4m 54s\n- openstack-tox-py37 http://logs.openstack.org/97/669197/1/check/openstack-tox-py37/43093d7/ : SUCCESS in 4m 14s\n- tox-molecule http://logs.openstack.org/97/669197/1/check/tox-molecule/6561fd8/tox/reports.html : SUCCESS in 18m 27s\n- openstack-tox-docs http://logs.openstack.org/97/669197/1/check/openstack-tox-docs/25956c7/html/ : SUCCESS in 4m 05s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/97/669197/1/check/tripleo-ci-centos-7-containers-multinode/45455ee/ : FAILURE in 25m 18s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/97/669197/1/check/tripleo-ci-centos-7-undercloud-containers/0b98aef/ : SUCCESS in 1h 41m 51s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/97/669197/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/ca9f776/ : FAILURE in 33m 08s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/97/669197/1/check/tripleo-ci-centos-7-standalone/19ac623/ : SUCCESS in 2h 07m 39s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/97/669197/1/check/tripleo-ci-fedora-28-standalone/3f52aab/ : FAILURE in 34m 20s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/97/669197/1/check/tripleo-ci-centos-7-standalone-upgrade/071d8e4/ : SUCCESS in 2h 21m 25s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/97/669197/1/check/tripleo-ci-centos-7-standalone-upgrade-stein/208bcbf/ : SUCCESS in 2h 24m 07s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/97/669197/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/5733597/ : FAILURE in 2h 06m 13s (non-voting)\n- tripleo-ci-centos-7-scenario010-standalone http://logs.openstack.org/97/669197/1/check/tripleo-ci-centos-7-scenario010-standalone/52626aa/ : SUCCESS in 1h 23m 56s\n- openstack-tox-linters http://logs.openstack.org/97/669197/1/check/openstack-tox-linters/fb79528/ : SUCCESS in 4m 18s\n- openstack-tox-py35 http://logs.openstack.org/97/669197/1/check/openstack-tox-py35/94f0016/ : SUCCESS in 4m 13s\n- openstack-tox-lower-constraints http://logs.openstack.org/97/669197/1/check/openstack-tox-lower-constraints/7ab829c/ : SUCCESS in 5m 16s","accounts_in_message":[],"_revision_number":1},{"id":"44e9e843348ad83f6852045484025e94c38fac9b","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-07-05 13:57:53.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"52abd42c8a4f16cd163084258afe48e0fc6d9006","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-05 16:46:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/97/669197/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/fe2b237/ : FAILURE in 1h 20m 44s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/97/669197/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/ad6e3d7/ : SUCCESS in 2h 46m 11s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/97/669197/1/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/f3e6bc4/ : SUCCESS in 1h 23m 12s","accounts_in_message":[],"_revision_number":1},{"id":"2124dbda7a55e2647fe6eaf8c228809360ec22f9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 17:08:28.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/97/669197/1/check/openstack-tox-cover/721d57b/cover/ : SUCCESS in 3m 51s\n- openstack-tox-pep8 http://logs.openstack.org/97/669197/1/check/openstack-tox-pep8/cae95ba/ : SUCCESS in 3m 06s\n- openstack-tox-py27 http://logs.openstack.org/97/669197/1/check/openstack-tox-py27/801ae55/ : SUCCESS in 3m 56s\n- openstack-tox-py36 http://logs.openstack.org/97/669197/1/check/openstack-tox-py36/c276fd6/ : SUCCESS in 3m 01s\n- openstack-tox-py37 http://logs.openstack.org/97/669197/1/check/openstack-tox-py37/65c7340/ : SUCCESS in 3m 22s\n- tox-molecule http://logs.openstack.org/97/669197/1/check/tox-molecule/00582dc/tox/reports.html : SUCCESS in 16m 40s\n- openstack-tox-docs http://logs.openstack.org/97/669197/1/check/openstack-tox-docs/07b772f/html/ : SUCCESS in 3m 31s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/97/669197/1/check/tripleo-ci-centos-7-containers-multinode/bec0eb5/ : SUCCESS in 2h 14m 11s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/97/669197/1/check/tripleo-ci-centos-7-undercloud-containers/7340eba/ : SUCCESS in 1h 45m 43s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/97/669197/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/f500e8b/ : SUCCESS in 2h 36m 06s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/97/669197/1/check/tripleo-ci-centos-7-standalone/b8cf743/ : SUCCESS in 2h 00m 26s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/97/669197/1/check/tripleo-ci-fedora-28-standalone/b63c87e/ : SUCCESS in 2h 00m 57s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/97/669197/1/check/tripleo-ci-centos-7-standalone-upgrade/9a98d53/ : SUCCESS in 2h 30m 08s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/97/669197/1/check/tripleo-ci-centos-7-standalone-upgrade-stein/277c20e/ : FAILURE in 1h 30m 40s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/97/669197/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/9b70003/ : FAILURE in 2h 30m 27s (non-voting)\n- tripleo-ci-centos-7-scenario010-standalone http://logs.openstack.org/97/669197/1/check/tripleo-ci-centos-7-scenario010-standalone/da37ca2/ : SUCCESS in 1h 22m 28s\n- openstack-tox-linters http://logs.openstack.org/97/669197/1/check/openstack-tox-linters/08a23b9/ : SUCCESS in 5m 23s\n- openstack-tox-py35 http://logs.openstack.org/97/669197/1/check/openstack-tox-py35/9ec7075/ : SUCCESS in 4m 43s\n- openstack-tox-lower-constraints http://logs.openstack.org/97/669197/1/check/openstack-tox-lower-constraints/aba8f0e/ : SUCCESS in 6m 21s","accounts_in_message":[],"_revision_number":1},{"id":"1dd7488c27fd55a6ed441e5f16d6f85bc370ad05","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-07-08 07:27:07.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"97e5881be1eaed46304d937db40f26f3c363a1c0","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-07-08 07:41:57.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c027fd362dcd38737353d8bbca0a59e8ebdbd52a","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-07-08 07:42:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9254430542ba59dbf36eb7c23cd81dca251e4a50","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-08 10:51:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/97/669197/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/d315cc3/ : SUCCESS in 3h 03m 12s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/97/669197/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/1851971/ : SUCCESS in 3h 03m 29s","accounts_in_message":[],"_revision_number":1},{"id":"6bba588f2c4e3246ada8a8728ef229dc063a9011","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 12:14:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/97/669197/1/check/openstack-tox-cover/0f173b3/cover/ : SUCCESS in 3m 33s\n- openstack-tox-pep8 http://logs.openstack.org/97/669197/1/check/openstack-tox-pep8/d39d565/ : SUCCESS in 3m 45s\n- openstack-tox-py27 http://logs.openstack.org/97/669197/1/check/openstack-tox-py27/7ce6452/ : SUCCESS in 3m 15s\n- openstack-tox-py36 http://logs.openstack.org/97/669197/1/check/openstack-tox-py36/b1306c6/ : SUCCESS in 3m 28s\n- openstack-tox-py37 http://logs.openstack.org/97/669197/1/check/openstack-tox-py37/e3a03c8/ : SUCCESS in 4m 00s\n- tox-molecule http://logs.openstack.org/97/669197/1/check/tox-molecule/a177ce8/tox/reports.html : SUCCESS in 19m 40s\n- openstack-tox-docs http://logs.openstack.org/97/669197/1/check/openstack-tox-docs/8c093c3/html/ : SUCCESS in 3m 11s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/97/669197/1/check/tripleo-ci-centos-7-containers-multinode/02bf311/ : SUCCESS in 2h 08m 11s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/97/669197/1/check/tripleo-ci-centos-7-undercloud-containers/685d242/ : SUCCESS in 1h 31m 55s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/97/669197/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/b553fbb/ : FAILURE in 2h 30m 11s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/97/669197/1/check/tripleo-ci-centos-7-standalone/080225b/ : SUCCESS in 2h 05m 20s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/97/669197/1/check/tripleo-ci-fedora-28-standalone/13f9cd5/ : SUCCESS in 2h 04m 29s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/97/669197/1/check/tripleo-ci-centos-7-standalone-upgrade/1130e4e/ : SUCCESS in 2h 28m 10s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/97/669197/1/check/tripleo-ci-centos-7-standalone-upgrade-stein/d1f9e64/ : SUCCESS in 2h 31m 01s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/97/669197/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/ee6bb02/ : FAILURE in 2h 28m 58s (non-voting)\n- tripleo-ci-centos-7-scenario010-standalone http://logs.openstack.org/97/669197/1/check/tripleo-ci-centos-7-scenario010-standalone/9c55e4f/ : SUCCESS in 1h 37m 33s\n- openstack-tox-linters http://logs.openstack.org/97/669197/1/check/openstack-tox-linters/c74fc60/ : SUCCESS in 4m 56s\n- openstack-tox-py35 http://logs.openstack.org/97/669197/1/check/openstack-tox-py35/ad65dc8/ : SUCCESS in 4m 07s\n- openstack-tox-lower-constraints http://logs.openstack.org/97/669197/1/check/openstack-tox-lower-constraints/b8e9261/ : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":1},{"id":"4fb3a6b2b36864fe08368ca64f7de9844f65f182","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-08 20:32:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"31560ff5ccd8ca3fe694314f15df62811e3b647f","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2019-07-09 08:03:44.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"abc3ebc8133e7e197ac077f0ff44e2de9719c558","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2019-07-10 12:24:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0d52936d48874dae1a15486dd3cc0c865d056671","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2019-07-10 13:21:19.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ce214de79283727df813147590b592eef26e003e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 13:21:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"841f7d00de283aa0fe5aa6844535a52518aae585","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2019-07-10 13:36:08.000000000","message":"Patch Set 1: -Workflow\n\npending CI tempest coverage","accounts_in_message":[],"_revision_number":1},{"id":"6228bac28775c3b39a5eaa6d1a7016ea7b61cff2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 15:45:18.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"50409d3b9aa9a698592b4337e5097bf322473b8e","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-07-11 13:45:44.000000000","message":"Patch Set 1: Code-Review-1\n\nplease add these changes to tripleo-ansible, these playbooks have already been ported over and are being removed from here.","accounts_in_message":[],"_revision_number":1},{"id":"074b2cd76732d78b7ea35c7d771f1f31e84c2989","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-07-15 11:58:26.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cf7f2ba9ea73d1fe30dbeebb3d8e8a1f515b859a","revisions":{"cf7f2ba9ea73d1fe30dbeebb3d8e8a1f515b859a":{"kind":"REWORK","_number":1,"created":"2019-07-04 14:38:06.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/97/669197/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/97/669197/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/97/669197/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/97/669197/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/97/669197/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/97/669197/1"}}},"commit":{"parents":[{"commit":"e1061bbecb71f80fa29dacc36678d61f781a4f5b","subject":"Merge \"Linting hardening via pre-commit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/e1061bbecb71f80fa29dacc36678d61f781a4f5b"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-07-04 14:11:19.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-07-04 14:36:49.000000000","tz":120},"subject":"Use internal endpoints for configuration in octavia nodes","message":"Use internal endpoints for configuration in octavia nodes\n\nWhen running configuration steps for octavia in octavia nodes, use\ninternal endpoints since public endpoints may not be available due to\nnetwork topology (spine-leaf network + octavia in networkers).\n\nDepends-On: https://review.opendev.org/#/c/669195/\n\nCloses-Bug: #1835171\nChange-Id: I242f7bb718b08724da21cafa86d3e8853d0ff453\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/cf7f2ba9ea73d1fe30dbeebb3d8e8a1f515b859a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/cf7f2ba9ea73d1fe30dbeebb3d8e8a1f515b859a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
