)]}'
{"id":"openstack%2Ftripleo-common~670152","triplet_id":"openstack%2Ftripleo-common~master~I6bbca8eb256f387c4f74378baa5253890cfe1abf","project":"openstack/tripleo-common","branch":"master","topic":"bp/reduce-deployment-resources","hashtags":[],"change_id":"I6bbca8eb256f387c4f74378baa5253890cfe1abf","subject":"Also write NetworkConfig to role dir","status":"MERGED","created":"2019-07-10 18:30:19.000000000","updated":"2019-07-15 19:03:36.000000000","submitted":"2019-07-15 19:03:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670152-1563217416500-31ef7618","meta_rev_id":"62d15fa84a22305cbe05148bd15cedf0476c616a","_number":670152,"virtual_id_number":670152,"owner":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-07-15 19:03:36.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-11 16:19:10.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-07-15 14:33:19.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-07-15 14:38:44.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2019-07-15 14:38:44.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-11 16:19:10.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-15 14:33:19.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-07-15 14:38:44.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-07-15 19:03:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"79d32216658920dcb22d95499a07e9a5aec1cd08","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2019-07-10 18:30:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"feb50e59cf235e3bcd8820a829b099a7699eb46f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-10 18:58:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/52/670152/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/f9ae4ed/ : FAILURE in 25m 47s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/52/670152/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/1032582/ : FAILURE in 25m 55s","accounts_in_message":[],"_revision_number":1},{"id":"db0488ea4f8e2c6ea0b95c9ef7088a2242b06ad0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 23:02:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/52/670152/1/check/openstack-tox-cover/a5f3207/cover/ : SUCCESS in 3m 48s\n- openstack-tox-pep8 http://logs.openstack.org/52/670152/1/check/openstack-tox-pep8/6742ac9/ : SUCCESS in 2m 54s\n- openstack-tox-py27 http://logs.openstack.org/52/670152/1/check/openstack-tox-py27/f5a6338/ : SUCCESS in 3m 35s\n- openstack-tox-py36 http://logs.openstack.org/52/670152/1/check/openstack-tox-py36/2d0ce5a/ : SUCCESS in 3m 21s\n- openstack-tox-py37 http://logs.openstack.org/52/670152/1/check/openstack-tox-py37/a241d19/ : SUCCESS in 3m 59s\n- openstack-tox-docs http://logs.openstack.org/52/670152/1/check/openstack-tox-docs/7e56769/html/ : SUCCESS in 3m 21s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/52/670152/1/check/tripleo-ci-centos-7-containers-multinode/7d720f5/ : FAILURE in 30m 41s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/52/670152/1/check/tripleo-ci-centos-7-undercloud-containers/25206a1/ : FAILURE in 33m 41s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/52/670152/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/051bb32/ : FAILURE in 37m 37s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/52/670152/1/check/tripleo-ci-centos-7-standalone/4dfc82c/ : FAILURE in 27m 00s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/52/670152/1/check/tripleo-ci-fedora-28-standalone/012b8f6/ : FAILURE in 27m 05s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/52/670152/1/check/tripleo-ci-centos-7-standalone-upgrade/ce42ec6/ : FAILURE in 1h 22m 47s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/52/670152/1/check/tripleo-ci-centos-7-standalone-upgrade-stein/7f30a7b/ : SUCCESS in 2h 06m 49s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/52/670152/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/941897c/ : FAILURE in 1h 39m 21s (non-voting)\n- openstack-tox-linters http://logs.openstack.org/52/670152/1/check/openstack-tox-linters/134d148/ : SUCCESS in 3m 39s\n- openstack-tox-py35 http://logs.openstack.org/52/670152/1/check/openstack-tox-py35/f41288f/ : SUCCESS in 3m 32s\n- openstack-tox-lower-constraints http://logs.openstack.org/52/670152/1/check/openstack-tox-lower-constraints/7f53080/ : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":1},{"id":"e5b15903619df9cb30d8837fe21fb259d57d2bdb","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2019-07-11 13:22:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d30fb15fa52222f012d59c8b08f522459b7195aa","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-11 16:19:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/52/670152/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/0d9e210/ : SUCCESS in 2h 54m 07s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/52/670152/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/f13a4a9/ : SUCCESS in 2h 43m 41s","accounts_in_message":[],"_revision_number":2},{"id":"56a4cdc5a7212d8a939cdf192d41bb1e9753e152","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 20:25:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/52/670152/2/check/openstack-tox-cover/f38bddb/cover/ : SUCCESS in 4m 58s\n- openstack-tox-pep8 http://logs.openstack.org/52/670152/2/check/openstack-tox-pep8/4260705/ : SUCCESS in 3m 20s\n- openstack-tox-py27 http://logs.openstack.org/52/670152/2/check/openstack-tox-py27/af9564d/ : SUCCESS in 5m 52s\n- openstack-tox-py36 http://logs.openstack.org/52/670152/2/check/openstack-tox-py36/bce6f67/ : SUCCESS in 3m 56s\n- openstack-tox-py37 http://logs.openstack.org/52/670152/2/check/openstack-tox-py37/c9f4a36/ : SUCCESS in 4m 19s\n- openstack-tox-docs http://logs.openstack.org/52/670152/2/check/openstack-tox-docs/ceed537/html/ : SUCCESS in 3m 14s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/52/670152/2/check/tripleo-ci-centos-7-containers-multinode/bb29e2c/ : SUCCESS in 2h 21m 06s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/52/670152/2/check/tripleo-ci-centos-7-undercloud-containers/665df3c/ : SUCCESS in 1h 43m 49s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/52/670152/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/99d1a17/ : FAILURE in 2h 44m 44s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/52/670152/2/check/tripleo-ci-centos-7-standalone/f011df3/ : SUCCESS in 2h 07m 44s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/52/670152/2/check/tripleo-ci-fedora-28-standalone/437dfe2/ : SUCCESS in 2h 07m 00s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/52/670152/2/check/tripleo-ci-centos-7-standalone-upgrade/99ed40c/ : FAILURE in 2h 04m 23s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/52/670152/2/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/225ec8f/ : FAILURE in 2h 27m 51s (non-voting)\n- openstack-tox-linters http://logs.openstack.org/52/670152/2/check/openstack-tox-linters/d9cfc1b/ : SUCCESS in 5m 06s\n- openstack-tox-py35 http://logs.openstack.org/52/670152/2/check/openstack-tox-py35/a794494/ : SUCCESS in 4m 48s\n- openstack-tox-lower-constraints http://logs.openstack.org/52/670152/2/check/openstack-tox-lower-constraints/0a398c6/ : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":2},{"id":"7e46c4e9744c9581a6e6a094e8810d71b1a8d486","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-07-15 14:33:19.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"cb1f5417952993f664572ed2827576148f781743","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-15 14:38:44.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"332815b8feb30a3bc0c40a0759acc4fea3b29a20","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 14:39:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b6b18e6ed75aa32d2cc673650776a781eeb250db","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 19:03:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"421ec1289993aca13790a6bb14d46072598ce810","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 19:03:36.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/52/670152/2/gate/openstack-tox-pep8/4bdf34a/ : SUCCESS in 5m 00s\n- openstack-tox-py27 http://logs.openstack.org/52/670152/2/gate/openstack-tox-py27/6b0d621/ : SUCCESS in 5m 03s\n- openstack-tox-py36 http://logs.openstack.org/52/670152/2/gate/openstack-tox-py36/17b29de/ : SUCCESS in 4m 38s\n- openstack-tox-py37 http://logs.openstack.org/52/670152/2/gate/openstack-tox-py37/351fe98/ : SUCCESS in 5m 44s\n- openstack-tox-docs http://logs.openstack.org/52/670152/2/gate/openstack-tox-docs/98d4a17/html/ : SUCCESS in 4m 46s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/52/670152/2/gate/tripleo-ci-centos-7-containers-multinode/b471cbd/ : SUCCESS in 2h 32m 36s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/52/670152/2/gate/tripleo-ci-centos-7-standalone/deb53fc/ : SUCCESS in 2h 02m 44s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/52/670152/2/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/d6f2e06/ : SUCCESS in 2h 38m 52s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/52/670152/2/gate/tripleo-ci-centos-7-undercloud-containers/25e545f/ : SUCCESS in 1h 49m 17s\n- openstack-tox-linters http://logs.openstack.org/52/670152/2/gate/openstack-tox-linters/5605410/ : SUCCESS in 4m 15s\n- openstack-tox-lower-constraints http://logs.openstack.org/52/670152/2/gate/openstack-tox-lower-constraints/c076196/ : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8ab13dd58cb77af9fd991744f10f8984ec456761","revisions":{"99620b64c72e5310c718d888a709bd3f7b7fe662":{"kind":"REWORK","_number":1,"created":"2019-07-10 18:30:19.000000000","uploader":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"ref":"refs/changes/52/670152/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/52/670152/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/52/670152/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/52/670152/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/52/670152/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/52/670152/1"}}},"commit":{"parents":[{"commit":"05460a665781e1094ebd18ef0614d9e8aa24f22e","subject":"Merge \"Remove roles that have been imported into tripleo-ansible\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/05460a665781e1094ebd18ef0614d9e8aa24f22e"}]}],"author":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2019-07-10 18:22:38.000000000","tz":-240},"committer":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2019-07-10 18:30:11.000000000","tz":-240},"subject":"Also write NetworkConfig to role dir","message":"Also write NetworkConfig to role dir\n\nThe NetworkConfig script is also written to the role directory within\nthe config-download directory, not just the individual server directory.\nAs we move towards making the NetworkConfig script generic for a role,\nit can be consumed from the role dir by any node.\n\nIt doesn\u0027t matter at this point which node\u0027s NetworkConfig script we\nwrite to the role dir, so it\u0027s just written whenever a node\u0027s script is\nwritten. Once we\u0027ve moved towards completely role generic Network\nconfiguration, we can remove writing it to the individual server dir as\nwell.\n\nChange-Id: I6bbca8eb256f387c4f74378baa5253890cfe1abf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/99620b64c72e5310c718d888a709bd3f7b7fe662"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/99620b64c72e5310c718d888a709bd3f7b7fe662"}]},"branch":"refs/heads/master"},"8ab13dd58cb77af9fd991744f10f8984ec456761":{"kind":"REWORK","_number":2,"created":"2019-07-11 13:22:47.000000000","uploader":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"ref":"refs/changes/52/670152/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/52/670152/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/52/670152/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/52/670152/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/52/670152/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/52/670152/2"}}},"commit":{"parents":[{"commit":"05460a665781e1094ebd18ef0614d9e8aa24f22e","subject":"Merge \"Remove roles that have been imported into tripleo-ansible\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/05460a665781e1094ebd18ef0614d9e8aa24f22e"}]}],"author":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2019-07-10 18:22:38.000000000","tz":-240},"committer":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2019-07-11 13:22:20.000000000","tz":-240},"subject":"Also write NetworkConfig to role dir","message":"Also write NetworkConfig to role dir\n\nThe NetworkConfig script is also written to the role directory within\nthe config-download directory, not just the individual server directory.\nAs we move towards making the NetworkConfig script generic for a role,\nit can be consumed from the role dir by any node.\n\nIt doesn\u0027t matter at this point which node\u0027s NetworkConfig script we\nwrite to the role dir, so it\u0027s just written whenever a node\u0027s script is\nwritten. Once we\u0027ve moved towards completely role generic Network\nconfiguration, we can remove writing it to the individual server dir as\nwell.\n\nChange-Id: I6bbca8eb256f387c4f74378baa5253890cfe1abf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/8ab13dd58cb77af9fd991744f10f8984ec456761"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/8ab13dd58cb77af9fd991744f10f8984ec456761"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
