)]}'
{"id":"openstack%2Ftripleo-common~670409","triplet_id":"openstack%2Ftripleo-common~master~I1addfab5e41eb48a08c048e1801bb654d20cf8a9","project":"openstack/tripleo-common","branch":"master","topic":"el8-ceilometermiddleware","hashtags":[],"change_id":"I1addfab5e41eb48a08c048e1801bb654d20cf8a9","subject":"Correct package name for ceilometermiddleware for swift","status":"MERGED","created":"2019-07-11 22:58:14.000000000","updated":"2019-07-27 21:46:34.000000000","submitted":"2019-07-27 21:46:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670409-1564263994420-94390bb6","meta_rev_id":"78cf7c5fabe35ad67c3bae9a68c8b4d5c9ac049c","_number":670409,"virtual_id_number":670409,"owner":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"date":"2019-07-23 17:52:05.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-07-27 21:46:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-07-15 22:50:52.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-07-15 13:14:20.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2019-07-15 13:14:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-15 13:14:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-07-27 18:10:29.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-11 22:58:52.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2019-07-12 07:17:48.000000000","updated_by":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"reviewer":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"state":"REVIEWER"},{"updated":"2019-07-15 22:50:52.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-27 18:10:29.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-07-27 21:46:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bfc2087087934d7d282ade31d0352abc14a0f7bb","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-11 22:58:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"651bbc44b676e6b1123024a215ce02a7647e2143","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-07-11 22:58:52.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"88d137738377d69838e27ed28ec9b26597e127f8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 02:23:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/09/670409/1/check/openstack-tox-cover/567512b/cover/ : SUCCESS in 3m 58s\n- openstack-tox-pep8 http://logs.openstack.org/09/670409/1/check/openstack-tox-pep8/c305fa9/ : SUCCESS in 2m 57s\n- openstack-tox-py27 http://logs.openstack.org/09/670409/1/check/openstack-tox-py27/88709de/ : SUCCESS in 5m 00s\n- openstack-tox-py36 http://logs.openstack.org/09/670409/1/check/openstack-tox-py36/77003a8/ : SUCCESS in 3m 28s\n- openstack-tox-py37 http://logs.openstack.org/09/670409/1/check/openstack-tox-py37/dd7c9fa/ : SUCCESS in 4m 50s\n- openstack-tox-docs http://logs.openstack.org/09/670409/1/check/openstack-tox-docs/e4460b7/html/ : SUCCESS in 3m 21s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/09/670409/1/check/tripleo-ci-centos-7-containers-multinode/0207778/ : SUCCESS in 2h 08m 01s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/09/670409/1/check/tripleo-ci-centos-7-undercloud-containers/4b05add/ : SUCCESS in 1h 34m 11s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/09/670409/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/3c60e7b/ : SUCCESS in 2h 38m 07s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/09/670409/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/d41e71c/ : FAILURE in 2h 34m 43s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/09/670409/1/check/tripleo-ci-centos-7-standalone/9e40703/ : SUCCESS in 2h 01m 20s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/09/670409/1/check/tripleo-ci-fedora-28-standalone/7deb8a3/ : SUCCESS in 1h 40m 38s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/09/670409/1/check/tripleo-ci-centos-7-standalone-upgrade/a8ee7bc/ : FAILURE in 1h 48m 22s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/09/670409/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/0246f17/ : FAILURE in 1h 58m 19s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/09/670409/1/check/tripleo-build-containers-centos-7/77f371b/ : SUCCESS in 50m 30s\n- tripleo-build-containers-centos-7-buildah http://logs.openstack.org/09/670409/1/check/tripleo-build-containers-centos-7-buildah/e6670d4/ : SUCCESS in 59m 31s\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/09/670409/1/check/tripleo-ci-centos-7-scenario001-standalone/75074f9/ : SUCCESS in 1h 29m 45s\n- tripleo-ci-centos-7-scenario002-standalone http://logs.openstack.org/09/670409/1/check/tripleo-ci-centos-7-scenario002-standalone/7aff35b/ : SUCCESS in 1h 45m 57s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/09/670409/1/check/tripleo-ci-centos-7-scenario003-standalone/159f107/ : SUCCESS in 1h 17m 08s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/09/670409/1/check/tripleo-ci-centos-7-scenario004-standalone/2ce2c41/ : SUCCESS in 1h 35m 14s\n- tripleo-ci-centos-7-scenario010-standalone http://logs.openstack.org/09/670409/1/check/tripleo-ci-centos-7-scenario010-standalone/60d4ac0/ : SUCCESS in 1h 15m 31s\n- tripleo-ci-centos-7-scenario012-standalone http://logs.openstack.org/09/670409/1/check/tripleo-ci-centos-7-scenario012-standalone/e64d5e1/ : FAILURE in 1h 25m 38s (non-voting)\n- openstack-tox-linters http://logs.openstack.org/09/670409/1/check/openstack-tox-linters/fce9e7b/ : SUCCESS in 3m 29s\n- openstack-tox-py35 http://logs.openstack.org/09/670409/1/check/openstack-tox-py35/560b1c2/ : SUCCESS in 3m 20s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/09/670409/1/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/a101dec/ : SUCCESS in 2h 08m 16s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/09/670409/1/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/6cffe53/ : SUCCESS in 2h 09m 56s\n- openstack-tox-lower-constraints http://logs.openstack.org/09/670409/1/check/openstack-tox-lower-constraints/454bb80/ : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":1},{"id":"d41288c2fe55a760a35372a89406b02534e6f28e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-12 02:39:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/09/670409/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/0dc59f9/ : TIMED_OUT in 3h 39m 17s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/09/670409/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/7ee6eb1/ : SUCCESS in 2h 44m 17s","accounts_in_message":[],"_revision_number":1},{"id":"4523013fcf1fa8d7b0fc190edd5514d5fca5884e","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-07-12 03:52:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"954b7329ddb632c785840762791cea39c609b317","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2019-07-12 07:17:48.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"fb37da6c942acaefc1dd669bf5f0ff1501416e69","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-07-13 02:21:17.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4035ffc64118c96965f5c10644b0664a5392f8c7","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-15 13:14:20.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"7df6b15a0e76f10b1065e50b5c3e519511a4766d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-15 16:04:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/09/670409/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/4a43a94/ : SUCCESS in 2h 35m 41s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/09/670409/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/1f655c4/ : SUCCESS in 2h 48m 13s","accounts_in_message":[],"_revision_number":2},{"id":"58e1be5d52fe0372321e53afa893fb3b52142133","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 17:17:02.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/09/670409/2/check/openstack-tox-cover/1bd6951/cover/ : SUCCESS in 4m 28s\n- openstack-tox-pep8 http://logs.openstack.org/09/670409/2/check/openstack-tox-pep8/d302644/ : SUCCESS in 3m 38s\n- openstack-tox-py27 http://logs.openstack.org/09/670409/2/check/openstack-tox-py27/6b73ca5/ : SUCCESS in 5m 13s\n- openstack-tox-py36 http://logs.openstack.org/09/670409/2/check/openstack-tox-py36/5cac5e7/ : SUCCESS in 4m 50s\n- openstack-tox-py37 http://logs.openstack.org/09/670409/2/check/openstack-tox-py37/aca6685/ : SUCCESS in 4m 09s\n- openstack-tox-docs http://logs.openstack.org/09/670409/2/check/openstack-tox-docs/80443bf/html/ : SUCCESS in 3m 42s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/09/670409/2/check/tripleo-ci-centos-7-containers-multinode/c727bf1/ : SUCCESS in 2h 07m 36s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/09/670409/2/check/tripleo-ci-centos-7-undercloud-containers/20a1bd2/ : SUCCESS in 1h 36m 31s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/09/670409/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/fb6ee73/ : SUCCESS in 2h 17m 36s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/09/670409/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/ef0ba6f/ : FAILURE in 2h 44m 56s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/09/670409/2/check/tripleo-ci-centos-7-standalone/db863f0/ : SUCCESS in 1h 54m 40s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/09/670409/2/check/tripleo-ci-fedora-28-standalone/b108464/ : SUCCESS in 1h 53m 45s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/09/670409/2/check/tripleo-ci-centos-7-standalone-upgrade/da68029/ : FAILURE in 1h 57m 07s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/09/670409/2/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/8b96717/ : FAILURE in 2h 05m 41s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/09/670409/2/check/tripleo-build-containers-centos-7/cc882de/ : SUCCESS in 48m 03s\n- tripleo-build-containers-centos-7-buildah http://logs.openstack.org/09/670409/2/check/tripleo-build-containers-centos-7-buildah/0a93ce2/ : SUCCESS in 59m 13s\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/09/670409/2/check/tripleo-ci-centos-7-scenario001-standalone/475525b/ : SUCCESS in 1h 44m 31s\n- tripleo-ci-centos-7-scenario002-standalone http://logs.openstack.org/09/670409/2/check/tripleo-ci-centos-7-scenario002-standalone/4ef27db/ : SUCCESS in 1h 35m 18s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/09/670409/2/check/tripleo-ci-centos-7-scenario003-standalone/95291fc/ : SUCCESS in 1h 10m 57s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/09/670409/2/check/tripleo-ci-centos-7-scenario004-standalone/45bdaa0/ : SUCCESS in 1h 58m 22s\n- tripleo-ci-centos-7-scenario010-standalone http://logs.openstack.org/09/670409/2/check/tripleo-ci-centos-7-scenario010-standalone/c0dd534/ : SUCCESS in 1h 19m 38s\n- tripleo-ci-centos-7-scenario012-standalone http://logs.openstack.org/09/670409/2/check/tripleo-ci-centos-7-scenario012-standalone/5e3f988/ : SUCCESS in 1h 48m 29s (non-voting)\n- openstack-tox-linters http://logs.openstack.org/09/670409/2/check/openstack-tox-linters/e2445fc/ : SUCCESS in 4m 07s\n- openstack-tox-py35 http://logs.openstack.org/09/670409/2/check/openstack-tox-py35/b0b7085/ : SUCCESS in 3m 13s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/09/670409/2/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/2a53901/ : SUCCESS in 2h 36m 11s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/09/670409/2/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/4196c1c/ : FAILURE in 2h 03m 19s\n- openstack-tox-lower-constraints http://logs.openstack.org/09/670409/2/check/openstack-tox-lower-constraints/bcba9c7/ : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":2},{"id":"96aa34660411f7507ebb15485fef0689cb1f17c1","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-15 17:37:23.000000000","message":"Patch Set 2:\n\nrecheck need to tweak the zuul conf because this file is only tested by \nthe container build jobs","accounts_in_message":[],"_revision_number":2},{"id":"fdcbcd1017a5e1bf33ea3718d5edb0e95776a25d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 21:55:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/09/670409/2/check/openstack-tox-cover/fb7ae06/cover/ : SUCCESS in 4m 30s\n- openstack-tox-pep8 http://logs.openstack.org/09/670409/2/check/openstack-tox-pep8/944c92e/ : SUCCESS in 3m 58s\n- openstack-tox-py27 http://logs.openstack.org/09/670409/2/check/openstack-tox-py27/49d36a3/ : SUCCESS in 4m 48s\n- openstack-tox-py36 http://logs.openstack.org/09/670409/2/check/openstack-tox-py36/348bcea/ : SUCCESS in 4m 09s\n- openstack-tox-py37 http://logs.openstack.org/09/670409/2/check/openstack-tox-py37/4e6f37e/ : SUCCESS in 4m 57s\n- openstack-tox-docs http://logs.openstack.org/09/670409/2/check/openstack-tox-docs/42a2307/html/ : SUCCESS in 4m 28s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/09/670409/2/check/tripleo-ci-centos-7-containers-multinode/490c205/ : SUCCESS in 2h 25m 19s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/09/670409/2/check/tripleo-ci-centos-7-undercloud-containers/84586da/ : SUCCESS in 1h 35m 45s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/09/670409/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/3a604ae/ : SUCCESS in 2h 29m 33s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/09/670409/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/1883114/ : FAILURE in 2h 32m 40s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/09/670409/2/check/tripleo-ci-centos-7-standalone/799039c/ : SUCCESS in 1h 55m 56s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/09/670409/2/check/tripleo-ci-fedora-28-standalone/b621ad4/ : SUCCESS in 1h 53m 09s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/09/670409/2/check/tripleo-ci-centos-7-standalone-upgrade/9043397/ : FAILURE in 2h 06m 28s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/09/670409/2/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/44b7e6d/ : FAILURE in 2h 11m 38s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/09/670409/2/check/tripleo-build-containers-centos-7/48eb9e5/ : SUCCESS in 52m 22s\n- tripleo-build-containers-centos-7-buildah http://logs.openstack.org/09/670409/2/check/tripleo-build-containers-centos-7-buildah/d9f4b30/ : SUCCESS in 52m 20s\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/09/670409/2/check/tripleo-ci-centos-7-scenario001-standalone/e80ead7/ : SUCCESS in 1h 35m 48s\n- tripleo-ci-centos-7-scenario002-standalone http://logs.openstack.org/09/670409/2/check/tripleo-ci-centos-7-scenario002-standalone/ac7c6c6/ : SUCCESS in 1h 42m 32s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/09/670409/2/check/tripleo-ci-centos-7-scenario003-standalone/90ad830/ : SUCCESS in 1h 24m 35s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/09/670409/2/check/tripleo-ci-centos-7-scenario004-standalone/33e8c9f/ : SUCCESS in 2h 07m 56s\n- tripleo-ci-centos-7-scenario010-standalone http://logs.openstack.org/09/670409/2/check/tripleo-ci-centos-7-scenario010-standalone/7d56b56/ : SUCCESS in 1h 25m 29s\n- tripleo-ci-centos-7-scenario012-standalone http://logs.openstack.org/09/670409/2/check/tripleo-ci-centos-7-scenario012-standalone/2e9cfae/ : SUCCESS in 2h 04m 56s (non-voting)\n- openstack-tox-linters http://logs.openstack.org/09/670409/2/check/openstack-tox-linters/92bc794/ : SUCCESS in 4m 03s\n- openstack-tox-py35 http://logs.openstack.org/09/670409/2/check/openstack-tox-py35/7161f54/ : SUCCESS in 3m 28s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/09/670409/2/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/f3c1d95/ : SUCCESS in 2h 23m 21s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/09/670409/2/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/4d44279/ : SUCCESS in 2h 19m 29s\n- openstack-tox-lower-constraints http://logs.openstack.org/09/670409/2/check/openstack-tox-lower-constraints/e463875/ : SUCCESS in 5m 21s","accounts_in_message":[],"_revision_number":2},{"id":"21b9886a0e0bbfb284a9095c93a5b9bda024baa4","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-15 22:50:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/09/670409/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/b72581d/ : SUCCESS in 2h 53m 21s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/09/670409/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/c9ac748/ : RETRY_LIMIT in 3m 49s","accounts_in_message":[],"_revision_number":2},{"id":"4a986912bacc06de714f3aed82e4c64060870685","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-23 17:52:05.000000000","message":"Patch Set 2:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":2},{"id":"22e1103d8fa6b255fad29a071db9cd939d548212","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-23 20:44:19.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/09/670409/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/02ec4f3/ : SUCCESS in 2h 50m 37s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/09/670409/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/3838b88/ : FAILURE in 1h 48m 49s","accounts_in_message":[],"_revision_number":2},{"id":"3c23606557da144e909f36f07fa92acb36a22152","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-07-27 18:10:29.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"623abb3a5c5e8c0a06fd3c1c4c44a5747caf8f2d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-27 18:10:46.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b0ff04223ef52fd7c2eb3fe0cd72647ba3bf41f7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-27 21:46:34.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/09/670409/2/gate/openstack-tox-pep8/9c7b0f5/ : SUCCESS in 2m 48s\n- openstack-tox-py27 http://logs.openstack.org/09/670409/2/gate/openstack-tox-py27/42e3b76/ : SUCCESS in 3m 43s\n- openstack-tox-py36 http://logs.openstack.org/09/670409/2/gate/openstack-tox-py36/07e6d68/ : SUCCESS in 3m 27s\n- openstack-tox-py37 http://logs.openstack.org/09/670409/2/gate/openstack-tox-py37/289677b/ : SUCCESS in 7m 00s\n- openstack-tox-docs http://logs.openstack.org/09/670409/2/gate/openstack-tox-docs/bfda39d/html/ : SUCCESS in 3m 16s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/09/670409/2/gate/tripleo-ci-centos-7-containers-multinode/2e410d2/ : SUCCESS in 2h 06m 49s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/09/670409/2/gate/tripleo-ci-centos-7-standalone/a01da4c/ : SUCCESS in 1h 56m 44s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/09/670409/2/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/3ceee9e/ : SUCCESS in 2h 36m 51s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/09/670409/2/gate/tripleo-ci-centos-7-undercloud-containers/9539c0c/ : SUCCESS in 1h 31m 30s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/09/670409/2/gate/tripleo-build-containers-centos-7/92493c3/ : SUCCESS in 38m 52s\n- tripleo-build-containers-centos-7-buildah http://logs.openstack.org/09/670409/2/gate/tripleo-build-containers-centos-7-buildah/1f90ac4/ : SUCCESS in 1h 02m 42s\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/09/670409/2/gate/tripleo-ci-centos-7-scenario001-standalone/a824874/ : SUCCESS in 1h 36m 12s\n- tripleo-ci-centos-7-scenario002-standalone http://logs.openstack.org/09/670409/2/gate/tripleo-ci-centos-7-scenario002-standalone/7c60f00/ : SUCCESS in 1h 47m 21s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/09/670409/2/gate/tripleo-ci-centos-7-scenario003-standalone/ad1d1c3/ : SUCCESS in 1h 11m 49s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/09/670409/2/gate/tripleo-ci-centos-7-scenario004-standalone/8ea9c35/ : SUCCESS in 2h 11m 44s\n- tripleo-ci-centos-7-scenario010-standalone http://logs.openstack.org/09/670409/2/gate/tripleo-ci-centos-7-scenario010-standalone/b7888ea/ : SUCCESS in 1h 10m 36s\n- openstack-tox-linters http://logs.openstack.org/09/670409/2/gate/openstack-tox-linters/729168d/ : SUCCESS in 3m 49s\n- openstack-tox-lower-constraints http://logs.openstack.org/09/670409/2/gate/openstack-tox-lower-constraints/e30a4a9/ : SUCCESS in 6m 16s","accounts_in_message":[],"_revision_number":2},{"id":"d542c0616a0e4ee398c73966986629876c660ae8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-27 21:46:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6a2c7848dd6493c1ce37ce0893218117a900059e","revisions":{"81def86c2ce01198b72ea29665b9ba9bca666d7f":{"kind":"REWORK","_number":1,"created":"2019-07-11 22:58:14.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/09/670409/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/09/670409/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/09/670409/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/09/670409/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/09/670409/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/09/670409/1"}}},"commit":{"parents":[{"commit":"a5cfe006ffaac0b1ac4840cc6c5613565da80c3a","subject":"Remove pika from container builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/a5cfe006ffaac0b1ac4840cc6c5613565da80c3a"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-07-11 22:57:20.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-07-11 22:57:20.000000000","tz":-360},"subject":"Correct package name for ceilometermiddleware for swift","message":"Correct package name for ceilometermiddleware for swift\n\nUnfortunately swift is still python2 based on el8. Add a workaround\nuntil it\u0027s cut over.\n\nChange-Id: I1addfab5e41eb48a08c048e1801bb654d20cf8a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/81def86c2ce01198b72ea29665b9ba9bca666d7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/81def86c2ce01198b72ea29665b9ba9bca666d7f"}]},"branch":"refs/heads/master"},"6a2c7848dd6493c1ce37ce0893218117a900059e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-07-15 13:14:20.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/09/670409/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/09/670409/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/09/670409/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/09/670409/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/09/670409/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/09/670409/2"}}},"commit":{"parents":[{"commit":"a97287c55c3de77e717e0a68028f464a30cfcecc","subject":"Merge \"Correct how we fetch External and InternalApi networks name\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/a97287c55c3de77e717e0a68028f464a30cfcecc"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-07-11 22:57:20.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-07-15 13:14:20.000000000","tz":0},"subject":"Correct package name for ceilometermiddleware for swift","message":"Correct package name for ceilometermiddleware for swift\n\nUnfortunately swift is still python2 based on el8. Add a workaround\nuntil it\u0027s cut over.\n\nChange-Id: I1addfab5e41eb48a08c048e1801bb654d20cf8a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/6a2c7848dd6493c1ce37ce0893218117a900059e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/6a2c7848dd6493c1ce37ce0893218117a900059e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
