)]}'
{"id":"openstack%2Ftripleo-common~670486","triplet_id":"openstack%2Ftripleo-common~master~I33effbf481682402d39436a9e97cbd3ee2044577","project":"openstack/tripleo-common","branch":"master","topic":"bp/reduce-deployment-resources","hashtags":[],"change_id":"I33effbf481682402d39436a9e97cbd3ee2044577","subject":"Create \u003cservice\u003e _cell_node_names if nova_additional_cell","status":"MERGED","created":"2019-07-12 10:09:57.000000000","updated":"2019-07-15 18:37:17.000000000","submitted":"2019-07-15 18:37:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670486-1563215837926-6d7173bf","meta_rev_id":"55a725bdffe3df93207c6ac03b3c73e616e3145d","_number":670486,"virtual_id_number":670486,"owner":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":23811,"name":"Oliver Walsh","email":"owalsh@redhat.com","username":"owalsh"},{"value":0,"_account_id":30133,"name":"Piotr Kopec","email":"pkopec17@gmail.com","username":"pkopec"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":2,"date":"2019-07-15 18:37:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-12 14:24:55.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-07-15 14:36:13.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":23811,"name":"Oliver Walsh","email":"owalsh@redhat.com","username":"owalsh"},{"value":0,"_account_id":30133,"name":"Piotr Kopec","email":"pkopec17@gmail.com","username":"pkopec"},{"value":2,"date":"2019-07-15 11:55:24.000000000","_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-07-15 14:36:13.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":23811,"name":"Oliver Walsh","email":"owalsh@redhat.com","username":"owalsh"},{"value":0,"date":"2019-07-15 07:12:09.000000000","_account_id":30133,"name":"Piotr Kopec","email":"pkopec17@gmail.com","username":"pkopec"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-15 07:12:14.000000000","_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":23811,"name":"Oliver Walsh","email":"owalsh@redhat.com","username":"owalsh"},{"_account_id":30133,"name":"Piotr Kopec","email":"pkopec17@gmail.com","username":"pkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-12 10:11:55.000000000","updated_by":{"_account_id":23811,"name":"Oliver Walsh","email":"owalsh@redhat.com","username":"owalsh"},"reviewer":{"_account_id":23811,"name":"Oliver Walsh","email":"owalsh@redhat.com","username":"owalsh"},"state":"REVIEWER"},{"updated":"2019-07-12 14:24:55.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-15 07:12:09.000000000","updated_by":{"_account_id":30133,"name":"Piotr Kopec","email":"pkopec17@gmail.com","username":"pkopec"},"reviewer":{"_account_id":30133,"name":"Piotr Kopec","email":"pkopec17@gmail.com","username":"pkopec"},"state":"REVIEWER"},{"updated":"2019-07-15 07:12:14.000000000","updated_by":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"reviewer":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"state":"REVIEWER"},{"updated":"2019-07-15 11:55:24.000000000","updated_by":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"reviewer":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"state":"REVIEWER"},{"updated":"2019-07-15 14:36:13.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-07-15 18:37:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e9fba2e4558a89793385ec9ae95e68c4eec6945e","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2019-07-12 10:09:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9daa415f8bab7c35303be9383f595cd92ecf8f7b","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2019-07-12 10:11:19.000000000","message":"Topic changed from cell_service_hostnames to bp/reduce-deployment-resources","accounts_in_message":[],"_revision_number":1},{"id":"86a17771ff1ec5e4634a77affc30351fc05eb095","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-12 10:40:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/86/670486/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/87b66a3/ : FAILURE in 26m 58s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/86/670486/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/cd82509/ : FAILURE in 27m 37s","accounts_in_message":[],"_revision_number":1},{"id":"2a816bde8592a9108bb857b804d1330a327a3683","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2019-07-12 11:39:57.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"d45ce796eaf9bddfc68612aba31b7dd8bbac30ef","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2019-07-12 12:02:28.000000000","message":"Topic changed from cell_service_hostnames to bp/reduce-deployment-resources","accounts_in_message":[],"_revision_number":2},{"id":"8b14e77f272a43031ff2e254587cb9299cbf1d89","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-12 14:24:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/86/670486/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/a6acc11/ : SUCCESS in 2h 40m 23s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/86/670486/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/1b9a16b/ : SUCCESS in 2h 43m 01s","accounts_in_message":[],"_revision_number":2},{"id":"2542b5e9245fca086c987abe48ab95689329e010","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 16:03:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/86/670486/2/check/openstack-tox-cover/28fed8c/cover/ : SUCCESS in 3m 47s\n- openstack-tox-pep8 http://logs.openstack.org/86/670486/2/check/openstack-tox-pep8/43e1798/ : SUCCESS in 2m 55s\n- openstack-tox-py27 http://logs.openstack.org/86/670486/2/check/openstack-tox-py27/e2d387f/ : SUCCESS in 4m 03s\n- openstack-tox-py36 http://logs.openstack.org/86/670486/2/check/openstack-tox-py36/d2b8809/ : SUCCESS in 3m 40s\n- openstack-tox-py37 http://logs.openstack.org/86/670486/2/check/openstack-tox-py37/fea6b67/ : SUCCESS in 3m 50s\n- openstack-tox-docs http://logs.openstack.org/86/670486/2/check/openstack-tox-docs/f4496f6/html/ : SUCCESS in 4m 42s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/86/670486/2/check/tripleo-ci-centos-7-containers-multinode/2b04ae3/ : SUCCESS in 2h 13m 25s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/86/670486/2/check/tripleo-ci-centos-7-undercloud-containers/aded61f/ : SUCCESS in 1h 33m 25s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/86/670486/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/39fc0c4/ : FAILURE in 2h 34m 39s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/86/670486/2/check/tripleo-ci-centos-7-standalone/7a90cc6/ : SUCCESS in 1h 44m 01s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/86/670486/2/check/tripleo-ci-fedora-28-standalone/379b92c/ : SUCCESS in 1h 46m 02s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/86/670486/2/check/tripleo-ci-centos-7-standalone-upgrade/3ff9a8f/ : FAILURE in 2h 26m 24s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/86/670486/2/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/759cd43/ : FAILURE in 2h 06m 25s (non-voting)\n- openstack-tox-linters http://logs.openstack.org/86/670486/2/check/openstack-tox-linters/958e279/ : SUCCESS in 3m 48s\n- openstack-tox-py35 http://logs.openstack.org/86/670486/2/check/openstack-tox-py35/6d91705/ : SUCCESS in 4m 37s\n- openstack-tox-lower-constraints http://logs.openstack.org/86/670486/2/check/openstack-tox-lower-constraints/9a485fb/ : SUCCESS in 4m 52s","accounts_in_message":[],"_revision_number":2},{"id":"21f198148700c42a2230bbd5045c5b13960ff124","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2019-07-15 11:55:24.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b3918e92f062020a32e762ec833edb8e29e4e2ee","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-07-15 14:36:13.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"83407e0a4943df851ffd19b00b6b6dd586057b68","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 14:36:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7000d4f410d28c87e9c2cb7752d1c9089c4b8452","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 18:37:17.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/86/670486/2/gate/openstack-tox-pep8/784c4bc/ : SUCCESS in 3m 12s\n- openstack-tox-py27 http://logs.openstack.org/86/670486/2/gate/openstack-tox-py27/bfbb558/ : SUCCESS in 3m 38s\n- openstack-tox-py36 http://logs.openstack.org/86/670486/2/gate/openstack-tox-py36/140d97f/ : SUCCESS in 4m 19s\n- openstack-tox-py37 http://logs.openstack.org/86/670486/2/gate/openstack-tox-py37/3779e21/ : SUCCESS in 5m 16s\n- openstack-tox-docs http://logs.openstack.org/86/670486/2/gate/openstack-tox-docs/0b5be0d/html/ : SUCCESS in 3m 28s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/86/670486/2/gate/tripleo-ci-centos-7-containers-multinode/d5a7f3b/ : SUCCESS in 2h 16m 40s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/86/670486/2/gate/tripleo-ci-centos-7-standalone/164ad19/ : SUCCESS in 1h 51m 59s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/86/670486/2/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/b0f6212/ : SUCCESS in 2h 14m 43s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/86/670486/2/gate/tripleo-ci-centos-7-undercloud-containers/d0ef5b3/ : SUCCESS in 1h 50m 28s\n- openstack-tox-linters http://logs.openstack.org/86/670486/2/gate/openstack-tox-linters/4db6c40/ : SUCCESS in 4m 24s\n- openstack-tox-lower-constraints http://logs.openstack.org/86/670486/2/gate/openstack-tox-lower-constraints/ec4a2f3/ : SUCCESS in 9m 36s","accounts_in_message":[],"_revision_number":2},{"id":"55a725bdffe3df93207c6ac03b3c73e616e3145d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 18:37:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"70b1daa42b41a441b34387a2e76ad3b39900ed8c","revisions":{"6a7e907625b24ea2d27b9847f94022925c3af505":{"kind":"REWORK","_number":1,"created":"2019-07-12 10:09:57.000000000","uploader":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"ref":"refs/changes/86/670486/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/86/670486/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/86/670486/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/86/670486/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/86/670486/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/86/670486/1"}}},"commit":{"parents":[{"commit":"6c381d55bda5b743767d55a802f779190dfb25ea","subject":"Merge \"Remove the tripleo.plan_management.v1.publish_ui_logs_to_swiftworkflow\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/6c381d55bda5b743767d55a802f779190dfb25ea"}]}],"author":{"name":"Martin Schuppert","email":"mschuppert@redhat.com","date":"2019-07-12 10:06:03.000000000","tz":120},"committer":{"name":"Martin Schuppert","email":"mschuppert@redhat.com","date":"2019-07-12 10:06:03.000000000","tz":120},"subject":"Create \u003cservice\u003e _cell_node_names if nova_additional_cell","message":"Create \u003cservice\u003e _cell_node_names if nova_additional_cell\n\nI9a37d1faec73a81a28d8f89d86375fb15ee765c7 moved to use ansible\nfor all_nodes hieradata. This also removed to create the\nSERVICE_cell_node_names in in net_ip_list_map.j2.yaml.\nThis change creates the \u003cservice\u003e _cell_node_names if\nnova_additional_cell is true.\n\nChange-Id: I33effbf481682402d39436a9e97cbd3ee2044577\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/6a7e907625b24ea2d27b9847f94022925c3af505"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/6a7e907625b24ea2d27b9847f94022925c3af505"}]},"branch":"refs/heads/master"},"70b1daa42b41a441b34387a2e76ad3b39900ed8c":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-07-12 11:39:57.000000000","uploader":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"ref":"refs/changes/86/670486/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/86/670486/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/86/670486/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/86/670486/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/86/670486/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/86/670486/2"}}},"commit":{"parents":[{"commit":"6c381d55bda5b743767d55a802f779190dfb25ea","subject":"Merge \"Remove the tripleo.plan_management.v1.publish_ui_logs_to_swiftworkflow\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/6c381d55bda5b743767d55a802f779190dfb25ea"}]}],"author":{"name":"Martin Schuppert","email":"mschuppert@redhat.com","date":"2019-07-12 10:06:03.000000000","tz":120},"committer":{"name":"Martin Schuppert","email":"mschuppert@redhat.com","date":"2019-07-12 11:39:39.000000000","tz":120},"subject":"Create \u003cservice\u003e _cell_node_names if nova_additional_cell","message":"Create \u003cservice\u003e _cell_node_names if nova_additional_cell\n\nI9a37d1faec73a81a28d8f89d86375fb15ee765c7 moved to use ansible\nfor all_nodes hieradata. This also removed to create the\nSERVICE_cell_node_names in in net_ip_list_map.j2.yaml.\nThis change creates the \u003cservice\u003e _cell_node_names if\nnova_additional_cell is true.\n\nDepends-On: https://review.opendev.org/670487\n\nChange-Id: I33effbf481682402d39436a9e97cbd3ee2044577\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/70b1daa42b41a441b34387a2e76ad3b39900ed8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/70b1daa42b41a441b34387a2e76ad3b39900ed8c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
