)]}'
{"id":"openstack%2Ftripleo-common~670974","triplet_id":"openstack%2Ftripleo-common~stable%2Fstein~Ib3d9ea8d9f85ea194a16feacb530d50419d7e54a","project":"openstack/tripleo-common","branch":"stable/stein","topic":"bp/validation-framework-stein","hashtags":[],"change_id":"Ib3d9ea8d9f85ea194a16feacb530d50419d7e54a","subject":"Get all the available Ansible variables for each validations","status":"MERGED","created":"2019-07-16 08:09:40.000000000","updated":"2019-07-19 12:17:45.000000000","submitted":"2019-07-19 01:34:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670974-1563500048351-a431606b","meta_rev_id":"9ddc30816a54fa25df9e8569e1f344cb72d414cc","_number":670974,"virtual_id_number":670974,"owner":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"date":"2019-07-17 09:16:01.000000000","_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"date":"2019-07-18 01:57:34.000000000","_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"value":2,"date":"2019-07-19 01:34:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-18 22:35:24.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},{"value":2,"date":"2019-07-16 13:25:29.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":2,"date":"2019-07-16 08:20:42.000000000","_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"date":"2019-07-16 08:13:51.000000000","_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"date":"2019-07-16 08:13:57.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2019-07-16 08:14:01.000000000","_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"date":"2019-07-16 08:13:43.000000000","_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},{"value":1,"date":"2019-07-16 13:25:29.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-16 08:13:35.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-16 08:13:35.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2019-07-16 08:13:43.000000000","updated_by":{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},"reviewer":{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},"state":"REVIEWER"},{"updated":"2019-07-16 08:13:51.000000000","updated_by":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"reviewer":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"state":"REVIEWER"},{"updated":"2019-07-16 08:13:57.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-07-16 08:14:01.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2019-07-16 08:20:42.000000000","updated_by":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"reviewer":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"state":"REVIEWER"},{"updated":"2019-07-16 13:25:29.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-07-18 01:57:34.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2019-07-18 22:35:24.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-19 01:34:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"22b6ab6a86ef6d44d244ac62105b5cadf8a122cf","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-16 08:09:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2846b1fb3a81d3e093464be56e6f6319ce8a039f","author":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"date":"2019-07-16 08:20:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f776474edc15ad3d191fa1a83abb274ace015271","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 11:12:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/74/670974/1/check/openstack-tox-cover/9250ec1/cover/ : SUCCESS in 3m 19s\n- openstack-tox-pep8 http://logs.openstack.org/74/670974/1/check/openstack-tox-pep8/b5bab5b/ : SUCCESS in 3m 13s\n- openstack-tox-py27 http://logs.openstack.org/74/670974/1/check/openstack-tox-py27/2aad3c0/ : SUCCESS in 3m 02s\n- openstack-tox-py36 http://logs.openstack.org/74/670974/1/check/openstack-tox-py36/7442890/ : SUCCESS in 3m 10s\n- openstack-tox-py37 http://logs.openstack.org/74/670974/1/check/openstack-tox-py37/055568c/ : SUCCESS in 3m 43s\n- openstack-tox-docs http://logs.openstack.org/74/670974/1/check/openstack-tox-docs/586b0bf/html/ : SUCCESS in 3m 22s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/74/670974/1/check/tripleo-ci-centos-7-containers-multinode/7fb0111/ : SUCCESS in 2h 25m 23s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/74/670974/1/check/tripleo-ci-centos-7-undercloud-containers/a9bebf8/ : SUCCESS in 1h 49m 54s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/74/670974/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/6a110d2/ : SUCCESS in 2h 47m 14s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/74/670974/1/check/tripleo-ci-centos-7-standalone/b35d32d/ : SUCCESS in 2h 00m 27s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/74/670974/1/check/tripleo-ci-fedora-28-standalone/8ae3143/ : SUCCESS in 1h 52m 26s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/74/670974/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/5652a07/ : FAILURE in 1h 33m 54s (non-voting)\n- openstack-tox-py35 http://logs.openstack.org/74/670974/1/check/openstack-tox-py35/3796ad8/ : SUCCESS in 4m 48s\n- openstack-tox-lower-constraints http://logs.openstack.org/74/670974/1/check/openstack-tox-lower-constraints/f35d179/ : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":1},{"id":"d8075ace73e103aa78f75bfe5642677903150b3b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-16 11:14:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/74/670974/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/f63ea67/ : SUCCESS in 3h 02m 58s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/74/670974/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/891afdb/ : SUCCESS in 3h 02m 56s","accounts_in_message":[],"_revision_number":1},{"id":"494e870ba1743067f10215dccf8579f16119cafa","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-16 13:25:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d9ca58a846038b7159a59dfdf71640879250e8e1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 13:25:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e9c34bb1e3263a5bcb28cc9ebf8e6c2ac118a21e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 20:43:02.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/74/670974/1/gate/openstack-tox-pep8/164a1c1/ : SUCCESS in 4m 36s\n- openstack-tox-py27 http://logs.openstack.org/74/670974/1/gate/openstack-tox-py27/2746f53/ : SUCCESS in 4m 45s\n- openstack-tox-py36 http://logs.openstack.org/74/670974/1/gate/openstack-tox-py36/44fca72/ : SUCCESS in 3m 41s\n- openstack-tox-py37 http://logs.openstack.org/74/670974/1/gate/openstack-tox-py37/df7677e/ : SUCCESS in 4m 56s\n- openstack-tox-docs http://logs.openstack.org/74/670974/1/gate/openstack-tox-docs/cb94460/html/ : SUCCESS in 5m 13s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/74/670974/1/gate/tripleo-ci-centos-7-containers-multinode/9a56607/ : FAILURE in 2h 34m 27s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/74/670974/1/gate/tripleo-ci-centos-7-standalone/865ffb0/ : SUCCESS in 2h 05m 26s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/74/670974/1/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/ef43582/ : SUCCESS in 2h 35m 24s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/74/670974/1/gate/tripleo-ci-centos-7-undercloud-containers/50c8037/ : SUCCESS in 1h 53m 00s\n- openstack-tox-lower-constraints http://logs.openstack.org/74/670974/1/gate/openstack-tox-lower-constraints/52237e9/ : SUCCESS in 6m 42s","accounts_in_message":[],"_revision_number":1},{"id":"cf91c76a4f57e8b2f28ae96fd7dfe989dae582b5","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-17 09:16:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e5b0e7d8a5a5998bd687cab56818afbab2bc981b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-17 12:09:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/74/670974/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/b859b15/ : SUCCESS in 2h 51m 54s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/74/670974/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/72d7d78/ : SUCCESS in 2h 49m 55s","accounts_in_message":[],"_revision_number":1},{"id":"940a8c862aeaced8dadb5b214216893649aaacf8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 13:27:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/74/670974/1/check/openstack-tox-cover/3b52c18/cover/ : SUCCESS in 3m 40s\n- openstack-tox-pep8 http://logs.openstack.org/74/670974/1/check/openstack-tox-pep8/76c3d77/ : SUCCESS in 3m 01s\n- openstack-tox-py27 http://logs.openstack.org/74/670974/1/check/openstack-tox-py27/e3c3a3d/ : SUCCESS in 4m 55s\n- openstack-tox-py36 http://logs.openstack.org/74/670974/1/check/openstack-tox-py36/588258f/ : SUCCESS in 4m 25s\n- openstack-tox-py37 http://logs.openstack.org/74/670974/1/check/openstack-tox-py37/38f8253/ : SUCCESS in 3m 34s\n- openstack-tox-docs http://logs.openstack.org/74/670974/1/check/openstack-tox-docs/2aaac32/html/ : SUCCESS in 2m 37s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/74/670974/1/check/tripleo-ci-centos-7-containers-multinode/ce97160/ : FAILURE in 2h 19m 33s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/74/670974/1/check/tripleo-ci-centos-7-undercloud-containers/5d62de4/ : SUCCESS in 1h 46m 50s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/74/670974/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/d9cfa48/ : SUCCESS in 2h 56m 28s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/74/670974/1/check/tripleo-ci-centos-7-standalone/c76a99b/ : SUCCESS in 1h 52m 05s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/74/670974/1/check/tripleo-ci-fedora-28-standalone/c7ab368/ : SUCCESS in 1h 44m 07s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/74/670974/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/69b5dab/ : FAILURE in 1h 48m 46s (non-voting)\n- openstack-tox-py35 http://logs.openstack.org/74/670974/1/check/openstack-tox-py35/d2dab7d/ : SUCCESS in 3m 17s\n- openstack-tox-lower-constraints http://logs.openstack.org/74/670974/1/check/openstack-tox-lower-constraints/1510e24/ : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":1},{"id":"8a5d744abc577099837989863db202faef92cd72","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-07-18 01:57:34.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9bd8527aeb386e6652b0768a7aa9bed299b90fd6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-18 04:18:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/74/670974/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/3f55605/ : RETRY_LIMIT in 5m 32s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/74/670974/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/1185681/ : RETRY_LIMIT in 3m 59s","accounts_in_message":[],"_revision_number":1},{"id":"5dd5611f8ce8192d166158b551f8da20162430b8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 05:14:40.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/74/670974/1/check/openstack-tox-cover/355de1e/cover/ : SUCCESS in 5m 58s\n- openstack-tox-pep8 http://logs.openstack.org/74/670974/1/check/openstack-tox-pep8/2376331/ : SUCCESS in 5m 22s\n- openstack-tox-py27 http://logs.openstack.org/74/670974/1/check/openstack-tox-py27/469c204/ : SUCCESS in 5m 10s\n- openstack-tox-py36 http://logs.openstack.org/74/670974/1/check/openstack-tox-py36/bc4bb8d/ : SUCCESS in 5m 46s\n- openstack-tox-py37 http://logs.openstack.org/74/670974/1/check/openstack-tox-py37/02b584b/ : SUCCESS in 5m 43s\n- openstack-tox-docs http://logs.openstack.org/74/670974/1/check/openstack-tox-docs/5842598/html/ : SUCCESS in 4m 55s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/74/670974/1/check/tripleo-ci-centos-7-containers-multinode/f1a6fe6/ : FAILURE in 2h 27m 15s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/74/670974/1/check/tripleo-ci-centos-7-undercloud-containers/28755ec/ : SUCCESS in 2h 05m 04s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/74/670974/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/def628c/ : SUCCESS in 2h 58m 31s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/74/670974/1/check/tripleo-ci-centos-7-standalone/6ea71aa/ : SUCCESS in 2h 21m 37s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/74/670974/1/check/tripleo-ci-fedora-28-standalone/c3acf57/ : SUCCESS in 1h 50m 34s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/74/670974/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/061840e/ : FAILURE in 1h 51m 06s (non-voting)\n- openstack-tox-py35 http://logs.openstack.org/74/670974/1/check/openstack-tox-py35/4b37241/ : SUCCESS in 4m 12s\n- openstack-tox-lower-constraints http://logs.openstack.org/74/670974/1/check/openstack-tox-lower-constraints/d286bf2/ : SUCCESS in 7m 02s","accounts_in_message":[],"_revision_number":1},{"id":"014eb589dce0077e963640763e0c974eaca2417a","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-18 13:38:47.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"08e97a3349d07fe580cff444d0ae62c5d143c555","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-18 17:14:23.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/74/670974/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/ac071f2/ : FAILURE in 2h 00m 05s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/74/670974/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/02f8dc3/ : SUCCESS in 3h 33m 12s","accounts_in_message":[],"_revision_number":1},{"id":"48d1a1bb12a295219705c110bac30997a6705b88","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 18:35:50.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/74/670974/1/check/openstack-tox-cover/a60e418/cover/ : SUCCESS in 4m 20s\n- openstack-tox-pep8 http://logs.openstack.org/74/670974/1/check/openstack-tox-pep8/6ce4839/ : SUCCESS in 3m 45s\n- openstack-tox-py27 http://logs.openstack.org/74/670974/1/check/openstack-tox-py27/23e7a47/ : SUCCESS in 3m 42s\n- openstack-tox-py36 http://logs.openstack.org/74/670974/1/check/openstack-tox-py36/d80abf9/ : SUCCESS in 2m 54s\n- openstack-tox-py37 http://logs.openstack.org/74/670974/1/check/openstack-tox-py37/db974ef/ : SUCCESS in 3m 35s\n- openstack-tox-docs http://logs.openstack.org/74/670974/1/check/openstack-tox-docs/6ecc980/html/ : SUCCESS in 2m 57s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/74/670974/1/check/tripleo-ci-centos-7-containers-multinode/b9bd753/ : FAILURE in 1h 49m 37s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/74/670974/1/check/tripleo-ci-centos-7-undercloud-containers/4cb9c13/ : SUCCESS in 1h 45m 23s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/74/670974/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/28e127c/ : SUCCESS in 2h 47m 45s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/74/670974/1/check/tripleo-ci-centos-7-standalone/818959c/ : SUCCESS in 2h 09m 21s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/74/670974/1/check/tripleo-ci-fedora-28-standalone/650f7f4/ : SUCCESS in 1h 57m 04s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/74/670974/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/0d904cf/ : FAILURE in 1h 23m 36s (non-voting)\n- openstack-tox-py35 http://logs.openstack.org/74/670974/1/check/openstack-tox-py35/a07c775/ : SUCCESS in 5m 15s\n- openstack-tox-lower-constraints http://logs.openstack.org/74/670974/1/check/openstack-tox-lower-constraints/9dcab2e/ : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":1},{"id":"6c47849c3ff42cb28177e384291a7e9446389475","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-18 19:28:34.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"de32af5662d99ceb74d6e971fde6df3294839073","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-18 22:35:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/74/670974/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/bac014b/ : SUCCESS in 3h 03m 28s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/74/670974/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/d2f0b1e/ : SUCCESS in 3h 05m 21s","accounts_in_message":[],"_revision_number":1},{"id":"dcff675a39d3b61585d7c36970e9d4825ab8e4a0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 22:41:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/74/670974/1/check/openstack-tox-cover/beab25e/cover/ : SUCCESS in 4m 01s\n- openstack-tox-pep8 http://logs.openstack.org/74/670974/1/check/openstack-tox-pep8/465e401/ : SUCCESS in 3m 38s\n- openstack-tox-py27 http://logs.openstack.org/74/670974/1/check/openstack-tox-py27/d136f56/ : SUCCESS in 3m 30s\n- openstack-tox-py36 http://logs.openstack.org/74/670974/1/check/openstack-tox-py36/83cbdc6/ : SUCCESS in 2m 46s\n- openstack-tox-py37 http://logs.openstack.org/74/670974/1/check/openstack-tox-py37/6c75520/ : SUCCESS in 3m 32s\n- openstack-tox-docs http://logs.openstack.org/74/670974/1/check/openstack-tox-docs/f8af354/html/ : SUCCESS in 3m 17s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/74/670974/1/check/tripleo-ci-centos-7-containers-multinode/8016246/ : SUCCESS in 2h 20m 10s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/74/670974/1/check/tripleo-ci-centos-7-undercloud-containers/7ab04e4/ : SUCCESS in 1h 26m 17s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/74/670974/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/2ae8445/ : SUCCESS in 2h 54m 30s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/74/670974/1/check/tripleo-ci-centos-7-standalone/7b81e9a/ : SUCCESS in 1h 42m 15s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/74/670974/1/check/tripleo-ci-fedora-28-standalone/d2521ec/ : SUCCESS in 1h 56m 15s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/74/670974/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/c330d74/ : FAILURE in 1h 32m 57s (non-voting)\n- openstack-tox-py35 http://logs.openstack.org/74/670974/1/check/openstack-tox-py35/fb0471c/ : SUCCESS in 4m 11s\n- openstack-tox-lower-constraints http://logs.openstack.org/74/670974/1/check/openstack-tox-lower-constraints/c614d6f/ : SUCCESS in 5m 31s","accounts_in_message":[],"_revision_number":1},{"id":"8f737dd6a5d1abed2f795ac14e81a96655a81145","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 22:41:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7c6b236a7c5ca4a8e16dc02d9a91a594bc6f5a2e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 01:34:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/74/670974/1/gate/openstack-tox-pep8/388283c/ : SUCCESS in 3m 25s\n- openstack-tox-py27 http://logs.openstack.org/74/670974/1/gate/openstack-tox-py27/5f6884e/ : SUCCESS in 4m 47s\n- openstack-tox-py36 http://logs.openstack.org/74/670974/1/gate/openstack-tox-py36/465159e/ : SUCCESS in 4m 53s\n- openstack-tox-py37 http://logs.openstack.org/74/670974/1/gate/openstack-tox-py37/d2c7f23/ : SUCCESS in 4m 39s\n- openstack-tox-docs http://logs.openstack.org/74/670974/1/gate/openstack-tox-docs/16bdb81/html/ : SUCCESS in 3m 01s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/74/670974/1/gate/tripleo-ci-centos-7-containers-multinode/396b728/ : SUCCESS in 2h 24m 21s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/74/670974/1/gate/tripleo-ci-centos-7-standalone/9da27d4/ : SUCCESS in 2h 17m 36s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/74/670974/1/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/63d66dc/ : SUCCESS in 2h 48m 32s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/74/670974/1/gate/tripleo-ci-centos-7-undercloud-containers/a1b5db8/ : SUCCESS in 1h 38m 22s\n- openstack-tox-lower-constraints http://logs.openstack.org/74/670974/1/gate/openstack-tox-lower-constraints/6969f32/ : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":1},{"id":"da6e95ec650a0495ea5b1ab1415dd02740a0937f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 01:34:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"bf81b1949e014eb10b7e97ca5b8e466dfc8b2949","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-19 12:17:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/74/670974/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/04a1c1a/ : SUCCESS in 2h 33m 23s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/74/670974/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/87990fe/ : SUCCESS in 2h 37m 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9d5225101fc51c20df45f2744f42f2d776362955","revisions":{"9d5225101fc51c20df45f2744f42f2d776362955":{"kind":"REWORK","_number":1,"created":"2019-07-16 08:09:40.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/74/670974/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/74/670974/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/670974/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/670974/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/670974/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/74/670974/1"}}},"commit":{"parents":[{"commit":"d564b1752d56ca0815005b0e722c1c16e0e3485e","subject":"Merge \"Install dumb-init rpm\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/d564b1752d56ca0815005b0e722c1c16e0e3485e"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-05 10:22:59.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-16 08:08:52.000000000","tz":120},"subject":"Get all the available Ansible variables for each validations","message":"Get all the available Ansible variables for each validations\n\nAll the Ansible variables in the validations playbooks must be at the\nsame level as the metadata and not part of them.\n\n```\n- hosts: myhost\n  vars:\n    metadata:\n      name: Validation name\n      description: What should this validation do...\n      groups:\n        - prep\n        - pre-deployment\n    variable1: \"value1\"\n    variable2: \"value2\"\n    ...\n```\n\nChange-Id: Ib3d9ea8d9f85ea194a16feacb530d50419d7e54a\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n(cherry picked from commit 569217425974c47314fe27fb91f6c006c3445bb3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/9d5225101fc51c20df45f2744f42f2d776362955"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/9d5225101fc51c20df45f2744f42f2d776362955"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
