)]}'
{"id":"openstack%2Ftripleo-common~671482","triplet_id":"openstack%2Ftripleo-common~master~I45adc2b610783c42394ed701fec4e7abbf3de974","project":"openstack/tripleo-common","branch":"master","hashtags":[],"change_id":"I45adc2b610783c42394ed701fec4e7abbf3de974","subject":"Bump ceph/daemon container image to v4.0.1","status":"MERGED","created":"2019-07-18 13:21:18.000000000","updated":"2019-07-29 21:09:06.000000000","submitted":"2019-07-29 21:09:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"671482-1564434546276-94390bb6","meta_rev_id":"735de4f60f94f21b9dcccd60bb615bd691013bb2","_number":671482,"virtual_id_number":671482,"owner":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"date":"2019-07-18 14:48:49.000000000","_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"value":2,"date":"2019-07-29 21:09:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-18 16:44:01.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":14270,"name":"Dimitri Savineau","email":"dsavinea@redhat.com","username":"dsavineau"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-07-29 18:03:37.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-07-23 22:50:44.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2019-07-24 07:17:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":14270,"name":"Dimitri Savineau","email":"dsavinea@redhat.com","username":"dsavineau"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-07-29 18:03:37.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-18 17:56:09.000000000","_account_id":14270,"name":"Dimitri Savineau","email":"dsavinea@redhat.com","username":"dsavineau"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"_account_id":14270,"name":"Dimitri Savineau","email":"dsavinea@redhat.com","username":"dsavineau"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-18 16:44:01.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-18 17:56:09.000000000","updated_by":{"_account_id":14270,"name":"Dimitri Savineau","email":"dsavinea@redhat.com","username":"dsavineau"},"reviewer":{"_account_id":14270,"name":"Dimitri Savineau","email":"dsavinea@redhat.com","username":"dsavineau"},"state":"REVIEWER"},{"updated":"2019-07-23 22:50:44.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-07-24 07:17:56.000000000","updated_by":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"reviewer":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"state":"REVIEWER"},{"updated":"2019-07-29 18:03:37.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-07-29 21:09:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"172cf0d0421779f4d09e5be3947429947d314b19","author":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"date":"2019-07-18 13:21:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ee260ac6c6eea8d1031779c768837086cca08e50","author":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"date":"2019-07-18 14:48:49.000000000","message":"Patch Set 1:\n\neventually this needs to be backported into stable/stein as well","accounts_in_message":[],"_revision_number":1},{"id":"0770ce870c7d8479b460bf7289a4ba369a22aec7","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-18 16:44:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/82/671482/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/998ce45/ : SUCCESS in 3h 15m 42s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/82/671482/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/2852473/ : SUCCESS in 3h 21m 07s","accounts_in_message":[],"_revision_number":1},{"id":"3b649686293d75f1847411d206a3c12ae22df7c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 17:31:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/82/671482/1/check/openstack-tox-cover/166e6b1/cover/ : SUCCESS in 5m 08s\n- openstack-tox-pep8 http://logs.openstack.org/82/671482/1/check/openstack-tox-pep8/1ebb161/ : SUCCESS in 4m 26s\n- openstack-tox-py27 http://logs.openstack.org/82/671482/1/check/openstack-tox-py27/34c2af3/ : SUCCESS in 3m 20s\n- openstack-tox-py36 http://logs.openstack.org/82/671482/1/check/openstack-tox-py36/077d778/ : SUCCESS in 3m 25s\n- openstack-tox-py37 http://logs.openstack.org/82/671482/1/check/openstack-tox-py37/3cd5dd7/ : SUCCESS in 4m 12s\n- openstack-tox-docs http://logs.openstack.org/82/671482/1/check/openstack-tox-docs/30ab772/html/ : SUCCESS in 5m 32s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/82/671482/1/check/tripleo-ci-centos-7-containers-multinode/d83ca28/ : SUCCESS in 2h 31m 07s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/82/671482/1/check/tripleo-ci-centos-7-undercloud-containers/c39ca22/ : SUCCESS in 1h 42m 56s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/82/671482/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/dc317e3/ : SUCCESS in 2h 20m 00s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/82/671482/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/2420d39/ : FAILURE in 2h 42m 54s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/82/671482/1/check/tripleo-ci-centos-7-standalone/19c2158/ : SUCCESS in 1h 47m 00s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/82/671482/1/check/tripleo-ci-fedora-28-standalone/521cb4f/ : SUCCESS in 1h 46m 05s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/82/671482/1/check/tripleo-ci-centos-7-standalone-upgrade/d9b16d1/ : FAILURE in 2h 06m 14s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/82/671482/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/08eb499/ : FAILURE in 1h 52m 17s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/82/671482/1/check/tripleo-build-containers-centos-7/5f64012/ : SUCCESS in 46m 58s\n- tripleo-build-containers-centos-7-buildah http://logs.openstack.org/82/671482/1/check/tripleo-build-containers-centos-7-buildah/6f05dea/ : SUCCESS in 1h 05m 53s\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/82/671482/1/check/tripleo-ci-centos-7-scenario001-standalone/257d70b/ : SUCCESS in 1h 41m 49s\n- tripleo-ci-centos-7-scenario002-standalone http://logs.openstack.org/82/671482/1/check/tripleo-ci-centos-7-scenario002-standalone/e5eed00/ : SUCCESS in 1h 38m 10s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/82/671482/1/check/tripleo-ci-centos-7-scenario003-standalone/147995e/ : SUCCESS in 1h 15m 44s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/82/671482/1/check/tripleo-ci-centos-7-scenario004-standalone/46b3f57/ : SUCCESS in 1h 49m 24s\n- tripleo-ci-centos-7-scenario010-standalone http://logs.openstack.org/82/671482/1/check/tripleo-ci-centos-7-scenario010-standalone/5285c9d/ : SUCCESS in 1h 23m 14s\n- tripleo-ci-centos-7-scenario012-standalone http://logs.openstack.org/82/671482/1/check/tripleo-ci-centos-7-scenario012-standalone/6fefe28/ : SUCCESS in 1h 57m 00s (non-voting)\n- openstack-tox-linters http://logs.openstack.org/82/671482/1/check/openstack-tox-linters/9dc8f04/ : SUCCESS in 3m 54s\n- openstack-tox-py35 http://logs.openstack.org/82/671482/1/check/openstack-tox-py35/7295106/ : SUCCESS in 3m 42s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/82/671482/1/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/59b6a16/ : SUCCESS in 2h 33m 00s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/82/671482/1/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/dadaa1a/ : FAILURE in 2h 02m 53s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/82/671482/1/check/openstack-tox-lower-constraints/9a4d139/ : SUCCESS in 5m 31s","accounts_in_message":[],"_revision_number":1},{"id":"8e9d01bc9b1fe52191f2cebacfad035f27791dce","author":{"_account_id":14270,"name":"Dimitri Savineau","email":"dsavinea@redhat.com","username":"dsavineau"},"date":"2019-07-18 17:56:09.000000000","message":"Patch Set 1:\n\nI don\u0027t know if you\u0027re interesting but you could use the container image tag without the \u0027-x86_64\u0027 suffix.\nv4.0.1-stable-4.0-nautilus-centos-7 is a manifest list supporting both x86_64 and aarch64 architectures.","accounts_in_message":[],"_revision_number":1},{"id":"a5238c1b132737835d8d2ebd9ab5ce41e72536dc","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-23 22:50:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"86ac7250dcd81cd487482fe1da1590ae1dbf466c","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2019-07-24 07:17:56.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c9956045400da815957f298544a70c592a9ab013","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-07-29 18:03:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e5da70f9491c07a01e8692d4cd4ab3d2ec309fb0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 18:03:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6e469b3a6297ec7af096296a56911598744cc23b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 21:09:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/82/671482/1/gate/openstack-tox-pep8/6d17492/ : SUCCESS in 4m 04s\n- openstack-tox-py27 http://logs.openstack.org/82/671482/1/gate/openstack-tox-py27/fe21d7c/ : SUCCESS in 4m 36s\n- openstack-tox-py36 http://logs.openstack.org/82/671482/1/gate/openstack-tox-py36/1009455/ : SUCCESS in 4m 38s\n- openstack-tox-py37 http://logs.openstack.org/82/671482/1/gate/openstack-tox-py37/02c4460/ : SUCCESS in 4m 33s\n- openstack-tox-docs http://logs.openstack.org/82/671482/1/gate/openstack-tox-docs/737466d/html/ : SUCCESS in 4m 56s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/82/671482/1/gate/tripleo-ci-centos-7-containers-multinode/b32e0ae/ : SUCCESS in 2h 41m 13s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/82/671482/1/gate/tripleo-ci-centos-7-standalone/d082dd4/ : SUCCESS in 2h 10m 15s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/82/671482/1/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/6dc7d2f/ : SUCCESS in 2h 32m 35s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/82/671482/1/gate/tripleo-ci-centos-7-undercloud-containers/73afb17/ : SUCCESS in 1h 37m 53s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/82/671482/1/gate/tripleo-build-containers-centos-7/eeaa06b/ : SUCCESS in 51m 08s\n- tripleo-build-containers-centos-7-buildah http://logs.openstack.org/82/671482/1/gate/tripleo-build-containers-centos-7-buildah/78e95f0/ : SUCCESS in 55m 12s\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/82/671482/1/gate/tripleo-ci-centos-7-scenario001-standalone/bdedaf1/ : SUCCESS in 1h 44m 56s\n- tripleo-ci-centos-7-scenario002-standalone http://logs.openstack.org/82/671482/1/gate/tripleo-ci-centos-7-scenario002-standalone/48377e5/ : SUCCESS in 1h 48m 12s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/82/671482/1/gate/tripleo-ci-centos-7-scenario003-standalone/3312283/ : SUCCESS in 1h 12m 09s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/82/671482/1/gate/tripleo-ci-centos-7-scenario004-standalone/4b524c3/ : SUCCESS in 1h 53m 22s\n- tripleo-ci-centos-7-scenario010-standalone http://logs.openstack.org/82/671482/1/gate/tripleo-ci-centos-7-scenario010-standalone/4c7d6bf/ : SUCCESS in 1h 27m 35s\n- openstack-tox-linters http://logs.openstack.org/82/671482/1/gate/openstack-tox-linters/13d3798/ : SUCCESS in 5m 10s\n- openstack-tox-lower-constraints http://logs.openstack.org/82/671482/1/gate/openstack-tox-lower-constraints/c2178d4/ : SUCCESS in 6m 34s","accounts_in_message":[],"_revision_number":1},{"id":"735de4f60f94f21b9dcccd60bb615bd691013bb2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 21:09:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9448a6380a473687145571e7d8934e1626c0cb33","revisions":{"9448a6380a473687145571e7d8934e1626c0cb33":{"kind":"REWORK","_number":1,"created":"2019-07-18 13:21:18.000000000","uploader":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"ref":"refs/changes/82/671482/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/82/671482/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/82/671482/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/82/671482/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/82/671482/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/82/671482/1"}}},"commit":{"parents":[{"commit":"fd15364734adbfccc9c8c8790002bf8a7bedb15d","subject":"Merge \"Set scenario009 job voting to false\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/fd15364734adbfccc9c8c8790002bf8a7bedb15d"}]}],"author":{"name":"Giulio Fidente","email":"gfidente@redhat.com","date":"2019-07-18 13:20:29.000000000","tz":120},"committer":{"name":"Giulio Fidente","email":"gfidente@redhat.com","date":"2019-07-18 13:20:29.000000000","tz":120},"subject":"Bump ceph/daemon container image to v4.0.1","message":"Bump ceph/daemon container image to v4.0.1\n\nTested via https://review.opendev.org/562213\n\nChange-Id: I45adc2b610783c42394ed701fec4e7abbf3de974\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/9448a6380a473687145571e7d8934e1626c0cb33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/9448a6380a473687145571e7d8934e1626c0cb33"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
