)]}'
{"id":"openstack%2Ftripleo-common~672632","triplet_id":"openstack%2Ftripleo-common~master~I3fd8465a718ac90fd23bd3e9fe99a4477c968c10","project":"openstack/tripleo-common","branch":"master","topic":"ovn_vip","hashtags":[],"change_id":"I3fd8465a718ac90fd23bd3e9fe99a4477c968c10","subject":"WIP Separate VIP for OVN DBs","status":"ABANDONED","created":"2019-07-25 06:27:44.000000000","updated":"2019-08-27 05:52:37.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8503b163a1bc38cb168b976978a1933c3f2bb06a","_number":672632,"virtual_id_number":672632,"owner":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"disliked":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"all":[{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"date":"2019-08-27 05:52:23.000000000","_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":1,"date":"2019-08-26 09:47:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-08-26 09:27:54.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"all":[{"value":-1,"date":"2019-08-26 06:40:59.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-11 01:11:36.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2019-08-26 09:27:54.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-26 09:47:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d66e1722ea239e8af4c4bcffd62adb63bf4bd795","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-07-25 06:27:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"98fd20b2f239db1908957f141419ab3b2168bf69","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-25 06:56:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/32/672632/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/b99fa32/ : FAILURE in 26m 54s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/32/672632/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/2a961aa/ : FAILURE in 27m 14s","accounts_in_message":[],"_revision_number":1},{"id":"c6bc808a61c7c856a7567e0c3dab88574d112638","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 08:54:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/32/672632/1/check/openstack-tox-cover/5da1a70/cover/ : SUCCESS in 5m 28s\n- openstack-tox-pep8 http://logs.openstack.org/32/672632/1/check/openstack-tox-pep8/f4bd977/ : SUCCESS in 4m 19s\n- openstack-tox-py27 http://logs.openstack.org/32/672632/1/check/openstack-tox-py27/420a6a1/ : SUCCESS in 3m 29s\n- openstack-tox-py36 http://logs.openstack.org/32/672632/1/check/openstack-tox-py36/ce1fe32/ : SUCCESS in 4m 11s\n- openstack-tox-py37 http://logs.openstack.org/32/672632/1/check/openstack-tox-py37/c5f4c73/ : SUCCESS in 5m 08s\n- openstack-tox-docs http://logs.openstack.org/32/672632/1/check/openstack-tox-docs/25f94c4/html/ : SUCCESS in 4m 09s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/32/672632/1/check/tripleo-ci-centos-7-containers-multinode/c4aefde/ : FAILURE in 40m 11s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/32/672632/1/check/tripleo-ci-centos-7-undercloud-containers/4a810d5/ : FAILURE in 35m 08s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/32/672632/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/0742540/ : FAILURE in 39m 43s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/32/672632/1/check/tripleo-ci-centos-7-standalone/39f4e9b/ : FAILURE in 30m 41s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/32/672632/1/check/tripleo-ci-fedora-28-standalone/ebc935e/ : FAILURE in 32m 43s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/32/672632/1/check/tripleo-ci-centos-7-standalone-upgrade/53ed1f3/ : FAILURE in 1h 31m 46s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/32/672632/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/720da5e/ : FAILURE in 1h 30m 21s (non-voting)\n- openstack-tox-linters http://logs.openstack.org/32/672632/1/check/openstack-tox-linters/d4ed959/ : SUCCESS in 5m 02s\n- openstack-tox-py35 http://logs.openstack.org/32/672632/1/check/openstack-tox-py35/8deb994/ : SUCCESS in 4m 33s\n- openstack-tox-lower-constraints http://logs.openstack.org/32/672632/1/check/openstack-tox-lower-constraints/3a368ce/ : SUCCESS in 7m 26s","accounts_in_message":[],"_revision_number":1},{"id":"8ae2688ceee79d14c0597800ec27c52da8b65c2f","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-07-25 09:23:08.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"45a21fc5c373fae0ae8d1bb59dd50d43d298a85a","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-07-25 12:08:49.000000000","message":"Topic set to ovn_vip","accounts_in_message":[],"_revision_number":2},{"id":"27833ccf13153473afccb013098051631923ec86","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-25 12:16:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/32/672632/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/8cf034f/ : FAILURE in 2h 30m 43s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/32/672632/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/6c34216/ : FAILURE in 2h 51m 40s","accounts_in_message":[],"_revision_number":2},{"id":"8123b258679fdb50244fdcad3b2a476dcef3bab3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 13:22:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/32/672632/2/check/openstack-tox-cover/7b52edb/cover/ : SUCCESS in 5m 48s\n- openstack-tox-pep8 http://logs.openstack.org/32/672632/2/check/openstack-tox-pep8/79dea70/ : SUCCESS in 5m 16s\n- openstack-tox-py27 http://logs.openstack.org/32/672632/2/check/openstack-tox-py27/7520167/ : SUCCESS in 6m 12s\n- openstack-tox-py36 http://logs.openstack.org/32/672632/2/check/openstack-tox-py36/73ce282/ : SUCCESS in 4m 23s\n- openstack-tox-py37 http://logs.openstack.org/32/672632/2/check/openstack-tox-py37/eda48f4/ : SUCCESS in 5m 11s\n- openstack-tox-docs http://logs.openstack.org/32/672632/2/check/openstack-tox-docs/02c0634/html/ : SUCCESS in 3m 28s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/32/672632/2/check/tripleo-ci-centos-7-containers-multinode/e708a6d/ : FAILURE in 2h 05m 15s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/32/672632/2/check/tripleo-ci-centos-7-undercloud-containers/16f37c8/ : SUCCESS in 1h 41m 29s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/32/672632/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/4d8e6b5/ : SUCCESS in 3h 01m 32s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/32/672632/2/check/tripleo-ci-centos-7-standalone/ff9fb01/ : SUCCESS in 1h 52m 25s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/32/672632/2/check/tripleo-ci-fedora-28-standalone/1d576b1/ : SUCCESS in 1h 52m 13s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/32/672632/2/check/tripleo-ci-centos-7-standalone-upgrade/f53dee4/ : FAILURE in 1h 57m 54s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/32/672632/2/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/9240481/ : FAILURE in 2h 15m 50s (non-voting)\n- openstack-tox-linters http://logs.openstack.org/32/672632/2/check/openstack-tox-linters/ed11ed9/ : SUCCESS in 4m 23s\n- openstack-tox-py35 http://logs.openstack.org/32/672632/2/check/openstack-tox-py35/986578b/ : SUCCESS in 4m 38s\n- openstack-tox-lower-constraints http://logs.openstack.org/32/672632/2/check/openstack-tox-lower-constraints/9058c5b/ : SUCCESS in 6m 04s","accounts_in_message":[],"_revision_number":2},{"id":"11ca42e0cc2416326a78aab1c3a85eba49eadcd0","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-07-27 12:08:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"cd60a9542447a1082ce624d85615bd017a2f61ce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-27 12:22:12.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"9d9c4d13573dea50175dde8dfa1c2f500fac31e7","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-27 12:22:13.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"c15aeeee1af9849eadcb61a111330c8b5c284040","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-07-29 15:06:16.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"80d4640f5dfba091eae98a30fb9e3c96a8b00c92","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-29 17:44:41.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/32/672632/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/46fc249/ : FAILURE in 2h 36m 07s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/32/672632/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/d1d5c97/ : FAILURE in 2h 31m 19s","accounts_in_message":[],"_revision_number":4},{"id":"5beb906337861bff01c8e96961a4307cbe1b5f8d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 19:07:04.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/32/672632/4/check/openstack-tox-cover/c0d0617/cover/ : SUCCESS in 3m 43s\n- openstack-tox-pep8 http://logs.openstack.org/32/672632/4/check/openstack-tox-pep8/c7bf99e/ : SUCCESS in 2m 47s\n- openstack-tox-py27 http://logs.openstack.org/32/672632/4/check/openstack-tox-py27/6d74e9f/ : SUCCESS in 4m 28s\n- openstack-tox-py36 http://logs.openstack.org/32/672632/4/check/openstack-tox-py36/af812c3/ : SUCCESS in 4m 40s\n- openstack-tox-py37 http://logs.openstack.org/32/672632/4/check/openstack-tox-py37/201bf3e/ : SUCCESS in 5m 45s\n- openstack-tox-docs http://logs.openstack.org/32/672632/4/check/openstack-tox-docs/152827e/html/ : SUCCESS in 6m 14s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/32/672632/4/check/tripleo-ci-centos-7-containers-multinode/449699a/ : FAILURE in 2h 15m 51s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/32/672632/4/check/tripleo-ci-centos-7-undercloud-containers/c7bbd9f/ : SUCCESS in 1h 43m 01s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/32/672632/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/25ce61e/ : SUCCESS in 2h 36m 07s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/32/672632/4/check/tripleo-ci-centos-7-standalone/deb4208/ : SUCCESS in 2h 03m 59s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/32/672632/4/check/tripleo-ci-fedora-28-standalone/2637674/ : SUCCESS in 1h 46m 53s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/32/672632/4/check/tripleo-ci-centos-7-standalone-upgrade/f448be4/ : SUCCESS in 2h 09m 22s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/32/672632/4/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/5202a3c/ : FAILURE in 2h 18m 36s (non-voting)\n- openstack-tox-linters http://logs.openstack.org/32/672632/4/check/openstack-tox-linters/94a9363/ : SUCCESS in 5m 05s\n- openstack-tox-py35 http://logs.openstack.org/32/672632/4/check/openstack-tox-py35/3c69555/ : SUCCESS in 8m 13s\n- openstack-tox-lower-constraints http://logs.openstack.org/32/672632/4/check/openstack-tox-lower-constraints/28e1874/ : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":4},{"id":"6ad42d4d2ff05ea9873cf8df92f29ba0cab5781e","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-08-02 08:05:10.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"3cb30a76fda178f3d6d2ec3b3bc9ebbf4114a7c1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-02 10:46:59.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/32/672632/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/c7775cf/ : FAILURE in 2h 22m 24s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/32/672632/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/c146bb0/ : FAILURE in 2h 39m 56s","accounts_in_message":[],"_revision_number":5},{"id":"b9eafdbeb87c6f1f81997ec537b6dbf8413be4a1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 11:38:46.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://logs.opendev.org/32/672632/5/check/openstack-tox-cover/2141f2c/cover/ : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://logs.opendev.org/32/672632/5/check/openstack-tox-pep8/ec4af93/ : SUCCESS in 4m 43s\n- openstack-tox-py27 https://logs.opendev.org/32/672632/5/check/openstack-tox-py27/c72e985/ : SUCCESS in 5m 16s\n- openstack-tox-py36 https://logs.opendev.org/32/672632/5/check/openstack-tox-py36/14d85f6/ : SUCCESS in 3m 08s\n- openstack-tox-py37 https://logs.opendev.org/32/672632/5/check/openstack-tox-py37/4fa4c45/ : SUCCESS in 5m 41s\n- openstack-tox-docs https://logs.opendev.org/32/672632/5/check/openstack-tox-docs/8e0ccf1/html/ : SUCCESS in 4m 58s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/32/672632/5/check/tripleo-ci-centos-7-containers-multinode/d3ea6dd/ : FAILURE in 2h 00m 40s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/32/672632/5/check/tripleo-ci-centos-7-undercloud-containers/805ebba/ : SUCCESS in 1h 48m 36s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/32/672632/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/98bdb81/ : SUCCESS in 2h 43m 01s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/32/672632/5/check/tripleo-ci-centos-7-standalone/c27659f/ : SUCCESS in 2h 06m 34s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/32/672632/5/check/tripleo-ci-fedora-28-standalone/f77246a/ : FAILURE in 57m 09s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/32/672632/5/check/tripleo-ci-centos-7-standalone-upgrade/aeab945/ : FAILURE in 1h 49m 54s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/32/672632/5/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/cf58ed3/ : TIMED_OUT in 3h 13m 31s (non-voting)\n- openstack-tox-linters https://logs.opendev.org/32/672632/5/check/openstack-tox-linters/fa68089/ : SUCCESS in 3m 46s\n- openstack-tox-py35 https://logs.opendev.org/32/672632/5/check/openstack-tox-py35/2321ed0/ : SUCCESS in 6m 16s\n- openstack-tox-lower-constraints https://logs.opendev.org/32/672632/5/check/openstack-tox-lower-constraints/74555a0/ : SUCCESS in 5m 59s","accounts_in_message":[],"_revision_number":5},{"id":"44aba0129bb0679a4e06ee211a4a692859355d2f","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-08-06 08:09:48.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"34fa22fb611b0ae171cad24ebe10e53ab329404f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-06 11:00:15.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/32/672632/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/500894f/ : FAILURE in 2h 47m 54s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/32/672632/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/6e7a693/ : FAILURE in 2h 34m 42s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/32/672632/6/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/d3284a4/ : FAILURE in 9m 53s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/32/672632/6/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/45269cd/ : FAILURE in 9m 25s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"a5cf3555a127f05d7b7e92315a6a0a38ecde92de","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 14:24:10.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://logs.opendev.org/32/672632/6/check/openstack-tox-cover/2d18951/cover/ : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://logs.opendev.org/32/672632/6/check/openstack-tox-pep8/e7a4eba/ : SUCCESS in 4m 06s\n- openstack-tox-py27 https://logs.opendev.org/32/672632/6/check/openstack-tox-py27/849745f/ : SUCCESS in 6m 23s\n- openstack-tox-py36 https://logs.opendev.org/32/672632/6/check/openstack-tox-py36/02744f8/ : SUCCESS in 3m 42s\n- openstack-tox-py37 https://logs.opendev.org/32/672632/6/check/openstack-tox-py37/d4ac716/ : SUCCESS in 4m 11s\n- openstack-tox-docs https://logs.opendev.org/32/672632/6/check/openstack-tox-docs/29bc1a9/html/ : SUCCESS in 4m 43s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/32/672632/6/check/tripleo-ci-centos-7-containers-multinode/be20bc3/ : FAILURE in 1h 17m 06s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/32/672632/6/check/tripleo-ci-centos-7-undercloud-containers/fbe9dde/ : SUCCESS in 1h 52m 11s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/32/672632/6/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/bf8cc90/ : TIMED_OUT in 3h 27m 40s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/32/672632/6/check/tripleo-ci-centos-7-standalone/69d241b/ : SUCCESS in 2h 02m 45s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/32/672632/6/check/tripleo-ci-fedora-28-standalone/3d10565/ : SUCCESS in 2h 15m 17s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/32/672632/6/check/tripleo-ci-centos-7-standalone-upgrade/b208980/ : SUCCESS in 2h 14m 12s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/32/672632/6/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/16a329d/ : FAILURE in 2h 23m 21s (non-voting)\n- openstack-tox-linters https://logs.opendev.org/32/672632/6/check/openstack-tox-linters/cf56255/ : SUCCESS in 4m 31s\n- openstack-tox-py35 https://logs.opendev.org/32/672632/6/check/openstack-tox-py35/e76fef3/ : SUCCESS in 4m 28s\n- openstack-tox-lower-constraints https://logs.opendev.org/32/672632/6/check/openstack-tox-lower-constraints/920efbc/ : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":6},{"id":"9aed3b1ead8b45dbe33dbb590a41ef25d0b5226f","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-08-09 04:00:14.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"6bce2ddd3dcb85826b4da6642a8c4621a213d4d0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-09 06:51:37.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/32/672632/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/beb3793/ : FAILURE in 2h 49m 39s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/32/672632/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/e6e819d/ : FAILURE in 2h 47m 29s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/32/672632/7/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/c173c66/ : SUCCESS in 19m 42s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/32/672632/7/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/d6905d8/ : SUCCESS in 17m 47s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"d208cf63b67592ee7630c6ce5a6cd2fecfc7f413","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 06:58:50.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://logs.opendev.org/32/672632/7/check/openstack-tox-cover/d41671b/cover/ : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://logs.opendev.org/32/672632/7/check/openstack-tox-pep8/cec60fb/ : SUCCESS in 4m 08s\n- openstack-tox-py27 https://logs.opendev.org/32/672632/7/check/openstack-tox-py27/7a67900/ : SUCCESS in 3m 40s\n- openstack-tox-py36 https://logs.opendev.org/32/672632/7/check/openstack-tox-py36/a36fb36/ : SUCCESS in 3m 14s\n- openstack-tox-py37 https://logs.opendev.org/32/672632/7/check/openstack-tox-py37/20f1f4b/ : SUCCESS in 4m 14s\n- openstack-tox-docs https://logs.opendev.org/32/672632/7/check/openstack-tox-docs/498f766/html/ : SUCCESS in 3m 23s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/32/672632/7/check/tripleo-ci-centos-7-containers-multinode/6e910f9/ : FAILURE in 2h 05m 26s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/32/672632/7/check/tripleo-ci-centos-7-undercloud-containers/5f1b83c/ : SUCCESS in 1h 30m 33s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/32/672632/7/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/9e77661/ : SUCCESS in 2h 43m 35s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/32/672632/7/check/tripleo-ci-centos-7-standalone/b43bc71/ : SUCCESS in 1h 49m 26s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/32/672632/7/check/tripleo-ci-fedora-28-standalone/0138ff4/ : SUCCESS in 1h 39m 44s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/32/672632/7/check/tripleo-ci-centos-7-standalone-upgrade/f37ece7/ : POST_FAILURE in 2h 25m 30s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/32/672632/7/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/2c13bb7/ : FAILURE in 2h 11m 54s (non-voting)\n- openstack-tox-linters https://logs.opendev.org/32/672632/7/check/openstack-tox-linters/27c3e15/ : SUCCESS in 4m 14s\n- openstack-tox-py35 https://logs.opendev.org/32/672632/7/check/openstack-tox-py35/a2abad8/ : SUCCESS in 4m 30s\n- openstack-tox-lower-constraints https://logs.opendev.org/32/672632/7/check/openstack-tox-lower-constraints/1915565/ : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":7},{"id":"93d57032c877d322c348092426d27df9725e58b5","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-08-11 01:11:36.000000000","message":"Patch Set 7: Code-Review-1\n\nThis role is being moved to tripleo-ansible. I\u0027d rather see this work land there as a dependent patch.","accounts_in_message":[],"_revision_number":7},{"id":"6cf7c0ea36dc2a14c655fa952ae447fadbd773e1","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-08-23 07:17:24.000000000","message":"Abandoned\n\nMove to https://review.opendev.org/#/c/678168/","accounts_in_message":[],"_revision_number":7},{"id":"8f17ced99b7e7b3208931126d584cf88e08263cb","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-08-26 06:39:31.000000000","message":"Restored","accounts_in_message":[],"_revision_number":7},{"id":"2319c6e788060446f975eb77722cc941a9b5e908","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-08-26 06:40:59.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"10289adfb77a016872aa712e53d1be4f12bbfbe8","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-26 09:27:54.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/32/672632/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/a48cc3c/ : SUCCESS in 2h 44m 45s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/32/672632/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/28898a1/ : FAILURE in 28m 12s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/32/672632/8/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/f0f45a4/ : SUCCESS in 22m 43s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/32/672632/8/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/c74d26b/ : SUCCESS in 18m 43s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"da077a29f5a2cfca3698f414ad0384eeb759d77a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 09:47:13.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/79a628df0e0d4d4688aa1a2a295cce2a : SUCCESS in 8m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6454682acc9044e296e1c3dbabb1408c : SUCCESS in 5m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ba42f920b3bf435db0e0a8f1670a9c9a : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0842024379424973b6017395c1ec291c : SUCCESS in 4m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/aa490881e1a24b43ad03edc8e1898f71 : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b911a1b0ca3341e084cff38cf6b19b59 : SUCCESS in 5m 25s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/e67fe034e47f4bdc9d074e694a97d079 : SUCCESS in 2h 37m 11s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/7498701d7dca49fba2e5668996b28ed2 : SUCCESS in 1h 34m 37s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/254cf43d267248a4bff70336865ab5dd : SUCCESS in 2h 40m 31s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/a61f04593b1845e5a4e5dabe42d743f0 : SUCCESS in 1h 56m 09s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/742b122225344fa2ad95a9ac238c2446 : SUCCESS in 2h 04m 04s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/174bdaf822034c5cb20d097695839f40 : FAILURE in 2h 29m 54s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f3760137cbdb4a8398d3f90aa1492202 : SUCCESS in 4m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7cec27486c7a481db5a8627e405a019f : SUCCESS in 5m 39s","accounts_in_message":[],"_revision_number":8},{"id":"296f7c855d474654d8ee168654d7a1afcf392f59","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-08-27 05:52:23.000000000","message":"Patch Set 8:\n\nAfter I1bf74a4fb0adcae632c06b38e06a572ebc434201 this is not needed here anymore","accounts_in_message":[],"_revision_number":8},{"id":"8503b163a1bc38cb168b976978a1933c3f2bb06a","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-08-27 05:52:37.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"0900de2453da1f2ab062d1bf6fdbf4336fd6ed1d","revisions":{"28f1c775eeef2ea5da7abefb8b822dd422359739":{"kind":"REWORK","_number":1,"created":"2019-07-25 06:27:44.000000000","uploader":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"ref":"refs/changes/32/672632/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/32/672632/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/32/672632/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/32/672632/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/32/672632/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/32/672632/1"}}},"commit":{"parents":[{"commit":"8c427ecbc7dd382dcd1672e678be6a4fa27cae8a","subject":"Rename _short_bootstrap_node_ip to _bootstrap_node_ip","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/8c427ecbc7dd382dcd1672e678be6a4fa27cae8a"}]}],"author":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-07-25 06:27:10.000000000","tz":120},"committer":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-07-25 06:27:10.000000000","tz":120},"subject":"WIP Separate VIP for OVN DBs","message":"WIP Separate VIP for OVN DBs\n\nChange-Id: I3fd8465a718ac90fd23bd3e9fe99a4477c968c10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/28f1c775eeef2ea5da7abefb8b822dd422359739"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/28f1c775eeef2ea5da7abefb8b822dd422359739"}]},"branch":"refs/heads/master"},"901ab67e7fb1084fd888034c435cfa5641c19627":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-07-25 09:23:08.000000000","uploader":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"ref":"refs/changes/32/672632/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/32/672632/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/32/672632/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/32/672632/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/32/672632/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/32/672632/2"}}},"commit":{"parents":[{"commit":"8c427ecbc7dd382dcd1672e678be6a4fa27cae8a","subject":"Rename _short_bootstrap_node_ip to _bootstrap_node_ip","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/8c427ecbc7dd382dcd1672e678be6a4fa27cae8a"}]}],"author":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-07-25 06:27:10.000000000","tz":120},"committer":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-07-25 09:21:47.000000000","tz":120},"subject":"WIP Separate VIP for OVN DBs","message":"WIP Separate VIP for OVN DBs\n\nDepends-On: I620e37117c26b5b51bf9e1eda91daeb00fdf0f43\nChange-Id: I3fd8465a718ac90fd23bd3e9fe99a4477c968c10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/901ab67e7fb1084fd888034c435cfa5641c19627"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/901ab67e7fb1084fd888034c435cfa5641c19627"}]},"branch":"refs/heads/master"},"9432b12a233f7113df64877572b9c7aa2f83d3f5":{"kind":"REWORK","_number":3,"created":"2019-07-27 12:08:40.000000000","uploader":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"ref":"refs/changes/32/672632/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/32/672632/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/32/672632/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/32/672632/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/32/672632/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/32/672632/3"}}},"commit":{"parents":[{"commit":"70c7bb233411318bfad4f7ab715f35f6b637397d","subject":"Merge \"Move redis_vip to all_nodes.j2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/70c7bb233411318bfad4f7ab715f35f6b637397d"}]}],"author":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-07-25 06:27:10.000000000","tz":120},"committer":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-07-27 12:07:58.000000000","tz":120},"subject":"WIP Separate VIP for OVN DBs","message":"WIP Separate VIP for OVN DBs\n\nDepends-On: I620e37117c26b5b51bf9e1eda91daeb00fdf0f43\nChange-Id: I3fd8465a718ac90fd23bd3e9fe99a4477c968c10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/9432b12a233f7113df64877572b9c7aa2f83d3f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/9432b12a233f7113df64877572b9c7aa2f83d3f5"}]},"branch":"refs/heads/master"},"3041b26753003e216cda538af1b4e3128c8b198f":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-07-29 15:06:16.000000000","uploader":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"ref":"refs/changes/32/672632/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/32/672632/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/32/672632/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/32/672632/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/32/672632/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/32/672632/4"}}},"commit":{"parents":[{"commit":"3784506054df93a097a8970ee5472f53a4e531aa","subject":"Merge \"Properly close web request\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/3784506054df93a097a8970ee5472f53a4e531aa"}]}],"author":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-07-25 06:27:10.000000000","tz":120},"committer":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-07-29 15:06:16.000000000","tz":0},"subject":"WIP Separate VIP for OVN DBs","message":"WIP Separate VIP for OVN DBs\n\nDepends-On: I620e37117c26b5b51bf9e1eda91daeb00fdf0f43\nChange-Id: I3fd8465a718ac90fd23bd3e9fe99a4477c968c10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/3041b26753003e216cda538af1b4e3128c8b198f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/3041b26753003e216cda538af1b4e3128c8b198f"}]},"branch":"refs/heads/master"},"0644835ff2a7397bbbaf8cf4861e1b17ad9c01f3":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-08-02 08:05:10.000000000","uploader":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"ref":"refs/changes/32/672632/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/32/672632/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/32/672632/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/32/672632/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/32/672632/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/32/672632/5"}}},"commit":{"parents":[{"commit":"9a1bdd8cf64e74d2303bd9afd6731cd0ad8c2fe6","subject":"Merge \"Add docker/podman packages to nova_libvirt container\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/9a1bdd8cf64e74d2303bd9afd6731cd0ad8c2fe6"}]}],"author":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-07-25 06:27:10.000000000","tz":120},"committer":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-08-02 08:05:10.000000000","tz":0},"subject":"WIP Separate VIP for OVN DBs","message":"WIP Separate VIP for OVN DBs\n\nDepends-On: I620e37117c26b5b51bf9e1eda91daeb00fdf0f43\nChange-Id: I3fd8465a718ac90fd23bd3e9fe99a4477c968c10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/0644835ff2a7397bbbaf8cf4861e1b17ad9c01f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/0644835ff2a7397bbbaf8cf4861e1b17ad9c01f3"}]},"branch":"refs/heads/master"},"180480cd4dcc21ef28befa474bc0f90d67f44fb0":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-08-06 08:09:48.000000000","uploader":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"ref":"refs/changes/32/672632/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/32/672632/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/32/672632/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/32/672632/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/32/672632/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/32/672632/6"}}},"commit":{"parents":[{"commit":"a1bd494fc79d3ad06c53a9c324ce293ce2803b15","subject":"Merge \"Add service_configs template to tripleo-hieradata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/a1bd494fc79d3ad06c53a9c324ce293ce2803b15"}]}],"author":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-07-25 06:27:10.000000000","tz":120},"committer":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-08-06 08:09:48.000000000","tz":0},"subject":"WIP Separate VIP for OVN DBs","message":"WIP Separate VIP for OVN DBs\n\nDepends-On: I620e37117c26b5b51bf9e1eda91daeb00fdf0f43\nChange-Id: I3fd8465a718ac90fd23bd3e9fe99a4477c968c10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/180480cd4dcc21ef28befa474bc0f90d67f44fb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/180480cd4dcc21ef28befa474bc0f90d67f44fb0"}]},"branch":"refs/heads/master"},"0d868f978aa82aecae3991a780e6a1b9ea5bfb42":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-08-09 04:00:14.000000000","uploader":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"ref":"refs/changes/32/672632/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/32/672632/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/32/672632/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/32/672632/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/32/672632/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/32/672632/7"}}},"commit":{"parents":[{"commit":"960e6da0dc8b7702484e4607298715356c55641d","subject":"Merge \"Add extraconfig/role_extraconfig template to hieradata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/960e6da0dc8b7702484e4607298715356c55641d"}]}],"author":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-07-25 06:27:10.000000000","tz":120},"committer":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-08-09 04:00:14.000000000","tz":0},"subject":"WIP Separate VIP for OVN DBs","message":"WIP Separate VIP for OVN DBs\n\nDepends-On: I620e37117c26b5b51bf9e1eda91daeb00fdf0f43\nChange-Id: I3fd8465a718ac90fd23bd3e9fe99a4477c968c10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/0d868f978aa82aecae3991a780e6a1b9ea5bfb42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/0d868f978aa82aecae3991a780e6a1b9ea5bfb42"}]},"branch":"refs/heads/master"},"0900de2453da1f2ab062d1bf6fdbf4336fd6ed1d":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2019-08-26 06:40:59.000000000","uploader":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"ref":"refs/changes/32/672632/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/32/672632/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/32/672632/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/32/672632/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/32/672632/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/32/672632/8"}}},"commit":{"parents":[{"commit":"29b7c8a1ac13cf2df5ab84e8acfd14b5ed92b370","subject":"Merge \"Remove disable_constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/29b7c8a1ac13cf2df5ab84e8acfd14b5ed92b370"}]}],"author":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-07-25 06:27:10.000000000","tz":120},"committer":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-08-26 06:40:59.000000000","tz":0},"subject":"WIP Separate VIP for OVN DBs","message":"WIP Separate VIP for OVN DBs\n\nDepends-On: I620e37117c26b5b51bf9e1eda91daeb00fdf0f43\nChange-Id: I3fd8465a718ac90fd23bd3e9fe99a4477c968c10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/0900de2453da1f2ab062d1bf6fdbf4336fd6ed1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/0900de2453da1f2ab062d1bf6fdbf4336fd6ed1d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
