)]}'
{"id":"openstack%2Ftripleo-common~673366","triplet_id":"openstack%2Ftripleo-common~master~I1bf74a4fb0adcae632c06b38e06a572ebc434201","project":"openstack/tripleo-common","branch":"master","topic":"hieraroles","hashtags":[],"change_id":"I1bf74a4fb0adcae632c06b38e06a572ebc434201","subject":"Remove hiera roles","status":"MERGED","created":"2019-07-29 21:12:04.000000000","updated":"2019-08-27 02:49:00.000000000","submitted":"2019-08-27 02:47:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"673366-1566874040113-80306961","meta_rev_id":"3b87a4d176f952577dc45196b4744ede9edecdf1","_number":673366,"virtual_id_number":673366,"owner":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":2,"date":"2019-08-27 02:47:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-26 21:39:05.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"all":[{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":2,"date":"2019-08-26 18:09:42.000000000","_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":2,"date":"2019-08-26 18:09:42.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":1,"date":"2019-08-26 18:09:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":1,"date":"2019-08-26 21:58:12.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-05 14:49:16.000000000","updated_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"reviewer":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"state":"REVIEWER"},{"updated":"2019-08-14 14:59:44.000000000","updated_by":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"reviewer":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"state":"REVIEWER"},{"updated":"2019-08-15 18:54:19.000000000","updated_by":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"reviewer":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"state":"REVIEWER"},{"updated":"2019-08-22 07:07:17.000000000","updated_by":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"reviewer":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"state":"REVIEWER"},{"updated":"2019-08-26 16:44:51.000000000","updated_by":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"reviewer":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"state":"REVIEWER"},{"updated":"2019-08-26 21:39:05.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-26 21:58:12.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-08-27 02:47:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c1e28f614acba62c33bd18c337f350f0ff2525be","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-07-29 21:12:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"60193305fcd22838fddd48e3782426a029b85d2b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 21:15:39.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"b07bd8c8707264482145b9b99128b5df459e012f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-29 21:15:39.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"d66e937bc319bdbc67d4f666acf8138c466c09b8","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-07-29 21:15:59.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2b94e31f7ab204a27fd97fd99452a6302e71992c","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-07-29 21:16:04.000000000","message":"Patch Set 1:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":1},{"id":"9b1ba72cb7d82881bfe76a034645e2e30801bbfb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 21:16:54.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"0602ee58705067a79b4d848050c70e082e96b198","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-29 21:16:55.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"31990a9a5869b1e516504203bf31fac4983d57b1","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-07-29 21:17:08.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"47476f4ee2b25a78f08d6e73f91fd966d0a767c2","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-07-29 21:17:12.000000000","message":"Patch Set 1:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":1},{"id":"a415c404b0b677b062871ea9c4b5a9ffea8bd639","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-07-29 21:37:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ac7794b810e5a1d60e9217ed89cec3066b496086","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-29 22:06:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/66/673366/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/da91dfc/ : FAILURE in 27m 06s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/66/673366/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/7d50e67/ : FAILURE in 27m 11s","accounts_in_message":[],"_revision_number":2},{"id":"e93fdc253dfcda4cf86e069ab9cb48dfc5617b31","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 22:06:31.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"f7c11bd6d3fe7de33bf88af66ada98d78f3b5b11","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-07-29 22:14:26.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d8b602a1eb032c3d2ccae77b96a7d00c616c891f","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-07-29 22:14:29.000000000","message":"Patch Set 2:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":2},{"id":"7980321f5f3bf008ee9020a03d3f5308e823432d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-29 23:03:41.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/66/673366/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/630ca1a/ : FAILURE in 47m 27s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/66/673366/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/0cd480f/ : FAILURE in 42m 11s","accounts_in_message":[],"_revision_number":2},{"id":"e97ef34ea26f6b1983a64934ceeea09b3318e9cf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 00:31:57.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/66/673366/2/check/test-release-openstack/ad59cf0/ : SUCCESS in 4m 09s\n- openstack-tox-cover http://logs.openstack.org/66/673366/2/check/openstack-tox-cover/c8cd52c/cover/ : SUCCESS in 4m 54s\n- openstack-tox-pep8 http://logs.openstack.org/66/673366/2/check/openstack-tox-pep8/38d2802/ : SUCCESS in 4m 08s\n- openstack-tox-py27 http://logs.openstack.org/66/673366/2/check/openstack-tox-py27/1ace4a3/ : SUCCESS in 3m 36s\n- openstack-tox-py36 http://logs.openstack.org/66/673366/2/check/openstack-tox-py36/5e5ff3b/ : SUCCESS in 3m 22s\n- openstack-tox-py37 http://logs.openstack.org/66/673366/2/check/openstack-tox-py37/50dd833/ : SUCCESS in 4m 46s\n- openstack-tox-docs http://logs.openstack.org/66/673366/2/check/openstack-tox-docs/0165ab0/html/ : SUCCESS in 4m 14s\n- build-openstack-releasenotes http://logs.openstack.org/66/673366/2/check/build-openstack-releasenotes/c56b4bb/html/ : SUCCESS in 3m 42s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/66/673366/2/check/tripleo-ci-centos-7-containers-multinode/ea762f9/ : FAILURE in 54m 50s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/66/673366/2/check/tripleo-ci-centos-7-undercloud-containers/93fba09/ : FAILURE in 55m 43s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/66/673366/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/e2bd01c/ : FAILURE in 1h 02m 19s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/66/673366/2/check/tripleo-ci-centos-7-standalone/894b44d/ : FAILURE in 46m 49s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/66/673366/2/check/tripleo-ci-fedora-28-standalone/2659771/ : FAILURE in 53m 38s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/66/673366/2/check/tripleo-ci-centos-7-standalone-upgrade/9ef23ee/ : FAILURE in 1h 28m 12s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/66/673366/2/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/a933aeb/ : FAILURE in 1h 37m 14s (non-voting)\n- tripleo-build-containers-centos-7 http://logs.openstack.org/66/673366/2/check/tripleo-build-containers-centos-7/fe1cbab/ : SUCCESS in 48m 53s\n- tripleo-build-containers-centos-7-buildah http://logs.openstack.org/66/673366/2/check/tripleo-build-containers-centos-7-buildah/6020987/ : SUCCESS in 49m 43s\n- openstack-tox-linters http://logs.openstack.org/66/673366/2/check/openstack-tox-linters/f6b11a2/ : SUCCESS in 5m 50s\n- openstack-tox-py35 http://logs.openstack.org/66/673366/2/check/openstack-tox-py35/76be1c8/ : SUCCESS in 8m 14s\n- openstack-tox-lower-constraints http://logs.openstack.org/66/673366/2/check/openstack-tox-lower-constraints/edcafeb/ : SUCCESS in 5m 53s","accounts_in_message":[],"_revision_number":2},{"id":"276bad910c565694ec679cb33ea8da1333339944","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-07-30 21:14:18.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"2c2965fd46134d97c8fd02504be574d2cade34c0","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-07-30 21:14:25.000000000","message":"Patch Set 2:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":2},{"id":"e5e817e2f74cc839d520767caaf161f27c719885","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-30 22:10:32.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/66/673366/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/f4fb51d/ : FAILURE in 37m 04s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/66/673366/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/af180b8/ : FAILURE in 49m 11s","accounts_in_message":[],"_revision_number":2},{"id":"2261e36a28ede0a82254fca83c4368a684613a47","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 22:53:45.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"b543c52cbad322f6b72585bc618dfab9ae2bf17e","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-07-31 13:23:48.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"7fec9f9eac814082b1e637762c9bf76a88fab80d","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-07-31 13:23:52.000000000","message":"Patch Set 2:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":2},{"id":"c029afcf2765c6e0f8a8fe8beb0222bcca7f5ddd","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-31 15:29:04.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/66/673366/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/6c75416/ : FAILURE in 2h 03m 12s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/66/673366/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/3420d9b/ : FAILURE in 1h 54m 48s","accounts_in_message":[],"_revision_number":2},{"id":"9047848c9d3f3268ccd44581a00e171376a69059","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 19:12:01.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"497344bf7a408045cd89cddb24702ae1c97953cb","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-07-31 19:12:10.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"dd8194144c1a7d608354afc068e46a354fda3007","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-07-31 19:12:14.000000000","message":"Patch Set 2:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":2},{"id":"5d2c76637faea72b0b6da288f3d67a6995eadcd2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-31 21:11:22.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/66/673366/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/375e303/ : FAILURE in 1h 48m 51s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/66/673366/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/3037ee2/ : FAILURE in 1h 55m 03s","accounts_in_message":[],"_revision_number":2},{"id":"0f8fbec4564d6c5cf27b8a25e946d265de5cdefd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 21:51:06.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"a0db19884e09011db702bf4091dcfe1bdfa0fcac","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-08-01 00:25:53.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"00228f109559a190f709d177c5cb3797e6171322","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-08-01 00:25:56.000000000","message":"Patch Set 2:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":2},{"id":"32ab887a8074ca100ea2198d83bf96e4d5157e7c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-01 03:11:59.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/66/673366/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/22dd72b/ : FAILURE in 1h 50m 50s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/66/673366/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/cb05180/ : SUCCESS in 2h 45m 07s","accounts_in_message":[],"_revision_number":2},{"id":"aa17af124697ba5d2352f90490b051f9f13b7640","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 03:24:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://logs.opendev.org/66/673366/2/check/test-release-openstack/f474b5d/ : SUCCESS in 2m 21s\n- openstack-tox-cover https://logs.opendev.org/66/673366/2/check/openstack-tox-cover/dd0151c/cover/ : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://logs.opendev.org/66/673366/2/check/openstack-tox-pep8/d968a19/ : SUCCESS in 2m 57s\n- openstack-tox-py27 https://logs.opendev.org/66/673366/2/check/openstack-tox-py27/d815fb6/ : SUCCESS in 4m 23s\n- openstack-tox-py36 https://logs.opendev.org/66/673366/2/check/openstack-tox-py36/5041c10/ : SUCCESS in 3m 49s\n- openstack-tox-py37 https://logs.opendev.org/66/673366/2/check/openstack-tox-py37/1b466a5/ : SUCCESS in 3m 57s\n- openstack-tox-docs https://logs.opendev.org/66/673366/2/check/openstack-tox-docs/b6af37c/html/ : SUCCESS in 3m 08s\n- build-openstack-releasenotes https://logs.opendev.org/66/673366/2/check/build-openstack-releasenotes/e867360/html/ : SUCCESS in 3m 24s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/66/673366/2/check/tripleo-ci-centos-7-containers-multinode/7da1bd1/ : SUCCESS in 2h 24m 43s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/66/673366/2/check/tripleo-ci-centos-7-undercloud-containers/fb32e57/ : SUCCESS in 1h 40m 03s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/66/673366/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/24bf342/ : SUCCESS in 2h 43m 13s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/66/673366/2/check/tripleo-ci-centos-7-standalone/a346922/ : SUCCESS in 1h 56m 06s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/66/673366/2/check/tripleo-ci-fedora-28-standalone/b69c2b4/ : FAILURE in 59m 15s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/66/673366/2/check/tripleo-ci-centos-7-standalone-upgrade/c240635/ : FAILURE in 1h 32m 03s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/66/673366/2/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/5c3a258/ : FAILURE in 1h 53m 14s (non-voting)\n- tripleo-build-containers-centos-7 https://logs.opendev.org/66/673366/2/check/tripleo-build-containers-centos-7/a2284ad/ : SUCCESS in 40m 28s\n- tripleo-build-containers-centos-7-buildah https://logs.opendev.org/66/673366/2/check/tripleo-build-containers-centos-7-buildah/60abdd4/ : SUCCESS in 55m 10s\n- openstack-tox-linters https://logs.opendev.org/66/673366/2/check/openstack-tox-linters/23289ff/ : SUCCESS in 3m 58s\n- openstack-tox-py35 https://logs.opendev.org/66/673366/2/check/openstack-tox-py35/667e044/ : SUCCESS in 3m 43s\n- openstack-tox-lower-constraints https://logs.opendev.org/66/673366/2/check/openstack-tox-lower-constraints/3b2e8d9/ : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":2},{"id":"42920b4c70d2c0912160b2a2e4199a73a1c8ae5f","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-05 14:49:16.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"28a928bc7ce31e900acfe093dd27944406afcf93","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-08-07 12:32:17.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"efa67ef020dd2286d00ddef1a981f8efe55e4b98","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-07 15:40:53.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/66/673366/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/3781e21/ : SUCCESS in 3h 06m 32s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/66/673366/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/4089d00/ : FAILURE in 3h 03m 28s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/66/673366/3/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/d7da902/ : FAILURE in 9m 46s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/66/673366/3/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/efd2ec8/ : FAILURE in 9m 28s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a1ecae083a304edb01dc8390037c8cf736c066ac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 19:12:02.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://logs.opendev.org/66/673366/3/check/test-release-openstack/171610a/ : SUCCESS in 2m 33s\n- openstack-tox-cover https://logs.opendev.org/66/673366/3/check/openstack-tox-cover/52e96e2/cover/ : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://logs.opendev.org/66/673366/3/check/openstack-tox-pep8/231ab8a/ : SUCCESS in 2m 28s\n- openstack-tox-py27 https://logs.opendev.org/66/673366/3/check/openstack-tox-py27/83b1db7/ : SUCCESS in 3m 57s\n- openstack-tox-py36 https://logs.opendev.org/66/673366/3/check/openstack-tox-py36/68ad6e2/ : SUCCESS in 3m 34s\n- openstack-tox-py37 https://logs.opendev.org/66/673366/3/check/openstack-tox-py37/c2d408c/ : SUCCESS in 5m 07s\n- openstack-tox-docs https://logs.opendev.org/66/673366/3/check/openstack-tox-docs/9683297/html/ : SUCCESS in 4m 52s\n- build-openstack-releasenotes https://logs.opendev.org/66/673366/3/check/build-openstack-releasenotes/fddacfb/html/ : SUCCESS in 3m 30s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/66/673366/3/check/tripleo-ci-centos-7-containers-multinode/5486514/ : SUCCESS in 2h 40m 12s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/66/673366/3/check/tripleo-ci-centos-7-undercloud-containers/33ff4a3/ : SUCCESS in 1h 43m 06s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/66/673366/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/7e93220/ : SUCCESS in 2h 48m 22s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/66/673366/3/check/tripleo-ci-centos-7-standalone/433f831/ : FAILURE in 55m 44s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/66/673366/3/check/tripleo-ci-fedora-28-standalone/b688859/ : SUCCESS in 1h 44m 08s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/66/673366/3/check/tripleo-ci-centos-7-standalone-upgrade/ee4f7b1/ : FAILURE in 1h 26m 28s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/66/673366/3/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/b86e2e3/ : FAILURE in 2h 02m 20s (non-voting)\n- tripleo-build-containers-centos-7 https://logs.opendev.org/66/673366/3/check/tripleo-build-containers-centos-7/a61927f/ : SUCCESS in 1h 02m 56s\n- tripleo-build-containers-centos-7-buildah https://logs.opendev.org/66/673366/3/check/tripleo-build-containers-centos-7-buildah/ca46f11/ : SUCCESS in 48m 13s\n- openstack-tox-linters https://logs.opendev.org/66/673366/3/check/openstack-tox-linters/a64d5a8/ : SUCCESS in 4m 58s\n- openstack-tox-py35 https://logs.opendev.org/66/673366/3/check/openstack-tox-py35/61c4b08/ : SUCCESS in 4m 34s\n- openstack-tox-lower-constraints https://logs.opendev.org/66/673366/3/check/openstack-tox-lower-constraints/eca9ece/ : SUCCESS in 6m 24s","accounts_in_message":[],"_revision_number":3},{"id":"bda9656961dcdc422195f0e1eb59a9453ad39440","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-08-08 04:33:36.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"bfb5b93c5d47ca748f5750d3a349e2cafbb273c4","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-08 05:01:43.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/66/673366/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/efc5ace/ : FAILURE in 18m 22s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/66/673366/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/1e63852/ : FAILURE in 26m 46s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/66/673366/4/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/b4e7919/ : FAILURE in 8m 42s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/66/673366/4/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/57536eb/ : FAILURE in 8m 31s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5cc310d5709eae023b20c1af1f5931deed5bfc18","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 06:34:03.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://logs.opendev.org/66/673366/4/check/test-release-openstack/23b234c/ : SUCCESS in 3m 15s\n- openstack-tox-cover https://logs.opendev.org/66/673366/4/check/openstack-tox-cover/24692f1/cover/ : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://logs.opendev.org/66/673366/4/check/openstack-tox-pep8/2f40e0e/ : SUCCESS in 4m 50s\n- openstack-tox-py27 https://logs.opendev.org/66/673366/4/check/openstack-tox-py27/f03837f/ : SUCCESS in 4m 18s\n- openstack-tox-py36 https://logs.opendev.org/66/673366/4/check/openstack-tox-py36/53ed800/ : SUCCESS in 3m 31s\n- openstack-tox-py37 https://logs.opendev.org/66/673366/4/check/openstack-tox-py37/d712524/ : SUCCESS in 5m 55s\n- openstack-tox-docs https://logs.opendev.org/66/673366/4/check/openstack-tox-docs/18cba1d/html/ : SUCCESS in 3m 47s\n- build-openstack-releasenotes https://logs.opendev.org/66/673366/4/check/build-openstack-releasenotes/51f4f45/html/ : SUCCESS in 4m 10s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/66/673366/4/check/tripleo-ci-centos-7-containers-multinode/fdbac2c/ : FAILURE in 38m 53s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/66/673366/4/check/tripleo-ci-centos-7-undercloud-containers/3dd0908/ : FAILURE in 35m 14s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/66/673366/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/d91dd22/ : FAILURE in 40m 44s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/66/673366/4/check/tripleo-ci-centos-7-standalone/ca7a38a/ : FAILURE in 33m 35s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/66/673366/4/check/tripleo-ci-fedora-28-standalone/5c46b52/ : FAILURE in 36m 49s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/66/673366/4/check/tripleo-ci-centos-7-standalone-upgrade/645083b/ : FAILURE in 1h 30m 35s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/66/673366/4/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/6b2d23e/ : FAILURE in 1h 36m 17s (non-voting)\n- tripleo-build-containers-centos-7 https://logs.opendev.org/66/673366/4/check/tripleo-build-containers-centos-7/56bd7cb/ : SUCCESS in 43m 00s\n- tripleo-build-containers-centos-7-buildah https://logs.opendev.org/66/673366/4/check/tripleo-build-containers-centos-7-buildah/2ade027/ : SUCCESS in 51m 36s\n- openstack-tox-linters https://logs.opendev.org/66/673366/4/check/openstack-tox-linters/467e997/ : SUCCESS in 5m 31s\n- openstack-tox-py35 https://logs.opendev.org/66/673366/4/check/openstack-tox-py35/f17cc75/ : SUCCESS in 3m 28s\n- openstack-tox-lower-constraints https://logs.opendev.org/66/673366/4/check/openstack-tox-lower-constraints/2491711/ : SUCCESS in 6m 07s","accounts_in_message":[],"_revision_number":4},{"id":"3fe7c626f8ed00fe72cb7ea32833a5c4cc45531c","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-08-08 13:56:03.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"042f47f275347c7a9fe2429e0040d5c8b6f97276","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-08 14:29:40.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 finger://ze05.softwarefactory-project.io:7979/42c228df3acd49d1b108ca34d44fb9e1 : RETRY_LIMIT in 7s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/66/673366/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/6d1cca1/ : FAILURE in 19m 25s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/66/673366/5/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/b6768c5/ : FAILURE in 9m 09s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/66/673366/5/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/5949aa4/ : FAILURE in 9m 27s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"05c9818dd01b76cf32f926a5c11646b6bfb52d11","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-08-08 15:05:47.000000000","message":"Patch Set 5:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":5},{"id":"c581896947d0bc0a7224d7c507e1d4cd33253fe6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-08 15:49:24.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/66/673366/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/868c652/ : FAILURE in 20m 36s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/66/673366/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/66ef49d/ : FAILURE in 22m 56s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/66/673366/5/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/e293595/ : FAILURE in 9m 00s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent tripleo-rhel-8-buildimage-ironic-python-agent : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"44cce8434b0beee3433e5e872dd631770c9bd26e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 19:55:13.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://logs.opendev.org/66/673366/5/check/test-release-openstack/3852c59/ : SUCCESS in 2m 31s\n- openstack-tox-cover https://logs.opendev.org/66/673366/5/check/openstack-tox-cover/66cc21e/cover/ : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://logs.opendev.org/66/673366/5/check/openstack-tox-pep8/200b32e/ : SUCCESS in 2m 53s\n- openstack-tox-py27 https://logs.opendev.org/66/673366/5/check/openstack-tox-py27/106d596/ : SUCCESS in 5m 54s\n- openstack-tox-py36 https://logs.opendev.org/66/673366/5/check/openstack-tox-py36/6e8a42a/ : SUCCESS in 4m 03s\n- openstack-tox-py37 https://logs.opendev.org/66/673366/5/check/openstack-tox-py37/cfbcf6d/ : SUCCESS in 4m 29s\n- openstack-tox-docs https://logs.opendev.org/66/673366/5/check/openstack-tox-docs/d02db53/html/ : SUCCESS in 4m 00s\n- build-openstack-releasenotes https://logs.opendev.org/66/673366/5/check/build-openstack-releasenotes/f9d7e6f/html/ : SUCCESS in 4m 40s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/66/673366/5/check/tripleo-ci-centos-7-containers-multinode/68aee17/ : FAILURE in 28m 27s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/66/673366/5/check/tripleo-ci-centos-7-undercloud-containers/5cc3a8b/ : FAILURE in 25m 25s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/66/673366/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/51100cf/ : FAILURE in 28m 44s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/66/673366/5/check/tripleo-ci-centos-7-standalone/854af52/ : FAILURE in 23m 17s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/66/673366/5/check/tripleo-ci-fedora-28-standalone/7b9da1a/ : FAILURE in 25m 29s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/66/673366/5/check/tripleo-ci-centos-7-standalone-upgrade/fbc2283/ : FAILURE in 1h 34m 25s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/66/673366/5/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/1c86860/ : FAILURE in 1h 17m 21s (non-voting)\n- tripleo-build-containers-centos-7 https://logs.opendev.org/66/673366/5/check/tripleo-build-containers-centos-7/9f05a02/ : SUCCESS in 49m 18s\n- tripleo-build-containers-centos-7-buildah https://logs.opendev.org/66/673366/5/check/tripleo-build-containers-centos-7-buildah/135c9c9/ : SUCCESS in 1h 02m 23s\n- openstack-tox-linters https://logs.opendev.org/66/673366/5/check/openstack-tox-linters/14693c0/ : SUCCESS in 4m 09s\n- openstack-tox-py35 https://logs.opendev.org/66/673366/5/check/openstack-tox-py35/f32a84b/ : SUCCESS in 4m 51s\n- openstack-tox-lower-constraints https://logs.opendev.org/66/673366/5/check/openstack-tox-lower-constraints/fc981e3/ : SUCCESS in 6m 20s","accounts_in_message":[],"_revision_number":5},{"id":"bba974626c2805152b831d99e219a3cc4d336d13","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-08-08 20:54:38.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"807a484d9f1ff0dedf4d80b4f6aee9c1e3e3dc85","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-08-08 20:54:42.000000000","message":"Patch Set 5:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":5},{"id":"ccc8c4f716646e314a2c682a3ed4fe50647917fc","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-08 21:16:37.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/66/673366/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/9fcaa64/ : FAILURE in 18m 01s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/66/673366/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/3b018d1/ : FAILURE in 17m 20s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/66/673366/5/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/6a77a62/ : SUCCESS in 20m 45s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/66/673366/5/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/2385335/ : SUCCESS in 16m 46s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"99ddd71f4b13b59136c030a3e4d23a0eae056fd5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 00:16:07.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://logs.opendev.org/66/673366/5/check/test-release-openstack/5082a7a/ : SUCCESS in 3m 09s\n- openstack-tox-cover https://logs.opendev.org/66/673366/5/check/openstack-tox-cover/3f89aa4/cover/ : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://logs.opendev.org/66/673366/5/check/openstack-tox-pep8/dc8b817/ : SUCCESS in 2m 41s\n- openstack-tox-py27 https://logs.opendev.org/66/673366/5/check/openstack-tox-py27/03fae9e/ : SUCCESS in 5m 17s\n- openstack-tox-py36 https://logs.opendev.org/66/673366/5/check/openstack-tox-py36/76dc473/ : SUCCESS in 5m 13s\n- openstack-tox-py37 https://logs.opendev.org/66/673366/5/check/openstack-tox-py37/527f5ed/ : SUCCESS in 5m 09s\n- openstack-tox-docs https://logs.opendev.org/66/673366/5/check/openstack-tox-docs/b639553/html/ : SUCCESS in 4m 14s\n- build-openstack-releasenotes https://logs.opendev.org/66/673366/5/check/build-openstack-releasenotes/afeedc5/html/ : SUCCESS in 5m 04s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/66/673366/5/check/tripleo-ci-centos-7-containers-multinode/1c75cd8/ : FAILURE in 31m 48s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/66/673366/5/check/tripleo-ci-centos-7-undercloud-containers/443670b/ : FAILURE in 24m 33s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/66/673366/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/0d6c140/ : FAILURE in 26m 05s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/66/673366/5/check/tripleo-ci-centos-7-standalone/0b5d1ea/ : FAILURE in 29m 18s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/66/673366/5/check/tripleo-ci-fedora-28-standalone/81a719e/ : FAILURE in 25m 14s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/66/673366/5/check/tripleo-ci-centos-7-standalone-upgrade/a8ec291/ : FAILURE in 1h 18m 40s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/66/673366/5/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/7f93bcf/ : FAILURE in 1h 35m 56s (non-voting)\n- tripleo-build-containers-centos-7 https://logs.opendev.org/66/673366/5/check/tripleo-build-containers-centos-7/4218788/ : SUCCESS in 52m 25s\n- tripleo-build-containers-centos-7-buildah https://logs.opendev.org/66/673366/5/check/tripleo-build-containers-centos-7-buildah/9b8ece7/ : SUCCESS in 50m 56s\n- openstack-tox-linters https://logs.opendev.org/66/673366/5/check/openstack-tox-linters/18b19e1/ : SUCCESS in 5m 07s\n- openstack-tox-py35 https://logs.opendev.org/66/673366/5/check/openstack-tox-py35/b627ad5/ : SUCCESS in 5m 16s\n- openstack-tox-lower-constraints https://logs.opendev.org/66/673366/5/check/openstack-tox-lower-constraints/e817ea8/ : SUCCESS in 6m 32s","accounts_in_message":[],"_revision_number":5},{"id":"584389c6f8349682ff008adbff229748686ef8d3","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-08-09 02:09:01.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"bb568d2214b97026c9832c90eb16c0917c023fde","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 04:52:09.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://logs.opendev.org/66/673366/6/check/openstack-tox-cover/96466e0/cover/ : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://logs.opendev.org/66/673366/6/check/openstack-tox-pep8/b83359e/ : SUCCESS in 4m 16s\n- openstack-tox-py27 https://logs.opendev.org/66/673366/6/check/openstack-tox-py27/1ca7de3/ : SUCCESS in 3m 12s\n- openstack-tox-py36 https://logs.opendev.org/66/673366/6/check/openstack-tox-py36/2763f79/ : SUCCESS in 4m 10s\n- openstack-tox-py37 https://logs.opendev.org/66/673366/6/check/openstack-tox-py37/6b2e8e4/ : SUCCESS in 4m 16s\n- openstack-tox-docs https://logs.opendev.org/66/673366/6/check/openstack-tox-docs/bfb556c/html/ : SUCCESS in 4m 46s\n- build-openstack-releasenotes https://logs.opendev.org/66/673366/6/check/build-openstack-releasenotes/d9fc91e/html/ : SUCCESS in 3m 27s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/66/673366/6/check/tripleo-ci-centos-7-containers-multinode/b098163/ : SUCCESS in 2h 18m 04s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/66/673366/6/check/tripleo-ci-centos-7-undercloud-containers/eedb4b2/ : SUCCESS in 1h 48m 25s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/66/673366/6/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/bb0324e/ : SUCCESS in 2h 32m 10s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/66/673366/6/check/tripleo-ci-centos-7-standalone/589fd2a/ : SUCCESS in 1h 53m 38s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/66/673366/6/check/tripleo-ci-fedora-28-standalone/87f24be/ : SUCCESS in 1h 54m 11s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/66/673366/6/check/tripleo-ci-centos-7-standalone-upgrade/7f40fe4/ : FAILURE in 1h 30m 31s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/66/673366/6/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/67aa62d/ : FAILURE in 1h 47m 53s (non-voting)\n- tripleo-build-containers-centos-7 https://logs.opendev.org/66/673366/6/check/tripleo-build-containers-centos-7/c3cbb27/ : SUCCESS in 1h 00m 10s\n- tripleo-build-containers-centos-7-buildah https://logs.opendev.org/66/673366/6/check/tripleo-build-containers-centos-7-buildah/c38d37f/ : SUCCESS in 1h 01m 25s\n- openstack-tox-linters https://logs.opendev.org/66/673366/6/check/openstack-tox-linters/cd99f30/ : SUCCESS in 4m 54s\n- openstack-tox-py35 https://logs.opendev.org/66/673366/6/check/openstack-tox-py35/ea84917/ : SUCCESS in 3m 16s\n- openstack-tox-lower-constraints https://logs.opendev.org/66/673366/6/check/openstack-tox-lower-constraints/3870ada/ : SUCCESS in 6m 36s","accounts_in_message":[],"_revision_number":6},{"id":"37dd0308e40f252f4eb1e67cc8fba80bc449448f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-09 06:42:21.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/66/673366/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/21b0fd0/ : TIMED_OUT in 4h 32m 34s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/66/673366/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/251b695/ : SUCCESS in 2h 55m 23s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/66/673366/6/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/05c634d/ : SUCCESS in 19m 12s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/66/673366/6/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/93578c3/ : SUCCESS in 15m 45s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"92881a4b05bc7a4a3195b6ee9168cfe3ebaa0fb0","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-08-12 21:53:57.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"536804d718edfb860f0dc578d4bd17ea4a0d7cf0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-13 00:59:18.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/66/673366/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/683ceb7/ : SUCCESS in 3h 04m 31s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/66/673366/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/4b1b03a/ : SUCCESS in 2h 53m 30s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/66/673366/7/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/db144c9/ : SUCCESS in 20m 18s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/66/673366/7/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/a36817f/ : SUCCESS in 16m 53s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"c1dafca781012898878c3e2b2139731762f2ec2f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 01:25:35.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5960e7534d6546f999bdea01c46d76ff : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33ba92ed896b431484e5df1acd91c23d : SUCCESS in 2m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/af5a6f43a75d4e32a09242ca9296e808 : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b66c2aa5053490eaf42d69884a71018 : SUCCESS in 3m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/115c755cc16d4fc48be94d9ca91fe0ee : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a5efd290a5045d7895882d77af12d8d : SUCCESS in 4m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/28210044aa8b46f39e36a8abf6b26145 : SUCCESS in 4m 05s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/1b7192c0e21e45e694180afe6ced5e33 : SUCCESS in 2h 23m 45s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/1ad18c44f9e24f719821e2d0b60b690a : SUCCESS in 1h 36m 05s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/165735739a794ff1954972a647a9290b : SUCCESS in 3h 12m 10s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/7b07431277bb4bee8f3cdf034e6ed81e : SUCCESS in 1h 48m 03s\n- tripleo-ci-fedora-28-standalone https://zuul.opendev.org/t/openstack/build/b3929c99a51645aa8846c949a1feb86b : SUCCESS in 1h 47m 29s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/e7708502b373466286044d67110be267 : SUCCESS in 2h 10m 30s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/f79e3a23db974c36978b133322fc1117 : FAILURE in 1h 55m 07s (non-voting)\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/60b6bb4fcc1942b0b94d7966aaa11df5 : SUCCESS in 47m 50s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/0ecc4ea165a44ea993e374453c3336c6 : SUCCESS in 1h 03m 05s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/074309351e2f4e9c8d255232d65dbe5d : SUCCESS in 4m 48s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7309b10bbd8246b29aacebf7f2cc630c : SUCCESS in 3m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/81e75e28c8314953af3ec3ddd4c06e17 : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":7},{"id":"bc75819ca5a9af708c2253e0da3ab4777d309a80","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-08-14 14:29:59.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"eba6779b92fd88f6e37dc04fb3ef75a21d05b321","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2019-08-14 14:59:44.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"7e5e7db959b72d209fb6620d444e59379e003dfd","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-08-14 19:34:02.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"21578a9b83dd237fea3fa1a35f22b76811b0d405","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 22:23:15.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d9e62ea33484cdd96f41c211f684861 : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d777a47362d4afe904d09dc374de594 : SUCCESS in 3m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3ef4955232d940adb3c38acbbad7947c : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b43776855a224265aef3ec9c4714242e : SUCCESS in 5m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a8298d80f6f545109625fbed627ec61f : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b1a7f2e567648f9b41967eedaeb6476 : SUCCESS in 3m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/775fb9dabb694af8a37dda3cc1b628d5 : SUCCESS in 4m 56s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/5b93b49f3d9c4a3cab9910c121992121 : SUCCESS in 2h 29m 34s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/5bb21766d947416ebe4d2a1b893af711 : FAILURE in 1h 10m 43s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/70cbc3e8e04646c6bc09d62f4c7ee129 : FAILURE in 1h 11m 25s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/c16365d0a6d84931a83a741de13213cc : SUCCESS in 2h 05m 45s\n- tripleo-ci-fedora-28-standalone https://zuul.opendev.org/t/openstack/build/c6c4ae1e6fcc45298afc2857d61eb03d : SUCCESS in 1h 44m 52s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/308e72f778ac44dcb1e1a90da5b80033 : SUCCESS in 2h 23m 36s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/7725d055ba3b4618912b0a60406b2e7f : FAILURE in 2h 30m 00s (non-voting)\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/c20249f423074bf0bf543ea14896e939 : SUCCESS in 57m 16s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/60029188fcff4fedacf3c392ede55bf8 : SUCCESS in 1h 06m 49s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/57c5bf27a76a42d5949615f550be1231 : SUCCESS in 4m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/361bd704f92c4a7dbc1eabebbdbac501 : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":8},{"id":"5fe983bd35393e8972b38a0f3a14716d5ea4ef5e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-14 22:39:32.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/66/673366/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/f2c04fc/ : SUCCESS in 2h 49m 41s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/66/673366/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/7dcf7fc/ : SUCCESS in 3h 03m 50s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/66/673366/8/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/47ba5a9/ : SUCCESS in 19m 58s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/66/673366/8/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/28214c9/ : SUCCESS in 18m 53s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"16ce30670c3bb68ae084b76c6960a0321d350692","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-08-15 18:54:19.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"94a545f57176a0b9f5d778c04bddbbb9cd8d4e4f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 21:56:58.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6a3b1a1e2d69452295b59c84560f1864 : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba82d0d218574683b7827bad79337404 : SUCCESS in 3m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9a3ebbc6a71d4cdbb35fcc1e8c79a1c6 : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/65cec0ee0a3e44b494a84d68bd722c68 : SUCCESS in 5m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e0a1581e85864232867e7480eb1a519b : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70a4071849ad4146ad28f1f94110cb63 : SUCCESS in 3m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4309fbfe738a424f8847c6cf4c88b903 : SUCCESS in 3m 54s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/c397c99298504797befb99551589339b : SUCCESS in 2h 39m 27s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/8bc03095c72f4f929265d999c57fc8bd : SUCCESS in 1h 43m 28s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/75acf691bf5f4189b3182bbd17b46b00 : SUCCESS in 2h 34m 52s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/37bfd923a0394f2da2e38e7ed25edba6 : SUCCESS in 2h 01m 25s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/96308760ebfc42d3ab0aed6d11bcc8c1 : SUCCESS in 2h 33m 41s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/893ba88701344a2a9020ad067123ebe0 : FAILURE in 2h 19m 16s (non-voting)\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/208f239f1f414f6bbbcb99cba53ce254 : SUCCESS in 52m 53s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/06eadc69aad24a818bed042e2936e651 : SUCCESS in 1h 17m 46s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7c91f8e402d14b66b1d37e1fa7d3d7c4 : SUCCESS in 4m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9e83ae4c51ca4d2e8c6fc1471e93fca9 : SUCCESS in 6m 42s","accounts_in_message":[],"_revision_number":8},{"id":"bd4d237bc4a0f657d23af9295315c378eff6a969","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-16 00:33:07.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 finger://ze02.softwarefactory-project.io:7979/5127de88e7aa4b1eb263f19f347120cf : POST_FAILURE in 4h 15m 26s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 finger://ze05.softwarefactory-project.io:7979/7068d31453134acd82b879250b97efe5 : POST_FAILURE in 4h 01m 50s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/66/673366/8/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/a921155/ : SUCCESS in 21m 45s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/66/673366/8/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/262be4b/ : SUCCESS in 18m 11s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"42768b3c835cac95fdc15b9e6eba13e81f353a3a","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-20 15:37:44.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"80427ea04511a3c8d14d7a3fd83d3e6ddf9f8e1d","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2019-08-22 07:07:17.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"ecb4849cc6a820b88e5244b68821db83c1296d65","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2019-08-26 16:44:51.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1\n\ndon\u0027t see the setup.cfg change included. can be a followup patch","accounts_in_message":[],"_revision_number":8},{"id":"a1a2fd02591eafdbdb20e3b4f5820f5cd4be72fb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 16:45:17.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"77712bd19f136187fc72b0c9a3c6cbb773d6ef9f","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-08-26 18:09:42.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"5a0c599f911d53af0a0eaf47ebe12305e605219d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 21:20:35.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/506a58848fb5406da02fac5e870d8152 : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b86ecbb38c04da9a7e2e7e6e78de163 : SUCCESS in 3m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/52991624160d4299a89d31eee04f2f1c : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd0a8a1d55074b4aa47e33039859da25 : SUCCESS in 4m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/74e201254f674905ad72507515766020 : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a19818b4a0d4c27ac6b37a1af27de9a : SUCCESS in 4m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6d6efc14af02443b8ae28ce1eb1a6f4f : SUCCESS in 5m 06s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/f49c7d69f8d649ad97ff407489188711 : SUCCESS in 2h 42m 50s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/d94484bc7ac344d680ad06d7580bab4d : SUCCESS in 1h 54m 15s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/c6982aa240cd4615bf99b58832802144 : SUCCESS in 2h 39m 21s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/4ba1d11bc6e346cb89ce63370a395cce : SUCCESS in 2h 17m 20s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/e51f713e193a4606b04c9995ed53a9b2 : SUCCESS in 2h 27m 47s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/a4da4ba009bb4573b1ede71cecdfcbfb : FAILURE in 2h 11m 57s (non-voting)\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/203a89e833644024953718a46c668c67 : SUCCESS in 48m 29s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/cc807cec8a144f809036c4c6215fd523 : SUCCESS in 56m 09s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/dea3fb9d0d0d40f19e965f87fc66346a : SUCCESS in 4m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/77c0969a9763495aa479eb8830520b3b : SUCCESS in 5m 27s","accounts_in_message":[],"_revision_number":9},{"id":"71cffda825d8ea8944317e65a96a19c0c9b8d482","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-26 21:39:05.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/66/673366/9/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/9974455/ : SUCCESS in 3h 27m 46s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/66/673366/9/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/520f897/ : SUCCESS in 3h 10m 41s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/66/673366/9/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/d61661e/ : SUCCESS in 21m 40s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/66/673366/9/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/4709867/ : SUCCESS in 17m 52s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"393c7cecdb7316ca2895673b3d87eea98f64b370","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-26 21:58:12.000000000","message":"Patch Set 9: Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"83c99f074d5ccfe3d21f1a44b560132716c5dfd7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 21:58:32.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"42e5ecce4d5d25e3148959c40b7b52984f206247","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 02:47:19.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df9d7c7239494082b12b960b467945eb : SUCCESS in 4m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9e4be6dffad1488ea757558b089e5f76 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/75ead10c02f647d3a3731ae861e17a65 : SUCCESS in 5m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bf44d8c57da048a7bc193114ca2a9190 : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71249bf05a124732bbcb1151984647fe : SUCCESS in 6m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4032951518a44c2090852879ebf2a9da : SUCCESS in 4m 21s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/719d750195d6410ab99c12e46bc014a9 : SUCCESS in 2h 49m 37s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/6d71d5656b794bdcac136e905df7b1af : SUCCESS in 2h 12m 58s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/66fbf2bbe5ae4f42ae367a55b26a84ed : SUCCESS in 2h 37m 41s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/4b7deb054ff04c1abeb16296d351235f : SUCCESS in 1h 42m 24s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/392842cfb7a4497298a5890ff990375c : SUCCESS in 55m 06s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/a42dc2cc31da4892a41b919849de8454 : SUCCESS in 1h 08m 05s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9fafbff6e30d49a696af7d0a9ede2d95 : SUCCESS in 5m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b7dc3e5a02894477b7be245b9ebd5889 : SUCCESS in 6m 02s","accounts_in_message":[],"_revision_number":9},{"id":"5b9c41a876dfb7ff88c46086452d6151924a226c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 02:47:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"51dcb96cbe62b406ee3c04c6aca098c65209a998","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 02:49:00.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49fc084e7a174877ac82b2238eb30e25 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"920b2124d9884bdec039b3bc9c19557b1b6a23fd","revisions":{"284e027cbbbb7a341f9a2f43fa32f1b040c8df87":{"kind":"REWORK","_number":1,"created":"2019-07-29 21:12:04.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/66/673366/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/66/673366/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/66/673366/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/66/673366/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/66/673366/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/66/673366/1"}}},"commit":{"parents":[{"commit":"3784506054df93a097a8970ee5472f53a4e531aa","subject":"Merge \"Properly close web request\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/3784506054df93a097a8970ee5472f53a4e531aa"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2019-07-29 21:05:30.000000000","tz":-300},"committer":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2019-07-29 21:09:30.000000000","tz":-300},"subject":"Remove hiera roles","message":"Remove hiera roles\n\nThe hiera roles have been moved into tripleo-ansible and are no longer\nneeded within this repository. This change removes the roles that have\nbeen moved.\n\nBecause this change removes the last of the roles within this repo\nthe setup.cfg file has been modified to remove the no longer relevant\ninstall path(s).\n\nStory: 2006044\nTask: 34726\nStory: 2006044\nTask: 34730\nDepends-On: I7f9e993735a0347aac12f728393639d88c80ff0f\nChange-Id: I1bf74a4fb0adcae632c06b38e06a572ebc434201\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/284e027cbbbb7a341f9a2f43fa32f1b040c8df87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/284e027cbbbb7a341f9a2f43fa32f1b040c8df87"}]},"branch":"refs/heads/master"},"d18358bf8752d4f97b87feb410a9b9f790a58023":{"kind":"REWORK","_number":2,"created":"2019-07-29 21:37:29.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/66/673366/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/66/673366/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/66/673366/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/66/673366/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/66/673366/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/66/673366/2"}}},"commit":{"parents":[{"commit":"3784506054df93a097a8970ee5472f53a4e531aa","subject":"Merge \"Properly close web request\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/3784506054df93a097a8970ee5472f53a4e531aa"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2019-07-29 21:05:30.000000000","tz":-300},"committer":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2019-07-29 21:35:04.000000000","tz":-300},"subject":"Remove hiera roles","message":"Remove hiera roles\n\nThe hiera roles have been moved into tripleo-ansible and are no longer\nneeded within this repository. This change removes the roles that have\nbeen moved.\n\nBecause this change removes the last of the roles within this repo\nthe setup.cfg file has been modified to remove the no longer relevant\ninstall path(s), the molecule related config has been pulled out of\nthe tox.ini file, and the molecule related packages have been purged\nfrom the bindep.txt file.\n\nStory: 2006044\nTask: 34726\nStory: 2006044\nTask: 34730\nDepends-On: I7f9e993735a0347aac12f728393639d88c80ff0f\nChange-Id: I1bf74a4fb0adcae632c06b38e06a572ebc434201\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/d18358bf8752d4f97b87feb410a9b9f790a58023"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/d18358bf8752d4f97b87feb410a9b9f790a58023"}]},"branch":"refs/heads/master"},"10f3c1fa4bd7b3e9b42ca99fc3f27dc494a49b08":{"kind":"REWORK","_number":3,"created":"2019-08-07 12:32:17.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/66/673366/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/66/673366/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/66/673366/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/66/673366/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/66/673366/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/66/673366/3"}}},"commit":{"parents":[{"commit":"a3dc7aa00cd3e61cc66c12c678fa2a4dad8b2182","subject":"Restore AllNodesExtraMapData","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/a3dc7aa00cd3e61cc66c12c678fa2a4dad8b2182"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2019-07-29 21:05:30.000000000","tz":-300},"committer":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2019-08-07 12:32:08.000000000","tz":-300},"subject":"Remove hiera roles","message":"Remove hiera roles\n\nThe hiera roles have been moved into tripleo-ansible and are no longer\nneeded within this repository. This change removes the roles that have\nbeen moved.\n\nBecause this change removes the last of the roles within this repo\nthe setup.cfg file has been modified to remove the no longer relevant\ninstall path(s), the molecule related config has been pulled out of\nthe tox.ini file, and the molecule related packages have been purged\nfrom the bindep.txt file.\n\nStory: 2006044\nTask: 34726\nStory: 2006044\nTask: 34730\nDepends-On: I7f9e993735a0347aac12f728393639d88c80ff0f\nChange-Id: I1bf74a4fb0adcae632c06b38e06a572ebc434201\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/10f3c1fa4bd7b3e9b42ca99fc3f27dc494a49b08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/10f3c1fa4bd7b3e9b42ca99fc3f27dc494a49b08"}]},"branch":"refs/heads/master"},"c139e74e7457d0eaf20e9fb45f13cdc274243066":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-08-08 04:33:36.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/66/673366/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/66/673366/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/66/673366/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/66/673366/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/66/673366/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/66/673366/4"}}},"commit":{"parents":[{"commit":"d9412384b928bf1c5ca8d4aedf8c6fa10f01c114","subject":"Merge \"Allows to disable in-flight validations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/d9412384b928bf1c5ca8d4aedf8c6fa10f01c114"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2019-07-29 21:05:30.000000000","tz":-300},"committer":{"name":"Kevin Carter (cloudnull)","email":"kevin@cloudnull.com","date":"2019-08-08 04:33:36.000000000","tz":0},"subject":"Remove hiera roles","message":"Remove hiera roles\n\nThe hiera roles have been moved into tripleo-ansible and are no longer\nneeded within this repository. This change removes the roles that have\nbeen moved.\n\nBecause this change removes the last of the roles within this repo\nthe setup.cfg file has been modified to remove the no longer relevant\ninstall path(s), the molecule related config has been pulled out of\nthe tox.ini file, and the molecule related packages have been purged\nfrom the bindep.txt file.\n\nStory: 2006044\nTask: 34726\nStory: 2006044\nTask: 34730\nDepends-On: I7f9e993735a0347aac12f728393639d88c80ff0f\nChange-Id: I1bf74a4fb0adcae632c06b38e06a572ebc434201\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/c139e74e7457d0eaf20e9fb45f13cdc274243066"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/c139e74e7457d0eaf20e9fb45f13cdc274243066"}]},"branch":"refs/heads/master"},"742a6c7a98718b5aafa0401a60120f3dd0b9d8f0":{"kind":"REWORK","_number":5,"created":"2019-08-08 13:56:03.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/66/673366/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/66/673366/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/66/673366/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/66/673366/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/66/673366/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/66/673366/5"}}},"commit":{"parents":[{"commit":"960e6da0dc8b7702484e4607298715356c55641d","subject":"Merge \"Add extraconfig/role_extraconfig template to hieradata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/960e6da0dc8b7702484e4607298715356c55641d"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2019-07-29 21:05:30.000000000","tz":-300},"committer":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2019-08-08 13:55:56.000000000","tz":-300},"subject":"Remove hiera roles","message":"Remove hiera roles\n\nThe hiera roles have been moved into tripleo-ansible and are no longer\nneeded within this repository. This change removes the roles that have\nbeen moved.\n\nBecause this change removes the last of the roles within this repo\nthe setup.cfg file has been modified to remove the no longer relevant\ninstall path(s), the molecule related config has been pulled out of\nthe tox.ini file, and the molecule related packages have been purged\nfrom the bindep.txt file.\n\nStory: 2006044\nTask: 34726\nStory: 2006044\nTask: 34730\nDepends-On: I7f9e993735a0347aac12f728393639d88c80ff0f\nChange-Id: I1bf74a4fb0adcae632c06b38e06a572ebc434201\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/742a6c7a98718b5aafa0401a60120f3dd0b9d8f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/742a6c7a98718b5aafa0401a60120f3dd0b9d8f0"}]},"branch":"refs/heads/master"},"1b4120b69b2be92e51a5863db3d28b4cb1b68af7":{"kind":"REWORK","_number":6,"created":"2019-08-09 02:09:01.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/66/673366/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/66/673366/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/66/673366/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/66/673366/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/66/673366/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/66/673366/6"}}},"commit":{"parents":[{"commit":"960e6da0dc8b7702484e4607298715356c55641d","subject":"Merge \"Add extraconfig/role_extraconfig template to hieradata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/960e6da0dc8b7702484e4607298715356c55641d"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2019-07-29 21:05:30.000000000","tz":-300},"committer":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2019-08-09 02:08:39.000000000","tz":-300},"subject":"Remove hiera roles","message":"Remove hiera roles\n\nThe hiera roles have been moved into tripleo-ansible and are no longer\nneeded within this repository. This change removes the roles that have\nbeen moved.\n\nBecause this change removes the last of the roles within this repo\nthe setup.cfg file has been modified to remove the no longer relevant\ninstall path(s), the molecule related config has been pulled out of\nthe tox.ini file, and the molecule related packages have been purged\nfrom the bindep.txt file.\n\nStory: 2006044\nTask: 34726\nStory: 2006044\nTask: 34730\nDepends-On: I7f9e993735a0347aac12f728393639d88c80ff0f\nChange-Id: I1bf74a4fb0adcae632c06b38e06a572ebc434201\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/1b4120b69b2be92e51a5863db3d28b4cb1b68af7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/1b4120b69b2be92e51a5863db3d28b4cb1b68af7"}]},"branch":"refs/heads/master"},"b5207ba452d716007c243cacf9efc5d27a759beb":{"kind":"REWORK","_number":7,"created":"2019-08-12 21:53:57.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/66/673366/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/66/673366/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/66/673366/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/66/673366/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/66/673366/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/66/673366/7"}}},"commit":{"parents":[{"commit":"4d4fe71cc7b2d3c1b42d809bf711bd5f6acef5d5","subject":"Merge \"Remove hieradata ansible tasks for datafiles\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/4d4fe71cc7b2d3c1b42d809bf711bd5f6acef5d5"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2019-07-29 21:05:30.000000000","tz":-300},"committer":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2019-08-12 21:53:36.000000000","tz":-300},"subject":"Remove hiera roles","message":"Remove hiera roles\n\nThe hiera roles have been moved into tripleo-ansible and are no longer\nneeded within this repository. This change removes the roles that have\nbeen moved.\n\nBecause this change removes the last of the roles within this repo\nthe setup.cfg file has been modified to remove the no longer relevant\ninstall path(s), the molecule related config has been pulled out of\nthe tox.ini file, and the molecule related packages have been purged\nfrom the bindep.txt file.\n\nStory: 2006044\nTask: 34726\nStory: 2006044\nTask: 34730\nDepends-On: I7f9e993735a0347aac12f728393639d88c80ff0f\nChange-Id: I1bf74a4fb0adcae632c06b38e06a572ebc434201\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/b5207ba452d716007c243cacf9efc5d27a759beb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/b5207ba452d716007c243cacf9efc5d27a759beb"}]},"branch":"refs/heads/master"},"9b8b737e7eb72cd6db1c8ad7e8a21788fdf80072":{"kind":"REWORK","_number":8,"created":"2019-08-14 19:34:02.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/66/673366/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/66/673366/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/66/673366/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/66/673366/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/66/673366/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/66/673366/8"}}},"commit":{"parents":[{"commit":"314917b2a9effbf26460ce6ca786362dd9dd2a4f","subject":"Merge \"Add Python 3 Train unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/314917b2a9effbf26460ce6ca786362dd9dd2a4f"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2019-07-29 21:05:30.000000000","tz":-300},"committer":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2019-08-14 19:33:57.000000000","tz":-300},"subject":"Remove hiera roles","message":"Remove hiera roles\n\nThe hiera roles have been moved into tripleo-ansible and are no longer\nneeded within this repository. This change removes the roles that have\nbeen moved.\n\nBecause this change removes the last of the roles within this repo\nthe setup.cfg file has been modified to remove the no longer relevant\ninstall path(s), the molecule related config has been pulled out of\nthe tox.ini file, and the molecule related packages have been purged\nfrom the bindep.txt file.\n\nStory: 2006044\nTask: 34726\nStory: 2006044\nTask: 34730\nDepends-On: I7f9e993735a0347aac12f728393639d88c80ff0f\nChange-Id: I1bf74a4fb0adcae632c06b38e06a572ebc434201\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/9b8b737e7eb72cd6db1c8ad7e8a21788fdf80072"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/9b8b737e7eb72cd6db1c8ad7e8a21788fdf80072"}]},"branch":"refs/heads/master"},"920b2124d9884bdec039b3bc9c19557b1b6a23fd":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2019-08-26 18:09:42.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/66/673366/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/66/673366/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/66/673366/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/66/673366/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/66/673366/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/66/673366/9"}}},"commit":{"parents":[{"commit":"29b7c8a1ac13cf2df5ab84e8acfd14b5ed92b370","subject":"Merge \"Remove disable_constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/29b7c8a1ac13cf2df5ab84e8acfd14b5ed92b370"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2019-07-29 21:05:30.000000000","tz":-300},"committer":{"name":"Kevin Carter (cloudnull)","email":"kevin@cloudnull.com","date":"2019-08-26 18:09:42.000000000","tz":0},"subject":"Remove hiera roles","message":"Remove hiera roles\n\nThe hiera roles have been moved into tripleo-ansible and are no longer\nneeded within this repository. This change removes the roles that have\nbeen moved.\n\nBecause this change removes the last of the roles within this repo\nthe setup.cfg file has been modified to remove the no longer relevant\ninstall path(s), the molecule related config has been pulled out of\nthe tox.ini file, and the molecule related packages have been purged\nfrom the bindep.txt file.\n\nStory: 2006044\nTask: 34726\nStory: 2006044\nTask: 34730\nDepends-On: I7f9e993735a0347aac12f728393639d88c80ff0f\nChange-Id: I1bf74a4fb0adcae632c06b38e06a572ebc434201\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/920b2124d9884bdec039b3bc9c19557b1b6a23fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/920b2124d9884bdec039b3bc9c19557b1b6a23fd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
