)]}'
{"id":"openstack%2Ftripleo-common~674064","triplet_id":"openstack%2Ftripleo-common~master~I19f4b2a789a5e0996b18afa52d7a8443964f6e1f","project":"openstack/tripleo-common","branch":"master","topic":"mistral-removal","hashtags":[],"change_id":"I19f4b2a789a5e0996b18afa52d7a8443964f6e1f","subject":"Remove the unused tripleo.plan.gather_roles Mistral action","status":"MERGED","created":"2019-08-01 15:00:25.000000000","updated":"2019-08-30 05:25:23.000000000","submitted":"2019-08-30 05:23:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674064-1567142618479-22bc38ef","meta_rev_id":"ab844165446f56418be1eff6f8197b102747bb48","_number":674064,"virtual_id_number":674064,"owner":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-08-29 12:35:24.000000000","_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-08-30 05:23:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-29 15:33:59.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":2,"date":"2019-08-28 18:24:19.000000000","_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":2,"date":"2019-08-22 08:22:03.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":1,"date":"2019-08-28 18:24:19.000000000","_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-05 23:53:20.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-08-28 18:24:19.000000000","updated_by":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"reviewer":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"state":"REVIEWER"},{"updated":"2019-08-29 15:33:59.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-30 05:23:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e3e9e81e4566362f1c694c1ecb3df382586fa058","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-08-01 15:00:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"010d1b0cacda78f93cfd99faac8b85bfa55ac97e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-01 16:27:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/64/674064/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/f5fe368/ : FAILURE in 1h 22m 54s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/64/674064/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/fab5f73/ : FAILURE in 1h 18m 13s","accounts_in_message":[],"_revision_number":1},{"id":"a08202a3b9e7988dfd682e5174c9be4578241597","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 19:48:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://logs.opendev.org/64/674064/1/check/test-release-openstack/3b53326/ : SUCCESS in 5m 51s\n- openstack-tox-cover https://logs.opendev.org/64/674064/1/check/openstack-tox-cover/4eb729c/cover/ : SUCCESS in 9m 03s\n- openstack-tox-pep8 https://logs.opendev.org/64/674064/1/check/openstack-tox-pep8/c314424/ : SUCCESS in 5m 26s\n- openstack-tox-py27 https://logs.opendev.org/64/674064/1/check/openstack-tox-py27/438d3af/ : SUCCESS in 5m 34s\n- openstack-tox-py36 https://logs.opendev.org/64/674064/1/check/openstack-tox-py36/34f469c/ : SUCCESS in 5m 26s\n- openstack-tox-py37 https://logs.opendev.org/64/674064/1/check/openstack-tox-py37/d04bf30/ : SUCCESS in 7m 12s\n- openstack-tox-docs https://logs.opendev.org/64/674064/1/check/openstack-tox-docs/74de28d/html/ : SUCCESS in 5m 50s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/64/674064/1/check/tripleo-ci-centos-7-containers-multinode/feb3af1/ : FAILURE in 1h 12m 35s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/64/674064/1/check/tripleo-ci-centos-7-undercloud-containers/8810f14/ : SUCCESS in 1h 35m 01s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/64/674064/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/b2a6ed3/ : FAILURE in 1h 20m 43s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/64/674064/1/check/tripleo-ci-centos-7-standalone/b927dd5/ : SUCCESS in 1h 43m 07s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/64/674064/1/check/tripleo-ci-fedora-28-standalone/fa8be31/ : FAILURE in 54m 44s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/64/674064/1/check/tripleo-ci-centos-7-standalone-upgrade/7ea51b7/ : FAILURE in 1h 45m 00s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/64/674064/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/7f67579/ : TIMED_OUT in 3h 13m 28s (non-voting)\n- tripleo-build-containers-centos-7 https://logs.opendev.org/64/674064/1/check/tripleo-build-containers-centos-7/c3f8f23/ : SUCCESS in 47m 01s\n- tripleo-build-containers-centos-7-buildah https://logs.opendev.org/64/674064/1/check/tripleo-build-containers-centos-7-buildah/715fdeb/ : SUCCESS in 56m 55s\n- openstack-tox-linters https://logs.opendev.org/64/674064/1/check/openstack-tox-linters/cb58cd5/ : SUCCESS in 7m 44s\n- openstack-tox-py35 https://logs.opendev.org/64/674064/1/check/openstack-tox-py35/61ec85a/ : SUCCESS in 5m 27s\n- openstack-tox-lower-constraints https://logs.opendev.org/64/674064/1/check/openstack-tox-lower-constraints/56c9b8a/ : SUCCESS in 6m 58s","accounts_in_message":[],"_revision_number":1},{"id":"8abc48f23627977ee747b65bff84ba09f7f4cdd4","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-08-02 08:25:51.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"07918b19751b394d511ed6ff33ed940e53d21206","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 11:57:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://logs.opendev.org/64/674064/2/check/test-release-openstack/67e9940/ : SUCCESS in 2m 33s\n- openstack-tox-cover https://logs.opendev.org/64/674064/2/check/openstack-tox-cover/2139104/cover/ : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://logs.opendev.org/64/674064/2/check/openstack-tox-pep8/f52d70f/ : SUCCESS in 2m 24s\n- openstack-tox-py27 https://logs.opendev.org/64/674064/2/check/openstack-tox-py27/dcdb235/ : SUCCESS in 4m 14s\n- openstack-tox-py36 https://logs.opendev.org/64/674064/2/check/openstack-tox-py36/e71dbde/ : SUCCESS in 3m 07s\n- openstack-tox-py37 https://logs.opendev.org/64/674064/2/check/openstack-tox-py37/bfe64ca/ : SUCCESS in 4m 50s\n- openstack-tox-docs https://logs.opendev.org/64/674064/2/check/openstack-tox-docs/7ab271a/html/ : SUCCESS in 3m 21s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/64/674064/2/check/tripleo-ci-centos-7-containers-multinode/81196de/ : SUCCESS in 2h 18m 54s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/64/674064/2/check/tripleo-ci-centos-7-undercloud-containers/420ad65/ : SUCCESS in 1h 48m 06s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/64/674064/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/6edb4ca/ : FAILURE in 2h 09m 33s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/64/674064/2/check/tripleo-ci-centos-7-standalone/c3328f0/ : SUCCESS in 2h 03m 36s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/64/674064/2/check/tripleo-ci-fedora-28-standalone/feb1fd2/ : FAILURE in 57m 11s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/64/674064/2/check/tripleo-ci-centos-7-standalone-upgrade/bfec0e5/ : FAILURE in 1h 45m 11s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/64/674064/2/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/5d06b4a/ : TIMED_OUT in 3h 12m 03s (non-voting)\n- tripleo-build-containers-centos-7 https://logs.opendev.org/64/674064/2/check/tripleo-build-containers-centos-7/57b21a1/ : SUCCESS in 48m 02s\n- tripleo-build-containers-centos-7-buildah https://logs.opendev.org/64/674064/2/check/tripleo-build-containers-centos-7-buildah/bfacea6/ : SUCCESS in 58m 53s\n- openstack-tox-linters https://logs.opendev.org/64/674064/2/check/openstack-tox-linters/241626b/ : SUCCESS in 4m 23s\n- openstack-tox-py35 https://logs.opendev.org/64/674064/2/check/openstack-tox-py35/4ecaa60/ : SUCCESS in 3m 30s\n- openstack-tox-lower-constraints https://logs.opendev.org/64/674064/2/check/openstack-tox-lower-constraints/7368ecf/ : SUCCESS in 5m 52s","accounts_in_message":[],"_revision_number":2},{"id":"8d65ccb014e617cd6a8eff58c580ed1829919030","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-02 12:26:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 finger://ze04.softwarefactory-project.io:7979/43dd8dda378b495b85fdd95e84053080 : POST_FAILURE in 3h 55m 17s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 finger://ze05.softwarefactory-project.io:7979/de6c9049edac4dec9b30448702f72c29 : POST_FAILURE in 3h 58m 20s","accounts_in_message":[],"_revision_number":2},{"id":"61aa8bdeea4d6e72215c88b432f923850a173cda","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-05 23:53:20.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"49bf5ed0fddd2364273c1bb074ad870e775f418a","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-08-22 08:22:03.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"44d6c719772bac9f5d4bb17df5ba7a8b2d5e9063","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 11:17:45.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/74a45b91a2d3494ca4c4dea381088bb9 : SUCCESS in 3m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8f8607e0493547efba044e0e7e85eb2e : SUCCESS in 7m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6eea561fae44a2b9f9f8e91e8b832f1 : SUCCESS in 4m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dd637e315f82405bbd3aac132ab08de1 : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d3491320a3fd42d5981bf0bbe85537bd : SUCCESS in 4m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5c064b54caa649569abb8b809e4bf0b6 : SUCCESS in 6m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e83c8dfa9c6416788d9d8d4e03c56d3 : SUCCESS in 4m 24s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/9e9db2a7e47a4859b1903a6874e7e9a9 : SUCCESS in 2h 23m 18s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/6b694ef4146a4307b8fd9de3647b0087 : SUCCESS in 1h 38m 54s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/92b73db0718645a9a97f72ba7bb25287 : SUCCESS in 2h 31m 36s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/1780dd7a64264bcc9052a1da335d32e9 : SUCCESS in 2h 09m 16s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/9240d5a27cf9496a9b786f72eb4f3b65 : SUCCESS in 2h 14m 10s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/7787b1824f9b47cd9743779ab4b5387d : FAILURE in 2h 06m 34s (non-voting)\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/ae23b53d8634469781caf5940057bab1 : SUCCESS in 55m 00s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/bc9a11d9446d4242beb1511d49e43ed6 : SUCCESS in 54m 24s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/996cc70a13f44be0bccca4b96599e556 : SUCCESS in 4m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/12439ae46e154771a381ed33aed292b2 : SUCCESS in 6m 01s","accounts_in_message":[],"_revision_number":3},{"id":"f39237a6b14ff32cc417524b565bb9f89fc6f895","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-22 12:48:03.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/64/674064/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/ebfb834/ : FAILURE in 2h 21m 28s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 finger://ze06.softwarefactory-project.io:7979/e1c373fb7dfa46a0b5c61014f7195a73 : POST_FAILURE in 4h 24m 07s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/64/674064/3/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/1a91f98/ : SUCCESS in 24m 30s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/64/674064/3/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/331cc8b/ : SUCCESS in 18m 56s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e00cdf1dab7bdb06cdf20bc78952c79d9fedc618","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2019-08-28 18:24:19.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"53a55b2881425bcb2b7993e95a933efa8e84b326","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 22:04:50.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"747a8776f49073cbadc930e278a40bd4c67874d6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 03:42:43.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c2e120f502dd418f91366a285301b95b : SUCCESS in 2m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78ecd5c124234e2497c86632f3bd8c4b : SUCCESS in 3m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3998798365944a4888cefaf3c69e36aa : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d314342e0e34504b38ecfecbd56e48e : SUCCESS in 4m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3bef8c4c8b4e43c185b515973085bead : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efd6ead4c9704325bbc7a6931a894957 : SUCCESS in 4m 34s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/3ce3e988045d47f3a01e85930e067ce7 : SUCCESS in 3h 03m 59s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/decb9764f6a34ce2ad1bdc14dc7b3809 : FAILURE in 1h 34m 21s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/7de4900df5c14df8bf9509a5dbfd1c7d : SUCCESS in 2h 44m 25s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/da814c1c8cdd44ef819e309aef68259d : SUCCESS in 1h 53m 22s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/288c2d0e1376420c941ca7135bca7cc9 : SUCCESS in 52m 22s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/c786882e938b454c85c9c2cc8913e9c2 : SUCCESS in 1h 01m 22s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d0d7aa7940694b9183fe2243a6cdc1e5 : SUCCESS in 4m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/22dba93c247f47ca8140b0117ac53bf7 : SUCCESS in 5m 57s","accounts_in_message":[],"_revision_number":3},{"id":"52aa1068f424f15f9245e110bcbfa77044fd1f64","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-08-29 12:35:24.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"5cf5fce5b58e8ca57439a46d7c3a8b100301f8d6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-29 15:33:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/64/674064/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/2c1688a/ : SUCCESS in 2h 57m 31s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/64/674064/3/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/4fd6784/ : SUCCESS in 21m 26s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/64/674064/3/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/303274b/ : SUCCESS in 17m 50s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"900c4640793bf1bcda8563ffd7a2b21eba5c705d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 16:14:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/644945ef6c014ef9b6b94b1a22a768da : SUCCESS in 3m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0b013f10a3c244f09d98fa897f23bcec : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0c02aca6a8b42269d4487b812e23a63 : SUCCESS in 3m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0d25024136b841b986948478e6d31405 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6ba0d5b365234f39a6de9304e1deb3a5 : SUCCESS in 3m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/101b89852c2548e1a579fd714fa70069 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a7aa8c9b89b4c4ebef3ecccea005e22 : SUCCESS in 4m 08s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/44e28d92ff514a5189edc33e49cc1b73 : SUCCESS in 2h 52m 56s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/780745ace40e418cad8c6e3d194808e0 : SUCCESS in 1h 45m 42s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/72ff97ebf7434af885d471cd0f862a04 : SUCCESS in 3h 17m 52s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/c445f1f4a10b4802b64b246400538093 : SUCCESS in 2h 11m 03s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/67fe3ee08bbb457387dae3fa0b33bf3a : SUCCESS in 2h 22m 40s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/70907e8eaf6d41a49f3155d52498ddfa : FAILURE in 2h 14m 15s (non-voting)\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/e4ce23faa5fd4967b229ef380460fce5 : SUCCESS in 53m 53s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/be265de294ec4ca4988761002516c1a9 : SUCCESS in 1h 04m 20s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/597d947d39f14408a44d06445d9aac64 : SUCCESS in 5m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c74aa140f55349f0a5667b59e942592e : SUCCESS in 6m 44s","accounts_in_message":[],"_revision_number":3},{"id":"c544335d83c9b623469f63b801708519249f8838","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 16:15:10.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e3fb420305c681a869c703c00800949389293636","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 05:23:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"45f2dbb8c1da38afb872792db6c7845591f98087","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 05:23:38.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/71513c7f91a144b680058a98960624ce : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15d01551ef8e43f7984dde71a0a39e16 : SUCCESS in 4m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/92186b554a66488ebddea8a3498d1d8e : SUCCESS in 6m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6db8eae63ffe4b0086be1db24e00e595 : SUCCESS in 6m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ca67fb4041bc436b913a58f6e0748310 : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4defb34bdcd14cf8ba4a06277a8bb282 : SUCCESS in 5m 32s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/e64938b2b35242b69b8ecf65eb6a570f : SUCCESS in 3h 17m 33s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/d927c5429452432b99f320f993364718 : SUCCESS in 2h 39m 54s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/d3e475265adb4d4e8e72b158787e61ef : SUCCESS in 2h 57m 41s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/c2dd7143931f414db2530867baa27fad : SUCCESS in 1h 56m 13s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/88e2ff8c301145fd9a5d8e166a0cda8a : SUCCESS in 53m 17s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/8d56e9492a814dc282c9b8a744c5d3b6 : SUCCESS in 1h 13m 34s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2d10799dc6f54d2f92965b41e9ba458b : SUCCESS in 4m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f7d7f537277c4596ae647d0277b0175a : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":3},{"id":"90b8f560680a467c3b2d8b1b4498cf98414876c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 05:25:23.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3eb03a5e2bda4a0cb203ebc309ddc533 : SUCCESS in 1m 28s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2ec71e3a533d48e44d426a293c628cb87dbece0e","revisions":{"c3a97901c98ae6f2eb2dff016ece6c419dc2cfab":{"kind":"REWORK","_number":1,"created":"2019-08-01 15:00:25.000000000","uploader":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"ref":"refs/changes/64/674064/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/64/674064/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/64/674064/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/64/674064/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/64/674064/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/64/674064/1"}}},"commit":{"parents":[{"commit":"0b41d1cf256852515798408496e4a11b5d34fbe6","subject":"Remove the file exists action and save and restore to swift actions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/0b41d1cf256852515798408496e4a11b5d34fbe6"}]}],"author":{"name":"Dougal Matthews","email":"dougal@redhat.com","date":"2019-08-01 14:45:38.000000000","tz":60},"committer":{"name":"Dougal Matthews","email":"dougal@redhat.com","date":"2019-08-01 14:45:38.000000000","tz":60},"subject":"Remove the unused tripleo.plan.gather_roles Mistral action","message":"Remove the unused tripleo.plan.gather_roles Mistral action\n\nChange-Id: I19f4b2a789a5e0996b18afa52d7a8443964f6e1f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/c3a97901c98ae6f2eb2dff016ece6c419dc2cfab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/c3a97901c98ae6f2eb2dff016ece6c419dc2cfab"}]},"branch":"refs/heads/master"},"2562e36eb1c93fa8943b21ff8f070185c4c2e537":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-08-02 08:25:51.000000000","uploader":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"ref":"refs/changes/64/674064/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/64/674064/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/64/674064/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/64/674064/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/64/674064/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/64/674064/2"}}},"commit":{"parents":[{"commit":"2d51fd285c24e42d1cf923e4ef3ad257f8503aaf","subject":"Remove the file exists action and save and restore to swift actions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/2d51fd285c24e42d1cf923e4ef3ad257f8503aaf"}]}],"author":{"name":"Dougal Matthews","email":"dougal@redhat.com","date":"2019-08-01 14:45:38.000000000","tz":60},"committer":{"name":"Dougal Matthews","email":"dougal@redhat.com","date":"2019-08-02 08:25:39.000000000","tz":60},"subject":"Remove the unused tripleo.plan.gather_roles Mistral action","message":"Remove the unused tripleo.plan.gather_roles Mistral action\n\nChange-Id: I19f4b2a789a5e0996b18afa52d7a8443964f6e1f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/2562e36eb1c93fa8943b21ff8f070185c4c2e537"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/2562e36eb1c93fa8943b21ff8f070185c4c2e537"}]},"branch":"refs/heads/master"},"2ec71e3a533d48e44d426a293c628cb87dbece0e":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-08-22 08:22:03.000000000","uploader":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"ref":"refs/changes/64/674064/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/64/674064/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/64/674064/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/64/674064/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/64/674064/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/64/674064/3"}}},"commit":{"parents":[{"commit":"5c19663bc49fa0884c807a043331f1fe20907531","subject":"Remove the file exists action and save and restore to swift actions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/5c19663bc49fa0884c807a043331f1fe20907531"}]}],"author":{"name":"Dougal Matthews","email":"dougal@redhat.com","date":"2019-08-01 14:45:38.000000000","tz":60},"committer":{"name":"Dougal Matthews","email":"dougal@redhat.com","date":"2019-08-22 08:21:47.000000000","tz":60},"subject":"Remove the unused tripleo.plan.gather_roles Mistral action","message":"Remove the unused tripleo.plan.gather_roles Mistral action\n\nChange-Id: I19f4b2a789a5e0996b18afa52d7a8443964f6e1f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/2ec71e3a533d48e44d426a293c628cb87dbece0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/2ec71e3a533d48e44d426a293c628cb87dbece0e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
