)]}'
{"id":"openstack%2Ftripleo-common~674065","triplet_id":"openstack%2Ftripleo-common~master~Iaa57ce63e2ec9b3c72312500f53f35d3ca1a2849","project":"openstack/tripleo-common","branch":"master","topic":"mistral-removal","hashtags":[],"change_id":"Iaa57ce63e2ec9b3c72312500f53f35d3ca1a2849","subject":"Remove the unused tripleo.plan.update_plan_environment Mistral action","status":"MERGED","created":"2019-08-01 15:00:25.000000000","updated":"2019-09-10 05:20:30.000000000","submitted":"2019-09-10 05:19:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674065-1568092768602-1a0e83b5","meta_rev_id":"89a8b0ca87e39aa9ff1f4c01148b581c7ec5de8f","_number":674065,"virtual_id_number":674065,"owner":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":2,"date":"2019-09-10 05:19:28.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-09 10:38:16.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":2,"date":"2019-09-09 13:13:03.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-09-09 07:38:04.000000000","_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":2,"date":"2019-09-09 07:38:04.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2019-09-09 19:57:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":1,"date":"2019-09-10 01:59:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":1,"date":"2019-09-09 13:05:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":1,"date":"2019-09-09 13:13:03.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-05 23:53:36.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-08-28 18:25:03.000000000","updated_by":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"reviewer":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"state":"REVIEWER"},{"updated":"2019-09-09 10:38:16.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-09 13:05:04.000000000","updated_by":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"reviewer":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"state":"REVIEWER"},{"updated":"2019-09-09 13:13:03.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-09-09 19:57:30.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2019-09-10 01:59:53.000000000","updated_by":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"reviewer":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"state":"REVIEWER"},{"updated":"2019-09-10 05:19:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"13c634547a51d80b4740ef15431d524201daa970","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-08-01 15:00:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"afe22e6665d4b394569b5172fd3653a0df27355c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-01 15:32:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/65/674065/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/7e1a81d/ : RETRY_LIMIT in 7m 05s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 finger://ze03.softwarefactory-project.io:7979/05202d769dad48319120c4655cf98372 : RETRY_LIMIT in 7s","accounts_in_message":[],"_revision_number":1},{"id":"7403435909182754b7120f95c3bedc527bd7c7dd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 19:58:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://logs.opendev.org/65/674065/1/check/test-release-openstack/3422831/ : SUCCESS in 5m 28s\n- openstack-tox-cover https://logs.opendev.org/65/674065/1/check/openstack-tox-cover/4c09ad4/cover/ : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://logs.opendev.org/65/674065/1/check/openstack-tox-pep8/61663f8/ : SUCCESS in 5m 19s\n- openstack-tox-py27 https://logs.opendev.org/65/674065/1/check/openstack-tox-py27/b3fb9ca/ : SUCCESS in 8m 10s\n- openstack-tox-py36 https://logs.opendev.org/65/674065/1/check/openstack-tox-py36/35a410d/ : SUCCESS in 5m 06s\n- openstack-tox-py37 https://logs.opendev.org/65/674065/1/check/openstack-tox-py37/716c5f8/ : SUCCESS in 7m 01s\n- openstack-tox-docs https://logs.opendev.org/65/674065/1/check/openstack-tox-docs/a2a2546/html/ : SUCCESS in 7m 01s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/65/674065/1/check/tripleo-ci-centos-7-containers-multinode/a1ba960/ : FAILURE in 1h 11m 21s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/65/674065/1/check/tripleo-ci-centos-7-undercloud-containers/7d0a53d/ : SUCCESS in 1h 51m 42s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/65/674065/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/018244b/ : FAILURE in 1h 31m 06s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/65/674065/1/check/tripleo-ci-centos-7-standalone/7c9b755/ : SUCCESS in 1h 45m 50s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/65/674065/1/check/tripleo-ci-fedora-28-standalone/bdcf42f/ : FAILURE in 55m 08s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/65/674065/1/check/tripleo-ci-centos-7-standalone-upgrade/24adccf/ : FAILURE in 1h 59m 17s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/65/674065/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/a3aaeda/ : TIMED_OUT in 3h 12m 25s (non-voting)\n- tripleo-build-containers-centos-7 https://logs.opendev.org/65/674065/1/check/tripleo-build-containers-centos-7/f0074e3/ : SUCCESS in 51m 39s\n- tripleo-build-containers-centos-7-buildah https://logs.opendev.org/65/674065/1/check/tripleo-build-containers-centos-7-buildah/53a7309/ : SUCCESS in 55m 13s\n- openstack-tox-linters https://logs.opendev.org/65/674065/1/check/openstack-tox-linters/0fe02fd/ : SUCCESS in 5m 32s\n- openstack-tox-py35 https://logs.opendev.org/65/674065/1/check/openstack-tox-py35/5de847c/ : SUCCESS in 7m 53s\n- openstack-tox-lower-constraints https://logs.opendev.org/65/674065/1/check/openstack-tox-lower-constraints/f26fea4/ : SUCCESS in 9m 30s","accounts_in_message":[],"_revision_number":1},{"id":"2f6b000b792404213dc4ea0175a1b9c256d94286","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-08-02 08:25:51.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"950aebc662ce62d963924314001aa3c82c057931","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-02 12:15:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 finger://ze03.softwarefactory-project.io:7979/c276011a5da04933b03279f739d8b804 : POST_FAILURE in 3h 47m 53s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 finger://ze01.softwarefactory-project.io:7979/5c92a70124814297b4b31f3b15c93c16 : POST_FAILURE in 3h 45m 55s","accounts_in_message":[],"_revision_number":2},{"id":"4ef005345b11fc70ba014bf784e6b848e5ed4fbb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 12:27:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://logs.opendev.org/65/674065/2/check/test-release-openstack/6195266/ : SUCCESS in 2m 24s\n- openstack-tox-cover https://logs.opendev.org/65/674065/2/check/openstack-tox-cover/b7dc75c/cover/ : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://logs.opendev.org/65/674065/2/check/openstack-tox-pep8/829321f/ : SUCCESS in 2m 42s\n- openstack-tox-py27 https://logs.opendev.org/65/674065/2/check/openstack-tox-py27/e19a337/ : SUCCESS in 3m 32s\n- openstack-tox-py36 https://logs.opendev.org/65/674065/2/check/openstack-tox-py36/e16013c/ : SUCCESS in 4m 18s\n- openstack-tox-py37 https://logs.opendev.org/65/674065/2/check/openstack-tox-py37/86fc935/ : SUCCESS in 6m 49s\n- openstack-tox-docs https://logs.opendev.org/65/674065/2/check/openstack-tox-docs/f170489/html/ : SUCCESS in 3m 47s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/65/674065/2/check/tripleo-ci-centos-7-containers-multinode/c4e4b7c/ : SUCCESS in 2h 38m 46s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/65/674065/2/check/tripleo-ci-centos-7-undercloud-containers/da5a044/ : SUCCESS in 1h 38m 04s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/65/674065/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/8b67ce8/ : SUCCESS in 2h 43m 45s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/65/674065/2/check/tripleo-ci-centos-7-standalone/3656e7f/ : SUCCESS in 1h 59m 22s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/65/674065/2/check/tripleo-ci-fedora-28-standalone/d11a655/ : FAILURE in 1h 00m 33s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/65/674065/2/check/tripleo-ci-centos-7-standalone-upgrade/db8a0ce/ : FAILURE in 1h 44m 27s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/65/674065/2/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/1c11589/ : TIMED_OUT in 3h 12m 42s (non-voting)\n- tripleo-build-containers-centos-7 https://logs.opendev.org/65/674065/2/check/tripleo-build-containers-centos-7/b587ff6/ : SUCCESS in 47m 37s\n- tripleo-build-containers-centos-7-buildah https://logs.opendev.org/65/674065/2/check/tripleo-build-containers-centos-7-buildah/f469801/ : SUCCESS in 59m 55s\n- openstack-tox-linters https://logs.opendev.org/65/674065/2/check/openstack-tox-linters/a743ce9/ : SUCCESS in 4m 44s\n- openstack-tox-py35 https://logs.opendev.org/65/674065/2/check/openstack-tox-py35/b2b3f32/ : SUCCESS in 3m 38s\n- openstack-tox-lower-constraints https://logs.opendev.org/65/674065/2/check/openstack-tox-lower-constraints/870ebcc/ : SUCCESS in 7m 17s","accounts_in_message":[],"_revision_number":2},{"id":"963a4a13e1adf059a92cd552bc0c7297c4ed856b","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-05 23:53:36.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"90f5e722da4f1563fb5a57961d291add83a2ee9d","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-08-22 08:22:03.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"8c56676b5654637e16aba1eccd7b381c15caa526","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 11:54:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7c23da19f1904740b926e4376a00d559 : SUCCESS in 3m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ccbdf21518444c6d971a61e63b188c1e : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d98bb2a5f2f84565bf470bea50ad37fb : SUCCESS in 3m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/90907604e3f549479dccee285525c499 : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4c602f62d474689ba5cfbc06c8d2bdd : SUCCESS in 4m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6c5f44845ffa440ab394d39b1e47d1af : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/751b19c2878e4f2baa0602782f720e5d : SUCCESS in 4m 53s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/50724f56aeeb4d839bdee164f1e17a5f : SUCCESS in 2h 56m 24s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/460ff306f226450d9439c21dbe44025f : SUCCESS in 1h 47m 50s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/8767fda18ce34efeaf8879874186afde : SUCCESS in 3h 06m 43s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/8030ae98a8214ac3bb51601912696606 : SUCCESS in 2h 00m 08s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/7622ec51732d428f95837d039273a2f0 : SUCCESS in 2h 11m 22s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/51408f52b8cd4a208cbec61132d63452 : FAILURE in 2h 34m 41s (non-voting)\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/2f07bd033c0046979487acb20cc71d78 : SUCCESS in 49m 39s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/e0392b2ea06a43aeb81def764c2de1cf : SUCCESS in 1h 01m 16s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fde5d278ba3e445ebb1e0f1b5731bcd5 : SUCCESS in 5m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/73ab92327c53437592d35dca40ed769d : SUCCESS in 7m 15s","accounts_in_message":[],"_revision_number":3},{"id":"a4051e189dde907d9be85a8decb31025a293fbb7","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-22 12:35:40.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 finger://ze07.softwarefactory-project.io:7979/35a38f212329415c8eff7914f6e1739f : POST_FAILURE in 4h 09m 24s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 finger://ze02.softwarefactory-project.io:7979/637ceef82d9648698790b7ae14e39c41 : POST_FAILURE in 4h 11m 37s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/65/674065/3/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/08259b7/ : FAILURE in 16m 27s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/65/674065/3/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/8252a7f/ : SUCCESS in 19m 40s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2f56a380ba95d84e38f51d62229f8578f5ed5c5e","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2019-08-28 18:25:03.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"f40c76682e1098a6574e5b4f615220dd889b7c16","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 22:04:50.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"47fdf1a5c9a1631429ea7934f864c856b938bc30","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 03:42:43.000000000","message":"Patch Set 3: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"98de7908a2cb1e5b01ec74b1b49c48249073d012","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-08-29 12:40:43.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"e561b1d1e7d9a8a75c3a9e783b5425b811ae33c8","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-29 15:43:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/65/674065/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/1515598/ : SUCCESS in 3h 00m 36s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/65/674065/3/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/ad13db6/ : SUCCESS in 21m 03s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/65/674065/3/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/ad632c9/ : SUCCESS in 18m 47s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"540f3e806b8febff3e4e9a7d9a57ae6fa92006c6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 16:08:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6b88d35587884603aeda343a903b7b1c : SUCCESS in 2m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2e5c74dc8bc14a61aba0b0b8c0ac4cff : SUCCESS in 7m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c8ee9b81ffb4a70b36e8832891b09f0 : SUCCESS in 4m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2ef815f08b274c6e88418f916d46e7e0 : SUCCESS in 3m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4bbf63b51d8a42b5a438b643a287cbac : SUCCESS in 4m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/aac1b8de85254ee1a5c20d83d7da4b3d : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/554819cb424f4f088dc83c3afc96fd8a : SUCCESS in 4m 37s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/41de2b8cff3a4a4e85795bc446c615d1 : SUCCESS in 3h 11m 32s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/0b57bac48efd4e0d9a4c808863d3c1a6 : SUCCESS in 1h 50m 04s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/d92891cce012477da616fc01f2140438 : SUCCESS in 2h 50m 06s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/6be86e6439cd49efbbe661c6422d17cf : SUCCESS in 2h 20m 18s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/3e39e33ac86848d185e12d9e16f056b4 : SUCCESS in 2h 14m 40s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/01f83f35b0e3422b97568113feedf129 : FAILURE in 2h 02m 42s (non-voting)\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/acbf2cd21d7642008c614d26dd6e33cf : SUCCESS in 54m 56s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/00f151844e124b8f9f0cb7d81384967a : SUCCESS in 58m 12s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/eb76fa38509847a98f9f86e7efd82126 : SUCCESS in 4m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/83493f63a6fa4d7b8e4ecdfe347658b2 : SUCCESS in 6m 26s","accounts_in_message":[],"_revision_number":3},{"id":"947e3f5e9f7f957fa3eee037228236ebaffa9fbb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 16:15:37.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"3d58f5ca7e6ddd2620549811659d3cce0c017d01","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 05:23:39.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d6ab125722c449aea2dc756d5131881f : SUCCESS in 2m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49048a54e9444bb482d1c0f9a02d7aee : SUCCESS in 4m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/138b7076cfe84b44825b0da70af09795 : SUCCESS in 6m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8b9d124e5f4942779530089b41051dcf : SUCCESS in 6m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a983d9be8a224d33a3da149f1533d8ac : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dee0c97d10cc497487a17ff660af0ebb : SUCCESS in 6m 12s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/c2cc773480ae4e1bbadbb51a7bf76469 : FAILURE in 2h 03m 29s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/d0a7af81e68d482db5ca1687e474e0cd : SUCCESS in 2h 54m 30s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/dedd867691204c35b0050a0bbc1748ef : SUCCESS in 2h 35m 53s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/9f1631c00af540a1a9a4732114f2c4e0 : SUCCESS in 1h 48m 00s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/e864457b2464418595f63438680976e3 : SUCCESS in 1h 00m 30s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/2ec43c483aaa4a1db308e13f28bfbcfc : SUCCESS in 1h 12m 26s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/296467c25a624a94bbd1fdd3eaee058b : SUCCESS in 3m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a55a3cd35fa94af3a173a72e075ab317 : SUCCESS in 8m 07s","accounts_in_message":[],"_revision_number":3},{"id":"c8033bded964a381e4f9ddcb4207a7aee3ba3a5b","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-30 14:10:51.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"b07e7ac0c198821f4223ed4996fb5ca1d3c868bc","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-30 17:10:54.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/65/674065/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/46f9ff9/ : SUCCESS in 2h 58m 37s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/65/674065/3/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/43d8e43/ : SUCCESS in 20m 58s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/65/674065/3/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/d0689e7/ : SUCCESS in 17m 24s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b3edb716d1d5d23cbe8aaacad5f3ee8bcfd32a5c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 19:06:10.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f02d0b611ad04d0297e4cc2aa98ee182 : SUCCESS in 3m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9fe1b02a02d9471184c72ac26d353c57 : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1f101b28f1a472f8fbab198c9516b57 : SUCCESS in 4m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/184610e59fdc4b6fafcbd92a25e02206 : SUCCESS in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86deaf4a91cd4bdaab157b96830062c6 : SUCCESS in 4m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/671f42d1aa754fd2b2c8dda9e989284c : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56ad22a8931b4d2da70c5c0ce9a248ed : SUCCESS in 5m 03s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/ae388977c799494e87b772d866716aa2 : TIMED_OUT in 3h 36m 33s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/492df213f5ca47eeb7b55819c56795c7 : SUCCESS in 1h 55m 27s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/5a3f6ffaf3b64dd1a53d84db8491a0b8 : SUCCESS in 3h 17m 35s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/f5c4c3224b044323a5e8233a594ed0eb : SUCCESS in 2h 09m 03s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/bb8b194a034843408b3c5b074ed0a678 : SUCCESS in 1h 57m 18s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/a649ea73473e44c08bec17a9aebd6fde : FAILURE in 2h 18m 32s (non-voting)\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/708e612162cb404bb90dd84c13b28773 : SUCCESS in 51m 38s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/0281a055308144748402278bca5f3010 : SUCCESS in 1h 09m 28s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0722adec09f141f7b4a24c37609392f7 : SUCCESS in 4m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0c52f06270fb4408b628cea4c2c27bf7 : SUCCESS in 7m 01s","accounts_in_message":[],"_revision_number":3},{"id":"3a10c89f4ac9f9808be9db1338f3db19200fb711","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-09-09 07:38:04.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"5b7f0dcec7cd9a563f16c4b4d4965ffbbb420727","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-09 10:38:16.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/65/674065/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/53de7e8/ : SUCCESS in 2h 58m 41s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/65/674065/4/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/7f770e1/ : RETRY_LIMIT in 7m 49s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/65/674065/4/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/ceebb79/ : RETRY_LIMIT in 7m 43s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"021977ad1846e8d06ac31abb5cdf7a6a1f6ffb3c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 11:01:52.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0a4cdc2309d54193b42d2aa4cb708cfe : SUCCESS in 2m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef2b54773d804bda8226fb7cd09347b2 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/635e97ab08ff4610acb61cb75ac47e37 : SUCCESS in 2m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f74496e762244133a1164387daa2105c : SUCCESS in 3m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e9dfd5e957a441ea8b4a8bc254821105 : SUCCESS in 3m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b648664170f647b08968b3eb1e551b3c : SUCCESS in 3m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a1b2f95bfca40579766787e4c2c4b85 : SUCCESS in 3m 33s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/faf52fafa91c425b87f95afbe01ae70d : SUCCESS in 2h 59m 33s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/82c844b9a13f41e9b0bd2c15d9889f9f : SUCCESS in 1h 59m 09s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/b01db435dddd4c87ab59a78af9fb4d25 : SUCCESS in 2h 39m 07s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/b2a955bff49246b4b807b39f9fce5760 : SUCCESS in 2h 09m 34s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/0b804cf1b199445a88abb1a04d895fe6 : SUCCESS in 2h 27m 42s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/4b993bccda194bdb8449b048269f186d : SUCCESS in 1h 57m 08s (non-voting)\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/3b9f3da74bea4da98d68af7ff1200248 : SUCCESS in 47m 42s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/5ad72d30f3684d43b436b81336186354 : SUCCESS in 1h 06m 05s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/36057198b3ba49f7815d408bc30afb06 : SUCCESS in 4m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2a09d0e1464e477d8ad2f784767254fd : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":4},{"id":"23fcc23f5e9c7429e5bb620d449bcfa340a8c291","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2019-09-09 13:05:04.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"77737078cc4fb92c554aae117f230e1ab119b0c7","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-09-09 13:13:03.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"ab6dd6215df8ddebfe6457fca38f162a716c01c8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 13:13:29.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"aff74f1b330158b34a3e2e748e34ecd8b2a1096a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 18:09:11.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7d6464a7fc644911a2d166fb0e9df1dd : SUCCESS in 3m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbdf20c605384907a35164b352a6b66b : SUCCESS in 3m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/847aad5fda8c457bb067f8e386fe7060 : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/48d7200bad6c428ba00ad56f50fd7737 : SUCCESS in 5m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9a9640345e9949009107a70947e4c48a : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea362b21cb8b416da2d2fd0a6d34f4d2 : SUCCESS in 5m 37s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/ae2ab4d9d8db4a50972b1012cbb7ec16 : FAILURE in 2h 50m 48s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/ff8c9d9f447c472381900c69bac14402 : SUCCESS in 2h 36m 54s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/b18d1d2b091247f6ba29469dff3347db : SUCCESS in 2h 19m 07s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/6aeb2770b95b4e238694adca83c1af50 : SUCCESS in 44m 36s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/ca816fb4a32b44caad0fad74689a0578 : SUCCESS in 57m 15s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/cbb28563cf3d4649a66cc7c2e6173501 : SUCCESS in 5m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c4b76ae4dda243a3b7d5b62d2c0d2eed : SUCCESS in 6m 38s","accounts_in_message":[],"_revision_number":4},{"id":"12bfa320058bfb4c091e3cd822f628f637efbc17","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-09-09 18:23:50.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"16ef82da86cb696b9c7d1079baa2e83231f99aab","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-09-09 19:57:30.000000000","message":"Patch Set 4: Code-Review+1\n\nexcellent","accounts_in_message":[],"_revision_number":4},{"id":"0a1f5c9e2dcd5989f1a8a4db9e2e38c9970ba0ce","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-09 21:29:27.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/65/674065/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/3072b26/ : SUCCESS in 3h 04m 48s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/65/674065/4/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/6cb51ff/ : RETRY_LIMIT in 8m 29s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/65/674065/4/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/d86c379/ : RETRY_LIMIT in 8m 08s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f3c22190d7672d2dcf1ae8f7ee3b1273d80554c7","author":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"date":"2019-09-10 01:59:53.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"453e57365311076eb07155863f3a7571ece59d3d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 02:13:23.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b628b31b177a46bf82cdd0ed00e1f6fe : SUCCESS in 2m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/54ea7fad781f49369e1daad806fb5bf3 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b23133be7e54316985b20a0daed30f5 : SUCCESS in 3m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9f289227767b4527890e7c9f7823e3f0 : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a72ea57d03847e5bea5d9675924a6d4 : SUCCESS in 3m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/97e58df34fb645119437741e33021b36 : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fdce86b2d6c4889aa6f647496566d24 : SUCCESS in 3m 42s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/d83070ec0eea46b4835ca08f0ca3c9c1 : SUCCESS in 2h 50m 49s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/445b23263e9942ac9fbef1c1d3745553 : SUCCESS in 1h 39m 55s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/269d7b5073b548d7833223653b7e13c8 : SUCCESS in 2h 37m 26s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/3fac9eccdeac47ada2525574f540dbf7 : SUCCESS in 2h 06m 38s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/5aa644fb8bc84c1daf1505fe80bfc7dd : SUCCESS in 2h 01m 20s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/87b8ee3ac9774a6cbc24a00a8ec8e1bf : FAILURE in 1h 57m 28s (non-voting)\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/9f8dca4b7f694d69a4ed6a2420ebcebe : SUCCESS in 48m 23s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/e64f156007e5490ea4c1f17178e1b89f : SUCCESS in 1h 11m 30s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ae46f28b0b4943b383db0f7965b2ef50 : SUCCESS in 3m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0c0cca927b5942809a3e12b8806e340e : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":4},{"id":"b5114ede70e866592b5e861bea3c60cfe3ec0d27","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 02:13:36.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"e417c02799d83ce1ce30488eea89b5d01d0ffd12","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 05:19:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"74202270cfb85532b0327c1a13045b547413c795","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 05:19:28.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/abd940cee7cd4d159e23d4a12581843e : SUCCESS in 1m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd80c7084ccb4b5a99e22d9d4aef4e3c : SUCCESS in 2m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7f5f77cd21a84b73aa904f32f8aadc88 : SUCCESS in 2m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0cf2621e79f946c891e2a4ba3250a875 : SUCCESS in 3m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c2131e1c7ce34e77a5c87c7200773071 : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ff253bd649a4fb8b77aef5a6143579f : SUCCESS in 3m 15s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/9c8130d28b7f4e22ad816ca1a3b09a5e : SUCCESS in 3h 00m 46s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/c972ad81a14a434cbb393f3781ca2928 : SUCCESS in 2h 09m 37s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/8394acc4654f4d4aad742a813fc18b08 : SUCCESS in 2h 31m 45s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/e7807dc5555f4a6185a36eaf469da808 : SUCCESS in 2h 04m 10s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/0d2fe57251dd4d6c93003726b87fba65 : SUCCESS in 45m 41s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/9845458e02e641c9a819a437baf9726e : SUCCESS in 1h 16m 07s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/118cc7a8f42c40768a720866c9867740 : SUCCESS in 3m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0ab1fdcbb79c452d8c9f94015901da8d : SUCCESS in 4m 52s","accounts_in_message":[],"_revision_number":4},{"id":"31756b27a03c8d71a04b6f4fcd07127393601b94","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 05:20:30.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e80a7bb06ad34b01862eec80e0d6e88d : SUCCESS in 44s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"6cb8048a2525c5ecb1005666057902de191590ef","revisions":{"87c9b616d2a4d5cb7d6348437233e5ec3ea5170f":{"kind":"REWORK","_number":1,"created":"2019-08-01 15:00:25.000000000","uploader":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"ref":"refs/changes/65/674065/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/65/674065/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/65/674065/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/65/674065/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/65/674065/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/65/674065/1"}}},"commit":{"parents":[{"commit":"c3a97901c98ae6f2eb2dff016ece6c419dc2cfab","subject":"Remove the unused tripleo.plan.gather_roles Mistral action","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/c3a97901c98ae6f2eb2dff016ece6c419dc2cfab"}]}],"author":{"name":"Dougal Matthews","email":"dougal@redhat.com","date":"2019-08-01 14:47:06.000000000","tz":60},"committer":{"name":"Dougal Matthews","email":"dougal@redhat.com","date":"2019-08-01 14:47:06.000000000","tz":60},"subject":"Remove the unused tripleo.plan.update_plan_environment Mistral action","message":"Remove the unused tripleo.plan.update_plan_environment Mistral action\n\nThis action isn\u0027t used or tested.\n\nChange-Id: Iaa57ce63e2ec9b3c72312500f53f35d3ca1a2849\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/87c9b616d2a4d5cb7d6348437233e5ec3ea5170f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/87c9b616d2a4d5cb7d6348437233e5ec3ea5170f"}]},"branch":"refs/heads/master"},"aef5f4fb27ddf43613d17b353ca5f056d5cc6205":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-08-02 08:25:51.000000000","uploader":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"ref":"refs/changes/65/674065/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/65/674065/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/65/674065/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/65/674065/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/65/674065/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/65/674065/2"}}},"commit":{"parents":[{"commit":"2562e36eb1c93fa8943b21ff8f070185c4c2e537","subject":"Remove the unused tripleo.plan.gather_roles Mistral action","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/2562e36eb1c93fa8943b21ff8f070185c4c2e537"}]}],"author":{"name":"Dougal Matthews","email":"dougal@redhat.com","date":"2019-08-01 14:47:06.000000000","tz":60},"committer":{"name":"Dougal Matthews","email":"dougal@redhat.com","date":"2019-08-02 08:25:39.000000000","tz":60},"subject":"Remove the unused tripleo.plan.update_plan_environment Mistral action","message":"Remove the unused tripleo.plan.update_plan_environment Mistral action\n\nThis action isn\u0027t used or tested.\n\nChange-Id: Iaa57ce63e2ec9b3c72312500f53f35d3ca1a2849\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/aef5f4fb27ddf43613d17b353ca5f056d5cc6205"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/aef5f4fb27ddf43613d17b353ca5f056d5cc6205"}]},"branch":"refs/heads/master"},"fe87f50c15241107a106521d3e4b1d9dd02401dd":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-08-22 08:22:03.000000000","uploader":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"ref":"refs/changes/65/674065/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/65/674065/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/65/674065/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/65/674065/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/65/674065/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/65/674065/3"}}},"commit":{"parents":[{"commit":"2ec71e3a533d48e44d426a293c628cb87dbece0e","subject":"Remove the unused tripleo.plan.gather_roles Mistral action","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/2ec71e3a533d48e44d426a293c628cb87dbece0e"}]}],"author":{"name":"Dougal Matthews","email":"dougal@redhat.com","date":"2019-08-01 14:47:06.000000000","tz":60},"committer":{"name":"Dougal Matthews","email":"dougal@redhat.com","date":"2019-08-22 08:21:47.000000000","tz":60},"subject":"Remove the unused tripleo.plan.update_plan_environment Mistral action","message":"Remove the unused tripleo.plan.update_plan_environment Mistral action\n\nThis action isn\u0027t used or tested.\n\nChange-Id: Iaa57ce63e2ec9b3c72312500f53f35d3ca1a2849\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/fe87f50c15241107a106521d3e4b1d9dd02401dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/fe87f50c15241107a106521d3e4b1d9dd02401dd"}]},"branch":"refs/heads/master"},"6cb8048a2525c5ecb1005666057902de191590ef":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-09-09 07:38:04.000000000","uploader":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"ref":"refs/changes/65/674065/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/65/674065/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/65/674065/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/65/674065/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/65/674065/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/65/674065/4"}}},"commit":{"parents":[{"commit":"0a8190f21b080b4e279355b6ce042526598afa9a","subject":"Merge \"NeutronPhysnetNUMANodesMapping parameter issue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/0a8190f21b080b4e279355b6ce042526598afa9a"}]}],"author":{"name":"Dougal Matthews","email":"dougal@redhat.com","date":"2019-08-01 14:47:06.000000000","tz":60},"committer":{"name":"Dougal Matthews","email":"dougal@redhat.com","date":"2019-09-09 07:38:04.000000000","tz":0},"subject":"Remove the unused tripleo.plan.update_plan_environment Mistral action","message":"Remove the unused tripleo.plan.update_plan_environment Mistral action\n\nThis action isn\u0027t used or tested.\n\nChange-Id: Iaa57ce63e2ec9b3c72312500f53f35d3ca1a2849\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/6cb8048a2525c5ecb1005666057902de191590ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/6cb8048a2525c5ecb1005666057902de191590ef"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
