)]}'
{"id":"openstack%2Ftripleo-common~674293","triplet_id":"openstack%2Ftripleo-common~stable%2Fstein~Ie41ba6de1170a9c9ffc94889e47bce9ef1200521","project":"openstack/tripleo-common","branch":"stable/stein","topic":"bug/1838536-stable/stein","hashtags":[],"change_id":"Ie41ba6de1170a9c9ffc94889e47bce9ef1200521","subject":"Fix ansible-generate-inventory action","status":"MERGED","created":"2019-08-02 12:25:29.000000000","updated":"2019-08-05 20:06:47.000000000","submitted":"2019-08-05 20:06:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674293-1565035607742-d227193f","meta_rev_id":"56ea85aeeec6114616ade7f04364aec5f36425b2","_number":674293,"virtual_id_number":674293,"owner":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"date":"2019-08-05 09:46:11.000000000","_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":2,"date":"2019-08-05 20:06:47.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-05 08:08:07.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"all":[{"value":2,"date":"2019-08-05 08:15:40.000000000","_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2019-08-05 13:01:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2019-08-05 14:22:53.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-02 12:25:29.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2019-08-05 08:08:07.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-05 08:15:40.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2019-08-05 09:46:11.000000000","updated_by":{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},"reviewer":{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},"state":"REVIEWER"},{"updated":"2019-08-05 13:01:51.000000000","updated_by":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"reviewer":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"state":"REVIEWER"},{"updated":"2019-08-05 14:22:53.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-08-05 20:06:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e542b526bc2cb5c726787cec3288c4652d529492","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-02 12:25:29.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"b187a1bbaca51e3f3decee038806369d17e0fd75","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-02 12:47:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/93/674293/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/080a71d/ : FAILURE in 19m 49s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/93/674293/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/d9a2e95/ : FAILURE in 19m 43s","accounts_in_message":[],"_revision_number":1},{"id":"f74e4f50eb285bf9299e92d3e7cb42d13c463344","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 15:14:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://logs.opendev.org/93/674293/1/check/openstack-tox-cover/7b053e8/ : FAILURE in 4m 00s\n- openstack-tox-pep8 https://logs.opendev.org/93/674293/1/check/openstack-tox-pep8/3ba17f7/ : SUCCESS in 3m 52s\n- openstack-tox-py27 https://logs.opendev.org/93/674293/1/check/openstack-tox-py27/77e0523/ : FAILURE in 4m 32s\n- openstack-tox-py36 https://logs.opendev.org/93/674293/1/check/openstack-tox-py36/cc4f0ed/ : FAILURE in 4m 40s\n- openstack-tox-py37 https://logs.opendev.org/93/674293/1/check/openstack-tox-py37/d7f4536/ : FAILURE in 4m 58s\n- openstack-tox-docs https://logs.opendev.org/93/674293/1/check/openstack-tox-docs/6957fef/html/ : SUCCESS in 3m 20s\n- tripleo-ci-centos-7-containers-multinode tripleo-ci-centos-7-containers-multinode : SKIPPED\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/93/674293/1/check/tripleo-ci-centos-7-undercloud-containers/3e785d4/ : FAILURE in 22m 43s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades : SKIPPED (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/93/674293/1/check/tripleo-ci-centos-7-standalone/6050ac3/ : FAILURE in 26m 06s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/93/674293/1/check/tripleo-ci-fedora-28-standalone/35dd1d2/ : SUCCESS in 1h 46m 42s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades tripleo-ci-centos-7-containerized-undercloud-upgrades : SKIPPED (non-voting)\n- openstack-tox-py35 https://logs.opendev.org/93/674293/1/check/openstack-tox-py35/6ab5b0a/ : FAILURE in 3m 34s\n- openstack-tox-lower-constraints https://logs.opendev.org/93/674293/1/check/openstack-tox-lower-constraints/1cd203a/ : FAILURE in 5m 02s","accounts_in_message":[],"_revision_number":1},{"id":"20fc3cae5ef0be447ca72a0a784b79e0c745770f","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-08-05 03:41:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9dc266aec8282539ef7b3c7571bfd392d45a6978","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-05 04:48:48.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/93/674293/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/65ac683/ : FAILURE in 1h 06m 21s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/93/674293/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/2d445b7/ : FAILURE in 1h 02m 23s","accounts_in_message":[],"_revision_number":2},{"id":"adf93578ef42ee389ba5dcb0adb1f9612ccd311d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 05:20:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://logs.opendev.org/93/674293/2/check/openstack-tox-cover/618194f/cover/ : SUCCESS in 3m 03s\n- openstack-tox-pep8 https://logs.opendev.org/93/674293/2/check/openstack-tox-pep8/f885b9d/ : SUCCESS in 3m 16s\n- openstack-tox-py27 https://logs.opendev.org/93/674293/2/check/openstack-tox-py27/75b2a15/ : SUCCESS in 3m 39s\n- openstack-tox-py36 https://logs.opendev.org/93/674293/2/check/openstack-tox-py36/8d347e3/ : SUCCESS in 3m 11s\n- openstack-tox-py37 https://logs.opendev.org/93/674293/2/check/openstack-tox-py37/87a9815/ : SUCCESS in 4m 43s\n- openstack-tox-docs https://logs.opendev.org/93/674293/2/check/openstack-tox-docs/1d651ca/html/ : SUCCESS in 2m 54s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/93/674293/2/check/tripleo-ci-centos-7-containers-multinode/3d7802e/ : FAILURE in 1h 24m 09s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/93/674293/2/check/tripleo-ci-centos-7-undercloud-containers/8a0d3b5/ : FAILURE in 1h 10m 54s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/93/674293/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/066fdf0/ : FAILURE in 1h 05m 51s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/93/674293/2/check/tripleo-ci-centos-7-standalone/0b86d76/ : FAILURE in 50m 40s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/93/674293/2/check/tripleo-ci-fedora-28-standalone/c36e559/ : FAILURE in 51m 59s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/93/674293/2/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/b5077bf/ : FAILURE in 1h 23m 29s (non-voting)\n- openstack-tox-py35 https://logs.opendev.org/93/674293/2/check/openstack-tox-py35/c3bb84f/ : SUCCESS in 3m 28s\n- openstack-tox-lower-constraints https://logs.opendev.org/93/674293/2/check/openstack-tox-lower-constraints/76e6ad7/ : SUCCESS in 5m 31s","accounts_in_message":[],"_revision_number":2},{"id":"666ef9bf00e8ff5e34da789e17fb6f0e853ff057","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-08-05 05:26:58.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"f189f08581de66b75d57fe170dc21ee2863c73a3","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-05 08:08:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/93/674293/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/3a43987/ : SUCCESS in 2h 33m 01s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/93/674293/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/2e165e3/ : SUCCESS in 2h 39m 49s","accounts_in_message":[],"_revision_number":3},{"id":"04210ba92bff4c2e5469014e310a6d66032c420f","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2019-08-05 08:15:40.000000000","message":"Patch Set 3: Code-Review+2\n\nlgtm but waiting for zuul to vote otherwise would merge if it was green","accounts_in_message":[],"_revision_number":3},{"id":"5f925c8131844992771ab9a8e92d76c16cb18a19","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 08:31:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://logs.opendev.org/93/674293/3/check/openstack-tox-cover/d2cf52b/cover/ : SUCCESS in 2m 52s\n- openstack-tox-pep8 https://logs.opendev.org/93/674293/3/check/openstack-tox-pep8/1bd4d16/ : SUCCESS in 2m 38s\n- openstack-tox-py27 https://logs.opendev.org/93/674293/3/check/openstack-tox-py27/616e770/ : SUCCESS in 2m 36s\n- openstack-tox-py36 https://logs.opendev.org/93/674293/3/check/openstack-tox-py36/b6da998/ : SUCCESS in 3m 06s\n- openstack-tox-py37 https://logs.opendev.org/93/674293/3/check/openstack-tox-py37/f97159b/ : SUCCESS in 2m 52s\n- openstack-tox-docs https://logs.opendev.org/93/674293/3/check/openstack-tox-docs/a8ace1e/html/ : SUCCESS in 2m 23s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/93/674293/3/check/tripleo-ci-centos-7-containers-multinode/6814962/ : SUCCESS in 2h 12m 58s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/93/674293/3/check/tripleo-ci-centos-7-undercloud-containers/189834a/ : SUCCESS in 1h 44m 38s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/93/674293/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/069a7ac/ : SUCCESS in 2h 53m 58s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/93/674293/3/check/tripleo-ci-centos-7-standalone/ca01cb4/ : SUCCESS in 1h 36m 51s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/93/674293/3/check/tripleo-ci-fedora-28-standalone/337010f/ : SUCCESS in 2h 35m 56s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/93/674293/3/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/5ce8387/ : FAILURE in 2h 02m 11s (non-voting)\n- openstack-tox-py35 https://logs.opendev.org/93/674293/3/check/openstack-tox-py35/831af84/ : SUCCESS in 4m 27s\n- openstack-tox-lower-constraints https://logs.opendev.org/93/674293/3/check/openstack-tox-lower-constraints/8fafeb2/ : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":3},{"id":"6fc6dd2e0d813e2a25cb47ae41b476ec40c46347","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2019-08-05 13:01:51.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"138c01887b665800d88df475bbb5a08151fd88ee","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-05 14:22:53.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a237f6319a89f9b39926093394952f2a10f85260","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 14:23:12.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"cc60257a737581c3631baaef86cc231bb71931cd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 20:06:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"a72cabc1a01068509e3603254473a35e07edd465","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 20:06:47.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://logs.opendev.org/93/674293/3/gate/openstack-tox-pep8/d8268dd/ : SUCCESS in 4m 19s\n- openstack-tox-py27 https://logs.opendev.org/93/674293/3/gate/openstack-tox-py27/d7f5dec/ : SUCCESS in 5m 21s\n- openstack-tox-py36 https://logs.opendev.org/93/674293/3/gate/openstack-tox-py36/092a833/ : SUCCESS in 5m 16s\n- openstack-tox-py37 https://logs.opendev.org/93/674293/3/gate/openstack-tox-py37/35ae679/ : SUCCESS in 4m 19s\n- openstack-tox-docs https://logs.opendev.org/93/674293/3/gate/openstack-tox-docs/ed3f678/html/ : SUCCESS in 3m 32s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/93/674293/3/gate/tripleo-ci-centos-7-containers-multinode/1f671f2/ : SUCCESS in 2h 16m 59s\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/93/674293/3/gate/tripleo-ci-centos-7-standalone/441bc8a/ : SUCCESS in 2h 04m 34s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/93/674293/3/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/fe8346e/ : SUCCESS in 2h 24m 04s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/93/674293/3/gate/tripleo-ci-centos-7-undercloud-containers/f2cb21e/ : SUCCESS in 1h 46m 21s\n- openstack-tox-lower-constraints https://logs.opendev.org/93/674293/3/gate/openstack-tox-lower-constraints/74048c2/ : SUCCESS in 5m 04s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"df8f0243d0cb52806a05a8aff6dfdd9c875f02d9","revisions":{"2bf761f2e8530166f35a9f55be4a547b597cb6d8":{"kind":"REWORK","_number":1,"created":"2019-08-02 12:25:29.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/93/674293/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/93/674293/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/93/674293/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/93/674293/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/93/674293/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/93/674293/1"}}},"commit":{"parents":[{"commit":"5525d4d6d265b6be0ebcb36ba553dcb01082469b","subject":"Properly close web request","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/5525d4d6d265b6be0ebcb36ba553dcb01082469b"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2019-07-31 15:31:05.000000000","tz":330},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-08-02 12:25:29.000000000","tz":0},"subject":"Fix ansible-generate-inventory action","message":"Fix ansible-generate-inventory action\n\nansible-generate-inventory action raises Exception\nwhen \u0027ansible_python_interpreter\u0027 input is passed to it.\n\nThis patch fixes it and also update unit test to cover\nmissing part.\n\nCloses-Bug: #1838536\nChange-Id: Ie41ba6de1170a9c9ffc94889e47bce9ef1200521\n(cherry picked from commit 6a35576e9e6154a33e64413750e9da01603b7722)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/2bf761f2e8530166f35a9f55be4a547b597cb6d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/2bf761f2e8530166f35a9f55be4a547b597cb6d8"}]},"branch":"refs/heads/stable/stein"},"dd0d43a0032aa3041c91a58f816d916d14c1474b":{"kind":"REWORK","_number":2,"created":"2019-08-05 03:41:34.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/93/674293/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/93/674293/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/93/674293/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/93/674293/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/93/674293/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/93/674293/2"}}},"commit":{"parents":[{"commit":"5525d4d6d265b6be0ebcb36ba553dcb01082469b","subject":"Properly close web request","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/5525d4d6d265b6be0ebcb36ba553dcb01082469b"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2019-07-31 15:31:05.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2019-08-05 03:40:49.000000000","tz":330},"subject":"Fix ansible-generate-inventory action","message":"Fix ansible-generate-inventory action\n\nansible-generate-inventory action raises Exception\nwhen \u0027ansible_python_interpreter\u0027 input is passed to it.\n\nThis patch fixes it and also update unit test to cover\nmissing part.\n\nAdjusts Unit test for stable/stein as [1] is not in\nstable/stein.\n\n[1] https://review.opendev.org/#/c/663466/\n\nCloses-Bug: #1838536\nChange-Id: Ie41ba6de1170a9c9ffc94889e47bce9ef1200521\n(cherry picked from commit 6a35576e9e6154a33e64413750e9da01603b7722)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/dd0d43a0032aa3041c91a58f816d916d14c1474b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/dd0d43a0032aa3041c91a58f816d916d14c1474b"}]},"branch":"refs/heads/stable/stein"},"df8f0243d0cb52806a05a8aff6dfdd9c875f02d9":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-08-05 05:26:58.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/93/674293/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/93/674293/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/93/674293/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/93/674293/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/93/674293/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/93/674293/3"}}},"commit":{"parents":[{"commit":"5525d4d6d265b6be0ebcb36ba553dcb01082469b","subject":"Properly close web request","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/5525d4d6d265b6be0ebcb36ba553dcb01082469b"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2019-07-31 15:31:05.000000000","tz":330},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2019-08-05 05:26:54.000000000","tz":0},"subject":"Fix ansible-generate-inventory action","message":"Fix ansible-generate-inventory action\n\nansible-generate-inventory action raises Exception\nwhen \u0027ansible_python_interpreter\u0027 input is passed to it.\n\nThis patch fixes it and also update unit test to cover\nmissing part.\n\nAdjusts Unit test for stable/stein as [1] is not in\nstable/stein.\n\n[1] https://review.opendev.org/#/c/663466/\n\nCloses-Bug: #1838536\nDepends-On: https://review.opendev.org/#/c/674530/\nChange-Id: Ie41ba6de1170a9c9ffc94889e47bce9ef1200521\n(cherry picked from commit 6a35576e9e6154a33e64413750e9da01603b7722)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/df8f0243d0cb52806a05a8aff6dfdd9c875f02d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/df8f0243d0cb52806a05a8aff6dfdd9c875f02d9"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
