)]}'
{"id":"openstack%2Ftripleo-common~674431","triplet_id":"openstack%2Ftripleo-common~stable%2Fstein~I169b07553b171fcacb532bf58acc4200e270bfbb","project":"openstack/tripleo-common","branch":"stable/stein","topic":"swift-proxy-healthcheck-stable/stein","hashtags":[],"change_id":"I169b07553b171fcacb532bf58acc4200e270bfbb","subject":"Fix swift-proxy healthcheck","status":"MERGED","created":"2019-08-02 22:09:28.000000000","updated":"2019-08-06 16:09:52.000000000","submitted":"2019-08-06 16:09:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674431-1565107792371-3bded89f","meta_rev_id":"074eec1e7fddd99818fc3112ece2aa6f97addadc","_number":674431,"virtual_id_number":674431,"owner":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-08-06 06:44:05.000000000","_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},{"value":0,"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":2,"date":"2019-08-06 16:09:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-06 02:48:52.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},{"value":0,"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":2,"date":"2019-08-05 23:55:43.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},{"value":0,"date":"2019-08-06 11:06:54.000000000","_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":1,"date":"2019-08-06 13:30:40.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"date":"2019-08-06 11:06:23.000000000","_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-06 02:48:52.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-06 11:06:23.000000000","updated_by":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"reviewer":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"state":"REVIEWER"},{"updated":"2019-08-06 11:06:54.000000000","updated_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"reviewer":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"state":"REVIEWER"},{"updated":"2019-08-06 13:30:40.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-08-06 16:09:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"05e4c4651a0adbe9aba10cc33fc802909422540a","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2019-08-02 22:09:28.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"6476c6c17253eaeced0c09cbc3246bf40c705b09","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-02 23:12:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/31/674431/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/f3fd690/ : FAILURE in 1h 02m 23s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/31/674431/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/da83fb1/ : FAILURE in 59m 21s","accounts_in_message":[],"_revision_number":1},{"id":"690243f2c913e888895bea39f2539a4c65576275","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-03 00:11:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://logs.opendev.org/31/674431/1/check/openstack-tox-cover/c02bef9/cover/ : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://logs.opendev.org/31/674431/1/check/openstack-tox-pep8/b66020d/ : SUCCESS in 3m 22s\n- openstack-tox-py27 https://logs.opendev.org/31/674431/1/check/openstack-tox-py27/15065e7/ : SUCCESS in 3m 27s\n- openstack-tox-py36 https://logs.opendev.org/31/674431/1/check/openstack-tox-py36/b3a1855/ : SUCCESS in 3m 17s\n- openstack-tox-py37 https://logs.opendev.org/31/674431/1/check/openstack-tox-py37/8af0e22/ : SUCCESS in 3m 55s\n- openstack-tox-docs https://logs.opendev.org/31/674431/1/check/openstack-tox-docs/db01afe/html/ : SUCCESS in 3m 05s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/31/674431/1/check/tripleo-ci-centos-7-containers-multinode/23dd825/ : FAILURE in 1h 16m 11s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/31/674431/1/check/tripleo-ci-centos-7-undercloud-containers/d67263f/ : FAILURE in 1h 09m 29s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/31/674431/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/74693be/ : FAILURE in 1h 18m 47s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/31/674431/1/check/tripleo-ci-centos-7-standalone/65f5555/ : FAILURE in 48m 07s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/31/674431/1/check/tripleo-ci-fedora-28-standalone/da3692d/ : FAILURE in 46m 50s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/31/674431/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/4184b57/ : FAILURE in 1h 35m 20s (non-voting)\n- openstack-tox-py35 https://logs.opendev.org/31/674431/1/check/openstack-tox-py35/1dbd3dc/ : SUCCESS in 3m 51s\n- openstack-tox-lower-constraints https://logs.opendev.org/31/674431/1/check/openstack-tox-lower-constraints/bcecc29/ : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":1},{"id":"82e81cb3dd3b323c09ffee27f93b0d257234f3d1","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2019-08-05 07:42:50.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8e869eb443e66b21e4a99ad6b241ae1add161b92","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-05 08:49:30.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/31/674431/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/85b8e72/ : FAILURE in 1h 05m 51s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/31/674431/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/32136b5/ : FAILURE in 1h 04m 41s","accounts_in_message":[],"_revision_number":1},{"id":"9f51e9e1e3d255a4c1f754acfbd22e0ebab1fc6b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 09:35:21.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://logs.opendev.org/31/674431/1/check/openstack-tox-cover/48ef97d/cover/ : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://logs.opendev.org/31/674431/1/check/openstack-tox-pep8/1fc1ebc/ : SUCCESS in 3m 04s\n- openstack-tox-py27 https://logs.opendev.org/31/674431/1/check/openstack-tox-py27/8ee156a/ : SUCCESS in 4m 54s\n- openstack-tox-py36 https://logs.opendev.org/31/674431/1/check/openstack-tox-py36/0cc5e9f/ : SUCCESS in 2m 51s\n- openstack-tox-py37 https://logs.opendev.org/31/674431/1/check/openstack-tox-py37/37f455f/ : SUCCESS in 4m 34s\n- openstack-tox-docs https://logs.opendev.org/31/674431/1/check/openstack-tox-docs/a4dabeb/html/ : SUCCESS in 3m 39s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/31/674431/1/check/tripleo-ci-centos-7-containers-multinode/b07c769/ : FAILURE in 1h 23m 35s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/31/674431/1/check/tripleo-ci-centos-7-undercloud-containers/9e25d76/ : FAILURE in 1h 08m 59s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/31/674431/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/cf71439/ : FAILURE in 1h 26m 14s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/31/674431/1/check/tripleo-ci-centos-7-standalone/5e8b12c/ : FAILURE in 52m 58s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/31/674431/1/check/tripleo-ci-fedora-28-standalone/0ceeb52/ : FAILURE in 57m 06s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/31/674431/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/64c1a76/ : FAILURE in 1h 38m 44s (non-voting)\n- openstack-tox-py35 https://logs.opendev.org/31/674431/1/check/openstack-tox-py35/3a14af2/ : SUCCESS in 4m 18s\n- openstack-tox-lower-constraints https://logs.opendev.org/31/674431/1/check/openstack-tox-lower-constraints/3dbef1a/ : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":1},{"id":"eaab7e8a7876abc49598bba11f0ee481409be491","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2019-08-05 13:11:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0b7ef31f4c44479b6fcddae4b26db416624e71b9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-05 14:25:37.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/31/674431/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/a57adf2/ : FAILURE in 1h 12m 35s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/31/674431/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/69caf9d/ : FAILURE in 1h 04m 39s","accounts_in_message":[],"_revision_number":1},{"id":"5f6ad1041e4a27765659f6cbf43a5683389eb7ed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 14:56:51.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://logs.opendev.org/31/674431/1/check/openstack-tox-cover/ffaec65/cover/ : SUCCESS in 3m 12s\n- openstack-tox-pep8 https://logs.opendev.org/31/674431/1/check/openstack-tox-pep8/3c6d670/ : SUCCESS in 2m 57s\n- openstack-tox-py27 https://logs.opendev.org/31/674431/1/check/openstack-tox-py27/4367cfe/ : SUCCESS in 2m 59s\n- openstack-tox-py36 https://logs.opendev.org/31/674431/1/check/openstack-tox-py36/80ed79e/ : SUCCESS in 2m 59s\n- openstack-tox-py37 https://logs.opendev.org/31/674431/1/check/openstack-tox-py37/efd7b1a/ : SUCCESS in 3m 21s\n- openstack-tox-docs https://logs.opendev.org/31/674431/1/check/openstack-tox-docs/bab416e/html/ : SUCCESS in 2m 42s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/31/674431/1/check/tripleo-ci-centos-7-containers-multinode/d94e71d/ : FAILURE in 1h 06m 02s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/31/674431/1/check/tripleo-ci-centos-7-undercloud-containers/f207515/ : FAILURE in 1h 12m 48s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/31/674431/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/40a14fa/ : FAILURE in 1h 16m 49s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/31/674431/1/check/tripleo-ci-centos-7-standalone/4435b91/ : FAILURE in 51m 37s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/31/674431/1/check/tripleo-ci-fedora-28-standalone/8abf7a2/ : FAILURE in 1h 01m 15s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/31/674431/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/9696408/ : FAILURE in 1h 30m 10s (non-voting)\n- openstack-tox-py35 https://logs.opendev.org/31/674431/1/check/openstack-tox-py35/bf79380/ : SUCCESS in 5m 15s\n- openstack-tox-lower-constraints https://logs.opendev.org/31/674431/1/check/openstack-tox-lower-constraints/8ff10af/ : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":1},{"id":"8ee8e7e74e38a3684728d8bdf56fcc4ac498d4bc","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-05 23:55:43.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1cf3291da259b320566c96266958e72165030905","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-06 02:48:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/31/674431/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/b2904ed/ : SUCCESS in 2h 45m 16s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/31/674431/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/c092657/ : SUCCESS in 2h 50m 56s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/31/674431/1/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/4d2cc35/ : RETRY_LIMIT in 6m 56s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/31/674431/1/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/b85563c/ : RETRY_LIMIT in 6m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"682e9f6e41b1da4bcee2049bca9bc0f973e819b8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 02:48:59.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://logs.opendev.org/31/674431/1/check/openstack-tox-cover/4bbedae/cover/ : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://logs.opendev.org/31/674431/1/check/openstack-tox-pep8/16616c3/ : SUCCESS in 3m 52s\n- openstack-tox-py27 https://logs.opendev.org/31/674431/1/check/openstack-tox-py27/7c6bbf6/ : SUCCESS in 4m 08s\n- openstack-tox-py36 https://logs.opendev.org/31/674431/1/check/openstack-tox-py36/505a3c2/ : SUCCESS in 3m 57s\n- openstack-tox-py37 https://logs.opendev.org/31/674431/1/check/openstack-tox-py37/242a654/ : SUCCESS in 5m 01s\n- openstack-tox-docs https://logs.opendev.org/31/674431/1/check/openstack-tox-docs/56717de/html/ : SUCCESS in 4m 11s\n- tripleo-ci-centos-7-containers-multinode finger://ze03.openstack.org/34f1014406194fcbb9b35e928fd97d47 : POST_FAILURE in 2h 14m 25s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/31/674431/1/check/tripleo-ci-centos-7-undercloud-containers/9a9d7fc/ : SUCCESS in 1h 41m 51s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades finger://ze03.openstack.org/4077572be6e34ef4831a61479b2d7eed : POST_FAILURE in 2h 38m 30s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/31/674431/1/check/tripleo-ci-centos-7-standalone/2c051bc/ : SUCCESS in 1h 54m 21s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/31/674431/1/check/tripleo-ci-fedora-28-standalone/aacb1dc/ : SUCCESS in 1h 29m 41s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/31/674431/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/c4fd78d/ : FAILURE in 1h 39m 45s (non-voting)\n- openstack-tox-py35 https://logs.opendev.org/31/674431/1/check/openstack-tox-py35/291d33f/ : SUCCESS in 4m 03s\n- openstack-tox-lower-constraints https://logs.opendev.org/31/674431/1/check/openstack-tox-lower-constraints/623c0e3/ : SUCCESS in 5m 49s","accounts_in_message":[],"_revision_number":1},{"id":"dc86fcd3b36d3872823f8ad3c983118ba99ff1b9","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2019-08-06 06:44:05.000000000","message":"Patch Set 1:\n\nrecheck - POST_FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"ba30789094453b980784ee9898640399d11a4fec","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-06 09:57:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/31/674431/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/ce0ae82/ : SUCCESS in 2h 38m 23s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/31/674431/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/dd39900/ : SUCCESS in 3h 10m 43s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/31/674431/1/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/9f2e23f/ : RETRY_LIMIT in 6m 26s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/31/674431/1/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/517fdc9/ : RETRY_LIMIT in 6m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"57716a0f8393e4166b6c6bece92537a94c895e12","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 10:46:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://logs.opendev.org/31/674431/1/check/openstack-tox-cover/050545d/cover/ : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://logs.opendev.org/31/674431/1/check/openstack-tox-pep8/98ca818/ : SUCCESS in 4m 59s\n- openstack-tox-py27 https://logs.opendev.org/31/674431/1/check/openstack-tox-py27/ab64185/ : SUCCESS in 3m 07s\n- openstack-tox-py36 https://logs.opendev.org/31/674431/1/check/openstack-tox-py36/0ea44d6/ : SUCCESS in 4m 39s\n- openstack-tox-py37 https://logs.opendev.org/31/674431/1/check/openstack-tox-py37/230d259/ : SUCCESS in 3m 52s\n- openstack-tox-docs https://logs.opendev.org/31/674431/1/check/openstack-tox-docs/17694f3/html/ : SUCCESS in 4m 42s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/31/674431/1/check/tripleo-ci-centos-7-containers-multinode/b33cdb6/ : SUCCESS in 2h 11m 53s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/31/674431/1/check/tripleo-ci-centos-7-undercloud-containers/04e6b2d/ : SUCCESS in 1h 28m 04s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/31/674431/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/c41d62f/ : SUCCESS in 2h 34m 44s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/31/674431/1/check/tripleo-ci-centos-7-standalone/6d1444c/ : SUCCESS in 2h 04m 12s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/31/674431/1/check/tripleo-ci-fedora-28-standalone/a0ddc0d/ : FAILURE in 2h 00m 51s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/31/674431/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/3043fa1/ : FAILURE in 1h 41m 07s (non-voting)\n- openstack-tox-py35 https://logs.opendev.org/31/674431/1/check/openstack-tox-py35/fa2603e/ : SUCCESS in 5m 11s\n- openstack-tox-lower-constraints https://logs.opendev.org/31/674431/1/check/openstack-tox-lower-constraints/22d3651/ : SUCCESS in 5m 23s","accounts_in_message":[],"_revision_number":1},{"id":"22061ed6b354ffaad79a4f0d89f5621c3180bd4d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-06 13:30:40.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"315c0f847716892c292ad50ca7173e303d01f1e1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 13:30:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"39733f526aa9bf96be10d240464a3f8c48bad256","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 16:09:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://logs.opendev.org/31/674431/1/gate/openstack-tox-pep8/412eb26/ : SUCCESS in 4m 32s\n- openstack-tox-py27 https://logs.opendev.org/31/674431/1/gate/openstack-tox-py27/a295990/ : SUCCESS in 4m 24s\n- openstack-tox-py36 https://logs.opendev.org/31/674431/1/gate/openstack-tox-py36/f1dbcb5/ : SUCCESS in 4m 01s\n- openstack-tox-py37 https://logs.opendev.org/31/674431/1/gate/openstack-tox-py37/0c7e1e8/ : SUCCESS in 6m 50s\n- openstack-tox-docs https://logs.opendev.org/31/674431/1/gate/openstack-tox-docs/3b6be1b/html/ : SUCCESS in 5m 41s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/31/674431/1/gate/tripleo-ci-centos-7-containers-multinode/d12f805/ : SUCCESS in 2h 29m 48s\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/31/674431/1/gate/tripleo-ci-centos-7-standalone/d08dd3a/ : SUCCESS in 2h 06m 20s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/31/674431/1/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/382cdf0/ : SUCCESS in 2h 18m 43s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/31/674431/1/gate/tripleo-ci-centos-7-undercloud-containers/42671ff/ : SUCCESS in 1h 40m 59s\n- openstack-tox-lower-constraints https://logs.opendev.org/31/674431/1/gate/openstack-tox-lower-constraints/5f552df/ : SUCCESS in 7m 22s","accounts_in_message":[],"_revision_number":1},{"id":"41f86c3d7534c2245a8b75f5a1ce46d440983ab6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 16:09:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e2c5deba1d1fce3c9001364f51aa09565223aee3","revisions":{"e2c5deba1d1fce3c9001364f51aa09565223aee3":{"kind":"REWORK","_number":1,"created":"2019-08-02 22:09:28.000000000","uploader":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"ref":"refs/changes/31/674431/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/31/674431/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/31/674431/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/31/674431/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/31/674431/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/31/674431/1"}}},"commit":{"parents":[{"commit":"2d015a5fa9e2a071d88c762649d766c5d35a8158","subject":"Bump ceph/daemon container image to v4.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/2d015a5fa9e2a071d88c762649d766c5d35a8158"}]}],"author":{"name":"Martin Schuppert","email":"mschuppert@redhat.com","date":"2019-08-01 14:41:40.000000000","tz":120},"committer":{"name":"Martin Schuppert","email":"mschuppert@redhat.com","date":"2019-08-02 22:09:28.000000000","tz":0},"subject":"Fix swift-proxy healthcheck","message":"Fix swift-proxy healthcheck\n\nWith the current swift-proxy healthcheck the container status is\nunhealty even the service is up.\n\nThe used ps/grep won\u0027t return the swift proxy processes as the\noutput gets truncated:\n\n()[root@overcloud-controller-0 /]$ ps -e\n    PID TTY          TIME CMD\n      1 ?        00:00:00 dumb-init\n      6 ?        00:02:52 swift-proxy-ser\n     28 ?        00:00:25 swift-proxy-ser\n   4753 pts/1    00:00:00 sh\n   4797 pts/1    00:00:00 ps\n\nAs a result we end up in the else, checking the cache.\n\nThis moves the if to use pgrep instead.\n\nChange-Id: I169b07553b171fcacb532bf58acc4200e270bfbb\n(cherry picked from commit 3ddb77e34096a095909ea4dc5a18d478eeb93cab)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/e2c5deba1d1fce3c9001364f51aa09565223aee3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/e2c5deba1d1fce3c9001364f51aa09565223aee3"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
