)]}'
{"id":"openstack%2Ftripleo-common~674476","triplet_id":"openstack%2Ftripleo-common~master~I260501e6061c0d6652b22504e7284940b2c64709","project":"openstack/tripleo-common","branch":"master","hashtags":[],"change_id":"I260501e6061c0d6652b22504e7284940b2c64709","subject":"Fix expensive api call to get deployment_name","status":"MERGED","created":"2019-08-04 00:51:38.000000000","updated":"2019-08-05 12:09:20.000000000","submitted":"2019-08-05 11:38:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674476-1565005127310-682773cd","meta_rev_id":"c018ac01361b0315c2b192205b1432660f20c28c","_number":674476,"virtual_id_number":674476,"owner":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-08-05 09:01:15.000000000","_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":2,"date":"2019-08-05 11:38:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-04 06:48:29.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"all":[{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":2,"date":"2019-08-04 20:47:13.000000000","_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-05 09:04:09.000000000","_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"all":[{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-05 09:04:09.000000000","_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-04 06:48:29.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-04 20:47:13.000000000","updated_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"reviewer":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"state":"REVIEWER"},{"updated":"2019-08-05 09:04:09.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"REVIEWER"},{"updated":"2019-08-05 11:38:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6184018c61162f085bc8bcffe006b20c7d7c9492","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2019-08-04 00:51:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2cdab31e1efee5bf45c48e6fd44558dccaa36359","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-04 01:15:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover finger://ze02.openstack.org/c58d7794a9b445e597ab9292a67daccc : RETRY_LIMIT in 3m 59s\n- openstack-tox-pep8 finger://ze02.openstack.org/c51a3e7573cf4a22934a8ae568ff1e15 : POST_FAILURE in 2m 00s\n- openstack-tox-py27 finger://ze03.openstack.org/663063d89b74424aadda9eb004948ab0 : POST_FAILURE in 2m 42s\n- openstack-tox-py36 finger://ze02.openstack.org/12a433890297450fb8c6043ba1d34080 : POST_FAILURE in 2m 39s\n- openstack-tox-py37 finger://ze03.openstack.org/6c1f6330cf854b9dade2f469eb196e00 : POST_FAILURE in 2m 52s\n- openstack-tox-docs finger://ze01.openstack.org/8655e4f96eb449cbb5642de46171b2b3 : POST_FAILURE in 2m 31s\n- tripleo-ci-centos-7-containers-multinode tripleo-ci-centos-7-containers-multinode : SKIPPED\n- tripleo-ci-centos-7-undercloud-containers tripleo-ci-centos-7-undercloud-containers : SKIPPED\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades : SKIPPED (non-voting)\n- tripleo-ci-centos-7-standalone tripleo-ci-centos-7-standalone : SKIPPED\n- tripleo-ci-fedora-28-standalone tripleo-ci-fedora-28-standalone : SKIPPED (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade tripleo-ci-centos-7-standalone-upgrade : SKIPPED (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades tripleo-ci-centos-7-containerized-undercloud-upgrades : SKIPPED (non-voting)\n- openstack-tox-linters finger://ze05.openstack.org/65c81611e9fc4be5ae5cd68da8cfdc1a : POST_FAILURE in 2m 53s\n- openstack-tox-py35 finger://ze02.openstack.org/2253ca6affc249f19ead433342a6a7d6 : POST_FAILURE in 2m 35s\n- openstack-tox-lower-constraints finger://ze02.openstack.org/357fe397533b4d14bf807cf1564b1c6d : POST_FAILURE in 3m 39s","accounts_in_message":[],"_revision_number":1},{"id":"7fc8d2ab5787791666a9eebd2b2f79f0bf1e73fd","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2019-08-04 04:00:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d2841e854fe5e18cf9e6290357eba8dd403a2fcf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-04 04:49:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover finger://ze02.openstack.org/0b2d55eb1ec64f96a8ef60a442e84731 : RETRY_LIMIT in 2m 51s\n- openstack-tox-pep8 finger://ze07.openstack.org/4d69f21e08554356a5eacfb4c89413b9 : RETRY_LIMIT in 3m 36s\n- openstack-tox-py27 finger://ze03.openstack.org/8323a927f57d4e6781980f4ea3241eef : RETRY_LIMIT in 3m 12s\n- openstack-tox-py36 finger://ze02.openstack.org/4b9a9b7000d84b4a8cacab2f17c1e35e : RETRY_LIMIT in 4m 13s\n- openstack-tox-py37 finger://ze05.openstack.org/35d0a27fa4c74f1d8d47ddb957ef497f : RETRY_LIMIT in 3m 29s\n- openstack-tox-docs finger://ze02.openstack.org/629e49145245435f9ffcc288deadcbb8 : RETRY_LIMIT in 3m 18s\n- tripleo-ci-centos-7-containers-multinode tripleo-ci-centos-7-containers-multinode : SKIPPED\n- tripleo-ci-centos-7-undercloud-containers tripleo-ci-centos-7-undercloud-containers : SKIPPED\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades : SKIPPED (non-voting)\n- tripleo-ci-centos-7-standalone tripleo-ci-centos-7-standalone : SKIPPED\n- tripleo-ci-fedora-28-standalone tripleo-ci-fedora-28-standalone : SKIPPED (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade tripleo-ci-centos-7-standalone-upgrade : SKIPPED (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades tripleo-ci-centos-7-containerized-undercloud-upgrades : SKIPPED (non-voting)\n- openstack-tox-linters finger://ze02.openstack.org/60e9b6286d254549ad6879bf20796e90 : RETRY_LIMIT in 3m 49s\n- openstack-tox-py35 finger://ze03.openstack.org/ed80467edfed44168eabb3fdff3153cf : RETRY_LIMIT in 2m 50s\n- openstack-tox-lower-constraints finger://ze03.openstack.org/bfc36bf76522417d80f00397815829ea : RETRY_LIMIT in 5m 02s","accounts_in_message":[],"_revision_number":2},{"id":"f765bf275333ae8777c67d710358463f8417bfec","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-04 06:48:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/76/674476/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/5608502/ : SUCCESS in 2h 34m 46s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/76/674476/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/48e619f/ : SUCCESS in 2h 42m 18s","accounts_in_message":[],"_revision_number":2},{"id":"5ebe3a07378204ef2a763e9692794c19ffc49714","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2019-08-04 14:06:13.000000000","message":"Patch Set 2:\n\nrecheck ci fixed","accounts_in_message":[],"_revision_number":2},{"id":"3d76f677e4c3f8fe3316f1a2ba985b67022cedb7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-04 16:42:28.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://logs.opendev.org/76/674476/2/check/openstack-tox-cover/be02be3/cover/ : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://logs.opendev.org/76/674476/2/check/openstack-tox-pep8/a10a332/ : SUCCESS in 2m 28s\n- openstack-tox-py27 https://logs.opendev.org/76/674476/2/check/openstack-tox-py27/8200b9d/ : SUCCESS in 2m 59s\n- openstack-tox-py36 https://logs.opendev.org/76/674476/2/check/openstack-tox-py36/033cfce/ : SUCCESS in 2m 56s\n- openstack-tox-py37 https://logs.opendev.org/76/674476/2/check/openstack-tox-py37/8304a90/ : SUCCESS in 3m 12s\n- openstack-tox-docs https://logs.opendev.org/76/674476/2/check/openstack-tox-docs/cf77149/html/ : SUCCESS in 2m 40s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/76/674476/2/check/tripleo-ci-centos-7-containers-multinode/21070dd/ : FAILURE in 13m 03s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/76/674476/2/check/tripleo-ci-centos-7-undercloud-containers/056c93f/ : FAILURE in 9m 56s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/76/674476/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/99f9121/ : SUCCESS in 2h 25m 56s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/76/674476/2/check/tripleo-ci-centos-7-standalone/af7650a/ : SUCCESS in 1h 46m 47s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/76/674476/2/check/tripleo-ci-fedora-28-standalone/b4c490a/ : FAILURE in 1h 03m 49s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/76/674476/2/check/tripleo-ci-centos-7-standalone-upgrade/e90ad30/ : FAILURE in 41m 12s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/76/674476/2/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/9aeee01/ : FAILURE in 55m 05s (non-voting)\n- openstack-tox-linters https://logs.opendev.org/76/674476/2/check/openstack-tox-linters/fbea92a/ : SUCCESS in 3m 34s\n- openstack-tox-py35 https://logs.opendev.org/76/674476/2/check/openstack-tox-py35/915c5ad/ : SUCCESS in 2m 55s\n- openstack-tox-lower-constraints https://logs.opendev.org/76/674476/2/check/openstack-tox-lower-constraints/6c45aac/ : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":2},{"id":"95fe2a0204cd35f749cbdc256f8e09cb1dc3e4a1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-04 16:54:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/76/674476/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/6088693/ : SUCCESS in 2h 33m 49s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/76/674476/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/1c8bbb3/ : SUCCESS in 2h 47m 29s","accounts_in_message":[],"_revision_number":2},{"id":"281f049255d662b0716698caa629023f22b9ddaa","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2019-08-04 20:47:13.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"05430b1c73c058bbad94b6c537a2f82f65728895","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2019-08-05 02:11:54.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"52f22793599443edfcc266f9d129f4c18329d02a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-05 05:07:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/76/674476/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/10751f6/ : SUCCESS in 2h 32m 54s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/76/674476/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/8633146/ : SUCCESS in 2h 54m 33s","accounts_in_message":[],"_revision_number":2},{"id":"c57529d347c67bb3cf55f443e177bddefda6c8b0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 05:34:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://logs.opendev.org/76/674476/2/check/openstack-tox-cover/d42d3f0/cover/ : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://logs.opendev.org/76/674476/2/check/openstack-tox-pep8/3b9fa9b/ : SUCCESS in 3m 54s\n- openstack-tox-py27 https://logs.opendev.org/76/674476/2/check/openstack-tox-py27/1cc5c5a/ : SUCCESS in 4m 10s\n- openstack-tox-py36 https://logs.opendev.org/76/674476/2/check/openstack-tox-py36/fd5bb50/ : SUCCESS in 4m 17s\n- openstack-tox-py37 https://logs.opendev.org/76/674476/2/check/openstack-tox-py37/ec279c1/ : SUCCESS in 4m 59s\n- openstack-tox-docs https://logs.opendev.org/76/674476/2/check/openstack-tox-docs/486dd73/html/ : SUCCESS in 3m 47s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/76/674476/2/check/tripleo-ci-centos-7-containers-multinode/0b55128/ : SUCCESS in 2h 09m 06s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/76/674476/2/check/tripleo-ci-centos-7-undercloud-containers/32d8571/ : SUCCESS in 1h 25m 34s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/76/674476/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/77c98d2/ : SUCCESS in 3h 02m 59s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/76/674476/2/check/tripleo-ci-centos-7-standalone/45b7642/ : SUCCESS in 2h 19m 59s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/76/674476/2/check/tripleo-ci-fedora-28-standalone/5e47bfe/ : FAILURE in 46m 11s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/76/674476/2/check/tripleo-ci-centos-7-standalone-upgrade/72582ba/ : FAILURE in 53m 51s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/76/674476/2/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/81a3401/ : FAILURE in 1h 20m 17s (non-voting)\n- openstack-tox-linters https://logs.opendev.org/76/674476/2/check/openstack-tox-linters/9ddb432/ : SUCCESS in 4m 35s\n- openstack-tox-py35 https://logs.opendev.org/76/674476/2/check/openstack-tox-py35/769eee4/ : SUCCESS in 4m 29s\n- openstack-tox-lower-constraints https://logs.opendev.org/76/674476/2/check/openstack-tox-lower-constraints/3f2f8d4/ : SUCCESS in 5m 57s","accounts_in_message":[],"_revision_number":2},{"id":"e7301b056337c9ea32420ab92771ee0c36651207","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-08-05 08:55:39.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"748ec946bb77eb76c4a73a0a279f492a654aea24","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-08-05 08:56:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"aefc4e990238f4a2c8146442fe80b47e967dd972","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2019-08-05 09:01:15.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"eee614dd188d4259643f70cb4c877de1046f1b15","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-08-05 09:04:09.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nthanks!","accounts_in_message":[],"_revision_number":2},{"id":"583adbed81d555f919c96457e0eb7b40a0743c4e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 09:04:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8999accb9b16b5ad583a2df2b98de4e6dfea42ff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 11:38:47.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://logs.opendev.org/76/674476/2/gate/openstack-tox-pep8/63e54fb/ : SUCCESS in 3m 51s\n- openstack-tox-py27 https://logs.opendev.org/76/674476/2/gate/openstack-tox-py27/49bf1b3/ : SUCCESS in 3m 37s\n- openstack-tox-py36 https://logs.opendev.org/76/674476/2/gate/openstack-tox-py36/c783495/ : SUCCESS in 4m 12s\n- openstack-tox-py37 https://logs.opendev.org/76/674476/2/gate/openstack-tox-py37/dd8547f/ : SUCCESS in 6m 15s\n- openstack-tox-docs https://logs.opendev.org/76/674476/2/gate/openstack-tox-docs/d08ca4e/html/ : SUCCESS in 3m 39s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/76/674476/2/gate/tripleo-ci-centos-7-containers-multinode/724c1fb/ : SUCCESS in 2h 27m 10s\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/76/674476/2/gate/tripleo-ci-centos-7-standalone/3bd856b/ : SUCCESS in 1h 59m 29s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/76/674476/2/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/4e934e6/ : SUCCESS in 2h 24m 36s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/76/674476/2/gate/tripleo-ci-centos-7-undercloud-containers/ae0cde5/ : SUCCESS in 1h 49m 57s\n- openstack-tox-linters https://logs.opendev.org/76/674476/2/gate/openstack-tox-linters/766e92c/ : SUCCESS in 5m 07s\n- openstack-tox-lower-constraints https://logs.opendev.org/76/674476/2/gate/openstack-tox-lower-constraints/6597070/ : SUCCESS in 7m 22s","accounts_in_message":[],"_revision_number":2},{"id":"317a857ad65722521fd6c954c9484e722f8ef328","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 11:38:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"8cb9b9f9789b352300cadc03502cf82be3caa2ba","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-05 12:08:49.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 142ae92f8ba20e1dfca811182d92f74553cb546d","accounts_in_message":[],"_revision_number":2},{"id":"5bccf845798f47447c9b05ed94b477a42a36b340","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-05 12:09:07.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit 652aad411fe217247b5612b533a9a5a1ff1aa013","accounts_in_message":[],"_revision_number":2},{"id":"c018ac01361b0315c2b192205b1432660f20c28c","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-05 12:09:20.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/queens as commit e3ca555245cf2a6eec22d3473a3ea2dc4538bbb6","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3df3719e82bedb0a167d5b981668403e2bd4e316","revisions":{"5091d87e305053af15106a17c24d3b0414ea6526":{"kind":"REWORK","_number":1,"created":"2019-08-04 00:51:38.000000000","uploader":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"ref":"refs/changes/76/674476/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/76/674476/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/76/674476/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/76/674476/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/76/674476/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/76/674476/1"}}},"commit":{"parents":[{"commit":"e4cea93c4ce3193974338d161fb8b190c25da679","subject":"Merge \"support conductor group feature in ironic conductor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/e4cea93c4ce3193974338d161fb8b190c25da679"}]}],"author":{"name":"Rabi Mishra","email":"ramishra@redhat.com","date":"2019-08-04 00:03:19.000000000","tz":330},"committer":{"name":"Rabi Mishra","email":"ramishra@redhat.com","date":"2019-08-04 00:50:14.000000000","tz":330},"subject":"Fix expensive api call to get deployment_name","message":"Fix expensive api call to get deployment_name\n\nWe filter all resources for overcloud root stack with a\nnon-indexed field \u0027physical_resource_id\u0027 (nova_instance).\nThis is very inefficient.\n\nAfter this, though there would still be heat api calls\nfor every deployment, it won\u0027t be so expensive.\n\nChange-Id: I260501e6061c0d6652b22504e7284940b2c64709\nCloses-Bug: #1838861\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/5091d87e305053af15106a17c24d3b0414ea6526"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/5091d87e305053af15106a17c24d3b0414ea6526"}]},"branch":"refs/heads/master"},"3df3719e82bedb0a167d5b981668403e2bd4e316":{"kind":"REWORK","_number":2,"created":"2019-08-04 04:00:32.000000000","uploader":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"ref":"refs/changes/76/674476/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/76/674476/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/76/674476/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/76/674476/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/76/674476/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/76/674476/2"}}},"commit":{"parents":[{"commit":"e4cea93c4ce3193974338d161fb8b190c25da679","subject":"Merge \"support conductor group feature in ironic conductor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/e4cea93c4ce3193974338d161fb8b190c25da679"}]}],"author":{"name":"Rabi Mishra","email":"ramishra@redhat.com","date":"2019-08-04 00:03:19.000000000","tz":330},"committer":{"name":"Rabi Mishra","email":"ramishra@redhat.com","date":"2019-08-04 03:59:03.000000000","tz":330},"subject":"Fix expensive api call to get deployment_name","message":"Fix expensive api call to get deployment_name\n\nWe filter all resources for overcloud root stack with a\nnon-indexed field \u0027physical_resource_id\u0027 (nova_instance).\nThis is very inefficient.\n\nWith this patch, though there would still be making heat api\ncalls for every deployment, it won\u0027t be so expensive.\n\nChange-Id: I260501e6061c0d6652b22504e7284940b2c64709\nCloses-Bug: #1838861\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/3df3719e82bedb0a167d5b981668403e2bd4e316"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/3df3719e82bedb0a167d5b981668403e2bd4e316"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
