)]}'
{"id":"openstack%2Ftripleo-common~674584","triplet_id":"openstack%2Ftripleo-common~stable%2Fqueens~I260501e6061c0d6652b22504e7284940b2c64709","project":"openstack/tripleo-common","branch":"stable/queens","hashtags":[],"change_id":"I260501e6061c0d6652b22504e7284940b2c64709","subject":"Fix expensive api call to get deployment_name","status":"MERGED","created":"2019-08-05 12:09:20.000000000","updated":"2019-08-06 16:41:54.000000000","submitted":"2019-08-06 16:41:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674584-1565109714127-85351d20","meta_rev_id":"3c07d7bec5c0b81aa73f8d9ce2f7818cd9b557ae","_number":674584,"virtual_id_number":674584,"owner":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-08-05 17:59:24.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-08-06 16:41:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-06 11:44:48.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":2,"date":"2019-08-06 14:17:44.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2019-08-06 09:09:16.000000000","_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":1,"date":"2019-08-06 14:17:44.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-06 09:09:16.000000000","updated_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"reviewer":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"state":"REVIEWER"},{"updated":"2019-08-06 11:44:48.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-06 14:17:44.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-08-06 16:41:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c8b08fda49805cd8fd3dd0ade6d3ba09037a5834","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-05 12:09:20.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"13df6810f755fb176747e87d98784868a4dd2e16","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 14:11:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-oooq https://logs.opendev.org/84/674584/1/check/tripleo-ci-centos-7-undercloud-oooq/b4ef3db/ : SUCCESS in 1h 23m 51s\n- tripleo-ci-centos-7-undercloud-upgrades https://logs.opendev.org/84/674584/1/check/tripleo-ci-centos-7-undercloud-upgrades/1954fb2/ : SUCCESS in 1h 20m 05s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/84/674584/1/check/tripleo-ci-centos-7-containers-multinode/1b0f6db/ : SUCCESS in 2h 00m 15s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/84/674584/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/9f3d060/ : FAILURE in 1h 47m 23s (non-voting)\n- build-openstack-sphinx-docs https://logs.opendev.org/84/674584/1/check/build-openstack-sphinx-docs/73c3d59/html/ : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://logs.opendev.org/84/674584/1/check/openstack-tox-pep8/770df86/ : SUCCESS in 3m 10s\n- openstack-tox-py27 https://logs.opendev.org/84/674584/1/check/openstack-tox-py27/37fb5d7/ : SUCCESS in 4m 09s\n- openstack-tox-py35 https://logs.opendev.org/84/674584/1/check/openstack-tox-py35/d641a51/ : SUCCESS in 3m 10s\n- openstack-tox-cover https://logs.opendev.org/84/674584/1/check/openstack-tox-cover/0fad1f1/cover/ : SUCCESS in 3m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5aca5c35e7b6f24317eaef6dca34d286cf8f64fe","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-05 14:37:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/84/674584/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/3a22f33/ : FAILURE in 2h 26m 29s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/84/674584/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/330ece1/ : SUCCESS in 2h 16m 51s","accounts_in_message":[],"_revision_number":1},{"id":"417c885d79527b0d101f0857893dbb0cd9845442","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-05 15:17:54.000000000","message":"Patch Set 1:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":1},{"id":"8c63ee28e06224648c3bea5c42906bb538d0b76e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-05 17:41:02.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/84/674584/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/bf75724/ : FAILURE in 2h 17m 40s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/84/674584/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/b0f1196/ : SUCCESS in 2h 21m 19s","accounts_in_message":[],"_revision_number":1},{"id":"f63af2fea78fc81529ce93fddbbd3c9fc30cfea0","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-05 17:59:24.000000000","message":"Patch Set 1:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":1},{"id":"39c05fbcc578a161a1833da0717dd12c7a955a63","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-05 20:40:41.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/84/674584/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/e0ba56f/ : FAILURE in 2h 39m 38s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/84/674584/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/a866af3/ : SUCCESS in 2h 17m 54s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/84/674584/1/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/ae8980a/ : RETRY_LIMIT in 6m 54s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/84/674584/1/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/2afab6e/ : RETRY_LIMIT in 6m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"97b7bd8d705c6f2d1c6f7b38ecb0c7dddd0664d2","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2019-08-06 09:09:16.000000000","message":"Patch Set 1:\n\nrecheck rdo","accounts_in_message":[],"_revision_number":1},{"id":"6b559930c965f69d6223be8d788b897990d931dd","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-06 11:44:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/84/674584/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/7cd0bb0/ : SUCCESS in 2h 34m 00s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/84/674584/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/1c9d037/ : SUCCESS in 2h 18m 18s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/84/674584/1/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/6053827/ : RETRY_LIMIT in 6m 46s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/84/674584/1/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/9f82395/ : RETRY_LIMIT in 6m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7866687b75c3f66530958c52fe26b38d57563fe9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 13:36:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-oooq https://logs.opendev.org/84/674584/1/check/tripleo-ci-centos-7-undercloud-oooq/9fd610a/ : SUCCESS in 1h 17m 21s\n- tripleo-ci-centos-7-undercloud-upgrades https://logs.opendev.org/84/674584/1/check/tripleo-ci-centos-7-undercloud-upgrades/37724f3/ : SUCCESS in 1h 17m 15s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/84/674584/1/check/tripleo-ci-centos-7-containers-multinode/4629b18/ : SUCCESS in 2h 18m 04s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/84/674584/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/ff363ba/ : FAILURE in 1h 53m 24s (non-voting)\n- build-openstack-sphinx-docs https://logs.opendev.org/84/674584/1/check/build-openstack-sphinx-docs/a1596fd/html/ : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://logs.opendev.org/84/674584/1/check/openstack-tox-pep8/1a2d37f/ : SUCCESS in 4m 12s\n- openstack-tox-py27 https://logs.opendev.org/84/674584/1/check/openstack-tox-py27/84c6ab5/ : SUCCESS in 3m 53s\n- openstack-tox-py35 https://logs.opendev.org/84/674584/1/check/openstack-tox-py35/032977e/ : SUCCESS in 4m 23s\n- openstack-tox-cover https://logs.opendev.org/84/674584/1/check/openstack-tox-cover/8104e21/cover/ : SUCCESS in 4m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c23e3cadcf7e5a7cb0c6c08479d002b8ae1d7976","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-06 14:17:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"792b36fa519c969880e64b350e09c90ba9f380bd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 14:18:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9fb24da6837ec18857ac6c8c728dd9c2f09ea8d8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 16:41:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-undercloud-oooq https://logs.opendev.org/84/674584/1/gate/tripleo-ci-centos-7-undercloud-oooq/b8a108a/ : SUCCESS in 1h 10m 51s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/84/674584/1/gate/tripleo-ci-centos-7-containers-multinode/495355d/ : SUCCESS in 2h 17m 46s\n- build-openstack-sphinx-docs https://logs.opendev.org/84/674584/1/gate/build-openstack-sphinx-docs/38adf30/html/ : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://logs.opendev.org/84/674584/1/gate/openstack-tox-pep8/18ae998/ : SUCCESS in 5m 44s\n- openstack-tox-py27 https://logs.opendev.org/84/674584/1/gate/openstack-tox-py27/b334424/ : SUCCESS in 5m 11s\n- openstack-tox-py35 https://logs.opendev.org/84/674584/1/gate/openstack-tox-py35/c1c52b6/ : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":1},{"id":"3c07d7bec5c0b81aa73f8d9ce2f7818cd9b557ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 16:41:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e3ca555245cf2a6eec22d3473a3ea2dc4538bbb6","revisions":{"e3ca555245cf2a6eec22d3473a3ea2dc4538bbb6":{"kind":"REWORK","_number":1,"created":"2019-08-05 12:09:20.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/84/674584/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/84/674584/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/84/674584/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/84/674584/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/84/674584/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/84/674584/1"}}},"commit":{"parents":[{"commit":"341b3cbb48fc2df2740595436a5156f90d79190f","subject":"Merge \"Ensure that OctaviaServerCertsKeyPassphrase is 32-byte long\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/341b3cbb48fc2df2740595436a5156f90d79190f"}]}],"author":{"name":"Rabi Mishra","email":"ramishra@redhat.com","date":"2019-08-04 00:03:19.000000000","tz":330},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-08-05 12:09:20.000000000","tz":0},"subject":"Fix expensive api call to get deployment_name","message":"Fix expensive api call to get deployment_name\n\nWe filter all resources for overcloud root stack with a\nnon-indexed field \u0027physical_resource_id\u0027 (nova_instance).\nThis is very inefficient.\n\nWith this patch, though there would still be making heat api\ncalls for every deployment, it won\u0027t be so expensive.\n\nChange-Id: I260501e6061c0d6652b22504e7284940b2c64709\nCloses-Bug: #1838861\n(cherry picked from commit 3df3719e82bedb0a167d5b981668403e2bd4e316)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/e3ca555245cf2a6eec22d3473a3ea2dc4538bbb6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/e3ca555245cf2a6eec22d3473a3ea2dc4538bbb6"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
