)]}'
{"id":"openstack%2Ftripleo-common~674801","triplet_id":"openstack%2Ftripleo-common~master~I6f3e4fb279146a0f292818929092d1aeb1d3fced","project":"openstack/tripleo-common","branch":"master","topic":"in-flight","hashtags":[],"change_id":"I6f3e4fb279146a0f292818929092d1aeb1d3fced","subject":"Allows to disable in-flight validations","status":"MERGED","created":"2019-08-06 12:50:23.000000000","updated":"2019-08-07 21:06:14.000000000","submitted":"2019-08-07 21:06:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674801-1565211974801-62203689","meta_rev_id":"28de1724d9ff76f0d55ca28bd1977d3ce1c24343","_number":674801,"virtual_id_number":674801,"owner":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-08-07 21:06:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-06 16:02:47.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-08-07 08:25:57.000000000","_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":1,"date":"2019-08-06 12:57:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":2,"date":"2019-08-06 13:35:16.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"date":"2019-08-06 12:56:39.000000000","_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"date":"2019-08-06 12:56:36.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":1,"date":"2019-08-07 13:58:37.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-06 12:56:36.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-08-06 12:56:39.000000000","updated_by":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"reviewer":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"state":"REVIEWER"},{"updated":"2019-08-06 12:57:35.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2019-08-06 16:02:47.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-07 08:25:57.000000000","updated_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"reviewer":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"state":"REVIEWER"},{"updated":"2019-08-07 13:58:37.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-08-07 21:06:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"378a9e5d6dc710e7191e5c3886bc893df3f5c633","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-08-06 12:50:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"15c0bbca6adc56d7506ddcb70689a262b871578e","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-08-06 12:57:35.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e663f06a2dfc2a060f8d5902098dfd24db59bd3d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-06 13:35:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"369e6593459f039a8053ea22903407016029c08c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-06 16:02:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/01/674801/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/4553682/ : SUCCESS in 3h 10m 35s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/01/674801/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/39b3e28/ : SUCCESS in 2h 56m 11s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/01/674801/1/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/e116149/ : FAILURE in 9m 36s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/01/674801/1/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/44bd3d6/ : FAILURE in 9m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1dd1c0928b777237cbe3960d1ce64dbad9b12d62","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 17:02:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://logs.opendev.org/01/674801/1/check/openstack-tox-cover/14579fc/cover/ : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://logs.opendev.org/01/674801/1/check/openstack-tox-pep8/9650a41/ : SUCCESS in 3m 57s\n- openstack-tox-py27 https://logs.opendev.org/01/674801/1/check/openstack-tox-py27/b700c8c/ : SUCCESS in 5m 13s\n- openstack-tox-py36 https://logs.opendev.org/01/674801/1/check/openstack-tox-py36/4d3f599/ : SUCCESS in 4m 45s\n- openstack-tox-py37 https://logs.opendev.org/01/674801/1/check/openstack-tox-py37/4f4f443/ : SUCCESS in 5m 16s\n- openstack-tox-docs https://logs.opendev.org/01/674801/1/check/openstack-tox-docs/1883060/html/ : SUCCESS in 4m 25s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/01/674801/1/check/tripleo-ci-centos-7-containers-multinode/38d8a0c/ : SUCCESS in 2h 25m 21s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/01/674801/1/check/tripleo-ci-centos-7-undercloud-containers/8d10de8/ : SUCCESS in 1h 42m 53s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/01/674801/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/75a831c/ : SUCCESS in 2h 49m 14s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/01/674801/1/check/tripleo-ci-centos-7-standalone/d8a15ef/ : SUCCESS in 1h 48m 55s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/01/674801/1/check/tripleo-ci-fedora-28-standalone/5f695ed/ : FAILURE in 59m 42s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/01/674801/1/check/tripleo-ci-centos-7-standalone-upgrade/5c3bb79/ : SUCCESS in 2h 13m 10s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/01/674801/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/7cf95e8/ : FAILURE in 2h 23m 24s (non-voting)\n- openstack-tox-linters https://logs.opendev.org/01/674801/1/check/openstack-tox-linters/4c05930/ : SUCCESS in 5m 59s\n- openstack-tox-py35 https://logs.opendev.org/01/674801/1/check/openstack-tox-py35/3170b06/ : SUCCESS in 6m 16s\n- openstack-tox-lower-constraints https://logs.opendev.org/01/674801/1/check/openstack-tox-lower-constraints/127695e/ : SUCCESS in 7m 21s","accounts_in_message":[],"_revision_number":1},{"id":"3ec7a8ce3c4d49faa49b6735cecaed45fbcb4332","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-07 08:25:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bd7333b6e36b75a3e40826bf9de33626354b7d91","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-07 13:58:37.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"93757c574b7bf3b34e41893348dc7e78377cf27e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 13:59:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"42c63051c5b06ea33516f90ba20978644ee936ff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 21:06:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://logs.opendev.org/01/674801/1/gate/openstack-tox-pep8/890ecd5/ : SUCCESS in 2m 56s\n- openstack-tox-py27 https://logs.opendev.org/01/674801/1/gate/openstack-tox-py27/6c1a52e/ : SUCCESS in 4m 22s\n- openstack-tox-py36 https://logs.opendev.org/01/674801/1/gate/openstack-tox-py36/6a392fb/ : SUCCESS in 3m 39s\n- openstack-tox-py37 https://logs.opendev.org/01/674801/1/gate/openstack-tox-py37/0cd4344/ : SUCCESS in 5m 01s\n- openstack-tox-docs https://logs.opendev.org/01/674801/1/gate/openstack-tox-docs/d1ab8a8/html/ : SUCCESS in 3m 52s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/01/674801/1/gate/tripleo-ci-centos-7-containers-multinode/c28a7f1/ : SUCCESS in 2h 47m 16s\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/01/674801/1/gate/tripleo-ci-centos-7-standalone/656af61/ : SUCCESS in 1h 52m 04s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/01/674801/1/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/9c9acc3/ : SUCCESS in 2h 19m 40s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/01/674801/1/gate/tripleo-ci-centos-7-undercloud-containers/7e26775/ : SUCCESS in 1h 50m 00s\n- openstack-tox-linters https://logs.opendev.org/01/674801/1/gate/openstack-tox-linters/55dad82/ : SUCCESS in 4m 16s\n- openstack-tox-lower-constraints https://logs.opendev.org/01/674801/1/gate/openstack-tox-lower-constraints/b48f91e/ : SUCCESS in 5m 50s","accounts_in_message":[],"_revision_number":1},{"id":"866c831c9eaf9b9227956afee1c84ccbb66e3c79","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 21:06:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"04b106cb97aea0a5a09d6fcddc5e42a42a543e69","revisions":{"04b106cb97aea0a5a09d6fcddc5e42a42a543e69":{"kind":"REWORK","_number":1,"created":"2019-08-06 12:50:23.000000000","uploader":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"ref":"refs/changes/01/674801/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/01/674801/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/01/674801/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/01/674801/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/01/674801/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/01/674801/1"}}},"commit":{"parents":[{"commit":"a1bd494fc79d3ad06c53a9c324ce293ce2803b15","subject":"Merge \"Add service_configs template to tripleo-hieradata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/a1bd494fc79d3ad06c53a9c324ce293ce2803b15"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-08-06 08:05:58.000000000","tz":120},"committer":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-08-06 12:50:07.000000000","tz":120},"subject":"Allows to disable in-flight validations","message":"Allows to disable in-flight validations\n\nThis patches adds the possibility to deactivate in-flight validations by\nskipping the \"opendev-validation\" ansible tag.\n\nIt does it in a generic way, meaning we might pass other tags to skip.\n\nChange-Id: I6f3e4fb279146a0f292818929092d1aeb1d3fced\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/04b106cb97aea0a5a09d6fcddc5e42a42a543e69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/04b106cb97aea0a5a09d6fcddc5e42a42a543e69"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
