)]}'
{"id":"openstack%2Ftripleo-common~674822","triplet_id":"openstack%2Ftripleo-common~stable%2Fstein~Ie9880d7b6702b3c5f0e407e4ffe6f297912a1d10","project":"openstack/tripleo-common","branch":"stable/stein","topic":"ceph-ansible-inventory-stable/stein","hashtags":[],"change_id":"Ie9880d7b6702b3c5f0e407e4ffe6f297912a1d10","subject":"Add ceph-ansible support to tripleo-ansible-inventory","status":"MERGED","created":"2019-08-06 14:52:22.000000000","updated":"2019-08-07 21:06:08.000000000","submitted":"2019-08-07 21:06:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674822-1565211968645-e657b839","meta_rev_id":"37f35f908821dbf52fb6cd5b5b2703f962ab0864","_number":674822,"virtual_id_number":674822,"owner":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"value":2,"date":"2019-08-07 21:06:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-06 17:51:11.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-08-07 13:56:02.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"value":1,"date":"2019-08-07 06:40:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":2,"date":"2019-08-06 15:10:03.000000000","_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-08-07 13:56:02.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2019-08-06 15:04:45.000000000","_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-06 15:10:03.000000000","updated_by":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"reviewer":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"state":"REVIEWER"},{"updated":"2019-08-06 17:51:11.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-07 06:40:14.000000000","updated_by":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"reviewer":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"state":"REVIEWER"},{"updated":"2019-08-07 13:56:02.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-08-07 21:06:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c660dbac516d16a28a84711dcffd7c5b604bf74b","author":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"date":"2019-08-06 14:52:22.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"1f01083de5e9171ba5663c3df0e1d102f7972c41","author":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"date":"2019-08-06 14:52:42.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"ca539dd9a23d4c51025812016accdb1bca20adfa","author":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"date":"2019-08-06 15:04:45.000000000","message":"Patch Set 1: -Workflow\n\nThis will result in new entries in the inventory which are benign and shouldn\u0027t hurt stein users. I don\u0027t intend to change the inventory which is used for the ceph-ansible run, but I do want to give stein users a ceph-ansible-compatible inventory for when they upgrade to stein and then need to run ceph-ansible playbooks directly for their move to bluestore when they use ceph nautilus. Waiting for CI results next.","accounts_in_message":[],"_revision_number":1},{"id":"e237813d70cb97082b4b78ec39bef5dab397ac1f","author":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"date":"2019-08-06 15:10:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"692b14b41ef5411f1111e49f50329ebb0409c2a1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-06 17:51:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/22/674822/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/a3192a7/ : SUCCESS in 2h 57m 46s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/22/674822/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/e45302d/ : SUCCESS in 2h 50m 25s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/22/674822/1/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/3a91744/ : RETRY_LIMIT in 6m 01s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/22/674822/1/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/c7d5037/ : RETRY_LIMIT in 6m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"df619ee85eb39f10c67cdbdfa5458fe09b3860ba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 21:39:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://logs.opendev.org/22/674822/1/check/openstack-tox-cover/cee6ac7/cover/ : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://logs.opendev.org/22/674822/1/check/openstack-tox-pep8/d254a21/ : SUCCESS in 4m 21s\n- openstack-tox-py27 https://logs.opendev.org/22/674822/1/check/openstack-tox-py27/05b6301/ : SUCCESS in 5m 29s\n- openstack-tox-py36 https://logs.opendev.org/22/674822/1/check/openstack-tox-py36/e3b2813/ : SUCCESS in 4m 27s\n- openstack-tox-py37 https://logs.opendev.org/22/674822/1/check/openstack-tox-py37/75e8b7d/ : SUCCESS in 5m 02s\n- openstack-tox-docs https://logs.opendev.org/22/674822/1/check/openstack-tox-docs/e3880d5/html/ : SUCCESS in 4m 21s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/22/674822/1/check/tripleo-ci-centos-7-containers-multinode/fb5876b/ : SUCCESS in 2h 07m 26s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/22/674822/1/check/tripleo-ci-centos-7-undercloud-containers/7a8ac60/ : SUCCESS in 1h 24m 12s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/22/674822/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/7c35b47/ : FAILURE in 2h 13m 33s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/22/674822/1/check/tripleo-ci-centos-7-standalone/d7fba98/ : SUCCESS in 1h 54m 28s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/22/674822/1/check/tripleo-ci-fedora-28-standalone/bdb3512/ : SUCCESS in 1h 45m 02s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/22/674822/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/4119af2/ : FAILURE in 1h 36m 16s (non-voting)\n- openstack-tox-py35 https://logs.opendev.org/22/674822/1/check/openstack-tox-py35/42f358e/ : SUCCESS in 4m 23s\n- openstack-tox-lower-constraints https://logs.opendev.org/22/674822/1/check/openstack-tox-lower-constraints/25f7165/ : SUCCESS in 5m 49s","accounts_in_message":[],"_revision_number":1},{"id":"7918079c84dcd396bd6e99aaf7b76937000e1315","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2019-08-07 06:40:14.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"350555cfb09a1d3228a197185c5c3c4f546d9eed","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-07 13:56:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f1584119fd0646d6179892c3a2cfe960b216f85c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 13:56:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a261f9834e4372bae7d8896d7bd30f25240373d8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 21:06:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://logs.opendev.org/22/674822/1/gate/openstack-tox-pep8/d17f937/ : SUCCESS in 3m 55s\n- openstack-tox-py27 https://logs.opendev.org/22/674822/1/gate/openstack-tox-py27/9d2ff45/ : SUCCESS in 4m 03s\n- openstack-tox-py36 https://logs.opendev.org/22/674822/1/gate/openstack-tox-py36/d653508/ : SUCCESS in 4m 13s\n- openstack-tox-py37 https://logs.opendev.org/22/674822/1/gate/openstack-tox-py37/d812ff9/ : SUCCESS in 4m 13s\n- openstack-tox-docs https://logs.opendev.org/22/674822/1/gate/openstack-tox-docs/e5fb016/html/ : SUCCESS in 2m 53s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/22/674822/1/gate/tripleo-ci-centos-7-containers-multinode/5a3507a/ : SUCCESS in 2h 21m 27s\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/22/674822/1/gate/tripleo-ci-centos-7-standalone/72bd50d/ : SUCCESS in 2h 13m 47s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/22/674822/1/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/e09de3b/ : SUCCESS in 3h 06m 46s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/22/674822/1/gate/tripleo-ci-centos-7-undercloud-containers/f073315/ : SUCCESS in 2h 17m 37s\n- openstack-tox-lower-constraints https://logs.opendev.org/22/674822/1/gate/openstack-tox-lower-constraints/e58e2ca/ : SUCCESS in 7m 30s","accounts_in_message":[],"_revision_number":1},{"id":"f929efe6ae2fe182cb89bbb51040f956b6af340b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 21:06:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"65ab6278f1515fa13252af2832cbf3287bbc9932","revisions":{"65ab6278f1515fa13252af2832cbf3287bbc9932":{"kind":"REWORK","_number":1,"created":"2019-08-06 14:52:22.000000000","uploader":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"ref":"refs/changes/22/674822/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/22/674822/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/22/674822/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/22/674822/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/22/674822/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/22/674822/1"}}},"commit":{"parents":[{"commit":"b33182d4233b8b69e7dab0cbea8321d8bce8f6d5","subject":"Merge \"Fix ansible-generate-inventory action\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/b33182d4233b8b69e7dab0cbea8321d8bce8f6d5"}]}],"author":{"name":"John Fulton","email":"fulton@redhat.com","date":"2019-06-06 01:20:42.000000000","tz":-240},"committer":{"name":"John Fulton","email":"fulton@redhat.com","date":"2019-08-06 14:52:22.000000000","tz":0},"subject":"Add ceph-ansible support to tripleo-ansible-inventory","message":"Add ceph-ansible support to tripleo-ansible-inventory\n\nThe tripleo-ansible-inventory command produces an inventory\nwhich works with TripleO but not with ceph-ansible. TripleO\nceph-ansible integration then creates its own inventory for\nthe ceph-ansible playbook run. This patch makes generating\na separate inventory for ceph-ansible no longer necessary.\n\nTripleO services already map to the same inventory groups\nused by ceph-ansible, e.g. ceph_clients:clients, ceph_mons:\nmons, ceph_osds:osds, etc. This patch maps the names of the\nTripleO inventory Ceph groups to the names of the inventory\ngroups used by ceph-ansible and creates a duplicate group\nentry in the inventory, but with the names that ceph-ansible\nexpects.\n\nChange-Id: Ie9880d7b6702b3c5f0e407e4ffe6f297912a1d10\n(cherry picked from commit 3f48630791f5051b97afba1365d8f5defb797951)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/65ab6278f1515fa13252af2832cbf3287bbc9932"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/65ab6278f1515fa13252af2832cbf3287bbc9932"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
