)]}'
{"id":"openstack%2Ftripleo-common~674933","triplet_id":"openstack%2Ftripleo-common~master~Ia64ceb1f05d96d2358509392bb3cbc3b9cb61030","project":"openstack/tripleo-common","branch":"master","topic":"openshift/removal","hashtags":[],"change_id":"Ia64ceb1f05d96d2358509392bb3cbc3b9cb61030","subject":"Remove OpenShift containers","status":"MERGED","created":"2019-08-06 21:21:34.000000000","updated":"2019-08-22 00:57:03.000000000","submitted":"2019-08-22 00:54:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674933-1566435288204-0f3fa659","meta_rev_id":"1fc5e2b17d9b9a0878e25b5254565a74e961fea7","_number":674933,"virtual_id_number":674933,"owner":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":2,"date":"2019-08-22 00:54:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-08-21 21:00:20.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"recommended":{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2019-08-13 17:37:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"value":2,"date":"2019-08-12 15:10:41.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2019-08-14 06:24:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-08-21 14:06:13.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-12 15:10:41.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-08-13 17:37:06.000000000","updated_by":{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},"reviewer":{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},"state":"REVIEWER"},{"updated":"2019-08-14 06:24:41.000000000","updated_by":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"reviewer":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"state":"REVIEWER"},{"updated":"2019-08-21 21:00:20.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-22 00:54:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"502b4a9b313c51cf4c2df52bb58f9bb269106959","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-06 21:21:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d1a2fde4565e0d20cd182fc20a902f60dcebc71e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-06 21:43:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/33/674933/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/514af2e/ : FAILURE in 17m 45s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/33/674933/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/ac71315/ : FAILURE in 19m 02s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/33/674933/1/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/d9f723d/ : FAILURE in 8m 55s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/33/674933/1/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/bd0dd6c/ : FAILURE in 9m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3b1a343e556c3c47ae3c272a89b404f20a7549cb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 00:24:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://logs.opendev.org/33/674933/1/check/test-release-openstack/ad928c6/ : SUCCESS in 3m 14s\n- openstack-tox-cover https://logs.opendev.org/33/674933/1/check/openstack-tox-cover/a4c31dd/cover/ : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://logs.opendev.org/33/674933/1/check/openstack-tox-pep8/228012f/ : SUCCESS in 4m 13s\n- openstack-tox-py27 https://logs.opendev.org/33/674933/1/check/openstack-tox-py27/0c443c3/ : SUCCESS in 3m 42s\n- openstack-tox-py36 https://logs.opendev.org/33/674933/1/check/openstack-tox-py36/c26a2bb/ : SUCCESS in 4m 47s\n- openstack-tox-py37 https://logs.opendev.org/33/674933/1/check/openstack-tox-py37/25474a8/ : SUCCESS in 5m 13s\n- openstack-tox-docs https://logs.opendev.org/33/674933/1/check/openstack-tox-docs/cd63c32/html/ : SUCCESS in 4m 02s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/33/674933/1/check/tripleo-ci-centos-7-containers-multinode/ecbc78c/ : FAILURE in 26m 34s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/33/674933/1/check/tripleo-ci-centos-7-undercloud-containers/da664e9/ : FAILURE in 25m 49s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/33/674933/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/5210a2f/ : FAILURE in 31m 00s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/33/674933/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/25c2095/ : FAILURE in 29m 46s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/33/674933/1/check/tripleo-ci-centos-7-standalone/e9bfd03/ : FAILURE in 24m 36s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/33/674933/1/check/tripleo-ci-fedora-28-standalone/8ab7a6c/ : FAILURE in 29m 49s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/33/674933/1/check/tripleo-ci-centos-7-standalone-upgrade/f4cdc75/ : FAILURE in 1h 21m 02s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/33/674933/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/6c6f6bd/ : FAILURE in 1h 16m 10s (non-voting)\n- tripleo-build-containers-centos-7 https://logs.opendev.org/33/674933/1/check/tripleo-build-containers-centos-7/688209f/ : SUCCESS in 40m 47s\n- tripleo-build-containers-centos-7-buildah https://logs.opendev.org/33/674933/1/check/tripleo-build-containers-centos-7-buildah/8d4db7c/ : SUCCESS in 1h 02m 02s\n- tripleo-ci-centos-7-scenario001-standalone https://logs.opendev.org/33/674933/1/check/tripleo-ci-centos-7-scenario001-standalone/510aed3/ : FAILURE in 24m 42s\n- tripleo-ci-centos-7-scenario002-standalone https://logs.opendev.org/33/674933/1/check/tripleo-ci-centos-7-scenario002-standalone/b4adc0a/ : FAILURE in 24m 35s\n- tripleo-ci-centos-7-scenario003-standalone https://logs.opendev.org/33/674933/1/check/tripleo-ci-centos-7-scenario003-standalone/e4387e9/ : FAILURE in 21m 21s\n- tripleo-ci-centos-7-scenario004-standalone https://logs.opendev.org/33/674933/1/check/tripleo-ci-centos-7-scenario004-standalone/2e763e8/ : FAILURE in 26m 26s\n- tripleo-ci-centos-7-scenario010-standalone https://logs.opendev.org/33/674933/1/check/tripleo-ci-centos-7-scenario010-standalone/dfba6eb/ : FAILURE in 21m 58s\n- tripleo-ci-centos-7-scenario012-standalone https://logs.opendev.org/33/674933/1/check/tripleo-ci-centos-7-scenario012-standalone/4c2846c/ : FAILURE in 24m 37s (non-voting)\n- tripleo-ci-centos-7-scenario007-standalone https://logs.opendev.org/33/674933/1/check/tripleo-ci-centos-7-scenario007-standalone/0392f8c/ : FAILURE in 19m 24s (non-voting)\n- openstack-tox-linters https://logs.opendev.org/33/674933/1/check/openstack-tox-linters/fcb8822/ : SUCCESS in 5m 15s\n- openstack-tox-py35 https://logs.opendev.org/33/674933/1/check/openstack-tox-py35/8cf7c5a/ : SUCCESS in 4m 03s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://logs.opendev.org/33/674933/1/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/0ff8b89/ : FAILURE in 27m 35s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://logs.opendev.org/33/674933/1/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/a619e8c/ : FAILURE in 25m 36s (non-voting)\n- openstack-tox-lower-constraints https://logs.opendev.org/33/674933/1/check/openstack-tox-lower-constraints/e5e60a0/ : SUCCESS in 5m 17s","accounts_in_message":[],"_revision_number":1},{"id":"f7b9a2cb2f2e77d5c5e52d505f6f0339b4c79e7e","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-07 14:08:59.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"0ff7a8771636e46393af7a5dc7450c266d3176e8","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-07 14:32:10.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/33/674933/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/d677d5a/ : FAILURE in 21m 02s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/33/674933/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/41100fb/ : FAILURE in 20m 42s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/33/674933/2/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/3e5dbbe/ : FAILURE in 9m 01s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/33/674933/2/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/08c18a9/ : FAILURE in 9m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b67a1adad0471b21b3cb054a046cb75ae24d8dbb","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-07 15:19:46.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8a49f14273a1c8d21acec76bf68c41b3fac93ae5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-07 18:24:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/33/674933/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/1ed3fa9/ : SUCCESS in 3h 00m 36s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/33/674933/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/fe4681d/ : SUCCESS in 2h 51m 26s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/33/674933/3/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/0eddab4/ : FAILURE in 9m 37s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/33/674933/3/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/0dcbef1/ : FAILURE in 21m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e25718513aa72781c27aca6f1e0741389248ecab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 21:31:37.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://logs.opendev.org/33/674933/3/check/openstack-tox-cover/d311bb3/cover/ : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://logs.opendev.org/33/674933/3/check/openstack-tox-pep8/0263bdf/ : SUCCESS in 2m 53s\n- openstack-tox-py27 https://logs.opendev.org/33/674933/3/check/openstack-tox-py27/452f962/ : SUCCESS in 4m 24s\n- openstack-tox-py36 https://logs.opendev.org/33/674933/3/check/openstack-tox-py36/6a129fc/ : SUCCESS in 4m 09s\n- openstack-tox-py37 https://logs.opendev.org/33/674933/3/check/openstack-tox-py37/5e7e532/ : SUCCESS in 7m 17s\n- openstack-tox-docs https://logs.opendev.org/33/674933/3/check/openstack-tox-docs/a9876d5/html/ : SUCCESS in 4m 08s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/33/674933/3/check/tripleo-ci-centos-7-containers-multinode/15ef9a5/ : SUCCESS in 2h 35m 28s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/33/674933/3/check/tripleo-ci-centos-7-undercloud-containers/44b3c14/ : SUCCESS in 1h 55m 14s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/33/674933/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/10e0270/ : SUCCESS in 2h 47m 45s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/33/674933/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/127b9ae/ : SUCCESS in 2h 32m 37s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/33/674933/3/check/tripleo-ci-centos-7-standalone/4321ed4/ : SUCCESS in 1h 55m 56s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/33/674933/3/check/tripleo-ci-fedora-28-standalone/1512b6f/ : SUCCESS in 1h 58m 20s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/33/674933/3/check/tripleo-ci-centos-7-standalone-upgrade/dfd39cd/ : SUCCESS in 2h 22m 30s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/33/674933/3/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/79dad50/ : FAILURE in 1h 58m 47s (non-voting)\n- tripleo-build-containers-centos-7 https://logs.opendev.org/33/674933/3/check/tripleo-build-containers-centos-7/a4bc3bb/ : SUCCESS in 47m 51s\n- tripleo-build-containers-centos-7-buildah https://logs.opendev.org/33/674933/3/check/tripleo-build-containers-centos-7-buildah/cf089eb/ : SUCCESS in 51m 10s\n- tripleo-ci-centos-7-scenario001-standalone https://logs.opendev.org/33/674933/3/check/tripleo-ci-centos-7-scenario001-standalone/d91bd0a/ : SUCCESS in 1h 48m 46s\n- tripleo-ci-centos-7-scenario002-standalone https://logs.opendev.org/33/674933/3/check/tripleo-ci-centos-7-scenario002-standalone/e5b30cc/ : SUCCESS in 1h 52m 36s\n- tripleo-ci-centos-7-scenario003-standalone https://logs.opendev.org/33/674933/3/check/tripleo-ci-centos-7-scenario003-standalone/af050bd/ : SUCCESS in 1h 33m 21s\n- tripleo-ci-centos-7-scenario004-standalone https://logs.opendev.org/33/674933/3/check/tripleo-ci-centos-7-scenario004-standalone/d0fdaf7/ : SUCCESS in 1h 53m 12s\n- tripleo-ci-centos-7-scenario010-standalone https://logs.opendev.org/33/674933/3/check/tripleo-ci-centos-7-scenario010-standalone/9dc2397/ : FAILURE in 59m 22s\n- tripleo-ci-centos-7-scenario012-standalone https://logs.opendev.org/33/674933/3/check/tripleo-ci-centos-7-scenario012-standalone/f90cae1/ : SUCCESS in 1h 55m 35s (non-voting)\n- tripleo-ci-centos-7-scenario007-standalone https://logs.opendev.org/33/674933/3/check/tripleo-ci-centos-7-scenario007-standalone/c1bfc2d/ : SUCCESS in 1h 53m 30s (non-voting)\n- openstack-tox-linters https://logs.opendev.org/33/674933/3/check/openstack-tox-linters/7a71479/ : SUCCESS in 4m 11s\n- openstack-tox-py35 https://logs.opendev.org/33/674933/3/check/openstack-tox-py35/247e12b/ : SUCCESS in 5m 56s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://logs.opendev.org/33/674933/3/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/aa46e1e/ : TIMED_OUT in 3h 15m 33s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://logs.opendev.org/33/674933/3/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/07c9ee0/ : FAILURE in 1h 38m 34s (non-voting)\n- openstack-tox-lower-constraints https://logs.opendev.org/33/674933/3/check/openstack-tox-lower-constraints/7a964e2/ : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":3},{"id":"436a9c7e379f80e4afb91dab4143b552a042ba11","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-07 22:29:04.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"17cf75b750874b3db11c473544009e4d199d68f8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 01:45:09.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://logs.opendev.org/33/674933/3/check/openstack-tox-cover/9033902/cover/ : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://logs.opendev.org/33/674933/3/check/openstack-tox-pep8/15833d6/ : SUCCESS in 3m 17s\n- openstack-tox-py27 https://logs.opendev.org/33/674933/3/check/openstack-tox-py27/c91567b/ : SUCCESS in 4m 11s\n- openstack-tox-py36 https://logs.opendev.org/33/674933/3/check/openstack-tox-py36/2222935/ : SUCCESS in 4m 02s\n- openstack-tox-py37 https://logs.opendev.org/33/674933/3/check/openstack-tox-py37/e34a290/ : SUCCESS in 4m 26s\n- openstack-tox-docs https://logs.opendev.org/33/674933/3/check/openstack-tox-docs/6fd6146/html/ : SUCCESS in 5m 36s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/33/674933/3/check/tripleo-ci-centos-7-containers-multinode/1d6fd18/ : SUCCESS in 2h 40m 37s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/33/674933/3/check/tripleo-ci-centos-7-undercloud-containers/4da5478/ : SUCCESS in 1h 51m 00s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/33/674933/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/d1fa794/ : SUCCESS in 2h 30m 42s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/33/674933/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/6c19166/ : SUCCESS in 2h 54m 51s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/33/674933/3/check/tripleo-ci-centos-7-standalone/34abf10/ : SUCCESS in 2h 02m 51s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/33/674933/3/check/tripleo-ci-fedora-28-standalone/710d358/ : SUCCESS in 2h 02m 53s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/33/674933/3/check/tripleo-ci-centos-7-standalone-upgrade/b3604b3/ : SUCCESS in 2h 09m 11s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/33/674933/3/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/f93d5fe/ : FAILURE in 2h 04m 11s (non-voting)\n- tripleo-build-containers-centos-7 https://logs.opendev.org/33/674933/3/check/tripleo-build-containers-centos-7/7384995/ : SUCCESS in 48m 01s\n- tripleo-build-containers-centos-7-buildah https://logs.opendev.org/33/674933/3/check/tripleo-build-containers-centos-7-buildah/05a9fd4/ : SUCCESS in 1h 03m 14s\n- tripleo-ci-centos-7-scenario001-standalone https://logs.opendev.org/33/674933/3/check/tripleo-ci-centos-7-scenario001-standalone/6d47a20/ : SUCCESS in 1h 52m 55s\n- tripleo-ci-centos-7-scenario002-standalone https://logs.opendev.org/33/674933/3/check/tripleo-ci-centos-7-scenario002-standalone/8f6c64f/ : SUCCESS in 1h 42m 47s\n- tripleo-ci-centos-7-scenario003-standalone https://logs.opendev.org/33/674933/3/check/tripleo-ci-centos-7-scenario003-standalone/df5253c/ : SUCCESS in 1h 15m 44s\n- tripleo-ci-centos-7-scenario004-standalone https://logs.opendev.org/33/674933/3/check/tripleo-ci-centos-7-scenario004-standalone/687c06e/ : SUCCESS in 1h 59m 14s\n- tripleo-ci-centos-7-scenario010-standalone https://logs.opendev.org/33/674933/3/check/tripleo-ci-centos-7-scenario010-standalone/0f5204d/ : FAILURE in 1h 57m 28s\n- tripleo-ci-centos-7-scenario012-standalone https://logs.opendev.org/33/674933/3/check/tripleo-ci-centos-7-scenario012-standalone/8e04d48/ : SUCCESS in 1h 54m 35s (non-voting)\n- tripleo-ci-centos-7-scenario007-standalone https://logs.opendev.org/33/674933/3/check/tripleo-ci-centos-7-scenario007-standalone/1a98dd8/ : SUCCESS in 1h 30m 24s (non-voting)\n- openstack-tox-linters https://logs.opendev.org/33/674933/3/check/openstack-tox-linters/aa88a66/ : SUCCESS in 4m 21s\n- openstack-tox-py35 https://logs.opendev.org/33/674933/3/check/openstack-tox-py35/12e9ba1/ : SUCCESS in 4m 50s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://logs.opendev.org/33/674933/3/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/a13337b/ : SUCCESS in 2h 28m 19s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://logs.opendev.org/33/674933/3/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/dbaf30f/ : FAILURE in 1h 45m 32s (non-voting)\n- openstack-tox-lower-constraints https://logs.opendev.org/33/674933/3/check/openstack-tox-lower-constraints/7711a74/ : SUCCESS in 6m 01s","accounts_in_message":[],"_revision_number":3},{"id":"bc1b300fd9cfc83ab7299e8ab76050ddbf7ef8bb","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-08 02:19:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/33/674933/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/4470076/ : SUCCESS in 2h 47m 09s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/33/674933/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/7a102f8/ : SUCCESS in 2h 47m 51s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/33/674933/3/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/533c3ab/ : FAILURE in 9m 06s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/33/674933/3/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/f410ee9/ : FAILURE in 9m 16s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5a02dd64cfb0524a265af8fca32bb6c439e9be88","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-08 12:41:36.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"e1b7717bfa6cf9641785556d5a2342b8f125c466","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-08 15:53:44.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/33/674933/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/090299e/ : SUCCESS in 2h 53m 36s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/33/674933/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/9da02ac/ : SUCCESS in 2h 58m 37s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/33/674933/3/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/e605968/ : FAILURE in 8m 34s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/33/674933/3/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/39edc1b/ : FAILURE in 10m 23s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"bf69ca4097b06472a0b5fc7f53243a32dd3ab524","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 18:33:00.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://logs.opendev.org/33/674933/3/check/openstack-tox-cover/8b670a1/cover/ : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://logs.opendev.org/33/674933/3/check/openstack-tox-pep8/2605d32/ : SUCCESS in 4m 09s\n- openstack-tox-py27 https://logs.opendev.org/33/674933/3/check/openstack-tox-py27/c625c9a/ : SUCCESS in 4m 27s\n- openstack-tox-py36 https://logs.opendev.org/33/674933/3/check/openstack-tox-py36/492c1ab/ : SUCCESS in 4m 13s\n- openstack-tox-py37 https://logs.opendev.org/33/674933/3/check/openstack-tox-py37/1b0b346/ : SUCCESS in 4m 33s\n- openstack-tox-docs https://logs.opendev.org/33/674933/3/check/openstack-tox-docs/2872921/html/ : SUCCESS in 4m 09s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/33/674933/3/check/tripleo-ci-centos-7-containers-multinode/972a4c4/ : SUCCESS in 2h 29m 52s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/33/674933/3/check/tripleo-ci-centos-7-undercloud-containers/ccf8af6/ : SUCCESS in 1h 37m 16s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/33/674933/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/e638823/ : SUCCESS in 2h 14m 49s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/33/674933/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/ccf8918/ : SUCCESS in 2h 31m 32s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/33/674933/3/check/tripleo-ci-centos-7-standalone/bb7f68a/ : SUCCESS in 1h 52m 36s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/33/674933/3/check/tripleo-ci-fedora-28-standalone/d8fb7ce/ : SUCCESS in 1h 44m 32s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/33/674933/3/check/tripleo-ci-centos-7-standalone-upgrade/d3bcf26/ : SUCCESS in 2h 24m 11s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/33/674933/3/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/3ea13c4/ : FAILURE in 2h 19m 13s (non-voting)\n- tripleo-build-containers-centos-7 https://logs.opendev.org/33/674933/3/check/tripleo-build-containers-centos-7/c6ba7ad/ : SUCCESS in 47m 37s\n- tripleo-build-containers-centos-7-buildah https://logs.opendev.org/33/674933/3/check/tripleo-build-containers-centos-7-buildah/7ac8cad/ : SUCCESS in 1h 05m 09s\n- tripleo-ci-centos-7-scenario001-standalone https://logs.opendev.org/33/674933/3/check/tripleo-ci-centos-7-scenario001-standalone/4f1dae5/ : SUCCESS in 1h 28m 03s\n- tripleo-ci-centos-7-scenario002-standalone https://logs.opendev.org/33/674933/3/check/tripleo-ci-centos-7-scenario002-standalone/53f362c/ : SUCCESS in 2h 06m 18s\n- tripleo-ci-centos-7-scenario003-standalone https://logs.opendev.org/33/674933/3/check/tripleo-ci-centos-7-scenario003-standalone/92261e8/ : SUCCESS in 1h 21m 18s\n- tripleo-ci-centos-7-scenario004-standalone https://logs.opendev.org/33/674933/3/check/tripleo-ci-centos-7-scenario004-standalone/bed6450/ : SUCCESS in 1h 37m 19s\n- tripleo-ci-centos-7-scenario010-standalone https://logs.opendev.org/33/674933/3/check/tripleo-ci-centos-7-scenario010-standalone/940e5e8/ : FAILURE in 1h 23m 39s\n- tripleo-ci-centos-7-scenario012-standalone https://logs.opendev.org/33/674933/3/check/tripleo-ci-centos-7-scenario012-standalone/a9f9361/ : SUCCESS in 1h 51m 01s (non-voting)\n- tripleo-ci-centos-7-scenario007-standalone https://logs.opendev.org/33/674933/3/check/tripleo-ci-centos-7-scenario007-standalone/f6177b7/ : SUCCESS in 1h 35m 36s (non-voting)\n- openstack-tox-linters https://logs.opendev.org/33/674933/3/check/openstack-tox-linters/31e7909/ : SUCCESS in 3m 58s\n- openstack-tox-py35 https://logs.opendev.org/33/674933/3/check/openstack-tox-py35/c4c1ce8/ : SUCCESS in 4m 23s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://logs.opendev.org/33/674933/3/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/5baebee/ : SUCCESS in 2h 26m 08s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://logs.opendev.org/33/674933/3/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/fc76470/ : FAILURE in 1h 53m 45s (non-voting)\n- openstack-tox-lower-constraints https://logs.opendev.org/33/674933/3/check/openstack-tox-lower-constraints/3cfbc24/ : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":3},{"id":"97bec5304b20460fa41b023f269f35af41ecac2d","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-09 15:49:55.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"6460f2e14aa9f3a93368db703a78524a688a34b2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-09 18:46:50.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/33/674933/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/4c1675b/ : SUCCESS in 2h 35m 22s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/33/674933/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/dae310f/ : SUCCESS in 2h 54m 49s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/33/674933/3/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/d865bc1/ : SUCCESS in 21m 22s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/33/674933/3/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/71c9b68/ : SUCCESS in 16m 13s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1f95ee9a7eaa993e72f6190ddd06ca412cc09507","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-10 02:10:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://logs.opendev.org/33/674933/3/check/openstack-tox-cover/6e93029/cover/ : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://logs.opendev.org/33/674933/3/check/openstack-tox-pep8/cfac71a/ : SUCCESS in 5m 27s\n- openstack-tox-py27 https://logs.opendev.org/33/674933/3/check/openstack-tox-py27/bff7e75/ : SUCCESS in 4m 32s\n- openstack-tox-py36 https://logs.opendev.org/33/674933/3/check/openstack-tox-py36/33f02ba/ : SUCCESS in 3m 59s\n- openstack-tox-py37 https://logs.opendev.org/33/674933/3/check/openstack-tox-py37/36d96e7/ : SUCCESS in 3m 30s\n- openstack-tox-docs https://logs.opendev.org/33/674933/3/check/openstack-tox-docs/60982c4/html/ : SUCCESS in 3m 27s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/33/674933/3/check/tripleo-ci-centos-7-containers-multinode/f645e9c/ : SUCCESS in 2h 19m 08s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/33/674933/3/check/tripleo-ci-centos-7-undercloud-containers/b116202/ : SUCCESS in 1h 39m 23s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/33/674933/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/86e6808/ : SUCCESS in 2h 07m 00s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/33/674933/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/1aa843b/ : SUCCESS in 2h 21m 21s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/33/674933/3/check/tripleo-ci-centos-7-standalone/b8aa964/ : SUCCESS in 2h 20m 31s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/33/674933/3/check/tripleo-ci-fedora-28-standalone/5f69e78/ : SUCCESS in 1h 44m 15s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/33/674933/3/check/tripleo-ci-centos-7-standalone-upgrade/62144cd/ : SUCCESS in 2h 03m 56s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/33/674933/3/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/b2d75b1/ : FAILURE in 2h 06m 08s (non-voting)\n- tripleo-build-containers-centos-7 https://logs.opendev.org/33/674933/3/check/tripleo-build-containers-centos-7/8079bca/ : SUCCESS in 52m 28s\n- tripleo-build-containers-centos-7-buildah https://logs.opendev.org/33/674933/3/check/tripleo-build-containers-centos-7-buildah/8824c3e/ : SUCCESS in 1h 02m 29s\n- tripleo-ci-centos-7-scenario001-standalone https://logs.opendev.org/33/674933/3/check/tripleo-ci-centos-7-scenario001-standalone/bfa02cc/ : SUCCESS in 1h 46m 14s\n- tripleo-ci-centos-7-scenario002-standalone https://logs.opendev.org/33/674933/3/check/tripleo-ci-centos-7-scenario002-standalone/fedc136/ : SUCCESS in 2h 04m 15s\n- tripleo-ci-centos-7-scenario003-standalone https://logs.opendev.org/33/674933/3/check/tripleo-ci-centos-7-scenario003-standalone/f755955/ : SUCCESS in 1h 14m 44s\n- tripleo-ci-centos-7-scenario004-standalone https://logs.opendev.org/33/674933/3/check/tripleo-ci-centos-7-scenario004-standalone/6585c28/ : SUCCESS in 1h 49m 11s\n- tripleo-ci-centos-7-scenario010-standalone https://logs.opendev.org/33/674933/3/check/tripleo-ci-centos-7-scenario010-standalone/48c785f/ : FAILURE in 1h 15m 37s (non-voting)\n- tripleo-ci-centos-7-scenario012-standalone https://logs.opendev.org/33/674933/3/check/tripleo-ci-centos-7-scenario012-standalone/1b855df/ : SUCCESS in 1h 41m 18s (non-voting)\n- tripleo-ci-centos-7-scenario007-standalone https://logs.opendev.org/33/674933/3/check/tripleo-ci-centos-7-scenario007-standalone/ff322e7/ : SUCCESS in 1h 27m 44s (non-voting)\n- openstack-tox-linters https://logs.opendev.org/33/674933/3/check/openstack-tox-linters/dfb633b/ : SUCCESS in 3m 48s\n- openstack-tox-py35 https://logs.opendev.org/33/674933/3/check/openstack-tox-py35/194e700/ : SUCCESS in 5m 49s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://logs.opendev.org/33/674933/3/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/334a41c/ : SUCCESS in 3h 04m 12s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://logs.opendev.org/33/674933/3/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/6443073/ : FAILURE in 1h 21m 40s (non-voting)\n- openstack-tox-lower-constraints https://logs.opendev.org/33/674933/3/check/openstack-tox-lower-constraints/623465e/ : SUCCESS in 7m 22s","accounts_in_message":[],"_revision_number":3},{"id":"a7231d8c547a5bafb196d88c3b041b37f4172e54","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-12 15:10:41.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"996fbf67a5b504c86ab9330a14c8a803af98b6d3","author":{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},"date":"2019-08-13 17:37:06.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"33fe975a884f25b2f2cf6feff0224a85378ea1fd","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2019-08-14 06:24:41.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"3164c7753f63489a8abd58984af23f65b1c060d6","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-21 14:06:13.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"619594e4b37b712c9589c84db122eef8b045f5ed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 14:06:30.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"05caa6ef958b54d866a51f8b10f89d42758b9782","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 16:47:03.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e12cf5a38e041b5a2cefde206e2623d : SUCCESS in 3m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cb3c7cd26b1647b6ba8a887d28be0f84 : SUCCESS in 6m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f96eba9a53ee4a99bda842e9289d2786 : SUCCESS in 4m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bb74e67e14e64ecba59c35adc2fee8a9 : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42fa7fa0371244ff8fe45dcb31bc0e58 : SUCCESS in 4m 17s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/2862ce6cf5ac4f339035f2c14b4401c3 : FAILURE in 19m 20s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/e113952c9a734869980ed5990da09b78 : SUCCESS in 2h 02m 50s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/e326ddc9a85442d9ba3defa9043c7b26 : SUCCESS in 2h 24m 30s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/ceb5704071d449229faf0c8b318c1768 : SUCCESS in 1h 52m 29s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/1ef6f36a37fe4c26aa504a3d36ae0a1c : SUCCESS in 49m 04s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/a3bf8ce4977342de8b9ac728f2075b3a : SUCCESS in 58m 39s\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/e47e0249471a48ad81178f1e908bfe50 : SUCCESS in 2h 01m 34s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/b16e4a44fb214439928def809e50baeb : SUCCESS in 1h 38m 35s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/c19df5e7a43147dd89a6f280c139c1e5 : SUCCESS in 1h 27m 22s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/4cac030b7dd24cf28279e829b9803b83 : SUCCESS in 1h 58m 07s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/ca1ddba878a044908428c0949a98a191 : SUCCESS in 1h 21m 41s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d8fe5b9bb40642fcb8c30ae1cb2dc5ad : SUCCESS in 6m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/337a7437618d4150961f6e24e633a72c : SUCCESS in 7m 11s","accounts_in_message":[],"_revision_number":3},{"id":"acefce0e5efea17a9e0449aa68d0867f3fcf6205","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-21 18:21:22.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"1bada352624a21056a26325468ef2801f7dbad20","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-21 21:00:20.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/33/674933/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/98c4035/ : FAILURE in 1h 48m 21s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/33/674933/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/349fda5/ : SUCCESS in 2h 37m 13s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/33/674933/3/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/e1b9d89/ : SUCCESS in 23m 48s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/33/674933/3/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/c2cc049/ : SUCCESS in 18m 49s (non-voting)\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/33/674933/3/openstack-check/tripleo-build-containers-rhel-8-master/28ee6a8/ : SUCCESS in 43m 25s","accounts_in_message":[],"_revision_number":3},{"id":"c421c8da5280739d1a78d42266bf604afead1f8d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 21:28:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ddfd33c20a5b4a28b04a75c4c2c2ae6b : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af1c11fe873e44dabc51b7e1d1d36f5e : SUCCESS in 4m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/899a4d7b98e74017abe2844026de5413 : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3d55f6da6eb5492a9e592ae1e6bf03af : SUCCESS in 3m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e6870d0c1cff4e03b8b4c595b5ff1826 : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffc34b789e324af79c3bbc66444d3c2b : SUCCESS in 4m 38s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/a3a02e7bf69e41d3afc792e58762d7c8 : SUCCESS in 2h 32m 47s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/f58a6546707e4442a49c3654061c4cda : SUCCESS in 1h 31m 10s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/4e8ed3068f9245869476f2c588d279d0 : SUCCESS in 2h 11m 37s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/16abb4954eb94df8858116c50859da22 : SUCCESS in 2h 31m 30s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/3c31c6b8ad4c4600bf45a395bdddb328 : SUCCESS in 2h 08m 56s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/22948974d4f94e63b2ed634b081ead78 : SUCCESS in 2h 20m 25s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/ed161fd0fdc1493a98c9a75eec80c925 : FAILURE in 2h 13m 34s (non-voting)\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/5b09362afdd54830ba032793bbdd8ec4 : SUCCESS in 44m 13s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/f5d07dce5dcd4e099d344ed15ad4ee07 : SUCCESS in 53m 48s\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/39aba009e2ba4e3abbd29854df4c0bfc : SUCCESS in 1h 33m 48s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/7f16e828d7374b7ea393c8597b7d05ff : SUCCESS in 1h 45m 55s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/6f16de6d3833490aa49bfbeb7315dbfb : SUCCESS in 1h 28m 03s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/b7b81ec386dd4494abf61138cd7aeece : SUCCESS in 1h 50m 28s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/73e02c7352734cd7887afba86ec2832c : SUCCESS in 1h 24m 20s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/00edf783c5fa49f09e20b99fefe8f784 : FAILURE in 1h 25m 40s (non-voting)\n- tripleo-ci-centos-7-scenario007-standalone https://zuul.opendev.org/t/openstack/build/4c5c1525e20c4532976bb79788593efa : SUCCESS in 58m 41s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/415439b338aa4127a82f22eaf946ce9e : SUCCESS in 7m 19s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/2f6a0f0207f94da6b66ca6bc14fa359b : SUCCESS in 2h 21m 51s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/5e595d85242747ebb69dd7f56fea122b : FAILURE in 1h 20m 47s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9cdb8777a47b4d3d835a3c755d3a98fa : SUCCESS in 5m 41s","accounts_in_message":[],"_revision_number":3},{"id":"2f7b8ac1a71ba9195ef5f7a94743f5b55461048d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 21:28:26.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"da925bcc3d5ac106e81587305afa48d903c10648","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 00:54:47.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b8c5e1bf12b499a8df5a5f54c91a2f2 : SUCCESS in 3m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/54ef570adffa4440952489a98d182a94 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/50f9087c0e8f4e6282dfdaddb0a17d22 : SUCCESS in 4m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/82eb8cb4bcaa4dd592cb9eb1b6969626 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc958140ffbc4391824041f5fce4322d : SUCCESS in 4m 21s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/f2e4b805d6db499282e3fb56794d1993 : SUCCESS in 2h 48m 53s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/232c6b014c534088839d257ceea7a456 : SUCCESS in 1h 56m 39s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/8d64a8e5721f4da8816bdf81e4b644d2 : SUCCESS in 2h 42m 51s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/eb3497adefab4e3f8564dac4a429379c : SUCCESS in 1h 37m 18s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/bee770d4bb294f04804297e0f534669f : SUCCESS in 43m 36s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/fc5ab7114538477c8bd1021b8580ed7c : SUCCESS in 58m 26s\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/5bd3201e0b3c4ec8b8002de90e552a49 : SUCCESS in 2h 06m 51s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/b3e7c11f4d114e21a6e93938621660c3 : SUCCESS in 1h 44m 25s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/185a06f0535947949af3b0975c950e1c : SUCCESS in 2h 09m 01s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/2e4f741f193c40a68a0770d90dfac78a : SUCCESS in 1h 47m 53s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/9a9afe3b5f36446f9bf0b96f04fc1446 : SUCCESS in 1h 23m 36s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/db76b283c6624a9aa63d82ea36fb996a : SUCCESS in 4m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/850b981e0b224d22af24d9e13db71a27 : SUCCESS in 6m 20s","accounts_in_message":[],"_revision_number":3},{"id":"f1be062c793394555d2d56cb29e3b7d9aa23a7d6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 00:54:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"35d34e770b04a9ff0cc2beddb50c22f9d86ec12d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 00:57:03.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbf5c48d3ad142eaa5b7eab016193454 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"220c6dec7fa533c77f3e6965108f71fc725ed2d8","revisions":{"a2408131972fdaca3ff7fa7bb3dee0a4c401d826":{"kind":"REWORK","_number":1,"created":"2019-08-06 21:21:34.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/33/674933/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/33/674933/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/33/674933/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/33/674933/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/33/674933/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/33/674933/1"}}},"commit":{"parents":[{"commit":"a3dc7aa00cd3e61cc66c12c678fa2a4dad8b2182","subject":"Restore AllNodesExtraMapData","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/a3dc7aa00cd3e61cc66c12c678fa2a4dad8b2182"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-08-06 21:20:42.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-08-06 21:20:42.000000000","tz":-240},"subject":"Remove OpenShift containers","message":"Remove OpenShift containers\n\nSee I6a73f7f22dda69bef324ffdaecdcd6be693c1257 for the whole context\nbut we are removing OpenShift from TripleO.\n\nChange-Id: Ia64ceb1f05d96d2358509392bb3cbc3b9cb61030\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/a2408131972fdaca3ff7fa7bb3dee0a4c401d826"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/a2408131972fdaca3ff7fa7bb3dee0a4c401d826"}]},"branch":"refs/heads/master"},"ee193cb9ee260d6192f3da6284dea34da897467e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-08-07 14:08:59.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/33/674933/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/33/674933/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/33/674933/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/33/674933/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/33/674933/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/33/674933/2"}}},"commit":{"parents":[{"commit":"a3dc7aa00cd3e61cc66c12c678fa2a4dad8b2182","subject":"Restore AllNodesExtraMapData","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/a3dc7aa00cd3e61cc66c12c678fa2a4dad8b2182"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-08-06 21:20:42.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-08-07 14:08:57.000000000","tz":0},"subject":"Remove OpenShift containers","message":"Remove OpenShift containers\n\nSee I6a73f7f22dda69bef324ffdaecdcd6be693c1257 for the whole context\nbut we are removing OpenShift from TripleO.\n\nDepends-On: https://review.rdoproject.org/r/21749\nChange-Id: Ia64ceb1f05d96d2358509392bb3cbc3b9cb61030\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/ee193cb9ee260d6192f3da6284dea34da897467e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/ee193cb9ee260d6192f3da6284dea34da897467e"}]},"branch":"refs/heads/master"},"220c6dec7fa533c77f3e6965108f71fc725ed2d8":{"kind":"REWORK","_number":3,"created":"2019-08-07 15:19:46.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/33/674933/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/33/674933/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/33/674933/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/33/674933/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/33/674933/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/33/674933/3"}}},"commit":{"parents":[{"commit":"a3dc7aa00cd3e61cc66c12c678fa2a4dad8b2182","subject":"Restore AllNodesExtraMapData","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/a3dc7aa00cd3e61cc66c12c678fa2a4dad8b2182"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-08-06 21:20:42.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-08-07 15:19:40.000000000","tz":-240},"subject":"Remove OpenShift containers","message":"Remove OpenShift containers\n\nSee I6a73f7f22dda69bef324ffdaecdcd6be693c1257 for the whole context\nbut we are removing OpenShift from TripleO.\n\nNote: the scripts/tripleo-deploy-openshift script hasn\u0027t been removed\nyet because we need to update the packaging first in\nhttps://review.rdoproject.org/r/#/c/21749 but first it requires the unit\ntests to be removed.\n\nChange-Id: Ia64ceb1f05d96d2358509392bb3cbc3b9cb61030\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/220c6dec7fa533c77f3e6965108f71fc725ed2d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/220c6dec7fa533c77f3e6965108f71fc725ed2d8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
