)]}'
{"id":"openstack%2Ftripleo-common~676270","triplet_id":"openstack%2Ftripleo-common~master~I953c5648ac79dc8a59fd4bc798c21e6e383e1afc","project":"openstack/tripleo-common","branch":"master","topic":"hardened-images-python3","hashtags":[],"change_id":"I953c5648ac79dc8a59fd4bc798c21e6e383e1afc","subject":"Add python3 file for hardened images","status":"MERGED","created":"2019-08-13 20:04:34.000000000","updated":"2019-08-20 20:37:48.000000000","submitted":"2019-08-20 20:35:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"676270-1566333304097-ac68746e","meta_rev_id":"1fd579aac9866d1f617b12137f5d3a7a9798a3f0","_number":676270,"virtual_id_number":676270,"owner":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":8161,"name":"Lon Hohberger","email":"lhh@redhat.com","username":"lon"},{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"date":"2019-08-20 14:47:32.000000000","_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-08-20 20:35:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-08-20 18:31:37.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"recommended":{"_account_id":8161,"name":"Lon Hohberger","email":"lhh@redhat.com","username":"lon"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2019-08-16 20:04:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8161,"name":"Lon Hohberger","email":"lhh@redhat.com","username":"lon"},{"value":2,"date":"2019-08-19 19:50:59.000000000","_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":2,"date":"2019-08-16 14:02:48.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"all":[{"value":0,"date":"2019-08-16 13:03:29.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":8161,"name":"Lon Hohberger","email":"lhh@redhat.com","username":"lon"},{"value":1,"date":"2019-08-19 19:50:59.000000000","_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},{"value":0,"date":"2019-08-19 13:34:51.000000000","_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"_account_id":8161,"name":"Lon Hohberger","email":"lhh@redhat.com","username":"lon"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-16 13:03:29.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-08-16 14:02:48.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-08-16 20:04:28.000000000","updated_by":{"_account_id":8161,"name":"Lon Hohberger","email":"lhh@redhat.com","username":"lon"},"reviewer":{"_account_id":8161,"name":"Lon Hohberger","email":"lhh@redhat.com","username":"lon"},"state":"REVIEWER"},{"updated":"2019-08-19 13:34:51.000000000","updated_by":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"reviewer":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"state":"REVIEWER"},{"updated":"2019-08-19 19:50:59.000000000","updated_by":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"reviewer":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"state":"REVIEWER"},{"updated":"2019-08-20 18:31:37.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-20 20:35:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"36a2c81a6f1963e32c9072494970db87abe4e356","author":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"date":"2019-08-13 20:04:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"739db261ee3f35b87469f66a3773cbe4cdfc79b3","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-13 22:56:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/70/676270/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/973bb80/ : SUCCESS in 2h 46m 21s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/70/676270/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/e8fdad3/ : SUCCESS in 2h 49m 51s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/70/676270/1/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/5fbb9ef/ : SUCCESS in 21m 17s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/70/676270/1/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/b6bc385/ : SUCCESS in 17m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5880f552367298111e6ebe2c015c936d63dd89ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 23:02:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fb86aa4c24304ee9a75dba7d35d04a71 : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f817bcc6fb354138bba1d7b336a9bf6a : SUCCESS in 3m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fdaca4198c564df9801871d7730f13fa : SUCCESS in 3m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d77f4aa8ea4a434fbee36c72445e5856 : SUCCESS in 4m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/80b8fcce77e7412e91191241f301d8ce : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7627717ec0fa4057abc6e4b0d12bac75 : SUCCESS in 5m 05s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/fa1ce2844f0142828a3c172620017aaa : SUCCESS in 2h 36m 16s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/f93b5b5e3c51408f805d425151f143ec : SUCCESS in 1h 35m 56s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/8404f86230564b3ea7a308dc46d25822 : SUCCESS in 2h 37m 51s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/338beb92c3b94865b40a6143d0c546cd : SUCCESS in 1h 45m 59s\n- tripleo-ci-fedora-28-standalone https://zuul.opendev.org/t/openstack/build/851bf0e446b945099c0f463a93fc2cbb : SUCCESS in 1h 50m 58s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/42ae32f8515740b1bef28e7c3e90eae5 : SUCCESS in 2h 09m 51s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/7a4c91c2b072471990832dd1c2a11680 : FAILURE in 1h 53m 32s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/47d6fdabec32497dad4091e05dd11d90 : SUCCESS in 4m 38s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9769ae3360bd406197c4c00ac146fa2a : SUCCESS in 6m 36s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/cfafee3dfb6d4b5c9e0e481d22b0533d : SUCCESS in 21m 11s\n- tripleo-buildimage-ironic-python-agent-centos-7 https://zuul.opendev.org/t/openstack/build/5a12a21589174c898432bc073387a93f : SUCCESS in 16m 16s (non-voting)\n- tripleo-buildimage-overcloud-hardened-full-centos-7 https://zuul.opendev.org/t/openstack/build/8798a46d583848dda0d5d01cf674cf15 : SUCCESS in 24m 19s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cfaaccadd8684ca9b891f85e70003471 : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":1},{"id":"f4c12c17516cb83d16f79e3a3931a7d76c87a09e","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-16 14:02:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"45e5256fbd96a40bba3b424e86de0bb84f754991","author":{"_account_id":8161,"name":"Lon Hohberger","email":"lhh@redhat.com","username":"lon"},"date":"2019-08-16 20:04:28.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d70fc28871b1552e97cc25d5c1c1c2d3390c466b","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2019-08-19 19:50:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1811ca9cdbbb9f6e492a23b2b2151470738eced1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 19:51:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0cc8ac24965ebd932401d3a14f00a6d01fc2af04","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 22:48:47.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d29aa90419ef43dd8b8f369f357528a1 : SUCCESS in 3m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bcabe650fed64610974e0dce527f43ce : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/73320e680780438c9857ffe020513f37 : SUCCESS in 4m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e550b960054b46979ff44ba51337c96e : SUCCESS in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db6eab6f013940ea9b313362de77eef4 : SUCCESS in 3m 47s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/95007ccb1142428ab054a08ea337052f : SUCCESS in 2h 48m 03s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/7f507d6b753446cab8b98f9bb480cfbe : SUCCESS in 1h 55m 50s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/cde73f042a0044a4a2728f1b5a1394e6 : SUCCESS in 2h 33m 52s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/301bb35564a04a1cb108edf7b725f19f : SUCCESS in 1h 30m 50s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/47616c2307c2443abddbb6743949eaea : SUCCESS in 5m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/600fd857642745479937a026c40ad959 : SUCCESS in 6m 26s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/aff571598145425ca29f99bdaa6a671b : RETRY_LIMIT in 9m 34s\n- tripleo-buildimage-ironic-python-agent-centos-7 https://zuul.opendev.org/t/openstack/build/cafa0902e7d446b290144a16963ea2f9 : RETRY_LIMIT in 7m 14s (non-voting)\n- tripleo-buildimage-overcloud-hardened-full-centos-7 https://zuul.opendev.org/t/openstack/build/861c515149d641848210835f12ee011f : RETRY_LIMIT in 9m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6b6fcae65c16d20205d535503ff0b9a5b335d9f2","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2019-08-20 11:16:26.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"176446c5c9f4920d06a77a00ca4a918f2c0576f3","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-20 14:04:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/70/676270/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/7fa2e37/ : SUCCESS in 2h 46m 26s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/70/676270/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/1843744/ : SUCCESS in 2h 45m 00s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/70/676270/1/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/3a7f865/ : SUCCESS in 22m 13s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/70/676270/1/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/fa2a570/ : SUCCESS in 19m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"feb1590c8e6700c8b22892ff34ed34b565084f4b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 14:37:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/25ca6737e0f044dbbe5dc3c44b1cd2bc : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/add2d5aa79f64831b6925b93c24c1b92 : SUCCESS in 5m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cf1f6c3e016047139263d9e6341462c6 : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7381cbdb7e3b4948859e0a85e3fd88f6 : SUCCESS in 5m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dea56a60e56146c58964072e7c065bcc : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1105b7a01e5649ef805818031ee1ed1e : SUCCESS in 5m 47s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/74cc23dfbb744e168def81e642f958ca : SUCCESS in 2h 41m 03s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/473a4ff27b5e4dd984193a248840b176 : FAILURE in 23m 21s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/5d8ea8c200dd494792b06522ac9db832 : SUCCESS in 2h 47m 07s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/617eb4fc8ee84603b5f89e519437cb71 : FAILURE in 2h 50m 33s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/ea52344bdefb4d36af9047f2474527b6 : SUCCESS in 2h 15m 18s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/6ace26907c2944a693a740a7e27c23ae : FAILURE in 2h 16m 16s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/930f84c6a40748de9440b1112b0c399d : SUCCESS in 5m 39s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/e5ded1a8d9ca428b8217c5ade15c8b0f : SUCCESS in 21m 29s\n- tripleo-buildimage-ironic-python-agent-centos-7 https://zuul.opendev.org/t/openstack/build/4cfc9bb561d043e89925cc7353e67a23 : SUCCESS in 16m 04s (non-voting)\n- tripleo-buildimage-overcloud-hardened-full-centos-7 https://zuul.opendev.org/t/openstack/build/18108abf6a5044959ad78f2cb5bbab0e : SUCCESS in 20m 27s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5a7c38706bce4d9aab48e7348df4aa09 : SUCCESS in 6m 03s","accounts_in_message":[],"_revision_number":1},{"id":"7e2f3465c6acae009512b3ce322a5c855f1c38c4","author":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"date":"2019-08-20 14:47:32.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"af18a6a5c62bb95d1b310813b0bc3d8db1c0314a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 17:41:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f452f0f5b24740c8b0e4e36faa8b7969 : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb4f14a31a724f6dbe22f7274756fc6d : SUCCESS in 4m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f6905c47996d42319c5b98c89108cab4 : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ecb5563f6df8492d8c0a3c5da7e23110 : SUCCESS in 7m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7456d38bfe1b4947b3febf2e972a2975 : SUCCESS in 7m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4f030a6f6f94a79836804318ea85966 : SUCCESS in 6m 48s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/d25ee95a65c342ea9ef86148d9cc2cda : SUCCESS in 2h 34m 32s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/752cc1ab2c24427eb04b4111d87c5532 : SUCCESS in 1h 49m 14s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/bb1541b2f9f64c6e9e3455d79bd88d83 : SUCCESS in 2h 36m 07s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/304a8b090ae04056aafed48b0aac44b2 : SUCCESS in 1h 46m 24s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/cde390dcb00d459bb89417eb45b26f31 : SUCCESS in 2h 19m 26s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/c5abd11cde394c0bb7e1c50dd0498573 : FAILURE in 2h 09m 55s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0a6abee4e9764a23b846c719d68a6f4f : SUCCESS in 6m 28s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/4b805b0d8ec64c729b0c7b27a74f52cf : SUCCESS in 23m 31s\n- tripleo-buildimage-ironic-python-agent-centos-7 https://zuul.opendev.org/t/openstack/build/2c68517e33b949c5b03aa5d0ff91bee9 : SUCCESS in 23m 05s (non-voting)\n- tripleo-buildimage-overcloud-hardened-full-centos-7 https://zuul.opendev.org/t/openstack/build/7ac876a3656c4b28838d0ae92d4dfc01 : SUCCESS in 21m 52s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d7270ea8f34a497688749726ba8d05c9 : SUCCESS in 5m 21s","accounts_in_message":[],"_revision_number":1},{"id":"0ad9d6cd461427dbb23a73280f06e7fac9aad0d0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 17:41:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"09e50a418684ca39d9eb0085f02da2fb75989f60","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-20 18:31:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/70/676270/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/785ccd2/ : TIMED_OUT in 3h 40m 38s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/70/676270/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/ffcce37/ : FAILURE in 3h 31m 25s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/70/676270/1/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/6fe1d99/ : SUCCESS in 22m 23s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/70/676270/1/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/ef3b284/ : SUCCESS in 18m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"729f921afcb9e46745b2267747a2b93b65c17d43","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 20:35:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/096a1443a2b14f03ad484de90ab13f47 : SUCCESS in 5m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1ef7c80d5acf488f962f511ab4adb37a : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0e4c881720914d8da11560c479159e02 : SUCCESS in 5m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c5b8017db04c4a49a0e4beabfad0bb45 : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/372375d57b614e82bd3912dae1f5d20a : SUCCESS in 4m 11s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/51e8ce4bec464a9e869c7e8f8bdee0ca : SUCCESS in 2h 46m 15s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/04ffc22f761e47d687951a19f19a9bf7 : SUCCESS in 1h 52m 29s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/8034f0561cdb4e5a846640e517d37307 : SUCCESS in 2h 23m 34s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/48e9ad74d3474ce4892d20bbb4a51dbc : SUCCESS in 1h 49m 18s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b2a23a3703d94863a1fcf28d5473255e : SUCCESS in 4m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b8875568ec56428eb7a570744114a849 : SUCCESS in 7m 48s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/891a1ed347744f61ab56d6b7c0baa3cd : SUCCESS in 21m 51s\n- tripleo-buildimage-ironic-python-agent-centos-7 https://zuul.opendev.org/t/openstack/build/66282773c3834f98b0d99625930f7508 : SUCCESS in 20m 28s (non-voting)\n- tripleo-buildimage-overcloud-hardened-full-centos-7 https://zuul.opendev.org/t/openstack/build/399cef261c1e487ca37a90c0edff2716 : SUCCESS in 28m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"569cf9d95757a61a58eb8bebfc85888c712a6fe0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 20:35:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"92e7b0de48e9df3d814ad9dc829013817017ca95","author":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"date":"2019-08-20 20:36:52.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit f4e5b139c9579706200f4efd3bef9e5316ea4080","accounts_in_message":[],"_revision_number":1},{"id":"58a22af5897ab92dbcdd902c2fb08042a6fd4885","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 20:37:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8f9709dbd0d489e86e4a39d87a2ae3b : SUCCESS in 2m 23s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"af23dfd61be98107aa9a7c8c180e805bb4030849","revisions":{"af23dfd61be98107aa9a7c8c180e805bb4030849":{"kind":"REWORK","_number":1,"created":"2019-08-13 20:04:34.000000000","uploader":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"ref":"refs/changes/70/676270/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/70/676270/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/70/676270/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/70/676270/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/70/676270/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/70/676270/1"}}},"commit":{"parents":[{"commit":"bb73c03f5fd691874b1d758fdd9f96376790041b","subject":"Remove the tripleo.plan_management.v1.list_roles workflow","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/bb73c03f5fd691874b1d758fdd9f96376790041b"}]}],"author":{"name":"Bob Fournier","email":"bfournie@redhat.com","date":"2019-08-13 19:59:01.000000000","tz":-240},"committer":{"name":"Bob Fournier","email":"bfournie@redhat.com","date":"2019-08-13 20:04:32.000000000","tz":-240},"subject":"Add python3 file for hardened images","message":"Add python3 file for hardened images\n\nAdd a python3 file to support RHEL 8 for hardened images.  In addition\nremove the screen package as its no longer supported.\n\nChange-Id: I953c5648ac79dc8a59fd4bc798c21e6e383e1afc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/af23dfd61be98107aa9a7c8c180e805bb4030849"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/af23dfd61be98107aa9a7c8c180e805bb4030849"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
