)]}'
{"id":"openstack%2Ftripleo-common~677705","triplet_id":"openstack%2Ftripleo-common~master~I8c449731915e4eb44ff92d85786f47d7707a5651","project":"openstack/tripleo-common","branch":"master","topic":"ceph_dashboard","hashtags":[],"change_id":"I8c449731915e4eb44ff92d85786f47d7707a5651","subject":"Change ceph dashboard service name to meet puppet requirements","status":"MERGED","created":"2019-08-21 11:42:52.000000000","updated":"2019-08-23 02:25:19.000000000","submitted":"2019-08-23 02:23:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"677705-1566527029346-8367d9e9","meta_rev_id":"e26af8f7e47f0cbb1f6aef1fbca7c6a67dc9d96e","_number":677705,"virtual_id_number":677705,"owner":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"value":0,"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"date":"2019-08-22 10:33:09.000000000","_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"value":2,"date":"2019-08-23 02:23:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-22 09:54:40.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"value":0,"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"value":2,"date":"2019-08-22 20:01:58.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"date":"2019-08-22 06:43:38.000000000","_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"value":0,"date":"2019-08-22 06:43:50.000000000","_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"value":1,"date":"2019-08-22 20:01:58.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"date":"2019-08-22 06:43:43.000000000","_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"date":"2019-08-21 12:18:23.000000000","_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-21 12:18:23.000000000","updated_by":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"reviewer":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"state":"REVIEWER"},{"updated":"2019-08-22 06:43:38.000000000","updated_by":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"reviewer":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"state":"REVIEWER"},{"updated":"2019-08-22 06:43:43.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2019-08-22 06:43:50.000000000","updated_by":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"reviewer":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"state":"REVIEWER"},{"updated":"2019-08-22 09:54:40.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-22 20:01:58.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-08-23 02:23:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"56191e8c2ec172d4d30ed031c231e7398173fdf8","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2019-08-21 11:42:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"075a6913aed8a66a7d2897045c14cb44f0430519","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 14:42:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d30a6939f954ce59cb08e15378a478e : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c650287c0ac4ed48d69034ad9713530 : SUCCESS in 5m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3fe9ee24f2e64a60aaec9385b44ecf0a : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a39d4b1f0e14724b4ff2c30e0a28843 : SUCCESS in 4m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9f4fd9fb86874cbb886e78f000f28b55 : SUCCESS in 7m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7aad5aed31b4d21b79d65e55747853f : SUCCESS in 5m 28s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/cdb67e3673a048f9af12e727bd0fc7cc : SUCCESS in 2h 37m 09s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/ff924c00f5914f9c98aa98b64a845b9f : SUCCESS in 1h 46m 25s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/e25a6ac95397404a8a8588e761d7469c : SUCCESS in 2h 33m 18s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/76896300b2aa440ab93a252f357c168b : SUCCESS in 2h 00m 03s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/98ec17b5eb43446c8a8e3011c8b80d4f : SUCCESS in 2h 28m 58s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/aec49903e3b34cb69a9a51f66ce2dcc0 : FAILURE in 2h 19m 37s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c8792477cb9249bca5e8ee1fc8c5e6ed : SUCCESS in 5m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4eb42e38b4324bcd9eba2cd3b6e5a89f : SUCCESS in 6m 49s","accounts_in_message":[],"_revision_number":1},{"id":"b5c31ba07dd641e484b85bb9132b5d97580441cf","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-21 15:40:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 finger://ze01.softwarefactory-project.io:7979/8272642b95e4474e8c36d0d6fb14e674 : POST_FAILURE in 3h 56m 11s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/05/677705/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/5790e60/ : FAILURE in 2h 26m 45s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/05/677705/1/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/291aaf4/ : SUCCESS in 23m 55s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/05/677705/1/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/d872ab4/ : SUCCESS in 18m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"468c4df2467be9568a7c8c6f0e795245da32ea1f","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2019-08-22 06:43:21.000000000","message":"Patch Set 1:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":1},{"id":"d98e2491309b64e37298c99806e10f918c1308a0","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2019-08-22 07:45:03.000000000","message":"Patch Set 1:\n\nTested on review [1] and as expected we can see the generated inventory with ceph_grafana and grafana-server sections[2] on scenario001 (where dashboard_enabled: true)\n\n[1]https://review.opendev.org/#/c/674556/\n[2]https://storage.bhs1.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/logs_56/674556/12/check/tripleo-ci-centos-7-scenario001-standalone/569d09c/logs/undercloud/home/zuul/undercloud-ansible-fa9AO7/inventory.yaml.txt.gz","accounts_in_message":[],"_revision_number":1},{"id":"c4a3a27fb6cd8bd7bb62393f4500c36d814b5dfe","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-22 09:54:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/05/677705/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/3c8a2e7/ : SUCCESS in 2h 53m 25s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/05/677705/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/d94b6dd/ : SUCCESS in 3h 10m 10s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/05/677705/1/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/2264bbd/ : SUCCESS in 23m 00s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/05/677705/1/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/c49cb34/ : SUCCESS in 19m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ca65428d5e12cefec30b94d36004b14faec51e1d","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2019-08-22 10:33:09.000000000","message":"Patch Set 1:\n\nThe last execution of [1] confirms the inventory works as expected [2]\n\n[1] https://review.opendev.org/#/c/674556/\n[2] https://12942841482e92c0cd55-0acfe5e696fda6aec28b6c718c3c4841.ssl.cf1.rackcdn.com/674556/13/check/tripleo-ci-centos-7-scenario001-standalone/d957dce/logs/undercloud/home/zuul/undercloud-ansible-sJqhxi/inventory.yaml.txt.gz","accounts_in_message":[],"_revision_number":1},{"id":"11570f63deaa978fb6cf922e911603db929fa9c1","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-22 20:01:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"42b08332d93aa4070138bbbfca8a50c13183c430","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 20:02:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e87af7471a13f3b53a374dc0aba5d5003f853f2e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 02:23:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ddd6f75c43d442ea7bc51e0d4ace168 : SUCCESS in 4m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/97edee418b8c4e4689c14d008f13517c : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4cf7894e74f24d06a43a1ba3428dbb80 : SUCCESS in 4m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ca01c014c8384e8fbf4882e238e6c7f9 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f214763cc374a31beb4485f87c5e6f3 : SUCCESS in 4m 55s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/a1ee241ad6d7464bae16ad4d88dc53c1 : SUCCESS in 2h 41m 45s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/03d5a350c06440738a75dedcaa90ea32 : SUCCESS in 1h 59m 02s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/5fe73ed7735c4333a7c895994cc38f12 : SUCCESS in 2h 40m 34s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/87351d0afdbe4199954ed29418e3249a : SUCCESS in 2h 04m 25s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a5c81494c6a44037bcca72878fedd565 : SUCCESS in 5m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e996bdeef4a648fba50d5020db5bc969 : SUCCESS in 6m 25s","accounts_in_message":[],"_revision_number":1},{"id":"ebe7e743ecc80bf45a849375e75f3273d7177803","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 02:23:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6d329b96e15526b5fb8a9f1370947dbb7f11c45a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 02:25:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c56d00d150fa41d08757fa0ff75f6413 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"990f0f25f458f7d89a304f1191d2847af1baeee2","revisions":{"990f0f25f458f7d89a304f1191d2847af1baeee2":{"kind":"REWORK","_number":1,"created":"2019-08-21 11:42:52.000000000","uploader":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"ref":"refs/changes/05/677705/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/05/677705/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/05/677705/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/05/677705/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/05/677705/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/05/677705/1"}}},"commit":{"parents":[{"commit":"80013b09eee807b319f265e720b1dfe0abdc1384","subject":"Merge \"Close the http sessions of registry on image prepare\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/80013b09eee807b319f265e720b1dfe0abdc1384"}]}],"author":{"name":"fpantano","email":"fpantano@redhat.com","date":"2019-08-21 07:52:23.000000000","tz":120},"committer":{"name":"fpantano","email":"fpantano@redhat.com","date":"2019-08-21 11:47:26.000000000","tz":120},"subject":"Change ceph dashboard service name to meet puppet requirements","message":"Change ceph dashboard service name to meet puppet requirements\n\nSince we have some issues on puppet using the \"-\" notation, this\nchange replace the service name removing the last part. This should\nnot affect the ceph-ansible inventory since the group name is still\ntranslated into grafana-server.\n\nChange-Id: I8c449731915e4eb44ff92d85786f47d7707a5651\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/990f0f25f458f7d89a304f1191d2847af1baeee2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/990f0f25f458f7d89a304f1191d2847af1baeee2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
