)]}'
{"id":"openstack%2Ftripleo-common~682855","triplet_id":"openstack%2Ftripleo-common~master~Ie7adf41cef0fafde646f840b2bf3e01ba42a17d5","project":"openstack/tripleo-common","branch":"master","topic":"performance_improvements","hashtags":[],"change_id":"Ie7adf41cef0fafde646f840b2bf3e01ba42a17d5","subject":"Implement Ansible fact cache for Mistral executor","status":"MERGED","created":"2019-09-18 11:07:38.000000000","updated":"2019-10-04 07:45:48.000000000","submitted":"2019-10-03 19:40:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682855-1570131616499-b42f53c0","meta_rev_id":"9b261d35c79a1e6160bfa0799751db183ac99793","_number":682855,"virtual_id_number":682855,"owner":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-10-03 09:36:38.000000000","_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-03 19:40:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":-1,"date":"2019-10-02 18:28:41.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":0,"_account_id":11166,"name":"Lukas Bezdicka","email":"social@v3.sk","username":"xbezdick"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},{"value":2,"date":"2019-10-02 14:46:27.000000000","_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":2,"date":"2019-10-03 13:15:27.000000000","_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":0,"_account_id":11166,"name":"Lukas Bezdicka","email":"social@v3.sk","username":"xbezdick"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":1,"date":"2019-10-03 13:15:27.000000000","_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":0,"_account_id":11166,"name":"Lukas Bezdicka","email":"social@v3.sk","username":"xbezdick"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},{"_account_id":11166,"name":"Lukas Bezdicka","email":"social@v3.sk","username":"xbezdick"},{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-18 11:08:39.000000000","updated_by":{"_account_id":11166,"name":"Lukas Bezdicka","email":"social@v3.sk","username":"xbezdick"},"reviewer":{"_account_id":11166,"name":"Lukas Bezdicka","email":"social@v3.sk","username":"xbezdick"},"state":"REVIEWER"},{"updated":"2019-09-18 11:26:41.000000000","updated_by":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"reviewer":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"state":"REVIEWER"},{"updated":"2019-09-18 14:29:33.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2019-09-18 14:52:15.000000000","updated_by":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"reviewer":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"state":"REVIEWER"},{"updated":"2019-09-20 19:06:59.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-09-21 18:08:54.000000000","updated_by":{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},"reviewer":{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},"state":"REVIEWER"},{"updated":"2019-09-26 13:32:13.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"REVIEWER"},{"updated":"2019-10-02 14:46:27.000000000","updated_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"reviewer":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"state":"REVIEWER"},{"updated":"2019-10-02 18:28:41.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-03 13:15:27.000000000","updated_by":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"reviewer":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"state":"REVIEWER"},{"updated":"2019-10-03 19:40:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"639b4649d2c06ef8ecfb96a11dbfd553d1cd0d0a","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-09-18 11:07:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"71279be9fa1ef65743d58cab6ebeead6d1412366","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-09-18 11:26:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"92f2a31ff51c8da923afda48220c1c87dc6be771","author":{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},"date":"2019-09-18 11:57:04.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f9dda9632752c619458fb76f7bb28f0ec1a8cc12","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 13:30:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2ae238ed3ca84f4ca1bbeaeb92901b8b : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7beccab8a8b4cbb953de4317578a0e7 : FAILURE in 2m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/99c9d0cd6c804608be7244d95b479e44 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/564f745e840247f1bf913124330b23f3 : SUCCESS in 2m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1a9d141d74cc4359b593d6e1f90041c1 : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e1060528b4d42dc8a03da434b95919a : SUCCESS in 3m 26s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/50178c362a834640a460fd2a96596554 : SUCCESS in 1h 36m 37s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario007-standalone https://zuul.opendev.org/t/openstack/build/d86eb4bfc0e84676bcf34ed24963708d : SUCCESS in 1h 33m 54s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3772b70020d144519684db87e0ca1739 : FAILURE in 4m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eba7f7a739f24531821176997ac760a8 : SUCCESS in 5m 17s","accounts_in_message":[],"_revision_number":1},{"id":"03d6bb94328422fb9642b71021b3202ac60394c3","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-18 14:19:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/55/682855/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/bf159d7/ : FAILURE in 3h 10m 12s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/55/682855/1/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/1b33358/ : SUCCESS in 21m 48s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/55/682855/1/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/8f49f89/ : SUCCESS in 18m 57s (non-voting)\n- tripleo-ci-rhel-8-scenario001-standalone-rdo http://logs.rdoproject.org/55/682855/1/openstack-check/tripleo-ci-rhel-8-scenario001-standalone-rdo/d15cd51/ : SUCCESS in 1h 36m 24s\n- tripleo-ci-rhel-8-scenario002-standalone-rdo http://logs.rdoproject.org/55/682855/1/openstack-check/tripleo-ci-rhel-8-scenario002-standalone-rdo/f8e86ec/ : SUCCESS in 1h 24m 47s","accounts_in_message":[],"_revision_number":1},{"id":"800af33390968136fe3159f9c25bd8b54e9fa577","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-09-18 14:26:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a2204f337d23625cbb7183688b362cd384b49a62","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-18 16:40:08.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/55/682855/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/f0d3837/ : FAILURE in 2h 10m 26s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/55/682855/2/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/2d4692d/ : SUCCESS in 20m 56s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/55/682855/2/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/fd09b27/ : SUCCESS in 17m 05s (non-voting)\n- tripleo-ci-rhel-8-scenario001-standalone-rdo http://logs.rdoproject.org/55/682855/2/openstack-check/tripleo-ci-rhel-8-scenario001-standalone-rdo/1896cf1/ : SUCCESS in 1h 25m 07s\n- tripleo-ci-rhel-8-scenario002-standalone-rdo http://logs.rdoproject.org/55/682855/2/openstack-check/tripleo-ci-rhel-8-scenario002-standalone-rdo/bc9c372/ : FAILURE in 1h 15m 43s","accounts_in_message":[],"_revision_number":2},{"id":"c30fe8cc79cc29ecd314b7b94becee640b4c9462","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 21:12:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c92929c3809e433c8a39e6fe8beeaab0 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5110cc50fa4404fa870db6e878db660 : SUCCESS in 3m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1388ca82cd844ee0953fdd32a765ffe7 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c49be22ac1e443a78dc8a019c068132c : SUCCESS in 5m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1ab6fa024ac742eba6682d3d5dc211e8 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f074b022416947efa681d4b52d2b0966 : SUCCESS in 3m 08s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/895d57d1ddf54c77b5a7c77d3f0af275 : TIMED_OUT in 3h 28m 52s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/18aede3a877c49c8ba01ac2afd13d1d1 : SUCCESS in 2h 11m 03s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/ade88891fdf84b489601f3e822ebc5b0 : SUCCESS in 2h 43m 01s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/a9c097fb0a264aebad0628f55f5b79e0 : SUCCESS in 2h 23m 06s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/dcfad1680ca54484bb53a1bf9d4425b6 : SUCCESS in 2h 44m 33s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/f4bff66b01194db3a27467d9af8bbf57 : FAILURE in 2h 38m 09s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/70233d50a0124220b61db4f2e17f247e : FAILURE in 56m 11s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/3b7b9583898c48359575cd3f92fb8a62 : SUCCESS in 2h 15m 14s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/887f6f3d5f484c778e655faa2082b0f7 : SUCCESS in 2h 02m 36s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/2782cca88cb2463184a4cd7f10c7c6ef : SUCCESS in 1h 51m 02s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/4cef92da24554d6eabd9f8dd37a6a1c2 : SUCCESS in 2h 25m 25s (non-voting)\n- tripleo-ci-centos-7-scenario007-standalone https://zuul.opendev.org/t/openstack/build/a2b02e69ea3d44d19ba2abeb50d6a6c2 : SUCCESS in 1h 27m 08s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5280f85c695347bbb5342c6ec4483116 : SUCCESS in 4m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a670fb03f8a64d5f9edfcad69fdf6a18 : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":2},{"id":"576fb9d97733ba02087cdd034468e343e91bafae","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2019-09-19 08:41:59.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c3fe78ac1d7a6eead35bc7b125b4f21ea828e4e3","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-19 12:05:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/55/682855/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/f4dc701/ : SUCCESS in 3h 13m 50s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/55/682855/2/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/40e0d3f/ : SUCCESS in 22m 19s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/55/682855/2/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/e06aa36/ : SUCCESS in 20m 02s (non-voting)\n- tripleo-ci-rhel-8-scenario001-standalone-rdo http://logs.rdoproject.org/55/682855/2/openstack-check/tripleo-ci-rhel-8-scenario001-standalone-rdo/2f6cb39/ : SUCCESS in 1h 26m 10s\n- tripleo-ci-rhel-8-scenario002-standalone-rdo http://logs.rdoproject.org/55/682855/2/openstack-check/tripleo-ci-rhel-8-scenario002-standalone-rdo/293c241/ : SUCCESS in 1h 20m 16s","accounts_in_message":[],"_revision_number":2},{"id":"c38020c55ddc2376d63d743ecbda063376be2dbf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 01:37:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d1c4c25c9d064b60ae0641498e4494b7 : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6180f50c41a46c3bc46bb476a93dde8 : SUCCESS in 2m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/921ead5376954fb4a572056b294a74eb : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b6a7e8bbce2e4de6ac9e69e01adb659e : SUCCESS in 3m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a909819f4a584d56bbb0c5c06d3be963 : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4559e8c910f48ff955b5ef46af84119 : SUCCESS in 3m 43s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/902f0732c5ed4e1f98a6d27f28d222d9 : SUCCESS in 3h 31m 52s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/d18a74e6cd6f4fa29c4c6d8ea6d990a0 : SUCCESS in 2h 15m 23s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/12e2255dd7634a53994a056a132e966b : SUCCESS in 3h 05m 18s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/8cf41e2252d444d5bf50cd76793665b3 : SUCCESS in 2h 35m 48s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/38b18a917afb4f53b770a8ac8a28fe25 : SUCCESS in 3h 22m 26s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/4d62a81e040c48d9bf1a9bd531d4eed0 : FAILURE in 2h 18m 02s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/9e5e6b8a36ae4433b04d0c51172ea34a : SUCCESS in 2h 47m 43s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/7c15876b03704392b77488615f30fe7c : SUCCESS in 2h 20m 16s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/566225f760784f2ca40c2e53c3b3427d : SUCCESS in 2h 49m 48s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/31a84a3adcb84e7195af6c009cf8a2fc : SUCCESS in 1h 47m 01s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/d7fd404aefcf45059d0590bda3407c90 : SUCCESS in 2h 23m 53s (non-voting)\n- tripleo-ci-centos-7-scenario007-standalone https://zuul.opendev.org/t/openstack/build/267a7865ad8e477eb11c9152f427f057 : SUCCESS in 1h 45m 56s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4568fcb1fb38421f9801c982c7ca06f6 : SUCCESS in 3m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4c4d311bd82d4e7fb7d23d15365ca58d : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":2},{"id":"4ce7d997ac6d07c8204c607ace2a3879b8ad3fc0","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2019-09-20 07:20:16.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"000d989cd9bca3fc51fd5dc315ce6e6cf030792c","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-09-20 09:14:45.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a422163090a7b88432870ac8f10caecb1b97955c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 09:15:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b7566d8c6c8feef037e47c44b9123472d21a7001","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 13:06:15.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f45adaacdaa47e6811583af1f52cd37 : SUCCESS in 4m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ad34d8bab71a467cb8deae0956ef4fc2 : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a02d65fffa4242de8c593ce36d9ceab1 : SUCCESS in 5m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2f99188c93f54dd38250aa1b482524fc : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ae7d91d80a34252a9fbd5decca53594 : SUCCESS in 6m 28s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/12a22c82c03b494495ba94c3b09976c3 : SUCCESS in 3h 23m 23s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/df6a8840ad2b474ea8a9628e8af56ca0 : SUCCESS in 2h 33m 54s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/c04db632182f476ab5cdb8dcffb881b4 : SUCCESS in 3h 03m 04s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/90c3d17246914d92842337a0207e68fe : SUCCESS in 2h 15m 04s\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/4f694c5fd61d4531805fea9300edc9c4 : FAILURE in 1h 13m 06s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/f00da5d022e14504b75484f04eb14eb9 : SUCCESS in 2h 40m 11s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/3520af2c795a4511989c6f9499bae4cd : TIMED_OUT in 3h 33m 14s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/7827da4bf8fd40aea6b5c5d39432d00c : SUCCESS in 1h 57m 28s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/41c0e5f456ed4c9fa16e960b83c66939 : SUCCESS in 5m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/746a51af13c143aaa61fef63897b18d4 : SUCCESS in 5m 21s","accounts_in_message":[],"_revision_number":2},{"id":"1cce477dd0ca9ef90bc72902540facfff73d8d55","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-09-20 13:07:30.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"dc125dca6ea0e0a2bc0f877fecbec558a285d4ed","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-20 16:18:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/55/682855/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/58f2913/ : SUCCESS in 3h 09m 10s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/55/682855/2/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/07047d8/ : SUCCESS in 23m 48s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/55/682855/2/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/7e12655/ : SUCCESS in 18m 08s (non-voting)\n- tripleo-ci-rhel-8-scenario001-standalone-rdo http://logs.rdoproject.org/55/682855/2/openstack-check/tripleo-ci-rhel-8-scenario001-standalone-rdo/fa266cd/ : SUCCESS in 1h 26m 15s\n- tripleo-ci-rhel-8-scenario002-standalone-rdo http://logs.rdoproject.org/55/682855/2/openstack-check/tripleo-ci-rhel-8-scenario002-standalone-rdo/d06456c/ : SUCCESS in 1h 13m 12s","accounts_in_message":[],"_revision_number":2},{"id":"d9aa76db5fbf2cf72344c38d1d00e1f8dfadf37a","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-09-20 19:06:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6ba0b98f64b5fb807ee349e59b3269edfb0a5bbd","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-09-20 19:06:59.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"bfc10c340dd7438921e703d1b2b8f8e52d900269","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-20 22:26:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/55/682855/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/cf9988e/ : SUCCESS in 3h 17m 56s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/55/682855/2/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/69596ee/ : SUCCESS in 22m 01s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/55/682855/2/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/cf11532/ : SUCCESS in 18m 40s (non-voting)\n- tripleo-ci-rhel-8-scenario001-standalone-rdo http://logs.rdoproject.org/55/682855/2/openstack-check/tripleo-ci-rhel-8-scenario001-standalone-rdo/574ec0f/ : SUCCESS in 1h 27m 59s\n- tripleo-ci-rhel-8-scenario002-standalone-rdo http://logs.rdoproject.org/55/682855/2/openstack-check/tripleo-ci-rhel-8-scenario002-standalone-rdo/3141b1a/ : SUCCESS in 1h 13m 15s","accounts_in_message":[],"_revision_number":2},{"id":"ec114f2fd1d1bf2f70a0cb7649655a3a29bd25a9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-21 01:44:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/59f7410f84954c72b68f4634533e95c7 : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97e32803bc244b5fb9fc7fac4fd393d2 : SUCCESS in 2m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1cc152e8fcf642a5ab516ae744ac7829 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f3f65d614ddc43d4bed96517ff77f49d : SUCCESS in 4m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/78dfed4f74da4bcabd87be2f1d858b48 : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07f88f2bcbac43ec867f7c9e7d69f421 : SUCCESS in 3m 51s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/d1a9718da0e3467d9c5de7907f7fe389 : SUCCESS in 3h 34m 12s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/b59e782342c04b27b5134a3d1689830d : SUCCESS in 2h 35m 20s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/0f3bcfc4de46473b8bc349118b7d763b : TIMED_OUT in 3h 49m 59s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/9153f11c9ac046eba6b8b55ceba3de48 : SUCCESS in 2h 54m 39s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/fd47629eec744afdb38f33f5e175a10d : SUCCESS in 3h 25m 29s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/e9d6de49363f4eb782f5b5189e687521 : FAILURE in 3h 06m 50s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/0a67df4d50394c7bb79c738c42d7b121 : FAILURE in 1h 04m 46s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/648720c58bbd4fd48aea0221d28d944c : SUCCESS in 2h 03m 36s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/f18b6ed8800b41e48d53cf61114ccb36 : SUCCESS in 2h 38m 28s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/7ebc76c92fbe477497f8bf57a929fd44 : SUCCESS in 1h 56m 15s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/2a15a395a6c04801b5594de596465531 : FAILURE in 3h 11m 20s (non-voting)\n- tripleo-ci-centos-7-scenario007-standalone https://zuul.opendev.org/t/openstack/build/dbc1b4f5a091433ab6fb5763515d8d2a : SUCCESS in 1h 47m 26s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c90708ea82fb4cfc94928f67aefb2e6d : SUCCESS in 3m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a78ed9e4983d45d3b312f72b3c56f5b9 : SUCCESS in 5m 36s","accounts_in_message":[],"_revision_number":2},{"id":"7f5a1b7c50f0ae96dfd76a346676aea589466e3f","author":{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},"date":"2019-09-21 18:08:54.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"7ba10886868cb32499803036a70c232d0c5ac75e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-21 21:16:59.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/55/682855/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/2ec8ddc/ : FAILURE in 3h 06m 35s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/55/682855/2/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/dcefe64/ : SUCCESS in 21m 10s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/55/682855/2/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/9e9f02d/ : SUCCESS in 17m 58s (non-voting)\n- tripleo-ci-rhel-8-scenario001-standalone-rdo http://logs.rdoproject.org/55/682855/2/openstack-check/tripleo-ci-rhel-8-scenario001-standalone-rdo/189b448/ : SUCCESS in 1h 32m 12s\n- tripleo-ci-rhel-8-scenario002-standalone-rdo http://logs.rdoproject.org/55/682855/2/openstack-check/tripleo-ci-rhel-8-scenario002-standalone-rdo/7252f03/ : SUCCESS in 1h 19m 52s","accounts_in_message":[],"_revision_number":2},{"id":"1aa568bb633d8162ca1167794a89a706f8855132","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-21 22:18:53.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/07448ac3f63e43888ead572d0d4957e7 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e28aba8f4204a47b75447859015ad70 : SUCCESS in 3m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/eddacca79def4ea5b61586abf1f95be9 : SUCCESS in 3m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1093d9dd201c417985257aaefe4935dd : SUCCESS in 3m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3ba77e5999764e5088d78dba464e0a1c : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5194b2caadde4ad4b5611384533ac0c4 : SUCCESS in 4m 07s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/47b7ec0961af4d42a74723c0f28aa1fd : SUCCESS in 3h 38m 23s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/18a583a71f3941d6b2ec405d5cc8c944 : SUCCESS in 2h 16m 29s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/09e712a19d394b6d997903f51744edf4 : SUCCESS in 3h 03m 43s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/495b1d336e5b4d378d22815278f8bdfd : SUCCESS in 3h 01m 28s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/b4421d85f9cd4ccabb0495ed22911397 : SUCCESS in 3h 52m 59s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/af23d2efe59e47329ce609f47194fe79 : FAILURE in 2h 24m 36s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/d6735895ae0c4362b4402e9984cb8153 : FAILURE in 1h 13m 47s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/5176eac023d24ee88849f124bdfd09dd : SUCCESS in 2h 22m 04s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/6ab3d0bd36704f33984b76e6028c9eb3 : SUCCESS in 3h 23m 31s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/7d3185f288fc4a52b53613bb55dd18c4 : SUCCESS in 1h 50m 43s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/fdece13c37374dd88e6493bb7968c007 : FAILURE in 2h 18m 29s (non-voting)\n- tripleo-ci-centos-7-scenario007-standalone https://zuul.opendev.org/t/openstack/build/0107b63f78b44e24bbf34c57dddcb7d6 : SUCCESS in 1h 28m 11s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/dd1cb7baacc549109aef64b1572db77d : SUCCESS in 3m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4f28cd84cb5c4d3a9b7a8cdd3c4213a4 : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":2},{"id":"0d4d6ad58c8ee0aed83faa194db3eaf27001a0eb","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-09-23 11:59:44.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"5e2d4c8bb12679b32da00236c75b1dec51428870","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-23 15:17:20.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/55/682855/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/5670031/ : SUCCESS in 3h 15m 54s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/55/682855/2/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/9e06178/ : SUCCESS in 22m 00s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/55/682855/2/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/75716a2/ : SUCCESS in 16m 59s (non-voting)\n- tripleo-ci-rhel-8-scenario001-standalone-rdo http://logs.rdoproject.org/55/682855/2/openstack-check/tripleo-ci-rhel-8-scenario001-standalone-rdo/6ee198b/ : FAILURE in 20m 17s\n- tripleo-ci-rhel-8-scenario002-standalone-rdo http://logs.rdoproject.org/55/682855/2/openstack-check/tripleo-ci-rhel-8-scenario002-standalone-rdo/7ff9b90/ : FAILURE in 24m 29s","accounts_in_message":[],"_revision_number":2},{"id":"19aabd86052d9f842203f2a0030064d4a0ab10e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 21:14:25.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f86c9ed7a06245ce96d71e90b1a5758d : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4ded87de72b467a82995cd6d88be1cd : SUCCESS in 2m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f0d535059c6d492ba5b455d4082183b9 : SUCCESS in 3m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c910c5d568664618a680ac49e4be1fd8 : SUCCESS in 3m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4a03b491f36745199db01ab61e84f434 : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf65066f97e3482faa3b3ac6409161ff : SUCCESS in 4m 43s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/f2635ba2b75c46cc9a2913e98f352092 : TIMED_OUT in 3h 51m 25s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/6104f901b25a4925881a1d99f3517be5 : SUCCESS in 3h 02m 22s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/2ee62cff489c4575bb982014ee829ac4 : SUCCESS in 3h 37m 12s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/3e64f2ac7dcf4835933fa22dfd9c885e : SUCCESS in 2h 28m 35s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/03dcaf10f4d14524939ed63a116fdeeb : SUCCESS in 3h 13m 19s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/dc5a5005623b41779f4abbb747c1d310 : FAILURE in 2h 40m 27s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/f9b244d522f741019956edea3d0f1cfc : FAILURE in 1h 16m 23s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/29e7b28d54494398a1ad09e36d288feb : SUCCESS in 2h 46m 12s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/b23fc195a6694f19b1e4ae16973e8cca : SUCCESS in 2h 46m 31s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/b353199aeaea4ab89b2bab504df185da : SUCCESS in 1h 57m 41s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/e73dbb6526c3451e892416fa1fcfe62b : FAILURE in 2h 34m 03s (non-voting)\n- tripleo-ci-centos-7-scenario007-standalone https://zuul.opendev.org/t/openstack/build/880c82f0c83c489eb6572746f4814c34 : SUCCESS in 1h 55m 16s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5eed5a063b164699b3c4284a5c9aa276 : SUCCESS in 3m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ab2fd2a529ac43e583a25e400dc02626 : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":2},{"id":"615d3f514f71ffbf72bf3d0c4997fe7afeff9fd2","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2019-09-24 09:26:06.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"e4c27b9bbf824b887f0f1510e349ba60b3bd343d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-24 12:44:37.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/55/682855/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/182d52a/ : SUCCESS in 3h 17m 48s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/55/682855/3/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/5ec82aa/ : SUCCESS in 20m 31s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/55/682855/3/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/31dc29a/ : SUCCESS in 18m 10s (non-voting)\n- tripleo-ci-rhel-8-scenario001-standalone-rdo http://logs.rdoproject.org/55/682855/3/openstack-check/tripleo-ci-rhel-8-scenario001-standalone-rdo/0039635/ : FAILURE in 20m 51s\n- tripleo-ci-rhel-8-scenario002-standalone-rdo http://logs.rdoproject.org/55/682855/3/openstack-check/tripleo-ci-rhel-8-scenario002-standalone-rdo/116801f/ : FAILURE in 20m 51s","accounts_in_message":[],"_revision_number":3},{"id":"44ef716a3423b9bff563430e17ae78b9926b61f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 17:49:26.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb2e481219e24eca9c88f060e39589de : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ea32ae5e0954bee8a25d9586a602b2e : SUCCESS in 3m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8acc1610595f45fba84eb259b32f4af8 : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b8450cdc45724d448992e3aa84776d8c : SUCCESS in 3m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2216abb021dd493bb1a090e2a58f13c4 : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0de381f33aa24459b80165017ecb58fc : SUCCESS in 4m 09s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/c2a549f96639418f83001499a13dc3c3 : SUCCESS in 3h 46m 24s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/930f4022254643d7add480e293f27407 : SUCCESS in 2h 47m 23s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/94622d10a1224a0ea47aab07b502f5f6 : SUCCESS in 3h 19m 44s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/5e03d9837fe04cd082f61a934cf829bd : SUCCESS in 2h 30m 12s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/e3fbde86f6e34badbcd953463f716c11 : FAILURE in 2h 28m 30s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/9bdf139b887d444d875998306166e039 : FAILURE in 2h 34m 16s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/0d5a46836b8e4e8c8369cbbe27f4160f : SUCCESS in 3h 34m 06s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/8a0d7500773449679f0ea1d402798408 : FAILURE in 1h 28m 34s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/2cc9479c8ea24b52b040f6273fcf3cee : SUCCESS in 2h 28m 37s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/141f5663e6b14171b68a77d7296a3cba : SUCCESS in 2h 34m 07s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/46ed8beae6524913b4f6a3b4a3d08601 : FAILURE in 2h 28m 06s (non-voting)\n- tripleo-ci-centos-7-scenario007-standalone https://zuul.opendev.org/t/openstack/build/44a30861fd144c60a46c4c04f9fa4b00 : SUCCESS in 1h 49m 43s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d4aefbea7e37476bafe9d4def741fc16 : SUCCESS in 4m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fbc5502fed2e4a5383f30882a7c6a080 : SUCCESS in 5m 40s","accounts_in_message":[],"_revision_number":3},{"id":"dc71cef4c3ccc37926f3a1f90ebbf09cdb314a66","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-09-24 18:00:00.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"529fe3aaedd230dc94c3d75b0c5788c27e40211d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-24 21:16:25.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/55/682855/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/bb95fc9/ : FAILURE in 3h 12m 10s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/55/682855/3/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/e8bee66/ : SUCCESS in 19m 34s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/55/682855/3/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/2dcd8cc/ : SUCCESS in 18m 46s (non-voting)\n- tripleo-ci-rhel-8-scenario001-standalone-rdo http://logs.rdoproject.org/55/682855/3/openstack-check/tripleo-ci-rhel-8-scenario001-standalone-rdo/d23147a/ : FAILURE in 21m 25s\n- tripleo-ci-rhel-8-scenario002-standalone-rdo http://logs.rdoproject.org/55/682855/3/openstack-check/tripleo-ci-rhel-8-scenario002-standalone-rdo/2073417/ : FAILURE in 21m 39s","accounts_in_message":[],"_revision_number":3},{"id":"d3db2e24ef14c387cd5f9f8036dc527fa72713b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 01:34:40.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/36dbe34452114e9db013a194cc5cfa79 : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33b9915627ec4a28b26cef377e62260d : SUCCESS in 3m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/93ed51c209ce4142a5ccd97ef2d19aa6 : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe60ee41d3bc4855b872878eb6b3f701 : SUCCESS in 3m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b813d4a2af1a41c4b58542499cd06b91 : SUCCESS in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9b18f1cf60f4166870b94e8ab792ee5 : SUCCESS in 4m 20s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/d23f873e90294ab2a8f0aa7b150f5ca9 : TIMED_OUT in 3h 52m 57s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/0c07a5453fa54a188747b84902d75586 : SUCCESS in 2h 18m 28s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/b7330032dee441f0af98ea1397a8d248 : SUCCESS in 3h 24m 04s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/ede0777826e9490a9e7ccdb25cecaa03 : SUCCESS in 2h 32m 40s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/9cf2c5357ca442308f89537f76b9c544 : FAILURE in 2h 49m 39s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/b2d2cae3ec0641beb23d90832c02f79b : FAILURE in 2h 11m 00s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/0a18cee800734c56a5c8732030050b06 : SUCCESS in 3h 21m 37s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/60bbef40795246148ae46373ff1483a6 : SUCCESS in 2h 56m 28s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/04fbf99bcd4241dbae054d6a460473fe : SUCCESS in 2h 18m 38s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/78caa80216de4dfc838f1d03c0125163 : SUCCESS in 1h 51m 34s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/6cdb682cf8cb480087bfa6d3da903f17 : FAILURE in 2h 31m 47s (non-voting)\n- tripleo-ci-centos-7-scenario007-standalone https://zuul.opendev.org/t/openstack/build/e9b051ef32d342ecaf367788791ed47d : SUCCESS in 1h 23m 27s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f5660f5efc064d9a9b7d422266bb1e91 : SUCCESS in 3m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/53ffc4b84f1c4dc7a5a209125e1bfab4 : SUCCESS in 6m 05s","accounts_in_message":[],"_revision_number":3},{"id":"3f38cd60088e85f454ee0351c30e326efab8fa6c","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-09-25 08:38:05.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"24c90475bcab6e8a8bd4670ee656b1ae48f4ea1b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-25 11:43:43.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/55/682855/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/ce38c46/ : FAILURE in 3h 04m 44s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/55/682855/3/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/632c8d6/ : SUCCESS in 22m 04s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/55/682855/3/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/619aeff/ : SUCCESS in 17m 59s (non-voting)\n- tripleo-ci-rhel-8-scenario001-standalone-rdo http://logs.rdoproject.org/55/682855/3/openstack-check/tripleo-ci-rhel-8-scenario001-standalone-rdo/6ec8314/ : FAILURE in 20m 33s\n- tripleo-ci-rhel-8-scenario002-standalone-rdo http://logs.rdoproject.org/55/682855/3/openstack-check/tripleo-ci-rhel-8-scenario002-standalone-rdo/d76f179/ : FAILURE in 20m 34s","accounts_in_message":[],"_revision_number":3},{"id":"8aacd150009c2878d4abcf61f069fbc6f3d21d25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 14:53:59.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cdaa12aaedf848d1b8048e5ee44827a3 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74d0e667e18b4eeb8d946d4a5dcd9fd8 : SUCCESS in 4m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/42f9dbe670cd435caedc9afadca9b1a6 : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6beafbf9e2354625b8bef24ba3da99b6 : SUCCESS in 3m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/470028098f1240dc8463d3b9ba6c2ad9 : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/293a5e6d8c69426a82f2561f618cea51 : SUCCESS in 4m 37s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/1d7e7e72b963423ab8fd54439a45597e : SUCCESS in 3h 32m 22s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/aa15783ee7a043f19d6681f9a7fc1997 : SUCCESS in 2h 59m 29s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/5a22c75cdcaa4573a5e29c044ab4f67a : SUCCESS in 3h 17m 43s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/da299037d57f41438ff6be093c34e54a : SUCCESS in 2h 32m 43s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/a14e0c7ad55346749711d60f17a04506 : FAILURE in 2h 36m 38s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/d7f0a9202e4e49d18dd4501bf64ba70e : FAILURE in 2h 18m 38s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/e15517b7d9ac4476841cd0ad842efbed : FAILURE in 1h 08m 06s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/9792e378c0ce4c528b70de6246207c11 : SUCCESS in 2h 47m 26s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/7ed1df1e10d94305aa1c43719452bb95 : SUCCESS in 2h 44m 37s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/c5adcca406c24ba0be09629f584a8167 : SUCCESS in 2h 08m 10s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/b022426db10442f3b6966c4d1fbf1a89 : FAILURE in 3h 00m 28s (non-voting)\n- tripleo-ci-centos-7-scenario007-standalone https://zuul.opendev.org/t/openstack/build/cc1130b8cde3403d99a31ac7f584936b : SUCCESS in 1h 53m 51s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e340a14465774fbd92bbb63aac9b2648 : SUCCESS in 4m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b237e6d3777241f696400a612b0e0eb2 : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":3},{"id":"188ccbba017e4c0cca9245c9d621efc7366335d7","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-09-25 16:29:47.000000000","message":"Patch Set 3: Workflow-1\n\nIn testing, this may be causing some issues due to ownership changes for /var/tmp/ansible_facts. I\u0027m working through a more thorough set of tests to verify what\u0027s going on and to see if there\u0027s more needed to get this right.","accounts_in_message":[],"_revision_number":3},{"id":"b20228ea7c46c99cbc19b0b24c70c53acda94460","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-09-26 13:32:13.000000000","message":"Patch Set 3: -Code-Review","accounts_in_message":[],"_revision_number":3},{"id":"fe98efdaa71d9a0fe0eb266729c8fe909a9869a7","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-10-02 13:27:33.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c1769704e48b7be6a9747449212312f793b5a4a2","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2019-10-02 14:17:36.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"2c61be16b1338dd9304a1f9870ece38bd1ca59f6","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-10-02 14:35:01.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ac807934f2cd692d8e6a66355941fde4eb0e1731","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-10-02 14:37:23.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"cf3714a6deb51081b74b84dbc7f7833a825b4d28","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-10-02 14:46:27.000000000","message":"Patch Set 6: Code-Review+2\n\nTested in master and it works pretty smoothly.","accounts_in_message":[],"_revision_number":6},{"id":"8195e16dd9a57247d945318aaf53a315126ee352","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-02 18:28:41.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/55/682855/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/6b678ac/ : FAILURE in 3h 46m 19s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/55/682855/6/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/c0eb820/ : SUCCESS in 22m 17s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/55/682855/6/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/9a0015c/ : SUCCESS in 17m 30s (non-voting)\n- tripleo-ci-rhel-8-scenario001-standalone-rdo http://logs.rdoproject.org/55/682855/6/openstack-check/tripleo-ci-rhel-8-scenario001-standalone-rdo/a29ae47/ : FAILURE in 1h 35m 30s\n- tripleo-ci-rhel-8-scenario002-standalone-rdo http://logs.rdoproject.org/55/682855/6/openstack-check/tripleo-ci-rhel-8-scenario002-standalone-rdo/0aad258/ : FAILURE in 22m 53s","accounts_in_message":[],"_revision_number":6},{"id":"f643668eda4e8b27069fed3f36a7274909d29304","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 20:01:24.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/268c74d9b78d4501a7a6bf30861f3fde : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc585be11b7246289f1613cf8a9c856b : SUCCESS in 3m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/af8d456050424edeb6b8f340c35dfbd7 : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4fc1f8bcf3c4fe5ae793f08fac5ad06 : SUCCESS in 3m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bf1e68759e1e4ada866adb1428b55631 : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/298697e336454549b3f96dd46d694685 : SUCCESS in 5m 01s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/bf47a44264d646a1aee679b8b0c00a1e : SUCCESS in 3h 58m 19s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/6aad449f31ab4ca3ba79dc97e5072150 : SUCCESS in 2h 24m 57s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/6f0f403c200242a9ab28ea52e3f5fdb7 : SUCCESS in 2h 27m 00s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/2539e7c54fb24751961138925cb8d125 : SUCCESS in 2h 46m 54s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/196bfaf056f64345938dcd840648d446 : SUCCESS in 2h 22m 44s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/d3cba7ec27904fc581335ea08d3a2655 : FAILURE in 2h 08m 20s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/f9a03b31dc4d46fe867b40faa488df99 : FAILURE in 2h 06m 38s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/2cd38ba4001545d0957f106c48720f0e : SUCCESS in 2h 44m 45s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/2ef473bb310646e7a59ef4fd10a9790b : SUCCESS in 2h 20m 38s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/d98f02bcd4f54ef7882bba0135ffa2f2 : SUCCESS in 2h 07m 44s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/c7100c5f20984375bd9c5a78be39cd8c : FAILURE in 1h 32m 29s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/7da7fadcaddd4130a4e73803531efe8f : FAILURE in 1h 21m 52s (non-voting)\n- tripleo-ci-centos-7-scenario007-standalone https://zuul.opendev.org/t/openstack/build/8e30f42689b34af6a027f6940a1c36e5 : SUCCESS in 1h 37m 12s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/414014a6003a4c6c841ff0f53f76c5b3 : SUCCESS in 4m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ff6afc387ce54f7caa96dda61efe8fdc : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":6},{"id":"4987491261b3e5a02c43ec91ccfc1c02ebfc4889","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-10-03 09:36:38.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"8cfe5c5776a9ace13920bdc46fee7bbd37891be3","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-03 12:34:59.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/55/682855/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/c8c0523/ : SUCCESS in 2h 57m 25s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/55/682855/6/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/66c395f/ : SUCCESS in 23m 02s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/55/682855/6/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/7c2f23e/ : SUCCESS in 20m 13s (non-voting)\n- tripleo-ci-rhel-8-scenario001-standalone-rdo http://logs.rdoproject.org/55/682855/6/openstack-check/tripleo-ci-rhel-8-scenario001-standalone-rdo/d3c7efa/ : FAILURE in 20m 35s\n- tripleo-ci-rhel-8-scenario002-standalone-rdo http://logs.rdoproject.org/55/682855/6/openstack-check/tripleo-ci-rhel-8-scenario002-standalone-rdo/1ee9784/ : FAILURE in 21m 12s","accounts_in_message":[],"_revision_number":6},{"id":"19d3a38c4d2f76c071b3719895d3888efc95b514","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 13:11:31.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a943615cb2a04cef88e496fbd8c41327 : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ace6aaba58d74a97b1ca69a84cb000cb : SUCCESS in 3m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dc352427c0c1464eb2078f4f3dc98905 : SUCCESS in 3m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a896d5594634e768286a0a0f63b0f4f : SUCCESS in 3m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/95f94a7658f44ccfa150371dcde4fb68 : SUCCESS in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3df1d7bf9ee45d9a0f826cce2193db1 : SUCCESS in 3m 41s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/7fe83ba723314ea8b26de481b5cb4bed : SUCCESS in 3h 14m 56s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/4a31709a34b744628a55006b3d18826c : SUCCESS in 2h 03m 59s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/c77c1ce5ba6d45318ed0c5850f9ac46c : SUCCESS in 2h 47m 21s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/c7adc61117e64b249c6d11d74a0de6ce : SUCCESS in 2h 46m 13s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/26141d1357594984a7e0f5b4a8ab5cb2 : SUCCESS in 2h 29m 23s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/1d331b6cf03e453988eb3c90c5d07075 : FAILURE in 2h 01m 19s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/34cadcbecade4bc68aedea13847eb380 : FAILURE in 1h 56m 16s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/b9265cb12391441eb89e7543638b05a0 : SUCCESS in 2h 27m 18s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/436398e5f51345c4aa61d575f88f93af : SUCCESS in 2h 15m 48s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/ff688f3959a5494082f75a51732dc614 : SUCCESS in 2h 14m 37s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/2cec9954e7da41de9dabdfdcf30f8385 : FAILURE in 1h 44m 16s (non-voting)\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/bbb74b541858407fa454302d7a998da7 : FAILURE in 1h 39m 47s (non-voting)\n- tripleo-ci-centos-7-scenario007-standalone https://zuul.opendev.org/t/openstack/build/83ecc27f238042b3bea01408578d8e9b : SUCCESS in 1h 25m 17s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b6ab157faebf4f8087797bcb193900dc : SUCCESS in 3m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c422febc24a24f019286d6848d288375 : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":6},{"id":"d14e06d49a9cfda79092b265c62ad24c690e8281","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2019-10-03 13:15:27.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"2ba31aa1c4fd977e2bfe35c0d4da433dd908f898","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 13:15:42.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"b0b2d44de97a54d98487bb6388ddce646592772a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 19:40:16.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4805c5fde3c84dbcbf6cfdd640820d23 : SUCCESS in 4m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8c67a21842a142b39924aa6535e926fd : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/882944e84fc341c9a937e12a6de7755f : SUCCESS in 3m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5b517a6ff7dd404c8b84ed23fb9eb98a : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f20f8292a6514bd9aad1cc8fcdbe4d5e : SUCCESS in 4m 38s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/a7013d907b524294839b9969c1a731ab : SUCCESS in 3h 12m 47s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/db60c8d64b024ddf8ae93766eb7790c9 : SUCCESS in 2h 17m 14s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/1e253cd6b29a47aab0f3b9d83e13d9e6 : SUCCESS in 2h 56m 36s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/bda266ccce994f49a710cc1199f811ef : SUCCESS in 2h 14m 42s\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/53513366cd39467ba4fd9c0fa0adefe0 : SUCCESS in 2h 45m 32s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/b3df38eba41947e5baaf3e6d7d8ea6bb : SUCCESS in 1h 48m 32s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/21ed4ecfbf414edea4852f49be179439 : SUCCESS in 2h 34m 16s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0f42f8b649ba471f9ab8442fac7f9ba0 : SUCCESS in 5m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/94fc6d7727654442927a300de763e11f : SUCCESS in 5m 37s","accounts_in_message":[],"_revision_number":6},{"id":"3024cbe14db62fe3b39a26eb9e05ba274202137e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 19:40:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"ec5441608f1d5cd4111eb5878e034879ff556456","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 19:41:26.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/291374959153489c812acfbe10c92f05 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":6},{"id":"46926a7f48efab35a80238d8aa63fc1caa9be388","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2019-10-04 07:45:48.000000000","message":"Patch Set 6: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 6048a985a59e40a1dd7bb6dedda4293d2b8a885d","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"bbc0d767bbd87f10d69c71b35652d65bb631ba6c","revisions":{"1f3bab5efd0527793f4cae737c50fd56b148d8a4":{"kind":"REWORK","_number":1,"created":"2019-09-18 11:07:38.000000000","uploader":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"ref":"refs/changes/55/682855/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/55/682855/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/55/682855/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/55/682855/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/55/682855/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/55/682855/1"}}},"commit":{"parents":[{"commit":"8ced75035593d1653df3f8442be9487ba29c7840","subject":"Merge \"Ignore failure calling cell_v2_discover_hosts from deployment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/8ced75035593d1653df3f8442be9487ba29c7840"}]}],"author":{"name":"Jesse Pretorius (odyssey4me)","email":"jesse@odyssey4.me","date":"2019-09-18 11:02:05.000000000","tz":60},"committer":{"name":"Jesse Pretorius (odyssey4me)","email":"jesse@odyssey4.me","date":"2019-09-18 11:02:05.000000000","tz":60},"subject":"Implement Ansible fact cache for Mistral executor","message":"Implement Ansible fact cache for Mistral executor\n\nMost of the playbooks executed by Mistral gather facts which, by default,\nare stored in memory until the conclusion of the playbook execution. This\nmeans that for every new playbook run the facts for each host are gathered\nagain. Fact gathering can take a long time, especially when the target host\nhas a lot of devices, which is often the case for compute hosts (many\nnetwork devices and many storage devices) and for controller hosts (many\nnetwork devices).\n\nTo optimise fact gathering, we set the default behaviour for Mistral\u0027s\nAnsible execution to store gathered facts using the jsonfile format in a\ntemporary location, and we tell it to use smart gathering so that it will\nonly gather new facts when a new playbook executes.\n\nThis should cut down the deployment time a bit, which saves time in CI and\nin production environments.\n\nChange-Id: Ie7adf41cef0fafde646f840b2bf3e01ba42a17d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/1f3bab5efd0527793f4cae737c50fd56b148d8a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/1f3bab5efd0527793f4cae737c50fd56b148d8a4"}]},"branch":"refs/heads/master"},"2f6b8ad1e0eca4dd1739a18f4a55ae7c1f8431b9":{"kind":"REWORK","_number":2,"created":"2019-09-18 14:26:49.000000000","uploader":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"ref":"refs/changes/55/682855/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/55/682855/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/55/682855/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/55/682855/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/55/682855/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/55/682855/2"}}},"commit":{"parents":[{"commit":"8ced75035593d1653df3f8442be9487ba29c7840","subject":"Merge \"Ignore failure calling cell_v2_discover_hosts from deployment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/8ced75035593d1653df3f8442be9487ba29c7840"}]}],"author":{"name":"Jesse Pretorius (odyssey4me)","email":"jesse@odyssey4.me","date":"2019-09-18 11:02:05.000000000","tz":60},"committer":{"name":"Jesse Pretorius (odyssey4me)","email":"jesse@odyssey4.me","date":"2019-09-18 14:24:27.000000000","tz":60},"subject":"Implement Ansible fact cache for Mistral executor","message":"Implement Ansible fact cache for Mistral executor\n\nMost of the playbooks executed by Mistral gather facts which, by default,\nare stored in memory until the conclusion of the playbook execution. This\nmeans that for every new playbook run the facts for each host are gathered\nagain. Fact gathering can take a long time, especially when the target host\nhas a lot of devices, which is often the case for compute hosts (many\nnetwork devices and many storage devices) and for controller hosts (many\nnetwork devices).\n\nTo optimise fact gathering, we set the default behaviour for Mistral\u0027s\nAnsible execution to store gathered facts using the jsonfile format in a\ntemporary location, and we tell it to use smart gathering so that it will\nonly gather new facts when a new playbook executes.\n\nThis should cut down the deployment time a bit, which saves time in CI and\nin production environments.\n\nRelated: rhbz#1749406\nChange-Id: Ie7adf41cef0fafde646f840b2bf3e01ba42a17d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/2f6b8ad1e0eca4dd1739a18f4a55ae7c1f8431b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/2f6b8ad1e0eca4dd1739a18f4a55ae7c1f8431b9"}]},"branch":"refs/heads/master"},"b3911fe94c47066a29aad7792a979ac999180e35":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-09-24 09:26:06.000000000","uploader":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"ref":"refs/changes/55/682855/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/55/682855/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/55/682855/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/55/682855/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/55/682855/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/55/682855/3"}}},"commit":{"parents":[{"commit":"84d3e3425451fe2c6fb220f5801c3acdcbdb5244","subject":"Merge \"Add profile_tasks to ansible runs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/84d3e3425451fe2c6fb220f5801c3acdcbdb5244"}]}],"author":{"name":"Jesse Pretorius (odyssey4me)","email":"jesse@odyssey4.me","date":"2019-09-18 11:02:05.000000000","tz":60},"committer":{"name":"mathieu bultel","email":"mat.bultel@gmail.com","date":"2019-09-24 09:26:06.000000000","tz":0},"subject":"Implement Ansible fact cache for Mistral executor","message":"Implement Ansible fact cache for Mistral executor\n\nMost of the playbooks executed by Mistral gather facts which, by default,\nare stored in memory until the conclusion of the playbook execution. This\nmeans that for every new playbook run the facts for each host are gathered\nagain. Fact gathering can take a long time, especially when the target host\nhas a lot of devices, which is often the case for compute hosts (many\nnetwork devices and many storage devices) and for controller hosts (many\nnetwork devices).\n\nTo optimise fact gathering, we set the default behaviour for Mistral\u0027s\nAnsible execution to store gathered facts using the jsonfile format in a\ntemporary location, and we tell it to use smart gathering so that it will\nonly gather new facts when a new playbook executes.\n\nThis should cut down the deployment time a bit, which saves time in CI and\nin production environments.\n\nRelated: rhbz#1749406\nChange-Id: Ie7adf41cef0fafde646f840b2bf3e01ba42a17d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/b3911fe94c47066a29aad7792a979ac999180e35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/b3911fe94c47066a29aad7792a979ac999180e35"}]},"branch":"refs/heads/master"},"eff1374f711ead4d5ff64f535f74fd6bdc5c791b":{"kind":"REWORK","_number":4,"created":"2019-10-02 13:27:33.000000000","uploader":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"ref":"refs/changes/55/682855/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/55/682855/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/55/682855/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/55/682855/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/55/682855/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/55/682855/4"}}},"commit":{"parents":[{"commit":"33cb21a89b190c0458f78f9c283f96063ae5ae01","subject":"Merge \"Warn when looking up grandparent resource name\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/33cb21a89b190c0458f78f9c283f96063ae5ae01"}]}],"author":{"name":"Jesse Pretorius (odyssey4me)","email":"jesse@odyssey4.me","date":"2019-09-18 11:02:05.000000000","tz":60},"committer":{"name":"Jesse Pretorius (odyssey4me)","email":"jesse@odyssey4.me","date":"2019-10-02 13:24:12.000000000","tz":60},"subject":"Implement Ansible fact cache for Mistral executor","message":"Implement Ansible fact cache for Mistral executor\n\nMost of the playbooks executed by Mistral gather facts which, by default,\nare stored in memory until the conclusion of the playbook execution. This\nmeans that for every new playbook run the facts for each host are gathered\nagain. Fact gathering can take a long time, especially when the target host\nhas a lot of devices, which is often the case for compute hosts (many\nnetwork devices and many storage devices) and for controller hosts (many\nnetwork devices).\n\nTo optimise fact gathering, we set the default behaviour for Mistral\u0027s\nAnsible execution to store gathered facts using the jsonfile format in a\ntemporary location, and we tell it to use smart gathering so that it will\nonly gather new facts when a new playbook executes.\n\nThe validations are run using the \u0027validations\u0027 user, it cannot share the\nsame fact cache because the /var/tmp/ansible_fact_cache folder will be owned\nby the \u0027mistral\u0027 user, and is therefore not writable by the \u0027validations\u0027\nuser. We therefore set the validations to use a different path.\n\nThis should cut down the deployment time a bit, which saves a little time in\nCI and saves a lot of time in production environments.\n\nRelated: rhbz#1749406\nChange-Id: Ie7adf41cef0fafde646f840b2bf3e01ba42a17d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/eff1374f711ead4d5ff64f535f74fd6bdc5c791b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/eff1374f711ead4d5ff64f535f74fd6bdc5c791b"}]},"branch":"refs/heads/master"},"a78866c35231fdacb18b07bd26fffd614110c434":{"kind":"REWORK","_number":5,"created":"2019-10-02 14:35:01.000000000","uploader":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"ref":"refs/changes/55/682855/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/55/682855/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/55/682855/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/55/682855/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/55/682855/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/55/682855/5"}}},"commit":{"parents":[{"commit":"33cb21a89b190c0458f78f9c283f96063ae5ae01","subject":"Merge \"Warn when looking up grandparent resource name\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/33cb21a89b190c0458f78f9c283f96063ae5ae01"}]}],"author":{"name":"Jesse Pretorius (odyssey4me)","email":"jesse@odyssey4.me","date":"2019-09-18 11:02:05.000000000","tz":60},"committer":{"name":"Jesse Pretorius (odyssey4me)","email":"jesse@odyssey4.me","date":"2019-10-02 14:34:56.000000000","tz":60},"subject":"Implement Ansible fact cache for Mistral executor","message":"Implement Ansible fact cache for Mistral executor\n\nMost of the playbooks executed by Mistral gather facts which, by default,\nare stored in memory until the conclusion of the playbook execution. This\nmeans that for every new playbook run the facts for each host are gathered\nagain. Fact gathering can take a long time, especially when the target host\nhas a lot of devices, which is often the case for compute hosts (many\nnetwork devices and many storage devices) and for controller hosts (many\nnetwork devices).\n\nTo optimise fact gathering, we set the default behaviour for Mistral\u0027s\nAnsible execution to store gathered facts using the jsonfile format in a\ntemporary location, and we tell it to use smart gathering so that it will\nonly gather new facts when a new playbook executes.\n\nThe validations are run using the \u0027validations\u0027 user, it cannot share the\nsame fact cache because the /var/tmp/ansible_fact_cache folder will be owned\nby the \u0027mistral\u0027 user, and is therefore not writable by the \u0027validations\u0027\nuser. We therefore set the validations to use a different path.\n\nThis should cut down the deployment time a bit, which saves a little time in\nCI and saves a lot of time in production environments.\n\nRelated: rhbz#1749406\nChange-Id: Ie7adf41cef0fafde646f840b2bf3e01ba42a17d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/a78866c35231fdacb18b07bd26fffd614110c434"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/a78866c35231fdacb18b07bd26fffd614110c434"}]},"branch":"refs/heads/master"},"bbc0d767bbd87f10d69c71b35652d65bb631ba6c":{"kind":"REWORK","_number":6,"created":"2019-10-02 14:37:23.000000000","uploader":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"ref":"refs/changes/55/682855/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/55/682855/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/55/682855/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/55/682855/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/55/682855/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/55/682855/6"}}},"commit":{"parents":[{"commit":"33cb21a89b190c0458f78f9c283f96063ae5ae01","subject":"Merge \"Warn when looking up grandparent resource name\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/33cb21a89b190c0458f78f9c283f96063ae5ae01"}]}],"author":{"name":"Jesse Pretorius (odyssey4me)","email":"jesse@odyssey4.me","date":"2019-09-18 11:02:05.000000000","tz":60},"committer":{"name":"Jesse Pretorius (odyssey4me)","email":"jesse@odyssey4.me","date":"2019-10-02 14:37:18.000000000","tz":60},"subject":"Implement Ansible fact cache for Mistral executor","message":"Implement Ansible fact cache for Mistral executor\n\nMost of the playbooks executed by Mistral gather facts which, by default,\nare stored in memory until the conclusion of the playbook execution. This\nmeans that for every new playbook run the facts for each host are gathered\nagain. Fact gathering can take a long time, especially when the target host\nhas a lot of devices, which is often the case for compute hosts (many\nnetwork devices and many storage devices) and for controller hosts (many\nnetwork devices).\n\nTo optimise fact gathering, we set the default behaviour for Mistral\u0027s\nAnsible execution to store gathered facts using the jsonfile format in a\ntemporary location, and we tell it to use smart gathering so that it will\nonly gather new facts when a new playbook executes.\n\nThe validations are run using the \u0027validations\u0027 user, it cannot share the\nsame fact cache because the /var/tmp/ansible_fact_cache folder will be owned\nby the \u0027mistral\u0027 user, and is therefore not writable by the \u0027validations\u0027\nuser. We therefore set the validations to use a different path.\n\nThis should cut down the deployment time a bit, which saves a little time in\nCI and saves a lot of time in production environments.\n\nRelated: rhbz#1749406\nChange-Id: Ie7adf41cef0fafde646f840b2bf3e01ba42a17d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/bbc0d767bbd87f10d69c71b35652d65bb631ba6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/bbc0d767bbd87f10d69c71b35652d65bb631ba6c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
