)]}'
{"id":"openstack%2Ftripleo-common~685692","triplet_id":"openstack%2Ftripleo-common~stable%2Fqueens~If3841210a5d5cf125058cc3e5cf4de2905fae42b","project":"openstack/tripleo-common","branch":"stable/queens","topic":"numa-aware-vswitch-parameters-issue-stable/stein-stable/rocky-stable/queens","hashtags":[],"change_id":"If3841210a5d5cf125058cc3e5cf4de2905fae42b","subject":"OvS DPDK derive parameters timeout issue","status":"MERGED","created":"2019-09-30 11:45:20.000000000","updated":"2019-10-01 03:11:25.000000000","submitted":"2019-10-01 03:11:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"685692-1569899485629-8f0a65ef","meta_rev_id":"5082b0c2c47d16945432544a2c153086cf362956","_number":685692,"virtual_id_number":685692,"owner":{"_account_id":22865,"name":"Jaganathan Palanisamy","email":"jpalanis@redhat.com","username":"jaganathancse"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-01 03:11:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-09-30 13:58:36.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-09-30 19:29:08.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-09-30 19:29:08.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2019-09-30 11:46:59.000000000","_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"value":0,"date":"2019-09-30 11:46:53.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"date":"2019-09-30 11:47:15.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-30 11:46:48.000000000","_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-30 11:46:48.000000000","updated_by":{"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"},"reviewer":{"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"},"state":"REVIEWER"},{"updated":"2019-09-30 11:46:53.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2019-09-30 11:46:59.000000000","updated_by":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"reviewer":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"state":"REVIEWER"},{"updated":"2019-09-30 11:47:15.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-09-30 13:58:36.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-30 19:29:08.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-10-01 03:11:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ffb46d0a229c7e5fda48b38c3751bb3dfcc812b6","author":{"_account_id":22865,"name":"Jaganathan Palanisamy","email":"jpalanis@redhat.com","username":"jaganathancse"},"date":"2019-09-30 11:45:20.000000000","message":"Patch Set 1: Cherry Picked from branch stable/rocky.","accounts_in_message":[],"_revision_number":1},{"id":"e8fac62858e5442ab077d6d524a1d58c551b1b1f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-30 13:58:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/92/685692/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/576dec2/ : FAILURE in 2h 11m 44s","accounts_in_message":[],"_revision_number":1},{"id":"4ca428d977321e5144a074b8498f533cd2eb5b72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-30 14:58:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-oooq https://zuul.opendev.org/t/openstack/build/cf93c3fae67c4dd293f2855d04f636b7 : SUCCESS in 1h 44m 05s\n- tripleo-ci-centos-7-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/1175fdcc514b4b568272f91d69c006c7 : SUCCESS in 1h 35m 18s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/df77371506304e05a010a1efb5f94e67 : SUCCESS in 3h 05m 58s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/56a5b70c3a424f54989d17b0cbf4a147 : SUCCESS in 2h 44m 32s (non-voting)\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/ded2da07de154df7b8e5ea248297d43c : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6eb55fa660fc4833b248cdeedc35fd49 : SUCCESS in 2m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f5d3e41eeed54c1da70bf856f5002d5e : SUCCESS in 3m 00s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e907bb1bacdd4f5788b779d57a44d47e : SUCCESS in 3m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9130de67d8b14c4ba932ea802e691754 : SUCCESS in 3m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"50601cee6c5cdb14314d5d3950639fd22563aacc","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-09-30 19:29:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a024d2615f804c0e74c9bb62b99fe880cfe16837","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-30 19:29:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d916aa200e3ffbeed52fe6ab5c778d025262cfa7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 03:11:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-undercloud-oooq https://zuul.opendev.org/t/openstack/build/73fa8ace416040c4b5e103fa5f1803f6 : SUCCESS in 1h 45m 00s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/76b6531bfb2146cca766d6973892a281 : SUCCESS in 2h 43m 02s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/7c10d6a52a1c476cbd18c6c0a828211a : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/835e3f2ff62e4bdbb99b0e39092c2479 : SUCCESS in 4m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3e46bf326e5446df991428f84c09ca48 : SUCCESS in 4m 27s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e8b0013c03a349a29289f00ef2ed0f11 : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":1},{"id":"9d4e13a12754874fc06442324e63ba020f168aab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 03:11:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"513a1607702934ee223a9e5c25c733841ab41c7d","revisions":{"513a1607702934ee223a9e5c25c733841ab41c7d":{"kind":"REWORK","_number":1,"created":"2019-09-30 11:45:20.000000000","uploader":{"_account_id":22865,"name":"Jaganathan Palanisamy","email":"jpalanis@redhat.com","username":"jaganathancse"},"ref":"refs/changes/92/685692/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/92/685692/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/92/685692/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/92/685692/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/92/685692/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/92/685692/1"}}},"commit":{"parents":[{"commit":"e2208117a31eea79c50330333501b6581f9faafe","subject":"Merge \"Derives NUMA aware vSwitches parameters\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/e2208117a31eea79c50330333501b6581f9faafe"}]}],"author":{"name":"Jaganathan Palanisamy","email":"jpalanis@redhat.com","date":"2019-08-30 09:13:29.000000000","tz":330},"committer":{"name":"Jaganathan Palanisamy","email":"jpalanis@redhat.com","date":"2019-09-30 11:45:20.000000000","tz":0},"subject":"OvS DPDK derive parameters timeout issue","message":"OvS DPDK derive parameters timeout issue\n\nThis change is to fix the OvS DPDK derive parameters timeout\nissue. OvS DPDK parameters and NUMA aware vSwitch parameters are not\nderived and getting timeout issue.\n\nChange-Id: If3841210a5d5cf125058cc3e5cf4de2905fae42b\nCloses-Bug: #1842042\n(cherry picked from commit 135e65e02f9382dd0e1aaa131b694dd5f5c8d29a)\n(cherry picked from commit 7efce38a7b754d0999a0a1580598a78d30289d99)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/513a1607702934ee223a9e5c25c733841ab41c7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/513a1607702934ee223a9e5c25c733841ab41c7d"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
