)]}'
{"id":"openstack%2Ftripleo-common~697214","triplet_id":"openstack%2Ftripleo-common~master~I78656961fc915964477e0ba9bb766c679361b60e","project":"openstack/tripleo-common","branch":"master","topic":"mistral_to_ansible","hashtags":[],"change_id":"I78656961fc915964477e0ba9bb766c679361b60e","subject":"Convert shell ansible calls to ansible-runner","status":"ABANDONED","created":"2019-12-04 01:23:58.000000000","updated":"2020-04-16 14:05:25.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ada2053a86068ebd8b0e832ddb47c9deea9de549","_number":697214,"virtual_id_number":697214,"owner":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"all":[{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-12-07 07:25:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-12-07 06:08:17.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-04 01:29:46.000000000","updated_by":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"reviewer":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"state":"REVIEWER"},{"updated":"2019-12-04 10:49:08.000000000","updated_by":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"reviewer":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"state":"REVIEWER"},{"updated":"2019-12-05 15:50:57.000000000","updated_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"reviewer":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"state":"REVIEWER"},{"updated":"2019-12-05 22:09:03.000000000","updated_by":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"reviewer":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"state":"REVIEWER"},{"updated":"2019-12-07 06:08:17.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-07 07:25:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"66ef91d78f7cde9af0360b1458457fd94d91c224","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-12-04 01:23:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"34a9f1f3c49e6f16dda48c751c86e912a5206eb6","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-12-04 01:27:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a6f4a69f63216f36cf76c2491bcde2ea8fb3d4fe","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-04 01:57:05.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/14/697214/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/9001ff9/ : FAILURE in 22m 19s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/14/697214/2/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/d84ecc5/ : FAILURE in 19m 13s\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/14/697214/2/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/1f0c1ae/ : SUCCESS in 20m 24s (non-voting)\n- tripleo-ci-rhel-8-scenario001-standalone-rdo http://logs.rdoproject.org/14/697214/2/openstack-check/tripleo-ci-rhel-8-scenario001-standalone-rdo/35aa99e/ : FAILURE in 28m 29s\n- tripleo-ci-rhel-8-scenario002-standalone-rdo http://logs.rdoproject.org/14/697214/2/openstack-check/tripleo-ci-rhel-8-scenario002-standalone-rdo/9487774/ : FAILURE in 27m 51s\n- tripleo-ci-rhel-8-scenario003-standalone-rdo http://logs.rdoproject.org/14/697214/2/openstack-check/tripleo-ci-rhel-8-scenario003-standalone-rdo/aa1cd58/ : FAILURE in 25m 49s\n- tripleo-ci-rhel-8-scenario004-standalone-rdo http://logs.rdoproject.org/14/697214/2/openstack-check/tripleo-ci-rhel-8-scenario004-standalone-rdo/f78ca22/ : FAILURE in 27m 09s","accounts_in_message":[],"_revision_number":2},{"id":"29266c4c10e423c8d7fea6442e2b7e401be5bcbb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 02:52:55.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/15baeb4b206c4e6cbce60ce549fe395a : SUCCESS in 1m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/93b08651238744a8a66a62ec6d7d5689 : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01f66290153a4d88ae589a7b3c0febae : SUCCESS in 2m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/68ed4d806b844c87920775cf0c0c6e33 : FAILURE in 3m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/436d88aed7df46ed86b7d3a5ec0d70b9 : SUCCESS in 3m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/826307f0b5c84a2b82890ad4bb6e733a : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fa6713560624debb73aedef88604183 : SUCCESS in 3m 08s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/3667167784f2420ab89889fdc2c7fa2b : SUCCESS in 48m 46s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/b9d49274616640a0a292e39c7c1633c3 : SUCCESS in 1h 18m 07s\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/af827c97ee8947188d7cc3b274a1e141 : FAILURE in 26m 33s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario007-standalone https://zuul.opendev.org/t/openstack/build/f3ee7c25154144229197a3ee7e0d8f9b : FAILURE in 25m 35s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a47c204e660e4c619a5078501edf6657 : SUCCESS in 4m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/02d5673d22ec4340b3d04e9145b2a1d7 : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":2},{"id":"e9227882c4d7f1cbcda50c9843c9f03c8555bf66","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-12-04 03:44:15.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"755baa568d7f20540ee69cb950a9bf1647fb0ecd","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-04 04:17:38.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/14/697214/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/883f8b1/ : FAILURE in 24m 27s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/14/697214/3/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/9becf0e/ : SUCCESS in 24m 24s\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/14/697214/3/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/932e2eb/ : SUCCESS in 18m 08s (non-voting)\n- tripleo-ci-rhel-8-scenario001-standalone-rdo http://logs.rdoproject.org/14/697214/3/openstack-check/tripleo-ci-rhel-8-scenario001-standalone-rdo/d48c0db/ : FAILURE in 26m 34s\n- tripleo-ci-rhel-8-scenario002-standalone-rdo http://logs.rdoproject.org/14/697214/3/openstack-check/tripleo-ci-rhel-8-scenario002-standalone-rdo/a03b80c/ : FAILURE in 26m 57s\n- tripleo-ci-rhel-8-scenario003-standalone-rdo http://logs.rdoproject.org/14/697214/3/openstack-check/tripleo-ci-rhel-8-scenario003-standalone-rdo/f1f5843/ : FAILURE in 27m 14s\n- tripleo-ci-rhel-8-scenario004-standalone-rdo http://logs.rdoproject.org/14/697214/3/openstack-check/tripleo-ci-rhel-8-scenario004-standalone-rdo/cc7e0f0/ : FAILURE in 27m 21s","accounts_in_message":[],"_revision_number":3},{"id":"b231c46046af4ee9f94de1d6623da397ac48a741","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 05:25:27.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/8695f550a56147aebb308725189cac4c : SUCCESS in 2m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1e99a4a07a7249a1862cb57ddcefc6ca : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91be3c461e464256b5579855b60ee8a0 : SUCCESS in 3m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e15a980374ea4b5097431855a74193fe : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44a05b89bca94921904e1adcbcec46c0 : SUCCESS in 4m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3532010ce8a84c5eab29c190d1fb2038 : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27134dffea9b4aa08bbdaea8005407de : SUCCESS in 4m 19s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/4840e615e3fe4e55960df5b9eee36573 : FAILURE in 29m 39s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/40b990e29513420ebdf39c567ceea7ee : FAILURE in 25m 14s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/beec6581e86e43d989139e9e37d24199 : FAILURE in 33m 02s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/d548de31a6d64f6d81288e0db45b2fa0 : FAILURE in 24m 30s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/46f7d98db93044f798f51d0a026bad9c : FAILURE in 1h 18m 29s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/ade81570321b46d8957f7beddce3f690 : FAILURE in 1h 24m 31s (non-voting)\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/8197f3f6afbe4d609115f3523ccd2f74 : SUCCESS in 40m 48s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/acaede2867804a36a93a9c72f929144e : SUCCESS in 1h 09m 33s\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/22a34a5846aa41e391b52226442315f0 : FAILURE in 24m 29s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/126dc8cc6284410bbd011cde52716f7d : FAILURE in 25m 20s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/05fee404d2c84ae6ac39cd095aa8f92d : FAILURE in 24m 31s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/3bf75ed7d24c4b16a8101946826d6084 : FAILURE in 24m 28s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/38b5b7db352947b7ab104753d520ad1a : FAILURE in 24m 36s (non-voting)\n- tripleo-ci-centos-7-scenario007-standalone https://zuul.opendev.org/t/openstack/build/470d718836b84cb09ada289030793722 : FAILURE in 22m 38s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ed38e0a78d1749bebc085d64e702aea3 : SUCCESS in 3m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3bb21af8f8364f97a39e97699e0ec6c1 : SUCCESS in 5m 48s","accounts_in_message":[],"_revision_number":3},{"id":"a31449f04c2b50ff2e9e2ef621557577e460f6d7","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-12-04 10:49:08.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"cef974eb1d6ff6ad5e37685f6299a7678cf8b48e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 12:20:07.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/9c5c0f09da4d4fc4b682f05a7ce31ba3 : SUCCESS in 3m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cb5fc7a61b5b4c558293338943ad02d7 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/780bc3f8c35545c99553e382c11ee995 : SUCCESS in 2m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fe440b7e16934e0d831b82e540df86bd : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ed4f5149092640c98e410575075e53a2 : SUCCESS in 3m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2cd8b3ca9bc949759b60d14db0eee86a : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7194d34785c5465dbeb42bd19923b1cb : SUCCESS in 5m 26s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/1b826d78f3b7476c8247d19b4bb97d26 : FAILURE in 29m 37s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/78f36d67e6a34c0f9d52598e04c73464 : FAILURE in 27m 31s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/7c3a96844c6142bab6b3ec623c96a707 : FAILURE in 25m 22s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/5b70997c76f74baf9c06d9ba57f6a33e : FAILURE in 21m 53s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/850441b8244943e09cf9e0c62e9bf97b : FAILURE in 1h 15m 35s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/d620a3b8ad514570b27ae3134145f9bc : FAILURE in 1h 15m 56s (non-voting)\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/fe29ff97a60e4eddb31350a6f20a9368 : SUCCESS in 58m 50s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/85780408484f45269428b3cf0910b91c : SUCCESS in 1h 25m 24s\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/2c37487a452a4d09a18a0e12e690f0cd : FAILURE in 20m 31s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/9cea5447feca4d72882fef1417c98593 : FAILURE in 25m 14s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/c01f9016ee6a4d7480df6b20dd39b49c : FAILURE in 28m 38s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/e5e255499fcc4d9f8aefc5804102223f : FAILURE in 30m 25s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/3e4a4c41ed164831b886457eacfa01ff : FAILURE in 28m 22s (non-voting)\n- tripleo-ci-centos-7-scenario007-standalone https://zuul.opendev.org/t/openstack/build/38a0867ca560409eb2612eb85ca40aed : FAILURE in 28m 01s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e7a9f797708c43f29d558818f2bf5c93 : SUCCESS in 3m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7030ac9d7d6d49e3a3a65a597b5421b7 : SUCCESS in 5m 48s","accounts_in_message":[],"_revision_number":3},{"id":"0ecdab987054f1ef7f6f4cfd27ee8c828379336f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-04 12:56:46.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/14/697214/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/c97d5f3/ : FAILURE in 26m 09s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/14/697214/3/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/dcea0c1/ : SUCCESS in 23m 07s\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/14/697214/3/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/099c389/ : SUCCESS in 20m 53s (non-voting)\n- tripleo-ci-rhel-8-scenario001-standalone-rdo http://logs.rdoproject.org/14/697214/3/openstack-check/tripleo-ci-rhel-8-scenario001-standalone-rdo/5b2f155/ : FAILURE in 30m 07s\n- tripleo-ci-rhel-8-scenario002-standalone-rdo http://logs.rdoproject.org/14/697214/3/openstack-check/tripleo-ci-rhel-8-scenario002-standalone-rdo/d199d19/ : FAILURE in 1h 53m 30s\n- tripleo-ci-rhel-8-scenario003-standalone-rdo http://logs.rdoproject.org/14/697214/3/openstack-check/tripleo-ci-rhel-8-scenario003-standalone-rdo/c2f3c99/ : FAILURE in 27m 15s\n- tripleo-ci-rhel-8-scenario004-standalone-rdo http://logs.rdoproject.org/14/697214/3/openstack-check/tripleo-ci-rhel-8-scenario004-standalone-rdo/219c31e/ : FAILURE in 2h 02m 17s","accounts_in_message":[],"_revision_number":3},{"id":"d814ef49e97503dbe3e3c1762fba674fa858e279","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-12-05 01:07:39.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"74b86eccc406e0b840db603249b75ce6582f7f63","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-05 01:38:26.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/14/697214/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/dc28fd2/ : FAILURE in 29m 08s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/14/697214/3/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/c91301e/ : FAILURE in 17m 30s\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/14/697214/3/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/91e5fd4/ : SUCCESS in 18m 28s (non-voting)\n- tripleo-ci-rhel-8-scenario001-standalone-rdo http://logs.rdoproject.org/14/697214/3/openstack-check/tripleo-ci-rhel-8-scenario001-standalone-rdo/c81f15d/ : FAILURE in 28m 39s\n- tripleo-ci-rhel-8-scenario002-standalone-rdo http://logs.rdoproject.org/14/697214/3/openstack-check/tripleo-ci-rhel-8-scenario002-standalone-rdo/d01d3d3/ : FAILURE in 26m 57s\n- tripleo-ci-rhel-8-scenario003-standalone-rdo http://logs.rdoproject.org/14/697214/3/openstack-check/tripleo-ci-rhel-8-scenario003-standalone-rdo/f0099c7/ : FAILURE in 17m 03s\n- tripleo-ci-rhel-8-scenario004-standalone-rdo http://logs.rdoproject.org/14/697214/3/openstack-check/tripleo-ci-rhel-8-scenario004-standalone-rdo/8458609/ : FAILURE in 25m 32s","accounts_in_message":[],"_revision_number":3},{"id":"8966855ca620a3e8dc6e0a21e02f2804c9eae0b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-05 03:03:39.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/e8b2af5c7c574470b2bb9fac9376ef46 : SUCCESS in 2m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3e0d9a2775774bc890fbf8cfc417fede : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9edaefb275d43b5b597e841ab5222a4 : SUCCESS in 2m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/abcd59705a204fd0bd75747339b8d205 : SUCCESS in 3m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a8919b0b2af249ba938551198351e6f7 : SUCCESS in 4m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9f24c6e302084610aae8097da3d0b39c : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ac33a72423c4c758c3b4f013a088437 : SUCCESS in 4m 28s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/7a960f2270cd4840937276bd5ba1a343 : FAILURE in 32m 31s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/d08a4b7aaed440dcaa9d55f8f4c83a5e : FAILURE in 30m 38s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/eeae1349c14c42ed9c3c0daa0367b402 : FAILURE in 33m 18s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/8beb045831fd497aa5d20d78dba0629c : FAILURE in 28m 04s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/6fd73538b06c427b8184a1d2343d435f : FAILURE in 1h 33m 47s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/6e5efe3ee983407a872e4d88eec31d3f : FAILURE in 1h 37m 07s (non-voting)\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/598dedb00c204e658048d52e548c5175 : SUCCESS in 49m 03s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/367978f79e244b89b2890ccc76f2df8a : SUCCESS in 1h 17m 14s\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/17e5de71d2524d7b8c34d686f694c762 : FAILURE in 30m 24s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/1c66cea66d304c54864b19cc950e5700 : FAILURE in 35m 18s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/af0199ff7a5042188dff87bffd988586 : FAILURE in 27m 24s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/4482d075e0354676b76ff99da2632ab3 : FAILURE in 33m 21s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/d2f33d89d66d49e7a694e59e44e0b9c6 : FAILURE in 28m 48s (non-voting)\n- tripleo-ci-centos-7-scenario007-standalone https://zuul.opendev.org/t/openstack/build/21df9046c6c843a1bb6a90c1e15ad8bb : FAILURE in 31m 35s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b26c5ea3fdd242c29824cd096178dbef : SUCCESS in 5m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/430e80834cb5431d8ceefa47d4a2d741 : SUCCESS in 5m 33s","accounts_in_message":[],"_revision_number":3},{"id":"a0cdb03805bf354fe06d524abe9b9436f46c65d8","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2019-12-05 22:09:03.000000000","message":"Patch Set 3:\n\nwe can migrate to ansible-runner, but we have to be able to maintain the same featureset we have today with ansible-playbook.\n\nAs a user, you need to be able to cd to the work dir under /var/lib/mistral and easily reproduce the ansible execution. Right now, we write out a bash script that contains the exact ansible-playbook command that the mistral action uses.\n\nThe bash script also accepts $@, so as a user, i can easily pass additional args to debug if needed (such as --tags, --limit). This is a well known interface for ansible users, so that they don\u0027t have to learn something new.\n\nIt doesn\u0027t have to be exactly *like* that, but we at least need the equivalent functionality somehow. I\u0027ve not tried the patch yet, but will do so soon. Just wanted to leave this initial feedback.","accounts_in_message":[],"_revision_number":3},{"id":"edfb6a492351135b47ad98cd2f22e1b81d80b9d1","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-12-07 05:33:25.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"66d4c57b8b575ddad3ba17125ff251a239ea7cf7","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-07 06:08:17.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/14/697214/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/9546902/ : FAILURE in 28m 56s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/14/697214/4/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/58c7136/ : FAILURE in 19m 49s\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/14/697214/4/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/9f91512/ : SUCCESS in 19m 06s (non-voting)\n- tripleo-ci-rhel-8-scenario001-standalone-rdo http://logs.rdoproject.org/14/697214/4/openstack-check/tripleo-ci-rhel-8-scenario001-standalone-rdo/6427b34/ : FAILURE in 32m 49s\n- tripleo-ci-rhel-8-scenario002-standalone-rdo http://logs.rdoproject.org/14/697214/4/openstack-check/tripleo-ci-rhel-8-scenario002-standalone-rdo/813e3d1/ : FAILURE in 27m 09s\n- tripleo-ci-rhel-8-scenario003-standalone-rdo http://logs.rdoproject.org/14/697214/4/openstack-check/tripleo-ci-rhel-8-scenario003-standalone-rdo/f350eaa/ : FAILURE in 27m 49s\n- tripleo-ci-rhel-8-scenario004-standalone-rdo http://logs.rdoproject.org/14/697214/4/openstack-check/tripleo-ci-rhel-8-scenario004-standalone-rdo/64ecac2/ : FAILURE in 28m 55s","accounts_in_message":[],"_revision_number":4},{"id":"e3a309152e8a49fc5def4655ee3eb2aaf3de14d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-07 07:25:48.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/8fe3c636e13a4aa280eed33ce75c676b : SUCCESS in 1m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5790fb3a41ac40c1b68e53adf9670111 : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05cb07e070a74e16a4d7678fd62fc62e : SUCCESS in 2m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/19ab6409a20445058b102ecd772a45b3 : SUCCESS in 3m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/72de57252eb34fb4a1824b19c0568391 : SUCCESS in 3m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/09e9c692e07e42a79f94baa6909589c3 : SUCCESS in 3m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1da486bf496b4c9c990bcf188c49e82e : SUCCESS in 3m 29s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/d32e2281452c4ecabbd49499ad54d72b : FAILURE in 37m 55s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/ab34a91c9e93497ab95cd45c01e9c9e8 : FAILURE in 30m 09s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/ed5eaa9a84b549979e3a782dc77f83fd : FAILURE in 40m 31s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/7227b6979efe4d23bb8c3128c9db104d : FAILURE in 32m 47s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/c4e751e9fe7343e9912d519c3de0caf6 : FAILURE in 1h 23m 11s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/ba1dcf0ac93d4f29ba572853b30f3d38 : FAILURE in 1h 38m 59s (non-voting)\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/fc02cc1f3884497a818c81d57d2c5a4e : SUCCESS in 55m 47s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/56e7abb4e4794a348147b7b93098da1f : SUCCESS in 1h 14m 39s\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/22872bfdc47140319f4f5a1a0d11781c : FAILURE in 35m 31s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/7604657877d24aebbda863d706599089 : FAILURE in 35m 59s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/66d89e7cbc504bcc9f80304ddcd2a727 : FAILURE in 29m 30s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/94660a37903945479e6276b504258adf : FAILURE in 35m 39s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/1702a4a36d8c42ec80bd45e2eb5759c8 : FAILURE in 28m 23s (non-voting)\n- tripleo-ci-centos-7-scenario007-standalone https://zuul.opendev.org/t/openstack/build/9d3369e9d9cf49d8a00efa35f5e92805 : FAILURE in 30m 08s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/885e5cd6684b4df2ba62fa242c5ba85e : SUCCESS in 3m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1769aad7b39248cd9df98193d07c347d : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":4},{"id":"8cc256dc8c162ff0ed702eb0dadc1acbe6a0c113","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2020-04-16 14:05:25.000000000","message":"Abandoned\n\nthis is no longer needed","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"b9b031c7d77f52cd4f2ed6d3cbe521f4e88485e6","revisions":{"4c2e8606410aa15bf145c4569410b646a5f649b9":{"kind":"REWORK","_number":1,"created":"2019-12-04 01:23:58.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/14/697214/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/14/697214/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/14/697214/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/14/697214/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/14/697214/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/14/697214/1"}}},"commit":{"parents":[{"commit":"d8679da62e38d02cd0dca7248a2506793e954e18","subject":"Update hostvars lookup to fix regression","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/d8679da62e38d02cd0dca7248a2506793e954e18"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2019-12-04 01:21:59.000000000","tz":-360},"committer":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2019-12-04 01:21:59.000000000","tz":-360},"subject":"Convert shell ansible calls to ansible-runner","message":"Convert shell ansible calls to ansible-runner\n\nThe ansible-runner library is now being used to run all ansible interactions\nwithin tripleo-common. This interface will ensure that we\u0027re able to maintain\na consistent and well scoped interface for ansible playbooks and commands\nwhich will assist us as we transition mistral workflows to ansible playbooks.\n\nChange-Id: I78656961fc915964477e0ba9bb766c679361b60e\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/4c2e8606410aa15bf145c4569410b646a5f649b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/4c2e8606410aa15bf145c4569410b646a5f649b9"}]},"branch":"refs/heads/master"},"a9ccbf1335ac509f90131da9be8a2938d74781d9":{"kind":"REWORK","_number":2,"created":"2019-12-04 01:27:18.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/14/697214/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/14/697214/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/14/697214/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/14/697214/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/14/697214/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/14/697214/2"}}},"commit":{"parents":[{"commit":"d8679da62e38d02cd0dca7248a2506793e954e18","subject":"Update hostvars lookup to fix regression","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/d8679da62e38d02cd0dca7248a2506793e954e18"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2019-12-04 01:21:59.000000000","tz":-360},"committer":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2019-12-04 01:27:12.000000000","tz":-360},"subject":"Convert shell ansible calls to ansible-runner","message":"Convert shell ansible calls to ansible-runner\n\nThe ansible-runner library is now being used to run all ansible interactions\nwithin tripleo-common. This interface will ensure that we\u0027re able to maintain\na consistent and well scoped interface for ansible playbooks and commands\nwhich will assist us as we transition mistral workflows to ansible playbooks.\n\nChange-Id: I78656961fc915964477e0ba9bb766c679361b60e\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/a9ccbf1335ac509f90131da9be8a2938d74781d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/a9ccbf1335ac509f90131da9be8a2938d74781d9"}]},"branch":"refs/heads/master"},"1b40d807badb0387d2bc8dab49feae0cbb037451":{"kind":"REWORK","_number":3,"created":"2019-12-04 03:44:15.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/14/697214/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/14/697214/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/14/697214/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/14/697214/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/14/697214/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/14/697214/3"}}},"commit":{"parents":[{"commit":"d8679da62e38d02cd0dca7248a2506793e954e18","subject":"Update hostvars lookup to fix regression","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/d8679da62e38d02cd0dca7248a2506793e954e18"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2019-12-04 01:21:59.000000000","tz":-360},"committer":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2019-12-04 03:44:08.000000000","tz":-360},"subject":"Convert shell ansible calls to ansible-runner","message":"Convert shell ansible calls to ansible-runner\n\nThe ansible-runner library is now being used to run all ansible interactions\nwithin tripleo-common. This interface will ensure that we\u0027re able to maintain\na consistent and well scoped interface for ansible playbooks and commands\nwhich will assist us as we transition mistral workflows to ansible playbooks.\n\nChange-Id: I78656961fc915964477e0ba9bb766c679361b60e\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/1b40d807badb0387d2bc8dab49feae0cbb037451"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/1b40d807badb0387d2bc8dab49feae0cbb037451"}]},"branch":"refs/heads/master"},"b9b031c7d77f52cd4f2ed6d3cbe521f4e88485e6":{"kind":"REWORK","_number":4,"created":"2019-12-07 05:33:25.000000000","uploader":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"ref":"refs/changes/14/697214/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/14/697214/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/14/697214/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/14/697214/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/14/697214/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/14/697214/4"}}},"commit":{"parents":[{"commit":"d8679da62e38d02cd0dca7248a2506793e954e18","subject":"Update hostvars lookup to fix regression","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/d8679da62e38d02cd0dca7248a2506793e954e18"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2019-12-04 01:21:59.000000000","tz":-360},"committer":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-12-07 05:33:15.000000000","tz":0},"subject":"Convert shell ansible calls to ansible-runner","message":"Convert shell ansible calls to ansible-runner\n\nThe ansible-runner library is now being used to run all ansible interactions\nwithin tripleo-common. This interface will ensure that we\u0027re able to maintain\na consistent and well scoped interface for ansible playbooks and commands\nwhich will assist us as we transition mistral workflows to ansible playbooks.\n\nChange-Id: I78656961fc915964477e0ba9bb766c679361b60e\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/b9b031c7d77f52cd4f2ed6d3cbe521f4e88485e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/b9b031c7d77f52cd4f2ed6d3cbe521f4e88485e6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
