)]}'
{"id":"openstack%2Ftripleo-common~709767","triplet_id":"openstack%2Ftripleo-common~master~Iec115bfa1ef00112705785a5b3e6ed7dc222070e","project":"openstack/tripleo-common","branch":"master","topic":"bug/1864660","hashtags":[],"change_id":"Iec115bfa1ef00112705785a5b3e6ed7dc222070e","subject":"Use config for layers","status":"MERGED","created":"2020-02-25 16:03:35.000000000","updated":"2020-03-02 15:57:08.000000000","submitted":"2020-02-27 04:42:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"709767-1582778535579-20d59637","meta_rev_id":"707dbd2dbac33064d0acd69d844fdbbd319baa2c","_number":709767,"virtual_id_number":709767,"owner":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2020-02-26 16:40:42.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-27 04:42:15.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-02-26 01:45:03.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-02-25 17:13:24.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2020-02-25 21:32:20.000000000","_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-02-26 13:53:53.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-25 21:32:20.000000000","updated_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"reviewer":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"state":"REVIEWER"},{"updated":"2020-02-26 01:45:03.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-26 13:53:53.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-02-27 04:42:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d3ab81620f529dcbb25c381030a2b69b9544deac","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-02-25 16:03:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"be30c71a7de8e191d66a3abb20ce77a82fd4af1e","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-02-25 16:18:37.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"420cea2059e9673a33e56666fa107bf39f0876f8","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-02-25 16:44:43.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1207cb39f2a3e4a552d29556e1d9db4ec1c2037e","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-02-25 17:13:24.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e362573f36950f8308e2ea7680e3d3d5c980ab4e","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2020-02-25 21:32:20.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"0acfeb554bb61850f7199548ceeac6388008f5aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 22:06:58.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5e379d3efec045fe998a753ed16581de : SUCCESS in 7m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/119aea2064d14fb481c8ed72f8d70944 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb3cd9ccabbd49098865364774caa9b3 : SUCCESS in 5m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/826b73186c19478eb799064f2e4f6335 : SUCCESS in 8m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c6fb2b233b80415eb6f23277d9639c50 : SUCCESS in 7m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5edb966cad954475af6049a591fd1b5d : SUCCESS in 6m 33s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/ee936539d97741f5a12e08b73d1076c7 : SUCCESS in 2h 29m 51s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/350cab8af2ee47f48c475912a5665272 : SUCCESS in 2h 04m 54s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/824706eb4f17488dbf74fe51b33d40e9 : SUCCESS in 2h 45m 27s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/62cb48a0c6a14041ac75031424436f08 : SUCCESS in 1h 49m 24s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/538bb0f676ef4f7daae9dfafc1f1556a : SUCCESS in 2h 23m 58s (non-voting)\n- tripleo-ci-centos-7-containers-undercloud-minion https://zuul.opendev.org/t/openstack/build/23af9fc083e94156a567743f81806c99 : FAILURE in 1h 39m 13s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/ac886c06c991427e919eda19ae57385b : SUCCESS in 3h 11m 12s (non-voting)\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/f6c87920dbab4edf90eb7b5b2006d9e0 : SUCCESS in 46m 50s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8c275d13957942f0b0386f53183744bc : SUCCESS in 7m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c1610f216c78494d9e1ba92dc3fd6953 : SUCCESS in 6m 59s","accounts_in_message":[],"_revision_number":2},{"id":"af78abda27af52a1d1d46848becfa7e3867385b7","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-25 22:16:21.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/4756e2785ac149f9aeec1393ba7b3375 : SUCCESS in 3h 20m 15s\n- tripleo-rhel-8-buildimage-overcloud-full https://review.rdoproject.org/zuul/build/4b365d81b5af41b6b205416bff311ebe : FAILURE in 15m 08s\n- tripleo-rhel-8-buildimage-ironic-python-agent https://review.rdoproject.org/zuul/build/e6dfd575320a4a02ac846af5f326c857 : SUCCESS in 16m 37s (non-voting)\n- tripleo-ci-rhel-8-scenario001-standalone-rdo https://review.rdoproject.org/zuul/build/4b01c5be2874469e9db888b03d0e624a : SUCCESS in 1h 26m 14s\n- tripleo-ci-rhel-8-scenario002-standalone-rdo https://review.rdoproject.org/zuul/build/b4044550708b42a49f8d60bf10fc1474 : SUCCESS in 1h 27m 09s\n- tripleo-ci-rhel-8-scenario003-standalone-rdo https://review.rdoproject.org/zuul/build/aac3661149df4b1a9fea49a2bac14515 : SUCCESS in 1h 00m 12s\n- tripleo-ci-rhel-8-scenario004-standalone-rdo https://review.rdoproject.org/zuul/build/92538f0ad3d543fead4ed97a523468d8 : SUCCESS in 1h 13m 47s","accounts_in_message":[],"_revision_number":2},{"id":"cab5940f61ba615debe46229c95fb7661d511a0e","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-02-25 22:21:53.000000000","message":"Patch Set 2:\n\nrecheck minion failed on something unrelated to this change\n\nTASK [tripleo_container_manage : Check podman create status] *******************\nTuesday 25 February 2020  20:01:15 +0000 (0:00:00.469)       0:06:23.595 ****** \nFAILED - RETRYING: Check podman create status (30 retries left).\nFAILED - RETRYING: Check podman create status (29 retries left).\nFAILED - RETRYING: Check podman create status (28 retries left).\nFAILED - RETRYING: Check podman create status (27 retries left).\nfailed: [minion] (item\u003dNone) \u003d\u003e changed\u003dfalse \n  attempts: 5\n  censored: \u0027the output has been hidden due to the fact that \u0027\u0027no_log: true\u0027\u0027 was specified for this result\u0027\nfatal: [minion]: FAILED! \u003d\u003e changed\u003dfalse \n  censored: \u0027the output has been hidden due to the fact that \u0027\u0027no_log: true\u0027\u0027 was specified for this result\u0027","accounts_in_message":[],"_revision_number":2},{"id":"1bb31dc63ae98195a70795b7ec565b31a6f4bcdd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 01:29:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6ca1f95d4fcb466784dc33e13ed0b078 : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ada26bf69ee842d8a02c79fb33f0cdac : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/85e58e0ceac24f7395efa6fe5db7b3b3 : SUCCESS in 6m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a8cf3f42f9ee4d559b0513193ef5a72d : SUCCESS in 8m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/71fedb63e6f042acbe18334b5db2074e : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42261ab4d36542c794dd03d6e202056e : SUCCESS in 6m 01s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/a8bc87f4f71f48b993fd8eea87c7c04f : SUCCESS in 2h 52m 07s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/258489e0b0a34daf8f1b5bd7096e6e07 : SUCCESS in 1h 51m 34s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/b0cc7c9398c641b7989dcf25948a7461 : FAILURE in 1h 08m 59s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/49bf882ef75348b0881f1e45d8c83c89 : SUCCESS in 2h 05m 04s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/d108eddcb78f48bda6bed526c6af820c : SUCCESS in 2h 40m 56s (non-voting)\n- tripleo-ci-centos-7-containers-undercloud-minion https://zuul.opendev.org/t/openstack/build/177ca86c2091435397e966d97a81d8b7 : SUCCESS in 1h 43m 21s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/d1aec5cf90c244bc973d94c45b1bd406 : SUCCESS in 2h 36m 11s (non-voting)\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/b2a33224e9354d04b975ebed7033990e : SUCCESS in 40m 57s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/bdc495b35ccf4c299f5151201c3a1507 : SUCCESS in 5m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ba02b98f716d4f8089196eddd6e1886e : SUCCESS in 7m 21s","accounts_in_message":[],"_revision_number":2},{"id":"11a2466b67a779bff4023459ccd9895a7b5365dd","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-26 01:45:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/9a198466f21241dcadd2bc724f202004 : SUCCESS in 3h 05m 55s\n- tripleo-rhel-8-buildimage-overcloud-full https://review.rdoproject.org/zuul/build/c5286c4aa0cd404189a28e17c74ab480 : SUCCESS in 20m 18s\n- tripleo-rhel-8-buildimage-ironic-python-agent https://review.rdoproject.org/zuul/build/8207421e89994d5db910b74b663d9a8a : SUCCESS in 15m 55s (non-voting)\n- tripleo-ci-rhel-8-scenario001-standalone-rdo https://review.rdoproject.org/zuul/build/9f87edad0e5a4491a1c682e6e11e9d9b : SUCCESS in 1h 25m 42s\n- tripleo-ci-rhel-8-scenario002-standalone-rdo https://review.rdoproject.org/zuul/build/115817e2954b4594a83b6fd718f57fa4 : SUCCESS in 1h 19m 57s\n- tripleo-ci-rhel-8-scenario003-standalone-rdo https://review.rdoproject.org/zuul/build/007f2ba75db64284a1264fa527d5d120 : SUCCESS in 1h 07m 22s\n- tripleo-ci-rhel-8-scenario004-standalone-rdo https://review.rdoproject.org/zuul/build/05f99cfd811945a19c7f00ba2308bd1c : SUCCESS in 1h 11m 12s","accounts_in_message":[],"_revision_number":2},{"id":"434b1f6b187121b611ca483fcaf2719f9b1b9988","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-02-26 13:53:53.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"072e8b47f3c443d1ac7f0901037d29aeb32d27ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 13:55:33.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"761356a5f79fd1dc042e062105b0d76c02edeaca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 16:34:47.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e09a6fe11b574b1cb52af26fc76eaa02 : SUCCESS in 6m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f0d446a9c248456581a7ffe1b25a777a : SUCCESS in 7m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/514a6285b4934b96bd438f36d1956a43 : SUCCESS in 7m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5db35a2019bb49e997e98dd98a805dda : SUCCESS in 6m 12s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/dbc8eb60f1ee410da23f180764986b80 : FAILURE in 1h 51m 22s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/59aced1e39804f0ba7f3501d5545818e : SUCCESS in 1h 46m 48s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/6884f0ee00c2481d9d4cce8b8b04aeec : SUCCESS in 2h 33m 46s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/21f9afac1b8e4dd293b5631f14f33ece : SUCCESS in 2h 02m 35s\n- tripleo-ci-centos-7-containers-undercloud-minion https://zuul.opendev.org/t/openstack/build/c2bd62edaada48bf81ec5b78a591c1f4 : SUCCESS in 1h 41m 14s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/65cf1bbf0a034378bd4b2f816848b4e4 : SUCCESS in 8m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cd0366f9e70b45a18d9891c4e27d3d5a : SUCCESS in 8m 15s","accounts_in_message":[],"_revision_number":2},{"id":"d0f39212bcc3378468b1f34d1d2e964c27b3cf45","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-02-26 16:40:42.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f2d0c5d13aa56443b6f6645b9acd49db1d8a5bd8","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-26 20:27:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/be7f2f39c0a64ed2968286e486f3ca04 : SUCCESS in 3h 03m 25s\n- tripleo-rhel-8-buildimage-overcloud-full https://review.rdoproject.org/zuul/build/5aec631daea04e53b0df73e143d9a46e : SUCCESS in 20m 11s\n- tripleo-rhel-8-buildimage-ironic-python-agent https://review.rdoproject.org/zuul/build/7d0a7d8103f44d94a78f2874e217db3a : SUCCESS in 18m 48s (non-voting)\n- tripleo-ci-rhel-8-scenario001-standalone-rdo https://review.rdoproject.org/zuul/build/572d6e500a4b45828db7c70c33eb84e0 : SUCCESS in 1h 23m 54s\n- tripleo-ci-rhel-8-scenario002-standalone-rdo https://review.rdoproject.org/zuul/build/d214a2f7d17c40cfad091ffc62d63822 : SUCCESS in 1h 14m 21s\n- tripleo-ci-rhel-8-scenario003-standalone-rdo https://review.rdoproject.org/zuul/build/74f907015b61444aad9a15a9bf43c787 : SUCCESS in 1h 04m 22s\n- tripleo-ci-rhel-8-scenario004-standalone-rdo https://review.rdoproject.org/zuul/build/b3cbc50b12454278801ede73f00e5c78 : SUCCESS in 1h 08m 25s","accounts_in_message":[],"_revision_number":2},{"id":"bb956981baa81c345db8ec16ae2a390cc20fd1ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 23:48:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f63094d0ea44f02b25f9953aa877d45 : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab20e944e69a49c3a184c319217abb55 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/80a91fb1dceb49998971a8ee87cb90cc : SUCCESS in 5m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e19b906e1e3f40ecb8dae2652419485b : SUCCESS in 6m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6647cc50af10482c8793f3ca195c79ff : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbb189f826754d329a25335f5759db0b : SUCCESS in 5m 05s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/77f745cceba9443a96ed84120f324b7f : SUCCESS in 2h 32m 48s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/f97af46a87014e62bb2214b22c34e510 : SUCCESS in 1h 57m 13s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/a0d53a6fa4dd491b97d28d30b410e046 : SUCCESS in 2h 41m 55s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/e751fbd01ee343fbb93758e6580ea66b : SUCCESS in 1h 58m 50s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/e113ef615bee4b4d80066f9b24528e93 : SUCCESS in 2h 22m 53s (non-voting)\n- tripleo-ci-centos-7-containers-undercloud-minion https://zuul.opendev.org/t/openstack/build/294d3b607367481cb82d9b3d0357137b : SUCCESS in 1h 56m 58s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/8eec0bac64304fb8b4e2ae9be3509789 : SUCCESS in 2h 53m 15s (non-voting)\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/e268bec8ade44fad81f8141095742922 : SUCCESS in 49m 01s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/18cd7aff34c54326b2ba5d84816eb875 : SUCCESS in 5m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c9047e5e98714f8486626de9e5f3b5a8 : SUCCESS in 6m 23s","accounts_in_message":[],"_revision_number":2},{"id":"0904975d51391eb91b6bf69cc86942d010c31948","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 23:49:21.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f47ed6877650d3a5345b03f75f3884cd3279c240","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 04:42:15.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05848585c51842f28f275f8f8fb16f9e : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d6a8f0e416a845d381eea3e11ac15978 : SUCCESS in 6m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/88c277d6f9874f30a991483a09f36f3c : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6173158677594bf8b4f790b7bac02d46 : SUCCESS in 4m 51s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/51e6e736ea14487da75d7e9238c749b0 : SUCCESS in 2h 55m 12s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/5102b75099b9498383f861aad52e24dc : SUCCESS in 1h 56m 13s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/79bcd4aeacc1427f90000adde20de7dc : SUCCESS in 2h 34m 40s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/b146e88e6b924570bd98c8c3d9aee72f : SUCCESS in 1h 52m 32s\n- tripleo-ci-centos-7-containers-undercloud-minion https://zuul.opendev.org/t/openstack/build/975ecd34c0ce40dd80c9a0846f914f8a : SUCCESS in 1h 45m 46s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/810c8f81814f42088b559cae4440bfac : SUCCESS in 5m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b848dbbb2f914fd5a510c0c4bfbd4b47 : SUCCESS in 8m 21s","accounts_in_message":[],"_revision_number":2},{"id":"5de1334818673b258b70e8ac0d3718520bc90765","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 04:42:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"d6ef4e14407c1972d45de035390ff02602ccfb88","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 04:44:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44bf03f6ca124f43bce4844fc6b2a2ea : SUCCESS in 1m 29s","accounts_in_message":[],"_revision_number":2},{"id":"0335874b724d915a5818c8a37acc4814263c224d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-02-27 14:49:54.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 89a7df07a2b555de9163f699d12661b185b4a920","accounts_in_message":[],"_revision_number":2},{"id":"ce70d17ce9f7edd6bdd2f802b0e7c0f3c12ad43f","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2020-03-02 15:48:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"46057d14fbb8900b96b2dcd798de269116ba1646","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-03-02 15:57:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2950738526d7b93553102ceb77717d31c445caa2","revisions":{"d7287fb07042f2b1e44ae761db539ef2f4f0e23a":{"kind":"REWORK","_number":1,"created":"2020-02-25 16:03:35.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/67/709767/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/67/709767/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/67/709767/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/67/709767/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/67/709767/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/67/709767/1"}}},"commit":{"parents":[{"commit":"2e9332488c8c91b69045049da58e5bf23531b154","subject":"Merge \"Move away from \"ss\" and drop default verbose mode\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/2e9332488c8c91b69045049da58e5bf23531b154"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2020-02-25 16:03:12.000000000","tz":-420},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2020-02-25 16:03:21.000000000","tz":-420},"subject":"WIP: Use config for layers","message":"WIP: Use config for layers\n\nCloses-Bug: #1864660\nChange-Id: Iec115bfa1ef00112705785a5b3e6ed7dc222070e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/d7287fb07042f2b1e44ae761db539ef2f4f0e23a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/d7287fb07042f2b1e44ae761db539ef2f4f0e23a"}]},"branch":"refs/heads/master"},"2950738526d7b93553102ceb77717d31c445caa2":{"kind":"REWORK","_number":2,"created":"2020-02-25 16:44:43.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/67/709767/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/67/709767/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/67/709767/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/67/709767/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/67/709767/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/67/709767/2"}}},"commit":{"parents":[{"commit":"2e9332488c8c91b69045049da58e5bf23531b154","subject":"Merge \"Move away from \"ss\" and drop default verbose mode\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/2e9332488c8c91b69045049da58e5bf23531b154"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2020-02-25 16:03:12.000000000","tz":-420},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2020-02-25 16:42:12.000000000","tz":-420},"subject":"Use config for layers","message":"Use config for layers\n\nPodman keeps a copy of the metadata around in the local container\nstorage however this may not actually represent the actual layers being\nused in the storage. Instead of using the manifest file as the source of\ntruth for the container layers, we need to look into the container image\nconfig for the layer ids for the local storage. This change adds a\nfunction to swap out the layer metadata with layer information from the\nimage config if available. If ultimately we cannot find the layers we\nthrow an exception indicating that it is not found.\n\nCloses-Bug: #1864660\nChange-Id: Iec115bfa1ef00112705785a5b3e6ed7dc222070e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/2950738526d7b93553102ceb77717d31c445caa2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/2950738526d7b93553102ceb77717d31c445caa2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
