)]}'
{"id":"openstack%2Ftripleo-common~731739","triplet_id":"openstack%2Ftripleo-common~master~I131aca7a03802766a791e6912fad6ebf143d31db","project":"openstack/tripleo-common","branch":"master","topic":"buildtestpackages","hashtags":[],"change_id":"I131aca7a03802766a791e6912fad6ebf143d31db","subject":"[DNM] build test packages downstream testing","status":"ABANDONED","created":"2020-05-29 12:04:06.000000000","updated":"2020-07-04 15:07:59.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e39672153c65ddbcb9a729cb3755ba8247fb5e81","_number":731739,"virtual_id_number":731739,"owner":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-06-30 11:43:27.000000000","_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-06-30 10:12:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-30 08:58:20.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-30 08:58:20.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-30 10:12:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4c6849dbd142d4747d0d740390a49a747a890302","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-05-29 12:04:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"23b5440421531940371cf8561325531935c0d618","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 14:47:31.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"746ef8f3c1ae3225294f990225e719440c594115","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-29 14:47:33.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"ac4c6ecdfd8db8ad6413f50c30af7383acfec9f9","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-06-02 13:14:36.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"d757ce9a651548bfd236be4db2489a26ec10eba5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 15:37:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1dc102b952ba4345a4b34891fb4f9fcb : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a171ca0d5c474a12a2a89286a14bd27f : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f87510a32aad42f785b2e6045307a192 : SUCCESS in 6m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a8260aec0105420e8195ae8b09f1cdb5 : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/578a8459f1454ad4972711b614115e40 : SUCCESS in 5m 55s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/7c7725f3001f4cb59563270b4555a5a0 : SUCCESS in 2h 13m 55s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/680fcf6140ab4d369b8db83a9ecfd1af : SUCCESS in 1h 58m 36s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/4559ad97453f4863991a245d77843688 : SUCCESS in 1h 51m 09s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/444d9c27b55f4d5b84ebbbda2e08bd56 : SUCCESS in 3m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e135f6ec6ef841859026ee2d4873d0c4 : SUCCESS in 6m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1a2581989aba4ee68208c78d00ff9574 : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":2},{"id":"c9f55a30a5f8ec1fa93ccd55dd68b0329857fa8d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-02 17:34:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/5538865f061144f5bda0cd751a120afb : SUCCESS in 3h 22m 14s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/ceae326da87940a985a29d3443fc6849 : SUCCESS in 3h 21m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"49d202d1b54b42c318e4b5f14827429e897eb7ad","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-06-05 15:07:30.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"c9a1dd655caf936240e72695aa56ce3d69496c26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 15:13:57.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"4603c4c15bfb9635260791657354a31d22921b9b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-05 15:13:59.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"82bf4ef4c147f48379cf66a143bcda4ccfff0d7b","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-06-05 15:15:02.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"7ec7fb634e86862582203002ab34973cdb9e7bbb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 18:24:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/48744d3b78bb4439b28eafb064fa323d : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ff46bb758ed489ea1205b1d29d978d5 : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ec2751253b694d6091406d3d0d4113c6 : SUCCESS in 7m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e5fd562f48894b85bb5e533efc27cf10 : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/185deadf56194035b730e5e39b061cf9 : SUCCESS in 5m 34s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/a5d551413b91411eb232dd8748c29045 : SUCCESS in 2h 21m 51s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/27780a5fcd8b446684d8d5cd343a20d6 : SUCCESS in 1h 54m 49s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/cfd9d70d54d0449180c4b39e405ea76a : SUCCESS in 1h 34m 11s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/79d3247dcf1648adae3ca4a20d1b1264 : SUCCESS in 5m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5fce7d514df94e678f9b26d0ec6be749 : SUCCESS in 7m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/382398cdd4344403a36cb52a0cf6a3ac : SUCCESS in 7m 32s","accounts_in_message":[],"_revision_number":3},{"id":"0f079a95c3bfa7b9e2dff32072b5802ecbe952e5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-05 21:34:29.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/7b23b37d93f74b74adec29bf9bb376cc : TIMED_OUT in 4h 10m 47s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/aca5c317a5a840cb90c65bf7c913b7f2 : TIMED_OUT in 5h 14m 52s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e3ca391edb3a236775d65e4c60211269f9a64914","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-06-09 05:51:43.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"3b59ec4faf924aec52449c7cc9e01726b74ead4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 08:28:19.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e29557fb544d47a9b27f83ed7dedb354 : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3472bddaaa224ded8face39274b5577e : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c00911fbb9cf418d8e29e80a40bdc78c : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/61c0fd3891e546769535a30ad03143ff : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39f3a26b34a349ebbd00281082e4f7b6 : SUCCESS in 4m 50s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/8af70a6675f94ebfbdc22d78f0b6f903 : SUCCESS in 2h 24m 51s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/b7df708f53734d6d8859b2b867c59a40 : SUCCESS in 1h 46m 33s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/f7b241fddd88495d8c3895b6344744cd : SUCCESS in 1h 41m 40s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/32fa14f5acb64591be3d620d024ce859 : SUCCESS in 3m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/25d1c787512d49eb8143edb6208da77c : SUCCESS in 5m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/681d6ea596d94b71ab16f487bae87ab4 : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":4},{"id":"2f0abc6210d0b31956ea8048579342ca6559dada","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-09 08:53:32.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/1c9f15b1a49144ee9047f81edab1eb54 : SUCCESS in 2h 59m 58s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/288d512aec314fc79ee0de6e957dd584 : FAILURE in 1h 45m 42s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"714d4c8b2609dab2a8d772fdf054e45ae6223bab","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-06-09 09:19:34.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"c31446e4343e0dc87d08feba7c6099eb26ba203b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-09 10:59:02.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/7277c77625a64a62ace942ac6beb137b : FAILURE in 22m 08s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/d0147b50500f49458e511450b378f2ab : FAILURE in 30m 52s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"55cda9ab936ee1e7b8858b0e8d0d4bebc70097df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 11:42:25.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2dece83e412b4d0783eca4b42cc36478 : SUCCESS in 7m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5615602567834d31afc5f73b3c7bfe3e : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0bb50d931ecf4f4ea3f1cee833488c80 : SUCCESS in 6m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d19802a8d2b34504ad091ce3fff7e999 : SUCCESS in 6m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dcbc0f9f6b9415bb5ec354c7099b7fe : SUCCESS in 7m 00s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/5083d3fd27464063a960b4b3b0226fa6 : FAILURE in 23m 33s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/81d3593899c14b1bb9525c062cab18af : FAILURE in 21m 54s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/530d96de576245f18f0cc5356a112411 : FAILURE in 23m 49s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/28132962289b4e73a30979479ffb2a25 : SUCCESS in 4m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ad9143280c4f41059cb5cebd6e5d5664 : SUCCESS in 8m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4400c9ce882b450fb073d812a29a243e : SUCCESS in 7m 09s","accounts_in_message":[],"_revision_number":4},{"id":"9d46d2b36e4f2bf2a0c5c0ba12418faed96503c2","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-06-09 12:00:50.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"bd612aa3dc66fba858fb0c4a4fec0c58e34d2363","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 16:32:51.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/87069ae074db40c495ad9af2f6e0f0e3 : SUCCESS in 6m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6632a7aebc1d40118b9962974f46db00 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6019c78b925b42708ac20c0be06a643a : SUCCESS in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7f76355648bd440e97cbeed32de6aa8c : SUCCESS in 6m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61ce135dcf3e4bd2b5accaf7a94dc906 : SUCCESS in 7m 18s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/8a20e80514a44f6fb70df9aaad25bba2 : SUCCESS in 2h 41m 49s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/dc491c19996a47d1a2943dc82189463a : SUCCESS in 1h 49m 52s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/42239048141d44f3b6a23a01ca58df3e : SUCCESS in 1h 37m 45s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fa6d7cba7cfd40338ab4b22ee2f8f01e : SUCCESS in 10m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a908fdfa1ff247769290f06434c2875e : SUCCESS in 11m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e3e5ae5aa4e147aaa211bf138d0a8bb4 : SUCCESS in 12m 14s","accounts_in_message":[],"_revision_number":4},{"id":"ece2c20f480febb7711758c3f757bb94700f4d52","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-09 17:04:41.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/c86bc45e4e784f9da4c975c02514acc0 : SUCCESS in 2h 46m 09s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/4ee70ffd10754c15abda003b276f4a6e : SUCCESS in 3h 22m 23s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"161964cddfeb29023e4755e106427f6d06f7c990","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-06-10 11:29:55.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"5f562df9ff8c7cd025be6e23a1d7744ee187eabb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 14:46:01.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cd6632dc1f994794bf2f14bb4823029c : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db42b9f1126047c9971192cb6ece0e6b : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d64097134d9849cdae8afb76ffa7c391 : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c6e345156d7c4f5b99c48892f89c7e9e : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43e934bcb77240e4a7b6ca547a60bcd7 : SUCCESS in 5m 03s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/2511351327f040d2819d5525237804c2 : SUCCESS in 3h 08m 05s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/b4d08ca469374a53bfb9ba5c5459018a : SUCCESS in 2h 15m 30s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/a17cf34db63940f183c2f9726d90d75e : SUCCESS in 1h 57m 51s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/38314c0f597b4dbaaa884e8891cfbe24 : SUCCESS in 4m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/618e77a53cf3470e85e3f6f8b6da68cb : SUCCESS in 5m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5880a8d6fee44279b6c53e687fe49433 : SUCCESS in 6m 40s","accounts_in_message":[],"_revision_number":5},{"id":"b6ad661752b5dd1f4c1e2a40e16652f3474e813a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-10 15:37:05.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/b6962fe1b4ea4e66a78c698e88763f89 : SUCCESS in 2h 50m 12s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/a9a6e943eeb148c182428cc6022fbd6f : SUCCESS in 3h 19m 14s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2d660d313341caafffcfeb34df04ba05a94ce0a5","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-06-11 10:46:24.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"d15f57c3ed7f2bcdd4f903f83c3a35631bc38775","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-11 11:05:51.000000000","message":"Patch Set 5: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":5},{"id":"1ad96f6f8da401a366d37ce7d39898771fa7a5af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 11:05:52.000000000","message":"Patch Set 5: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":5},{"id":"cb9fda9eeaa68072d433d32fa7f6e6780620da37","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-06-11 11:06:16.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"58bed1762a0433e215c7d863e7b67e8d272e12af","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-06-11 11:08:31.000000000","message":"Patch Set 5: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 90075de7c8754336c5b28ac6553bf89603ead0eb","accounts_in_message":[],"_revision_number":5},{"id":"90aaa5328676533e7aea5c0a020c0273a4cb7590","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 11:30:57.000000000","message":"Patch Set 5:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":5},{"id":"f512b91f023cbc030e0196383a7606fd0c45b167","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-11 11:30:59.000000000","message":"Patch Set 5:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":5},{"id":"5bbab86f3b83b1259364433f48868289062486b4","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-06-11 11:32:24.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"fef98bbdee42b3f3227dd9564775eaf0742f2ee9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-11 12:13:22.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/6c12d1c3e8ea48d1b60ffde1b59d5e09 : FAILURE in 24m 17s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/e04b1a1ecec94c68b3388ada4e2f9758 : FAILURE in 40m 09s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"44b8da735d5544292d30b2868a7f01b648a3ae16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 12:19:04.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c0405be031f64a618f1350aaaa58baf0 : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d6ae685f99845d4ba6351d5cfde5a34 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/29cb6ea073f642adbf0e0cf7640284c1 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af8cdd92697d48ea8e0909ac10e459d0 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd85fdeb565f4c2980fea491868b2b0d : SUCCESS in 6m 20s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/31f40efee0854bad873a81d0346151c4 : FAILURE in 21m 35s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/bbdc1d6e5873419e93887bc0872400a3 : FAILURE in 18m 41s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/9dc90a956b03465a80fc8886f347d070 : FAILURE in 36m 55s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2ba56df52d2048688fa6d1d82700c052 : SUCCESS in 3m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/74f1a19ed648499fbac068f3eab91900 : SUCCESS in 4m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fd60f671501e48ad8822d59825f3c1f5 : SUCCESS in 6m 29s","accounts_in_message":[],"_revision_number":5},{"id":"33e5ba6889e0f27683b1961ff09f3faf85678093","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-06-11 12:19:24.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"74563815c62216b4a7c3b231de14af83c7277c2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 13:15:07.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f57d1821c365416593114dc1c117a005 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4246cb9712574f2c99a05eb9fe47c5d4 : SUCCESS in 2m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0dfe562d657b4aa6b794332a3c55deb6 : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa9ebecc193a4992807710aa53c854b2 : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61e06c0dab5648f8a27225fe4f29f105 : SUCCESS in 4m 12s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/7fa9987d8f544387a92b9e5e09fd9e17 : FAILURE in 44m 25s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/9051142840a34b9c9112148a50405cde : FAILURE in 23m 16s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/bcd981e71f7345b38d533cbee2f36371 : FAILURE in 23m 41s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ce3418569ff84a7cabd4e58343b315b7 : SUCCESS in 3m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/327a922080a4427985eb8b459ce361ce : SUCCESS in 6m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4a7f1ec07b0f4cfa8727959fe37bec3d : SUCCESS in 6m 14s","accounts_in_message":[],"_revision_number":5},{"id":"7b33fd1fbc2b69397f57194f009496544beb9b7d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-11 13:32:39.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/89727953cbdc4dc1b664bf51d61eb443 : FAILURE in 22m 05s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/4e2405aaf8704cf487c331f99d9780a3 : FAILURE in 32m 52s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"35ae303049d0b5fd29bdc4facb9c5038c0b25aa1","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-06-15 05:16:25.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"d3009cb3fb00cfcbfe4c3fac4e1f01974ac99d4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 07:05:47.000000000","message":"Patch Set 6: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":6},{"id":"0cd73f587f97f0398393080e087bb331902e108e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-15 07:05:49.000000000","message":"Patch Set 6: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":6},{"id":"5d7c43374cc2f9fffe83da6fccc463a5a3818d4d","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-06-15 07:52:07.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"a368a7d639f668fae98b8bb43bec8bc82c2ba6ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 09:02:41.000000000","message":"Patch Set 6:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":6},{"id":"b62ae7d6c9d489bdd0963c6fa88190ef7f5787f8","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-15 09:02:45.000000000","message":"Patch Set 6:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":6},{"id":"d4c4cbec1f109258b3de9ef65eaee66a5c5cc383","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-06-15 09:03:16.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"914ca7b1fb6d56306fa53137e394b807fa4284ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 11:59:07.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f518cc0480d04aee8413aded6f1a9231 : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec4068a3ffa9421e818db8012ed69be2 : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/89e5d67b4ed54d6380d9b15b27ecbc29 : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0b3dc876394d45c5913843dc4cda6fc3 : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/665af12ae18847b5a94309e2113d2c8c : SUCCESS in 4m 12s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/d61a4c726ae64ddaab3877154ddb1650 : SUCCESS in 2h 46m 24s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/5685589227a4401cb3c350fa1b4f5c90 : FAILURE in 26m 03s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/7dd93c6fde3e416684fde985662192c5 : SUCCESS in 1h 30m 51s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/34b4f21275df425e82a3f09781c813f3 : SUCCESS in 3m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cee4aa6e97824316b51583e72c955e04 : SUCCESS in 4m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ec6054474d0a43908bab011c2d7e3cda : SUCCESS in 5m 43s","accounts_in_message":[],"_revision_number":6},{"id":"ff722819f7f36be63dba49e90082db1ad6971d5f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-15 12:22:08.000000000","message":"Patch Set 6:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":6},{"id":"e02e0e17a7b19d6b469c8edd82e361320279190d","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-06-15 12:22:52.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"48d2044c9fffa2b63041c0f001fc790ecbee01a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 15:18:53.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc14a2f61d0441c6a9f3c3a9aec0878c : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bc4694b57f3483f817480b641fa0025 : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd59f05a1da44db2bef1c49bbfb99b3b : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a7b0dfb9fee74a7f86fcad6074476f75 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc1d36789d094bb0afae8d915f2960fa : SUCCESS in 6m 31s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/3d9ed88533894a4c840ea5bdebc2e517 : SUCCESS in 2h 38m 36s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/89be78beedbc40078f2dc3d46e6d0b17 : SUCCESS in 1h 56m 33s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/7025a27994cb462ca32ff75a01835971 : SUCCESS in 1h 47m 27s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a001dc9656ff4dea912968573e8a6686 : SUCCESS in 4m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1229668493c843ea934674b41f4c926f : SUCCESS in 6m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bb0e88d0f0914424af86229ac6fedd24 : SUCCESS in 8m 16s","accounts_in_message":[],"_revision_number":6},{"id":"9d3c75e887e1fa689f4a0d5ce325a8b237163c53","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-15 16:37:08.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/65ffb0488a7841b091de57d23cc436af : SUCCESS in 2h 49m 10s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/f2eac22f36ea48c8bc3d046cd885288c : SUCCESS in 3h 32m 48s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"2e4861ffd8ca6ca37444b608db1c94559fc0cddd","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-06-18 08:24:17.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"ab8596bf3703c178ce13bcd09c8fc6ee8ceca2a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 09:24:45.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0fd7ce1b57ba449c84eece942251bcd3 : SUCCESS in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63a744eda4a349aaa7c3af127bd9e825 : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/62b8aabdeb1c41f987b8abe4d30d2cf8 : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/94f0cae617164490ae0d65732e85db75 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1de5833c08ce44a9a72586a02927176c : SUCCESS in 5m 21s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/b4d28eb0794c4068ae2aeabc61761b09 : FAILURE in 30m 32s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/150b493efce44b10bde87079e7f4d5eb : FAILURE in 30m 06s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/0ae23f6f960b456abb4c1efabb2a57a9 : FAILURE in 23m 30s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d5a6aa66fd8e4892b6744cfa834b36ec : SUCCESS in 6m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/071b2abb1a3a4680b25f3fe9c5b50a68 : SUCCESS in 6m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/467969ce2fd344a4bbeb6b779c96ca3a : SUCCESS in 7m 29s","accounts_in_message":[],"_revision_number":7},{"id":"02832d6797737c2f9bfe63a40ecde1a82ba3a7f9","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-06-18 10:06:41.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"fefd34184e349fbaa48113c93c599174da1083a9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-18 10:06:41.000000000","message":"Patch Set 7: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":7},{"id":"220ebef7ce50b07008dc50b890a79e677ed262fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 11:24:40.000000000","message":"Patch Set 8: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":8},{"id":"1418cf3906a6724d56ad77c8e597df08a02f22b0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-18 11:24:44.000000000","message":"Patch Set 8: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":8},{"id":"983d865ea1dd16d83ab96a40f0e549dd03d41be6","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-06-18 11:24:56.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"6ba4130f3e90c2d9f5c5e6b45bf4084c76a9cffa","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-18 11:31:12.000000000","message":"Patch Set 8:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":8},{"id":"c28a4425154661c799544976ccd69cdabd2e6496","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-06-18 11:31:30.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"fb5edc3d6353a35d39012766be8a1174172a57f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 11:33:14.000000000","message":"Patch Set 8:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":8},{"id":"ea45a9b71b01fea79533db4536ab28b976678df6","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-06-18 11:35:24.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"5d4258b76bc488fcd28ca294adb0214b83323dde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 13:23:39.000000000","message":"Patch Set 8:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":8},{"id":"ed1e887c8da7468d2f9421de4fe3c59d3a65f7f6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-18 13:23:44.000000000","message":"Patch Set 8:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":8},{"id":"d97f473ba2412977316cd5ac102921e009315e76","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-06-18 13:24:07.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"9c8e2d3fc58dac4f2dfee55e88f9dcd3d1963b1b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-18 18:11:35.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/1365a44410a44885824b9e70fc2d2dcb : TIMED_OUT in 4h 11m 29s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/8c1d7267995041319a7b4979bd2a4b24 : RETRY_LIMIT in 41m 28s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"64b4a2c0bc223f5f9ac110b46399676a7c66687e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 18:39:19.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a9d5890a38744090b82095d56515b49d : SUCCESS in 8m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60f4e9299caa4090b806d5b716b440d7 : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/57184928157a461c87cc3b8ffb7b1e57 : SUCCESS in 6m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8bb57e8385c049ed89d251f641354862 : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/538807310a3e4e01b1e5e303c9338617 : SUCCESS in 4m 29s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/98d50a346fc24ed28e96dd13dde3479c : SUCCESS in 2h 51m 06s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/410eb4c78d304767964d7191558fc7cd : SUCCESS in 2h 02m 22s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/637e63a3fa5c41b5921cbbe3515d278e : SUCCESS in 1h 55m 04s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b9b9bb3d25c14b43affc63e951a50c80 : SUCCESS in 3m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1705d4bafcaa46f7a1f465e083ea46a4 : SUCCESS in 6m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5d1a26604abc467d8967fe93503d4e69 : SUCCESS in 13m 23s","accounts_in_message":[],"_revision_number":8},{"id":"db7af0a2093de7a7850bdd77f5b15b3ef5d38f56","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-06-19 09:18:34.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"2f364cbe337c2dd7afa1672e197af59dd25fa5bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 12:15:39.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8cc05e17a99a40ffa71ca0e4aaa57f71 : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e61e1ed4bcbd4d2f94365fddcb148023 : FAILURE in 7m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e441f9e500e47b1ac03e4330749f16b : SUCCESS in 7m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/17f6df49b19447b8a0818970867679bb : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8683f497a4f845e289c855e9e8ff475c : SUCCESS in 7m 12s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/0d5ab898651242b1badd1680f71b7d92 : SUCCESS in 2h 37m 23s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/e514d883a07b4ef08486a34c602bacce : SUCCESS in 1h 58m 28s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/a9bac53efa8b4477b615e1ac3bc9f469 : SUCCESS in 1h 40m 04s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a5012aa37ef5400cabfdda304749dadf : SUCCESS in 5m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4ea5b9049cd6485b9a72f88055904f28 : SUCCESS in 6m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4d5527da6875426c9b4c05384759665c : SUCCESS in 5m 33s","accounts_in_message":[],"_revision_number":9},{"id":"9a0ae2c546467d834977e83dcea735c68e916a0f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-19 12:16:01.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/20391cd3eb6e4e128c51c3f3989b665c : FAILURE in 1h 26m 17s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/282b4144cd6d4e59a18513883a23b16f : FAILURE in 2h 54m 58s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"1dd9622f1a85c748b0ce9c582255a2cf303507c7","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-06-19 16:02:04.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"19d935b30ae0e43bae069c4114df921fc4ae427b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-19 19:47:27.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/e229aa37d81c49e6bc1e737d17ebe5dd : FAILURE in 1h 33m 20s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"b33307d0b521069bc5a1f0d3a6e6834c1d93a97d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 21:16:51.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/600f13046a5b46ffb72fec9be8a9a905 : SUCCESS in 8m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77dd214a7f21489b9a2d622f13d7c119 : FAILURE in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2dbe9c80fda843bdba0af50f3d26bca6 : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/29f4301f77e64284892402e3e5c7732a : SUCCESS in 7m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06e1bf3d1f724ff5bfb41ac32402d6d8 : SUCCESS in 4m 56s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/8c86a2dffb254cf1bda308daf46d9588 : SUCCESS in 2h 31m 10s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/ba8ea55a1d8d4ceba1347d31eba377e2 : SUCCESS in 1h 58m 53s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/5ac1db6e707b4c338544de1fc8519f7f : SUCCESS in 1h 48m 38s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/72bec6355ebd4d6fa9a3e95af8988174 : SUCCESS in 5m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/55a76189cb874af2bcf0c4fbefbeaccd : SUCCESS in 7m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c8f55d64182240de9818609b732df942 : SUCCESS in 7m 24s","accounts_in_message":[],"_revision_number":9},{"id":"e753bc7365077a2f82f07ef41def410ae8f913a2","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-06-22 08:23:13.000000000","message":"Patch Set 10: Commit message was updated.","accounts_in_message":[],"_revision_number":10},{"id":"5dbf87d8e20b9b7ad3de21ff1d5cc3cd01a0a6cd","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-22 10:46:44.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/80f73dddfefe440dbeef36a5755f20ce : RETRY_LIMIT in 45m 41s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"aa24b86596e74f7cb92313125ff238de5e74959f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 11:58:55.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a68bf25823944920ae2a6249d4594fea : SUCCESS in 7m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b986559ac7542448821f565f79dcf88 : FAILURE in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ac75459a54b4f2bb94ab05974fbbbef : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/199cfccdac5a41c69ecac5386433138c : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/100706f843ed423faa7b144de6cc15cf : SUCCESS in 4m 26s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/07e7435cbbaa47acbbf079a94f358fa6 : RETRY_LIMIT in 44m 09s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/c6d3918e87c741e4a17b0dab46b99f39 : SUCCESS in 2h 03m 28s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/709ace357a2f4460ba1571422f43871d : SUCCESS in 1h 41m 48s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8d7a5a861ec446f792db1eb493cbc0aa : SUCCESS in 5m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d80253028c7940c0b2d57c6c88ac864d : SUCCESS in 6m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/70385e235bce4e708b012705d2b711c1 : SUCCESS in 6m 16s","accounts_in_message":[],"_revision_number":10},{"id":"34eec252df3907296f54081450e4c055a39a3c73","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-06-29 08:29:03.000000000","message":"Patch Set 11: Commit message was updated.","accounts_in_message":[],"_revision_number":11},{"id":"75a24cc87e1ad610545635aaec85f0b7a2ddc936","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-29 09:16:33.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/5fc014dac39748619df7aeb216375ca7 : ERROR Failed to update project None in 4s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"e4aa61a278cbeb446cca6b52cacd148e3412ff67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 14:17:54.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8cd7b5872f09438d9a0c6a3aca50f4e6 : SUCCESS in 10m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d4dbee1266d4659885fdba77fe836ac : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/88c263f111834ee5b3ef3bfe0b81ff60 : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/64cf3b5dc31b4b1aa7456ea495432539 : SUCCESS in 7m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/623b033900bd4280b1cff6a5a7928493 : SUCCESS in 7m 35s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/695f8d6b9cb542859ed94eb7e26cb20d : FAILURE in 3h 02m 39s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/ab507731870e4b03b5af0d6beba6eac6 : SUCCESS in 1h 50m 28s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/6e992178d52643c1b593ed96d0f9506e : SUCCESS in 1h 52m 00s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/bae2273223284457b20f71efceb0520a : SUCCESS in 6m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/532fb3f2d8594390ae702660918d82cf : SUCCESS in 7m 31s\n- tripleo-cross-tripleoclient-py38 https://zuul.opendev.org/t/openstack/build/7ff7b80159af4fd9a78d86152fe2fa0f : SUCCESS in 6m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0dcdca6ef72d4dfeb9d71b00019fa9ae : SUCCESS in 6m 58s","accounts_in_message":[],"_revision_number":11},{"id":"faf6a4c538d014b82c0355b3bcc3c384d70324d0","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-06-30 07:03:04.000000000","message":"Patch Set 12: Patch Set 11 was rebased","accounts_in_message":[],"_revision_number":12},{"id":"e01183295619f46b8e6298b019b99a0e9b481c4c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-30 08:58:20.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/a907320764ea413d9fd8a083b2888e67 : RETRY_LIMIT in 15m 36s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"a6c543e8822667a72f5b7b56db0fcbabd1501971","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 10:12:25.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cd1d659e87ac41fbbb6ccfbbbfc082c2 : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17b18c3673b649cbbd785e4e44f8cbb3 : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2a79a8fa14224579aab1a5edb0966301 : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7e0214c8f8b843b88d51e52745df4562 : SUCCESS in 6m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/616fada6f8b54a8d93dbb9730acfcd2f : SUCCESS in 5m 08s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/f40ab2af14844ee88c80467a204e3549 : SUCCESS in 2h 58m 03s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/30ac60b33c2e4b43b5c6e48521a2ced4 : SUCCESS in 2h 09m 07s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/634525a51eb74ea2b28094312923eeee : SUCCESS in 1h 50m 27s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/eff73f411bef4381873c7ed841d3d598 : SUCCESS in 4m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/96833730e6494dd283a4a6a96e09b2bd : SUCCESS in 5m 55s\n- tripleo-cross-tripleoclient-py38 https://zuul.opendev.org/t/openstack/build/2e0df0e2fbdf46bfad8834c68bce2c93 : SUCCESS in 5m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d07cab88b95d4880b1f48c67cb255928 : SUCCESS in 6m 53s","accounts_in_message":[],"_revision_number":12},{"id":"a07986e119c3222be0271ca5a2564e78fd1216cb","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-06-30 11:43:27.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"94e8fdbccf2c5bafd31a78d4ae594d9b09d298fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 14:33:33.000000000","message":"Patch Set 12:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a55562bffe904d288ba8092870143e8d : SUCCESS in 7m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c905a7179bdb415fb94836cc3d9be76a : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d53cf55acd8f47d8a945cf457e2848e3 : SUCCESS in 6m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/22a174e9a6f64edaa5981e9157bfc01f : SUCCESS in 6m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db20f11d946d4b38a4fc95f1c038b86c : SUCCESS in 6m 04s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/0c00f367b5c840fc8d4623ec27a9e3aa : SUCCESS in 2h 35m 44s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/aac036eb5c0b432dbe7c1bce1f896875 : SUCCESS in 1h 21m 02s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/cb20b95c97544deaa0fee38464e07cfb : SUCCESS in 1h 43m 27s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c9df1c16503543018615f1e927720c7e : SUCCESS in 5m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d8f8677a6ee44af59bd4ca19eaf29e39 : SUCCESS in 6m 22s\n- tripleo-cross-tripleoclient-py38 https://zuul.opendev.org/t/openstack/build/5d72eed362814bfdb26d846905915c76 : SUCCESS in 6m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a77180eeb16b4445b8d4aa8bc06c2184 : SUCCESS in 6m 14s","accounts_in_message":[],"_revision_number":12},{"id":"cfda7188a9d28086ba21698afabda361f2eff457","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-30 15:32:53.000000000","message":"Patch Set 12:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/f27f95e2c45e4f17bd3e0a701470a247 : SUCCESS in 3h 46m 11s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"ae400c300698f4c1719183aca0bdbe314c7626bf","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-07-04 15:07:59.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"c0bc01d6c6375070d9343f33082317d18e88c009","revisions":{"00a9197071b923edde10788a00745580a785af75":{"kind":"REWORK","_number":1,"created":"2020-05-29 12:04:06.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/39/731739/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/39/731739/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/39/731739/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/39/731739/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/39/731739/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/39/731739/1"}}},"commit":{"parents":[{"commit":"fe8ab8cb92a0721d215f67710732c6619e74f19f","subject":"Merge \"Switch to tripleo stdout callback\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/fe8ab8cb92a0721d215f67710732c6619e74f19f"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2020-05-29 12:03:04.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2020-05-29 12:04:02.000000000","tz":330},"subject":"[DNM] build test packages downstream testing","message":"[DNM] build test packages downstream testing\n\nDepends-On: https://review.opendev.org/727200\n\nChange-Id: I131aca7a03802766a791e6912fad6ebf143d31db\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/00a9197071b923edde10788a00745580a785af75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/00a9197071b923edde10788a00745580a785af75"}]},"branch":"refs/heads/master"},"fb6b55b12036b87da9dd04254b57d404baf1045a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-06-02 13:14:36.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/39/731739/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/39/731739/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/39/731739/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/39/731739/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/39/731739/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/39/731739/2"}}},"commit":{"parents":[{"commit":"ed7049ee841f0449ec7b5745a5e99329a454dc2f","subject":"Merge \"Switch to tripleo_dense\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/ed7049ee841f0449ec7b5745a5e99329a454dc2f"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2020-05-29 12:03:04.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2020-06-02 13:14:36.000000000","tz":0},"subject":"[DNM] build test packages downstream testing","message":"[DNM] build test packages downstream testing\n\nDepends-On: https://review.opendev.org/727200\n\nChange-Id: I131aca7a03802766a791e6912fad6ebf143d31db\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/fb6b55b12036b87da9dd04254b57d404baf1045a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/fb6b55b12036b87da9dd04254b57d404baf1045a"}]},"branch":"refs/heads/master"},"0f0ee2dee7e4ae49cf03aadcab3eef11c7b3b93e":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-06-05 15:07:30.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/39/731739/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/39/731739/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/39/731739/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/39/731739/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/39/731739/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/39/731739/3"}}},"commit":{"parents":[{"commit":"46f3055e5ddd7dfb09f7ae10206eff8bb903a15a","subject":"Merge \"Revert \"Add kolla template override to remove python3-networking-mlnx\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/46f3055e5ddd7dfb09f7ae10206eff8bb903a15a"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2020-05-29 12:03:04.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2020-06-05 15:07:30.000000000","tz":0},"subject":"[DNM] build test packages downstream testing","message":"[DNM] build test packages downstream testing\n\nDepends-On: https://review.opendev.org/727200\n\nChange-Id: I131aca7a03802766a791e6912fad6ebf143d31db\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/0f0ee2dee7e4ae49cf03aadcab3eef11c7b3b93e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/0f0ee2dee7e4ae49cf03aadcab3eef11c7b3b93e"}]},"branch":"refs/heads/master"},"bef650436637032348acb7dc27fc0cc91deb2994":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-06-09 05:51:43.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/39/731739/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/39/731739/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/39/731739/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/39/731739/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/39/731739/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/39/731739/4"}}},"commit":{"parents":[{"commit":"871f69204d15eef050147f7e34ea76b3fca78af1","subject":"Merge \"tcib/ceilometer-base: remove panko dependency\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/871f69204d15eef050147f7e34ea76b3fca78af1"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2020-05-29 12:03:04.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2020-06-09 05:51:43.000000000","tz":0},"subject":"[DNM] build test packages downstream testing","message":"[DNM] build test packages downstream testing\n\nDepends-On: https://review.opendev.org/727200\n\nChange-Id: I131aca7a03802766a791e6912fad6ebf143d31db\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/bef650436637032348acb7dc27fc0cc91deb2994"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/bef650436637032348acb7dc27fc0cc91deb2994"}]},"branch":"refs/heads/master"},"a7f0a0d3f9508d7a2e3d3adc7184739719aa6bce":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-06-10 11:29:55.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/39/731739/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/39/731739/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/39/731739/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/39/731739/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/39/731739/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/39/731739/5"}}},"commit":{"parents":[{"commit":"ef3e212500d3f2de2d711e35449c2b935ab3c5f5","subject":"Merge \"Remove tripleoclient container\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/ef3e212500d3f2de2d711e35449c2b935ab3c5f5"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2020-05-29 12:03:04.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2020-06-10 11:29:55.000000000","tz":0},"subject":"[DNM] build test packages downstream testing","message":"[DNM] build test packages downstream testing\n\nDepends-On: https://review.opendev.org/727200\n\nChange-Id: I131aca7a03802766a791e6912fad6ebf143d31db\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/a7f0a0d3f9508d7a2e3d3adc7184739719aa6bce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/a7f0a0d3f9508d7a2e3d3adc7184739719aa6bce"}]},"branch":"refs/heads/master"},"4a99a1f9dc6c0824e12831c27cfc7247e85869b7":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-06-15 05:16:25.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/39/731739/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/39/731739/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/39/731739/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/39/731739/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/39/731739/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/39/731739/6"}}},"commit":{"parents":[{"commit":"ff9fe2edfb940bf983c880bbfb4d6e95ccdce7c5","subject":"Add interface-names to Rhel8 images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/ff9fe2edfb940bf983c880bbfb4d6e95ccdce7c5"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2020-05-29 12:03:04.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2020-06-15 05:16:25.000000000","tz":0},"subject":"[DNM] build test packages downstream testing","message":"[DNM] build test packages downstream testing\n\nDepends-On: https://review.opendev.org/727200\n\nChange-Id: I131aca7a03802766a791e6912fad6ebf143d31db\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/4a99a1f9dc6c0824e12831c27cfc7247e85869b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/4a99a1f9dc6c0824e12831c27cfc7247e85869b7"}]},"branch":"refs/heads/master"},"d77b6e5c8817569b9c40156092309a463a905d14":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":7,"created":"2020-06-18 08:24:17.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/39/731739/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/39/731739/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/39/731739/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/39/731739/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/39/731739/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/39/731739/7"}}},"commit":{"parents":[{"commit":"fe8ab8cb92a0721d215f67710732c6619e74f19f","subject":"Merge \"Switch to tripleo stdout callback\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/fe8ab8cb92a0721d215f67710732c6619e74f19f"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2020-05-29 12:03:04.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2020-06-18 08:22:27.000000000","tz":330},"subject":"[DNM] build test packages downstream testing","message":"[DNM] build test packages downstream testing\n\nDepends-On: https://review.opendev.org/727200\n\nTHT changes:\nDepends-On: https://review.opendev.org/724734\nDepends-On: https://review.opendev.org/#/c/731108\n\nChange-Id: I131aca7a03802766a791e6912fad6ebf143d31db\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/d77b6e5c8817569b9c40156092309a463a905d14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/d77b6e5c8817569b9c40156092309a463a905d14"}]},"branch":"refs/heads/master"},"a6051f1e0ff0fea857b74a0c73a5096711267177":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2020-06-18 10:06:41.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/39/731739/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/39/731739/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/39/731739/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/39/731739/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/39/731739/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/39/731739/8"}}},"commit":{"parents":[{"commit":"e695cd8fbb4c68081dc74aba812bbb8e461e6ee0","subject":"Merge \"Remove config-download related mistral actions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/e695cd8fbb4c68081dc74aba812bbb8e461e6ee0"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2020-05-29 12:03:04.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2020-06-18 10:06:41.000000000","tz":0},"subject":"[DNM] build test packages downstream testing","message":"[DNM] build test packages downstream testing\n\nDepends-On: https://review.opendev.org/727200\n\nTHT changes:\nDepends-On: https://review.opendev.org/724734\nDepends-On: https://review.opendev.org/#/c/731108\n\nChange-Id: I131aca7a03802766a791e6912fad6ebf143d31db\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/a6051f1e0ff0fea857b74a0c73a5096711267177"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/a6051f1e0ff0fea857b74a0c73a5096711267177"}]},"branch":"refs/heads/master"},"e6355b727ad4e854afc87bbf14cf120fa1415c2c":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2020-06-19 09:18:34.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/39/731739/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/39/731739/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/39/731739/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/39/731739/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/39/731739/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/39/731739/9"}}},"commit":{"parents":[{"commit":"c915208fee4d15d9450514a6a5de6bdc5cc6f12f","subject":"Revert \"Remove config-download related mistral actions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/c915208fee4d15d9450514a6a5de6bdc5cc6f12f"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2020-05-29 12:03:04.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2020-06-19 09:18:34.000000000","tz":0},"subject":"[DNM] build test packages downstream testing","message":"[DNM] build test packages downstream testing\n\nDepends-On: https://review.opendev.org/727200\n\nTHT changes:\nDepends-On: https://review.opendev.org/724734\nDepends-On: https://review.opendev.org/#/c/731108\n\nChange-Id: I131aca7a03802766a791e6912fad6ebf143d31db\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/e6355b727ad4e854afc87bbf14cf120fa1415c2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/e6355b727ad4e854afc87bbf14cf120fa1415c2c"}]},"branch":"refs/heads/master"},"bbd8a161fbad8c396a456cbd4132a4c625832630":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2020-06-22 08:23:13.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/39/731739/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/39/731739/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/39/731739/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/39/731739/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/39/731739/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/39/731739/10"}}},"commit":{"parents":[{"commit":"c915208fee4d15d9450514a6a5de6bdc5cc6f12f","subject":"Revert \"Remove config-download related mistral actions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/c915208fee4d15d9450514a6a5de6bdc5cc6f12f"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2020-05-29 12:03:04.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2020-06-22 08:23:08.000000000","tz":0},"subject":"[DNM] build test packages downstream testing","message":"[DNM] build test packages downstream testing\n\nDepends-On: https://review.opendev.org/737179\n\nTHT changes:\nDepends-On: https://review.opendev.org/724734\nDepends-On: https://review.opendev.org/#/c/731108\n\nChange-Id: I131aca7a03802766a791e6912fad6ebf143d31db\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/bbd8a161fbad8c396a456cbd4132a4c625832630"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/bbd8a161fbad8c396a456cbd4132a4c625832630"}]},"branch":"refs/heads/master"},"0c7bc7bf815b92a2c94c0713d31be54f3253dc0d":{"kind":"NO_CODE_CHANGE","_number":11,"created":"2020-06-29 08:29:03.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/39/731739/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/39/731739/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/39/731739/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/39/731739/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/39/731739/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/39/731739/11"}}},"commit":{"parents":[{"commit":"c915208fee4d15d9450514a6a5de6bdc5cc6f12f","subject":"Revert \"Remove config-download related mistral actions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/c915208fee4d15d9450514a6a5de6bdc5cc6f12f"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2020-05-29 12:03:04.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2020-06-29 08:28:54.000000000","tz":0},"subject":"[DNM] build test packages downstream testing","message":"[DNM] build test packages downstream testing\n\nDepends-On: https://review.opendev.org/738394\n\nTHT changes:\nDepends-On: https://review.opendev.org/724734\nDepends-On: https://review.opendev.org/#/c/731108\n\nChange-Id: I131aca7a03802766a791e6912fad6ebf143d31db\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/0c7bc7bf815b92a2c94c0713d31be54f3253dc0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/0c7bc7bf815b92a2c94c0713d31be54f3253dc0d"}]},"branch":"refs/heads/master"},"c0bc01d6c6375070d9343f33082317d18e88c009":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2020-06-30 07:03:04.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/39/731739/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/39/731739/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/39/731739/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/39/731739/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/39/731739/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/39/731739/12"}}},"commit":{"parents":[{"commit":"d7dfc59b09509306810ea939499da19e3eb7cad5","subject":"Merge \"[Q-\u003eT] Include manila in the Queens to Train Upgrade\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/d7dfc59b09509306810ea939499da19e3eb7cad5"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2020-05-29 12:03:04.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2020-06-30 07:03:04.000000000","tz":0},"subject":"[DNM] build test packages downstream testing","message":"[DNM] build test packages downstream testing\n\nDepends-On: https://review.opendev.org/738394\n\nTHT changes:\nDepends-On: https://review.opendev.org/724734\nDepends-On: https://review.opendev.org/#/c/731108\n\nChange-Id: I131aca7a03802766a791e6912fad6ebf143d31db\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/c0bc01d6c6375070d9343f33082317d18e88c009"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/c0bc01d6c6375070d9343f33082317d18e88c009"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
