)]}'
{"id":"openstack%2Ftripleo-common~741228","triplet_id":"openstack%2Ftripleo-common~stable%2Ftrain~Id13028ab911ba91d67787ca8247effc02017fcbc","project":"openstack/tripleo-common","branch":"stable/train","topic":"tcib/train","hashtags":[],"change_id":"Id13028ab911ba91d67787ca8247effc02017fcbc","subject":"tcib: move cradox to gnocchi-base \u0026 enable ci on scenario001","status":"MERGED","created":"2020-07-15 14:42:33.000000000","updated":"2020-08-26 23:05:47.000000000","submitted":"2020-08-26 23:01:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"741228-1598482888784-436c11b4","meta_rev_id":"d887a34b8869f66440ff59526d01bc741b8c73d6","_number":741228,"virtual_id_number":741228,"owner":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"date":"2020-08-26 13:02:54.000000000","_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-26 23:01:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-21 00:19:45.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"recommended":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":1,"date":"2020-08-21 06:42:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":2,"date":"2020-08-24 11:27:28.000000000","_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":2,"date":"2020-08-20 21:01:03.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2020-08-24 11:22:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-24 11:30:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-08-20 21:01:10.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-24 17:13:34.000000000","updated_by":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"reviewer":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"state":"REVIEWER"},{"updated":"2020-08-11 08:18:39.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2020-08-20 12:49:07.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-08-21 00:19:45.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-21 06:42:58.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2020-08-24 11:22:29.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"},{"updated":"2020-08-24 11:27:28.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2020-08-24 11:30:52.000000000","updated_by":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"reviewer":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"state":"REVIEWER"},{"updated":"2020-08-26 13:02:54.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2020-08-26 23:01:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c53fbf89dad07681082b40624542f57ad8b7a29b","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-07-15 14:42:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"220bab65dda60252383c62ff6fb034610f011795","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-07-15 14:42:57.000000000","message":"Topic changed from cradix/train to tcib/train","accounts_in_message":[],"_revision_number":1},{"id":"678d7fbd75cd58eadd12ce5acbd186c0a24eb695","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-07-15 19:45:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-rhel-8-buildimage-overcloud-full-train https://review.rdoproject.org/zuul/build/c0f1d6e4f2a345fd9d36f1a449160814 : RETRY_LIMIT in 6m 34s\n- tripleo-rhel-8-buildimage-ironic-python-agent-train https://review.rdoproject.org/zuul/build/70c456cf175a419483859bca34c4f8a3 : RETRY_LIMIT in 6m 23s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/34f1fcd12dbb48b1bb64d3ce9fa79d39 : SUCCESS in 3h 53m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5126cc2d41bc30f329bb4b014f7799bcdc6e5a78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 21:18:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7b6f1819034e4dbe832b6feddcfbbf52 : SUCCESS in 9m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d19badebca0e41ffa2f3bba82bf90d08 : SUCCESS in 5m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ddbf51d9fbec400ab6933b6327c38d57 : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/164ec05efe53408ba047fd84d85ad08c : SUCCESS in 5m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cc11be67f2794c22b8eb1927db4b9adb : SUCCESS in 6m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f3a3a7df9bf4331bae2dcd9280cb91f : SUCCESS in 5m 19s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/4b22fc8004fd4e6cb443477d23a8f35d : SUCCESS in 1h 57m 28s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/d5536df0afc44b8ca1cef5066cb7b71a : SUCCESS in 2h 33m 01s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/667c9172e60d4b0093fa273ca9a33aa3 : SUCCESS in 2h 22m 30s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/20bd8d67d6f8406bac4158b7cc9425fa : SUCCESS in 3h 00m 41s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/68884e67c5894feeb1c668d447c7363f : FAILURE in 47m 42s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/5ace8848b67e4900a30a3c32ba267609 : TIMED_OUT in 3h 22m 44s (non-voting)\n- tripleo-ci-centos-7-containers-undercloud-minion https://zuul.opendev.org/t/openstack/build/d1f2d759b4fe480fb90cfb41cb0ea7e5 : SUCCESS in 1h 50m 48s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/7d2450220b744218b5717058d7d31b9a : FAILURE in 2h 22m 36s (non-voting)\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/4055fd7fc16441ffb6fefe83f8df69c7 : SUCCESS in 56m 16s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/5efd455c30f0423fbbd7a23f375c498f : SUCCESS in 1h 06m 14s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/60005857cb7a43828d7b59a3f58c9c93 : RETRY_LIMIT in 18m 26s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/e1be07d0607a4a68b8e07e0658a59b89 : SUCCESS in 2h 52m 30s\n- tripleo-ci-centos-7-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/47774cd0ea35455d92071462c5c85094 : SUCCESS in 1h 22m 39s (non-voting)\n- tripleo-ci-centos-7-scenario007-standalone https://zuul.opendev.org/t/openstack/build/1eaafda875744e2dbfd9c662af172773 : SUCCESS in 1h 22m 15s\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/27ca6b157fcf45a1bd3a126e96301061 : FAILURE in 24m 43s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/3657be6f9ba24a4980d5786a40fb42b0 : SUCCESS in 1h 59m 21s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e72e6cfa97184573a0dc75747034f2ce : SUCCESS in 4m 22s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/3b9789b320b8496b84dd968cbb787d3a : SUCCESS in 2h 30m 18s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/321a0d7c852245c3a4bfa30fd05542b0 : SUCCESS in 8m 21s","accounts_in_message":[],"_revision_number":1},{"id":"61cf557e4f3c841db242b538963edcf68df3c15c","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-07-15 22:02:06.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"ef5cd1f19ba7c2fc6a58c09fe6345e1c1c2d9ee1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 01:46:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f0c1989391ee4a5a915514e97c2cc81d : SUCCESS in 6m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31190c36953d4984b8c81e8bbab6dc9a : SUCCESS in 3m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d929316e4c93439eba2183ccce368286 : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0505690488f24811a758e2d2eab05352 : SUCCESS in 4m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c02b0f816c884a658a23d1489763c93e : SUCCESS in 6m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2330779e599241ab81cd0773d24164a6 : SUCCESS in 5m 11s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/ea65ec0fbfb34814b8c12ed7540e4257 : SUCCESS in 1h 57m 22s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/320e934fac4a439c83e03dfb26f62c4d : SUCCESS in 2h 36m 10s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/190024f2896f40dcbe2fbca852e4d17e : SUCCESS in 2h 20m 21s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/da25ddc4dd7b4948a1c2429e7d2d74c4 : SUCCESS in 3h 30m 52s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/d402d6a2a41545e19b077b6c5b93fcf9 : FAILURE in 25m 52s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/bc64d4949743407995d3227e75a3fce2 : SUCCESS in 2h 26m 59s (non-voting)\n- tripleo-ci-centos-7-containers-undercloud-minion https://zuul.opendev.org/t/openstack/build/0db7349160e84137bdcbae4db6f1b194 : SUCCESS in 1h 53m 20s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/87ed37c0f0df4f64ae50fbf5fe11355d : FAILURE in 2h 09m 50s (non-voting)\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/7daa1a68aa3041ceb7cffe8f90a937a5 : SUCCESS in 54m 17s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/98420cf3068042569e1a6416781d90a1 : SUCCESS in 1h 01m 43s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/b263ec040a68498da0a48aa9080e012c : RETRY_LIMIT in 14m 11s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/28a8ae6f6a5445d1b0b64a4ffccf5c01 : SUCCESS in 2h 23m 34s\n- tripleo-ci-centos-7-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/a1e24b31968a4c478281533a8d7485dd : SUCCESS in 1h 49m 59s (non-voting)\n- tripleo-ci-centos-7-scenario007-standalone https://zuul.opendev.org/t/openstack/build/313ed455e8424e3d92d0847ab543471f : SUCCESS in 1h 08m 02s\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/0974305b6544493891b95be13e5b2af9 : FAILURE in 28m 06s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/2c90c22bcc064963995f24d58a7f87ea : SUCCESS in 1h 48m 25s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/60fe4d7ba0ae442e86d4c29999580115 : SUCCESS in 4m 39s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/b21ccded94484de8b2a73ae88638dfc0 : SUCCESS in 3h 02m 58s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4003de365e1543838c5246ebdf67c865 : SUCCESS in 7m 29s","accounts_in_message":[],"_revision_number":2},{"id":"d67776a6d4c45d229f67f324eaf315e128885ea8","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-07-16 02:17:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-rhel-8-buildimage-overcloud-full-train https://review.rdoproject.org/zuul/build/280636cc04c44e2f9169aad9a1d7e8ff : RETRY_LIMIT in 6m 11s\n- tripleo-rhel-8-buildimage-ironic-python-agent-train https://review.rdoproject.org/zuul/build/17676982669a4c8ab8f290657d12901a : RETRY_LIMIT in 6m 13s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/f1b3269a729a42bf99f741ebda1fc0f4 : SUCCESS in 3h 27m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fb4313a07d8e9b4f2c5aad9a78757d8bf52948c0","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-07-24 01:53:53.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"0039bd8096c0368352a86abc9c5167fb0d1affa8","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-07-24 04:50:44.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-rhel-8-buildimage-overcloud-full-train https://review.rdoproject.org/zuul/build/b7811c8bdb2a44c9b0ec375eb22c7dbf : RETRY_LIMIT in 6m 15s\n- tripleo-rhel-8-buildimage-ironic-python-agent-train https://review.rdoproject.org/zuul/build/8480c5eded4b446580952b37b95e3c5f : RETRY_LIMIT in 5m 59s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/88e6ed26ecbc4d488a0042ac87506ac1 : FAILURE in 2h 54m 45s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"173eb21e00d72152a187e831f5a22e143b40f945","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 05:21:59.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1a4f6f2a94754e48acdaab49388b4dfd : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ae221366a2b4f28876338e5ad4ce5fd : SUCCESS in 3m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3819923d52cd4cd193a2c38ae77f961e : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5148f48675f04e70bb77a045dd55dc07 : SUCCESS in 4m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/baab8ea8fcad4ea4b3853aad540d7ba9 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52721d787f3944ee9027f531db3fa2d6 : SUCCESS in 4m 14s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/2988cf7d4b4d4e9ba7ac5fef912ea213 : SUCCESS in 2h 12m 02s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/4433a70741e945d89ae9649b02cf7a79 : SUCCESS in 2h 42m 13s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/e4e1c4b5b98b4053b628f9a0b57ef971 : SUCCESS in 2h 11m 45s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/37da824aca1746d1b53e767fd96be429 : SUCCESS in 3h 08m 03s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/ac095a99dde8402bbc9424ec553b7078 : TIMED_OUT in 3h 14m 13s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/cdf7d3978a9d4683a560f7746976ad3e : FAILURE in 29m 00s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/9f792cb9898d4bbfb432859c5d0c9a27 : SUCCESS in 1h 41m 01s (non-voting)\n- tripleo-ci-centos-7-containers-undercloud-minion https://zuul.opendev.org/t/openstack/build/960cd79bef1941bf965ebcf6e988f7c8 : SUCCESS in 1h 19m 06s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/0c138614117d4fcf9bd6a634bc2e10b1 : FAILURE in 1h 37m 00s (non-voting)\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/1db9fe5d83de46cc870d1f442993f0ec : SUCCESS in 37m 03s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/f878cc55fee5436ab6fde535ad90c0a3 : SUCCESS in 39m 22s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/3f94b2605e0545d58e24fea80872fbb6 : SUCCESS in 29m 18s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/6a11823ad5fc4c11af6a68cc21083cd9 : SUCCESS in 1h 25m 30s\n- tripleo-ci-centos-7-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/29bc3738c3454ad5ad8ab1be6d65361f : SUCCESS in 1h 04m 39s (non-voting)\n- tripleo-ci-centos-7-scenario007-standalone https://zuul.opendev.org/t/openstack/build/9f867bbe358d40db83f2078f6c03b2fe : SUCCESS in 50m 50s\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/c052beb1fad148d9a324b15628c11bb7 : FAILURE in 22m 52s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/8c3c3ec91e1b4f5598bc21214d0d1c4d : SUCCESS in 1h 33m 00s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/706f7aa5348f45789aff28e7fd2e6eec : SUCCESS in 3m 36s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/0125a8bfa00646a6bd8bb387f11d59de : SUCCESS in 3h 19m 23s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e5a24a13799444f590a2870eddf9ea5e : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":3},{"id":"606fdbc25c24f1d97ed11bedd6febc35063d9bf0","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2020-07-24 17:13:34.000000000","message":"Patch Set 3: Code-Review+1\n\nrecheck The CentOS 7 Standalone CI error looks unrelated. Some containers (none of them were Gnocchi) failed to build. See http://paste.openstack.org/show/796294/ from https://storage.gra.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_cdf/741228/3/check/tripleo-ci-centos-7-standalone/cdf7d39/logs/undercloud/home/zuul/container_image_build.log","accounts_in_message":[],"_revision_number":3},{"id":"f6d18b59af4810221a6fc8d55bd9f7c3232c8b82","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-07-24 20:10:01.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-rhel-8-buildimage-overcloud-full-train https://review.rdoproject.org/zuul/build/ae11deae60fe462ca4a6f70107ff889e : RETRY_LIMIT in 6m 05s\n- tripleo-rhel-8-buildimage-ironic-python-agent-train https://review.rdoproject.org/zuul/build/e8fdb4ef908349e3a21ef7bf4bfca10e : RETRY_LIMIT in 5m 48s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/8a2da76038fd4915a4b5d506b8bd1fcf : FAILURE in 2h 55m 00s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e6c0fe8fe116049bd66a323915f4ce034c1d4b29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 20:42:07.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e7eddf393a76448eb85fb96cc9f39be5 : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05be6e18273c49acb62046a642e4e1ca : SUCCESS in 5m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/261a703e036c48cabd2af880898bb2f4 : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b24c4f202e74871956b08686f10100a : SUCCESS in 6m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/45fe2cbdb0bf43569c8ecd5a5ee0aa81 : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58d3c88392a840a8811f6eb1c761913d : SUCCESS in 5m 49s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/a0169b3ccba54ababf54b08e912a70d7 : SUCCESS in 2h 09m 22s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/cbfb7878af374f889a4bca91457bf29c : SUCCESS in 3h 14m 48s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/b110dfad2c334ec2a4a117d506c50222 : SUCCESS in 2h 00m 34s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/735a44cc732b46c59a6530598ef791d1 : SUCCESS in 3h 04m 27s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/f0b0b96cf80447c38d4060c3e57d694d : FAILURE in 2h 30m 51s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/d467ee09efab4ac7b434f41047f31a66 : FAILURE in 31m 36s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/6aa05fabdbf24c9eb009418c9643b195 : SUCCESS in 2h 25m 40s (non-voting)\n- tripleo-ci-centos-7-containers-undercloud-minion https://zuul.opendev.org/t/openstack/build/002958b9851345acb01faa9496e71ee8 : SUCCESS in 1h 56m 54s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/ead0ead644864bfa92a06a4298a0a54e : FAILURE in 2h 20m 48s (non-voting)\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/5c47a9db0b4548f38ddc1a63586e2ee5 : SUCCESS in 47m 11s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/617d4881dcaa4da69b4f0bb8d5d5179e : SUCCESS in 1h 17m 54s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/24ccd5de88d24e4a818236398b834935 : SUCCESS in 42m 22s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/f16baf5e2bab4e77ad380df1ca022d0d : SUCCESS in 2h 47m 40s\n- tripleo-ci-centos-7-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/e3de2b576dec4485a4e212f483da3f3b : SUCCESS in 1h 32m 22s (non-voting)\n- tripleo-ci-centos-7-scenario007-standalone https://zuul.opendev.org/t/openstack/build/1a33b40f44864190a1776b30943c9515 : SUCCESS in 1h 36m 18s\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/4583c2e89b4f4fa7acd41fab3e82b379 : FAILURE in 31m 58s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/5312297c9b884be8a2866425cb26d8ab : SUCCESS in 1h 29m 49s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c695010b14074f8abea0d1161640f77a : SUCCESS in 4m 14s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/59b265d2aa324e349c3b056eaa70d8c7 : SUCCESS in 3h 14m 27s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4f64bf0f143041b29d3c921afa6bf5af : SUCCESS in 7m 02s","accounts_in_message":[],"_revision_number":3},{"id":"252ce32bc0861566bd6d3fe9aaa4edd232fc7af0","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-07-27 15:24:38.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"8b636909338be3826365e4bd2fb8bae8f9f67c76","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-07-27 15:25:13.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"f9e36df68a4e0e84001545526e102f5e1c25cc16","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-07-27 18:23:17.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-rhel-8-buildimage-overcloud-full-train https://review.rdoproject.org/zuul/build/8e22a75028894f84835d0eb53a670075 : RETRY_LIMIT in 6m 00s\n- tripleo-rhel-8-buildimage-ironic-python-agent-train https://review.rdoproject.org/zuul/build/39669e3bd52540c0b9b54be19663f460 : RETRY_LIMIT in 6m 00s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/b83f7408a5a9419ea48e85a860f46a89 : SUCCESS in 2h 56m 07s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"8ab406cbe013ea6de88aa3723a2f71ccd6785681","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 20:25:52.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/96ef9aa5506b4868a85e73d3a156c47c : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8846bac6975a41ba813f3630cf4cd2a0 : SUCCESS in 4m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7a9b1735aa02475e9cbda38c2a7ac3d9 : SUCCESS in 6m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/10f59ca17d534c8cbbcb8b28611ef8a0 : SUCCESS in 6m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9cc209db61ff4229ad5c8b2b998d5263 : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf99852212094d24848accd12e586f9f : SUCCESS in 5m 02s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/b6f028e6da444ff9b5017fe41afb99df : SUCCESS in 2h 16m 58s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/82230451eb2f4cbf9d6a2d919870e37d : FAILURE in 1h 21m 18s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/f753c1706e5944fdb062370904737de1 : FAILURE in 1h 13m 49s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/f5a5c948330b465aa11c095c92c844fa : SUCCESS in 3h 16m 46s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/3d83953a2517478fbc6b98298dc65363 : FAILURE in 1h 24m 21s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/e12e3bd9ed05438095c58f1a21847a5a : FAILURE in 1h 22m 10s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/6bd1395a416a4912b9a32dd6283a4ce8 : SUCCESS in 2h 13m 51s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/507bfa687f594456adf84949a30a285a : FAILURE in 1h 27m 45s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/6452598c97a3449e98ec670351cb0b98 : FAILURE in 1h 53m 12s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/416c5d488c1046dbb41cac3aa2d6357d : FAILURE in 1h 27m 50s\n- tripleo-ci-centos-7-containers-undercloud-minion https://zuul.opendev.org/t/openstack/build/58a3eec7b2c44116b62fd82183aeb388 : FAILURE in 1h 15m 34s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/a8e75cb0873242eea45dd12c4232887c : FAILURE in 1h 16m 49s (non-voting)\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/2d2be36b67274754b9a1a9f5b6ef6cd9 : SUCCESS in 55m 43s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/7ca96bb87266468db4392bf0b6d77b0b : FAILURE in 1h 34m 10s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/2f50e106910847b181e5222766c5c957 : SUCCESS in 51m 15s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/2f83209c08cb4a6cbb64a359448e1eb9 : SUCCESS in 43m 33s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/a2a2d3d51dd14445a96a69b42c326b68 : FAILURE in 1h 52m 07s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/7bb5103cde0d4e8e876dbe0edb8d20de : SUCCESS in 1h 16m 48s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/6bb75acad42c4b0d946e70819461460e : SUCCESS in 1h 24m 32s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/0dd38fa4b3164ea3837a64c5ad8c9c09 : FAILURE in 1h 20m 06s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/f4167139855e4f1d9facf07cfccf76f4 : SUCCESS in 1h 26m 13s (non-voting)\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/811b026fd50d41d896cdee9f4ed6564c : FAILURE in 1h 19m 43s\n- tripleo-ci-centos-7-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/707569ea4d554b5a937d87d3f164eaed : FAILURE in 1h 16m 04s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/712a49f83ab04eb8875ec219ba20dc48 : FAILURE in 1h 25m 24s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/f5f2e1b2022d4c83a31f1e2573e10d5e : FAILURE in 1h 22m 53s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/b35a515ffe904889a30752d95b0fa787 : FAILURE in 1h 23m 34s\n- tripleo-ci-centos-7-scenario007-standalone https://zuul.opendev.org/t/openstack/build/dd9f48925e8b43d4bae15848989607b8 : FAILURE in 1h 17m 19s\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/60a002e7725d4f3d9698b695b4cd1e57 : FAILURE in 1h 21m 35s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/7aff4a2784754a93b6c66c27d150cfb3 : SUCCESS in 2h 16m 44s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/24f64bfee51847f69e9891265f00c0df : SUCCESS in 2h 24m 06s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/a8d804103efe4357a3f8cdc79c8073e3 : FAILURE in 1h 17m 20s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/cf4e48864ee44c8e919dba9de80c7833 : SUCCESS in 4m 19s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/dafe22f798d44077885ea302a8349732 : FAILURE in 1h 16m 42s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c2c2e77fe2d743b8a76259e1ee496824 : SUCCESS in 8m 09s","accounts_in_message":[],"_revision_number":5},{"id":"b351a9c8dc9d74990eae65ec6869446f77a53e63","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-07-29 02:32:52.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"94169c30072a20ac44ebcc1b069724d4196a4974","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-07-29 05:56:53.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-rhel-8-buildimage-overcloud-full-train https://review.rdoproject.org/zuul/build/8bab78d5be14455cb21c99c22e4867d5 : RETRY_LIMIT in 6m 09s\n- tripleo-rhel-8-buildimage-ironic-python-agent-train https://review.rdoproject.org/zuul/build/da51d1ad7b1e4b1b974d673b6d6e441f : RETRY_LIMIT in 6m 00s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/c1d7c5433c30470bb87c964c6f12ab48 : SUCCESS in 3h 21m 53s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"c0f5d9d6deb85756e789aa65983480e26ae170e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 06:05:37.000000000","message":"Patch Set 6: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":6},{"id":"548d4029dd508c9d21cc29f7143e45c5e0dcb3ff","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-07-30 12:49:19.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"885819e3e7b6f857f2f8286f033ceb9b56316dea","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2020-07-30 14:08:15.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"0c195d28944461bb3a826f36f94fc2277f108f88","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-07-30 16:22:03.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-rhel-8-buildimage-overcloud-full-train https://review.rdoproject.org/zuul/build/16f3247722bd46c49c6874320d5485ec : RETRY_LIMIT in 6m 32s\n- tripleo-rhel-8-buildimage-ironic-python-agent-train https://review.rdoproject.org/zuul/build/75d03058daed4cd6b2e667aed9fd3f7c : RETRY_LIMIT in 7m 13s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/ed7cd4e1fbe845e4991b5a7dd5b8b763 : FAILURE in 2h 58m 21s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"3537483125b1bf5ed783cd97cc3d72e1b8735fdf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 23:27:15.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cfe30d986d9845a6bff19f9ab61fc7c5 : SUCCESS in 7m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/381b851e59f741568c2297fdcff25699 : SUCCESS in 3m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a687fc90432947c8ac9fd5e828f1186f : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4c92801cc8364902984a9231d2c1fca1 : SUCCESS in 4m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b30a6bf810b54b0984d1b5f306d3bbbb : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6468d65b79da4502b99d1e62d74d705d : SUCCESS in 4m 58s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/51ceeb43bd044f4bbe64f5b1267e15ef : TIMED_OUT in 3h 06m 52s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/a13e282efb644c13a69b2ec8548385d2 : SUCCESS in 2h 10m 25s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/4f152d744d75498f95966e566843498d : SUCCESS in 3h 00m 18s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/ecfc92b5b40941aca50030ade905952a : SUCCESS in 1h 59m 14s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/5823fcf58c294893a18f83631a364284 : SUCCESS in 2h 27m 10s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/ed60e316a13e47179d44d0135a67a828 : FAILURE in 2h 30m 15s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/968abf855cfb4626aac6176d12a24acf : SUCCESS in 1h 41m 08s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/e7c4c1cf80e74d039a27497c91f5d31c : SUCCESS in 2h 06m 43s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/15b09c6fd65243bf8b1e46b9bbe337ca : SUCCESS in 2h 21m 41s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/1e0cbcf7899f4a0291f30441fe44cb93 : SUCCESS in 1h 58m 54s\n- tripleo-ci-centos-7-containers-undercloud-minion https://zuul.opendev.org/t/openstack/build/b298120116cf4a4d891091ef2f389584 : SUCCESS in 1h 36m 51s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/133032debc1c4e5e8f70f04c67ed93f5 : FAILURE in 2h 08m 51s (non-voting)\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/9098681889e743fda88b0d3800993755 : SUCCESS in 58m 09s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/bb117d982f15480fa4aecfe8b9a84105 : SUCCESS in 1h 34m 46s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/6c1c797626f34f4cbcfca4eb4ce25621 : SUCCESS in 49m 03s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/84449f782b8548d7a9d20fe602f649d5 : SUCCESS in 41m 57s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/1463782ae95249a88369fc5c6ec8cadc : FAILURE in 45m 54s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/b1959dc87ab749c4ab648a2d4f1f83db : SUCCESS in 1h 05m 03s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/b09d16d020db4e2d8475041fc90c96fd : SUCCESS in 1h 23m 38s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/ea2d53e901a64fd2b4a38a47ff7a96f0 : SUCCESS in 1h 47m 19s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/7ff65bab0c63404496bc36a42ce38624 : SUCCESS in 1h 54m 14s (non-voting)\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/cb69d0f190a24048834cb8bc64f9602e : SUCCESS in 2h 36m 00s\n- tripleo-ci-centos-7-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/9b5bea87a9384172991c8773132f5083 : SUCCESS in 1h 26m 52s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/0788695c8620496ab06c66b7cc9a54cd : SUCCESS in 1h 27m 18s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/43e63f89727b4e31978cd3d51b9aaa8a : SUCCESS in 1h 26m 59s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/88805a3985a842ce9f30a62248d0eca2 : SUCCESS in 1h 07m 02s\n- tripleo-ci-centos-7-scenario007-standalone https://zuul.opendev.org/t/openstack/build/9cc09a997f3448ea95822558f91827b6 : SUCCESS in 1h 05m 06s\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/39182fe98f274761a8223745343cd3bb : SUCCESS in 2h 05m 22s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/1ed152a7f47e418a856c94881475c201 : SUCCESS in 1h 43m 39s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/480d2c4417804d77986c20a719e9985b : SUCCESS in 1h 25m 14s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/5328c597f8b04d5dba7794eccad0aa69 : SUCCESS in 1h 37m 47s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/086b7e3919ee4071a1abe5ed91c19518 : SUCCESS in 3m 35s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/ad5ac988300240cba4990a249cd2f349 : SUCCESS in 2h 15m 23s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dedf4fc831b249f0a65eaeb609f1a1af : SUCCESS in 5m 56s","accounts_in_message":[],"_revision_number":6},{"id":"dce4aeaa9d85688b44e9f415d6462fdd67de9e69","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-07-31 05:41:09.000000000","message":"Patch Set 6:\n\nStandalone failing with panko-api not found, panko is not there in container-images/tcib.","accounts_in_message":[],"_revision_number":6},{"id":"f574b760f12a890930c7d413b7f2e8a8b796d318","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-07-31 12:23:30.000000000","message":"Patch Set 6: Workflow-1\n\nI need to add panko","accounts_in_message":[],"_revision_number":6},{"id":"536beeb6f9723943a66b1c13189a9ce07ae393c9","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-07-31 13:24:11.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"4f564087b0da73ac9d305e65451e031a4ddc361c","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2020-07-31 13:26:04.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"8d4ddcaef80fc5d87e67c3dff2e438aca345ee95","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-07-31 13:30:05.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"dbb85842167b9aa7e2b927c94efa183e79925dfd","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-07-31 16:13:30.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/c587212309ce45c6a331de139efa263a : FAILURE in 2h 42m 01s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"cc89a64f145467cf2b0b5d023a628dfbf4f86327","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 21:02:01.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/42e5ab4e6d064dd8b59642d81cbabc30 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/910e49ace19b4ffea24366c1aad39aae : SUCCESS in 3m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0aadd8cff0e64f018c3fc823921363cb : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b19316926ed04847a4aa52b58db37984 : SUCCESS in 5m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/eeb56a4ee1754eab9ea0222cb5233e8f : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/759db6a02ff54ef0ae30dd172cb1917f : SUCCESS in 4m 54s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/5c23ee5998f8448aba9528839eea9101 : SUCCESS in 2h 37m 21s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/4f45e6eae98e4d018f886c8aaca2057a : SUCCESS in 2h 03m 55s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/55e659a1fd8046cd96a3805df2ef7845 : SUCCESS in 3h 13m 23s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/af89f4a7f9524009a3c919860694e930 : SUCCESS in 2h 17m 36s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/8b34e3d6f7124fd7abcd10d8c981d83f : SUCCESS in 2h 32m 57s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/492997114c56479b8e9a9b944c89fe5a : FAILURE in 2h 11m 56s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/ba237931a6764fa6853ed22eb690bba5 : SUCCESS in 1h 39m 45s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/0dfab5ab600d4cf1a907506380ca5475 : SUCCESS in 1h 38m 15s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/9be1188ccdd146bcbc1fe87d4b00c876 : SUCCESS in 2h 15m 46s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/02c8106e0487474680451fac344019d1 : SUCCESS in 1h 49m 12s\n- tripleo-ci-centos-7-containers-undercloud-minion https://zuul.opendev.org/t/openstack/build/e02d701dbc2741ed862d2ef31fbcba3d : SUCCESS in 1h 33m 35s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/ba86ad606a6f4022b8140bfd5b88be12 : FAILURE in 2h 02m 16s (non-voting)\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/78a3f2acc2c74f7fb7ff3623fbade113 : TIMED_OUT in 2h 03m 26s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/48f04c4bd877494682d4186c5a7a122c : FAILURE in 1h 37m 11s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/3381d240f5a24eb7a2761d5632321615 : FAILURE in 1h 54m 07s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/d117a7bb1c3c4a3596af3e6d6b90aa7e : SUCCESS in 42m 18s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/bd715bbf71bb4dd79996baf312e818c6 : FAILURE in 3h 14m 19s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/e7aca61f5386491e98435817e7487e48 : SUCCESS in 1h 59m 40s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/d1929e6a3aba4c9b8885144b38042a33 : SUCCESS in 1h 10m 25s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/3040a0e87b4240a5abf1589af9e6a395 : SUCCESS in 1h 41m 29s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/6efa93e16d4e4aaf94184ad98e0865b9 : SUCCESS in 2h 17m 30s (non-voting)\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/9a0c73053d7d4a1a8888f9566ea8c60b : SUCCESS in 2h 21m 26s\n- tripleo-ci-centos-7-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/ae0c29a0954c46f0bd893f91737c7f32 : SUCCESS in 1h 28m 48s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/17cca03e46a644e3843665a64f565f3a : SUCCESS in 1h 38m 11s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/43aec5fdf6c8424595c5edd9a9e1d213 : SUCCESS in 1h 00m 19s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/636b5dfce44b4f259f81d6a4486ebc4c : SUCCESS in 1h 07m 18s\n- tripleo-ci-centos-7-scenario007-standalone https://zuul.opendev.org/t/openstack/build/9c9910ed108e4f00a5d17edacb0bd674 : SUCCESS in 1h 56m 57s\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/e394c8950ed649269677c96fb11664ef : SUCCESS in 2h 00m 55s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/66fe8f2e3c8c49b598d85c42f1037934 : SUCCESS in 1h 22m 17s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/e00bf43171e74f99b4456f6b57556b0d : SUCCESS in 1h 18m 28s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/7e0577fdee5e42e18bbe606ec6c0d7ff : SUCCESS in 1h 48m 53s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7061f041ca4646d0a55970fcb1a7c49e : SUCCESS in 3m 39s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/8045b15c60e94c67bbd1e2cc2cc827c6 : SUCCESS in 2h 53m 28s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/22d016c478244ea8bbce12e21e1d4ef0 : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":8},{"id":"197c5f1976c2a4b056f6d8ac1346a7e8b3b94697","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-07-31 21:52:31.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"5ac2149ebdc0bdb66b4a134edd90f2942c1e7480","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-01 00:51:38.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/983d49b0292b4174ab907cfb93a2b9e8 : FAILURE in 2h 58m 00s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"0b6301834f4b2cdd7fa57266902d186ac544ebab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-01 01:34:56.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/694190b5bd16465daef5722518da7bf2 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/963e25f1238d49b6948e9dc58b817873 : SUCCESS in 4m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/618572fce03941eeab24536bbf7e01b5 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44815cbc46804e8f8dc10ed55c00f57a : SUCCESS in 4m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a0a69d8ae6494d51940db4734a8e1152 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/784afba887b545218a843e3a5eeef3ca : SUCCESS in 4m 18s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/060a2f6a49814b5097534d08531cafd6 : SUCCESS in 2h 23m 49s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/a4138fa69ddb4ab191aee2d30c8f351b : SUCCESS in 1h 59m 24s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/a5427b9952064728b7f8517c0899c9ba : SUCCESS in 2h 34m 37s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/de56369362e04dee8c3539cd28179ef7 : SUCCESS in 2h 04m 58s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/363fc2b283934d3286aec27feef10033 : SUCCESS in 2h 26m 42s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/423b616e852446da85e3038764595eec : FAILURE in 2h 11m 13s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/431ef11e94614a8697cb5d24638f5286 : SUCCESS in 1h 21m 54s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/ca04964c6fa64199b85b3d005cd4f53c : SUCCESS in 1h 27m 17s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/6bfd4fa7f05f4ce683776037779b9d8e : SUCCESS in 2h 16m 17s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/cfc1e8dbedec480baa8ce7af5b92f202 : SUCCESS in 1h 35m 23s\n- tripleo-ci-centos-7-containers-undercloud-minion https://zuul.opendev.org/t/openstack/build/01587f2a220e43959114293cf3e5d0fe : SUCCESS in 1h 08m 50s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/e22e36d451c74e2c8c034ac5ce1e055c : FAILURE in 1h 59m 37s (non-voting)\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/c08912dff0f34e2ab1ac14e27195562d : SUCCESS in 55m 10s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/f34eeffe089844a2805b5727eae3ad6a : SUCCESS in 1h 26m 28s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/f075357a43b94a62ae43557578769df9 : SUCCESS in 46m 07s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/57d841a0846e4a258fe16f1ce238ae85 : SUCCESS in 42m 08s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/72b24b8eca94408c9e051e1b6e5b5f59 : FAILURE in 2h 57m 09s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/a013efcfbb5e466a92fde5103076ff71 : FAILURE in 1h 58m 46s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/f3d7029b7ed344e59a284d5635f63430 : SUCCESS in 1h 10m 52s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/b93266f418ec4f359b460286b5debb2d : SUCCESS in 1h 41m 09s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/22d12e31fa854e2989dbd65b6d3e6580 : SUCCESS in 2h 20m 33s (non-voting)\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/2b3381e9000a487aa272703755e1b08d : SUCCESS in 1h 57m 48s\n- tripleo-ci-centos-7-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/620e68bc7a0b4c1db82066ce13d1f259 : SUCCESS in 1h 16m 48s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/5d4af8bcb3f84f96a1540b1608a16d6c : SUCCESS in 58m 56s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/cbbab2f6e5744e7289f3cb002d3c92b1 : SUCCESS in 1h 39m 02s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/092094c21ee041e9bca7b788ceb1b36f : SUCCESS in 1h 23m 36s\n- tripleo-ci-centos-7-scenario007-standalone https://zuul.opendev.org/t/openstack/build/276caf7f8e204ad9b12e9107c5a87b43 : SUCCESS in 1h 04m 37s\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/a60d7c41694f441b8a019663b86cd049 : SUCCESS in 2h 08m 54s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/a22130fcbcd34cf8964458ea89a6756e : SUCCESS in 1h 45m 14s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/257704963b2340dba1d029a436afe250 : SUCCESS in 1h 03m 05s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/509bb4579d094abea6256821428b2030 : SUCCESS in 1h 47m 32s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2cc191eb2ea64a2fa9f0bc787abca3e8 : SUCCESS in 3m 31s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/d743ddff9ca74cfebf4f0a04ff7d5125 : SUCCESS in 1h 42m 17s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ed1a7d5d344f497d866e981b23b945ae : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":8},{"id":"fff3f889f7556dd15c3a61a63149c4003e5fb1ef","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-08-03 12:55:05.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"e8b9042a7198fce68e4bb8ab7a0c383d96a17162","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-08-03 12:55:32.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"9307b5d8d41b12df9efcca3050e61bb3d0443a37","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-03 15:22:02.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/5df84f6655944cf1ab754dbacdf68fc5 : FAILURE in 2h 26m 05s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"0d3ad3e1b02a41c1e26ddf9b2c6542f464731ba5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 19:46:26.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cb0405613ffd46e4ad79febb68668bc9 : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2dfef976c9f6455e851c3b6f0edc811f : SUCCESS in 3m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b2f86b31473f4ce3ad6a55a45396adf1 : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/691a39dabbe34d748ecaa16b4e8aca96 : SUCCESS in 4m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1cd0ab0453fa4c6b9991bfaace652740 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffeea8a798864a6e87f4b3a1059678c0 : SUCCESS in 6m 12s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/b4cc2bfd3ba944d7b7c1cca42b27b163 : SUCCESS in 2h 26m 18s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/11e57f2b29c9420e991182a4945aa903 : SUCCESS in 3h 12m 23s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/80593e203c1842f8b26b000cfe72b478 : SUCCESS in 2h 29m 14s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/83f5a940d5744e97834ecf2252f0e8bb : SUCCESS in 2h 16m 48s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/c38e503b8a2949d6b47050c21c434712 : FAILURE in 2h 21m 55s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/a7826065b95d4dd991b91abf39857e27 : SUCCESS in 1h 39m 44s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/c1843f3069fa436fb5cb8033ebd1e2f5 : SUCCESS in 1h 36m 50s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/d5ca182fa14f4667b11fc2e13dc042a9 : SUCCESS in 2h 00m 20s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/d5c3d92dc67a4181964e7dee9a1efcb1 : FAILURE in 2h 09m 55s (non-voting)\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/d9c9103c1e5b44f5af6535acbf11bd19 : SUCCESS in 59m 19s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/717b0b6cb8d146d0ac06aab6c4cd5ebd : SUCCESS in 58m 37s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/76028298ca5f4c3c89c010764146274f : SUCCESS in 1h 03m 59s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/734f5a6783ad41b6808c1ff929dd938f : SUCCESS in 40m 39s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/ff2a64173c6f40ec80831c99e743877a : FAILURE in 2h 47m 57s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/99a53c511ff047f99e07a852b883e7a8 : SUCCESS in 1h 45m 41s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/7d62474e18264b2aa28fe272dd0f6fa5 : SUCCESS in 1h 17m 14s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/a1d69b41315a46c1b30121713bf062f0 : SUCCESS in 1h 48m 37s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/4127f780681c4c5bbcdd8ad777db8884 : SUCCESS in 2h 10m 59s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/f70634cc7cc0420db582963ba63a6d2e : SUCCESS in 1h 27m 05s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/46932273382541cf862a6a2ae052d17d : SUCCESS in 1h 44m 10s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/11184507db1d41ca876a821eef702e29 : SUCCESS in 48m 12s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4ad55e2441e641f3ac782e3362fe2989 : SUCCESS in 3m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e899549bf7d44cf1ad2068e870653477 : SUCCESS in 6m 04s","accounts_in_message":[],"_revision_number":8},{"id":"5efb339063fdfd464a5f1755828a8b7929b4c4ee","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-08-03 20:01:46.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"bb7b02d51a6ecd51540bb09c34dad8029e4698a3","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-03 23:03:51.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/30f2f8390b0d48d7a15692fbe90bf148 : FAILURE in 2h 58m 10s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"fc28c70a070840c2e89b17e97e3388dc7ab5021b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 23:30:39.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/25d77d10401b4f24b59bfe9c32730922 : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74d6291210dd4152a4e080c1a6547f9d : SUCCESS in 3m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/974a4c7321e64851b6d2c590b605796d : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c9c7bd44145f4f9cb525a4e097da6efe : SUCCESS in 5m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2058465ba9844d2889fffc8eeefdac1b : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79fc8eabc84348a7be526c83ec12a01d : SUCCESS in 4m 10s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/d3bffc6a08f94fa093ac686cbe4b18a4 : SUCCESS in 2h 32m 33s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/dcf672a1335e4df199943400831f8c06 : SUCCESS in 2h 52m 27s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/17cdb290d2e4429886b0ca79282131c7 : SUCCESS in 3h 10m 29s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/45cf725d1e6245458e82c6b760232936 : SUCCESS in 2h 58m 44s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/24e1020a10e14f1d80955304958b9cfc : FAILURE in 2h 12m 33s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/467759acb4fd452eac10ca23d61a396a : SUCCESS in 2h 52m 13s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/bd9b14c0584942328c0224d5fa62ebd8 : SUCCESS in 2h 30m 54s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/721e9585eac8437e810a1db0a1af9d8d : SUCCESS in 1h 54m 34s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/ab560694080b45478e037625e840d77f : FAILURE in 2h 29m 47s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/5a08b44b28d84908a4d0ba94f885e5f9 : SUCCESS in 1h 18m 48s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/299ca161b6ca4947ae8cb6e048abfef4 : SUCCESS in 43m 45s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/8efaaa8b89114d8a9437a8c0c154e17c : TIMED_OUT in 3h 19m 20s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/269630840052411a99e2a00ff38a48cf : SUCCESS in 1h 50m 24s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/41e3aa9e936348a59ce7057811ee7f1c : SUCCESS in 1h 12m 36s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/620213fbe712447c849972031ab37f35 : SUCCESS in 1h 39m 57s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/9689fa29306746d2b9be3e546f80296a : SUCCESS in 2h 10m 15s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/5d18f2b64ed1461d8c76607ddf27dfcf : SUCCESS in 1h 31m 00s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/5bd0bd9ddbc742d493ce0e2be09d941f : SUCCESS in 1h 19m 14s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/2c18d5c80319454aae0101aae1e7a5fa : SUCCESS in 1h 22m 32s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/17a83aab13f445009473ae15ba17d2ec : SUCCESS in 3m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bfb12a4a7ee74721b7008325047bf893 : SUCCESS in 5m 35s","accounts_in_message":[],"_revision_number":8},{"id":"322ee75fe7b2bb08e8de319710f5803d52c541e3","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2020-08-04 07:00:26.000000000","message":"Patch Set 8: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"22a2105474063ca616ff083be8d211057ef12f4a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-04 09:32:09.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/b633d22cf1be48809e24a592d9264b28 : FAILURE in 2h 30m 29s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"f152765b6462b14fc4c865b5fc8783ddfefe0c3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 11:04:37.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/371039dd50db4ea9b7953ef0f303f613 : SUCCESS in 7m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07ef7ff885e24a3d9c0139c4d10277b4 : SUCCESS in 4m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/723088a93fdc4351b4759fb38b134be1 : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e7534a8699c48c3997c966606b7d39f : SUCCESS in 6m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/773c8881a9ef4cab872563be85831185 : SUCCESS in 6m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98899be2e032411aa0d50f2882b6309b : SUCCESS in 4m 06s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/76551836f2e04e53857301a9e15a71df : SUCCESS in 2h 26m 58s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/f44688e8b45c446287edf2a15b33d63e : FAILURE in 3h 33m 04s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/f791bc42a3044b38ac077d54bf62f2b0 : SUCCESS in 1h 58m 03s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/52a30a12b45447729c768d669d1cf03e : SUCCESS in 2h 12m 34s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/a37ee691f431455dbe92592b018e7c1a : FAILURE in 2h 13m 50s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/83df51ae821f4e83a91ee0d2e109e7df : SUCCESS in 2h 30m 23s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/e82766e0de7b4c1d9239d06f80f6bd0e : SUCCESS in 2h 24m 33s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/1cfe2c2ed7534c3c83b275ff89737727 : SUCCESS in 1h 46m 40s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/c166257290ce49ff83b4e67a076df8a8 : FAILURE in 2h 09m 09s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/4637bd6126f74746862ae2cbc3314c4c : SUCCESS in 1h 05m 50s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/a057de51c5744acba11f682972084171 : SUCCESS in 44m 09s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/54e807b1021349d8bce4eecbc3efae8a : FAILURE in 2h 43m 25s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/606ce7e933f344958c624fe6073122bf : SUCCESS in 1h 39m 36s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/ae6fad37b0ea4e149f59801e5bc43150 : FAILURE in 1h 58m 16s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/25b43aca995b4afab0d8c989cf0f2ef5 : FAILURE in 2h 39m 16s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/7ba56745bc64433bbf10b72d75d5ad68 : FAILURE in 1h 23m 01s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/b12247031e7b448ca024631550d2c9d4 : SUCCESS in 1h 35m 55s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/891b395293cb45de9e6a4b66131b567e : SUCCESS in 1h 19m 48s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/d54b6e1655c34538b83b19dd89996dce : SUCCESS in 1h 08m 21s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4ff140e0285240bd8f40cd2a292015da : SUCCESS in 4m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cfc5c1aca5dc4f30be263c7bc18b7d80 : SUCCESS in 6m 35s","accounts_in_message":[],"_revision_number":8},{"id":"6267f50c772605f65a0aeb2c5fe45a74ed364ac2","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2020-08-04 13:19:02.000000000","message":"Patch Set 8: Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"c33acbdd38a220587c84b01c33f0ee04983398ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 17:04:43.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c6c1bc02d23646399ca9ccf9c9e36c7e : SUCCESS in 8m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8bb5456c41cf4f98935c98ef394402c9 : SUCCESS in 5m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/99915441df89410680e6d960352d729d : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b7c69658d42404480406c88f21dbf67 : SUCCESS in 6m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/abbd720aa4094633be23eda5a7ca02cf : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02298328294c490fbe43d929be3e2ecb : SUCCESS in 5m 33s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/c592939ed4024d4daab3e57ad7a41bdf : SUCCESS in 2h 27m 52s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/532df4b4f2fe4423afb546cf1a622cdd : SUCCESS in 2h 57m 06s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/188869eb3c0a45a99b9d5c2030151da4 : SUCCESS in 1h 57m 42s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/d2b44e99e091403a9375202fbb40068f : SUCCESS in 2h 12m 50s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/0a49223f001d47ba97180d3731b788e0 : TIMED_OUT in 3h 34m 39s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/0c98bef90c28473ca08aa481a75b196a : FAILURE in 1h 39m 42s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/99f19587f3104a2cb556ced1d06737fa : FAILURE in 1h 40m 11s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/4ee8d7b21964488eb7b7fc99cd8e44cd : SUCCESS in 1h 46m 36s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/13e952b5bfd94314bf2df6c196570e59 : FAILURE in 2h 10m 35s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/2eef7860e7184109b4297404ca3dfb5b : SUCCESS in 1h 16m 41s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/6e25646b70a64283980f213694b0fc93 : SUCCESS in 52m 26s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/65afbaadec844303a18f5508263f9402 : FAILURE in 2h 57m 47s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/1157d9a810e54a788d0f23e961fe4a40 : SUCCESS in 1h 27m 15s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/b9806ebfb66b4aa5a27de6b9326a9e43 : SUCCESS in 1h 13m 59s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/1c0433cb0f324195af6e2a0df31fa782 : SUCCESS in 2h 14m 18s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/a77c064bc4634337819714982ab003e7 : SUCCESS in 2h 29m 16s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/86d001db094e4f07861be5932c10d533 : SUCCESS in 1h 30m 35s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/08b5c206bc794fc7bc33fdb6c6e419b0 : SUCCESS in 1h 12m 38s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/392e970171fd4034a1dde533e895f561 : SUCCESS in 1h 08m 41s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0fe87799688947e6ba5a63f24d776636 : SUCCESS in 4m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4997a7ae7c364627843fdd0f356421b8 : SUCCESS in 6m 24s","accounts_in_message":[],"_revision_number":8},{"id":"f028893b36fc9a24b033d719e953980026f4e294","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-08-05 20:12:21.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"ffee30446b32e8d71631aaaf8dbe7f587cc3bdd4","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-05 20:16:00.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"cfb34ae24854965ce01f3f69671b450f56882542","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 23:32:08.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a64aaab0526344e790dedc8267e257b9 : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c01d484483f44338d5384cd449cb12e : SUCCESS in 5m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/23918e5597bb44c1a6f16fd184dcab15 : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/781fa2cf8605450d83b18c69deeb856a : SUCCESS in 5m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2fb0ca98516a48329880d6217baab90e : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fef8c79aacd45cf95d1317282997e21 : SUCCESS in 6m 36s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/3d302e9ecd2b4f2abf82b08a92283d3c : SUCCESS in 2h 11m 40s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/274e56e7b0a4422199fa77ccc8b48160 : SUCCESS in 2h 37m 52s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/03bca4cd7c034e6dab07df7ba358ec94 : SUCCESS in 2h 11m 14s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/8a224ba1722e469483fd1043653614a0 : SUCCESS in 2h 30m 48s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/0599110de676418d94404e3290c4a465 : FAILURE in 2h 14m 14s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/e7c3cb5b39854280963fc7d682119e2f : SUCCESS in 2h 37m 54s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/16c74230df064397860602b9fddacbb8 : SUCCESS in 2h 11m 49s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/b091f45528ba4160b3667a01be8a2c1e : SUCCESS in 2h 01m 57s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/73053322ed4c43c2a4cb937c666e6406 : FAILURE in 2h 31m 13s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/4d183ea14f134921a6829dc548ded3bb : SUCCESS in 1h 17m 41s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/bfb71fe99b764f878198eababccbccdc : SUCCESS in 43m 40s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/b1babd906382458e95ee9b9ba66099ff : FAILURE in 2h 51m 49s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/d6f1058d40cc45ff9ed2a34def6a0ed9 : SUCCESS in 1h 34m 32s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/f12ede5779c94f5fb7fd27ff693dda96 : SUCCESS in 1h 09m 24s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/6bd9d4e332304e1e879e6ca4a6b1dfde : SUCCESS in 1h 43m 26s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/7ea1061ad4df4eafbbf642e6f6449de5 : FAILURE in 2h 17m 33s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/14bce6f0ec8440b298aefdde05a42948 : FAILURE in 2h 17m 48s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/bdcae52cc9a543ada91a11f535df71ad : SUCCESS in 1h 17m 26s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/fcad336d697547099a44fc3b03c2ae64 : SUCCESS in 1h 06m 20s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c120a67b872f4982b5796e151758f601 : SUCCESS in 6m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9e52632907ea4260b0359298ebc15f86 : SUCCESS in 6m 36s","accounts_in_message":[],"_revision_number":8},{"id":"08ad40276474363b4891e63ec8208b8e9ba13f82","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-08-06 20:51:11.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"ff52dd1fca1a1283f026d266b109b145688b7811","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-06 23:55:47.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/0f126d552fff4c11906426396a855a71 : SUCCESS in 3h 03m 34s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"db79fb21832b488f0708df0a325e5e945b773f39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 00:02:26.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bf69a49bb1a64f56ab85cc2f8abe79a4 : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5d22dfe0abe4f488ccc0ee83090e94a : SUCCESS in 7m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/25eb8df2c4c8454891c0073e8d674426 : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fcec576e4246405b8b3b6f135bd10a57 : SUCCESS in 6m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ba3ba08b1cf84d4297affad541adfa52 : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79af4294794e460f9a21b7618dffcea0 : SUCCESS in 6m 38s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/204a6c40f2344eed87cb518ce9627cbe : SUCCESS in 2h 18m 46s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/72e1e576770e42aeaecb899d2a688fb8 : SUCCESS in 2h 46m 46s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/4b9bb37531014f8cbdeae150803ccad5 : SUCCESS in 1h 54m 16s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/e41f4645621b4d59a6ba437e5f0c275d : SUCCESS in 2h 36m 38s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/dd16d7632a2447ff8c0bdd34fd793f3f : FAILURE in 2h 05m 29s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/b68122bcc00e425cb9caa7c5b3630f94 : SUCCESS in 1h 57m 11s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/56d134e0707f4239abf2c55f9962e560 : SUCCESS in 2h 49m 05s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/b6b6451f38e641b3925434b373fab177 : SUCCESS in 1h 50m 17s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/c4d6553004324fc79898dd5298bc35f8 : FAILURE in 2h 07m 00s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/5d705ae3f5a54e32bc58a7144c1748f2 : SUCCESS in 1h 14m 45s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/9d030da3e5024144a856c74b59ae095e : SUCCESS in 39m 23s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/6b9062e244724743b890183fa6b1c1e9 : FAILURE in 2h 48m 03s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/7899668573ce4ef8bf2b8291e47e9f22 : SUCCESS in 1h 39m 07s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/1287208aa1684df681f7c0ebf65d7478 : SUCCESS in 1h 37m 10s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/e76a9b851c2a42d29056623737771442 : SUCCESS in 1h 53m 07s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/75f320b44b034828a7ce7fc967b4e20f : SUCCESS in 2h 04m 30s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/7f6c9f2848564372a1001466b4fa6f94 : SUCCESS in 1h 51m 28s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/3e99e4f0c8bd4641a94b0acdaaafa3bb : SUCCESS in 1h 20m 19s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/765534fbce3d474294cffdbec197fa64 : SUCCESS in 1h 16m 49s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fc51eb5bebd84150aff9a9ece07683df : SUCCESS in 5m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ab52499fcf8d4031accf14d1143d3f51 : SUCCESS in 7m 46s","accounts_in_message":[],"_revision_number":8},{"id":"7bdb99c30f25ddaf7092d3fd2ccd5a4e72ed3989","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-08-07 06:00:16.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"8eb34fbdeec9e30f7193643f1cee639af5f24fb5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-07 09:22:22.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/fb0e5b48beb84481a437c0ce7c047131 : SUCCESS in 3h 17m 36s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"fba22df0089275f9628542becb596a710501e551","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 10:22:02.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f9ebd07d688240098cbcd01d25c5fa98 : SUCCESS in 8m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2889fb4f4cb843e68ba3deef7cd1070f : SUCCESS in 6m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b747306327d84f80bc22b562451d25e4 : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d790f630f46d4f06a65ee37aeb837160 : SUCCESS in 4m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/77490d817b6e4919948559eeda7cdd5e : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bd8483252354327b72a816186741703 : SUCCESS in 4m 59s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/fee4cbba48ef416cb34bf8a67c660655 : SUCCESS in 2h 16m 52s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/f9c4377b0292445084acbedf22e5385a : SUCCESS in 2h 59m 00s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/b280ab4dd76b41feadf7f5e510b1c7ad : SUCCESS in 2h 07m 52s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/b92e76f9642e4f4e867181c268f38599 : SUCCESS in 2h 59m 35s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/58ef6d3eaeb7422ba8b09931f49b3518 : FAILURE in 2h 06m 41s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/a0b0cae000e04ab5a9f99ae88f92a533 : SUCCESS in 1h 51m 02s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/f05585b2e7f0456ca8f4b909bc5a2674 : SUCCESS in 1h 36m 19s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/cbb61cf839004841b2b03b806741bc13 : SUCCESS in 1h 51m 36s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/239cf89c6c6149c390c092d7a3b6004c : FAILURE in 2h 24m 37s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/53259fb5f28343649b06be0f50658b8b : SUCCESS in 1h 19m 54s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/988cc90214c049fd9b4727d09f41b7c8 : SUCCESS in 40m 52s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/d517a4b263704ab087b43359afa73643 : FAILURE in 2h 50m 53s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/94f964574f8a4c5c83c1d85cedd6f571 : SUCCESS in 1h 56m 12s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/ddd1c96a68404f66bf1034ef03ab5585 : SUCCESS in 1h 11m 12s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/da264044ee61490699962251d2880bab : SUCCESS in 1h 43m 33s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/c12858b312ac41b8b737de4df60f52e5 : SUCCESS in 2h 11m 54s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/c71daa8b8c4d4d53a307f69b112ad069 : SUCCESS in 1h 32m 17s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/ab6309aee93043eab7657ff5c6bfafca : SUCCESS in 1h 24m 38s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/65eae1e738cc40fbbd8d5237e92c8ec5 : SUCCESS in 1h 00m 15s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/96568324f78142c78695f9f1027bc96a : SUCCESS in 4m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/92c6bba8a85b4d3299796c87410ef9a3 : SUCCESS in 8m 57s","accounts_in_message":[],"_revision_number":8},{"id":"380c20251539a345d5dd4978bf83f688879727d9","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-10 13:57:01.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"139e65a12836bc59ffc01584550cdc60b9633b70","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-10 14:16:07.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/ea172e2fb56a4b498fe3846717b1ab6f : FAILURE in 17m 19s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"0351d67c07cbcd38c194c5946464020b93258d22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 17:56:34.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/677c87b5c93f438ab92ed0a1b54b23db : SUCCESS in 7m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5a367dd7b774538bc26076bf3c272b5 : SUCCESS in 4m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/911e3c8e5ebf49b0b138e22ac6a910ca : SUCCESS in 6m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a507fe64909e4c0eb42b5d438b3e9a4e : SUCCESS in 6m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ffafbacd9bdd430787d048c53390d4ab : SUCCESS in 6m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60300fcfd559454b8bd537b0456891ba : SUCCESS in 5m 24s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/e53294518894439d8e632804670d09e2 : FAILURE in 1h 43m 28s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/3d99bbe430ed49698b2a95edc8d54264 : SUCCESS in 3h 10m 17s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/1cc77adab62c4918999d56e2bf76779f : SUCCESS in 2h 35m 23s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/c21f1ba16b55485cbbbce59d8fdb03aa : SUCCESS in 2h 30m 45s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/dcb6eaec3914417bafb5915d80c2b980 : FAILURE in 1h 36m 30s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/34c7624587aa4498b9633c576dff9f9b : SUCCESS in 1h 59m 28s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/75a5eac90766457fa6d7cbe9aaaeb1ec : FAILURE in 1h 32m 04s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/6582dac3a58f431b83acc1e8fd566742 : SUCCESS in 1h 52m 48s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/22856b9f852e497f93149981eea6dcff : FAILURE in 2h 23m 46s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/111faaf42af84b26af817df67c668876 : SUCCESS in 1h 18m 23s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/10db4b4bd18e4dec8f805e32384042d2 : SUCCESS in 32m 27s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/56c19fac4b984c74a95bf68b011f6b52 : SUCCESS in 2h 21m 25s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/53c7280d4fbe40b58fba24d46a94a9c5 : SUCCESS in 2h 14m 43s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/200787e3c71e4de48ec34c05be2ff421 : SUCCESS in 1h 10m 30s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/cc30fd3c021f4e9eb842bf01b01feecb : SUCCESS in 2h 38m 29s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/6abc48b59b94443fa2114f443678b7d4 : SUCCESS in 2h 02m 24s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/63f3fc75df68498c9779a84ded1f434a : SUCCESS in 1h 55m 48s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/a5ce0d8959f84aec959fb81cc72e17be : SUCCESS in 1h 50m 40s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/aec9ce1df17e4a71882cd33788b2d4f0 : SUCCESS in 1h 09m 38s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6af8733d853141bc86bd481d05ae4664 : SUCCESS in 5m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5748b19a348a41bc9ab6994275042df3 : SUCCESS in 7m 46s","accounts_in_message":[],"_revision_number":9},{"id":"965b66cdf40c1926e013ce59c0706ff7eb357157","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-08-10 18:02:10.000000000","message":"Patch Set 9:\n\nrecheck 2020-08-10 15:45:36 | Timed out waiting for messages from Execution (ID: e443bbba-f480-4c78-be59-a31c608e3236, State: RUNNING). The WebSocket timed out before the Workflow completed.","accounts_in_message":[],"_revision_number":9},{"id":"ceb2abc0540a10783bef8cf8f5b8156bf0d18b3a","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-10 20:26:23.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"fcc0b35a839fbb7e07539c98ef95c8e987735873","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 23:19:50.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/280f6c287ccf4988bfc27a8e9efeefa2 : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39677b393c01421d8889fbbb4401c5e5 : SUCCESS in 3m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f0b5041847ac4916b48f1eb1f8eaadfd : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/531eaef12ff741beb22ae08033022264 : SUCCESS in 4m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/aa31978ca3bb47f49028ffdf4f3aea48 : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0895863d48b74da7a9bf5f4accaf3860 : SUCCESS in 6m 09s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/fec9f22afc074479b75219ab12c860b4 : SUCCESS in 2h 28m 23s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/d42b592b3a8f4e14bc948b860eedbb35 : SUCCESS in 2h 41m 57s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/37f1d3772f514dd2a9a185f5f0466498 : SUCCESS in 2h 26m 12s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/92f3079a4a7345cc918b97768d0e0899 : SUCCESS in 2h 27m 42s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/5f33e65c1b90484abe01f17ac6d97230 : FAILURE in 2h 13m 40s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/b5ee1e8564e2462686b996106535c904 : SUCCESS in 1h 55m 16s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/3f4e6fa7349043ae9dfc2feb89c1d038 : SUCCESS in 2h 28m 23s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/90730b88660947208d0be8cbac67c60f : SUCCESS in 1h 52m 30s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/5f484e37baf545378d62661dedce6d9b : FAILURE in 2h 15m 10s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/f7cd41e5300d4305806693f3f1561ca7 : SUCCESS in 59m 03s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/45110a91102d458692c2934f46659115 : SUCCESS in 45m 20s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/40956968ca764911a490d0f420b314c6 : SUCCESS in 2h 04m 17s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/0e3515bf84b547039bf9a6e30dd8b06a : SUCCESS in 1h 38m 50s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/4c9f8eddfb5441d3b342d3494c8224c8 : SUCCESS in 1h 37m 59s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/24e16f1243734507b131768b95396510 : FAILURE in 2h 37m 43s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/9655043bd5204f6ca2bbb265a324b1f8 : SUCCESS in 2h 18m 41s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/470e8b988f394305812f05040b4dac9a : SUCCESS in 1h 33m 18s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/0dc783c7fa0145b5b00c09f5db9b0224 : SUCCESS in 1h 18m 03s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/fa10cc1ace9e4b259e3763172c60f720 : SUCCESS in 1h 05m 35s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4b946aea4cf141dda32cd12a035f1dbe : SUCCESS in 4m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fc63c89950134218981ddf668564f16a : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":10},{"id":"1d74cf7c268e475ab1a4d80fd38a82bce2070b96","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-10 23:39:56.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/989e7be2fc5e4dc382387ee4e4add214 : SUCCESS in 3h 12m 22s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"55555805c179e18b19d6c3ff654f1aaa94d5444a","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2020-08-11 08:18:39.000000000","message":"Patch Set 10: Code-Review+2\n\nrecheck \n\ncant see what is the problem there http://storage.bhs.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_24e/741228/10/check/tripleo-ci-centos-8-scenario004-standalone/24e16f1/logs/undercloud/home/zuul/standalone_deploy.log","accounts_in_message":[],"_revision_number":10},{"id":"930d59bd6b595a1ebcc3b2417d24d6531f3311de","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-11 11:44:55.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/5bb49c5ac0b848c29ac549ab5f848b4a : SUCCESS in 3h 25m 21s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"3289bfd5a3fed61772f01945874a3233ad33ff5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 11:51:27.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/18ca049c4bf54ceca0f0c89a9af27c5a : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/559b70a06f6b4a539d6c92e1d4883d5d : SUCCESS in 3m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5c39aecc70a14227be310222d1e4c195 : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b4de2afb03fb4b60a01322596556bbbc : SUCCESS in 4m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/24641c785150405f8d275cbeaed26677 : SUCCESS in 7m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c63739f52615435aa95ca7804515a2a3 : SUCCESS in 3m 46s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/019fa366fb7b432683f02a598ffc99ee : SUCCESS in 2h 27m 01s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/bc3ffa2d52b64a6cba9a495b7cdef4fd : SUCCESS in 3h 00m 25s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/8ed2314355cb45519e2dac8d13f874c6 : SUCCESS in 2h 07m 09s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/1940e2686dd14a72b4141dd1469ce93a : SUCCESS in 2h 40m 17s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/55eb84442eb34058a402b46f618e41d5 : FAILURE in 2h 15m 51s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/026a12bbf3c24b159433e642952338a9 : SUCCESS in 1h 59m 34s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/891aacca7b2347599f440b48bfb3451d : SUCCESS in 2h 18m 48s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/461af428cbcc4c3e8cddb57d688ca637 : SUCCESS in 1h 45m 52s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/83aa5611879c473185ec0fb1e29de7f4 : FAILURE in 2h 07m 09s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/37b1e3f20661405cb9edaa868fff4f05 : SUCCESS in 57m 54s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/30f789525cdb43f0a2164f8cb848a420 : SUCCESS in 41m 58s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/0f827418bdc640da8948e1505442e225 : SUCCESS in 2h 01m 23s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/874469f8f56a4285bfda31389ef16b3d : SUCCESS in 1h 34m 11s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/b043ac5428d84ed9bba568e719bafb18 : SUCCESS in 1h 11m 42s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/ced72fc5da704f1ea05a977098a49364 : SUCCESS in 1h 35m 58s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/522d63bbb73c4dc5b16709096b945112 : SUCCESS in 1h 52m 22s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/720739589e4d43cc81345ede180b8934 : FAILURE in 1h 26m 13s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/59d12ce88323411e97e67675ed37bee8 : FAILURE in 1h 14m 17s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/123b43c5b16245279c2fc1ccb8aeac7b : SUCCESS in 1h 11m 57s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/bbaeded0a16043b99f4e4c39bf6a562d : SUCCESS in 3m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e1c83ed8300e482fbfcb7d5cd14291b7 : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":10},{"id":"b7d5523f94a1329f262c7b2bfd56ea6897568b18","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-11 12:37:46.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"6127f3f1a2a9aa81ea8562a9deccfdffd7f8400e","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-11 15:30:57.000000000","message":"Patch Set 11:\n\nit\u0027ll need a backport of https://review.opendev.org/745575 to land","accounts_in_message":[],"_revision_number":11},{"id":"d7b49380553cf771467fca72ece3379a44074ca7","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2020-08-11 15:43:08.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"dabbd00246a3c11f8f9953a43e0a7bc485bbb622","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-11 17:33:30.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/284b0ddea6b74ea6b1eeef348ec8f59e : SUCCESS in 3h 42m 20s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"726f53c2a6eee4970cd27c279bf74887acb25992","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 18:54:44.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/85123b60372e47dfa4752fb0a1eda5ff : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/635c6eb1ea0246528833a15b66350386 : SUCCESS in 3m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/562152ccee7f4b81b35da4200ae46931 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3e8fd3d07e74628ae548f559fd86991 : SUCCESS in 4m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d77963823bf8406b8ccc035662e64e5a : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f36f6daf2c344b4b5bbd3becbe12d99 : SUCCESS in 5m 05s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/eae5c2a430c64d24a008b9d76c8dbcb9 : SUCCESS in 2h 13m 04s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/d24eb512feb2493cb1c0a92951720a37 : SUCCESS in 2h 22m 00s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/a30a0029459a45218be55aba257b3e19 : FAILURE in 1h 32m 17s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/da0c9a531174423db761eecb4a9e1e4a : SUCCESS in 1h 47m 41s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/3607cbf6556448b5a8c7b8a302b62bdd : FAILURE in 3h 29m 54s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/7dab838466c24f49b3f6b476cd892972 : SUCCESS in 1h 41m 35s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/fe6a47a664484b3ab3244a051149957e : SUCCESS in 2h 09m 33s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/8a49af260f4e440b8fa879074f5b588a : SUCCESS in 1h 39m 03s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/dbd483562b784b1d98bc8e0674e74c6a : FAILURE in 2h 06m 18s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/2b3f8402278349728900f6dff63af286 : SUCCESS in 1h 11m 00s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/fabd84cf8fb244bc9c3fbc18e217d8c8 : SUCCESS in 52m 42s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/578ad643a5b74b078bec46b47db87831 : SUCCESS in 1h 51m 29s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/ad3284ee117a45e0a10f935feeceb3ba : SUCCESS in 1h 39m 13s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/66a9eff393304322b77b008e28092b29 : SUCCESS in 1h 07m 16s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/6a47452f2d904a3fade7a2df6bd0cc10 : SUCCESS in 1h 41m 19s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/dbbb2088075043878771a47019e5e4fa : SUCCESS in 2h 01m 12s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/f8dcd0492b9a456388e010388bb0b0ee : FAILURE in 1h 40m 07s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/c2acb827f6664911a5f649d48c8ff74b : SUCCESS in 1h 47m 29s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/9daa8a9c7e1d4550ae049f6c75c43e74 : SUCCESS in 1h 17m 20s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/29f3cd204122430a9607fe255d42c8aa : SUCCESS in 3m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/270ac2cf8e414cdababde19bef70d43a : SUCCESS in 6m 14s","accounts_in_message":[],"_revision_number":11},{"id":"ffb0b2f115bfca00dc52557925633a5ec61e2fef","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-11 19:20:55.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"f775c7158ee727db6b9950594fb528f9e21e52b8","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-11 22:06:18.000000000","message":"Patch Set 11:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/4bedddc589d3450291d49f94170ee5fa : POST_FAILURE in 2h 44m 43s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"f2fce913099913d8125890becdc6e6052c7035b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 23:06:27.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a8390fd73daf4860b70d44b5566eba2b : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bf0294c5214438e9f5c798d4cbf13f2 : SUCCESS in 3m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8d5100cce16d4fe2a97f943ec2496b1c : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1346840c97574070b77d101c812ca5f8 : SUCCESS in 5m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4be5c3288aed4880a002aa4eb7ad20a8 : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc24b5c3623947d4988507683d1ae2c4 : SUCCESS in 4m 19s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/ccb1f80f0ce443958e3f731e5735777f : SUCCESS in 2h 17m 21s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/91e1dc4f5f094aeaaa22630b2efedff4 : SUCCESS in 3h 25m 27s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/2b0782780b654262b141f6ec1033f730 : SUCCESS in 2h 46m 50s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/08e899d2a37b4b0ab20e466ff8438dca : SUCCESS in 3h 10m 33s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/0a98c1ac46c14c6980ba68ae0fd29bdc : FAILURE in 2h 02m 45s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/eaf88d0c412c433784e9b57cd7c98cc4 : SUCCESS in 1h 51m 54s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/61ead716c9374d6c9fd30be8b0f290c9 : SUCCESS in 2h 56m 58s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/b028d46ad027483db02746bc1c597c1a : SUCCESS in 1h 43m 00s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/7345d29a25da4dda8b29ef19d7c3656e : FAILURE in 2h 47m 19s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/34b7cc2411df47fe87e309ac027e1f9e : FAILURE in 1h 45m 16s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/999da1e6a90f43f78852477fdca2f68d : SUCCESS in 1h 24m 41s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/0e1cf9205cac46da852c32803d678074 : SUCCESS in 2h 02m 45s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/f5e5486ca9de49998d9b0103b8377e4c : SUCCESS in 1h 39m 26s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/6516ee88bc914d0a9742e28c0d47b090 : SUCCESS in 1h 21m 40s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/73ed2990859c4326ac63d9b7e4fd1fbd : SUCCESS in 1h 34m 20s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/a3a8d05959b14d18a2f9e514aa429428 : SUCCESS in 2h 46m 23s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/687fee062b35405c9c1a43e0327651b1 : SUCCESS in 2h 10m 43s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/05483cb47b094701b7be4d024c61ddd8 : SUCCESS in 1h 11m 49s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/c66e92a740654b05bfcdd1362af360d1 : SUCCESS in 1h 54m 15s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/439172ec09bf42b9aff4c21a2c0c13e3 : SUCCESS in 3m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cd94cf6429d34c17b221b5b953dbf41f : SUCCESS in 6m 17s","accounts_in_message":[],"_revision_number":11},{"id":"608d9f74b58232af216a1beb28b2ce398c326952","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-13 01:32:36.000000000","message":"Patch Set 12: Patch Set 11 was rebased","accounts_in_message":[],"_revision_number":12},{"id":"c204ba7e42c2cd0fa382b7f60d2783c4e1156fee","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-13 02:45:15.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/dfcaced18298417babd80da2db65cb01 : FAILURE in 1h 11m 07s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"17f52fc9549fd0782bc6b21d466a16c73a6f2fa0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 04:57:21.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2bf9823d5e474895a9b3e5428a3a93d2 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1162fd9cafd40d68eae4e76f5a7d864 : SUCCESS in 3m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/57f367b591804fcbaa59484824f4e9bf : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a7194f481f704e998a5cf9167e6c98fb : SUCCESS in 3m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1bbf9b9cebbb470a89b4c9b837a52559 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9eb6ba1290bc42169430504540516ea6 : SUCCESS in 6m 23s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/c2766a2a2b2f4b0cb85c5bcbd2bd2608 : SUCCESS in 3h 00m 05s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/491fc5f9f61a47e4b982db31600e4afc : SUCCESS in 3h 01m 48s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/f78977bc87274114ad2f8252cd17ef7c : SUCCESS in 2h 33m 57s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/d52d78339ffb47e696e1bc45db47430f : SUCCESS in 2h 50m 52s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/cb34b7b067924254aa24cae37c27ce9a : FAILURE in 2h 38m 32s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/4bd559dad49a404191eab7673a6cd5a3 : SUCCESS in 2h 19m 31s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/8053f16dbea7414da2a2e22cce7f3953 : SUCCESS in 2h 26m 35s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/e65a0bf2e7f4459fb730034ae85cdbe1 : SUCCESS in 1h 46m 43s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/3dc88553ce7049b4ac8f6d11b56a8e74 : FAILURE in 2h 13m 06s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/8d61a6292944476f9ce5bc6b07e3b300 : TIMED_OUT in 2h 07m 26s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/38fe647c5f204d228c79d113e82e60e6 : SUCCESS in 43m 43s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/63c5fff884784272997a3aca49d9ea96 : SUCCESS in 3h 11m 10s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/d7e56edced7e402d8986d2c368270533 : SUCCESS in 2h 04m 32s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/992a6cf1f6154e2f9d023cd66594ee5f : SUCCESS in 1h 07m 52s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/1b3d67ac7eff48fea9f811905e217d95 : SUCCESS in 1h 31m 27s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/258573a5e1ad4018beceb8741b98324a : SUCCESS in 1h 22m 22s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/fda5609cfadc4005b546ced22c027547 : SUCCESS in 1h 24m 04s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/8b8ac7fea15f4bb0ae92a561d28e804d : SUCCESS in 1h 04m 09s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/bd10aac449be4d07b1ae96b411507282 : SUCCESS in 1h 08m 42s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3ecd7833518e45b5bb1046de1e31c8ef : SUCCESS in 3m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/251ccf4f5f7b4f88b0e0806bfd9de0d5 : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":12},{"id":"b35b04614fcbb93250768ce0ac4a89c64d703bea","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2020-08-17 07:48:54.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"49c52e392e061807580dbd58198cb9bc3b155cbd","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-17 11:20:19.000000000","message":"Patch Set 12:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/1d2e3cf59720423bb3b744375c30a566 : SUCCESS in 3h 30m 25s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"8bc0c10af1450463d9b7e5c167fe7fe4fb2a1ac7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 11:34:39.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3a254da82b5b4d94a168bd4ecef9e9ba : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87627f9907d44d8fb899b9d18880cdc7 : SUCCESS in 3m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5ea447edb4674d5d97562614addb228c : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f452c0993b246048fb30543b0fea4bc : SUCCESS in 5m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e69508f82b6f4c2f917a5cb70cc2c0bd : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d039fc0ead5a4d7b8bb2dbafe598aa8e : SUCCESS in 5m 39s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/069acf5f53d24766bb0d4a1035db0b1c : SUCCESS in 2h 23m 03s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/e0f40fa52579418bbb6c661a0919ab61 : SUCCESS in 2h 44m 14s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/05f74915920a40c69054488b860e292b : SUCCESS in 2h 13m 21s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/a91e5ec78fd2405a9b8f1d531dece14a : SUCCESS in 2h 13m 09s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/15c8988cf1fb4020ae0b251ffb7c2619 : FAILURE in 2h 08m 40s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/d2ee0fa541504ace82aeaa46bcd13220 : SUCCESS in 1h 44m 26s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/ab2b6160530044eeae48b8f0d02565dd : SUCCESS in 2h 29m 01s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/365230ce88d64cbaa6c97e0a22698726 : TIMED_OUT in 3h 26m 23s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/1247b9b486c540299b515c9f2b678ce4 : FAILURE in 1h 56m 33s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/e3bbd6ea6bbd4bcba60bf10c8c048969 : SUCCESS in 1h 08m 59s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/64b0a7c80e2f47599c97f1e5cbdc5bf4 : SUCCESS in 45m 19s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/1df8311102ca45ddb65096ec02ba8d79 : SUCCESS in 1h 56m 44s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/0a7cb0a2408a4121b0513d538ae7019d : SUCCESS in 1h 28m 39s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/db685e9253094538bdbd56c48178918d : SUCCESS in 1h 09m 48s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/eca10cf70c8149c7bd89ad4466a64f7c : SUCCESS in 1h 34m 04s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/f4c99076ccdc487bb553562e9d11ee23 : SUCCESS in 2h 13m 55s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/79936340e5a9420ea8a0b8328c2e0b2d : SUCCESS in 1h 29m 15s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/a898d414f7ae42d69fbf755ae96c1c89 : SUCCESS in 1h 15m 38s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/1801d5ae31f44045ab1ef8083431f091 : SUCCESS in 1h 04m 34s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/444febe3ff0647b9ad83c16fc9e97511 : SUCCESS in 3m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/345dab8283ba403295d960b8f9212fdc : SUCCESS in 5m 59s","accounts_in_message":[],"_revision_number":12},{"id":"073c7229d7ff94f9021373a2310a02d3539325a4","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-17 13:25:57.000000000","message":"Patch Set 13: Patch Set 12 was rebased","accounts_in_message":[],"_revision_number":13},{"id":"64edae254331c576df2d22ea67791784ab8d28ab","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-17 15:10:42.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"78438232e94972245741fd7dfd2a0e6302ea4d8b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-17 17:20:37.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/96d54ad710404a7bb0ab999b99720152 : FAILURE in 2h 08m 41s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"b28adaf3241b152515143977125720139fcb0e13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 21:33:03.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c05294acb6d74e228180c8858a6550fe : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/047cab6a88644a4880f7e24c9620d12f : SUCCESS in 4m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ba6d05216fe444d9832cd78c0cb02f08 : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c5c7155b627f41acb2b94bf407046a23 : SUCCESS in 4m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6f2e5048146a4fec9583014e8808678e : SUCCESS in 6m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ead370d70b41470395df046918581db4 : SUCCESS in 4m 51s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/f2dfaf41e7234dc2875191c467128505 : SUCCESS in 2h 25m 08s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/d385df5747cd45459e8678a8b82dd44a : SUCCESS in 2h 58m 23s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/cc77b53b279d4689be37ba014c1146a0 : SUCCESS in 2h 28m 58s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/11680a33377247adb74c44dc3c5a2e7f : SUCCESS in 2h 46m 00s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/7163f1a5b4cb44d1af237e759731ff73 : FAILURE in 2h 01m 35s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/2ba573fd57d441e5bde14b62f91a325c : FAILURE in 57m 20s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/0dbd2d78101f4e46a6a77dbba36bbb82 : SUCCESS in 2h 09m 36s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/1a11418418ea4d29be2bb3fcf27591c3 : SUCCESS in 1h 15m 11s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/db6ca96ed63c46cb9227435dba350929 : FAILURE in 1h 56m 04s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/60b80d93118d4b4fb36bab71bbac7b79 : FAILURE in 1h 43m 08s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/a6e6f63759324cdba53a36caefef920c : FAILURE in 40m 03s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/7abeecd9e78b4ed29d3c5de1f86a63e0 : FAILURE in 38m 56s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/0d86ca0101bb497eb3596d979f8d4271 : SUCCESS in 1h 30m 29s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/b93f6c387f3f4bc99995af0be75db381 : SUCCESS in 1h 09m 01s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/52a5dcf54783435fb59f55b7e409b681 : SUCCESS in 1h 35m 10s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/60b607bf8e7241f6b8017efec14d5064 : SUCCESS in 2h 05m 35s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/956934da782c4c3eb37a8d02855be550 : SUCCESS in 1h 22m 50s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/fd4f87a7ff5f4e8a80047c3f082c0d52 : SUCCESS in 1h 18m 32s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/6905de0c7daa485598b7ae2647e29152 : SUCCESS in 1h 08m 43s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c3096a13a0554cbfbc7fcfeabfa65733 : SUCCESS in 4m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f0c98d839b1a4f898e56a5caf06dcf8e : SUCCESS in 5m 45s","accounts_in_message":[],"_revision_number":14},{"id":"1341f3c0530b58eee3f60baad877a8d57661d8de","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-18 12:18:14.000000000","message":"Patch Set 15: Published edit on patch set 14.","accounts_in_message":[],"_revision_number":15},{"id":"e2a18c0538827a471453c08c77afa8315796912d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 15:06:14.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a915edb703aa4559a90ab81df230e8be : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e1997e0d4bb4af8b4b2aa0c1d459bd2 : SUCCESS in 5m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1227a771a4814dc3abd9258c5b9369d6 : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aad42f1032f2441c931023e9b949aa06 : SUCCESS in 6m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a40891e85c06439299235910eb0bc278 : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d813260ffa9458ab07f956f23f5a11c : SUCCESS in 5m 03s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/afc16b6e3346452093cf1df1aab013c8 : SUCCESS in 2h 29m 45s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/c48b35f9119d4edfb132810b33517d4f : SUCCESS in 2h 28m 57s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/2b592727546141ca837628d3c65579a3 : SUCCESS in 2h 19m 28s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/3c1fb8c22e2041b9bb7ecb725af67e9b : FAILURE in 31m 56s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/706e24fb17c14f73aad63d0450547da6 : FAILURE in 2h 04m 38s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/f6dee7175fd44494a364bda5ce05e05d : SUCCESS in 1h 43m 34s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/a4a85fe27e2041f0a712884b6a82d127 : FAILURE in 27m 28s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/2db3b333de924e76a6e3195c0c46b923 : SUCCESS in 1h 39m 17s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/943b5bd140a843999bfa40d2b2879029 : FAILURE in 25m 48s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/f2afd320df2d4b029f2676d34dbdeada : SUCCESS in 1h 04m 25s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/b95563e5c1814b9f8829891cc309fc75 : SUCCESS in 44m 16s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/f2e6e15195f04186869a09f803c4a6b8 : SUCCESS in 2h 02m 58s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/5fb64704d1254324be1c9af30c4fe656 : SUCCESS in 1h 46m 54s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/0a4cef104d9744a3a80bcc8f7b701989 : SUCCESS in 1h 11m 09s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/487d93603c72458ab8626cb4bd49d0f0 : SUCCESS in 1h 53m 59s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/5950d56bba1c466ea4a75f0956122d00 : SUCCESS in 2h 14m 52s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/e825aa8f62bb4903b72e25724889bf0e : SUCCESS in 1h 32m 19s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/e119127653544bb290cd84e293f94e58 : SUCCESS in 1h 29m 06s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/7fad5d6d2d28492c8d58270cd8eb3049 : SUCCESS in 1h 15m 04s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/50a638e9b16147e487587bd266dbd0b5 : SUCCESS in 4m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ec9efd2bc9fc49bf9ee85af41322d200 : SUCCESS in 5m 54s","accounts_in_message":[],"_revision_number":15},{"id":"676878c8ade578b1cb3629a49f827e7964ed2cc1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-18 15:45:16.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/e38fb615d9dd44c6b811d14617e6c394 : SUCCESS in 3h 26m 05s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"385c030ac1d38437b39d0a5ef61120af1ca8c64d","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-18 21:17:24.000000000","message":"Patch Set 15: Workflow-1\n\npanko-api failed to start: https://2fe422f590c57d27d902-9c4d070fefbaa4f90a8a69af5ec07c7e.ssl.cf5.rackcdn.com/741228/15/check/tripleo-ci-centos-8-scenario001-standalone/f2e6e15/logs/undercloud/var/log/extra/podman/containers/panko_api/stdout.log\n\nI need to figure out that one before we move on.","accounts_in_message":[],"_revision_number":15},{"id":"fac3e199787b930a7ae9843b9ae27c41961ef60f","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-19 21:04:02.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"40cd2064beb741a35674128a3455f751b24441e1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-20 00:11:52.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/f422bbbbd83a471093f0efdb5bf3cede : SUCCESS in 3h 06m 55s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"2322cf9014a2ebc35fb4499efb93b5a0d8c1aac3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 00:48:45.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/44afd9051bd5403d87a75c38105d2e5a : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c279bd7e0e594df2bd11bcbb9bd9d3f3 : SUCCESS in 3m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/75bc2f9e85cb4ed19f0329498579d8b7 : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5dcfa119449d4188a2b3939dee7df563 : SUCCESS in 8m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c40ffea95c574c3ca39fedf8a8cceb1b : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c60c32a5abd345298065356c6fd64f15 : SUCCESS in 5m 57s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/95d5a4793849468db7fb2020c972d72a : SUCCESS in 2h 15m 08s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/f11fe56625df4bf0adee8c69d6e53fb5 : SUCCESS in 2h 24m 30s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/c3d60dfbe35b4252993e9be338617fe8 : SUCCESS in 2h 02m 09s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/167f746f8c3b47da81ba0f257c27d757 : SUCCESS in 2h 29m 15s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/1ca5149a457d48b0b78bab1647a6061d : TIMED_OUT in 3h 19m 00s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/85dc5639f2fc48bc93235bdb8d46357d : SUCCESS in 1h 49m 06s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/7a813787b53d43c7873d1eb01656b86b : SUCCESS in 2h 44m 40s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/a96922d4860a4ccab4d0fb5672ef9a29 : SUCCESS in 1h 48m 06s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/d1b8cba379974a568bb2485224cb2490 : FAILURE in 2h 25m 09s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/815f0f5181d246fdb873cea284dcb7bf : SUCCESS in 44m 36s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/0983739d16e44760a648439ff9031766 : SUCCESS in 38m 02s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/85e25b7bdf574372b07aaa898377d0e0 : SUCCESS in 2h 00m 48s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/228083bc86c6441fa055092ebd7beb9d : SUCCESS in 1h 38m 59s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/a551599f52384c53ab5e773280188c30 : SUCCESS in 1h 06m 39s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/7eb9cb6bbacb4a3186d9ac6550b1d39e : SUCCESS in 1h 07m 35s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/d78a2488571e403c852af09234438724 : SUCCESS in 1h 57m 16s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/26ed339f8e2d484390d2ad9add18ee8a : SUCCESS in 1h 16m 29s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/d3c23bc610f8487e83593e02388939a5 : SUCCESS in 1h 14m 24s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/52e086bc187b43d8a47ee6bec2bfecec : SUCCESS in 59m 29s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a17ed0a77c394446a5382ab3bea6e357 : SUCCESS in 3m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e709044c002e4aea82a770f6a2f143f8 : SUCCESS in 8m 49s","accounts_in_message":[],"_revision_number":16},{"id":"6d0635406d9066eef1f5bd73c68e0723302911e8","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-08-20 12:49:07.000000000","message":"Patch Set 16: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":16},{"id":"1a0b4be11fe7bfed6f3228566b72df4798b497a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 14:51:12.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"8d0ba1d9eb36ffb3745dbda80e4e0e8e16df57e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 18:22:34.000000000","message":"Patch Set 16: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":16},{"id":"84bf5fe0a351364ab02eab8736993969a8675fe2","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-20 21:01:03.000000000","message":"Patch Set 17: Patch Set 16 was rebased","accounts_in_message":[],"_revision_number":17},{"id":"4f9d0cbc21a2091d38b805a596fad1dac73a861a","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-20 21:01:10.000000000","message":"Patch Set 17: Workflow+1","accounts_in_message":[],"_revision_number":17},{"id":"1936f208bf96bd08c27a3add9978256fb6006384","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-21 00:19:45.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/b3260024e5434089a6683f224c85127f : SUCCESS in 3h 17m 03s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"ed0fd3d8c28b30c0c361e0ee13d508be096f44ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 01:18:21.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/657a45fba6c84cd282ca1e9e6b32720d : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99f063829c014b4a89984f51fa9fc0de : SUCCESS in 5m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9d9af449fec647859bc7781b1ebf0ce0 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8edf8c0d916e436a8f9ecae998944b88 : SUCCESS in 8m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5ce9e60791f94875a12ceb83743a0753 : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29d288e64948460e9d5b9adaf5464c7a : SUCCESS in 4m 55s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/0bea46f19ab643e7a9393c98ed1a6528 : SUCCESS in 2h 22m 21s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/0e761d459bd24ec6a5c02888f2221f70 : SUCCESS in 2h 39m 33s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/2f982aef3caf433c88afd7ce265dc2a4 : SUCCESS in 1h 49m 57s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/dbdcc0dfb01a41f0a0a011c97e4caf6f : SUCCESS in 2h 16m 32s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/aaa0d39496a344228eac4c2c6a7da835 : FAILURE in 2h 02m 20s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/8a0ea08bdcd740a299fe222aa709c1a2 : SUCCESS in 1h 46m 53s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/e9ee8aa72408429c9c18f6e71e4044f6 : SUCCESS in 2h 20m 40s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/41cabcd117a244fd8305823fcfde3539 : SUCCESS in 1h 37m 53s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/552f956ac2204cce9541e976212d0387 : FAILURE in 2h 16m 38s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/aa0dbcd948b24ca0a01d5754015afd48 : SUCCESS in 1h 17m 40s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/fd9b34e7cc2c4c8d81c2bb17a5c7a555 : SUCCESS in 40m 44s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/e52ed87f5db04c1daa95ebeb3673876f : SUCCESS in 1h 28m 04s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/cb2e389ea4c14475badd29d3211f5370 : SUCCESS in 1h 03m 37s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/5a057b32847c4a0397d2880b585ae009 : SUCCESS in 1h 04m 18s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/7205004f3c224a0eb2fa033942186540 : SUCCESS in 1h 36m 36s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/f618353c46094e18802a51fc87f073be : SUCCESS in 2h 04m 27s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/5cc96de21a1043a2a7a709267a841058 : FAILURE in 2h 20m 35s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/33be1a74416645c5b386d68db3bed3a5 : SUCCESS in 1h 16m 57s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/74fe825dd3224b23abe80f9c8e6688bb : SUCCESS in 1h 07m 05s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/df1797427a1d4b8788ed40c111cc2bbc : SUCCESS in 4m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ef048f2b388e4801bcf987e2b119af99 : SUCCESS in 6m 13s","accounts_in_message":[],"_revision_number":17},{"id":"af43dc087c691c936f1a19eb21a649a5518dd72b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 01:20:19.000000000","message":"Patch Set 17: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":17},{"id":"eb3d834b1f130a6e336117308d512d61f0c12606","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 06:27:46.000000000","message":"Patch Set 17: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1d481da0c334f1781cb1da72db5929a : SUCCESS in 3m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3504b3fd39fd429a9c74c72ba86f4368 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eab9d7f679674d09bf687cd7f5b7367d : SUCCESS in 4m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/14229e8ce64f43a9bfad7585b45bdd9d : SUCCESS in 6m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d4b4c948e2d4e97904b07fbb96e6f3e : SUCCESS in 5m 29s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/66a65279eccd4838a4a7e3ff852723d6 : SUCCESS in 2h 25m 54s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/f5b75f0df9ed4e42936af344c9b7f7e4 : SUCCESS in 2h 17m 47s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/768f16ff6e6647f3a7ce99cac9d36541 : SUCCESS in 1h 27m 29s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/4bb14f32d4f9441ba47243f7d71e388d : FAILURE in 1h 43m 37s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/25173801b2ba469f81b872c94f192019 : SUCCESS in 2h 12m 39s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/961d650e4121464cbd656d8e56a04d98 : SUCCESS in 1h 49m 24s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/0a8affdf5fc9436299b65f0bb562c846 : SUCCESS in 56m 32s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/bb21979985514ea3b1e7f17e488c43d0 : SUCCESS in 32m 41s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/73ed4758cd0f446f871db25be09ea967 : SUCCESS in 1h 42m 45s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/c904ce5c57454c7a8b2df5571e811f3f : SUCCESS in 1h 17m 02s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/97a480afaaf043bb8b4358606af3ead5 : SUCCESS in 1h 01m 58s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/6568a97970af405082df9beda03efb8d : SUCCESS in 1h 37m 12s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/dc8581450e54431daf45542faea3765e : SUCCESS in 1h 12m 39s\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/e49f41d82f49453a849b5ecf02d92233 : SUCCESS in 1h 25m 06s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/77f70ba4413143a89a6ced2df1ab2190 : SUCCESS in 3m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3ac4ee5c4a554d61baabd74bab5b4afb : SUCCESS in 5m 22s","accounts_in_message":[],"_revision_number":17},{"id":"3938bd6ef663d6022839dc445af56ead8497c2f4","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-08-21 06:42:58.000000000","message":"Patch Set 17: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":17},{"id":"8bf8abbb1c39263a02cc36e436f7ca8c09c00140","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-21 09:48:21.000000000","message":"Patch Set 17:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/89c5a6481c114ef88595f2009fe00d59 : SUCCESS in 3h 04m 17s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"16688a0f9ef30f0cefb806a7f7d96b7cee86c8d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 12:06:15.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d23bc8e4c4dc4b79aae1d1813707e9ec : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/892913ea46d84a31809dfa8718002d55 : SUCCESS in 5m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e2b711fd31924fc28b269e9d368238a6 : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/34b8fc6eea2d430481b89774050bd479 : SUCCESS in 5m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8c21635a14784395bad21002255e3b05 : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72fd512faac543e6a09b74128438d347 : SUCCESS in 4m 52s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/575c6bd048184482877af7ed60308968 : SUCCESS in 2h 40m 21s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/ca06a4fa23ae4cc0bf937de05dc41dd6 : FAILURE in 1h 16m 05s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/58d5dfebeb6d49edadb52a52393020e6 : FAILURE in 1h 19m 50s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/a09dc0d966ff4c0c8c33c03cc04ea9e7 : SUCCESS in 2h 37m 18s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/ca7f01edca6c4c5ba1b5fbc88845d669 : FAILURE in 2h 22m 30s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/ac5495d2c89c4205a64d5e1b6137183e : FAILURE in 1h 37m 16s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/8e0e0b19013945a1ab0fb7f1786cf53c : SUCCESS in 2h 13m 07s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/c40bc241000a472e9ac00b0fe6d5ea5f : SUCCESS in 1h 18m 33s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/0d263859202f4d67a9c7705ba1e7e3e7 : FAILURE in 1h 20m 16s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/327fa477fba944d2bc893a280f771dd4 : SUCCESS in 1h 10m 02s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/6606c7d18c874605a8d1def891e92539 : SUCCESS in 1h 28m 09s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/06ae6938dd4343cf8b365e610bf3acdf : SUCCESS in 1h 40m 45s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/1213b78e900647149fc04cd6b585eaa8 : SUCCESS in 1h 49m 18s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/8edf5f2d3f4d43f596e36b204ac725a8 : SUCCESS in 1h 27m 00s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/a4fee642d11f4286b8ee74d2369ddf0d : FAILURE in 1h 48m 24s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/e65d0b67cf4e4cac94142f00791f0f32 : FAILURE in 1h 25m 21s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/6845769c231547a5b2c6c26f9c48f405 : FAILURE in 1h 24m 52s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/b69ef977552c4628a34b5fb9960dcc58 : FAILURE in 1h 15m 24s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/ad2d4be410f1418bb624a0594cae7148 : FAILURE in 1h 23m 22s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/99967773f8b742b9b72a1327d46ea6a1 : SUCCESS in 4m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9e5d5b5a06454b418a6e6be9316af126 : SUCCESS in 7m 52s","accounts_in_message":[],"_revision_number":17},{"id":"062e96f2dbbfaf10399142c79284fcf94a8d25a0","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-21 12:11:29.000000000","message":"Patch Set 17:\n\nrecheck","accounts_in_message":[],"_revision_number":17},{"id":"0a95bb2840d1cbb72853f79f96e2c93e93f17756","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 15:16:59.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d4d29976b0734ff18fe3c569745fa674 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2fd9fcbdd164c2c98a7fc559c98d518 : SUCCESS in 4m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/35ca3e788861427ea0b7f911fea28bb5 : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a36aed0a647440049f4751afc13a7d11 : SUCCESS in 4m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2162a4f860c242ab9bd85845493230aa : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d914f00ad49489da182e71b5ab1ea81 : SUCCESS in 4m 31s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/7e5d269018db4734bc0ab422649b9f63 : SUCCESS in 2h 29m 04s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/69e4a3f5de554f1ba27a7eb579b388d4 : SUCCESS in 2h 20m 08s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/4b2f2592261844e49f3a4e399edf79aa : SUCCESS in 2h 26m 07s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/18c712a3b25a410e98153449e187c30e : SUCCESS in 2h 42m 16s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/6bf0168981c64d2280fa6098e13ffc01 : FAILURE in 2h 07m 37s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/44cc9df9f0754f92bc2ce1a71dfc1e15 : SUCCESS in 1h 43m 33s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/565bbee55a1a458989194bcc8987914c : SUCCESS in 2h 35m 07s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/ce70c613ba664afeb8c941ce312c9e3f : SUCCESS in 2h 11m 17s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/ef3772639a1244e6be81694e5c4415fb : FAILURE in 2h 16m 31s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/7f406ce09e154a90beed8ff8950523da : SUCCESS in 1h 09m 12s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/c21d88f038a84987995f69a7e9cd2f9e : SUCCESS in 34m 14s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/9921cca217554a99822a6df98ca80019 : SUCCESS in 2h 16m 41s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/f3d541f50b0548458b3a13f8fe2264f3 : SUCCESS in 1h 29m 21s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/cc9eb3dbcb8848c9b06dbae439b78e1c : SUCCESS in 47m 03s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/dfde49a1229e405da1abffcc2eb1bb85 : SUCCESS in 1h 40m 20s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/64f02b92f2ef45678826a0bff490b326 : SUCCESS in 2h 07m 15s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/e074e1a7ffe04a699da2a6c7469edf63 : SUCCESS in 1h 28m 32s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/b7530f0390074b07bffa420c8d90daca : SUCCESS in 1h 16m 08s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/17f8baa397c943df89934edf2588ff06 : SUCCESS in 1h 08m 11s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/280b0a5c16c147a9860726b8933f0bc7 : SUCCESS in 3m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cf6bd1ff75b84816be4bac9cbd2a51b7 : SUCCESS in 5m 48s","accounts_in_message":[],"_revision_number":17},{"id":"649910ca93bade8c8c1e19d5d7fd3d50331508dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 15:17:35.000000000","message":"Patch Set 17: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":17},{"id":"c44a4ed99ead529c78995913f9cd3c8680dafed7","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-21 15:33:36.000000000","message":"Patch Set 17:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/f59e91cee69c4b90ac131c5aa4c8490b : SUCCESS in 3h 16m 36s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"0f79e1a02d271d225643d8f6f40a0cb2748385be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-22 01:24:13.000000000","message":"Patch Set 17: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e8a03526230454a80637f91d4c77ba5 : SUCCESS in 3m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/75a4e2929612401c88e896120f4d3ec7 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/648ea0cb803042e29b5c1d4b1f870df1 : SUCCESS in 4m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b6fff5bcf0fd4079b4695f5cefd7e85b : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e8ab25bd5a14604b3c3c6c5cafada4c : SUCCESS in 4m 20s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/398ec56af27048a3a358ba1dcd90d9f8 : SUCCESS in 2h 27m 40s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/5acb26abd3f94c95b20e2f1823be8f0e : SUCCESS in 2h 21m 02s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/e14e846d009e438b8c46f5f219178088 : SUCCESS in 1h 41m 01s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/4956799081fc4ef8947ee5aacca52f8c : SUCCESS in 2h 09m 16s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/fafc8938af754d89b28e2d231b6fb159 : SUCCESS in 2h 16m 57s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/1c4e549d1c8a4c1abe918b748bb84785 : SUCCESS in 1h 46m 36s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/ebeea96b8eb04f25b02a8213ed4fd06b : SUCCESS in 1h 01m 36s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/3effa3377aac4851804a51d7fa1599ab : SUCCESS in 39m 49s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/40cd1777a67343268bba8b2a860c3061 : SUCCESS in 1h 45m 35s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/17c583a619ea4f398ae8b8ea50417430 : SUCCESS in 1h 33m 56s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/b2d9673d99b049b6affd1e0dfd8fd1a7 : SUCCESS in 1h 09m 14s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/fa46969077154e35802a47a3367018d9 : SUCCESS in 1h 25m 50s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/23a9667a924c411392d1ca36623913f9 : SUCCESS in 1h 10m 25s\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/ff6b7e00258d4707836a8b55c7648ba3 : FAILURE in 1h 15m 09s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9f0318a7183c4e1085c76de378e6ff31 : SUCCESS in 4m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/af42b09a3e3f4b389441581aea3b415e : SUCCESS in 6m 36s","accounts_in_message":[],"_revision_number":17},{"id":"e7399a70f85cf820a3ac22d5f113e514f181c8d5","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-22 02:42:20.000000000","message":"Patch Set 17:\n\nrecheck","accounts_in_message":[],"_revision_number":17},{"id":"da02e42eda88fc404fdc7b56c81f5af72deee952","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-22 05:38:47.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b26a0a63866c4839b44c3087d198057a : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28776b86c6ee477cb8e933c6d38676ea : SUCCESS in 3m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/27854a390f224048886ef4dbca4c9a74 : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5bebb7da341d49eb9619ea22ac784d2b : SUCCESS in 4m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ca826bc48f99483f8a76992a0ef2fd29 : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bb36731c6c747c69ddef592e0df7e57 : SUCCESS in 4m 21s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/7f0e319f1b39457c8a648481fc616684 : SUCCESS in 2h 18m 45s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/0aa2e5f74a6a4b7e964efd90568db1dd : SUCCESS in 2h 44m 47s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/f12992cc38a840678185b1cb2211b934 : SUCCESS in 1h 53m 02s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/98c61c6e753e4b1684dc02400f6ab733 : SUCCESS in 2h 11m 53s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/60741595629a4e46aa713c2230feaf07 : FAILURE in 2h 01m 14s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/8ddf524fc3cd440a92eb9932764b0368 : SUCCESS in 1h 44m 08s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/9dd9c128424647cca4b0f0612bc691f5 : SUCCESS in 2h 14m 32s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/80476a4f46924da2bb469c0ed0b956e8 : SUCCESS in 1h 36m 08s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/0b70f957d6084ae596d5925416665a86 : FAILURE in 2h 01m 28s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/29fde86dc9464c4cb57292ab6f851957 : SUCCESS in 54m 48s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/e8cd4daae87d4c09b4f4e3e0dd1a4333 : SUCCESS in 40m 42s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/f8d4a5282b9c4114b66110ef6c737aca : SUCCESS in 2h 31m 17s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/e642edae1eff4b1c8691d27652956ff2 : SUCCESS in 1h 29m 44s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/a8dbb6baa2cc4370a90dbf331647efb5 : SUCCESS in 1h 02m 11s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/d80c8e9a3a694b2a84ba2ee907c6a941 : SUCCESS in 1h 58m 30s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/577e95f2b7e544a8b6ef7aae4d31f35b : SUCCESS in 1h 57m 05s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/245f5feabc6b4dcfba485e80b960fb7d : SUCCESS in 1h 19m 37s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/f6861b7fd88a4e0cb6927e29ae8049ce : FAILURE in 1h 15m 48s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/3c8cfe991cdf4aec89b09f64cc4520be : SUCCESS in 1h 02m 28s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0021ccbd292945028fa0f2cd87b6a849 : SUCCESS in 3m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0eccd4ecfd1f48219715d3453f67f9c6 : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":17},{"id":"cdf754d978c2d2535061fff77a611f9f127dbf0f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-22 05:49:09.000000000","message":"Patch Set 17:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/53443b0f52e04daabd229bc4a2d35945 : SUCCESS in 3h 05m 48s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"6e918d8c9c91f2df00ef96a7cc0cec8fc635bc5e","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-22 23:36:12.000000000","message":"Patch Set 17:\n\nrecheck","accounts_in_message":[],"_revision_number":17},{"id":"77b931ba57d2be8fdf921973ba52b0b2fb63518a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-23 02:25:27.000000000","message":"Patch Set 17:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fd922f1454f944e59e609cb72ee398de : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ede3fc195d404d54a5fadbe456fee510 : SUCCESS in 3m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6abb44a83fef48c2953e077b7d3b9906 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/da0ef78d8cab4901a3df48b4417a8d95 : SUCCESS in 4m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e12bb0e8f86445479c2e1be92a3b128e : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4895605621d6430799205296fd611a17 : SUCCESS in 3m 56s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/b43510bb5ae24f29908b47ff748655e1 : SUCCESS in 2h 14m 41s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/105e41d5ab304585aec6804881ad09e0 : SUCCESS in 2h 37m 16s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/8c0b2675756d47adbb6f2f757b6ecfa2 : SUCCESS in 2h 13m 50s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/1791cce24e494bf79021933ac580ddc7 : SUCCESS in 2h 19m 20s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/d3fe745b7a824988aa876d0d21e5fb98 : FAILURE in 2h 04m 54s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/9de8018910e74a768b6b67d8ad94b900 : SUCCESS in 1h 45m 15s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/4fbdd0d7338f4a06b6b7512af5bea568 : SUCCESS in 2h 12m 01s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/43c2fa1d5c574bc29158250fcce73096 : SUCCESS in 1h 46m 00s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/90f7b6ae62db4abfbc5ef85d34f13b1e : FAILURE in 1h 55m 32s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/c54a3f388d174b1392d44e10cb142813 : FAILURE in 1h 57m 05s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/9eeb38a8df824a7593f205f58d144a92 : SUCCESS in 1h 35m 38s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/b3267411656f4b59be74768f886f7318 : SUCCESS in 2h 01m 22s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/213b1f9f2b68444daf2a850bec652fab : SUCCESS in 1h 28m 14s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/bd989c067df445a3ac060839f7ee908e : SUCCESS in 1h 07m 45s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/4cc908ba82e546aea8c0b90d39a2358d : SUCCESS in 1h 31m 51s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/9092a10f1fe74165883f2462254ada13 : SUCCESS in 2h 25m 03s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/557f42f4e26f4a9abf025f4f317878da : SUCCESS in 1h 23m 43s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/7ab8b4254e7b4024becb5afdf58a77c1 : FAILURE in 1h 21m 29s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/dbd5413008df4b26866a880fd297ce4f : SUCCESS in 1h 00m 10s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/10103371c98146fa9acb2e8d82a98885 : SUCCESS in 3m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a5ac5c1f1ea94c4f90c727d963f9e396 : SUCCESS in 5m 28s","accounts_in_message":[],"_revision_number":17},{"id":"5092522bef71e3cb0d69381a69da8ace67a7d772","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-23 02:46:12.000000000","message":"Patch Set 17:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/38fc5798cc0842718d38476c6c7a2d92 : SUCCESS in 3h 08m 48s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"1c702599e5ff8ee247adf4ee5815bcfd54734a8f","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-08-23 10:22:58.000000000","message":"Patch Set 17:\n\nrecheck","accounts_in_message":[],"_revision_number":17},{"id":"358858b9f4b280f508584187e7b15f70c859c535","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-23 12:55:13.000000000","message":"Patch Set 17:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/63e5e3a2c633436fa4a76ab771c8c365 : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b561355643f44d8b14e34562abee8fa : SUCCESS in 3m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8455065fcd6a4b37a7b192b3106db999 : SUCCESS in 3m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ea4cf81641f54eeaa91d3c068ab85fb7 : SUCCESS in 3m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4eca05d75e434978a43a2332eeb32dd9 : SUCCESS in 3m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74dc68409d6c4982ba3e94bbc95b4573 : SUCCESS in 3m 36s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/5658c68f894841149f7060f3b971a488 : SUCCESS in 2h 17m 45s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/7ed2fbc217e8436ab9148b93b0a05c32 : SUCCESS in 1h 43m 35s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/d57f961a11424c1e9d4ea88a51edcf48 : SUCCESS in 2h 05m 16s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/90a0f45d301a490f90307b5200218cdf : SUCCESS in 2h 16m 26s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/96b86f597daa472d9206132adfc92957 : FAILURE in 1h 56m 43s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/a9640f926b73491a86501549a6859526 : SUCCESS in 1h 44m 22s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/b98594ddd87148f0abd0eb181b86bde4 : SUCCESS in 2h 19m 50s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/c683f528cbc2414dacfbb32f9c3f7542 : SUCCESS in 1h 36m 42s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/1fd77dd3a99d41aeb6533f109b11ac84 : FAILURE in 2h 09m 09s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/35498d31e7404957aa6318b6df8a29a8 : SUCCESS in 1h 20m 49s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/421a8e7cf3c64e21816f2dae721a7247 : SUCCESS in 38m 22s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/6a9f0dd042004d2f89dada612e9504ea : SUCCESS in 1h 48m 41s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/fe9a85970df9466d9cd7c9b2a380eca8 : SUCCESS in 1h 21m 43s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/593cffbb580943659445f7693483180f : SUCCESS in 1h 01m 01s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/6b44f6167f084498aee57a7948657839 : SUCCESS in 1h 27m 31s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/a93041b5c46a4e96ac9c6cb8e8df39fc : SUCCESS in 2h 04m 12s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/1f05dbb5445e472c9f4e3c1cffab631e : SUCCESS in 1h 19m 36s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/61ccba8511cd40979062bc3230dadcd7 : FAILURE in 1h 13m 43s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/154d10687bc44fabbbfaf63e65b35f69 : SUCCESS in 1h 05m 17s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7eafa313487b4df9bc5d4bcb13cfcd29 : SUCCESS in 2m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/45c139a07c834252911d65cc555891ab : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":17},{"id":"686c35ad7a9fb61c01769af151de4eb7abd5c7a8","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-23 13:16:37.000000000","message":"Patch Set 17:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/687dd3955226465e983275b3e4ca063a : SUCCESS in 2h 52m 19s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"8359b900b09f523f27b9e7a16df0a2df8cd731bc","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-23 15:42:23.000000000","message":"Patch Set 17:\n\nrecheck","accounts_in_message":[],"_revision_number":17},{"id":"d21fe67fb5588ceee5b20e8bcd415ccb1c8f94c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-23 18:30:27.000000000","message":"Patch Set 17:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ded947fcf7254469971c96c48e144e27 : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bc45b7ee67c49d791933669fe60dbba : SUCCESS in 2m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c32466c3a6434e6486c62d07249a8d97 : SUCCESS in 3m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/50dcf474409c473dba4e60f312bc4a08 : SUCCESS in 3m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0f5314fe07ff48d79158de12c672e3ff : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fceb1d4c8794b8284dbba9ccac20910 : SUCCESS in 3m 21s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/17d8157b89594f54a04f618daa3f403f : SUCCESS in 2h 11m 56s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/29f71571cac743948a77296c528f6a8f : SUCCESS in 2h 34m 27s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/7e65edecf81c42b795b8132c5d0b9019 : SUCCESS in 2h 00m 11s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/fbac5931f70f42f8be541dcadbfc2f82 : SUCCESS in 2h 23m 30s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/6ef9a0d8004543fb984ed6bba85978d3 : FAILURE in 1h 59m 11s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/50edfd1883f54704a18e00e4c730227c : SUCCESS in 1h 50m 48s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/861ae7d7bdac4f6b8d65e6ba36a46a33 : SUCCESS in 2h 08m 06s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/88a088b7ee4f4a16a92c6f6112570e2e : SUCCESS in 1h 38m 06s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/e5880f4e72074667af9052212b5deb64 : FAILURE in 1h 58m 40s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/f58c8ad1cf8846bd8c92dc4a20df90b6 : SUCCESS in 1h 10m 48s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/aa1ec81f260d45f0a0b09adb1c00937c : SUCCESS in 38m 23s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/f3b9039b7b6f465dba5fb98bfba57a91 : SUCCESS in 1h 49m 10s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/4b22429e90e548768142a5f3078d3f2a : SUCCESS in 1h 31m 06s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/a550c99b8d094b559350771c9d0de486 : SUCCESS in 1h 15m 08s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/a596c00b7c454aeaa84e369e42bf7650 : SUCCESS in 1h 34m 28s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/b46836d60f1b40dbae82f4ee82dcb040 : SUCCESS in 2h 02m 27s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/0d08fe18c5c74c6fb3c31e5383ef654d : SUCCESS in 1h 21m 45s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/7e2c77fec30e42b7bc34948ae91f2649 : FAILURE in 1h 17m 49s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/a7951465479740da856758a8dba2ef34 : SUCCESS in 1h 10m 38s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9b7089f6897045cf99541941272d72ab : SUCCESS in 2m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a6bae8edd1d8414e8eec327e86a0d5cd : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":17},{"id":"98b2938725790a4a4bddddbb2a179dabfdbf301a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-23 18:59:16.000000000","message":"Patch Set 17:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/cbf965b6154645aeb87090a4ebe79ae9 : SUCCESS in 3h 15m 41s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"569e704d661da8ee353f7e2cabc3cbf068ff3033","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-23 20:01:03.000000000","message":"Patch Set 17:\n\nrecheck","accounts_in_message":[],"_revision_number":17},{"id":"d57c026d50f86891bf5eda537f37f21514f2af5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-23 22:33:24.000000000","message":"Patch Set 17:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c0c09eed1a44bf198f265c9fcb4c940 : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7225faa7ebeb400d8e246c8fc8999df5 : SUCCESS in 3m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0329a9be7730411198d113985b332f2f : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/64f98954c96e41a9b4af7cc9e800beac : SUCCESS in 3m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cf0ef0047556421fa281762d3ded8304 : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4de634af5bd143cda3cf29740aadaa76 : SUCCESS in 3m 58s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/a1f236a228d34b4ab18ee83925ab1aec : SUCCESS in 2h 23m 40s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/81c86af72544409a940ffe9cb5a084c2 : SUCCESS in 2h 09m 43s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/96ac15100d0f48278acacb32b21e52fe : SUCCESS in 1h 56m 56s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/8953703bf9db4d4c9fdf58ccef62da3c : SUCCESS in 2h 13m 02s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/1561282b729d4c37aac5831becf09a1c : FAILURE in 2h 19m 54s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/65d1d617f2a6436ab8d9017bf4fbf63f : SUCCESS in 1h 28m 28s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/28e0c2226a0f42ceb16edeac2ed8c30c : SUCCESS in 2h 09m 22s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/992dce764abd41169e9167e5a538883f : SUCCESS in 1h 54m 57s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/baa5338e9d8444bb9ec5d73ad65b1db1 : FAILURE in 2h 19m 30s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/2f3981345e8f4a04901ef58f2e3f0757 : SUCCESS in 1h 00m 17s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/95b4434a0de64bf0932229993ddd1542 : SUCCESS in 58m 17s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/8c3f642a57b3423e9608c9ceb420df5f : SUCCESS in 1h 43m 19s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/a85e3decf7a040c0b76018dc8f7fb8a2 : SUCCESS in 1h 39m 46s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/1e13026962914115aa716e96b265ddf9 : SUCCESS in 1h 03m 53s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/e2ae3ea92465459a864126fa980e80b0 : SUCCESS in 1h 32m 59s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/d0a31d0a603d4eef986fe98b595656fe : SUCCESS in 1h 46m 49s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/3dc8ea0d0e844fd6b8e155f6b8f0d960 : SUCCESS in 1h 14m 28s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/2b05cdf37ebd41988e28edd7841f2784 : FAILURE in 1h 07m 43s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/1d6728692c6149bb81d7abfbd3affd72 : SUCCESS in 1h 01m 16s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7bcf83ef315d4bf4be7a901be1c89971 : SUCCESS in 3m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5625e40dca5d4c0b98e7bd840b4f94c8 : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":17},{"id":"0c8c76fb4688e21b6713848495cadc599cdb06f8","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-23 23:20:26.000000000","message":"Patch Set 17:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/81b69488e74e47db91de063c8ce3bf5f : SUCCESS in 3h 11m 47s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"4970ad99e6636b8a1d977d5a023f6f7433907d9e","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-08-24 11:22:29.000000000","message":"Patch Set 17: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":17},{"id":"14545e787c63c295eb6cec3ef7637be18df6362d","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-08-24 11:27:28.000000000","message":"Patch Set 17: Code-Review+2","accounts_in_message":[],"_revision_number":17},{"id":"0aae637187e54ce51155121e1a4f0bbcf779dd19","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2020-08-24 11:30:52.000000000","message":"Patch Set 17: Code-Review+1","accounts_in_message":[],"_revision_number":17},{"id":"9222f056539ab13c99714c19bee87ccf89b8d088","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-24 14:44:10.000000000","message":"Patch Set 17:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/e6e18ca7cb4b48f29887373919c001ec : SUCCESS in 3h 20m 49s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"8a77c322c29f9cb32df28eb5f9e8fec687e283cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 16:59:08.000000000","message":"Patch Set 17:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7c8793a374184aa19b366ed4f20be66b : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c90e0252f83e4ac9abae369bc8c150b3 : SUCCESS in 5m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c12d239fb0874b1e865187b85ac4a158 : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f8e95e027bb84e0dbf1238388cbc4499 : SUCCESS in 5m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/65bd4951a68a471b834d121e6a97a440 : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02794c410974415a8d1de1bb9d8fa855 : SUCCESS in 6m 36s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/a8d7bfefa71945f9b11d5ebfc66fd936 : SUCCESS in 2h 24m 51s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/2c84975ae80547879c8f2f5f623c0fca : SUCCESS in 2h 33m 16s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/00c4b401628f4cdc9c139cc1666ec37a : SUCCESS in 1h 55m 20s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/2bdc2bd95d8c4a65a235ba5e2f9ffd9a : SUCCESS in 2h 29m 02s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/eac004251d804bb7ba1ea41d6e246700 : FAILURE in 3h 07m 06s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/a4b07c057b934c8fa98b71d0048dcb82 : SUCCESS in 2h 48m 34s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/905874fcdcad4cadb25a93cc2e624358 : SUCCESS in 2h 08m 47s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/f7dc58f0eee94c72a73becb3acb218c2 : SUCCESS in 2h 37m 58s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/c8510769e8b844d2b2345b7f8af65080 : FAILURE in 1h 40m 42s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/3cadc73c3871488f96e33b129cc47a24 : SUCCESS in 45m 34s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/f10f9389bc054ac5a56c5919f183dfe5 : SUCCESS in 51m 21s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/6b8599d4986446518c82a5090b9f2780 : FAILURE in 2h 29m 21s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/3a1aa603d7a34558bae3eb0dba5f7a1c : FAILURE in 2h 22m 28s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/0d7ef04a1b8c44ba820e5207001bff8e : SUCCESS in 1h 52m 55s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/34475f9271e04c72a90f28b46d5424d7 : FAILURE in 2h 29m 37s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/b5556249620142eeb94d677dcffe3b15 : SUCCESS in 2h 13m 21s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/e5ef83fa173d461cb694cf9299c5875e : SUCCESS in 1h 29m 06s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/b3d8e90f6dbe4248a40019072cead9a7 : SUCCESS in 1h 23m 11s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/bde407cc5fe042a299d66e9ab59274ac : SUCCESS in 1h 07m 29s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/01589d4d9afc4c86a2370d50408570b5 : SUCCESS in 3m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/37dbf2fd20354be381135612eacfa47b : SUCCESS in 5m 53s","accounts_in_message":[],"_revision_number":17},{"id":"2cde8522b7312d7c64b670ec4ed9cb9666718570","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-08-24 20:12:40.000000000","message":"Patch Set 17:\n\nrecheck","accounts_in_message":[],"_revision_number":17},{"id":"0bb026ba192e23c6c98a1bcd446caa76845cf598","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 23:32:08.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f26af6f19af457dbbf341dfd13ffe5f : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e378a39b99ef4aa5bee382e691d77625 : SUCCESS in 3m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6db15b89d889438ca2b3de4691b2bc28 : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/461c2c762dfa4a0189a91ef7bda2c196 : SUCCESS in 4m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dfbaf4efa6994e159146998a4b7081d6 : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1164dc508af940bba75f1eec4846f9b9 : SUCCESS in 5m 13s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/8cbcfd176acc4ebda4881e9e03c24002 : SUCCESS in 2h 13m 59s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/f4803791468b4b48a82d3549098a9c28 : SUCCESS in 2h 30m 02s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/27f995ff53114b8d9a3c1d4e3c951ac8 : SUCCESS in 2h 26m 12s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/2c5721909cca4f44b8f818383f157b7c : SUCCESS in 2h 29m 43s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/5460156b836143558b96a497a2b5c3b8 : FAILURE in 2h 09m 53s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/c06594acf97e4589a910defc1c773718 : SUCCESS in 1h 54m 32s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/18bb8ae193424b4fb914c1d840af7f2d : SUCCESS in 2h 15m 39s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/b5bb64117d7a41d4b5d6b207b92f8238 : SUCCESS in 1h 52m 41s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/ff84155d69bd4f1dac67b408e6118886 : FAILURE in 2h 35m 10s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/f8f7327ea7fd458abfa03cb89fb1cafa : SUCCESS in 1h 20m 35s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/3af9d3ad581a4825a19800be6d87d445 : SUCCESS in 26m 39s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/9897a8779d274a2e8929574cfa19de39 : SUCCESS in 1h 30m 48s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/12d04a0fbdbb4152a18b6cef8e0283f2 : SUCCESS in 1h 53m 49s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/cd3b52a4ee3d4cb49bb1aa5157fa6ba2 : SUCCESS in 1h 32m 20s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/975b957080554f339ecf8cdf4d6c8b59 : SUCCESS in 1h 55m 26s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/b7086309a8c04f4c8a0b124096326927 : SUCCESS in 1h 24m 04s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/86d9eca002b24bdc89faa12f1c0efdb7 : SUCCESS in 1h 15m 45s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/83a2d076c86d4816b57ddf4af6fae359 : SUCCESS in 1h 28m 50s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/f68f7ac976d14812ac47ff2269dacedb : SUCCESS in 1h 32m 08s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/10208784e57f4f8bb22635b03190ac69 : SUCCESS in 4m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cb8f778951b54b3eb215f64e648c8b6c : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":17},{"id":"cc82c3166dcd2203e42b8a98d6070ca021de7cf5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-24 23:34:28.000000000","message":"Patch Set 17:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/5adfb3c920dd491fa406500b7adf0e89 : SUCCESS in 3h 20m 59s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"3e5f70b3adf7a3a0b91846495d3a71637eb4dbc2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 03:43:07.000000000","message":"Patch Set 17: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":17},{"id":"d85716842ea51b0b5872cfa08216f9678a9beb1a","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-25 14:29:53.000000000","message":"Abandoned\n\nThe gate is currently hitting the \"docker api 429\" issue, see #tripleo channel for more details. I\u0027ll abandon that patch so it\u0027s cleared from the gate. Please do not restore it as I\u0027ll take care of it when the gate is stable again. Thanks for your understanding and patience!","accounts_in_message":[],"_revision_number":17},{"id":"81ce303155f983ed1e4879793986055ba4eaffcd","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-26 00:08:19.000000000","message":"Restored\n\nrecheck\nthanks for your patience, you\u0027re awesome. Now let\u0027s hope for this patch to land.","accounts_in_message":[],"_revision_number":17},{"id":"9807af1930a19238931410e2953f476173c662f8","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-26 03:25:26.000000000","message":"Patch Set 17:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/281f15e9f9e84933a7b869a1a4264d8f : SUCCESS in 3h 15m 38s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"e68f86380866c2f0e05aab27c65dd5849ae07bbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 03:46:41.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8659db14410c474cb2e8fbb95face020 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/495ed7fd434c46188ae84abaa2a5bc5f : SUCCESS in 3m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/68b163200ba248e5bf01dcad5cfb5b4b : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e4bfe1fa3694435391047626f39170c7 : SUCCESS in 4m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3a38435c0e004f68980645d5bf370ebc : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef5aa984a2784b1e8c8be46635c02dc2 : SUCCESS in 4m 58s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/5f29f704314c4017921e6a6268ef5f44 : SUCCESS in 2h 08m 53s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/1489b51045f94a02b38165746d5fbd97 : SUCCESS in 2h 23m 55s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/cc5f4658dcc04326adacde7641156723 : SUCCESS in 1h 52m 51s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/d0efb5de19394dc9a0a2505b2278db01 : SUCCESS in 3h 13m 48s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/ec4f95c95836403f800ac55012631dab : FAILURE in 2h 07m 59s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/8cf23f98ef644b788b4f9b9d60b16fd9 : SUCCESS in 1h 10m 39s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/9c826a42db11417b952a3e0ed5728ab4 : SUCCESS in 2h 22m 42s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/4f65638e5f5c47a380b124af5e653eb4 : SUCCESS in 1h 39m 40s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/afb09884f5704dedbd6946ceeb772f43 : FAILURE in 2h 38m 42s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/fb68f4fca391412da87fd21b4dcf966b : SUCCESS in 1h 17m 55s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/e1ccf935f9cf49598a2fdae79636e2e1 : SUCCESS in 38m 04s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/c8a9cc8a4bc74c818ef4551913a992cd : SUCCESS in 1h 56m 41s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/87404f5af45d4ad091bc56e18a6100aa : SUCCESS in 1h 38m 01s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/1956cf650ed14dda88d22e7b44d27724 : SUCCESS in 1h 01m 28s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/98b19767b28c4c259f7083f07b112c75 : SUCCESS in 1h 29m 37s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/5c76ccdd8869411a8c997e393162bfd2 : SUCCESS in 2h 01m 50s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/b2517c68a01f422d849806d2fc0a8959 : SUCCESS in 1h 32m 19s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/f8c66053949240ae97e47ce729135de6 : SUCCESS in 1h 17m 57s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/828548c37e9849faa177bab410f1f20f : SUCCESS in 1h 12m 18s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/27fa42f8e81347718a16564d3625ce76 : SUCCESS in 5m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e526e051e60e470580265e26e681ad45 : SUCCESS in 5m 35s","accounts_in_message":[],"_revision_number":17},{"id":"c4b459f9b76c427590adeea1311c5591e265bda9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 03:47:13.000000000","message":"Patch Set 17: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":17},{"id":"a7d781b48970b76fd058057bd927bb698f83cd7e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 09:58:27.000000000","message":"Patch Set 17: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/baf9bfad53894dfda887c4bc19a773f2 : SUCCESS in 5m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7f05b06d090f432597a3ca241d399cc6 : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e62f871806c448cd8c75b3fceb3a15b4 : SUCCESS in 5m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/22b7305aa76d4962a77c304c30ec7c6d : SUCCESS in 6m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2dc5d7df1d24811b3d0ba8ffd0e4139 : SUCCESS in 4m 57s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/087c8d9472e244f9814aef276e5a775a : SUCCESS in 3h 01m 22s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/e617f5a63cbe499d9d7e8f77cb302133 : SUCCESS in 2h 39m 56s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/3bb711f833ff4530ba02ba5368a442d7 : SUCCESS in 1h 41m 18s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/ae637e6c1e3945a294216975e2f7f0ad : SUCCESS in 2h 39m 22s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/c4220af5062b45608f1e6e0279e6b6b1 : SUCCESS in 2h 16m 22s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/0f901eb9b6704aa09622ce46b5ed293f : SUCCESS in 1h 26m 10s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/64c2180d1722460aa2901e7e10f25b2c : SUCCESS in 55m 38s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/dce5eff1dd2e459f92ac8d009237da2e : SUCCESS in 31m 31s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/efa89192dc5e47e9bc0b6febb27d456e : SUCCESS in 1h 48m 46s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/890c1101c9d44237b7b0361cdd4235a7 : SUCCESS in 1h 21m 53s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/f13d77d8edaa45dea866ce3b008f15a6 : SUCCESS in 1h 21m 33s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/451746ac219942e886f33caa495b2c1e : FAILURE in 2h 59m 18s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/07e531daf2704948862ed4ef47de20a4 : SUCCESS in 1h 17m 52s\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/c865e13974c34d9b96c5316cedb7a927 : SUCCESS in 1h 26m 12s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/016c5840fd9c4097bf3b07fe145591ba : SUCCESS in 4m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cd41f2a208724d9db0873a1180c8075e : SUCCESS in 6m 51s","accounts_in_message":[],"_revision_number":17},{"id":"af0b3721a508d58feed343e4718a3cafe9f03c18","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-08-26 10:14:04.000000000","message":"Patch Set 17:\n\nrecheck 004 failed standalone deploy with async task did not complete within the requested time - 5700s","accounts_in_message":[],"_revision_number":17},{"id":"53316c19de7e6e5aae2634a245ab3f01da252f18","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-26 12:48:32.000000000","message":"Patch Set 17:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/ce2b0c65335e42ff824c4525ae234f7a : FAILURE in 2h 33m 12s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"4acf7de674c9b241f429842fc3a9d5c8806f3a90","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-08-26 12:56:51.000000000","message":"Patch Set 17:\n\nrecheck","accounts_in_message":[],"_revision_number":17},{"id":"f1bc0d16fed54ef596efa50735bd25ad36651526","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2020-08-26 13:02:54.000000000","message":"Patch Set 17:\n\nrecheck","accounts_in_message":[],"_revision_number":17},{"id":"6d97e3d3082702d657cfc6e1ef4998d82eeb32f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 14:09:44.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/744cbba416464a22bb0e59f3e6519f5b : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24e074891e0f46dd989e05c7b0f90d4d : SUCCESS in 4m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/da128372a21c4ddd8c8badddc045ab75 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5763a9e4f3c14d50a9f4fe2503b75c9e : SUCCESS in 5m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/53b7befd8d22415882cc8116b468f357 : SUCCESS in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e463861f4554e4982e65dc768324a54 : SUCCESS in 5m 41s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/d33a9f6969f64653b4d6e336bd5c1670 : SUCCESS in 2h 47m 20s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/12353275814945679a266167f9934fa3 : SUCCESS in 3h 36m 22s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/124ba2c798e94fdaaa15a0c7aa19e068 : SUCCESS in 2h 15m 37s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/44a227b8394f4ce49e96094fa5359334 : SUCCESS in 2h 57m 37s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/484ac45b079b407ea9c484093794cb99 : FAILURE in 2h 37m 59s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/5364a056d97848dfaf932f5ef68c8a77 : SUCCESS in 2h 25m 46s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/7c0ca073a29644ccbc0c17947ec307f3 : SUCCESS in 2h 45m 40s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/349b933d0a2845d8ae2f1c5e45354f99 : SUCCESS in 1h 25m 48s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/cbc50a059601463abcb1e21034887b23 : FAILURE in 1h 59m 15s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/e319b16c206e404c8ddf66aeac4e8e9d : SUCCESS in 50m 41s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/5b0189b7592447f38fee65d85b672582 : SUCCESS in 53m 05s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/3be326411e684734a00f03a3272daf03 : SUCCESS in 1h 32m 04s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/9a0cefd0dc8d491bb20a5a6a1003b756 : SUCCESS in 1h 40m 38s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/1d79fb17bf7243898848c38000eb3d58 : SUCCESS in 1h 16m 52s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/c452785a3c954d82942e8fc0232b7cc7 : SUCCESS in 1h 42m 01s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/413b558cc75b4e0fb26bc59049baed9a : SUCCESS in 2h 12m 05s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/544798462fe04dafa705342d850ac7bf : SUCCESS in 1h 18m 16s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/72deba8ed21c491386db97573e10b3d3 : SUCCESS in 1h 17m 04s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/2b2c2d189b28451ab309af73d15e195b : SUCCESS in 1h 00m 53s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4144da91b35848d0be009bd572c55567 : SUCCESS in 3m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3f535950fe6542dfa18ef8cde1a200bc : SUCCESS in 6m 24s","accounts_in_message":[],"_revision_number":17},{"id":"b5d9bb66673caaa1ae8162dd49e822f87620ff7d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 14:10:13.000000000","message":"Patch Set 17: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":17},{"id":"011fee5d0db527a21af3de357efe8083fc18f799","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-26 15:26:14.000000000","message":"Patch Set 17:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/3917a6b9788e41b490d60a87d8bb66ff : FAILURE in 2h 28m 26s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"89c3df9de7607c7429e75f8068738432013764d0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 23:01:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":17},{"id":"9a813839f155090b53f088c845f32697890bf973","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 23:01:28.000000000","message":"Patch Set 17: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20f0610be8a146a69c4390cd38e41d74 : SUCCESS in 4m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3d36f77548c04ea68cce8540ce5941eb : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2142d04329be45abb07089091544ad9c : SUCCESS in 4m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8f9d746c62c3407f8509958d3dd36b6d : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82e34b8176e54bb6bb153dadb8c04483 : SUCCESS in 5m 09s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/d373deaa4ee64da7831235c4218abc2a : SUCCESS in 2h 29m 06s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/562fc8ed7c314753a5add5f95fad3629 : SUCCESS in 3h 19m 34s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/1423c8e582164513bf62e8c31b19df57 : SUCCESS in 1h 35m 41s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/45b9221c474847bd84d006c7a94808be : SUCCESS in 2h 26m 21s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/2b9bd269509e4ebba747f7fdeb173d0e : SUCCESS in 2h 40m 54s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/5228539b26a84eb8bced3955394918d6 : SUCCESS in 1h 54m 29s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/04f91fad5eed45369b86c10a0f8f2fdd : SUCCESS in 58m 11s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/bed2f328b4dc41c5810f4d7c30ba1daf : SUCCESS in 1h 25m 37s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/73191b6b30ee4e2b9b83f9847257d826 : SUCCESS in 2h 17m 27s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/c8def0ece5f84b78bcdf47b0cfa45d5f : SUCCESS in 2h 02m 56s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/f88fc6e35ddb46e5bf0762200e4625f9 : SUCCESS in 1h 46m 13s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/a07b23a2e9644afd8ca1860a93d784b3 : SUCCESS in 2h 26m 01s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/b486a25bede34b60b49f8800a4b0025d : SUCCESS in 1h 38m 44s\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/280b470a793a4da2bb2c89820f3892e8 : SUCCESS in 1h 47m 02s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8cb350a4e6b245c1992d55d72bb4c091 : SUCCESS in 6m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2e4c69b1fba44495b4a2ad77f693b296 : SUCCESS in 5m 17s","accounts_in_message":[],"_revision_number":17},{"id":"d887a34b8869f66440ff59526d01bc741b8c73d6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 23:05:47.000000000","message":"Patch Set 17:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0be56a1b75b44158ba61d757a9edf6cc : SUCCESS in 1m 42s","accounts_in_message":[],"_revision_number":17}],"current_revision_number":17,"current_revision":"365bfd518f7c8b9d06b88f3751adf857a4a8edd9","revisions":{"8f8b025a9616a3abfff90924bbdf205e1526900c":{"kind":"REWORK","_number":1,"created":"2020-07-15 14:42:33.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/28/741228/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/28/741228/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/1"}}},"commit":{"parents":[{"commit":"e14e7135e4612f2d92a0b73c7bd16e4e659eef65","subject":"WIP - Run standalone jobs with locally built images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/e14e7135e4612f2d92a0b73c7bd16e4e659eef65"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-06-10 15:09:44.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-07-15 14:42:24.000000000","tz":-240},"subject":"tcib: move cradox to gnocchi-base \u0026 enable ci on scenario001","message":"tcib: move cradox to gnocchi-base \u0026 enable ci on scenario001\n\ngnocchi-db-sync requires cradox to be installed if RBD backend is used.\nUntil we fix it properly in the distgit, let\u0027s make sure cradox is in\nthe base image for gnocchi. It also requires httpd and deps for the\ndv_sync to operate when Ceph is installed.\n\nWith that patch, we can now test container builds on scenario001.\n\nChange-Id: Id13028ab911ba91d67787ca8247effc02017fcbc\n(cherry picked from commit 7ecc263401018ec657e3423edcd9974bd8a2754f)\n(cherry picked from commit 55573ec43fdc9ffddbbf9928ab4b16df8edc56c3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/8f8b025a9616a3abfff90924bbdf205e1526900c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/8f8b025a9616a3abfff90924bbdf205e1526900c"}]},"branch":"refs/heads/stable/train"},"babc6ccab990f4cdba63176ddba7ad0cc94bd958":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-07-15 22:02:06.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/28/741228/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/28/741228/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/2"}}},"commit":{"parents":[{"commit":"afffbe87392c91572c3bc24ec3ff86c4ea347db6","subject":"WIP - Run standalone jobs with locally built images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/afffbe87392c91572c3bc24ec3ff86c4ea347db6"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-06-10 15:09:44.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-07-15 22:02:06.000000000","tz":0},"subject":"tcib: move cradox to gnocchi-base \u0026 enable ci on scenario001","message":"tcib: move cradox to gnocchi-base \u0026 enable ci on scenario001\n\ngnocchi-db-sync requires cradox to be installed if RBD backend is used.\nUntil we fix it properly in the distgit, let\u0027s make sure cradox is in\nthe base image for gnocchi. It also requires httpd and deps for the\ndv_sync to operate when Ceph is installed.\n\nWith that patch, we can now test container builds on scenario001.\n\nChange-Id: Id13028ab911ba91d67787ca8247effc02017fcbc\n(cherry picked from commit 7ecc263401018ec657e3423edcd9974bd8a2754f)\n(cherry picked from commit 55573ec43fdc9ffddbbf9928ab4b16df8edc56c3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/babc6ccab990f4cdba63176ddba7ad0cc94bd958"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/babc6ccab990f4cdba63176ddba7ad0cc94bd958"}]},"branch":"refs/heads/stable/train"},"af7c8236b1da3f78730f5e3e9dce10ad9b190aaf":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-07-24 01:53:53.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/28/741228/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/28/741228/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/3"}}},"commit":{"parents":[{"commit":"be71ff5971238a168757966229b8b6c90a08ef55","subject":"WIP - Run standalone jobs with locally built images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/be71ff5971238a168757966229b8b6c90a08ef55"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-06-10 15:09:44.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-07-24 01:53:53.000000000","tz":0},"subject":"tcib: move cradox to gnocchi-base \u0026 enable ci on scenario001","message":"tcib: move cradox to gnocchi-base \u0026 enable ci on scenario001\n\ngnocchi-db-sync requires cradox to be installed if RBD backend is used.\nUntil we fix it properly in the distgit, let\u0027s make sure cradox is in\nthe base image for gnocchi. It also requires httpd and deps for the\ndv_sync to operate when Ceph is installed.\n\nWith that patch, we can now test container builds on scenario001.\n\nChange-Id: Id13028ab911ba91d67787ca8247effc02017fcbc\n(cherry picked from commit 7ecc263401018ec657e3423edcd9974bd8a2754f)\n(cherry picked from commit 55573ec43fdc9ffddbbf9928ab4b16df8edc56c3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/af7c8236b1da3f78730f5e3e9dce10ad9b190aaf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/af7c8236b1da3f78730f5e3e9dce10ad9b190aaf"}]},"branch":"refs/heads/stable/train"},"7ba653345a553c935cd5bae16671be45433465d1":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-07-27 15:24:38.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/28/741228/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/28/741228/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/4"}}},"commit":{"parents":[{"commit":"c61ffc182d73bec7dd86ad0c2f4f6a88d860ae7c","subject":"Run standalone jobs with locally built images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/c61ffc182d73bec7dd86ad0c2f4f6a88d860ae7c"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-06-10 15:09:44.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-07-27 15:24:38.000000000","tz":0},"subject":"tcib: move cradox to gnocchi-base \u0026 enable ci on scenario001","message":"tcib: move cradox to gnocchi-base \u0026 enable ci on scenario001\n\ngnocchi-db-sync requires cradox to be installed if RBD backend is used.\nUntil we fix it properly in the distgit, let\u0027s make sure cradox is in\nthe base image for gnocchi. It also requires httpd and deps for the\ndv_sync to operate when Ceph is installed.\n\nWith that patch, we can now test container builds on scenario001.\n\nChange-Id: Id13028ab911ba91d67787ca8247effc02017fcbc\n(cherry picked from commit 7ecc263401018ec657e3423edcd9974bd8a2754f)\n(cherry picked from commit 55573ec43fdc9ffddbbf9928ab4b16df8edc56c3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/7ba653345a553c935cd5bae16671be45433465d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/7ba653345a553c935cd5bae16671be45433465d1"}]},"branch":"refs/heads/stable/train"},"1035d25b60996f415ffc0378dbc3b8cc12947d25":{"kind":"REWORK","_number":5,"created":"2020-07-27 15:25:13.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/28/741228/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/28/741228/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/5"}}},"commit":{"parents":[{"commit":"c61ffc182d73bec7dd86ad0c2f4f6a88d860ae7c","subject":"Run standalone jobs with locally built images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/c61ffc182d73bec7dd86ad0c2f4f6a88d860ae7c"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-06-10 15:09:44.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-07-27 15:25:10.000000000","tz":0},"subject":"tcib: move cradox to gnocchi-base \u0026 enable ci on scenario001","message":"tcib: move cradox to gnocchi-base \u0026 enable ci on scenario001\n\ngnocchi-db-sync requires cradox to be installed if RBD backend is used.\nUntil we fix it properly in the distgit, let\u0027s make sure cradox is in\nthe base image for gnocchi. It also requires httpd and deps for the\ndv_sync to operate when Ceph is installed.\n\nWith that patch, we can now test container builds on scenario001.\n\nChange-Id: Id13028ab911ba91d67787ca8247effc02017fcbc\n(cherry picked from commit 7ecc263401018ec657e3423edcd9974bd8a2754f)\n(cherry picked from commit 55573ec43fdc9ffddbbf9928ab4b16df8edc56c3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/1035d25b60996f415ffc0378dbc3b8cc12947d25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/1035d25b60996f415ffc0378dbc3b8cc12947d25"}]},"branch":"refs/heads/stable/train"},"2992b70e6c4f8f92dbc7b13bc21c3479a1b40d7a":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-07-29 02:32:52.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/28/741228/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/28/741228/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/6"}}},"commit":{"parents":[{"commit":"344f3b6046b1a0dce639c1fb609228c42414e014","subject":"Run standalone jobs with locally built images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/344f3b6046b1a0dce639c1fb609228c42414e014"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-06-10 15:09:44.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-07-29 02:32:52.000000000","tz":0},"subject":"tcib: move cradox to gnocchi-base \u0026 enable ci on scenario001","message":"tcib: move cradox to gnocchi-base \u0026 enable ci on scenario001\n\ngnocchi-db-sync requires cradox to be installed if RBD backend is used.\nUntil we fix it properly in the distgit, let\u0027s make sure cradox is in\nthe base image for gnocchi. It also requires httpd and deps for the\ndv_sync to operate when Ceph is installed.\n\nWith that patch, we can now test container builds on scenario001.\n\nChange-Id: Id13028ab911ba91d67787ca8247effc02017fcbc\n(cherry picked from commit 7ecc263401018ec657e3423edcd9974bd8a2754f)\n(cherry picked from commit 55573ec43fdc9ffddbbf9928ab4b16df8edc56c3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/2992b70e6c4f8f92dbc7b13bc21c3479a1b40d7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/2992b70e6c4f8f92dbc7b13bc21c3479a1b40d7a"}]},"branch":"refs/heads/stable/train"},"981373244db19a1e434d7b2fb6b84c5362bf22df":{"kind":"REWORK","_number":7,"created":"2020-07-31 13:24:11.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/28/741228/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/28/741228/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/7"}}},"commit":{"parents":[{"commit":"344f3b6046b1a0dce639c1fb609228c42414e014","subject":"Run standalone jobs with locally built images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/344f3b6046b1a0dce639c1fb609228c42414e014"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-06-10 15:09:44.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-07-31 13:23:31.000000000","tz":-240},"subject":"tcib: move cradox to gnocchi-base \u0026 enable ci on scenario001","message":"tcib: move cradox to gnocchi-base \u0026 enable ci on scenario001\n\ngnocchi-db-sync requires cradox to be installed if RBD backend is used.\nUntil we fix it properly in the distgit, let\u0027s make sure cradox is in\nthe base image for gnocchi. It also requires httpd and deps for the\ndv_sync to operate when Ceph is installed.\n\nWith that patch, we can now test container builds on scenario001.\n\nNote: adding panko-api container image, required in Train.\n\nChange-Id: Id13028ab911ba91d67787ca8247effc02017fcbc\n(cherry picked from commit 7ecc263401018ec657e3423edcd9974bd8a2754f)\n(cherry picked from commit 55573ec43fdc9ffddbbf9928ab4b16df8edc56c3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/981373244db19a1e434d7b2fb6b84c5362bf22df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/981373244db19a1e434d7b2fb6b84c5362bf22df"}]},"branch":"refs/heads/stable/train"},"4d3e3b8a0f0e426f381afe499b663fb4487f65b6":{"kind":"REWORK","_number":8,"created":"2020-07-31 13:30:05.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/28/741228/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/28/741228/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/8"}}},"commit":{"parents":[{"commit":"344f3b6046b1a0dce639c1fb609228c42414e014","subject":"Run standalone jobs with locally built images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/344f3b6046b1a0dce639c1fb609228c42414e014"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-06-10 15:09:44.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-07-31 13:30:03.000000000","tz":0},"subject":"tcib: move cradox to gnocchi-base \u0026 enable ci on scenario001","message":"tcib: move cradox to gnocchi-base \u0026 enable ci on scenario001\n\ngnocchi-db-sync requires cradox to be installed if RBD backend is used.\nUntil we fix it properly in the distgit, let\u0027s make sure cradox is in\nthe base image for gnocchi. It also requires httpd and deps for the\ndv_sync to operate when Ceph is installed.\n\nWith that patch, we can now test container builds on scenario001.\n\nNote: adding panko-api container image, required in Train.\n\nChange-Id: Id13028ab911ba91d67787ca8247effc02017fcbc\n(cherry picked from commit 7ecc263401018ec657e3423edcd9974bd8a2754f)\n(cherry picked from commit 55573ec43fdc9ffddbbf9928ab4b16df8edc56c3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/4d3e3b8a0f0e426f381afe499b663fb4487f65b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/4d3e3b8a0f0e426f381afe499b663fb4487f65b6"}]},"branch":"refs/heads/stable/train"},"3fa8a199900921f0875624a77f9c4e784d33bf24":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2020-08-10 13:57:01.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/28/741228/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/28/741228/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/9"}}},"commit":{"parents":[{"commit":"e0c8d7770aa36aba353e5c70174c1f5ca7a59ae0","subject":"Revert \"Fix stdout_callback\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/e0c8d7770aa36aba353e5c70174c1f5ca7a59ae0"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-06-10 15:09:44.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-08-10 13:57:01.000000000","tz":0},"subject":"tcib: move cradox to gnocchi-base \u0026 enable ci on scenario001","message":"tcib: move cradox to gnocchi-base \u0026 enable ci on scenario001\n\ngnocchi-db-sync requires cradox to be installed if RBD backend is used.\nUntil we fix it properly in the distgit, let\u0027s make sure cradox is in\nthe base image for gnocchi. It also requires httpd and deps for the\ndv_sync to operate when Ceph is installed.\n\nWith that patch, we can now test container builds on scenario001.\n\nNote: adding panko-api container image, required in Train.\n\nChange-Id: Id13028ab911ba91d67787ca8247effc02017fcbc\n(cherry picked from commit 7ecc263401018ec657e3423edcd9974bd8a2754f)\n(cherry picked from commit 55573ec43fdc9ffddbbf9928ab4b16df8edc56c3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/3fa8a199900921f0875624a77f9c4e784d33bf24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/3fa8a199900921f0875624a77f9c4e784d33bf24"}]},"branch":"refs/heads/stable/train"},"9c32855ba5213558348c1f1e17a9c36ce4cc7d70":{"kind":"REWORK","_number":10,"created":"2020-08-10 20:26:23.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/28/741228/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/28/741228/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/10"}}},"commit":{"parents":[{"commit":"645dc20f30a32e6664380019cced4420fd6ec125","subject":"Merge \"Port jobs from centos7 to centos8\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/645dc20f30a32e6664380019cced4420fd6ec125"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-06-10 15:09:44.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-08-10 20:26:08.000000000","tz":-240},"subject":"tcib: move cradox to gnocchi-base \u0026 enable ci on scenario001","message":"tcib: move cradox to gnocchi-base \u0026 enable ci on scenario001\n\ngnocchi-db-sync requires cradox to be installed if RBD backend is used.\nUntil we fix it properly in the distgit, let\u0027s make sure cradox is in\nthe base image for gnocchi. It also requires httpd and deps for the\ndv_sync to operate when Ceph is installed.\n\nWith that patch, we can now test container builds on scenario001.\n\nNote: adding panko-api container image, required in Train.\n\nChange-Id: Id13028ab911ba91d67787ca8247effc02017fcbc\n(cherry picked from commit 7ecc263401018ec657e3423edcd9974bd8a2754f)\n(cherry picked from commit 55573ec43fdc9ffddbbf9928ab4b16df8edc56c3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/9c32855ba5213558348c1f1e17a9c36ce4cc7d70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/9c32855ba5213558348c1f1e17a9c36ce4cc7d70"}]},"branch":"refs/heads/stable/train"},"f290607d3e8d338e26cddf19dcdf384b6fdc22ea":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2020-08-11 12:37:46.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/28/741228/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/28/741228/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/11"}}},"commit":{"parents":[{"commit":"473d9cbc32163ce548e1483e0d5e1631c096c8d2","subject":"Merge \"Properly collect tasks with or conditional in step file.\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/473d9cbc32163ce548e1483e0d5e1631c096c8d2"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-06-10 15:09:44.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-08-11 12:37:46.000000000","tz":0},"subject":"tcib: move cradox to gnocchi-base \u0026 enable ci on scenario001","message":"tcib: move cradox to gnocchi-base \u0026 enable ci on scenario001\n\ngnocchi-db-sync requires cradox to be installed if RBD backend is used.\nUntil we fix it properly in the distgit, let\u0027s make sure cradox is in\nthe base image for gnocchi. It also requires httpd and deps for the\ndv_sync to operate when Ceph is installed.\n\nWith that patch, we can now test container builds on scenario001.\n\nNote: adding panko-api container image, required in Train.\n\nChange-Id: Id13028ab911ba91d67787ca8247effc02017fcbc\n(cherry picked from commit 7ecc263401018ec657e3423edcd9974bd8a2754f)\n(cherry picked from commit 55573ec43fdc9ffddbbf9928ab4b16df8edc56c3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/f290607d3e8d338e26cddf19dcdf384b6fdc22ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/f290607d3e8d338e26cddf19dcdf384b6fdc22ea"}]},"branch":"refs/heads/stable/train"},"07445de41bf22782d61121b37c5c015ee463350b":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2020-08-13 01:32:36.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/28/741228/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/28/741228/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/12"}}},"commit":{"parents":[{"commit":"21b4a1f3511fec6eab2c25db52acd85e1693ab84","subject":"Merge \"Switch to use virt:8.3 for non-kolla container builds\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/21b4a1f3511fec6eab2c25db52acd85e1693ab84"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-06-10 15:09:44.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-08-13 01:32:36.000000000","tz":0},"subject":"tcib: move cradox to gnocchi-base \u0026 enable ci on scenario001","message":"tcib: move cradox to gnocchi-base \u0026 enable ci on scenario001\n\ngnocchi-db-sync requires cradox to be installed if RBD backend is used.\nUntil we fix it properly in the distgit, let\u0027s make sure cradox is in\nthe base image for gnocchi. It also requires httpd and deps for the\ndv_sync to operate when Ceph is installed.\n\nWith that patch, we can now test container builds on scenario001.\n\nNote: adding panko-api container image, required in Train.\n\nChange-Id: Id13028ab911ba91d67787ca8247effc02017fcbc\n(cherry picked from commit 7ecc263401018ec657e3423edcd9974bd8a2754f)\n(cherry picked from commit 55573ec43fdc9ffddbbf9928ab4b16df8edc56c3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/07445de41bf22782d61121b37c5c015ee463350b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/07445de41bf22782d61121b37c5c015ee463350b"}]},"branch":"refs/heads/stable/train"},"eec6684d9218ee0099901cfe7bfd5c8e3a067ee9":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2020-08-17 13:25:57.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/28/741228/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/28/741228/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/13"}}},"commit":{"parents":[{"commit":"4ce40a6654b72e447cc1583f2f03c483767ae898","subject":"Merge \"Ensure tripleo ansible inventory file update is atomic\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/4ce40a6654b72e447cc1583f2f03c483767ae898"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-06-10 15:09:44.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-08-17 13:25:57.000000000","tz":0},"subject":"tcib: move cradox to gnocchi-base \u0026 enable ci on scenario001","message":"tcib: move cradox to gnocchi-base \u0026 enable ci on scenario001\n\ngnocchi-db-sync requires cradox to be installed if RBD backend is used.\nUntil we fix it properly in the distgit, let\u0027s make sure cradox is in\nthe base image for gnocchi. It also requires httpd and deps for the\ndv_sync to operate when Ceph is installed.\n\nWith that patch, we can now test container builds on scenario001.\n\nNote: adding panko-api container image, required in Train.\n\nChange-Id: Id13028ab911ba91d67787ca8247effc02017fcbc\n(cherry picked from commit 7ecc263401018ec657e3423edcd9974bd8a2754f)\n(cherry picked from commit 55573ec43fdc9ffddbbf9928ab4b16df8edc56c3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/eec6684d9218ee0099901cfe7bfd5c8e3a067ee9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/eec6684d9218ee0099901cfe7bfd5c8e3a067ee9"}]},"branch":"refs/heads/stable/train"},"cb746e39ebdd979c1b34b122ecf70d8cece703ae":{"kind":"REWORK","_number":14,"created":"2020-08-17 15:10:42.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/28/741228/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/28/741228/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/14"}}},"commit":{"parents":[{"commit":"4a17778ac0b0f33e1a372d028e4c27d8b79d8c9e","subject":"(backward compatibility) create kolla UIDs/GIDs for TCIB","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/4a17778ac0b0f33e1a372d028e4c27d8b79d8c9e"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-06-10 15:09:44.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-08-17 15:10:30.000000000","tz":-240},"subject":"tcib: move cradox to gnocchi-base \u0026 enable ci on scenario001","message":"tcib: move cradox to gnocchi-base \u0026 enable ci on scenario001\n\ngnocchi-db-sync requires cradox to be installed if RBD backend is used.\nUntil we fix it properly in the distgit, let\u0027s make sure cradox is in\nthe base image for gnocchi. It also requires httpd and deps for the\ndv_sync to operate when Ceph is installed.\n\nWith that patch, we can now test container builds on scenario001.\n\nNote: adding panko-api container image, required in Train.\n\nChange-Id: Id13028ab911ba91d67787ca8247effc02017fcbc\n(cherry picked from commit 7ecc263401018ec657e3423edcd9974bd8a2754f)\n(cherry picked from commit 55573ec43fdc9ffddbbf9928ab4b16df8edc56c3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/cb746e39ebdd979c1b34b122ecf70d8cece703ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/cb746e39ebdd979c1b34b122ecf70d8cece703ae"}]},"branch":"refs/heads/stable/train"},"7c2a0736f26895f73772910e11934af69587fab0":{"kind":"REWORK","_number":15,"created":"2020-08-18 12:18:14.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/28/741228/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/28/741228/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/15"}}},"commit":{"parents":[{"commit":"4a17778ac0b0f33e1a372d028e4c27d8b79d8c9e","subject":"(backward compatibility) create kolla UIDs/GIDs for TCIB","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/4a17778ac0b0f33e1a372d028e4c27d8b79d8c9e"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-06-10 15:09:44.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-08-18 12:18:13.000000000","tz":0},"subject":"tcib: move cradox to gnocchi-base \u0026 enable ci on scenario001","message":"tcib: move cradox to gnocchi-base \u0026 enable ci on scenario001\n\ngnocchi-db-sync requires cradox to be installed if RBD backend is used.\nUntil we fix it properly in the distgit, let\u0027s make sure cradox is in\nthe base image for gnocchi. It also requires httpd and deps for the\ndv_sync to operate when Ceph is installed.\n\nWith that patch, we can now test container builds on scenario001.\n\nNote: adding panko-api container image, required in Train.\n\nChange-Id: Id13028ab911ba91d67787ca8247effc02017fcbc\n(cherry picked from commit 7ecc263401018ec657e3423edcd9974bd8a2754f)\n(cherry picked from commit 55573ec43fdc9ffddbbf9928ab4b16df8edc56c3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/7c2a0736f26895f73772910e11934af69587fab0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/7c2a0736f26895f73772910e11934af69587fab0"}]},"branch":"refs/heads/stable/train"},"f8afb7338986d756b56bb6a9d5f83b1bdb5626e8":{"kind":"REWORK","_number":16,"created":"2020-08-19 21:04:02.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/28/741228/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/28/741228/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/16"}}},"commit":{"parents":[{"commit":"4a17778ac0b0f33e1a372d028e4c27d8b79d8c9e","subject":"(backward compatibility) create kolla UIDs/GIDs for TCIB","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/4a17778ac0b0f33e1a372d028e4c27d8b79d8c9e"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-06-10 15:09:44.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-08-19 21:03:51.000000000","tz":-240},"subject":"tcib: move cradox to gnocchi-base \u0026 enable ci on scenario001","message":"tcib: move cradox to gnocchi-base \u0026 enable ci on scenario001\n\ngnocchi-db-sync requires cradox to be installed if RBD backend is used.\nUntil we fix it properly in the distgit, let\u0027s make sure cradox is in\nthe base image for gnocchi. It also requires httpd and deps for the\ndv_sync to operate when Ceph is installed.\n\nWith that patch, we can now test container builds on scenario001.\n\nNote: adding panko-api container image, required in Train.\n\nChange-Id: Id13028ab911ba91d67787ca8247effc02017fcbc\n(cherry picked from commit 7ecc263401018ec657e3423edcd9974bd8a2754f)\n(cherry picked from commit 55573ec43fdc9ffddbbf9928ab4b16df8edc56c3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/f8afb7338986d756b56bb6a9d5f83b1bdb5626e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/f8afb7338986d756b56bb6a9d5f83b1bdb5626e8"}]},"branch":"refs/heads/stable/train"},"365bfd518f7c8b9d06b88f3751adf857a4a8edd9":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2020-08-20 21:01:03.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/28/741228/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/28/741228/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/28/741228/17"}}},"commit":{"parents":[{"commit":"5397e9eb4fcf2c95f50049995d433bca3ea3dcf9","subject":"(backward compatibility) create kolla UIDs/GIDs for TCIB","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/5397e9eb4fcf2c95f50049995d433bca3ea3dcf9"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-06-10 15:09:44.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-08-20 21:01:03.000000000","tz":0},"subject":"tcib: move cradox to gnocchi-base \u0026 enable ci on scenario001","message":"tcib: move cradox to gnocchi-base \u0026 enable ci on scenario001\n\ngnocchi-db-sync requires cradox to be installed if RBD backend is used.\nUntil we fix it properly in the distgit, let\u0027s make sure cradox is in\nthe base image for gnocchi. It also requires httpd and deps for the\ndv_sync to operate when Ceph is installed.\n\nWith that patch, we can now test container builds on scenario001.\n\nNote: adding panko-api container image, required in Train.\n\nChange-Id: Id13028ab911ba91d67787ca8247effc02017fcbc\n(cherry picked from commit 7ecc263401018ec657e3423edcd9974bd8a2754f)\n(cherry picked from commit 55573ec43fdc9ffddbbf9928ab4b16df8edc56c3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/365bfd518f7c8b9d06b88f3751adf857a4a8edd9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/365bfd518f7c8b9d06b88f3751adf857a4a8edd9"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
