)]}'
{"id":"openstack%2Ftripleo-common~743301","triplet_id":"openstack%2Ftripleo-common~stable%2Ftrain~I3c4e6e21259f97ea794111a9d68b1ffec1f733e0","project":"openstack/tripleo-common","branch":"stable/train","topic":"callback-improvement-stable/train","hashtags":[],"change_id":"I3c4e6e21259f97ea794111a9d68b1ffec1f733e0","subject":"Revert \"Fix stdout_callback\"","status":"MERGED","created":"2020-07-27 18:18:54.000000000","updated":"2020-08-06 23:35:39.000000000","submitted":"2020-08-06 23:34:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"743301-1596756850426-76a6aaaf","meta_rev_id":"bc0b6a4a68d39103ef114681d1f957c08cc4d55b","_number":743301,"virtual_id_number":743301,"owner":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-06 23:34:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-06 21:11:40.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-08-06 17:35:19.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":1,"date":"2020-08-06 17:35:21.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-27 18:18:54.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2020-07-27 19:03:45.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-07-29 10:47:03.000000000","updated_by":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"reviewer":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"state":"REVIEWER"},{"updated":"2020-08-06 21:11:40.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-06 23:34:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"57880ea82ed1be42a4951ea226c0de9d7b739567","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-07-27 18:18:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e97b0acece646295804f3e8ee7152e34248de3d4","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-07-27 19:03:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"602172a3c7c3c674a323908248b6c336fcb6f826","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-07-27 21:05:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-rhel-8-buildimage-overcloud-full-train https://review.rdoproject.org/zuul/build/eff6e6fe15b74010935c904e971ba3c8 : RETRY_LIMIT in 6m 04s\n- tripleo-rhel-8-buildimage-ironic-python-agent-train https://review.rdoproject.org/zuul/build/954258c894704f43afabb39865f47e0e : RETRY_LIMIT in 6m 12s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/69d5ea55228c4c528478ae54d4bdda4e : FAILURE in 2h 45m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7fa5801e955d15930b1e1eccc93eb0a0a0c30dd6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 23:28:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3af71dba9c8a4141819a4b4e397218ec : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1315bb0e189a4e80ba3fe8cf8901ba16 : SUCCESS in 3m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a50976b13cef43c78585e2920b6ebd8f : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8003141e75b54af1a988ad388610706a : SUCCESS in 4m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/41e968cc94154dde9983a05f518c67e2 : SUCCESS in 6m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5b90f19251444a88d00c049a5848886 : SUCCESS in 5m 03s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/58bd4f2a8b5f4474a18ce73c3ae664b0 : SUCCESS in 2h 29m 00s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/9934ac5821084aeb90c7641a80e7855e : SUCCESS in 2h 29m 57s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/062c9cd16ae14ed2b67dba77c8bdb2de : FAILURE in 2h 00m 05s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/66097beff80346b083272f344373fbdf : FAILURE in 1h 23m 59s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/262ceabec9d04f258d3d8242e975c5c6 : SUCCESS in 1h 29m 19s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/a3524a9c853b4219827d3c385a927afe : FAILURE in 1h 27m 55s (non-voting)\n- tripleo-ci-centos-7-containers-undercloud-minion https://zuul.opendev.org/t/openstack/build/4f81a94784154c5a98bc9c4f410ee4a7 : SUCCESS in 1h 34m 24s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/2eb0005074154ee881b20ec6b7c4b398 : FAILURE in 1h 07m 09s (non-voting)\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/a57e8110b90e45a595404f81eb873af1 : SUCCESS in 1h 58m 47s\n- tripleo-ci-centos-7-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/98cb38b7d5d74f6db9b68561170a7365 : FAILURE in 1h 16m 00s (non-voting)\n- tripleo-ci-centos-7-scenario007-standalone https://zuul.opendev.org/t/openstack/build/b5baa480effd47969719185c2bf9db0d : FAILURE in 2h 17m 07s\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/051a54a1b7704b049f58b6c2843765e8 : FAILURE in 2h 48m 48s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/d8ad738dccbb43b9a9e9047efa9004f3 : SUCCESS in 2h 06m 12s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/36b3d250e3fe459ba8b115a087d17128 : FAILURE in 1h 15m 01s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c80e61b340614ca7826bac1bf62f588f : SUCCESS in 3m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/783018f9d1e047319cbdb84e2ba0149a : SUCCESS in 6m 02s","accounts_in_message":[],"_revision_number":1},{"id":"2678ec39170db8518592f75ab2dc0aa065b88246","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-07-28 12:46:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7def588098aed9af7bbf420523d984060d7710da","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-07-28 16:15:34.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-rhel-8-buildimage-overcloud-full-train https://review.rdoproject.org/zuul/build/5897225f899143148061285e2247abb0 : RETRY_LIMIT in 7m 28s\n- tripleo-rhel-8-buildimage-ironic-python-agent-train https://review.rdoproject.org/zuul/build/1cbc03a1cb364aebb96e6b075b7dd227 : RETRY_LIMIT in 6m 03s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/e92d2ff092844617860e9ab14d3364e5 : SUCCESS in 3h 27m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2822ebaf309ddcdf6e246a036ea649b1d7a429d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 22:14:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e18dc797cac54c59aa584d061b4c1177 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9cefd85f2054733ba1fe31d32ccb6f1 : SUCCESS in 3m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4fd1d116c51046638a7bc22a429f008d : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0d999392a0aa4459b3ebf30d79a2a20d : SUCCESS in 4m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1295894bf6c44eadbe1a6b9daa77cc22 : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b902bb88bea5420f8a666efb96206060 : SUCCESS in 5m 50s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/742fef600e704cda97609fe006843a44 : SUCCESS in 1h 51m 49s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/df19268bc8a9419094590ff3b5eecbde : SUCCESS in 2h 53m 10s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/570260c81cd14bfbb61a66803fc799ed : SUCCESS in 2h 26m 22s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/5cd6ced4ed7b447da5d25c37040fd2aa : FAILURE in 2h 15m 15s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/763350bf56be446e9bfb356e0ff13201 : SUCCESS in 1h 44m 44s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/52602668c34e4a5e83d60da0b4004283 : SUCCESS in 3h 10m 29s (non-voting)\n- tripleo-ci-centos-7-containers-undercloud-minion https://zuul.opendev.org/t/openstack/build/d2be0029227842c58ba793c35a364fc7 : SUCCESS in 1h 46m 03s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/a7a280e32cf34d8dabb35f58ccba8852 : FAILURE in 2h 21m 57s (non-voting)\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/ca818bbe6edc4f82b6f964e435e02b9e : SUCCESS in 2h 28m 47s\n- tripleo-ci-centos-7-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/ca61658d65e148dd80b6128073251b8a : SUCCESS in 1h 20m 14s (non-voting)\n- tripleo-ci-centos-7-scenario007-standalone https://zuul.opendev.org/t/openstack/build/045762a33e524786be970ca6527b981f : SUCCESS in 1h 24m 27s\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/952b07672ede412496896e5297397edd : SUCCESS in 2h 03m 11s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/607bd67b220e40c18ee5bc007f6d2f51 : SUCCESS in 1h 32m 32s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/a0f81522da4543ac8c0b1d78a3360939 : SUCCESS in 1h 51m 54s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f06631ca30884a2bb23a87994ee1cca1 : SUCCESS in 5m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0fcdc56d00ef4329a3a9c6b7c045be8f : SUCCESS in 5m 20s","accounts_in_message":[],"_revision_number":1},{"id":"998e1a2fd811f2bf9e89470f9fd14742bdff67f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 22:14:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6993fea43022c25b30bf28ee65e8f62c7bd231f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 02:28:13.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8ed43b37417492c932edb1d83dfb9b0 : SUCCESS in 4m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f3c0c9e1ce1543f5b95f8cd1555ee22f : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/580cbda2e0e5499e852e906dba8cd357 : SUCCESS in 4m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3fa5c4576d0544118a2b3944521e022b : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d0c629a558448b1ba20384e69f1bb42 : SUCCESS in 5m 26s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/07faa43f96bc45398485a5138d8981f8 : SUCCESS in 2h 56m 09s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/aed5898158bb459085e661732bfffc36 : SUCCESS in 1h 28m 52s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/f9e08fe4d46e4fd2bafe8244d3d9924e : SUCCESS in 2h 14m 45s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/a451ac54bace474fb6c0edfbd7db5e84 : SUCCESS in 2h 17m 55s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/576a8e0924bf45dfbb15d772a9e95373 : SUCCESS in 1h 45m 29s\n- tripleo-ci-centos-7-containers-undercloud-minion https://zuul.opendev.org/t/openstack/build/2c68c1e1c38e44538a1638b3295734c2 : SUCCESS in 1h 50m 11s\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/a2e2282fc89a462e9702860f4c5b2d3f : FAILURE in 2h 25m 19s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/49d42e9e8ebe460394cfd77524249243 : SUCCESS in 1h 40m 17s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/8850a6e4b7b547009d83ce96b4d85b2f : SUCCESS in 1h 39m 03s\n- tripleo-ci-centos-7-scenario007-standalone https://zuul.opendev.org/t/openstack/build/f3431ec2023d443a94ed9502351f0e7e : SUCCESS in 1h 32m 38s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/395d99af27cd4969bc545dc031914545 : SUCCESS in 1h 47m 51s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/bc0c186808ab477491a0c53317b19b00 : SUCCESS in 3m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8962c0dc0d03400f800b92c06aa91546 : SUCCESS in 5m 57s","accounts_in_message":[],"_revision_number":1},{"id":"e519159a36b798085fca6438b292e35de0feb2d5","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-07-29 02:34:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1b1eccbae233b830e15e6cb1c9e92b7dcad60711","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 05:38:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/80bc965849b94da2a5709a6da04caff4 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69f258362149426e90e95f8e96622804 : SUCCESS in 3m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fd08fe1d7b264cbb81b10ef910eaea14 : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ce9ed03cb484796a2c42815cd34c231 : SUCCESS in 4m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d0116a0ef35d4d7d926cc932600fb121 : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7b34c92b3d64c72a5d0ec47ca6ffefe : SUCCESS in 3m 42s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/c873e2cb62df45f1a163909a3e1b9c3b : FAILURE in 15m 02s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/b8d316ec70374e5b93e10a55cc1471ef : FAILURE in 18m 22s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/e6a6fce15ec74322a81f5bdc2e7efdc9 : FAILURE in 12m 22s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/6ad00849a14443a6a8c692382a4297b9 : FAILURE in 1h 55m 41s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/cbe18a76af95445ab8d3b653a3672027 : FAILURE in 14m 55s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/1bb2fb035a234f818a860f1a36c6fa77 : SUCCESS in 2h 37m 53s (non-voting)\n- tripleo-ci-centos-7-containers-undercloud-minion https://zuul.opendev.org/t/openstack/build/5dc15c1cbd794f32bbb59c17d3646d83 : SUCCESS in 1h 47m 29s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/5cb5b09e4a3c436aae9addf506a4aa3a : FAILURE in 2h 20m 50s (non-voting)\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/1733997df97141da9e50399d761bbcd9 : SUCCESS in 2h 03m 41s\n- tripleo-ci-centos-7-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/580193bbf2d3423884b3c1cce381c517 : FAILURE in 2h 23m 13s (non-voting)\n- tripleo-ci-centos-7-scenario007-standalone https://zuul.opendev.org/t/openstack/build/080bbffb12884f0387f563479ee3bbf0 : FAILURE in 15m 38s\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/6cfe26932f784eaca98be1790401661c : FAILURE in 8m 34s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/38a502eb32b343fd8cc02ad701fae11a : FAILURE in 9m 21s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/cec014f8929443809bf216dc373ade1e : FAILURE in 16m 37s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7c883b87e0604215b1162070ea9eaf4d : SUCCESS in 3m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/132588bb8c7c44bfb96bff9f035246b0 : SUCCESS in 6m 27s","accounts_in_message":[],"_revision_number":1},{"id":"859ba0d5ca85a76df319518d7c0daa0a8851c09c","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2020-07-29 06:03:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5a2b4b6e9c71fc11708dc41f71594571ea6c5ce1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-07-29 06:06:38.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-rhel-8-buildimage-overcloud-full-train https://review.rdoproject.org/zuul/build/2741bff1b670469183797ce91ae71da8 : RETRY_LIMIT in 5m 55s\n- tripleo-rhel-8-buildimage-ironic-python-agent-train https://review.rdoproject.org/zuul/build/a5c839f391084fd8a1f6e86bf50b9cea : RETRY_LIMIT in 6m 13s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/812ecb98effe424eb0980fd74d6f8152 : SUCCESS in 3h 30m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"29718ccdccd35a8188c2c26182d621e02310fa3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 10:39:37.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6f41f52fccb54efc9b5243c468169f20 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac81569bc3c240a4878406ddcc76fdf8 : SUCCESS in 3m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f5336a05fc514e7cb596a7bd80da3e10 : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/086713bf7b7e4eaaa54ad56456fe7576 : SUCCESS in 3m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0d8ae3e4bca7403d8c40edc539d01270 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25e6f462ce02424b9aa577afe8548f92 : SUCCESS in 4m 26s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/8108d45f7e93475c81c64bfd89371b1d : FAILURE in 14m 14s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/2e47007069db41dba373398a51246291 : FAILURE in 11m 06s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/dd3421b17fc644eb86e7a0882b664257 : FAILURE in 17m 49s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/708c7d56cb414a2db9bd47b7c5c9b92d : FAILURE in 2h 09m 44s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/d691de19059b45c99b3902b1553b19fc : FAILURE in 13m 43s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/015649babf6d4f07a3a431aa1e92b161 : FAILURE in 8m 50s (non-voting)\n- tripleo-ci-centos-7-containers-undercloud-minion https://zuul.opendev.org/t/openstack/build/69376ff207e44417ba0b0fbe4f06c5ea : FAILURE in 11m 27s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/db1ebee5ecb246ddbdfbbc836854312a : FAILURE in 9m 41s (non-voting)\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/dd84f2a86aa44091b2745a17fd8c48de : FAILURE in 14m 00s\n- tripleo-ci-centos-7-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/80c27e31e2934741923ad1227d792f74 : FAILURE in 13m 54s (non-voting)\n- tripleo-ci-centos-7-scenario007-standalone https://zuul.opendev.org/t/openstack/build/fb7f71f544864b1fb04cea6a7802122e : FAILURE in 13m 58s\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/9886e3eef5414531ae7662a214a8b5aa : FAILURE in 8m 36s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/8e61f77bb30a4b1f9fd5396e3e90ecfa : FAILURE in 15m 04s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/ab534fc434df43cb83a678654e339e38 : FAILURE in 16m 03s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8517aaaaaee6455199e374ae6f51f0e8 : SUCCESS in 3m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c6170c97da754a38aad60a3b98eb761d : SUCCESS in 5m 51s","accounts_in_message":[],"_revision_number":1},{"id":"0ef43cc2c11e637cad0bf826a7bd401b169e4ce6","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2020-07-29 10:47:03.000000000","message":"Patch Set 1:\n\nLatest pip release break c7 jobs.\n\nbz: https://bugs.launchpad.net/tripleo/+bug/1889357\n\nWe have posted a patch to pin pip to last known good version.","accounts_in_message":[],"_revision_number":1},{"id":"38e33f31ae265677b27c23f1e141dc354fbb43d0","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-08-03 15:11:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"23ddafc0d9d870ee10c252fa1d475538d4c8ad3e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-03 18:36:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/aef4a1d8bc5249fc9ae405ce2e088ff1 : SUCCESS in 3h 23m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a945c653e457c55e57986d9d7b6a3830386f47ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 21:58:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9753a7081fe44b878b5103abbcbb2364 : SUCCESS in 11m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f39a29312a049baaf643173550609fe : SUCCESS in 3m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/50b1c6076a18497f86319ed8fee4a608 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ee9a6f5f872545f0b5aed8ef5519888a : SUCCESS in 4m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4188ed675ac2427cad1d12873fa6eb4d : SUCCESS in 6m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f0f16f3f0e84ea69ad7f9f710abfffa : SUCCESS in 5m 01s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/0ff0743cddf44004b0aa0ce78ad494ec : SUCCESS in 2h 32m 28s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/b1bf15fa6083491db71b488793fa125a : SUCCESS in 3h 04m 53s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/8c127292a8b046d6b4265cda854f7632 : SUCCESS in 2h 51m 29s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/1d0c33597dc9401a840f3ed72c923fbb : FAILURE in 45m 19s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/c281626cd12e4894879ebefac268ac16 : SUCCESS in 1h 49m 28s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/864c8eaa95bf4b78a7d7d3b90be2c34c : SUCCESS in 1h 37m 48s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/2e4dd0d0c7584fa19bc9db7d86546439 : SUCCESS in 2h 03m 34s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/0808bed83fb0423684f736de14804370 : FAILURE in 2h 11m 30s (non-voting)\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/99027628791a4f009673f528c18ee5f9 : SUCCESS in 2h 05m 17s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/c13638578081431c88c79dde6f8ac403 : SUCCESS in 1h 45m 47s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/7b578b966e904e758043402a25a635c2 : SUCCESS in 1h 34m 02s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/464699ca065b43b0adb6023525dda919 : SUCCESS in 1h 18m 56s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/4d9f1abb83bc4eb38114e4d0177ae3b7 : SUCCESS in 1h 21m 31s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/6a9306433d844ee58641a2c4f3ec4fa9 : SUCCESS in 1h 24m 07s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/99d3c523678149e18b6b08f29cec5533 : SUCCESS in 1h 23m 05s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e1ed6d951a404408b780cab67eb0d4fc : SUCCESS in 4m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/209b5e2407454c23bb065dd6466f1370 : SUCCESS in 7m 18s","accounts_in_message":[],"_revision_number":1},{"id":"18d1825462508867eedc70e2ec93dc158ad0f9fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 21:58:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fbb33432aed819f46b862e199e2b65ab79a3dbfb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 09:36:47.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d500a3a49624cdd80f42ace31ec5f28 : SUCCESS in 5m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/99856adfbc2e4fcdb09d7b5a89f3238a : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4f07066b31504e1a96088ba0e324611a : SUCCESS in 5m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/31f3a1a87b254042970668b5f4516482 : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/285384b4db9140a2b3d9fd75a29b4c0d : SUCCESS in 6m 07s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/ebd4e71e3ddc4b9fb864207188f84bb3 : SUCCESS in 3h 15m 23s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/bf749af817674449828a8b1cf381aecc : SUCCESS in 2h 23m 46s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/5754d97d0a4141779b2451d456a3d24a : SUCCESS in 2h 04m 22s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/5fdaeec9496f4909b5d40288bd60191c : SUCCESS in 3h 09m 08s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/4fe07ba146924b5497559c1df166926f : SUCCESS in 3h 13m 34s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/86708446a4f244dd8f31e3dde98a9534 : SUCCESS in 2h 37m 44s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/6a94d93fe2c3406e97f011fa0f569a45 : FAILURE in 2h 49m 22s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/5ef172b7973f403f915668204c7b0e76 : SUCCESS in 1h 54m 42s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/5773b21b6d1940a4963667182936c385 : SUCCESS in 2h 13m 02s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/7de1af4427e7406a87d6d767035d23e7 : SUCCESS in 1h 18m 53s\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/b1fec43c9cfc4c59a1ceab9d0590e44e : SUCCESS in 1h 10m 36s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/384900678d8c45df9f17b7c1933a3fe1 : SUCCESS in 3m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5b9911c389554b38a1a572c342f89227 : SUCCESS in 6m 36s","accounts_in_message":[],"_revision_number":1},{"id":"44aebc4519bfff93b0b5b380a3e7305921c0c310","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-08-04 12:51:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2e9ff08480b1e4fb7130b8f1e2ae4d34be043e54","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-04 15:52:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/cfc5fa33ff6d4783b2b6c1c4323cadb6 : FAILURE in 3h 00m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9e5c3c0eb2648c1e9674c29445d1d9072a176cc0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 16:07:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ec0df25072294262af76486950adc32d : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d85f5a79315d40a9b6dfe174121de56d : SUCCESS in 6m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7675b6999b264feca1e0f993e66c6e1b : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c7efca5030fe48c1a214d1add2bf6bfc : SUCCESS in 5m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4adb72e203ad47a4865f615e0de14ca5 : SUCCESS in 6m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93de309bcc5a4f278376adc8d5c952f9 : SUCCESS in 5m 07s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/a89abc85b647434b9765714fbed704e2 : SUCCESS in 2h 34m 21s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/24ad019ec8214f66af12febd76202391 : SUCCESS in 2h 28m 53s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/2ba09f7ec0544c56a23234044f8a1217 : SUCCESS in 3h 01m 03s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/b21dcc64a2bb4c7bb82dee0ef0bf45b7 : FAILURE in 2h 03m 17s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/e560f989b1164d5083458beae22677d8 : SUCCESS in 2h 13m 21s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/19ccadb11d00448dbbf486323e3f305c : SUCCESS in 2h 24m 13s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/a799c5d20417453c864be436d74f7aa3 : SUCCESS in 1h 57m 23s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/d5a5cf15d94e46ec8785140a2388ce39 : FAILURE in 2h 04m 05s (non-voting)\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/cba449bb951542f78de431843299fa39 : SUCCESS in 1h 49m 56s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/931f5749650042c6ae824e5dd75e99ff : SUCCESS in 1h 53m 43s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/91d0436a937c4f2caf6b3e8da44395af : SUCCESS in 1h 53m 30s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/74b2c88f6406415e9464a71bf74fd843 : SUCCESS in 2h 02m 35s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/8186a46928e64ccdb1f7ed1da50af301 : FAILURE in 2h 19m 05s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/e71f842a7c7440749184077e1142db0d : SUCCESS in 1h 25m 18s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/de55ae8e71df4819bc01583bee132611 : SUCCESS in 1h 23m 01s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d093b6033d5e48b9a3e65afd36213a3e : SUCCESS in 4m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/af4b0ed3b929471ab536a081691fb9e8 : SUCCESS in 7m 34s","accounts_in_message":[],"_revision_number":1},{"id":"420ac639b436c25d227ef1eae2a07f6e5a6b9776","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 16:08:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8024bc0c73636f29aba3765addfda95bfc96bca9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 02:13:05.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebd1e3af2c5a4b00b13a2d867a2dc081 : SUCCESS in 3m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/aba073caedbb456283c4a8b15727ef65 : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eeb9e14a001d495e9a15afa47f2551b7 : SUCCESS in 4m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/354ece82b55a4e24b0848834830e569c : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5f32ce47b944750a39603271e58dc79 : SUCCESS in 4m 11s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/19791f640ed64680aa3c36e519be0a8d : TIMED_OUT in 3h 23m 34s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/b0087d88440c4252946492f5877d707b : SUCCESS in 2h 32m 17s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/6d9df9a203524023aef9b3b730bba78e : SUCCESS in 1h 41m 09s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/7b250bc1f911412799dc3eaed1d50052 : SUCCESS in 2h 07m 32s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/bd98fc11cf454afbb1e399cc58fde754 : SUCCESS in 2h 20m 19s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/501af2a1816d44c7acb6318b13f32c18 : SUCCESS in 1h 50m 36s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/aea8745dde9d4019920893574cf754ec : SUCCESS in 1h 51m 16s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/bbfd79d974114a9bbf465a102450b1a9 : SUCCESS in 1h 06m 32s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/005f1c292edc46aeb60b2ed1db49f6d1 : SUCCESS in 1h 31m 36s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/8995c4885c3e47f78323a3f916aabb5a : SUCCESS in 1h 12m 44s\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/58d62ccaccbf4d4b9e48c3d991f72e40 : SUCCESS in 1h 31m 42s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d6e97f4aeece43248415f17ed6329aff : SUCCESS in 3m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dc0a3afc9fde451b94d53c4400a2ba6e : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":1},{"id":"72f6d9cadc54cf74cc33adfe4179b6e281d59351","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-08-05 13:08:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"464f10fc4486c202e39c210a13c5e4a6a7daaac4","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-05 16:10:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/89473f49d7a14d34a27dca21b1a407d9 : SUCCESS in 3h 00m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"30740ff7607fc1c5e1caba1827a7f510df7c433e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 16:52:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/050890313b874a428a4bb61220013ca7 : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b15c1fa6ee3e4ea6b5ef6024da6bcb2c : SUCCESS in 4m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1df89e3c99a844898167725aeececc17 : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dcea684c246a45ad917733cf8c7643c8 : SUCCESS in 5m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/400ea4f382a3427e9dd8f472f9a788a8 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccd15d02c04d44b99bdc270b1afcf7e5 : SUCCESS in 6m 31s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/64d40feb989e4f51950fbbe5abe96a5a : TIMED_OUT in 3h 26m 14s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/92006299f94b418f93e638500274558c : FAILURE in 1h 37m 58s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/74218d0c138246008da3fe99a6349551 : FAILURE in 1h 21m 30s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/ff548001f5ea47ffb9928acb859f6f13 : FAILURE in 1h 43m 56s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/80975c3a220d44209473208ff6b079a1 : SUCCESS in 2h 20m 26s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/4ec7789a18ac48fb8ef984dca58cd3e9 : SUCCESS in 2h 25m 12s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/17a5149b2b2446d9a22a9c2cf13f13d9 : SUCCESS in 1h 55m 44s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/225843dbbd5c4ff6b49274be77591862 : FAILURE in 2h 43m 28s (non-voting)\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/dfdd924e31f94c3a968c2f40c25d7e0f : SUCCESS in 2h 17m 58s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/e1d47ff98c6b4b8ea1bde5c6b3868790 : SUCCESS in 2h 34m 37s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/b71e03e2e33249d38358b4f8e92e55ed : SUCCESS in 2h 19m 20s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/a9789f6da3b14554abae81bcbbf53644 : SUCCESS in 2h 29m 02s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/58b09828cb364f3ea0db0dfb1f4d794a : SUCCESS in 1h 56m 21s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/45e4b537e7ff4f6a915c489713af5ae2 : SUCCESS in 1h 49m 54s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/fdfc5027a57c4fab84869b71ff19bfee : SUCCESS in 1h 40m 05s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5fc94ef707b74ae4a189254dda745a21 : SUCCESS in 4m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ff499f6165024ad3a6ef051ccc988a4d : SUCCESS in 10m 16s","accounts_in_message":[],"_revision_number":1},{"id":"0fb93f53107bdc6b0faea27114efe0eb1c5054e9","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-08-06 13:46:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5f47e5e9a61679afdb0cf79976c3611623a6f0ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 17:02:30.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/447b7121dbf548fb93914f6523ad1dc2 : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ef476ff1b0546638ba0ea861cb2f523 : SUCCESS in 3m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dc200fe69a0848198d0a0f6b8bf37caa : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b4ba465b1ba44becae21791941576348 : SUCCESS in 4m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/343748fe0ebf4e7195d67da41cedaaa4 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31c90d8683304483ae40db548b15decf : SUCCESS in 4m 12s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/52b3adac20784f90a16f3b13cf39e2d7 : SUCCESS in 2h 30m 55s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/fd3fcac75fd2420c9455ff4c3a87a855 : SUCCESS in 2h 47m 44s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/2d7da19c2a7c443a97278d6b01d99cbc : SUCCESS in 2h 26m 49s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/93c3fbef664f485e9ea4e30cfa610e2b : FAILURE in 2h 12m 28s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/9c3ff1b128be47faaa7f64b3fcfe38c3 : FAILURE in 1h 27m 01s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/f691280648fe477e820974ede475b97e : SUCCESS in 2h 50m 20s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/b0d891b01dcf4119bc64bee812b75ee1 : FAILURE in 1h 54m 06s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/6816c929d8384deabb0c0b8cedfbe15b : FAILURE in 2h 23m 00s (non-voting)\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/f48af35e34ec4d08b59fb7b898a887ed : FAILURE in 1h 51m 11s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/4df3d2cc226948229947a3621d782254 : SUCCESS in 1h 45m 26s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/e4feee94285e4b0da432a203236852c0 : FAILURE in 1h 37m 59s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/388559b0925746fcaafaa6a78fe0a664 : FAILURE in 29m 48s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/7b9e1d8f692146eb98789e0d9108c1bd : FAILURE in 25m 34s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/3cf9af625ebe46c98550460dad0323c4 : SUCCESS in 1h 33m 44s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/b7ecafbf30cc403b8b29436b77bd1247 : SUCCESS in 1h 25m 48s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1b2f268a49de4e71b4d1958161945019 : SUCCESS in 3m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ed4709a8b3a749a689855f99eba430f0 : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":1},{"id":"1ba8e8beb3f545ec38aadc9aa4c425cff8b9091e","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-08-06 17:35:19.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"12cc1dbeb4e5e3bc275ef2f782b809669960f1a7","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-08-06 17:35:21.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1e7590e9115044f2f92ddd89e792d86a5adaee1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 20:38:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/41fbe7a419f84cc9ba8efe02f2cc9646 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81b479a1b3b24782af54988e3a509389 : SUCCESS in 4m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1d436aa108cc4d3487c967be0897c19c : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d69498e5bf14bac8138c4338d58bb0c : SUCCESS in 4m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a84416b6467f45e7834a815e28036e99 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/264e88bd098946868928b86e8f4d4ca9 : SUCCESS in 5m 13s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/2b2d15455d614b9ab8f1d82032000163 : SUCCESS in 2h 10m 09s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/939fc588c95d4a759a84969a73d0ea79 : SUCCESS in 2h 53m 23s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/6e03976a9819400183948b18be540ea9 : SUCCESS in 2h 05m 52s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/172a462c14944b2f876b657e52ae53c0 : FAILURE in 1h 57m 21s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/b525590431b647f78c3566cf3c352003 : SUCCESS in 1h 28m 59s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/7d6b66bda91344aa966a40f2164774ff : SUCCESS in 2h 08m 46s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/44cbca84a7df47a4b4adfdd37491b4b7 : SUCCESS in 1h 36m 34s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/bc6c9b57ea3a40f28978b95eb2479de0 : FAILURE in 2h 08m 36s (non-voting)\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/eeea6a1bf5024913bc6d0a34919e2e5b : SUCCESS in 1h 40m 55s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/cea7e257217f4f0482c2475d36d92c1c : SUCCESS in 1h 54m 11s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/095ed1498d4c4f5890aefcb007f979c2 : SUCCESS in 1h 36m 08s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/8ee614edf80247258faaff3558c03fce : SUCCESS in 1h 51m 16s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/12b5d97d646241d0b24fc55ea3755c1f : SUCCESS in 1h 26m 44s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/84460d5b21334ca2a11800acb1adaaba : SUCCESS in 1h 20m 08s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/ccf2c7e77ecd436fbfbec0e7f4509585 : SUCCESS in 1h 17m 35s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a8fbbf1d70b84652991d5bcdc8ee9fb7 : SUCCESS in 3m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e3abfa9d290549fea85775e9c4230a40 : SUCCESS in 6m 46s","accounts_in_message":[],"_revision_number":2},{"id":"b2f6535813920842bc236805df6d07e984c1fb6b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 20:38:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"baf517717bbda2c8cabaa632059f78e54da1e4d3","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-06 21:10:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/18238ee77f284a77bfebf47d2bec23ba : SUCCESS in 3h 32m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"88c183adc8b1fed7d2cfd2fc2f9d40c4112f6e08","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-06 21:11:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/68c89cb7a4a4475b8906fefbb823dcd0 : SUCCESS in 3h 34m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b89530b506261ed760f6d333471d6a4f5b33539e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 23:34:10.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ab240697fdc4a9da4d5ce03476c69ac : SUCCESS in 6m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/76b431d46e16404098b9fe76619d8fac : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/50d5b1c8120c4a0bab9880201ee20e4a : SUCCESS in 5m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7634e49795484699aac7fdf0df79c1d2 : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e376a6f4bd424419ad33b46b4d759d12 : SUCCESS in 4m 24s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/56a96f091a614de28e1ee291b16ad873 : SUCCESS in 2h 48m 10s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/74f77af893c14b96a92f02ef7d9fbc50 : SUCCESS in 2h 25m 49s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/d8d1503a5eb0429fab0f53a8cd735920 : SUCCESS in 1h 27m 15s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/861dabee7e614a4f98e16c7d9444043d : SUCCESS in 2h 03m 00s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/70962c0cd2e141b0b9510c4b6ea9cced : SUCCESS in 2h 50m 53s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/4771707dc56248c0b425a1f14f6c1b03 : SUCCESS in 1h 40m 37s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/d21e6b743064447bb35bbde20ad84121 : SUCCESS in 1h 38m 31s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/52e0172c727e4f5ea2cd69bdfc369ac7 : SUCCESS in 1h 49m 02s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/23f1e3200060454190086eef85606310 : SUCCESS in 1h 42m 43s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/cb8fefd1b7624d15ae062bf8de93f4b4 : SUCCESS in 1h 12m 56s\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/436f7395184444fc96df86c916c0d8d7 : SUCCESS in 1h 17m 29s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c206334593c24629b11b773b1397522a : SUCCESS in 4m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8bd443fa836140f2aef78f4a3ba22298 : SUCCESS in 6m 27s","accounts_in_message":[],"_revision_number":2},{"id":"e32f44fd8357f98675318ef4f3ce1a9ba7a48abb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 23:34:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"8cc4f99a367c918f9165c01fa097fe72fb2f4725","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 23:35:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0447d75c929742c2b6224cc3d565227a : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e0c8d7770aa36aba353e5c70174c1f5ca7a59ae0","revisions":{"8c833c29ad2b66320e88de1fe25b2a4add08aefd":{"kind":"REWORK","_number":1,"created":"2020-07-27 18:18:54.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/01/743301/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/01/743301/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/01/743301/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/01/743301/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/01/743301/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/01/743301/1"}}},"commit":{"parents":[{"commit":"e91a484711d946a3f560b28398d5b493211da4c3","subject":"Fix stdout_callback","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/e91a484711d946a3f560b28398d5b493211da4c3"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2020-07-27 18:18:54.000000000","tz":0},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2020-07-27 18:18:54.000000000","tz":0},"subject":"Revert \"Fix stdout_callback\"","message":"Revert \"Fix stdout_callback\"\n\nThis reverts commit e91a484711d946a3f560b28398d5b493211da4c3.\n\nWe haven\u0027t switched to dense in tripleoclient yet so this\nmakes it a mixed output with profile_tasks instead of\ntripleo_profile_tasks. Let\u0027s revert this until we\u0027re ready\nto completely switch.\n\nChange-Id: I3c4e6e21259f97ea794111a9d68b1ffec1f733e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/8c833c29ad2b66320e88de1fe25b2a4add08aefd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/8c833c29ad2b66320e88de1fe25b2a4add08aefd"}]},"branch":"refs/heads/stable/train"},"e0c8d7770aa36aba353e5c70174c1f5ca7a59ae0":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-08-06 17:35:19.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/01/743301/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/01/743301/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/01/743301/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/01/743301/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/01/743301/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/01/743301/2"}}},"commit":{"parents":[{"commit":"bb678cea098d70128ff2856efb5e0f3752f499be","subject":"Merge \"Stop doing tag_from_label lookup with tag\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/bb678cea098d70128ff2856efb5e0f3752f499be"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2020-07-27 18:18:54.000000000","tz":0},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2020-08-06 17:35:19.000000000","tz":0},"subject":"Revert \"Fix stdout_callback\"","message":"Revert \"Fix stdout_callback\"\n\nThis reverts commit e91a484711d946a3f560b28398d5b493211da4c3.\n\nWe haven\u0027t switched to dense in tripleoclient yet so this\nmakes it a mixed output with profile_tasks instead of\ntripleo_profile_tasks. Let\u0027s revert this until we\u0027re ready\nto completely switch.\n\nChange-Id: I3c4e6e21259f97ea794111a9d68b1ffec1f733e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/e0c8d7770aa36aba353e5c70174c1f5ca7a59ae0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/e0c8d7770aa36aba353e5c70174c1f5ca7a59ae0"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
