)]}'
{"id":"openstack%2Ftripleo-common~743741","triplet_id":"openstack%2Ftripleo-common~stable%2Ftrain~Ic0bfdec756a23460f9257e78c508d167d21f0792","project":"openstack/tripleo-common","branch":"stable/train","hashtags":[],"change_id":"Ic0bfdec756a23460f9257e78c508d167d21f0792","subject":"container-images: add librados2 to gnocchi-base","status":"MERGED","created":"2020-07-29 12:18:50.000000000","updated":"2020-08-27 07:06:18.000000000","submitted":"2020-08-27 07:03:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"743741-1598511801447-038e877c","meta_rev_id":"248e8e4694b19c2b611411ee6f86739e8decc931","_number":743741,"virtual_id_number":743741,"owner":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-27 07:03:21.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-21 00:20:13.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"date":"2020-08-24 20:10:13.000000000","_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"date":"2020-08-26 13:02:47.000000000","_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"all":[{"value":2,"date":"2020-08-20 21:01:16.000000000","_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":1,"date":"2020-08-20 21:01:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2020-08-20 21:01:16.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2020-08-20 21:01:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2020-08-20 21:01:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":1,"date":"2020-08-20 21:01:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},{"value":1,"date":"2020-08-21 12:11:37.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-30 05:17:41.000000000","updated_by":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"reviewer":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"state":"REVIEWER"},{"updated":"2020-07-30 05:20:17.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"},{"updated":"2020-07-30 05:51:03.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2020-07-30 19:53:03.000000000","updated_by":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"reviewer":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"state":"REVIEWER"},{"updated":"2020-08-03 10:50:15.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2020-08-03 11:27:44.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2020-08-03 12:55:22.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-08-12 05:49:14.000000000","updated_by":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"reviewer":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"state":"REVIEWER"},{"updated":"2020-08-21 00:20:13.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-21 12:11:37.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-08-24 20:10:13.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2020-08-27 07:03:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"69466130de6138a05a014264d4e29cc4f03d703a","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2020-07-29 12:18:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"58a7165affd49eb3fc4a4092e37ac42236626cff","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-07-29 12:43:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-rhel-8-buildimage-overcloud-full-train https://review.rdoproject.org/zuul/build/94aab7bf9728441eb0c21e9d41defb9e : RETRY_LIMIT in 6m 17s\n- tripleo-rhel-8-buildimage-ironic-python-agent-train https://review.rdoproject.org/zuul/build/ef59ee4fb0994365a154f1e6d6c749b9 : RETRY_LIMIT in 5m 58s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/4d316f45df354db5b550e4177aaadff3 : FAILURE in 15m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"82ef3ba14559614a342fcd6400a5620021c3eab5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 16:13:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cce275740e6f43bb9154550106b20a99 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a4eb262a5214cb787e89660f1d00742 : SUCCESS in 3m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bb5b5ea96d4745ccb1e369d4d0cf7aa6 : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/67bcb03e756d42eda3a7496eab89e18e : SUCCESS in 4m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a233158837e844e58a65d3a4b0309309 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfd8ef471ea04bafb202b48f6eb43ba4 : SUCCESS in 3m 47s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/47b2c9815a2f41589240abb30a7652de : SUCCESS in 2h 21m 02s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/cd6855fb7fae4e9ba1528958479062db : FAILURE in 9m 29s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/ba8854a741564fb4b25042445cafde34 : FAILURE in 12m 40s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/ec13826216474fc8ba2c2699c0e9e841 : FAILURE in 12m 01s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/584e9028e8584a81b444d0f5c658428b : FAILURE in 19m 01s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/3a9f3ec7143e499d88b20d942744e0b7 : FAILURE in 2h 08m 39s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/b77d7cc101d74ddda31c50e15ae36c34 : FAILURE in 52m 31s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/ae65813ae567431e9437f50a8e087fd3 : FAILURE in 15m 06s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/e031d630fde144c8851dba47c0565cde : FAILURE in 16m 57s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/7a821ad7999e45df9b2b47158e7fb5a5 : SUCCESS in 1h 54m 55s\n- tripleo-ci-centos-7-containers-undercloud-minion https://zuul.opendev.org/t/openstack/build/e3b96120c83140968455e13e7ca0ca66 : FAILURE in 11m 04s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/3b5d59fbe02e4770a6c0be9792a3bbfa : FAILURE in 13m 37s (non-voting)\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/56450287c40c40a4a2b6c100ae830e0b : SUCCESS in 43m 14s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/e140a3aaadbd4fa899cb0b6cfe6f0fdb : SUCCESS in 53m 12s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/2f040b651ae34c2f9330f7a33c785d28 : SUCCESS in 38m 28s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/b4f773d538af43029621f230663c03b4 : SUCCESS in 33m 01s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/88bb525f0d8e497697c4f77cc9161cb3 : FAILURE in 1h 07m 11s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/69dac81be2dc492dab043bd310aa5915 : SUCCESS in 1h 40m 40s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/2694fad5db1544729e5fbf31ec92fa7b : SUCCESS in 1h 18m 07s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/cc1000b7104d49d99b770abcbdb53540 : SUCCESS in 2h 06m 53s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/d080a0353cc946e09df20b9971790161 : SUCCESS in 2h 02m 34s (non-voting)\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/7049a5e8212f4e7ab65c33be1b3cfbd6 : FAILURE in 13m 34s\n- tripleo-ci-centos-7-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/332e97637d914c7b951f289be0e9662c : FAILURE in 8m 28s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/7113b28ea54a44778915d774296d1882 : SUCCESS in 1h 31m 10s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/edbfdca66c784be7b55e7ff030318b4a : SUCCESS in 1h 23m 41s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/e8d575c22a10496fb97ec6a4cfa13a4b : SUCCESS in 1h 04m 26s\n- tripleo-ci-centos-7-scenario007-standalone https://zuul.opendev.org/t/openstack/build/423ce6365ee84b64a401102f6c863c2e : FAILURE in 10m 03s\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/2a8f96a2a9f44ea0ae12dc044bfad106 : FAILURE in 16m 34s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/4584da1eca5c4e2987be543148c4785c : FAILURE in 14m 21s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/c96ae277f8d941ed846a279803906d95 : FAILURE in 8m 55s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/f55c747cc6d14b2da5db1e1af022ac5f : FAILURE in 10m 02s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fb775bfa058f432c9dd260ffd4ae39c1 : SUCCESS in 3m 56s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/37c79ccaba70494db9effe4af69a1755 : FAILURE in 12m 06s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/32cd933cfbd24616b4cb1aefddbc870f : SUCCESS in 6m 13s","accounts_in_message":[],"_revision_number":1},{"id":"31359ca9eff45280fd8d7674ec16e8b39aea68cc","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-07-30 05:17:18.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b082e656a3f467c2569be592c181173dbf243e19","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-07-30 05:20:17.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a7aabf1bea066542142c43170824d009aa417df8","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-07-30 05:51:03.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ba65c6a78f121fcff981971397956d4e4e9b83ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 08:47:20.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd5d7a9e39f743c2a2db5edc6770ee47 : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7af2f66e9e4c41b394eeefb9549782de : SUCCESS in 3m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/028be30b28084217a9126e894d3c6b25 : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6cdff13c3999420cac28385ce284b1a8 : SUCCESS in 5m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/71a28c1516fb44fa8fa6bcd5f687aa22 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/585968c16860493ebc08e6d9c01cbfb0 : SUCCESS in 3m 51s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/02784742226d4f68bcc949ab900303e6 : SUCCESS in 3h 11m 34s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/24693faa9676493187acaf9dc5e4f97f : SUCCESS in 2h 19m 12s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/aa47d5f7d3d34f1ebe8ff16745a6aae9 : SUCCESS in 2h 43m 46s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/af0ed82114394be3a3967215002360ea : SUCCESS in 2h 01m 03s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/b797be308e8d4f94ac8ca87c923a9e59 : SUCCESS in 2h 23m 20s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/c38276b44f274fdb93e243d40db75a7a : FAILURE in 2h 01m 56s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/d7336ad605e64ed3a51c88ed6a1094b4 : SUCCESS in 2h 04m 58s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/78407746500744fba4b2f9f3e840fa5f : SUCCESS in 1h 42m 53s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/2aa2781716a7470fb303435479aa41e7 : SUCCESS in 2h 11m 42s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/4aa9889a41ed42578fefad6ec3934787 : SUCCESS in 1h 53m 41s\n- tripleo-ci-centos-7-containers-undercloud-minion https://zuul.opendev.org/t/openstack/build/52182032707542a097b03b31fb9f8b05 : SUCCESS in 1h 48m 32s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/33ceddeaf8ab4a738685ea945413480e : FAILURE in 2h 47m 10s (non-voting)\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/3b5b9b7cf4764e7f9d34529cc195ae15 : SUCCESS in 46m 45s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/4f2078a355934d80bc91ebb56d4bcf8d : SUCCESS in 1h 29m 20s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/0abd414cd27546d5b6e4824539d0f66e : SUCCESS in 32m 13s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/136e6d38c105469d83e6a93f249535a2 : SUCCESS in 39m 59s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/90e83bacf6694b5b8c780a393577ea58 : FAILURE in 1h 08m 12s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/562d487f77934a569c5844acb8f1db6c : SUCCESS in 1h 32m 29s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/c43e41a5dcd444b691d2c33eb9e50d21 : SUCCESS in 1h 05m 16s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/eeef7285532740d5a022962b5954067a : SUCCESS in 1h 39m 00s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/dde2d740f5024c5e8a701b6a7032e528 : FAILURE in 2h 25m 44s (non-voting)\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/44daa3e56d044abda81ad0bae333e3db : SUCCESS in 2h 34m 08s\n- tripleo-ci-centos-7-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/c58edc19482e4dd196a5dc83cee651ab : SUCCESS in 1h 28m 15s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/9489473a56ae4444ba5bccd835a8746e : SUCCESS in 1h 17m 25s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/0190b63885bd45eead9e9000b5fc8868 : SUCCESS in 1h 25m 59s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/3470c99c2b6e48cbaa3e322ea4cf416b : SUCCESS in 1h 48m 41s\n- tripleo-ci-centos-7-scenario007-standalone https://zuul.opendev.org/t/openstack/build/aeeb0061794246838caeb857b99b72dc : SUCCESS in 1h 13m 38s\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/8d2ccd08ca23417bb0ba9fd2a7bea4a8 : SUCCESS in 1h 41m 11s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/f3871364f90f47ad9b6786b99a1a0443 : SUCCESS in 2h 02m 57s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/5b93416866ec46e8a28bdddba58cc95f : SUCCESS in 1h 21m 00s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/43df6f4d441840909178c32204a4401e : SUCCESS in 1h 36m 15s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/00bdf544b515465194f19e58320c1f79 : SUCCESS in 3m 07s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/ccff2bafbd9d455cbfe0eba6250ad102 : SUCCESS in 2h 29m 47s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/de938072fc954f9ca96c17eea0a27df6 : SUCCESS in 5m 57s","accounts_in_message":[],"_revision_number":1},{"id":"e6689d655d9c92c301d88ec0e6f375c2f244bbf5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-07-30 08:50:12.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-rhel-8-buildimage-overcloud-full-train https://review.rdoproject.org/zuul/build/5f0b014cfe9a406f9b5548094850dfec : RETRY_LIMIT in 7m 03s\n- tripleo-rhel-8-buildimage-ironic-python-agent-train https://review.rdoproject.org/zuul/build/e815e52c503041f3b4c03af8f259058e : RETRY_LIMIT in 6m 38s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/4b54a6bd88914cc5896a8aed42c9a15a : SUCCESS in 3h 31m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3f8fd720720c0624030d9a699fa596decb2323d7","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2020-07-30 14:08:12.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fad05e22d6aa80ba8d38298757309681120701bf","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2020-07-30 14:08:25.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ae3595edcfa8c90db21cae5ec6282530fc74767f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-07-30 17:30:23.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-rhel-8-buildimage-overcloud-full-train https://review.rdoproject.org/zuul/build/b5a2ef7123e947819bbefb939c32d0e6 : RETRY_LIMIT in 7m 36s\n- tripleo-rhel-8-buildimage-ironic-python-agent-train https://review.rdoproject.org/zuul/build/dd425034444d432fadb285b2efed8a92 : RETRY_LIMIT in 6m 37s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/cc71d81e417d4bb99d6f5c783b582dbb : FAILURE in 3h 02m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"af8e6cfbd1dfc3fa331ff62bb5f854be965523ab","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2020-07-30 19:53:03.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ab46b151fa694eefea9fbc05a95e1103732a5f87","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2020-07-31 00:10:23.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5067d77af2837df450b95ee4a09faf12451703e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 02:27:52.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e48a70e82a194c19ab4a9387826395c9 : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50be7274512b476190e655fc79e3b847 : SUCCESS in 3m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/41b00ac2596f4ff9802714c0bc6ef58f : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/243b14e9c0c34e368f3dfe57c23dc95e : SUCCESS in 5m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/445a05b264364839845e799152fc25af : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b403cfc4c2ac4b6b8b0ab330d1c3d1e4 : SUCCESS in 4m 50s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/e6b1f5ca38e64a7da34b60e6f181f068 : SUCCESS in 2h 22m 35s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/92851dc5b03a422daca8394a511398cf : SUCCESS in 1h 57m 32s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/a69812d9db624cb7b7ed8a90dc1523eb : SUCCESS in 2h 46m 10s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/8666fbb00822467c8eb6ee7f7a15ea5c : SUCCESS in 1h 57m 57s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/90814fddd67f4f91969f4f051a523c0d : SUCCESS in 2h 22m 34s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/fbf28cd8b35641f5a57360ec3ef6544b : FAILURE in 2h 17m 15s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/cf031ca613dd476bad0ba9734b6fcd16 : SUCCESS in 1h 50m 32s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/9ccc717f7c934ce7a089258a7be75b13 : SUCCESS in 1h 53m 41s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/54306b2fed4047b8aed222c2de9d3e16 : SUCCESS in 2h 41m 03s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/2ced130f5a894378bd7a3a1b673f74bb : SUCCESS in 1h 47m 27s\n- tripleo-ci-centos-7-containers-undercloud-minion https://zuul.opendev.org/t/openstack/build/7095040df8a4437baa95f98b16b1c89b : SUCCESS in 1h 57m 11s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/c9e53fa86cf241dca4609b179b68d980 : FAILURE in 3h 04m 44s (non-voting)\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/55b23b303db34b2385806698d0bbe2c5 : SUCCESS in 59m 35s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/1161874c6a91458eb5b5ae4b3749c5fe : SUCCESS in 1h 06m 18s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/64b3243ed857452d9054c2064a9bac38 : SUCCESS in 30m 27s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/b65cab6851f34f7382a1bf2ca36a629a : SUCCESS in 43m 53s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/27f94e5522e9404c96f7864a5f54f232 : FAILURE in 1h 02m 20s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/2bbb5ebe2a4642ccac2bd76e0f704310 : SUCCESS in 1h 56m 33s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/0c6572541e634020b0d4d45a3513f29e : SUCCESS in 1h 09m 07s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/1ff098b526c74f378b27fdb67e4e3d9a : SUCCESS in 1h 08m 41s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/d212b130bb5c4902aa4d1a7efc444691 : SUCCESS in 2h 20m 12s (non-voting)\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/a651bdf6da0e4d5d9115c011eae4a4ab : SUCCESS in 1h 53m 08s\n- tripleo-ci-centos-7-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/df3724b1f49342cea849711cdfb997aa : SUCCESS in 1h 44m 02s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/deae174e9cca4436a1ba67459d560818 : FAILURE in 2h 30m 30s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/3ee4506a6b1349759d62b65fbdf82277 : SUCCESS in 1h 14m 18s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/2094a20d3e324ba598c0d7bc9cc032e1 : FAILURE in 1h 36m 42s\n- tripleo-ci-centos-7-scenario007-standalone https://zuul.opendev.org/t/openstack/build/c35efcc420dd4aeb9f08f4d7bb4d969e : SUCCESS in 1h 08m 00s\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/e459a71ef46b4467be207f4212c97eb7 : SUCCESS in 1h 41m 46s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/75b6d5f66a0749d7a4076d804eed284a : SUCCESS in 1h 51m 26s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/242810648e284e788fd5adc7b772285f : SUCCESS in 1h 31m 09s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/32c7513b55724a87a195b484b8611721 : SUCCESS in 2h 03m 39s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a60787cbc0d54affbede137ed44b42c5 : SUCCESS in 3m 47s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/0c8c0439d5624f2b8854b2fc76e1fb92 : SUCCESS in 2h 47m 38s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/534886ed5214475fb3c474be204f537f : SUCCESS in 6m 18s","accounts_in_message":[],"_revision_number":1},{"id":"f16de82cc68ad29894aa93db01c280080abc7af2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-07-31 03:00:52.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-rhel-8-buildimage-overcloud-full-train https://review.rdoproject.org/zuul/build/1ee24af2d7fd431296ef973aa6c0f3ee : RETRY_LIMIT in 7m 22s\n- tripleo-rhel-8-buildimage-ironic-python-agent-train https://review.rdoproject.org/zuul/build/4dfb1256bc7b448285199fc69b64b049 : RETRY_LIMIT in 6m 00s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/1f34e646a2e24b8d8b2b8dafd7a4d817 : FAILURE in 2h 49m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0e90a1135d33fc859ee4c0ea570734b0d056d9f2","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-07-31 13:30:21.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"993d15439ff6b9b428cc0c79480532f0987ba4d9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-07-31 17:00:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/009981ce86d84e46a4a9dcfd61944be1 : SUCCESS in 3h 28m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bedbcd4bb90a9a080ab9bfc2a264ea899b256887","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 21:22:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2379322bcbf04bdeb69c7e95d02f3590 : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfd3af422b6c4e57b3b40304df22aeab : SUCCESS in 5m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9f9b5b825c6a4eec84c1e67280ab6177 : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/567a2357c0564f4c81e2b436b5e642a9 : SUCCESS in 4m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/11d71dbff9574a4b8de7de25fe37f882 : SUCCESS in 6m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2f2207cb63c44309db9a16ee90546c6 : SUCCESS in 5m 15s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/d6d1e8fdf9664d23af1b82ba03ac0870 : SUCCESS in 2h 23m 47s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/0feeec2788fd462a9cf888a0e989350e : SUCCESS in 2h 10m 22s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/a21073cd934248e4889a2a631170b1ce : SUCCESS in 2h 48m 40s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/2cceeed8561e4b8486e0dcec524c4c43 : SUCCESS in 1h 39m 09s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/1a57e20cebf84844a3f719db0e8a9126 : SUCCESS in 1h 47m 32s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/2da0f189f5104a589dd7025cda5e9c95 : FAILURE in 2h 15m 18s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/6122ab5fe2e1476b9af88863ddef2adc : SUCCESS in 2h 03m 05s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/046c8c499eda4b22870aee0505f88fd6 : SUCCESS in 1h 53m 35s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/4ddedf43e6c64aaa95b682cc0eefb6e1 : SUCCESS in 2h 49m 45s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/b160f7e6de1847328ffffc8e4bfdebee : SUCCESS in 1h 46m 58s\n- tripleo-ci-centos-7-containers-undercloud-minion https://zuul.opendev.org/t/openstack/build/29892abad009493d85e962b130942457 : SUCCESS in 1h 44m 57s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/15c48884852542739ac7a4ce13359968 : FAILURE in 2h 22m 36s (non-voting)\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/fa1ff888da5d43eb92b6ebbe9ea89520 : SUCCESS in 1h 28m 14s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/101031c548254d16934a04479bad13d3 : SUCCESS in 58m 14s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/4a98ead876604c3dba53ac1bf3e9473f : SUCCESS in 51m 47s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/d87a7b0e629b4bafb5228be2ab4dd05a : SUCCESS in 30m 08s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/013d3aa6e0eb4a61bc5bca8935c5a001 : SUCCESS in 2h 50m 28s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/3ce580c1847f4507961cfdb03c0ff3c4 : SUCCESS in 2h 01m 41s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/9603537e42e246b9ba36b4ebf5ae7530 : SUCCESS in 1h 14m 04s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/3fe0bb605ced4242b1a5a2ca5a3ef49a : SUCCESS in 1h 27m 30s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/7c7b038211f345f0b4e3464d50c3a78e : SUCCESS in 2h 13m 27s (non-voting)\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/fa9e77f8cd90436b93e3948f4c43d68c : SUCCESS in 2h 17m 48s\n- tripleo-ci-centos-7-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/32e0c7b9a64f40d8bfd0bf5f392ed7ce : SUCCESS in 1h 02m 20s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/e44e082277fd4d8dbf6b9fe4af751cec : FAILURE in 1h 13m 23s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/902e9940953243f78e6da27f80e1ec05 : SUCCESS in 1h 26m 40s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/91e8d6fbc8364b05b5724889b8bfe1bc : SUCCESS in 1h 13m 32s\n- tripleo-ci-centos-7-scenario007-standalone https://zuul.opendev.org/t/openstack/build/c68b300c61a44ff0a5ad04c0cdc6ac02 : SUCCESS in 1h 19m 49s\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/9d9e61b69f644f5cac0017c15880b767 : SUCCESS in 2h 08m 49s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/4640b6eb703e4081bd2574ee728ae9da : SUCCESS in 1h 55m 32s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/abbd4dabb4284bc7bb4c10c00c67adc4 : SUCCESS in 1h 17m 51s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/07134f74447f4e4d94ddf02975268ce7 : SUCCESS in 1h 53m 32s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1ceda5632de14078a0876860873b6ddd : SUCCESS in 4m 05s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/f8773d0bcabb4c7888af4b0a83a794fa : SUCCESS in 2h 14m 44s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5c0d642b82c643de9cfa872af8933fd8 : SUCCESS in 5m 57s","accounts_in_message":[],"_revision_number":2},{"id":"a4572f0fb5bc545e6d8152b9166c602091b5a261","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2020-08-03 10:50:15.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4071917f8c9f90d55a16e890c1058de1cf560fbc","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-08-03 11:27:44.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"43b33c90bfff693099a5d1c4b50879d75c5a485f","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-08-03 12:55:22.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nwon\u0027t land until the other two things land","accounts_in_message":[],"_revision_number":2},{"id":"6febf31a96bb7b8382fe38e8611819071c6059ee","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-10 13:57:06.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"687a99361834113d4aa2633c431c8f78ade4e32f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-10 14:15:42.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/7e6f90a3e82d4e98ab22ae1f29b9dd0b : FAILURE in 16m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3cc701a2c08b7117c32df58c5886a6d369f046b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 17:40:42.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d34fe5287aae481fa6d9de5a114ff644 : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7518eda2d5e6496484c93d4cdadd4e2c : SUCCESS in 4m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1ca38d92b46a4423974291a92c743954 : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/335e4e299bec414fbab1d169738bd21a : SUCCESS in 5m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/136e7ca9ecab468b89db4f4dbf8c832e : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e6a8bef96ad430f94d3d5d45079e8a9 : SUCCESS in 7m 10s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/c1e51dc3d5a34daeb7d3369d12676ab3 : FAILURE in 1h 36m 20s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/d1439abb764e4dce94c4983792aa4ee8 : SUCCESS in 2h 48m 26s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/d327a7fed8fa431fb5ab9efb7fc24852 : SUCCESS in 2h 01m 55s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/ea2f235d6d024a0eb7d1630f7f623409 : SUCCESS in 2h 05m 36s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/d58b1e2e61284748ab0011818fa972fd : FAILURE in 2h 00m 24s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/6fde71d67ba2405f8e75316b894d12be : SUCCESS in 1h 54m 53s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/9335efe561b74c4a9ef43617f40d40ed : SUCCESS in 2h 42m 50s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/36b4754c1ebd4d79bddeb1007f2f18e8 : SUCCESS in 1h 52m 00s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/a488d3d4974f4e3e97088e47f294bfd2 : FAILURE in 2h 15m 13s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/7a18421f85d9403da04a47b4feb71541 : SUCCESS in 53m 00s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/8eb3b5b7f44d46ee987908a3b7fe9915 : SUCCESS in 39m 55s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/0cb637e7566c47b0b9e1369cd8331177 : SUCCESS in 2h 12m 48s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/4fe0fb8957374dffbd31cea8eb69b385 : SUCCESS in 1h 35m 02s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/1630e254873648b3908ad54b13f88aef : SUCCESS in 1h 32m 08s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/d60a54058f1c4f03a876bb7301c90fdf : SUCCESS in 1h 43m 43s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/7abca969c2cf4c3e9456d086d888a32b : SUCCESS in 2h 32m 12s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/1c3869e9b2664535a9ed2697e8482c34 : SUCCESS in 1h 34m 52s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/1721deeb525a487d8b8f35c5a0ab2d9f : SUCCESS in 1h 15m 07s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/4897a26e0241412ea514348ed44ea2ae : SUCCESS in 1h 08m 33s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a70aae384cb042d1ab83966f8d8050df : SUCCESS in 5m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6418f2d1ad84405f9235a094c28e5998 : SUCCESS in 6m 04s","accounts_in_message":[],"_revision_number":3},{"id":"0f66c72970a1b7dc552a5ee814a8e69e2eafa023","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-10 20:26:26.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"805f578ed4ff6e194dc99a96524dd2b013034908","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-11 00:00:00.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/33e947d7ccfd427cbe8463a58ed911e6 : FAILURE in 3h 31m 13s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"14578ed416d960a2685c38c244e872c150923537","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 00:21:55.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c89c865113fe44e48900cb4bf5f22aee : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8feee85886d841fbbbe60c55ac6da1fd : SUCCESS in 3m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f132513585404f21a0385e43016d91fe : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9771976adfbb48f9a170ea7c780bf993 : SUCCESS in 5m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a123c0b638d04212a483973c596ea5bc : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34cca836b0f046938d81e6cb9e71e735 : SUCCESS in 4m 36s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/67d7c27eee874f6bbe5e00955ec781c2 : SUCCESS in 2h 10m 45s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/47830bba41b04844bc663b3c90cb6158 : TIMED_OUT in 3h 31m 16s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/c0508f6ac6764e9caaa0bbfae1d1576e : SUCCESS in 2h 26m 08s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/5aeb652584884b75ba0e82dc20be1c9e : SUCCESS in 2h 14m 35s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/25fdd9198e2f4b59898e68eda186677b : FAILURE in 2h 50m 17s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/6e419fa38166476dae19986cdaa4caad : SUCCESS in 3h 03m 54s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/0abbe4dd32294687824f6bd340c44b4c : SUCCESS in 2h 29m 12s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/4465ee62c757415a8e915c2745e2c0b5 : SUCCESS in 2h 17m 22s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/abd2ff80b6314bdc8d199efd03f14bd4 : FAILURE in 2h 22m 53s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/fafb79f46f7646c1a2edd34f84d055bd : SUCCESS in 1h 08m 58s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/3fc4e7f2c765488fbe2e2cefac4aaa65 : SUCCESS in 47m 47s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/d71217a1020944688e94492aa4c53c3e : TIMED_OUT in 3h 22m 14s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/0764046822df4a8cb8ed9e78efa6e49b : SUCCESS in 2h 12m 01s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/c9759391d4734533880d2067e7c4fd88 : SUCCESS in 1h 06m 00s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/dc5d475ec1724339989128fbc6cc8f8b : SUCCESS in 1h 52m 10s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/b7794ca8ce6447e99e454fb5882d828f : SUCCESS in 2h 24m 27s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/1b1541deb796459bb16783858e52eb8b : SUCCESS in 1h 31m 59s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/7cf58f974328496f9f2fbaf5411d26f0 : SUCCESS in 1h 21m 09s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/a3d28dcfbf7c4796a2a55cf6bcbe15a7 : SUCCESS in 1h 14m 32s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/26ad77bf7d3e46fa98e344db9a0ae727 : SUCCESS in 3m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8a1e0684569644798bd172ecd24ee14f : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":4},{"id":"0f810d21e5051079885a4dc0d450f9a9da627020","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-11 12:37:50.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"1c28fcac040f0b6ae6e9dd3cb1bab438e1300276","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 16:22:15.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a556369e44c9414db9e696b346c695c7 : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9328a5b3e0644fcd897277b22a404f50 : SUCCESS in 3m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/44ac579367274a4da240f1ff5c3cb129 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ab3ba4004f44f9bb0fcb778473c5761 : SUCCESS in 5m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c03baddf5ceb41639015a0c2d967fae4 : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b0217f854144c698c3cb95b664603fd : SUCCESS in 6m 27s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/a64b2c2cfac845989328c7b439bf8cfd : SUCCESS in 2h 18m 34s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/425f915ef3584881a3ebc7f6e6d85df5 : FAILURE in 1h 19m 23s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/98506db4f3a148438312dc2f69224850 : FAILURE in 3h 41m 15s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/2a31e043f1d04caa9157d5f44276fdde : SUCCESS in 2h 41m 49s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/c0a851a062104277a7571e0ab83c897b : FAILURE in 2h 20m 53s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/c1ab45ad47fa4ed68e358b80bd546f37 : SUCCESS in 1h 45m 01s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/15e907e1bb8b4b43b9112fa0357d0e01 : SUCCESS in 2h 43m 39s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/b857902b24514baaa5f05fba3cf06316 : SUCCESS in 2h 05m 00s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/84eede4227dc46589b8cc55ac4901185 : FAILURE in 2h 09m 32s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/8ff1e1904f0f49e78ec2ab522852a4b5 : SUCCESS in 1h 16m 04s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/2e417c58954449a396ef1f202e3b7009 : SUCCESS in 46m 13s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/d5b6db51236e497e92370a520e29ac29 : SUCCESS in 1h 46m 41s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/234e7fc58cfc4a0292d65e4340bfae07 : FAILURE in 1h 15m 03s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/d3c911ae7a424468827bb300a5bd02a9 : SUCCESS in 1h 31m 43s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/aad129d859be4cd493282d7c81ce97c6 : SUCCESS in 2h 00m 21s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/2c92918b003e48c38f059c0d15b1541d : FAILURE in 1h 17m 06s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/9be765235d9b4a199eba4e6b58b4967b : SUCCESS in 1h 30m 57s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/16e8a821fe30495eb1bebf59e46258ff : SUCCESS in 1h 10m 18s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/c0d189d415be418e94a658bc93f07d2a : SUCCESS in 1h 11m 50s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/10d0f52ec8d2494ab6f1b186c26adf13 : SUCCESS in 4m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d1cde5fd7942442aa63ea895655cd0c5 : SUCCESS in 6m 12s","accounts_in_message":[],"_revision_number":5},{"id":"be51b106a16ce6007ba344c3bbe2f609644eb63a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-11 17:21:00.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/ccd6240738294597aa4f299b8f086cb8 : SUCCESS in 3h 29m 32s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"29f35f6200c2b831657841af046e33ac811e04ee","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2020-08-12 05:49:14.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"8df1af8229ce7f3ea7185acce665583369a98540","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-12 09:07:48.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/ca46b0016d4f42519b48a274511fd003 : POST_FAILURE in 3h 17m 33s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"27b046134cb5a8bef585cf746395f369f88ff21b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 09:19:00.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db470f74afa346abaee847b291535c3c : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/591a580a300441999e387b329626625e : SUCCESS in 3m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f4394fc7582d4854b8cb8c5816cb34a8 : SUCCESS in 3m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/450a2c76326e4fc794907a4334e33279 : SUCCESS in 4m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/97a246b6e9294ea589dfbbefc3ac5fe0 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7232ec90ebce40108283c7a45fc51997 : SUCCESS in 3m 50s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/1fafe91a892f46f89d47d10b6bf85c39 : FAILURE in 2h 39m 31s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/4b631dbbb48049b9a0ba6e312ab45744 : SUCCESS in 2h 55m 22s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/d2de9176948b434fa7a47879db934980 : SUCCESS in 1h 59m 53s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/cacc8ec9eb8640189bf82a94408c22c4 : SUCCESS in 2h 41m 49s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/e29c3130dd7c474ea74292cb6834a562 : TIMED_OUT in 3h 18m 41s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/cc74f725e0d540a5afe103ab997e9eea : SUCCESS in 2h 00m 35s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/5d4d8867f7974399842d48c39edaa41a : SUCCESS in 2h 19m 32s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/705596a737094c8abd291fed391eee9b : SUCCESS in 1h 59m 10s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/0d05d31c0f044503971182929b8eeba6 : FAILURE in 2h 08m 50s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/4c386bc129674fe69f581f2075a0b319 : SUCCESS in 1h 01m 15s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/2b1add9096d04133ac37042e00a56bfb : SUCCESS in 1h 02m 55s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/d514b5adac814fe89ff51234b085bd40 : SUCCESS in 2h 08m 41s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/f937c008f7ee47eba523a27091913bae : SUCCESS in 1h 44m 11s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/cb07b45c77214c93904b2e4a0d8e7a9c : SUCCESS in 1h 16m 04s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/858abdbab454422dbc1f0d0529b64344 : SUCCESS in 1h 56m 55s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/358ffff8981d46228f8dd4c147f75750 : SUCCESS in 1h 48m 45s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/8fd1933cae894051bfa9d8f61b340e98 : SUCCESS in 1h 24m 42s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/3a29554e8576451f952da5fd95110843 : SUCCESS in 1h 23m 08s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/d55f3962415844979d30b3712c0802ab : SUCCESS in 1h 05m 19s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/36c6b5e38e514a40a567036b5fa4cb45 : SUCCESS in 3m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1866d0686f264dc587626589fecc6ffb : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":5},{"id":"e44dd5a698f90612af712dc4e783ed40f6740007","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-13 01:32:40.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"6f56ad4a8e571b99727611c5a5415a20e3775f2c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-13 02:47:04.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/06f0fca901f8473dbde16eed3936544e : FAILURE in 1h 13m 00s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f1e182a01cd54a331c4a771841caf7852b37437a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 04:33:20.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cfa9b733a768437087dbe1a70903cb8e : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1a22cbdd9244938807439319a7caa1d : SUCCESS in 3m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ce4f93f9692e4f0cb45e7b4c2a6f0f95 : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f928ff1cf3354872b958eb677a1fbc30 : SUCCESS in 3m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9fb8038f2fe04c93aa6c216aa7bfbf3f : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9764e242553e413e8079a5531f929672 : SUCCESS in 3m 46s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/c6c0d44ac6ba4f698f1a506827ab3ade : SUCCESS in 2h 17m 46s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/5a4561f53aa64c8ca430a73949a70c9f : SUCCESS in 2h 39m 36s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/03b62faad28341e49f3068ea90e2c190 : SUCCESS in 1h 52m 58s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/ef7918d814ec4647b9df89e4f5ce049b : SUCCESS in 2h 18m 31s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/38fd50c99c724821ba342db4eb60f229 : FAILURE in 2h 06m 47s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/f7a719d0afc642edba9a4c1e9add3129 : SUCCESS in 1h 42m 49s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/c101ae08998c48c4b4f3a77d2f93eec7 : SUCCESS in 2h 21m 57s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/af18c90cece94196845bef1dbdbd0944 : SUCCESS in 2h 47m 55s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/c24b8a2327fa4d63a083fd419252f976 : FAILURE in 2h 01m 13s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/cd20d01a5a2a47749573fc2a9c4199c5 : TIMED_OUT in 2h 05m 51s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/0478c9dec93543baa9a03aa372df710c : FAILURE in 21m 09s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/7db7d7c08334493fb287f941968ac631 : FAILURE in 2h 33m 26s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/4d44f366512a4fb9bc8b5b4e6b831874 : FAILURE in 2h 37m 44s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/2e9ee1856b9946dbb8a5e0b93aa7a82a : SUCCESS in 2h 15m 41s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/543f8f5d46314b9a9d85882aff51c15d : SUCCESS in 1h 26m 48s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/4633c4ece945424db8d0a24a1c7d1127 : SUCCESS in 2h 00m 43s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/801bcf87dbd24a5b9df7739636031b3b : SUCCESS in 1h 23m 37s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/92aa3875b92f4de7854a9c34cbdeaf90 : SUCCESS in 2h 27m 54s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/d8d13b15004d465ab733186a99dc7b15 : SUCCESS in 1h 52m 01s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/037498741ecd4f0cb53ef8a8b954b176 : SUCCESS in 4m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5650927461674f83b03a80a6a94e78bb : SUCCESS in 6m 06s","accounts_in_message":[],"_revision_number":6},{"id":"0bf8c2a366149d78104ed142cfb0bdd8cb597edf","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-17 13:26:03.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"61867c985dc6dbc06a0dd11608d2c1d69f8431c5","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-17 15:10:53.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"aaa63163ddd67c25a5a28536f22a633a64de2ce5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-17 17:23:43.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/964bbfb13b514410b305c2f5bbf43a58 : FAILURE in 2h 10m 49s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"d30d1d38a4f179b24a4ab9b0b53ad1153125c792","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 21:19:51.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/176f202b51854e0cb60973514c79efe5 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d555eee08164e5abf3535da5a03790d : SUCCESS in 9m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2376a67ba5d84f538db5ff6a44b62139 : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ce5c7bbe10e4d709a1645be8ac238ad : SUCCESS in 4m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f13739bce0334a0da5837f30f799feed : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47f29be2e2a249c8a83215496d86f361 : SUCCESS in 4m 53s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/496bb9a556f946e1b1a939f8867d8662 : SUCCESS in 2h 35m 52s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/4108286799bc47d097c157e20bb7bf12 : SUCCESS in 2h 41m 12s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/5bc968ab1dce49529f2fd09df55912bc : SUCCESS in 1h 59m 09s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/c7c9de2ab9784751bf84649f8115c09b : SUCCESS in 2h 09m 04s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/76c6973dc96447ec93eda5322da21e51 : FAILURE in 2h 00m 44s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/226632728b074e6d978a4dcc108b8682 : FAILURE in 57m 00s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/2fe36e21d9e1466ebd20eeab37b88533 : SUCCESS in 2h 16m 02s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/1740824dd33f4b1287adac6aec9169b2 : SUCCESS in 1h 34m 38s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/09af638a0aee48bb87bce07007196155 : FAILURE in 2h 12m 46s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/c8a7d9efc87e4407a744cd53e3c43bc6 : SUCCESS in 1h 14m 48s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/9941f434c5594e3b9bf682d78d4dfa11 : FAILURE in 39m 36s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/d87cd57a067e48fb91edf0112e7cb27d : FAILURE in 1h 36m 39s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/043ba0248a8442af843eba8bfc087f08 : SUCCESS in 1h 36m 55s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/0c62068494ad4c66b22dac24487668aa : SUCCESS in 1h 16m 14s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/4ad778f3721e4e038cc51570ecdfe083 : SUCCESS in 1h 34m 09s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/8249362c08084b1a92df346489a99f97 : SUCCESS in 2h 12m 30s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/2d7d030d0c8443d6b16fdd80807256c1 : SUCCESS in 1h 23m 22s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/61241ef3f6ee4de3a8aaec91b4e88d7f : SUCCESS in 1h 27m 57s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/ca8fe25e8a014f609d8de3c6018c22c2 : SUCCESS in 1h 13m 46s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6caea928bd4349bb8bae7228f5657ca6 : SUCCESS in 4m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ae9c7a83840e417cad0763eba475ee6c : SUCCESS in 5m 39s","accounts_in_message":[],"_revision_number":8},{"id":"1c6bde8e61d853b254a274380c8e211ecb2d4535","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-18 12:18:21.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"60b052ff9a89cbb267fe5a544f3cf01639c6971f","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2020-08-18 12:59:56.000000000","message":"Patch Set 9: Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"31e0f3d9d31a008e5cba13d0e85a754070e389ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 15:13:47.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ebe4a3203f6148b5ac782f5a8ae9ecbc : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92b1e0ea07514216b9e90fa18e26d710 : SUCCESS in 3m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/57b5adbf0fd740a9b1091c1b026e7edd : SUCCESS in 6m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/661e107dae684fb997b30ae0ab17970a : SUCCESS in 5m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/347ca44c1ec844da9f74dccbd1cc3b95 : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6dc2b38b17a14c54bb47ba147c5731d6 : SUCCESS in 5m 16s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/cac01a03e4c347b5a5728796d479679c : FAILURE in 2h 37m 20s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/1e6e9d1a060f452b9b5cf80121a459a7 : SUCCESS in 2h 34m 00s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/91f7447136a34ff69f092beac501ae4e : SUCCESS in 1h 31m 54s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/894e205c435546f48c9a671f30dfd25d : SUCCESS in 2h 21m 30s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/6a3a5d22cc1d4a7899a300c97cc6180c : FAILURE in 2h 01m 00s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/0650fc3f0f5a405f8d85993692dbdd79 : SUCCESS in 1h 39m 51s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/7aceed08eeaf40a7b55ca8f2d3275fed : SUCCESS in 2h 09m 39s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/7469c4d606a94f92b843babf9603e00d : SUCCESS in 1h 50m 01s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/f351e6dc93114427ae43541e75f952f4 : FAILURE in 26m 52s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/cd1caef66e38437480f70463b2d711b5 : SUCCESS in 1h 31m 41s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/efe6d5f688584c48a9df2646fac62600 : SUCCESS in 41m 39s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/bd6fc0d24fd54bde84ae79109b2e60c7 : SUCCESS in 1h 54m 45s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/d3966834c0a24602b1e84499922ea05e : SUCCESS in 1h 24m 52s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/47653815a4e4478fa2c06e90de939cb8 : SUCCESS in 1h 07m 08s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/f9bcbbd1ac2b4ce986505b4bb8c245fe : SUCCESS in 1h 42m 20s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/e899953bfa9a40a8afbe96573f05c1ce : SUCCESS in 2h 11m 16s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/3cc6717e04624dc7a9462718ed9b8754 : SUCCESS in 1h 30m 56s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/934c415d7f38466b9c8ea24ff930073c : SUCCESS in 1h 17m 23s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/ea2d31cb8802466188abf1a0d0c6ebfc : SUCCESS in 1h 19m 08s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f94762869982401ab4c328dd06bd8e48 : SUCCESS in 4m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d724729f05af4d6fad95a1990edd27fa : SUCCESS in 6m 31s","accounts_in_message":[],"_revision_number":9},{"id":"a7854d78bb9e6998ad2a34f476d8684716772cb9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-18 15:47:24.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/4c23f51f4f4d4ef3a53745c5af211ba7 : SUCCESS in 3h 27m 20s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"3cb2a299cec9310f9d3d8574fc0a34d9d6caf841","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-08-19 17:43:20.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"642ebf856e82fe56761ad35cd068c9e5932e8a72","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-19 21:00:09.000000000","message":"Patch Set 9:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/b23676c32fed4994b95bbb566bd866ea : SUCCESS in 3h 15m 35s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"b015e1b81586df9a24c41e6a397a25076ce8047f","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-19 21:09:02.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"55a8e6913125da5dac8bc163ec0e1c375ecaf591","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-20 00:06:06.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/da1beb50f2c1478cb7757eec52fc3b05 : SUCCESS in 2h 55m 41s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"d841713811e6ec173e3bf98cf905d53bf049440c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 00:56:31.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/78db52f8114b43119e358b96c57c5e0a : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c914ea8996864e4bb2bcb6da2e9e562d : SUCCESS in 3m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1c7ddc0e71cd4e1bb1186c3912e086c7 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7aecadd0e99f440196a9c2fce0b18bc8 : SUCCESS in 3m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f072fc17d1b149f29b7c6cd5ce59801b : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8aab927df8a45918b8184a7b4e885e4 : SUCCESS in 3m 56s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/38ec69030b8b46cf9fbc88be76db0fa7 : SUCCESS in 2h 14m 26s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/d1d42ca5b132477899331e1142ca6676 : SUCCESS in 2h 15m 41s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/538b51da237a4319839742437c7e4c15 : SUCCESS in 1h 58m 37s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/56b235fbd1774bc092cac688ac10fe56 : SUCCESS in 2h 06m 53s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/6fb772112a1940028f61853954f99af1 : FAILURE in 1h 59m 37s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/2eb3eae6809849d59b40153d21417125 : SUCCESS in 1h 45m 51s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/7e19a9c3a2f24fc3b944db466acc7630 : SUCCESS in 2h 38m 58s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/88c6ec2b5dc4408ca84a75a8163430a2 : FAILURE in 2h 49m 42s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/25124a7bbfeb4e368f4be03393ac7d38 : FAILURE in 2h 26m 49s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/a9523d7dc1b641cfaad5139793531f1c : SUCCESS in 1h 02m 18s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/90311060413642088f537a34323383c4 : SUCCESS in 39m 03s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/195bde8c17634f73bfce23dd39811c52 : SUCCESS in 1h 49m 51s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/bafdd8a4274b4b70a6f7b34ab514ac34 : SUCCESS in 1h 20m 02s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/86dbce9d1d8e4db788aafb4a51e0dbe9 : SUCCESS in 1h 04m 01s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/863561b89d754d3496da4e5fc676f5ad : SUCCESS in 1h 37m 16s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/ab3536748b1b4bae80e39815420d9df9 : SUCCESS in 2h 24m 45s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/ce4870944cd7413fade4609f1fae45e6 : SUCCESS in 1h 27m 13s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/3cffe53f3fad4e6a817a41f77e168c47 : SUCCESS in 1h 17m 38s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/046199c1df334c4eac5955a76009b773 : SUCCESS in 1h 17m 16s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/12700a135daa44e5ab0b7565aa67dff2 : SUCCESS in 5m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/451ab04891a147bb8408397b8eff49a8 : SUCCESS in 6m 36s","accounts_in_message":[],"_revision_number":10},{"id":"760ea3dcf69d7d5a0cfad0c92a3269e3faf9ca4b","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-20 02:28:46.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"fa5c93777662c2df1d415151b0d83bc2e9cad5ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 05:16:05.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7b69ceb76f74436a83abf80c8595e577 : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a5a167eb0e84a8dabc888e1ac42b6f3 : SUCCESS in 5m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/901402633a6f40969cef4ea8a100f4e3 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/039304102fa34f459871b815d3889fb6 : SUCCESS in 4m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d4e068019f8f4b68b5f9096969832c3f : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76d650f6933042449f7666db876799c0 : SUCCESS in 3m 29s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/7d0ca601e8864791bf80c5b36b8fe1cf : SUCCESS in 2h 02m 13s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/f9a565b811384d65a96b7d78159afd7c : FAILURE in 2h 39m 51s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/88203c0b4b3e49aa9b98f61fcf49f110 : SUCCESS in 2h 03m 02s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/d4d6b5b2bd46448ea7eb172b4399d6ac : SUCCESS in 2h 02m 24s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/36b059b3cb384d6c9103349c0a69b7a7 : FAILURE in 2h 01m 04s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/83850c37dade4e17b29e01bbc79a9cf3 : SUCCESS in 1h 15m 11s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/3e75ab26b50442e09272ea049134787d : SUCCESS in 2h 09m 19s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/fe36ea90f1334bb1a6199d1023a94a54 : SUCCESS in 1h 44m 15s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/953dfe3696ee47899620dff818a4e615 : FAILURE in 2h 10m 27s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/90cec6bde6dd431986d9199ededebde4 : SUCCESS in 1h 05m 50s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/e8e3d3645ec64a2e915ffb1bf1444602 : SUCCESS in 37m 09s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/b5a4e23f6b30477989a2a9fe1569ee4b : SUCCESS in 2h 05m 58s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/c9e1d6280b9f4651b7c8d6f191df9d97 : SUCCESS in 1h 08m 58s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/83fa9694dc7449e991560aca0c297fec : SUCCESS in 1h 11m 28s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/78153b2d0b154026880d499140e934f3 : SUCCESS in 1h 35m 34s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/57ec69182262476793a1aa149814b0bf : SUCCESS in 1h 22m 11s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/8cfe5cb4a5744fb286069711f5d455f1 : SUCCESS in 1h 21m 49s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/5aeca656d45144a1a392444099ee10f8 : SUCCESS in 54m 26s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/c3c7e82c55a9426c853225c28678f1fa : SUCCESS in 1h 04m 28s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d1094ba6ef9e4aec994d10db436cbc2e : SUCCESS in 3m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/77187619d82a44aba76e2d8f1891fcc1 : SUCCESS in 6m 26s","accounts_in_message":[],"_revision_number":10},{"id":"fca34f154fe84435bb22f06992da0d9defbc2659","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-20 05:35:21.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/d0eb699b612a47069ceb559e0d2d49cd : SUCCESS in 3h 05m 01s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"539256cd7088b9d13b631b1ac35b1eb03997c9b4","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2020-08-20 14:16:32.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"171e21b041a510df8773b09453e7085e7d5c4ffc","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-20 17:34:20.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/73f90418eaf44bc2b3b6cce30982a797 : SUCCESS in 3h 16m 40s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"92e7b77d2508826c02d3b00c7eed64b0e08e00d4","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-20 21:01:16.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"d58ecc689bc8fcb32e67f37e5b27b23917ac80cd","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-21 00:20:13.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/c6b90abfb12740199c177b136f8606c1 : SUCCESS in 3h 17m 22s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"b4ac1753855d5d9ffb1e6f126cb61ff90ee2423b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 01:47:59.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5ecc75f7d517402ebfd1c63774dd70a1 : SUCCESS in 7m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35001f3cc1d6454391a36472e308f68a : SUCCESS in 4m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5c1a9189e41641e9bac12e41d95cc6a0 : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86d8698cef8b4b9499a23348a49d549e : SUCCESS in 5m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8983975c23534a96a3f677bfa4275f30 : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61d4bd4b27be498d8c5439f76a37adaa : SUCCESS in 5m 50s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/93bff2d2390142b99cf163b18d2d54d0 : SUCCESS in 2h 09m 14s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/2a40b9a06efd4b2a903420567b1d0901 : SUCCESS in 2h 23m 02s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/2b81000e1d33437389cda64baf39277e : SUCCESS in 1h 59m 39s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/6bde25ff093a49339742ba2bf9b021b2 : SUCCESS in 2h 08m 53s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/c5ec9b34b3194729bd3608f9c88db381 : TIMED_OUT in 3h 13m 15s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/f57c5a1a917346b28e349121fc6e811d : SUCCESS in 1h 45m 07s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/7ab4b7af85ab473185df6005a5f794ab : SUCCESS in 2h 24m 32s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/18d8f00ede4141f7bf9d617c4b727acf : SUCCESS in 1h 46m 07s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/de0761b9132b440fb699f491f744067f : FAILURE in 2h 15m 10s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/52e1b6753cb14d7ab4aa4605dcee44dd : SUCCESS in 1h 25m 59s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/e418f45ffaec482c8ab75983b4a44121 : SUCCESS in 44m 41s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/c351e25c59b048598b9ccb31ab4c4b81 : SUCCESS in 2h 11m 29s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/5146aeafabad43f0b8e52ef352677065 : SUCCESS in 1h 29m 09s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/194617ad73de4c8bb85e14201e53a4d2 : SUCCESS in 1h 05m 38s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/1fd5af44fc78402ba7a4fa14951ceb25 : SUCCESS in 1h 36m 32s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/f68011cd98784b01b7e15abcbb3b8b2d : SUCCESS in 1h 55m 03s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/e68d8d01bba14d4f8021b86ad912d1ff : SUCCESS in 1h 20m 51s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/86c9bec23e094cf38167daa2c60a0248 : SUCCESS in 1h 47m 05s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/c72b9ac265e243f1bb25e771a143f04a : SUCCESS in 1h 06m 00s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f74787d94263497a8f0045231114b4aa : SUCCESS in 4m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/19ea4b6bd6ff4e5f9b7a66c6487525a7 : SUCCESS in 6m 36s","accounts_in_message":[],"_revision_number":11},{"id":"70cbc23de5e9112bab54c1aa40588ad6e64cb084","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-21 12:11:37.000000000","message":"Patch Set 11: Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"aaa7fcc93a0d9736d8a5fc185a7619c8c8925dab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 15:17:36.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"0b75f4502c31e0660e6f586d2ffc2fdaa9756505","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-22 01:24:14.000000000","message":"Patch Set 11: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":11},{"id":"3d905226f54b5aed741bc069de8811155dbe485d","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-24 12:35:32.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"c570468b2bb166d287774153a7cdb66005e02689","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-24 16:14:07.000000000","message":"Patch Set 11:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/9b834b9d99cc4748895fb922aa365ae0 : SUCCESS in 3h 37m 27s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"5810f35e6aa0484669fa1ddf8e7ebdfd555c0a03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 17:35:13.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b7520af6f7d245c1a780f2dce682867b : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a47df4deab2647e0a717464c0a93f5e3 : SUCCESS in 3m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bcdb6c0b33db4fcab5f617632b70aba6 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/82b0ec15a325430a8cfeab50579f1d69 : SUCCESS in 5m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ceea7b7a92d746639a66ec1b8088d820 : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff25f5c5995a450fa99bb2665ec38ff7 : SUCCESS in 4m 13s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/c32842e43453435181b79ce4ba8afe30 : SUCCESS in 2h 07m 55s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/6703095c51194f77a35e63038b3934e9 : SUCCESS in 2h 21m 23s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/601f73d493ab4ffca013173160238414 : SUCCESS in 2h 12m 55s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/c8935bd8bd8b47d7a82c1ed13c570965 : SUCCESS in 2h 24m 14s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/30d78131e6bb49809f3ca5a18ae5851e : FAILURE in 2h 06m 05s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/111712858ed34d62a32d653ed7b02dd8 : FAILURE in 1h 44m 55s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/48980ecbab294938a6bed996d358d087 : SUCCESS in 2h 43m 43s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/dc397034f99f40119fc0231f403584ff : SUCCESS in 1h 42m 45s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/aa46e28c681945fc8e68c75aa6de942b : FAILURE in 2h 32m 50s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/11c74456e17a4d8da9fd93694d23714b : SUCCESS in 1h 15m 34s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/70eb6fa9f5b248c9ab39c3ecb07125cb : SUCCESS in 45m 28s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/c373106e333d4685a648e8bab073beb9 : SUCCESS in 2h 46m 07s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/3ad8f81ced5b4966854242906c357d39 : SUCCESS in 1h 25m 54s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/2f9f7924f46d4d14a242740d902df3f8 : SUCCESS in 1h 08m 38s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/30934aa472f5451ba74fc8b77d5984f4 : SUCCESS in 1h 33m 17s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/3f94d0a8ec11404e942911d59bd2dbdb : SUCCESS in 2h 13m 27s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/08cbad4cc69f4405a4ded299af7772d7 : SUCCESS in 1h 23m 58s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/5f5b7e6d16ad46f78361599e885dcb33 : SUCCESS in 1h 34m 37s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/737895ed2eba4aa3a4db5b1ab83b662f : SUCCESS in 1h 04m 45s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b90030b8b0e942e6a1034e37c12e5b2d : SUCCESS in 4m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c99deb32a81a4f6fb4072902306c7191 : SUCCESS in 6m 34s","accounts_in_message":[],"_revision_number":11},{"id":"cb430f4c28718916fe3c351d54d95d1ee988eb07","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-08-24 20:10:13.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"c3fd17c5cb42e90a1eb5cb29a2ab6d5288290899","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-24 23:41:48.000000000","message":"Patch Set 11:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/fc12cac8b8864ddcbc6ad69e52163551 : SUCCESS in 3h 30m 31s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"99f5540c6f55cc69cbcf3e4338eac9d8772b9520","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 23:47:08.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e5453b64ba4f4982b277dfc20cb8feaf : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05858799d18f4e2e95c2e88e555c1f68 : SUCCESS in 6m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1b083c6710a047a4a1f934a0fee8c710 : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aacc9aa53fe54d3f87a71430a9acc0a4 : SUCCESS in 5m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a28f486235bf41698c4e6db22d417d9f : SUCCESS in 6m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1573b0fea7ab4b39aa8d40de547adfd4 : SUCCESS in 4m 51s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/f23e0a3c60184e32a3bcd3e9536e08cf : SUCCESS in 1h 42m 04s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/56d7e4a83b444a398fcfd32ff5aa30c5 : SUCCESS in 3h 00m 23s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/d177bef45f3946ba86d7fcad7250acf7 : FAILURE in 1h 43m 42s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/56ff3a06c18b402f9396ea3e65d1b58b : SUCCESS in 2h 17m 49s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/92a5da224f614d98a1bc90b03bd3b8d2 : FAILURE in 1h 54m 00s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/6ade61ee17e14419acc230b903b47140 : SUCCESS in 1h 51m 49s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/97341bdcb8e447e88346fa64a5f73e74 : SUCCESS in 1h 38m 15s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/65f3ec5e9f2f437fa2dcdcc59a2b6440 : SUCCESS in 1h 42m 13s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/0797e93897d645ac9664713155f593de : FAILURE in 2h 42m 38s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/e87e75fc1e2b4392917f02fd58376665 : SUCCESS in 1h 06m 12s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/96c857f558aa4a83b2585aec1aef64ef : SUCCESS in 31m 01s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/7ba0323165f14d2ebae5dbf50ca4c5a3 : SUCCESS in 2h 07m 59s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/736f71f77b4f4d1a9b9f46b47ad18a10 : FAILURE in 2h 15m 32s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/cea5e48f8ab74886ac5846e15014a014 : SUCCESS in 1h 14m 55s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/abd2ce804ef747609c826b8515d11942 : SUCCESS in 2h 13m 25s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/bcb4a8ad31ec4777bc3966fbd899019b : SUCCESS in 1h 56m 53s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/74afc36024174347b3b0befb32dcbaae : SUCCESS in 1h 24m 26s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/8464d9f05c464657a6386bd223003a53 : SUCCESS in 1h 33m 11s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/ee51b542dd684f4687ed0b24396d3a9d : SUCCESS in 1h 20m 52s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/aa8b076f35c64aafbf8e1d12fe7137eb : SUCCESS in 3m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0c19289dd0c542718eda6fa4e23fdd43 : SUCCESS in 5m 51s","accounts_in_message":[],"_revision_number":11},{"id":"47685b9ba0abc24d9d5f7d62188930e32125e0d6","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-26 02:09:51.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"b31ebe9b463347ece65b2e0dde32a480ad691535","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-26 04:41:34.000000000","message":"Patch Set 11:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/dcea5ab515f14ece9c91394aecda4dcd : FAILURE in 2h 30m 51s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"100e423767f188ce519c649583b7da76fdb51324","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 05:31:14.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fd723864c8cd4f7bb795d165c10d89fc : SUCCESS in 5m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1b4abc2e0414c56bc5fb416ff0bde70 : SUCCESS in 3m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4fbec2fbeb384cb294b478a67a81ddc5 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e277fc63ee1456693fd0b15ac032379 : SUCCESS in 4m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/60c1e43bfe294656863f9d220434169a : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/375dfda54b9b47acbef4699a0f15e826 : SUCCESS in 4m 52s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/806f686f11d94bf596c8fff340d30902 : SUCCESS in 2h 45m 31s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/bda010f110db44b5b657d62444266d3f : SUCCESS in 2h 50m 36s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/14563f1918a2481b9ecd11897760a6ec : SUCCESS in 2h 09m 54s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/63e777ae48ac473e8b1a70c3a0f5b758 : SUCCESS in 3h 11m 08s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/cb39e4eced3a41bc895296febbaa713f : FAILURE in 3h 02m 08s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/964b8dae1a464f168901143e54b512eb : FAILURE in 2h 23m 18s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/062f97b286cf4db6984abcae470de45d : SUCCESS in 2h 08m 36s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/638ed0ee711445c2a0beb21fd9cafdba : SUCCESS in 1h 54m 57s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/75652211bcb3428ca342fb16dbaaf615 : FAILURE in 1h 58m 14s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/8be351e99ffc4f5fb06f7eb5dbfba895 : SUCCESS in 57m 03s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/f87eae6f54a4437487baf4ed4ec2b916 : SUCCESS in 29m 38s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/793bb53d784a4e139dc525eeb56f0ab4 : SUCCESS in 2h 06m 59s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/c52a79dc021447e68fd68b25a5524cb1 : SUCCESS in 1h 41m 35s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/80440627fb2d4864a6b4a20365fadb08 : SUCCESS in 1h 17m 40s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/2c4f7335821242cfb1f1f773c07aeb81 : SUCCESS in 1h 38m 17s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/ee2462997e8b4dc1be66cd89943106c0 : SUCCESS in 1h 52m 51s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/2229157c8a5f42e68e73e6d24664a09c : SUCCESS in 1h 22m 49s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/9af2826493e74767a0c267678fae54f1 : SUCCESS in 1h 18m 20s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/ddc4dc51512b4cba8572e5db1ebd7717 : SUCCESS in 1h 58m 39s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c0a28d123e21449f8b032d338522bf2d : SUCCESS in 3m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f08ea3f666c544b080f3a68712f7784e : SUCCESS in 4m 53s","accounts_in_message":[],"_revision_number":11},{"id":"8039300386c4b2ce04ee0764748a9f960940fb35","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-08-26 05:48:07.000000000","message":"Patch Set 11:\n\nContainer builds timed out, strange?\nasync task did not complete within the requested time - 5700s","accounts_in_message":[],"_revision_number":11},{"id":"5f928799e2e974257e3ba8bb35be2d9eb5697cd1","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-08-26 05:48:16.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"f33ef97bd4b860127ea93107185ee9403988f713","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-26 08:46:55.000000000","message":"Patch Set 11:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/cdbd09e8cfb849cabc2fe2281373b1d7 : SUCCESS in 2h 57m 44s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"f2eaff3d27b490c775166da8162292bbd7849a5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 09:11:18.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/08ba1a6b781a45c9bfcb3379b5dce40d : SUCCESS in 6m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3b1aa30300044f68995c80c704de717 : SUCCESS in 4m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3293bb4fd8554059bf6f38e1f55b98de : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07c2a2110ee64438835bfbb7a4e975dc : SUCCESS in 4m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1f0b3042cb454cdb9dacb35f19246c7d : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d226fea9b7e3405eaa3ff22b2abd2c09 : SUCCESS in 4m 20s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/29e1246013c143fab6bd1b2ca8d5a76a : SUCCESS in 2h 48m 31s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/885987e826b745fe9af196739b4bb463 : SUCCESS in 3h 14m 44s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/a09e02050c5c471ea97fc6fd0d1468fe : SUCCESS in 3h 05m 16s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/e23faae516a14622af2075b0da65ca40 : SUCCESS in 3h 08m 22s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/d0fe9c1f3e7f4aae9859c4e3c850472c : FAILURE in 2h 52m 30s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/a520903bbbb24c8a9a434dfea7821fe5 : SUCCESS in 1h 28m 17s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/e54f501a5dd441f08a48a822aa9222bf : SUCCESS in 2h 35m 50s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/7d7df76e16e8410c9330c38f27f81d0f : SUCCESS in 1h 38m 24s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/be8629edc5ea4fc88a650d5eaa891b06 : FAILURE in 2h 54m 16s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/d010f9e69d924b18ae669f9467885310 : SUCCESS in 1h 11m 15s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/a003429c6806445dbe80d41751e806f1 : SUCCESS in 1h 09m 31s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/f54159d4dad049f992a592242c77d491 : SUCCESS in 1h 57m 36s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/8961cc5417a642cdac1b796b22e96612 : SUCCESS in 1h 42m 46s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/20bc35bd37434d079948a970ab3fb28d : SUCCESS in 1h 28m 10s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/8905ddcc28004cbf9323a0318c20f6ea : SUCCESS in 1h 29m 04s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/c2fbb11af19146178d300190846ac1af : SUCCESS in 2h 09m 50s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/0a893b96594d42dd980269df920510bf : SUCCESS in 1h 32m 19s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/e70f8ede2c2e4544b548979c5ade2338 : SUCCESS in 1h 36m 15s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/fce4d93a7f7c4be1bb2dee8fec4a1a24 : SUCCESS in 1h 03m 54s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/66f899c197a647e78c37202471e8b437 : SUCCESS in 3m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3437e05401654cb0b072262bab33768b : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":11},{"id":"ed599c181098906d715aa9bedd08693f05b375f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 09:11:51.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"b2770ac969e7581ebd8dc93063146b2b66e09e21","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 09:58:29.000000000","message":"Patch Set 11: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":11},{"id":"4d04e43b94dfde24284696c4da2cc4a41c3ea069","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-08-26 10:14:50.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"01aedf131fce2872049c91879b002811408b65f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 12:57:23.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d520cdf6014d4b0eaf70b35bef0f5d2d : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57a9b2d2b5b44ca09470c4c4ac222eb8 : SUCCESS in 3m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/99093d93508a4b26979bd0b6fa4de00d : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a45aa0c7070b464997afa5691bfeda08 : SUCCESS in 6m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9c4762cc15f04456909f340aa72ae519 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77efd1729e6a400b9ed790c3697cb934 : SUCCESS in 4m 26s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/7a2fcb7ba7c14a99a8845874f521b64e : FAILURE in 1h 47m 49s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/3b512867dfbf4eecaaf9d315fd32a71d : SUCCESS in 2h 24m 42s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/44689033e4024f24be25aba2f1738ef7 : SUCCESS in 2h 06m 23s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/6a7e327e68264a3bbce1c66e0950a909 : SUCCESS in 2h 08m 33s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/767b01711bbc4a13873bf0f21cc4d8b7 : FAILURE in 2h 04m 18s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/70c117f33ed04464a32daabb4e554ad6 : SUCCESS in 1h 20m 40s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/5b13b8857e474c918ee4592104cf4471 : SUCCESS in 2h 18m 36s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/8f33ef34e7604ac6b566da3903bbbd6f : SUCCESS in 1h 51m 12s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/6b1a5612549044a0991e87455ec48230 : FAILURE in 1h 27m 42s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/83257b650110481d9aab65001b3e91ca : SUCCESS in 46m 29s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/e4f23a180a294187af183279c6b9696c : SUCCESS in 42m 40s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/da25de8f4cf6479a89ba25bf1684c860 : SUCCESS in 1h 23m 52s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/15ed0f5a525e4c508b52e0a9adfd2472 : SUCCESS in 1h 35m 23s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/f912e0483bdf43edaff90b21de700050 : SUCCESS in 51m 01s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/c067b5d0564d405e8377ae468251eb8a : SUCCESS in 1h 33m 47s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/820ce17b2a754483a633714af0cf8aa8 : SUCCESS in 1h 53m 12s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/340ee55ffd0e405e98879ff6820709c0 : SUCCESS in 1h 19m 32s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/281783f68dcb41a8a616fbb2a585a863 : SUCCESS in 1h 04m 46s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/67b0c655570342d58b0554ff9dd129a8 : SUCCESS in 1h 12m 00s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/52c99e76220446db96c5f176b173daac : SUCCESS in 3m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/597c9ba968e04f47a6dfb661609db064 : SUCCESS in 6m 08s","accounts_in_message":[],"_revision_number":11},{"id":"b3e20208c9e77097bee277d7afe9a9567da5e586","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2020-08-26 13:02:47.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"3f5ec69144fa24b1d5fc88c5e0326e148aacccdf","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-26 13:34:40.000000000","message":"Patch Set 11:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/5177e9547b6442c5b0624530db92017c : SUCCESS in 3h 18m 12s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"bbe4d03511ddf7af85ba2847f21806c1c03297a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 19:30:53.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed2570f94d48435fbcab6d517c35091a : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3bce5aba9e9475eb396331028e16f80 : SUCCESS in 3m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c4f56560d4f846128abea3be07b25e1d : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/887b696abde34f80a81afe0bf4464d83 : SUCCESS in 8m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/24f4086e05aa475b99310379894d8249 : SUCCESS in 7m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b84cfa03c7f14f29aaf07d18b6f595d9 : SUCCESS in 6m 05s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/6bdd476b5c284f0e9537b0f75ebaa16f : SUCCESS in 2h 14m 55s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/a8a769d6324d43689440d4e3fce9f810 : SUCCESS in 2h 49m 06s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/da577ab3b62c435aa8861ba239a6506c : SUCCESS in 2h 01m 18s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/b76db66a19214b30b96ae7b7c53d12b0 : SUCCESS in 2h 23m 49s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/a8cbeea0227c4924bc8e05dbae12af77 : FAILURE in 2h 06m 35s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/12cb9ad933f14daf977d70d0054c6acd : SUCCESS in 1h 41m 09s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/c1c894234b91432380c36e6e104b2c3b : SUCCESS in 3h 10m 56s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/2434275a51b348fd9766445c21e541b7 : SUCCESS in 1h 43m 53s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/67852612ee70489e8480d97676ad5122 : FAILURE in 2h 40m 45s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/c3797d6b619e4039be941ce87d25b806 : SUCCESS in 51m 03s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/f737709220e640309f1ff9ca5602fb89 : SUCCESS in 42m 55s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/12376ed689034d62b9c3a977476191f2 : SUCCESS in 1h 58m 30s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/d41895da5c4a461bbdcbec2d1b77bfc3 : SUCCESS in 1h 49m 55s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/7c72262400fa499c8531d11ecbf55381 : SUCCESS in 1h 07m 32s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/580b67b75b7447579477c2fa285da749 : SUCCESS in 1h 34m 56s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/9b8d3c06a9734d0692439b896b5206f8 : SUCCESS in 2h 18m 40s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/360a3d59624a4198974f1c53d1d12cce : SUCCESS in 1h 02m 53s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/aae92931f49a40efacd288621252a4b0 : SUCCESS in 1h 24m 25s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/7e89f162d2814516828371a721da5efc : SUCCESS in 49m 08s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8940a71a115d4e7db0ef098c0bc41d8a : SUCCESS in 3m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6e7d02e2d8d2492aa01056838473a63f : SUCCESS in 5m 45s","accounts_in_message":[],"_revision_number":11},{"id":"7ce6b4ee2b19c187fd1cfe6a6c5511d51075d10b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 19:31:20.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"82d05f7db50f94a4c9cdf23303c8a179314c11af","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 07:03:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":11},{"id":"2ec5ff92d6235967c37274b45d468131dd674566","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 07:03:21.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8df26b00c3954f1688ad211ff93a5b37 : SUCCESS in 3m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fd31e4840c8947c282293ca4f5ab03c7 : SUCCESS in 6m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/11b3f1b97c9249ea9469d0ec0938808c : SUCCESS in 4m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d15aa8f4afa7410e9b53993c6772ba59 : SUCCESS in 6m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7399764664164f73995558756695fc40 : SUCCESS in 4m 16s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/5bb261b1661b4fe1a3e662843ed1d8c0 : SUCCESS in 2h 44m 11s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/d623e26b02a24f77b43bb7f3f6a5191b : SUCCESS in 2h 48m 43s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/9c9529891cb243a28d6bda220b00af11 : SUCCESS in 1h 25m 32s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/1b7ae9e9df53444b97676a530e1781dd : SUCCESS in 2h 50m 23s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/abdb28bec064415b8c868d9f302540e9 : SUCCESS in 2h 28m 35s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/02a6c0e6027c486aaea3a1866c5ca3d9 : SUCCESS in 1h 43m 00s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/c2d8b26390a540d095f1f3d9b507c1f4 : SUCCESS in 52m 00s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/3ebe5d51c583496aa2de3880dd311fe0 : SUCCESS in 29m 14s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/b6e218901c044949a4c10fc6dfb85efe : SUCCESS in 1h 55m 40s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/72a216bd2a0c4c02b44bce2990bdcfec : SUCCESS in 2h 10m 40s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/32c312cd535e43f0ae73e04484f09106 : SUCCESS in 1h 09m 34s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/e84a4885705441ddbc419f65c18c633e : SUCCESS in 2h 06m 04s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/bde4e3e1c75049efb0aed3de3657ea1f : SUCCESS in 46m 46s\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/2f240344acb04feb80f5373c4dc69557 : SUCCESS in 1h 18m 34s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7009771a380a439ab6369e8c4516d74e : SUCCESS in 4m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/321950eced9646a98f45ace9f222b7b8 : SUCCESS in 7m 47s","accounts_in_message":[],"_revision_number":11},{"id":"248e8e4694b19c2b611411ee6f86739e8decc931","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 07:06:18.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c870c1cf700847f586883e0364aa63e4 : SUCCESS in 1m 53s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"4dac873004a8eecb20274cc078cffd1ff3cd4ade","revisions":{"c8199240217c46e8cf58fc41c44924abb4bb8a40":{"kind":"REWORK","_number":1,"created":"2020-07-29 12:18:50.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/41/743741/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/41/743741/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/743741/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/743741/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/743741/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/41/743741/1"}}},"commit":{"parents":[{"commit":"2992b70e6c4f8f92dbc7b13bc21c3479a1b40d7a","subject":"tcib: move cradox to gnocchi-base \u0026 enable ci on scenario001","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/2992b70e6c4f8f92dbc7b13bc21c3479a1b40d7a"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-07-27 15:48:42.000000000","tz":-240},"committer":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2020-07-29 12:18:18.000000000","tz":-300},"subject":"container-images: add librados2 to gnocchi-base","message":"container-images: add librados2 to gnocchi-base\n\nThis package is needed so the right dependencies are installed in the\ngnocchi base container images and db_sync can run fine when ceph is\nused as a backend.\n\nPartial-Bug: #1889192\nChange-Id: Ic0bfdec756a23460f9257e78c508d167d21f0792\n(cherry picked from commit c6e680e180f457b8b9085d503ea8e8b6ccdad90d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/c8199240217c46e8cf58fc41c44924abb4bb8a40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/c8199240217c46e8cf58fc41c44924abb4bb8a40"}]},"branch":"refs/heads/stable/train"},"916bea722cce89153bd9d9adafec79e13c1d06f8":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-07-31 13:30:21.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/41/743741/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/41/743741/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/743741/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/743741/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/743741/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/41/743741/2"}}},"commit":{"parents":[{"commit":"4d3e3b8a0f0e426f381afe499b663fb4487f65b6","subject":"tcib: move cradox to gnocchi-base \u0026 enable ci on scenario001","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/4d3e3b8a0f0e426f381afe499b663fb4487f65b6"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-07-27 15:48:42.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-07-31 13:30:21.000000000","tz":0},"subject":"container-images: add librados2 to gnocchi-base","message":"container-images: add librados2 to gnocchi-base\n\nThis package is needed so the right dependencies are installed in the\ngnocchi base container images and db_sync can run fine when ceph is\nused as a backend.\n\nPartial-Bug: #1889192\nChange-Id: Ic0bfdec756a23460f9257e78c508d167d21f0792\n(cherry picked from commit c6e680e180f457b8b9085d503ea8e8b6ccdad90d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/916bea722cce89153bd9d9adafec79e13c1d06f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/916bea722cce89153bd9d9adafec79e13c1d06f8"}]},"branch":"refs/heads/stable/train"},"b04c268c55296fd613447980697eaa56653f0aad":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-08-10 13:57:06.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/41/743741/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/41/743741/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/743741/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/743741/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/743741/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/41/743741/3"}}},"commit":{"parents":[{"commit":"3fa8a199900921f0875624a77f9c4e784d33bf24","subject":"tcib: move cradox to gnocchi-base \u0026 enable ci on scenario001","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/3fa8a199900921f0875624a77f9c4e784d33bf24"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-07-27 15:48:42.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-08-10 13:57:06.000000000","tz":0},"subject":"container-images: add librados2 to gnocchi-base","message":"container-images: add librados2 to gnocchi-base\n\nThis package is needed so the right dependencies are installed in the\ngnocchi base container images and db_sync can run fine when ceph is\nused as a backend.\n\nPartial-Bug: #1889192\nChange-Id: Ic0bfdec756a23460f9257e78c508d167d21f0792\n(cherry picked from commit c6e680e180f457b8b9085d503ea8e8b6ccdad90d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/b04c268c55296fd613447980697eaa56653f0aad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/b04c268c55296fd613447980697eaa56653f0aad"}]},"branch":"refs/heads/stable/train"},"6c7e57557551bebb87bb5781aa7b44e104683b41":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-08-10 20:26:26.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/41/743741/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/41/743741/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/743741/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/743741/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/743741/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/41/743741/4"}}},"commit":{"parents":[{"commit":"9c32855ba5213558348c1f1e17a9c36ce4cc7d70","subject":"tcib: move cradox to gnocchi-base \u0026 enable ci on scenario001","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/9c32855ba5213558348c1f1e17a9c36ce4cc7d70"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-07-27 15:48:42.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-08-10 20:26:26.000000000","tz":0},"subject":"container-images: add librados2 to gnocchi-base","message":"container-images: add librados2 to gnocchi-base\n\nThis package is needed so the right dependencies are installed in the\ngnocchi base container images and db_sync can run fine when ceph is\nused as a backend.\n\nPartial-Bug: #1889192\nChange-Id: Ic0bfdec756a23460f9257e78c508d167d21f0792\n(cherry picked from commit c6e680e180f457b8b9085d503ea8e8b6ccdad90d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/6c7e57557551bebb87bb5781aa7b44e104683b41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/6c7e57557551bebb87bb5781aa7b44e104683b41"}]},"branch":"refs/heads/stable/train"},"cebad4f959546bafe9846e4c5d0f1e8ba94f7ed7":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-08-11 12:37:50.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/41/743741/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/41/743741/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/743741/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/743741/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/743741/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/41/743741/5"}}},"commit":{"parents":[{"commit":"f290607d3e8d338e26cddf19dcdf384b6fdc22ea","subject":"tcib: move cradox to gnocchi-base \u0026 enable ci on scenario001","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/f290607d3e8d338e26cddf19dcdf384b6fdc22ea"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-07-27 15:48:42.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-08-11 12:37:50.000000000","tz":0},"subject":"container-images: add librados2 to gnocchi-base","message":"container-images: add librados2 to gnocchi-base\n\nThis package is needed so the right dependencies are installed in the\ngnocchi base container images and db_sync can run fine when ceph is\nused as a backend.\n\nPartial-Bug: #1889192\nChange-Id: Ic0bfdec756a23460f9257e78c508d167d21f0792\n(cherry picked from commit c6e680e180f457b8b9085d503ea8e8b6ccdad90d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/cebad4f959546bafe9846e4c5d0f1e8ba94f7ed7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/cebad4f959546bafe9846e4c5d0f1e8ba94f7ed7"}]},"branch":"refs/heads/stable/train"},"79de88df364390b7eda854fe11c864330307ab5f":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-08-13 01:32:40.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/41/743741/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/41/743741/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/743741/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/743741/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/743741/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/41/743741/6"}}},"commit":{"parents":[{"commit":"07445de41bf22782d61121b37c5c015ee463350b","subject":"tcib: move cradox to gnocchi-base \u0026 enable ci on scenario001","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/07445de41bf22782d61121b37c5c015ee463350b"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-07-27 15:48:42.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-08-13 01:32:40.000000000","tz":0},"subject":"container-images: add librados2 to gnocchi-base","message":"container-images: add librados2 to gnocchi-base\n\nThis package is needed so the right dependencies are installed in the\ngnocchi base container images and db_sync can run fine when ceph is\nused as a backend.\n\nPartial-Bug: #1889192\nChange-Id: Ic0bfdec756a23460f9257e78c508d167d21f0792\n(cherry picked from commit c6e680e180f457b8b9085d503ea8e8b6ccdad90d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/79de88df364390b7eda854fe11c864330307ab5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/79de88df364390b7eda854fe11c864330307ab5f"}]},"branch":"refs/heads/stable/train"},"d376c4bce4283d11fa26ed4f6b01f119da71f357":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2020-08-17 13:26:03.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/41/743741/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/41/743741/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/743741/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/743741/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/743741/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/41/743741/7"}}},"commit":{"parents":[{"commit":"eec6684d9218ee0099901cfe7bfd5c8e3a067ee9","subject":"tcib: move cradox to gnocchi-base \u0026 enable ci on scenario001","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/eec6684d9218ee0099901cfe7bfd5c8e3a067ee9"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-07-27 15:48:42.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-08-17 13:26:03.000000000","tz":0},"subject":"container-images: add librados2 to gnocchi-base","message":"container-images: add librados2 to gnocchi-base\n\nThis package is needed so the right dependencies are installed in the\ngnocchi base container images and db_sync can run fine when ceph is\nused as a backend.\n\nPartial-Bug: #1889192\nChange-Id: Ic0bfdec756a23460f9257e78c508d167d21f0792\n(cherry picked from commit c6e680e180f457b8b9085d503ea8e8b6ccdad90d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/d376c4bce4283d11fa26ed4f6b01f119da71f357"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/d376c4bce4283d11fa26ed4f6b01f119da71f357"}]},"branch":"refs/heads/stable/train"},"3482c09c2007c622b825d19014d4e48929898c26":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2020-08-17 15:10:53.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/41/743741/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/41/743741/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/743741/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/743741/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/743741/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/41/743741/8"}}},"commit":{"parents":[{"commit":"cb746e39ebdd979c1b34b122ecf70d8cece703ae","subject":"tcib: move cradox to gnocchi-base \u0026 enable ci on scenario001","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/cb746e39ebdd979c1b34b122ecf70d8cece703ae"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-07-27 15:48:42.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-08-17 15:10:53.000000000","tz":0},"subject":"container-images: add librados2 to gnocchi-base","message":"container-images: add librados2 to gnocchi-base\n\nThis package is needed so the right dependencies are installed in the\ngnocchi base container images and db_sync can run fine when ceph is\nused as a backend.\n\nPartial-Bug: #1889192\nChange-Id: Ic0bfdec756a23460f9257e78c508d167d21f0792\n(cherry picked from commit c6e680e180f457b8b9085d503ea8e8b6ccdad90d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/3482c09c2007c622b825d19014d4e48929898c26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/3482c09c2007c622b825d19014d4e48929898c26"}]},"branch":"refs/heads/stable/train"},"1f629df0f4e9fb826ec37637d4799aa77e310b74":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2020-08-18 12:18:21.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/41/743741/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/41/743741/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/743741/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/743741/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/743741/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/41/743741/9"}}},"commit":{"parents":[{"commit":"7c2a0736f26895f73772910e11934af69587fab0","subject":"tcib: move cradox to gnocchi-base \u0026 enable ci on scenario001","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/7c2a0736f26895f73772910e11934af69587fab0"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-07-27 15:48:42.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-08-18 12:18:21.000000000","tz":0},"subject":"container-images: add librados2 to gnocchi-base","message":"container-images: add librados2 to gnocchi-base\n\nThis package is needed so the right dependencies are installed in the\ngnocchi base container images and db_sync can run fine when ceph is\nused as a backend.\n\nPartial-Bug: #1889192\nChange-Id: Ic0bfdec756a23460f9257e78c508d167d21f0792\n(cherry picked from commit c6e680e180f457b8b9085d503ea8e8b6ccdad90d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/1f629df0f4e9fb826ec37637d4799aa77e310b74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/1f629df0f4e9fb826ec37637d4799aa77e310b74"}]},"branch":"refs/heads/stable/train"},"cdaca862e156958cd0e7e67103446b204e9efeed":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2020-08-19 21:09:02.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/41/743741/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/41/743741/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/743741/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/743741/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/743741/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/41/743741/10"}}},"commit":{"parents":[{"commit":"f8afb7338986d756b56bb6a9d5f83b1bdb5626e8","subject":"tcib: move cradox to gnocchi-base \u0026 enable ci on scenario001","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/f8afb7338986d756b56bb6a9d5f83b1bdb5626e8"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-07-27 15:48:42.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-08-19 21:09:02.000000000","tz":0},"subject":"container-images: add librados2 to gnocchi-base","message":"container-images: add librados2 to gnocchi-base\n\nThis package is needed so the right dependencies are installed in the\ngnocchi base container images and db_sync can run fine when ceph is\nused as a backend.\n\nPartial-Bug: #1889192\nChange-Id: Ic0bfdec756a23460f9257e78c508d167d21f0792\n(cherry picked from commit c6e680e180f457b8b9085d503ea8e8b6ccdad90d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/cdaca862e156958cd0e7e67103446b204e9efeed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/cdaca862e156958cd0e7e67103446b204e9efeed"}]},"branch":"refs/heads/stable/train"},"4dac873004a8eecb20274cc078cffd1ff3cd4ade":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2020-08-20 21:01:16.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/41/743741/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/41/743741/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/743741/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/743741/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/41/743741/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/41/743741/11"}}},"commit":{"parents":[{"commit":"365bfd518f7c8b9d06b88f3751adf857a4a8edd9","subject":"tcib: move cradox to gnocchi-base \u0026 enable ci on scenario001","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/365bfd518f7c8b9d06b88f3751adf857a4a8edd9"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-07-27 15:48:42.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-08-20 21:01:16.000000000","tz":0},"subject":"container-images: add librados2 to gnocchi-base","message":"container-images: add librados2 to gnocchi-base\n\nThis package is needed so the right dependencies are installed in the\ngnocchi base container images and db_sync can run fine when ceph is\nused as a backend.\n\nPartial-Bug: #1889192\nChange-Id: Ic0bfdec756a23460f9257e78c508d167d21f0792\n(cherry picked from commit c6e680e180f457b8b9085d503ea8e8b6ccdad90d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/4dac873004a8eecb20274cc078cffd1ff3cd4ade"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/4dac873004a8eecb20274cc078cffd1ff3cd4ade"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
