)]}'
{"id":"openstack%2Ftripleo-common~744826","triplet_id":"openstack%2Ftripleo-common~stable%2Fussuri~If75ddd2fc569e7283235b8dec5d3807af5d26941","project":"openstack/tripleo-common","branch":"stable/ussuri","topic":"ansible-2.9.11","hashtags":[],"change_id":"If75ddd2fc569e7283235b8dec5d3807af5d26941","subject":"[DNM] Test ansible-2.9.11 Ussuri","status":"ABANDONED","created":"2020-08-05 06:34:06.000000000","updated":"2020-11-24 17:51:23.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ed73662f3b1cbd5e497e11151ce3011b33612e66","_number":744826,"virtual_id_number":744826,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-08-07 05:31:42.000000000","_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-08-07 09:09:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-06 06:09:17.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-06 06:09:17.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-07 09:09:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"39b4ef4c1b3aa6e1c140a8e548a436a8727ba0a7","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-08-05 06:34:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"57f61739f3569e6e7914969afbddb9a8af3fddba","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-05 08:57:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/2bdbbbc49056423da48364064d858f58 : SUCCESS in 2h 22m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cc7a03c7156929cb3b04e558377347a630a75e89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 10:57:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/501290b46fdc4f68a078a61e3344538f : SUCCESS in 5m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d54123ba9f54fa4a79d427ec0fd13ae : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9c9d029162594be9bccb1cd051b08e75 : SUCCESS in 5m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/155ecb72da8647fe8265a0fd9caa6ff2 : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1acff668e9a244209eb71ca06ccbb4d1 : SUCCESS in 4m 29s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/1a0225bb59714c668070f92abeec4fdd : TIMED_OUT in 3h 27m 15s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-ussuri https://zuul.opendev.org/t/openstack/build/24a458129e0b48fba459ce4e6ac2312c : SUCCESS in 2h 11m 08s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades-ussuri https://zuul.opendev.org/t/openstack/build/02d69b2d215343da8364a00fb8dff469 : SUCCESS in 2h 17m 06s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/912d02fdbcba459c87c2e63bbcf144b3 : SUCCESS in 2h 06m 08s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/89f6ca94839645089dd909ee215eafdd : SUCCESS in 1h 39m 04s\n- tripleo-ci-centos-8-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/f11a0731deab4e3888c801f9fe025f84 : FAILURE in 2h 15m 36s (non-voting)\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/799b4d44e188491d9a6d68871d86b866 : SUCCESS in 49m 39s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/a4ef87886288431f9c213dd5faea37cb : FAILURE in 8m 40s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/25ba33a95e1648f7af2ba5e06e76cb20 : FAILURE in 34m 41s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/fc3a7ffaa71a4dd0835e5e2df5a1b1e6 : SUCCESS in 2h 16m 23s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/867105db8066428bbb8af1a425290ae9 : SUCCESS in 2h 10m 00s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/2426efe812e8439ea82ea5e8c16b1557 : FAILURE in 12m 38s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/faa6ac0e58184179a425b3d3730872ff : FAILURE in 1h 36m 57s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/cc7a769cc5354731b300cf393b9ddf9c : SUCCESS in 1h 28m 02s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/7fa73a4da7f34c7f8d85971a5e99fcfe : SUCCESS in 2h 08m 58s\n- tripleo-ci-centos-8-standalone-upgrade-ussuri https://zuul.opendev.org/t/openstack/build/d2fe760344bd444088759756a8d8a6dc : FAILURE in 1h 31m 03s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e5229af286704c6791cd80275ff60772 : SUCCESS in 5m 06s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/7b1e580cfa544dce8e2e55fa802230b0 : SUCCESS in 1h 08m 04s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/8bb582d6b2ed4aaa9f03ef53182a4c40 : SUCCESS in 24m 17s\n- tripleo-buildimage-overcloud-hardened-full-centos-8 https://zuul.opendev.org/t/openstack/build/8914482d6950498f8345485bf5335c8e : SUCCESS in 29m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d6aad68d73dc4e6fb672fe546827b0d5 : SUCCESS in 5m 41s","accounts_in_message":[],"_revision_number":1},{"id":"19c43fae2ad50730c8d46d5da137af2ad6914b85","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-08-05 12:44:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8541d7cb9be4f10ed617ffc13a20c2818a151c77","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-05 15:14:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/65ea0c5e6fb54f229760c8349de81b48 : SUCCESS in 2h 28m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9d505abe260ecabad96c07fb2e0f665e7a25703f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 16:28:13.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e8ecfbb0ee7a466ca919a861a72179bc : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae5a3454b90f4fddb8754fb87076aa75 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f868ccaa917428595b49a3b52ee26f2 : SUCCESS in 6m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8f179646e7f744d5a0a9847d4695db13 : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/832b6a4782d64c11b0cf130b25a94292 : SUCCESS in 6m 02s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/b9e59ba3ccc849088e41536789511f83 : FAILURE in 1h 23m 19s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-ussuri https://zuul.opendev.org/t/openstack/build/47e6b9aab7de4b50af8b797a73c7a14e : FAILURE in 1h 16m 59s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades-ussuri https://zuul.opendev.org/t/openstack/build/16a7e8549c1f43c68e36f65a379d486b : FAILURE in 1h 20m 12s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/1999f2480d8447f08305166e22e2e341 : SUCCESS in 2h 21m 31s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/209bd795c34346399b58d6e8f2e170d7 : SUCCESS in 1h 49m 54s\n- tripleo-ci-centos-8-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/a59e281b0200420e817268d065d144ce : SUCCESS in 3h 12m 08s (non-voting)\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/5bd9c806d615434782f123f2f21fab7a : SUCCESS in 43m 29s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/0ee9ecb9773f4f2fb4c1c704c54f6afa : FAILURE in 3h 29m 19s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/89fe1f86a92748969f3cb18768afa1d0 : FAILURE in 1h 11m 17s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/62d8677c97564b078c6d26d803e90c39 : SUCCESS in 1h 36m 09s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/f26dd6bff08a4e7dbf1c5f41544731c3 : SUCCESS in 2h 37m 22s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/46a7ef6a480e470fb4d7e5f21eb10938 : FAILURE in 2h 34m 51s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/75c11325bfac40b69a4697f5ea01c126 : FAILURE in 1h 23m 20s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/9b3e7b22c9c74d11bb2cd4af00ace651 : FAILURE in 1h 11m 18s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/8211e121f44342379e503edeb49d91c4 : SUCCESS in 1h 47m 02s\n- tripleo-ci-centos-8-standalone-upgrade-ussuri https://zuul.opendev.org/t/openstack/build/bbc77ea9323448cd989d4b6801c70a9f : FAILURE in 1h 43m 21s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7af7c896ab084ae6a14d20b2b76a723e : SUCCESS in 5m 09s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/7f7b36206d60476d8c3f37fff2c0c19f : SUCCESS in 48m 23s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/47463135100a4b3195fee5d6fb25d944 : SUCCESS in 28m 16s\n- tripleo-buildimage-overcloud-hardened-full-centos-8 https://zuul.opendev.org/t/openstack/build/d28c570b7e524db5a1b6d59990199ee6 : SUCCESS in 28m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9d09703b3b9c4f65a69282f93f8734b1 : SUCCESS in 5m 28s","accounts_in_message":[],"_revision_number":1},{"id":"839e2ed93db45928eb90a69ba2c06028dddd098f","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-08-06 05:27:46.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"8eb31120358ac9673ddcae8f241affa5b44c2c24","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-06 06:09:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/188af2dde7fa402a8c885f0f33ae949c : FAILURE in 13m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5f56e25a3dd9427d4ac3866d908202046691cc1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 09:05:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/610b499315c9403e8bfde693ef6fd8e2 : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a0a3919962d4ad8a0709d223deee196 : SUCCESS in 3m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a0e13541902a49a5bc0e1bb5565dc586 : SUCCESS in 3m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/58f8de386f614aa28622a674f4b1c6aa : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fea23d3dc5de42ef9db313bbc9411e09 : SUCCESS in 3m 48s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/493ac31d194c49ce96c7274575d7f45a : SUCCESS in 2h 37m 16s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-ussuri https://zuul.opendev.org/t/openstack/build/53b8f3bbcc1042e7b529fe63d5119303 : TIMED_OUT in 3h 27m 47s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades-ussuri https://zuul.opendev.org/t/openstack/build/0d975e175fba4e3fbb90e1a67208985b : SUCCESS in 2h 19m 00s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/e9851df1efcc4b2e9dcfb95473486c57 : SUCCESS in 1h 24m 50s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/e1658557cde74be5909d03c4bfc2324b : SUCCESS in 1h 36m 36s\n- tripleo-ci-centos-8-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/53a1b246c0014eee87ef17979a7e6e57 : SUCCESS in 2h 37m 33s (non-voting)\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/7ccfed2c825646bbb0c0e542fb7d58fc : SUCCESS in 25m 29s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/37b0cbda402f4bc09f2111ad135a5acb : SUCCESS in 2h 38m 03s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/609db7058810495daba16d29ae674bfa : SUCCESS in 1h 15m 34s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/4e505831b7de4be19f8cd6dc23ea7066 : SUCCESS in 2h 04m 45s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/37bb988e2491484b967427bcdd57ef86 : SUCCESS in 1h 30m 20s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/4436fed5a79349af883b253fa8714f24 : SUCCESS in 3h 15m 33s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/eb022ba9cb4e452dab419c48f1c5f0ba : FAILURE in 3h 08m 51s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/16ede3e4493c4c00ab456282533b1ac6 : SUCCESS in 1h 28m 51s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/b28e3cf04d92409d9e197554e1c61b99 : SUCCESS in 1h 13m 46s\n- tripleo-ci-centos-8-standalone-upgrade-ussuri https://zuul.opendev.org/t/openstack/build/a3cc317aae5249dbb43dd3417d7e6cec : FAILURE in 1h 32m 09s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d96b4b4da38046e88d5202df7c384158 : SUCCESS in 3m 27s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/b9f9f899a66c486d9bed10725c8333c0 : SUCCESS in 1h 00m 14s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/12f4a1a804394e6da2effbb0527918f8 : SUCCESS in 21m 22s\n- tripleo-buildimage-overcloud-hardened-full-centos-8 https://zuul.opendev.org/t/openstack/build/e7a192f901524f4b8bfeee0dce6d8ad8 : SUCCESS in 20m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8f37066bc5f04182a8996f6ca0975ea2 : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":2},{"id":"8ee5fddb447a8f8a972a069094650df41da2e34b","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-08-07 05:31:42.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a99673d4f2ddfa47617d985c6d279e343fd9af88","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-07 07:56:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/6578a411286c4bb0a815759ac29027d3 : SUCCESS in 2h 23m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ee689cc49cbd77e462b854ae6b9c5299d9c3f23a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 09:09:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b123da29868b46d997ae2b1553b2056b : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca8dd9923c8f431ab8506f9ad78145c4 : SUCCESS in 2m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6bf231b5a931493788ffcb1b0ed6cf34 : SUCCESS in 3m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8a40afa1a8dc4f24b8998992a6fcb917 : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4da902307bf54129a6341088931c6e50 : SUCCESS in 4m 21s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/fafe5441df574150a74fb1f20e6b1fc6 : SUCCESS in 2h 30m 20s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-ussuri https://zuul.opendev.org/t/openstack/build/bf635c67f99545cfb620f2ef0b9cd681 : SUCCESS in 1h 54m 05s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades-ussuri https://zuul.opendev.org/t/openstack/build/d7490513ce844060b763a6ef043f1d81 : SUCCESS in 3h 20m 21s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/49296c808ce7439bb0ed41ec55f6b620 : SUCCESS in 1h 52m 00s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/b2d1d35f171741b696e35dc77b65b377 : SUCCESS in 1h 30m 42s\n- tripleo-ci-centos-8-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/69504fa25b3b4c30b854cb4fcb7f6d1e : TIMED_OUT in 3h 18m 15s (non-voting)\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/7a2afc17120b48089d69376a5292f02c : SUCCESS in 1h 38m 05s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/70aa97cb78034e588fcc06c24c181f8e : SUCCESS in 1h 57m 51s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/de5d48c4f5f34f5584d3a734579ed45a : SUCCESS in 1h 50m 57s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/b5771be6a93c48bd8ae2586cb309d5f8 : SUCCESS in 2h 00m 48s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/2a8e2a5171ad4a98aa8acd93444027d4 : SUCCESS in 1h 51m 25s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/5f3c35cdf5564cf091df7c689d78e573 : TIMED_OUT in 3h 30m 11s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/7e0b4704f4f94a259b39d5769a2b4452 : FAILURE in 2h 30m 44s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/f2b650dbe7ce42fe9ed46b91db37a79b : SUCCESS in 2h 22m 02s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/2bc9839aa84843f39cf72ea51f8c6393 : SUCCESS in 1h 33m 59s\n- tripleo-ci-centos-8-standalone-upgrade-ussuri https://zuul.opendev.org/t/openstack/build/fd19983f176549dca0d7f0e09d6b3020 : FAILURE in 1h 42m 28s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9365bdfe57a342c4ab0f356d85ccb2b5 : SUCCESS in 3m 57s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/d4c76e9c56144404990e9604ed4e12bd : SUCCESS in 50m 17s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/4fab1e7340ce4fe5af330f33c2dad31a : SUCCESS in 24m 05s\n- tripleo-buildimage-overcloud-hardened-full-centos-8 https://zuul.opendev.org/t/openstack/build/24c44ee9e9104b4496d58112ab6c67c7 : SUCCESS in 29m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e760c53e6baa458786ff6f7b19e0c5a7 : SUCCESS in 6m 28s","accounts_in_message":[],"_revision_number":2},{"id":"ed73662f3b1cbd5e497e11151ce3011b33612e66","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-11-24 17:51:23.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d6cea798cb2e9cd79fb6920ea017ca79abb5a836","revisions":{"08c15123ef44c13cd189a33cb2145f330d9e0675":{"kind":"REWORK","_number":1,"created":"2020-08-05 06:34:06.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/26/744826/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/26/744826/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/26/744826/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/26/744826/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/26/744826/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/26/744826/1"}}},"commit":{"parents":[{"commit":"aa4825cb53fc4d0b0740f4b4834cf702fcd19f66","subject":"Handle redirects for blobs better","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/aa4825cb53fc4d0b0740f4b4834cf702fcd19f66"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-08-05 06:33:24.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-08-05 06:33:26.000000000","tz":330},"subject":"[DNM] Test ansible-2.9.11 Ussuri","message":"[DNM] Test ansible-2.9.11 Ussuri\n\nDepends-On: https://review.opendev.org/#/c/722218/\nChange-Id: If75ddd2fc569e7283235b8dec5d3807af5d26941\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/08c15123ef44c13cd189a33cb2145f330d9e0675"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/08c15123ef44c13cd189a33cb2145f330d9e0675"}]},"branch":"refs/heads/stable/ussuri"},"d6cea798cb2e9cd79fb6920ea017ca79abb5a836":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-08-06 05:27:46.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/26/744826/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/26/744826/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/26/744826/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/26/744826/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/26/744826/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/26/744826/2"}}},"commit":{"parents":[{"commit":"d3c11cd581dfcd88b86e1a1d17c0c9826fc5676b","subject":"Stop doing tag_from_label lookup with tag","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/d3c11cd581dfcd88b86e1a1d17c0c9826fc5676b"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-08-05 06:33:24.000000000","tz":330},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2020-08-06 05:27:46.000000000","tz":0},"subject":"[DNM] Test ansible-2.9.11 Ussuri","message":"[DNM] Test ansible-2.9.11 Ussuri\n\nDepends-On: https://review.opendev.org/#/c/722218/\nChange-Id: If75ddd2fc569e7283235b8dec5d3807af5d26941\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/d6cea798cb2e9cd79fb6920ea017ca79abb5a836"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/d6cea798cb2e9cd79fb6920ea017ca79abb5a836"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
