)]}'
{"id":"openstack%2Ftripleo-common~744952","triplet_id":"openstack%2Ftripleo-common~stable%2Fussuri~Ib6fed59337f3dac07264052aaf2973ec5a69504f","project":"openstack/tripleo-common","branch":"stable/ussuri","topic":"bug/1889122","hashtags":[],"change_id":"Ib6fed59337f3dac07264052aaf2973ec5a69504f","subject":"Reduce workers for label lookups","status":"MERGED","created":"2020-08-05 17:05:31.000000000","updated":"2020-08-06 03:55:17.000000000","submitted":"2020-08-06 03:53:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"744952-1596686006808-88ec3396","meta_rev_id":"8b3fcff96c6cfa36d9a31edd89c12fc0227cd1d4","_number":744952,"virtual_id_number":744952,"owner":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-06 03:53:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-05 20:54:02.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"all":[{"value":2,"date":"2020-08-06 01:11:03.000000000","_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"all":[{"value":1,"date":"2020-08-06 01:12:19.000000000","_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-05 20:54:02.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-06 01:12:19.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2020-08-06 03:53:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b49ac0d5878a9d279974ebbb4916c653724efa69","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-08-05 17:05:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e0aa828f235826f4f411125443c918dcabed12e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 19:54:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2b16d04b93eb4af6998809f36b5dd9bc : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79c172b18dec4a7fb217afddd81093cf : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76109858456240b481fd4c49cd155d5c : SUCCESS in 4m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6ebf898a19334864896b34c486fa8ee6 : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3370f3cf763f4241996c63609ffe4763 : SUCCESS in 6m 35s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/0fd7b6a2d7fb4bcc8dbca4c53733d2c9 : SUCCESS in 2h 20m 03s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-ussuri https://zuul.opendev.org/t/openstack/build/a36ed9eb88bf4807a9284c2bc43522fe : SUCCESS in 2h 16m 15s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades-ussuri https://zuul.opendev.org/t/openstack/build/39f277f3d9494eb6ac2c447e9011c5fb : SUCCESS in 2h 24m 06s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/9943a43f0f5844149ce9199bb06985b3 : SUCCESS in 2h 06m 08s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/55ff0920fbbc44348bd8eb379f84543d : SUCCESS in 1h 33m 41s\n- tripleo-ci-centos-8-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/2fa9c9022dee4fd79f5d40b5a95ac8c3 : SUCCESS in 2h 46m 18s (non-voting)\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/8a7cb772e5534bf2b7d05655db8427a4 : SUCCESS in 48m 00s\n- tripleo-ci-centos-8-standalone-upgrade-ussuri https://zuul.opendev.org/t/openstack/build/6e2486e9273849cb93594c39f5b9426e : FAILURE in 1h 26m 18s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/bd0a72a7cbc1464685244aed652af40c : SUCCESS in 4m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8d62c97a905b418d826d81898bd6f482 : SUCCESS in 7m 15s","accounts_in_message":[],"_revision_number":1},{"id":"f80e948323e88d5ad9be4066b0b30d2940d21198","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-05 20:54:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/3a9461481d9041cca66f875be6ee87bd : RETRY_LIMIT in 1h 31m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f33e9a0c71b47a2dcecfc0cafd7d09b0808d8276","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-08-06 01:11:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9a4b7f7ad468d2f7e3feda2a6c835a1ac539ea44","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-08-06 01:12:19.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cc92bc64b00489268756b551d3c4e7f9e32a06f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 01:12:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6c3864d9ecada3d5df9ecac06340744c98facdae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 03:53:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5519824193a7008f90561e55998ecb87b81f7497","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 03:53:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2aae407703844ab8ed71005d3ae9d9a : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/960b3e9b148940418babaa13f3a81a3d : SUCCESS in 5m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/59eca8210a384e51acabea5bcfe24f98 : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2dd071570674df8937e2eda2f9a4a31 : SUCCESS in 4m 55s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/dc5c00eed7de4f32a20a476139786f07 : SUCCESS in 2h 29m 15s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/98074bb9bd874bdbbfcd0fdec6720f4b : SUCCESS in 1h 40m 25s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-ussuri https://zuul.opendev.org/t/openstack/build/90de200fa3f8404d91a51fc73c69c81f : SUCCESS in 2h 01m 19s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades-ussuri https://zuul.opendev.org/t/openstack/build/943646a9b3d945599e892a81fedd51da : SUCCESS in 2h 24m 36s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/0a39f7338d5e4074be258e0d9b37f75c : SUCCESS in 1h 25m 21s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/a0589c3162d14cd2bdaa68153e86f279 : SUCCESS in 43m 13s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e9e248d3981b4393b378a49cbef3bffc : SUCCESS in 4m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b33ae1e3b3814639805394b7f5499ef7 : SUCCESS in 5m 44s","accounts_in_message":[],"_revision_number":1},{"id":"10270d7fa73572d64536c81f0aa43ab4b857f8f9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 03:55:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0858270e89fe4d14a8c6cb790c18d0ec : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"84fc9430e261b41237fc92e58fc41aaea033ce8f","revisions":{"84fc9430e261b41237fc92e58fc41aaea033ce8f":{"kind":"REWORK","_number":1,"created":"2020-08-05 17:05:31.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/52/744952/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/52/744952/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/52/744952/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/52/744952/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/52/744952/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/52/744952/1"}}},"commit":{"parents":[{"commit":"aa4825cb53fc4d0b0740f4b4834cf702fcd19f66","subject":"Handle redirects for blobs better","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/aa4825cb53fc4d0b0740f4b4834cf702fcd19f66"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2020-08-05 14:27:53.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2020-08-05 17:05:20.000000000","tz":-360},"subject":"Reduce workers for label lookups","message":"Reduce workers for label lookups\n\nUsing 16 causes excessive auth requests and gets us rate limited\nupstream. Let\u0027s make this dynamic so that the upstream VMs will use less\nworkers.\n\nPartial-Bug: #1889122\nChange-Id: Ib6fed59337f3dac07264052aaf2973ec5a69504f\n(cherry picked from commit 5602dfd986939d4d2d19a124603c86504eb32b99)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/84fc9430e261b41237fc92e58fc41aaea033ce8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/84fc9430e261b41237fc92e58fc41aaea033ce8f"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
