)]}'
{"id":"openstack%2Ftripleo-common~744954","triplet_id":"openstack%2Ftripleo-common~stable%2Ftrain~Ib6fed59337f3dac07264052aaf2973ec5a69504f","project":"openstack/tripleo-common","branch":"stable/train","topic":"bug/1889122","hashtags":[],"change_id":"Ib6fed59337f3dac07264052aaf2973ec5a69504f","subject":"Reduce workers for label lookups","status":"MERGED","created":"2020-08-05 17:06:07.000000000","updated":"2020-08-06 03:55:45.000000000","submitted":"2020-08-06 03:53:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"744954-1596686005757-07a05074","meta_rev_id":"581627d8f0c07ec842b04cc199f64c93e7ad5f40","_number":744954,"virtual_id_number":744954,"owner":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-06 03:53:25.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-05 18:00:24.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"all":[{"value":2,"date":"2020-08-05 20:38:00.000000000","_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"all":[{"value":1,"date":"2020-08-06 01:08:53.000000000","_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-05 18:00:24.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-06 01:08:53.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2020-08-06 03:53:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2799d06718ef05cfcb598f1b3357444735208e1a","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-08-05 17:06:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dcc4b64074529b9ddb8e956809e488a2d6afd203","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-05 18:00:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/ac69ddac2afb4eb2a90ae273bf2b112c : FAILURE in 48m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"21a7e5429d3898263f0e7f303e6aeccb31b0d2a9","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-08-05 20:38:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"86ce55633e4d831bfcc2c37bcab15b5fd90eccc7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 20:43:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e071b2d568ce4ec69a69ea2466ed8c41 : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/544efd55aaaf43c38dba6561aec1c163 : SUCCESS in 4m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/20935a12ba8245749d00eb4a58a355e2 : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3175db3b0662489fa89ab35fa9f6530c : SUCCESS in 5m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6f09b1fa549349ab93720ef192df5a15 : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0860490e600491a95dd4770a14d0aa8 : SUCCESS in 5m 33s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/61346a777440481fabd9bdc3717a92c3 : SUCCESS in 2h 27m 33s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/c6b1f9a4a13b4a859f30710c21677646 : SUCCESS in 3h 05m 38s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/6d04aab4a7ce43e48d627ace3d86de87 : SUCCESS in 2h 42m 37s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/cd54597d4fe94454be900a189d19f9ca : TIMED_OUT in 3h 31m 08s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/d92cb7cf32c6434ba6f9d12e5862ab94 : SUCCESS in 2h 14m 45s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/0691ebfdec1d4e818373e6a57b82cea3 : SUCCESS in 2h 17m 54s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/dbea60f621944a68b8d1b04d8acd7e6f : SUCCESS in 2h 08m 24s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/40267062021240f7af9bd8aa050607c0 : FAILURE in 2h 42m 39s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/e99fa69b1199492483e635bc7f8afc77 : SUCCESS in 1h 23m 59s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/ab1f44cafeca4393aba44a443285071c : SUCCESS in 43m 08s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c7f6149ef91a41989421f8be2cdd7424 : SUCCESS in 6m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/65365a72e09d41d9a1c2af3ca061128c : SUCCESS in 6m 10s","accounts_in_message":[],"_revision_number":1},{"id":"70d9c9e07ac95fdd58ead71e9f0fd89f39c0affb","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-08-06 01:08:53.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"64087d176021d20138453bce69335ea8681e916c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 01:09:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"487c4aa33ade56ce321448dbabf33cb949b9c5aa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 03:53:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7eebf081c4379004843e1eb03cca0befbf367041","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 03:53:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/975538b7f9144ece9d4a7f27816c3d4f : SUCCESS in 3m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6717283d092e408793dabb525cb05eec : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8faaf8157c8548e4ad442c029c5d490c : SUCCESS in 3m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1763863f252749ba85915696240be657 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b92d252a1c1b49038cf94d6836457424 : SUCCESS in 4m 24s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/2a547e99ab144f6da644665aa1bcef93 : SUCCESS in 2h 29m 45s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/285596da0b0345b789d91deda1e854b5 : SUCCESS in 2h 25m 24s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/7ec3617784f64d09bbbdc18ff71a571e : SUCCESS in 1h 39m 57s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/5877c5fb0c5b45bb9bba530b7704b525 : SUCCESS in 2h 10m 38s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/c5c74427cb394911a7f619f87caeb1c5 : SUCCESS in 2h 13m 58s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/5ee289a1dfe44e1fb52709db6e1a5ed5 : SUCCESS in 1h 28m 32s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/552e53ae80b04a85a2fdd2960c965ae0 : SUCCESS in 1h 07m 49s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/fec2b92ceb0142339e03181642e556d9 : SUCCESS in 41m 47s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/67c22dbf85684131b9258fef104d44cf : SUCCESS in 3m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fa18e130dd5247218de8b1c7a43e47d4 : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":1},{"id":"55f22a8ea3723853edc9a73f2e3ebd0d4f24da9a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 03:55:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b27e9878b17d49e2abd59f1bd3a0d12f : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"933b52b1bd5178dd9fa1229b4b0175a2da911812","revisions":{"933b52b1bd5178dd9fa1229b4b0175a2da911812":{"kind":"REWORK","_number":1,"created":"2020-08-05 17:06:07.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/54/744954/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/54/744954/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/54/744954/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/54/744954/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/54/744954/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/54/744954/1"}}},"commit":{"parents":[{"commit":"d078e9bf142220ca9a3079938bd78cf73811735d","subject":"Merge \"Handle redirects for blobs better\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/d078e9bf142220ca9a3079938bd78cf73811735d"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2020-08-05 14:27:53.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2020-08-05 17:05:58.000000000","tz":-360},"subject":"Reduce workers for label lookups","message":"Reduce workers for label lookups\n\nUsing 16 causes excessive auth requests and gets us rate limited\nupstream. Let\u0027s make this dynamic so that the upstream VMs will use less\nworkers.\n\nPartial-Bug: #1889122\nChange-Id: Ib6fed59337f3dac07264052aaf2973ec5a69504f\n(cherry picked from commit 5602dfd986939d4d2d19a124603c86504eb32b99)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/933b52b1bd5178dd9fa1229b4b0175a2da911812"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/933b52b1bd5178dd9fa1229b4b0175a2da911812"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
