)]}'
{"id":"openstack%2Ftripleo-common~744955","triplet_id":"openstack%2Ftripleo-common~stable%2Ftrain~I4966641aed1a21be60d915ea58dda78b80fe0e1f","project":"openstack/tripleo-common","branch":"stable/train","topic":"bug/1889122","hashtags":[],"change_id":"I4966641aed1a21be60d915ea58dda78b80fe0e1f","subject":"Stop doing tag_from_label lookup with tag","status":"MERGED","created":"2020-08-05 17:06:07.000000000","updated":"2020-08-06 09:21:40.000000000","submitted":"2020-08-06 09:19:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"744955-1596705570887-afcfe1c4","meta_rev_id":"c9c4125a92e08ca515ae27fa312ebec9fa272d3a","_number":744955,"virtual_id_number":744955,"owner":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"date":"2020-08-05 21:14:18.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-06 09:19:30.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-05 18:02:34.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"all":[{"value":2,"date":"2020-08-06 06:07:58.000000000","_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":1,"date":"2020-08-05 20:37:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2020-08-06 06:20:17.000000000","_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"all":[{"value":1,"date":"2020-08-06 06:20:49.000000000","_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-05 18:02:34.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-05 20:37:37.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2020-08-06 06:20:17.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"},{"updated":"2020-08-06 06:20:49.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2020-08-06 09:19:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f5cc49829705e788f8035d0960c973ff65ec39dd","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-08-05 17:06:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9f8a02d6bac1a2ab04b135069251fd5ea6dd5878","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-05 18:02:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/eef419ba043b4c3784e2a6392710b3bd : FAILURE in 54m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"59d9295ea1f89a73d7cc31dadb3363655a1c91d2","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-08-05 20:37:37.000000000","message":"Patch Set 1: Code-Review+1\n\nscenario4 failed :(","accounts_in_message":[],"_revision_number":1},{"id":"ff56938c1c53997d65fca2185a59018223c8e09e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 20:58:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b53b08f6b431464bb6630afb4bad2b5e : SUCCESS in 5m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/320b9d5ec6e94ca6a9e528550da6a223 : SUCCESS in 3m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/91215cc24d874e06a2d3b845f2307aa8 : SUCCESS in 6m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/400cbbdbd9da4b2ab8eb3d5d5e92635e : SUCCESS in 7m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1362a5df860a4f42a1e58cd419c404fb : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0237bf3b8ad54b4991eb63a6679b8235 : SUCCESS in 4m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fb42f2037ad947a6a4ca684be62caea3 : SUCCESS in 4m 40s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/f6e33f96b37d4f15a11424f98ff58a9a : SUCCESS in 2h 27m 23s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/4c93f8fa427f40728fb5a55591a6f7f2 : SUCCESS in 3h 18m 10s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/602732cf7d5a48e3b71e938d17234cad : SUCCESS in 2h 46m 18s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/3ce6e6bdffb142bcb88b0c88ffb621b3 : SUCCESS in 3h 19m 58s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/a4b53862d6e84264a6f287d51b7aefcf : FAILURE in 2h 20m 17s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/a9753551dd24422faa299c33aaf9666b : SUCCESS in 2h 08m 39s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/92f6e14117844c4dbe86dc7734f9c7d0 : SUCCESS in 2h 17m 26s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/fb2899f58fa94bde9e6fc3dbaa109400 : SUCCESS in 1h 52m 56s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/dab53dc8c5494f99b42137dfd0d8a55c : FAILURE in 2h 11m 14s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/f231416f8dae4d8b97fdd382af0ebfa5 : SUCCESS in 1h 21m 16s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/1caf15cd70f842aba39436ecbb1627f2 : SUCCESS in 35m 43s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/a8aca6bd3c784b579c61a7f47ff82bea : SUCCESS in 1h 54m 12s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/333e5ca8f7f046ac978041ae0478a2d8 : SUCCESS in 1h 40m 05s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/1f1b51f7ee9d46778903ea343300a7a4 : SUCCESS in 1h 11m 49s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/09d03b28702d4ace9df416f015b67e36 : FAILURE in 2h 56m 39s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/51078e85e38e428d8439584d9ede996a : SUCCESS in 1h 44m 23s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/b5689a84bbe8429e912edc4f42aff069 : SUCCESS in 1h 48m 45s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/86980400eb5c48fcb3a11723f48c0c1d : SUCCESS in 1h 16m 39s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/cfc269aff01c4496b2f18c46237a159e : SUCCESS in 1h 22m 59s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d55649751cfe4922af684534905af0ab : SUCCESS in 4m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/996a7c94a1784196a29d0f491335b2e9 : SUCCESS in 6m 34s","accounts_in_message":[],"_revision_number":1},{"id":"ed831ce2d8b5d899185c5b2689f4ffef7f0c139a","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-08-05 21:14:18.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5a9f61b54db8ee18d0528d424297c530750025be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 00:16:48.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/56f476127be847708bb58025b296514f : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/484bf262d5aa46ebb34e769e25e0178f : SUCCESS in 3m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a9101d82919d4ca8bcbad508a21b7bac : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/75b51ac1dcc845f1bbe8e591a22082f3 : SUCCESS in 3m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/24bda4b229884f9cbcc8124009605bf8 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67403499fb234f61898080cb8da69947 : SUCCESS in 4m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d9ec404514514899b4cab1244de7d9a8 : SUCCESS in 3m 42s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/afcbcdb2c8934be3a53dc7ce73998f9b : SUCCESS in 2h 32m 06s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/31829a2fa68446bfaa7d95ddf58278f9 : SUCCESS in 2h 24m 26s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/23c554ccb65b43b498cabe4687c31d36 : SUCCESS in 1h 48m 53s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/ce9228feb5df4061a7e3b30bb2ef006e : SUCCESS in 2h 18m 08s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/0b1fecab4e70423d8a8f714b2b077e7c : FAILURE in 2h 06m 28s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/e78c826df61e486ebe6cbb03caca5688 : SUCCESS in 1h 41m 35s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/e3596f823ae74fb8b3cb3586f5824975 : SUCCESS in 2h 21m 17s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/77beb9420916443b89bf085b8313d71c : SUCCESS in 1h 44m 49s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/f5cd8cab26fa4cb7884fa50ebf5b080a : FAILURE in 2h 19m 20s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/a39ba7561b4d4b059a2375038f901728 : SUCCESS in 1h 04m 37s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/6dfe46ae608941bea0de64a4e9f4b52b : SUCCESS in 48m 13s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/6f4f4910ce4041c8a15fdc991f971680 : SUCCESS in 1h 53m 23s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/de4737f86fb740a8964f81e9a6b497e5 : SUCCESS in 1h 38m 38s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/8fcc63f07d0f4234ba937ab086308340 : SUCCESS in 1h 21m 36s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/f667ffed3a33404784be5238ec8c093d : FAILURE in 2h 52m 49s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/4abe6e3a686940db998bc7078f29ab9c : SUCCESS in 2h 14m 00s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/c9a435e5c4ee49c99f55dfd9d9e2b838 : SUCCESS in 1h 23m 55s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/71680ad3dda7425c9bd8e928cef2d342 : SUCCESS in 1h 22m 03s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/857ad370a4a747fdac5d2038e36175f4 : SUCCESS in 1h 11m 28s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/70a58308fa684319854e82fd8c77a30f : SUCCESS in 3m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/732eec0e54cc476783e1ca2a73adf678 : SUCCESS in 6m 43s","accounts_in_message":[],"_revision_number":1},{"id":"63c9ddfcb96bf12f41a9f35fabd78c9c153f025a","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-08-06 01:08:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"db803a19651d4e17c7ee6f53f25435d6e55e1c86","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-06 01:35:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/a22cd424c2874437b578f131904ace35 : FAILURE in 19m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a4de1e32eb4c99de8cd21c702959120be64d2cf5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 03:33:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f74a5cdb179b4daeba265d30184da73d : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0665c56f5d5b45be9d81e17a1c59c017 : SUCCESS in 3m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/63996e898df24fdea50a10e011c4c635 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1affdaf51a434fc1b254eb13dc3bf579 : SUCCESS in 4m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c8b2cdc28dbc490cbbab6174a752aaf8 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e18c24ead8447fbb3c9445af302eba6 : SUCCESS in 4m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/25e5e1e8287345bf99913b80da8607dd : SUCCESS in 3m 46s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/8f13218670a04ea69a4b3a6567c192f9 : SUCCESS in 2h 21m 21s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/e431d040aabc40e3b103330896d097fa : SUCCESS in 2h 10m 50s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/1e3a005ce21a4ec79f4692b097f66090 : SUCCESS in 1h 50m 09s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/983e0642545b4e549f834d7e463741fe : SUCCESS in 2h 07m 10s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/8de8a324ff284dea84da5366e99b757d : FAILURE in 2h 02m 07s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/f8ea9075832f4346a8ca2077fc146251 : SUCCESS in 1h 41m 41s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/aa0bb8178fb840958317968a73e03ad2 : SUCCESS in 2h 03m 22s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/f5b2e92fc22444ad9479e655938500ea : SUCCESS in 1h 36m 00s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/b6b5a596d7014250a38da2f89465aa9a : FAILURE in 1h 59m 55s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/1980c6c6655f4e2eaf72efc2feeab422 : SUCCESS in 1h 00m 07s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/f463b18b095b42438ad97146af5cddaf : SUCCESS in 42m 39s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/285b915752384460983d0a272e24a410 : SUCCESS in 1h 44m 48s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/692954bc9e314b2aaf32e67d7d2b7b81 : SUCCESS in 1h 33m 58s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/13560f5e65f04d39919e9b4022d535ae : SUCCESS in 1h 07m 39s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/71bd89dc9b48462f88379ae9b6a55e8a : SUCCESS in 1h 36m 14s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/547c1a1a7d3b45ac8c689b65dbc9df7f : SUCCESS in 2h 03m 51s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/28f0039af27a41b2acbed0e9ba5415f1 : SUCCESS in 1h 21m 37s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/42c6d3502782413dae20ed1aebaf9de6 : SUCCESS in 1h 20m 50s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/a077775d926043d2915d502d55948561 : SUCCESS in 1h 07m 46s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0b93cf2dbb2e44c58cc2a56e36bfd2c5 : SUCCESS in 3m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/100746f310a346ccb4fdb47914f4adc0 : SUCCESS in 5m 32s","accounts_in_message":[],"_revision_number":1},{"id":"ea7b79382e3a5a5157e4515220ff3f6697e3bf4a","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-08-06 05:58:49.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"27868943c54f340795918239e66af639f0fae023","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2020-08-06 06:07:58.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2407bfb88c428f43a414ed6bcad1863fa0a8fa36","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-08-06 06:20:17.000000000","message":"Patch Set 1: Code-Review+2\n\nSeeing max request issue https://66044ceaa9fa4a1fa395-906744ae5f779cf07a1bcffe2431d77c.ssl.cf2.rackcdn.com/741443/6/gate/tripleo-ci-centos-8-undercloud-containers/304105a/ and https://8fade4c2387c99a22d0f-6579984b68d810662ba94fe8b7947de3.ssl.cf5.rackcdn.com/741443/6/gate/tripleo-ci-centos-8-containers-multinode/4a4dec3/ and https://b136053d8c8f9c21fde6-af98ebf7fb9f80d504693b6d78e0e17e.ssl.cf1.rackcdn.com/741443/6/gate/tripleo-ci-centos-8-scenario004-standalone/8baacc5/logs/undercloud/home/zuul/standalone_deploy.log\n\nSo giving +2","accounts_in_message":[],"_revision_number":1},{"id":"b2b1979b8cbf59bf984f1abeb535600260b48628","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2020-08-06 06:20:49.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f3973d42c4a5ef5044200d888dff44ab7d3804b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 06:21:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b14e7d8e36737832b7a594702f5dd93cbf391b4e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 09:19:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c405343bd051402aabb93b309b6533e0 : SUCCESS in 6m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d9c6ad190f774c6184ffcf8afca40ac9 : SUCCESS in 7m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1605d89639d64f19a2adb1b300045740 : SUCCESS in 6m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/65713a10f9da458fa141a8cd380e2191 : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bd8d6e252eb477295c954f86c738a25 : SUCCESS in 6m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/38b7c149f9da4f339eb4f93596207a8c : SUCCESS in 6m 18s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/95ad422a074c49fb9d6e4da924116b40 : SUCCESS in 2h 51m 05s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/b2f26fad6891449fbba00a7b9b98f520 : SUCCESS in 2h 36m 58s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/7fd4f48a4e1a4c88aee00bf3613f0a02 : SUCCESS in 1h 09m 26s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/3c9333ef0bed41c298c9175549d74849 : SUCCESS in 1h 39m 57s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/0ef10076bf0e4c64bd1c5b12cbeba5ce : SUCCESS in 1h 42m 18s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/0439887b5c134a52b65bb849d19dff43 : SUCCESS in 1h 48m 13s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/a93e391f45f24cc1926ee48af38b08df : SUCCESS in 1h 15m 02s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/67e447b925584e28b3564f631e332546 : SUCCESS in 47m 51s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/8c90f365a3c14f7388ab2c7f85c8536a : SUCCESS in 1h 55m 15s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/facc03dfb8e34d9bb45c8c121fea559b : SUCCESS in 1h 38m 55s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/e5068c6d5bc74419b40c620b206cfc1e : SUCCESS in 1h 09m 03s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/5d869fc3a54543d9816a3c4c75e7b348 : SUCCESS in 2h 02m 14s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/19188d8e42804831b7c519e28e6ebd9c : SUCCESS in 1h 22m 48s\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/02691c80256d4469a518909bee09e084 : SUCCESS in 1h 25m 31s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/60e71d16afee4a70a7b0fcad6b07088b : SUCCESS in 6m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/43aa74421d2c47bdb1b55af6bfdb3ef7 : SUCCESS in 6m 01s","accounts_in_message":[],"_revision_number":1},{"id":"749dca920db396ae1a81c3f30e1aa1caa7bb4be1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 09:19:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e67b45706bc687d1e85488666408d38953f16724","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 09:21:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b25595021c954b4ebf9effa06eaf2387 : SUCCESS in 1m 25s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2ef9b091a410492598c2972de7550a6f : SUCCESS in 1m 28s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fff6f888a07316c4b2cd612098588a6ebd58a922","revisions":{"fff6f888a07316c4b2cd612098588a6ebd58a922":{"kind":"REWORK","_number":1,"created":"2020-08-05 17:06:07.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/55/744955/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/55/744955/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/55/744955/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/55/744955/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/55/744955/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/55/744955/1"}}},"commit":{"parents":[{"commit":"933b52b1bd5178dd9fa1229b4b0175a2da911812","subject":"Reduce workers for label lookups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/933b52b1bd5178dd9fa1229b4b0175a2da911812"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2020-08-05 14:42:17.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2020-08-05 17:06:03.000000000","tz":-360},"subject":"Stop doing tag_from_label lookup with tag","message":"Stop doing tag_from_label lookup with tag\n\nWhen using specifying ContainerImagePrepare if a tag is explicitly\nprovided in a set, the tag_from_label functionality will not be run as\nwe use the defined tag for the containers. Previously we would still\nattempt tag lookups even if we wanted a specific tag.\n\nThis would also cause failures during the deployment if using a\nnamespace that didn\u0027t have the defined tag_from_label format but we\ndefined a specific tag we wanted to use. e.g. using a tag with a md5 vs\na tag_from_label with numbers in it.\n\nChange-Id: I4966641aed1a21be60d915ea58dda78b80fe0e1f\nPartial-Bug: #1889122\n(cherry picked from commit 014d985f0836340b03d3c2cdeaae8717cfd1ea6b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/fff6f888a07316c4b2cd612098588a6ebd58a922"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/fff6f888a07316c4b2cd612098588a6ebd58a922"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
