)]}'
{"id":"openstack%2Ftripleo-common~745322","triplet_id":"openstack%2Ftripleo-common~stable%2Fussuri~I3832332abdfd7daaf373dc0924fec268f159d774","project":"openstack/tripleo-common","branch":"stable/ussuri","topic":"bug/1889372-stable/ussuri","hashtags":[],"change_id":"I3832332abdfd7daaf373dc0924fec268f159d774","subject":"Add exponential backoff to ratelimited requests","status":"MERGED","created":"2020-08-07 13:10:51.000000000","updated":"2020-08-10 16:10:18.000000000","submitted":"2020-08-10 16:07:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"745322-1597075677086-c9d102ff","meta_rev_id":"6137a5d1c32c3c8408ccf1d84f411e57290ce03e","_number":745322,"virtual_id_number":745322,"owner":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-10 16:07:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-07 21:58:12.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"recommended":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"all":[{"value":1,"date":"2020-08-10 05:55:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-08-07 13:33:15.000000000","_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":1,"date":"2020-08-10 12:36:16.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-07 13:33:15.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2020-08-07 21:58:12.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-10 05:55:43.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2020-08-10 12:36:16.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-08-10 16:07:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6a453fb29526d3e841885ec0362885cf6dc036c1","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-08-07 13:10:51.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"051aab90f644ee4c9c36974809c5e88fa120121c","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-08-07 13:33:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e09774309b31a86dd56c5653228854d3cc34f287","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 16:49:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c0f867bc140c4874945c9f53d506dfba : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/691c68daa145435da91d6e3d803cd915 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/99f46e580e1642fc8d83e4608b27cb92 : SUCCESS in 5m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/13e1318b5ef34e06b2f384bfaf8a2caa : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2748fba2d5544e599a76132fe775d24 : SUCCESS in 5m 02s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/abf00dcf201f40f38acc6bbca54ee469 : SUCCESS in 2h 14m 59s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-ussuri https://zuul.opendev.org/t/openstack/build/fd2f1a69b0db474d927cdc2987455795 : SUCCESS in 3h 07m 27s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades-ussuri https://zuul.opendev.org/t/openstack/build/b1f416089f674b4f8498090809980618 : TIMED_OUT in 3h 23m 47s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/99abd480226041b1aef8ff34ac67ba38 : SUCCESS in 2h 11m 38s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/cf470a74350c4182859e92ee36b38878 : SUCCESS in 1h 32m 20s\n- tripleo-ci-centos-8-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/e3ae1bbc842c49978b31899782fb07b7 : TIMED_OUT in 3h 35m 53s (non-voting)\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/88dd36e5274d4572a0f7787b881e9fd7 : SUCCESS in 1h 17m 02s\n- tripleo-ci-centos-8-standalone-upgrade-ussuri https://zuul.opendev.org/t/openstack/build/efc6e2d5f6d94c759a65ea54418c7fa1 : FAILURE in 2h 30m 21s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f9d4d1ce384644a78e32fdcfd3ebd97a : SUCCESS in 3m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/046cd975725c424bb559060cd4ff5048 : SUCCESS in 7m 08s","accounts_in_message":[],"_revision_number":1},{"id":"fd55320e4301daee57a0a3a75e5f081e6196c0b4","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-07 21:58:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/246904067a7a43bcabfbe4d1fb893739 : RETRY_LIMIT in 3h 29m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3a48b5f781e6ec38c85368bbcc84c96e36cd2652","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2020-08-10 05:55:43.000000000","message":"Patch Set 1: Code-Review+1\n\nrecheck \n\ncan\u0027t quickly see the problem with the ussuri upgrades job... let\u0027s see if the error persists","accounts_in_message":[],"_revision_number":1},{"id":"a0ce9d8d81500aaf98f30bc38ca80c0fff34b048","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2020-08-10 05:55:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"74cf17bec12968548b8c70a40831bfb8aee4e1e6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-10 08:27:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/5671b05428454aa78a72ec2024d77854 : SUCCESS in 2h 30m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a77484b02cdca67ada11798e8580c6b4ad4a87e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 09:16:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/105a4b07643241bb83d0b66655a8de89 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad561206c3ed4d56a1e0a195692fe059 : SUCCESS in 3m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e90ba9fc5c15468cb2e81d71efd3760f : SUCCESS in 4m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f8a0d8c020c2438f916ec3e9a6a80a11 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6460fa7498dc43b4a162f5f69f1157c2 : SUCCESS in 4m 18s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/e9941a5b1b3e4d13a6cf22b6a2686028 : SUCCESS in 2h 19m 54s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-ussuri https://zuul.opendev.org/t/openstack/build/908baba208524bcc9ba57937db0ef9f9 : SUCCESS in 2h 02m 22s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades-ussuri https://zuul.opendev.org/t/openstack/build/9326765ec871481880f931e8ebe2e2af : SUCCESS in 2h 13m 41s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/8647faf1c705409e8741967f91cc4a01 : SUCCESS in 2h 11m 56s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/c367d834829440fb8aecd00129b08184 : SUCCESS in 1h 55m 21s\n- tripleo-ci-centos-8-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/dd03896972cd401d83e619cf9c2cfa44 : FAILURE in 1h 34m 35s (non-voting)\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/2f4dd1778ca149de89eab7e1accd8124 : SUCCESS in 34m 29s\n- tripleo-ci-centos-8-standalone-upgrade-ussuri https://zuul.opendev.org/t/openstack/build/a2f59f93ca464680a01aa92f744cee31 : FAILURE in 1h 04m 13s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/bfeb25d6164b4a5e87ac6d4964ed4627 : SUCCESS in 3m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/403a188bdc144b5b93aeab9cdf629047 : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":1},{"id":"f80ea399a6285d80680733530c5b4dcc60aa951e","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-10 12:36:16.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"008f415aaff07e125c90cd40fb6b5cd1b6b6ecae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 12:36:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"354dcdd1d814fc98456fed658d44dc133f3d94d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 16:07:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c58381b1930f498381a2493a8a6078fe : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5cb15bccfd9740c192227dffe342bfc9 : SUCCESS in 6m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e86d6099489f4470a2d2e3fbf55586af : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c56c1765a234f60b47472e9206239e0 : SUCCESS in 4m 42s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/0aff17e41eaf4f7e80252bb4078a862b : SUCCESS in 2h 30m 37s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/8c31b34f00e3453ba7f357e4dbba2d75 : SUCCESS in 1h 44m 34s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-ussuri https://zuul.opendev.org/t/openstack/build/0ded002dac3d46b0aea2b4ebb09a268a : SUCCESS in 2h 11m 12s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades-ussuri https://zuul.opendev.org/t/openstack/build/94b8334341844c25bef482a265b53751 : SUCCESS in 2h 17m 51s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/ca4775a83b6d4dfcb03b1a16fecadd40 : SUCCESS in 1h 36m 03s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/d2e5937d58724bd6ab7a303161bded1c : SUCCESS in 46m 15s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/41c22ad789f043ae82cc47826333ca7d : SUCCESS in 3m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ab5c11bd9adf48169ace42027c496961 : SUCCESS in 6m 03s","accounts_in_message":[],"_revision_number":1},{"id":"dbef35d07cfba7c35de03de281c5c730a64c5298","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 16:07:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9d71fecdeeb5a19e9db70a70fa23e61c6780fc5e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 16:10:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b83827324dc464d89e786d0a0d4a263 : SUCCESS in 1m 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"be81e6ee690a8a5f8bc15309d333ea8266f43af2","revisions":{"be81e6ee690a8a5f8bc15309d333ea8266f43af2":{"kind":"REWORK","_number":1,"created":"2020-08-07 13:10:51.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/22/745322/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/22/745322/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/22/745322/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/22/745322/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/22/745322/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/22/745322/1"}}},"commit":{"parents":[{"commit":"d3c11cd581dfcd88b86e1a1d17c0c9826fc5676b","subject":"Stop doing tag_from_label lookup with tag","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/d3c11cd581dfcd88b86e1a1d17c0c9826fc5676b"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2020-08-06 15:21:59.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2020-08-07 13:10:51.000000000","tz":0},"subject":"Add exponential backoff to ratelimited requests","message":"Add exponential backoff to ratelimited requests\n\nIf we\u0027re ratelimited on a request, we should retry the request but back\noff significantly.  Previously we\u0027re not really handling 429 any\ndifferent than other requests which may lead to the requests being\nretried too quickly.\n\nChange-Id: I3832332abdfd7daaf373dc0924fec268f159d774\nRelated-Bug: #1889372\nRelated-bug: #1889122\n(cherry picked from commit d4f6eb64d4a54cfa90d4c1572ef54495f151fea5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/be81e6ee690a8a5f8bc15309d333ea8266f43af2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/be81e6ee690a8a5f8bc15309d333ea8266f43af2"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
