)]}'
{"id":"openstack%2Ftripleo-common~745516","triplet_id":"openstack%2Ftripleo-common~stable%2Fussuri~I28628100e1300e789114af420bf3ea0dfe9c0d70","project":"openstack/tripleo-common","branch":"stable/ussuri","hashtags":[],"change_id":"I28628100e1300e789114af420bf3ea0dfe9c0d70","subject":"Ensure tripleo ansible inventory file update is atomic","status":"MERGED","created":"2020-08-10 11:41:15.000000000","updated":"2020-11-18 00:57:13.000000000","submitted":"2020-08-11 04:32:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"745516-1597120324508-5080a3ed","meta_rev_id":"3c81d96856f7bf76cd5ebcac8759373b0f69f269","_number":745516,"virtual_id_number":745516,"owner":{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":23811,"name":"Oliver Walsh","email":"owalsh@redhat.com","username":"owalsh"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-11 04:32:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":1,"date":"2020-08-10 14:40:27.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":11166,"name":"Lukas Bezdicka","email":"social@v3.sk","username":"xbezdick"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-08-10 14:18:47.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":23811,"name":"Oliver Walsh","email":"owalsh@redhat.com","username":"owalsh"},{"value":2,"date":"2020-08-10 13:12:30.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-10 14:29:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":11166,"name":"Lukas Bezdicka","email":"social@v3.sk","username":"xbezdick"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2020-08-10 11:41:15.000000000","_account_id":23811,"name":"Oliver Walsh","email":"owalsh@redhat.com","username":"owalsh"},{"value":1,"date":"2020-08-10 15:17:33.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"date":"2020-08-10 11:41:44.000000000","_account_id":11166,"name":"Lukas Bezdicka","email":"social@v3.sk","username":"xbezdick"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":11166,"name":"Lukas Bezdicka","email":"social@v3.sk","username":"xbezdick"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":23811,"name":"Oliver Walsh","email":"owalsh@redhat.com","username":"owalsh"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-10 11:41:15.000000000","updated_by":{"_account_id":23811,"name":"Oliver Walsh","email":"owalsh@redhat.com","username":"owalsh"},"reviewer":{"_account_id":23811,"name":"Oliver Walsh","email":"owalsh@redhat.com","username":"owalsh"},"state":"REVIEWER"},{"updated":"2020-08-10 11:41:44.000000000","updated_by":{"_account_id":11166,"name":"Lukas Bezdicka","email":"social@v3.sk","username":"xbezdick"},"reviewer":{"_account_id":11166,"name":"Lukas Bezdicka","email":"social@v3.sk","username":"xbezdick"},"state":"REVIEWER"},{"updated":"2020-08-10 14:18:47.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-08-10 14:29:24.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2020-08-10 14:40:27.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-10 15:17:33.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-08-11 04:32:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c1bc15eb4e8352a9dc82359df6058f23e77c7b92","author":{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},"date":"2020-08-10 11:41:15.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"3fcef44ce9c9b15bd7ec875c4b03e5cf59450fc0","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-08-10 13:12:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1b5638a25d143c182b7b50a3a25f5e8b2689b838","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-10 14:18:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ffc1b3053a4aac060dd61d9b3ef9a1315dd0072b","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-08-10 14:29:24.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6967fc34544e12c5600cc2f30d93aa17ccd9b19d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-10 14:40:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/50c928b0edc646f2a50d49eeed3b1eca : SUCCESS in 2h 44m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bfb9f36d6cbe95d37cd473e9d64238b1a26f4bed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 15:10:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/385a6cc9086c4a0499862c5c72f25581 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f4b49b320324ff79ebfc417bbf31598 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/63484794e72a4fd4a298eb617dcaf7ab : SUCCESS in 4m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e0fc7b9559394725975dd186ffb5d38f : SUCCESS in 6m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c32c97fb88841b8a43f86405fa840a0 : SUCCESS in 6m 28s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/4d971f59b6814b5ab2f60e94c9fe3681 : SUCCESS in 2h 35m 15s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-ussuri https://zuul.opendev.org/t/openstack/build/3724242e428b41afa9b26144d76049cf : SUCCESS in 1h 59m 56s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades-ussuri https://zuul.opendev.org/t/openstack/build/335bdfa01ee245cca208ac60c8450924 : SUCCESS in 3h 02m 38s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/b2bc53dfa19d4c8fb98b108cdb17fcf7 : SUCCESS in 2h 20m 11s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/53a4e6b2817d4ad69c104009b383a259 : SUCCESS in 1h 39m 49s\n- tripleo-ci-centos-8-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/7a6489580d1046c4a5f29b8703eb96bf : TIMED_OUT in 3h 27m 10s (non-voting)\n- tripleo-ci-centos-8-standalone-upgrade-ussuri https://zuul.opendev.org/t/openstack/build/2f854277d7b040a5b663033d1f7b938b : FAILURE in 1h 25m 58s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/125d3dbe9dbd4e9cb0205f17ef1fe8db : SUCCESS in 4m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eb11f69c52e24f44b7d3cb6f61af850c : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":1},{"id":"0d4b72a73ec074038570aac2a4f28c1b06309b11","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-08-10 15:17:33.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2eb5daffebc9d15da1d74618d4d10061ed2a3600","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 15:17:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3cacf636c6ace6aae364e54de4d5e129f46981d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 04:32:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54eae117608f4b9487e62f408781ac62 : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b156dfa72024980b4cb885cf34b1bbd : SUCCESS in 3m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2e295e4d5c104ca0b1f6351f1e6b0ef0 : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d906415b085420b80630ab0ffaf2136 : SUCCESS in 4m 31s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/7b1a7d81b5eb438090d376be59ef7adb : SUCCESS in 2h 12m 48s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/152a4d106cc949d3a5292ff19ef41bfa : SUCCESS in 2h 07m 20s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-ussuri https://zuul.opendev.org/t/openstack/build/8aba4ff9fa3f48639b7a2639cb674c5b : SUCCESS in 2h 12m 26s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades-ussuri https://zuul.opendev.org/t/openstack/build/826e3717deb1412aa1e88fb8e6b7d806 : SUCCESS in 1h 43m 35s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/0f3fb4c5f1c5446ea391d2b2e59dd924 : SUCCESS in 1h 56m 26s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9a223ca3921e4fc6b83db5d66dcddb19 : SUCCESS in 4m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/67158922ebcf4aa88761878fe4cc5a5e : SUCCESS in 5m 47s","accounts_in_message":[],"_revision_number":1},{"id":"eaf063f4e32cc6be47bb4ba4a88a5285d1f5add9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 04:32:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4def806f8096df071443f6429d4caf306bc63e12","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 04:33:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1d029bdd820440aa089f6d09c7905e2 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1},{"id":"e48c48d4061b5779240778df73e1025e998b822f","author":{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},"date":"2020-11-17 20:01:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3c81d96856f7bf76cd5ebcac8759373b0f69f269","author":{"_account_id":23811,"name":"Oliver Walsh","email":"owalsh@redhat.com","username":"owalsh"},"date":"2020-11-18 00:57:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"62f4084537b38fa80cde1196a73a2c9b9494a510","revisions":{"62f4084537b38fa80cde1196a73a2c9b9494a510":{"kind":"REWORK","_number":1,"created":"2020-08-10 11:41:15.000000000","uploader":{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},"ref":"refs/changes/16/745516/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/16/745516/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/16/745516/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/16/745516/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/16/745516/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/16/745516/1"}}},"commit":{"parents":[{"commit":"4a473d743a952882873832adc61bdc5fb67583bc","subject":"Merge \"Revert \"Add kolla template override to remove python3-networking-mlnx\"\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/4a473d743a952882873832adc61bdc5fb67583bc"}]}],"author":{"name":"Oliver Walsh","email":"owalsh@redhat.com","date":"2020-08-07 15:15:43.000000000","tz":60},"committer":{"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","date":"2020-08-10 11:41:15.000000000","tz":0},"subject":"Ensure tripleo ansible inventory file update is atomic","message":"Ensure tripleo ansible inventory file update is atomic\n\nMultiple python-tripleoclient commands can run concurrently\nand all attempt to update the same inventory file in\n~/tripleo-ansible-inventory.yaml. Make the update atomic to\nensure the inventory file is always complete/valid.\n\nChange-Id: I28628100e1300e789114af420bf3ea0dfe9c0d70\n(cherry picked from commit 8e082f45ddfb8f29bd02feb1c9fd1ace4e9ec6dd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/62f4084537b38fa80cde1196a73a2c9b9494a510"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/62f4084537b38fa80cde1196a73a2c9b9494a510"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
