)]}'
{"id":"openstack%2Ftripleo-common~745732","triplet_id":"openstack%2Ftripleo-common~stable%2Ftrain~I7418ea45d89dd0da7ac39193200f47563acab4cb","project":"openstack/tripleo-common","branch":"stable/train","topic":"train/1889635","hashtags":[],"change_id":"I7418ea45d89dd0da7ac39193200f47563acab4cb","subject":"Correct the inventory generation","status":"MERGED","created":"2020-08-11 20:03:32.000000000","updated":"2020-08-12 16:57:55.000000000","submitted":"2020-08-12 16:55:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"745732-1597251311693-bd91aa04","meta_rev_id":"c25ddfa20233d8205944e3758c4b5956d252d239","_number":745732,"virtual_id_number":745732,"owner":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-12 16:55:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-11 23:16:10.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":2,"date":"2020-08-12 06:58:13.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":1,"date":"2020-08-12 13:04:20.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"date":"2020-08-12 06:58:22.000000000","_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-11 23:16:10.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-12 06:58:13.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2020-08-12 06:58:22.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"REVIEWER"},{"updated":"2020-08-12 13:04:20.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-08-12 16:55:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e902338131cfb2d1d9ec1415d09d0e00173875d4","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-11 20:03:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e0c2217c5c1fa896934668bebbc1a4f6244efdd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 23:10:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b9cf235a2db9464daf435fc8cf20c997 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d09cccddc2304ff8ad4bc6cd6677be27 : SUCCESS in 3m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c2cd58350d9c4047a15057474928d398 : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b2a8c3a5bcdc46888eadf9b352515913 : SUCCESS in 3m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d4147a42e8764f5ca88dc90f6172354f : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/daf8ea79a6534c5383ea9e6beb9ff223 : SUCCESS in 4m 02s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/31b97f6c87424446ba0a3ceac1f19b31 : SUCCESS in 2h 46m 08s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/bc0709382b824904aadf07a06b89e979 : SUCCESS in 2h 51m 08s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/1a9c7ef0130a4951ba5ff3fb6ed65ded : SUCCESS in 2h 13m 56s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/538a7addbd0a46d797555e331f2a80f6 : SUCCESS in 2h 35m 45s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/92042244f7d8464ca5af8472d8874db4 : SUCCESS in 1h 41m 33s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/59a14c3d8fbc4e88848e1268a422085b : SUCCESS in 2h 20m 11s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/847c649920db4f23b12664b4e6333937 : SUCCESS in 1h 51m 58s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/10429cac378449aba7bb440c04a7424b : FAILURE in 2h 20m 05s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f1e09e1b825443e3abca3983bed121db : SUCCESS in 3m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ed328d2c13124acd97060a915cba22cf : SUCCESS in 5m 28s","accounts_in_message":[],"_revision_number":1},{"id":"7c5302107e14dc04c12e6b03ded9924bdae333d9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-11 23:16:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/0ab306d460674825b627f52e0d5bfab6 : POST_FAILURE in 3h 11m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9e9044f4bd8ccbf6b2e45c049da8f4049b7ceb6c","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2020-08-12 06:58:13.000000000","message":"Patch Set 1: Code-Review+2\n\n(even if I\u0027m the one with the master change, let\u0027s merge it asap)","accounts_in_message":[],"_revision_number":1},{"id":"f6e5064014e8c1e49da47fe8cad8c9db02f0487c","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-08-12 13:04:20.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d4b0a17dcc6cbd72a676f4d41b37b047a29e3587","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 13:06:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6080b6bd8a520688fcc26b290f76ef71f1135ca6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 16:55:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1de4a7e3b02243a0ae978948fd86a77c : SUCCESS in 22m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c3fd20291351468ea0d4c1a3e38a1c44 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c90a70734a094200903703c523deae78 : SUCCESS in 4m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/67bc03b9c0384924ba0528b301ca9b76 : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c46259635554be48e8d8feaf2d13cc1 : SUCCESS in 5m 05s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/ee59b61f28af459c96975056d1c567c9 : SUCCESS in 3h 18m 33s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/cf368db00ee34e10a7c43a358fa0ade5 : SUCCESS in 2h 14m 13s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/6f496b2537664598a7db93bc21d1de24 : SUCCESS in 1h 49m 16s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/a05664e95bdf4ed19866fc07f9accf61 : SUCCESS in 1h 53m 48s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/4eea487954bf4541a49dfb2d23400c37 : SUCCESS in 2h 09m 45s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/32f7d24ae6a8499082f1310e83534965 : SUCCESS in 1h 39m 23s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/bbbe005d5c4f4d76acc749c2a84253dd : SUCCESS in 4m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8fe557ceed3c457b802477f6cd7a4738 : SUCCESS in 5m 50s","accounts_in_message":[],"_revision_number":1},{"id":"af26bb86771c86b23d444d8fc54fc25a0739054e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 16:55:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"817c018513deab617d955a4241b7c3ab4cb6c7c9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 16:57:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c33d35b3ac65469f98aca3cfc8c01ea9 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a51595922721eec1e155b88310068d1f7b1aa5a8","revisions":{"a51595922721eec1e155b88310068d1f7b1aa5a8":{"kind":"REWORK","_number":1,"created":"2020-08-11 20:03:32.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/32/745732/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/32/745732/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/32/745732/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/32/745732/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/32/745732/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/32/745732/1"}}},"commit":{"parents":[{"commit":"473d9cbc32163ce548e1483e0d5e1631c096c8d2","subject":"Merge \"Properly collect tasks with or conditional in step file.\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/473d9cbc32163ce548e1483e0d5e1631c096c8d2"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2020-07-30 14:36:53.000000000","tz":120},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-08-11 20:01:45.000000000","tz":-240},"subject":"Correct the inventory generation","message":"Correct the inventory generation\n\nThe missing quotes didn\u0027t trigger anything since \"vars\" is representing\nthe vars() method...\nAdding the missing quotes allows to actually get the injected host\nvariables in the inventory, among them the well needed \"container_cli\"\nfor the validations.\n\nThis patch also makes a huge update in the different inline dict, in\norder to make them more human readable. This should help for further\neditions.\n\nCloses-Bug: #1889635\nChange-Id: I7418ea45d89dd0da7ac39193200f47563acab4cb\n(cherry picked from commit e2613f0317721cf8231fe6a5a27f29f555b50a69)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/a51595922721eec1e155b88310068d1f7b1aa5a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/a51595922721eec1e155b88310068d1f7b1aa5a8"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
