)]}'
{"id":"openstack%2Ftripleo-common~745924","triplet_id":"openstack%2Ftripleo-common~stable%2Ftrain~I3ff5dc84b15a81d6afcc70ab5dde2d327b3c4714","project":"openstack/tripleo-common","branch":"stable/train","topic":"bug/1887176","hashtags":[],"change_id":"I3ff5dc84b15a81d6afcc70ab5dde2d327b3c4714","subject":"[TRAIN-Only] Dynamically add centos8 namespace","status":"MERGED","created":"2020-08-12 15:41:35.000000000","updated":"2020-08-25 14:11:47.000000000","submitted":"2020-08-25 14:07:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"745924-1598364465114-9c66d515","meta_rev_id":"51edf28437cc61a4011a6c9f98cfe239c11aa43c","_number":745924,"virtual_id_number":745924,"owner":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"date":"2020-08-24 13:13:59.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-25 14:07:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-12 19:41:12.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-08-12 16:32:41.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-08-13 08:09:02.000000000","_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2020-08-13 08:09:02.000000000","_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-12 15:49:37.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2020-08-12 16:32:41.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-08-12 19:41:12.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-13 08:09:02.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2020-08-25 14:07:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5d12288ab0cf329226b905272e31d26b415b4a20","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-08-12 15:41:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f0cbf8ef0cd47bbefbd4fc69e46fe0e71e8f8372","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2020-08-12 15:49:37.000000000","message":"Patch Set 1: Code-Review+1\n\nlgtm","accounts_in_message":[],"_revision_number":1},{"id":"ae33d15a3a9a9305b1facee7f031781ce1596265","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-08-12 16:00:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"877000bf99ab2437876ba4d5e84a1135e2adceb2","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-12 16:32:41.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"8332ed0ac0954804581ffa443e913acf2d2eadba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 19:36:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/42918cc4492b4bf0a939c05512342cf0 : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f34df41018c345b99f2616e25aeeb84c : SUCCESS in 3m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/684846684b87407f85d447ef365ee234 : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e14f831b7654ec3b2c5fd65a8c8eebf : SUCCESS in 3m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/de0750c92b9645ba8b7d8927937636b4 : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19f60e2621434cae819db657c19cc91d : SUCCESS in 4m 07s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/54cb1ed587d846db9948068614673c38 : SUCCESS in 2h 19m 36s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/8739244202de4b0dab8c0c612d501cba : SUCCESS in 3h 23m 11s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/cc250436e0fd4be0b2f9f4e5389403e3 : SUCCESS in 2h 11m 09s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/3090913c04324231864dcfdf116fd036 : SUCCESS in 2h 30m 35s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/12f64fd7429d422cb40070a1ff80d86c : TIMED_OUT in 3h 23m 55s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/9c590711e1fe436aa7ca01775afbcb13 : SUCCESS in 2h 11m 13s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/d0c68503335f4254b1c620cabba82efb : SUCCESS in 2h 27m 18s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/3a2f7e5772cf4436a3a218bd074e763e : FAILURE in 27m 57s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/6a4dfcdad69c4fd4a53e4398b6058e85 : FAILURE in 2h 16m 35s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/c412dd19ec5849568b332b3c6310230f : SUCCESS in 1h 10m 56s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/1227c191bebb459dac09d4f898323984 : SUCCESS in 41m 56s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/4cf06bec0acf4ca995bd1923b32fff98 : SUCCESS in 1h 36m 43s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/2922a60077154a039f7f4c9d2955a002 : SUCCESS in 1h 44m 05s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/2e99249066c84bf092e038a53f6581c0 : SUCCESS in 1h 07m 35s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/0cddb3666dad49dfab34b2640128af35 : SUCCESS in 1h 29m 32s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/d83bd01b3973433384fd5f5e254575db : SUCCESS in 2h 01m 18s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/336659a5c3f84086a99f28be5a96513e : SUCCESS in 1h 32m 58s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/16dea1ba5fde4004bf1030e4d7e9783b : SUCCESS in 1h 10m 36s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/8b6a78c820ca412f85b74efcd93a5996 : SUCCESS in 1h 11m 53s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d712e04024a04c6fb9dc8e6fd321b10d : SUCCESS in 3m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6c7cd0003fbf4cc3a7eccea0cf0ebc78 : SUCCESS in 5m 49s","accounts_in_message":[],"_revision_number":2},{"id":"0c25634dc92fd4abb5b72733f7a5e2fa43ce58bf","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-12 19:41:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/600810556e53416f845043422edc067e : SUCCESS in 3h 12m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"980c90684800ba4961e3940d4bb1e67b6b2c41f8","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-08-12 19:46:26.000000000","message":"Patch Set 2:\n\nrecheck WARNING: Retrying (Retry(total\u003d4, connect\u003dNone, read\u003dNone, redirect\u003dNone, status\u003dNone)) after connection broken by \u0027ReadTimeoutError(\"HTTPSConnectionPool(host\u003d\u0027mirror.ca-ymq-1.vexxhost.opendev.org\u0027, port\u003d443): Read timed out. (read timeout\u003d60.0)\",)\u0027: /pypi/simple/bindep/","accounts_in_message":[],"_revision_number":2},{"id":"ce75fd9cc1071d04b475a6eb215dad21e2b22f2a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-12 22:45:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/0180a08ff74c487dbabc46bd479099e4 : SUCCESS in 2h 58m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b366a362e244af5472e747ab67a0d25c24a72109","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 23:06:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9fce910c1637489ba7da39d1beb059c5 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2dc22d2d8d6f4e2f8a6aba4ec2b18759 : SUCCESS in 3m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b0440f1ba1c647fcb4b108c3ac7bdc41 : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ff3b1fd74b3486e80f51cc078793037 : SUCCESS in 4m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/26d72679b497425c91e9c1fd7df8d74f : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/597def59987d4148a6af0820801a11c5 : SUCCESS in 3m 56s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/6f61677499894b339fa3d8b6f107fd2c : SUCCESS in 2h 58m 24s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/3c4a8b2ff99e42508a15f1f5d86c51e4 : SUCCESS in 3h 05m 48s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/f1c8a8893d264e7895102ff7a4f5cef1 : SUCCESS in 2h 08m 56s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/27d3229dfb1b4c2d9230c7e4bcb00b50 : SUCCESS in 2h 02m 02s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/6d209bbf34ad4c7ab3dd293614435e13 : FAILURE in 2h 01m 58s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/da9b3239ace44a6980cb38bb423d7b53 : SUCCESS in 1h 37m 02s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/e31c9871c36e4c0fbd5304e474aec901 : SUCCESS in 2h 30m 31s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/b890b3596de642a1a0a588341a2088d1 : SUCCESS in 1h 38m 43s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/df6ba11060934592b596cb6674acbddf : FAILURE in 2h 07m 03s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/3c5309f9972b4fa8bb6a64f87eab6e6f : SUCCESS in 1h 13m 25s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/2d068517144f4ece86d8bfb99f49fbfd : SUCCESS in 41m 32s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/3d23b58f2b8f4a429e6cf74960c5d290 : SUCCESS in 1h 37m 09s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/9e3a0865d7cc4673b6c43649789274fe : SUCCESS in 1h 25m 46s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/f4ffa5d73f6b4a189c897f588be916e9 : SUCCESS in 1h 06m 43s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/498e50030f0a442da0deb5df7b81ecc8 : SUCCESS in 1h 28m 36s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/00d4da6fb27e4065ab0632936a6a6eba : SUCCESS in 2h 12m 19s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/55797882a5e9473a9929b29ba3290b41 : SUCCESS in 1h 16m 52s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/314353eb8a984053a131a4bd2b0a93f1 : SUCCESS in 1h 16m 59s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/5ee983d23c0549b49e72a4b13bc3a9c4 : SUCCESS in 1h 18m 48s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8a1965b1022d4de89c8801d68caafc02 : SUCCESS in 3m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/abd2254153ca4d14a80efa8625db2dac : SUCCESS in 6m 27s","accounts_in_message":[],"_revision_number":2},{"id":"6f32f7c4facd05850c3eb89c385388115fd1b8f3","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2020-08-13 08:09:02.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"62e7bbd1ddafe7e3deea3bc786b53a3e42507467","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 08:09:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"25f524e1571f1c18c7f6b736850e753853664411","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 11:01:05.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/721ff635b8884255b008aa7a15b870ae : SUCCESS in 4m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dd1f32df5abc470e9058c34ba38c3c95 : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/494ad7b6c6724ca89ec4659538363125 : SUCCESS in 6m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7dd862a717f44f6ca0ff0b6e6e5115f6 : SUCCESS in 6m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80883bd268f7401792a1743bde663959 : SUCCESS in 6m 20s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/dfecdfd229d640bca7da7c02f07d12c1 : FAILURE in 22m 59s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/cf0f7930f33d4341bb65dea987ae0c1a : FAILURE in 1h 17m 55s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/ca75fe590e8849049d410e060a5283f9 : SUCCESS in 1h 55m 12s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/dd20cdc578b74e57ba38dba42abe57a7 : FAILURE in 31m 05s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/738c120581c343e2b59eb6a04c5381e6 : SUCCESS in 2h 45m 36s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/e37c938b66fb4703965592e9f8eb7a54 : SUCCESS in 1h 54m 48s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/dd5e7cfff0cf4513b334bc3c43077ce5 : SUCCESS in 1h 04m 04s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/6ba5e0427d264e6faeced5651f801d79 : SUCCESS in 40m 11s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/43427021d8b14c5aa582993b5eb21c25 : FAILURE in 30m 01s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/f2e263df312249bea22e81fc9e85afde : SUCCESS in 1h 29m 05s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/d792400973d24788ae2fb7feb72e9c29 : SUCCESS in 1h 14m 39s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/fb951aae5e5a4452bcbce4e9ecfb36f3 : SUCCESS in 1h 36m 27s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/7b9b8c7379444c7e869d95f18b28948f : SUCCESS in 1h 16m 12s\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/7999ef57e18043b1b14945182bf00a14 : FAILURE in 24m 49s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4e01b9735dca46bfb6ce33cf83a60c35 : SUCCESS in 4m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b31ed4e2d3e846aa837ed5f7822c6172 : SUCCESS in 6m 38s","accounts_in_message":[],"_revision_number":2},{"id":"515e3bbb86787c4e22ee8e8d80754c080d94c86a","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2020-08-13 12:33:03.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"0ef60d1e391809640c44e8c9151fea2a7dc0ba15","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-13 15:48:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/1706eec278ab49c8a99518a2a509899a : SUCCESS in 3h 13m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1c393f8fba46e9913a75e3294ebff053326f2046","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 16:04:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d2daef2212d2497bbdbcf9b2e57bd785 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d00a0c36320c4130b8c50430b299ae0c : SUCCESS in 3m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1572f4b5e5c449e5884341ad61efe41e : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cca2fb171d5d4d178bfe738f062afbcf : SUCCESS in 4m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e24b217acc1d40848c8048ea9c16834e : SUCCESS in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/837523d7b8564c34882853ad5bf5a96d : SUCCESS in 5m 40s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/ecd5fc06148144a7b6671d4c186f8a90 : SUCCESS in 2h 35m 30s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/96c57479b2d841acb72dc282ab5ffa01 : SUCCESS in 2h 32m 00s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/630da643d6494a438e4eb596d3b00734 : SUCCESS in 2h 06m 21s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/015e837ea8624a6e97137c8f72d81c3e : SUCCESS in 2h 07m 17s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/b83f7ba52be84187ac85218d922a9633 : FAILURE in 2h 05m 57s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/b4e232ca43184f75aaac574c42aff883 : SUCCESS in 2h 37m 28s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/113994be7e094cfba28377f92a84dff3 : SUCCESS in 2h 28m 29s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/4efa4bdac35e43f89b4f1490bd3f8356 : SUCCESS in 1h 47m 28s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/602745184d9d477ab41f83a554d94034 : FAILURE in 2h 24m 39s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/8e68276945104216a1f0dc376d50be27 : SUCCESS in 1h 16m 38s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/59b9bd3cd701403fb4fabafda24702e7 : SUCCESS in 44m 10s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/e1c21945d496409e861280a410ba5ee2 : SUCCESS in 1h 47m 03s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/2ef19ba000c546578a641fb7a68871c4 : SUCCESS in 1h 39m 47s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/df9b40996dc048479baf468aa9d70238 : SUCCESS in 1h 12m 05s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/ff875fbc15434619833b0d0bd7e04171 : SUCCESS in 2h 11m 18s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/a001c54299e84b9586b0525fe6acf48f : SUCCESS in 2h 15m 07s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/5a913d2a34ab4f2a9b55a5c72a0cfab7 : SUCCESS in 1h 38m 04s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/7664d15f1a32444397f3764577484ff5 : SUCCESS in 1h 10m 47s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/886099bee6ab464783f930b3244116b0 : SUCCESS in 1h 03m 13s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d31f965185094f69891cc69975e788f5 : SUCCESS in 3m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f004b28cf1f444bfa3b3470ebf7fbbe0 : SUCCESS in 7m 06s","accounts_in_message":[],"_revision_number":2},{"id":"73202a0953d831b9ea2493667e4bf8c8d2f4310e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 16:04:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e54698d922271b2a6478c7e2dec2a1466da07969","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 19:21:49.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1608bbaf3c51433db9761ea9c4300e88 : SUCCESS in 4m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/eb9b8a73a6a8494ba6505c219d34c864 : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/15c0506382ad4e2b812235a9608fd577 : SUCCESS in 4m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ba78cce4325e40f09b03aaa411b6b081 : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbbac292dfe3402e8795f51ce4acdbbb : SUCCESS in 5m 47s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/d9802d827b0646218bc5333553e1a036 : SUCCESS in 2h 14m 43s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/fb09eeca2c1e4f32821ca92b13bf7fb4 : SUCCESS in 2h 47m 57s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/a4e4f3ad8f8640b5aa92ba60363d0ccc : SUCCESS in 1h 36m 35s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/6f6c2a6765ed4b219d5cf805d4b25857 : SUCCESS in 2h 42m 31s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/a15dbf4d64f44836a50f31153474997d : SUCCESS in 2h 55m 37s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/71d21f30ada7487db6e93021fc0abb6e : SUCCESS in 2h 40m 25s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/2726957a69bd4bab92f1046434d5925c : FAILURE in 1h 33m 06s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/793e2232b0434a12bad18bd0069a7826 : SUCCESS in 1h 14m 57s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/b200b33a641b461d9ae51f880e41bbb1 : SUCCESS in 2h 08m 37s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/757703ca9b7742579c6c73b3a1dbe331 : SUCCESS in 1h 33m 04s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/5d9307ba0699482c81959375bffe6007 : SUCCESS in 1h 44m 21s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/eeb8c14d39c9474db6bd6c876c6c1924 : SUCCESS in 1h 51m 04s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/6e6bae2de9074e198111ad9722ad9d72 : SUCCESS in 1h 41m 35s\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/cd46e3ec65d347f2bc1ca7117225ae45 : SUCCESS in 1h 47m 57s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ecfa42dbe8b64992b4e177bdac70bd2e : SUCCESS in 5m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9b40260826444099b359a75b5d22c3d3 : SUCCESS in 6m 47s","accounts_in_message":[],"_revision_number":2},{"id":"177dbbc6e886f0992f9672fff1c6a9339113d888","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-08-13 19:38:49.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"94b51d61298d66b8705e07103e7aa65b1a120808","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 22:34:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f62671e5b844ce0ac82a1c55e5b54c1 : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/286177e2dd304403af49043a13f431d0 : SUCCESS in 3m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8a82fba8797b4924a3f5b5987d768649 : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e04f7fd31dcc4812a06a1efc187ad501 : SUCCESS in 4m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/355f72dd1e5647d1a4c6becb448faffa : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e42247b70d654cc594b85b01cc370f2e : SUCCESS in 4m 37s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/c0ed48a1acf842e289674c35fc3b0123 : SUCCESS in 2h 19m 51s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/ab7165d2e9b14577aec08fd68a857cc4 : SUCCESS in 2h 40m 54s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/fd99ac8d95ae4d9f8382274816f7143c : SUCCESS in 2h 07m 21s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/ea21ab23533e47a99276911375d21104 : SUCCESS in 2h 27m 40s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/ca2c381b6c5948e0b481a721016187b4 : FAILURE in 2h 02m 01s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/510999dad14b41bdbaf9473d2e2bc551 : SUCCESS in 1h 47m 13s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/d1d547ca974b4682a134b52b3e007ea9 : SUCCESS in 2h 34m 38s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/17ee2a64c6f745c492d5b843c1b86cb4 : SUCCESS in 1h 44m 05s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/4614003cd9424f65b5c9126deed000ae : FAILURE in 2h 49m 14s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/f7f64c364ca8494883cdd4a95f4487a6 : SUCCESS in 1h 06m 44s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/96eb634aeee54c4c903648328ed4e3b8 : SUCCESS in 42m 09s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/7d58380d65a84856b0afebcbd135422e : SUCCESS in 1h 35m 33s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/d891f2f9383b46428a49e0c8afceb7f9 : SUCCESS in 1h 34m 26s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/6541810b995542d8bffe35c7cc6170a4 : SUCCESS in 1h 08m 52s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/1e0eb44d690f4dd8bb11895e11c6b075 : SUCCESS in 1h 35m 45s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/c9018ee3e8be429e93a07ca48a88e278 : SUCCESS in 1h 53m 05s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/bdf9247045824a24946a7db3d4793246 : SUCCESS in 1h 43m 06s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/8e11e5e214b641729e1c30837e8b1eb0 : SUCCESS in 1h 13m 01s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/9369c0c0af0d4f3aa540569e1d570b38 : SUCCESS in 1h 08m 12s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1208424ae1da416f9975bd2cd6c64b13 : SUCCESS in 3m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1a200c09e927463699a15796890ec391 : SUCCESS in 5m 49s","accounts_in_message":[],"_revision_number":2},{"id":"77a324b1ad38356728cd6a20f32b9e5e8bebd2e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 22:34:13.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"201321802c5254dc70c5943bc7388de5618c1aec","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-13 22:57:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/956505711a104ac793e8b74541e062f6 : SUCCESS in 3h 17m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"aa7256f8893b56279d6e75ca2be89b2ba9f9c16c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 23:58:49.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/684e599c7c3a42888a29e12fb49e8b7a : SUCCESS in 3m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/21e44fe3b4ad467fbf717a31e26173d3 : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d5554fde0aa348f5bded82d0494daedf : SUCCESS in 5m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3ced9113cf254fb88d3896568912f59d : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/968cecd36e3a49e08859dd257bff71eb : SUCCESS in 4m 54s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/6499a0f79242480a8a7deddaa33fccc0 : FAILURE in 15m 29s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/e765c01690c6440c972b48f116bedc08 : FAILURE in 16m 53s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/c4d23fe5945748f4ab9436ee46ba1be0 : FAILURE in 14m 08s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/b9d7385bf42b480a9dd1f0fb94d460d8 : FAILURE in 8m 55s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/f093aacc04084c59b925927f2bcca194 : FAILURE in 14m 35s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/8e78ec7ab70b4f598aff9dbf9fac8455 : FAILURE in 8m 05s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/d984036a50234ff68dba1c364a56d64e : SUCCESS in 1h 18m 22s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/c551982b936941dba9e228c86cdbaafe : SUCCESS in 43m 35s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/b3ef12a19a58482f98499a81436da7a3 : FAILURE in 14m 01s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/0f1d8899cf224c48a02b9c68c55daf14 : FAILURE in 14m 56s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/85d8772653774949a4aa11544b3a1a46 : FAILURE in 7m 34s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/ffcdc9742ea04c719ea3d263bb696045 : FAILURE in 15m 12s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/0786aaf63e5b4f3eba243e3a24f87576 : FAILURE in 14m 24s\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/3117ac399edc4f118073ea1e390c09b3 : FAILURE in 16m 26s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/eef5fc46b6a14b4fb0b0507fa60970e1 : SUCCESS in 3m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/27cffc78517546aca933d64a5a0ad13d : SUCCESS in 5m 51s","accounts_in_message":[],"_revision_number":2},{"id":"5564794a98cc605936feb41330f843c8116fe4a0","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-08-17 05:42:59.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d0a9e3aa59644c86f2a4d4011ea986abf338c0b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 08:40:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3d66b8f224343e0addb418be4254cbc : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c94398d3c2d49c581f721421e0f1cff : SUCCESS in 3m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/780c85d3c6d24551bc1a946f27dfb594 : SUCCESS in 3m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ad59ce91f3141ed97dbe4643c2b0450 : SUCCESS in 4m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d69025f24a674d83b0fdb365719a79f9 : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cccb90dd2844b28b4ab9fd415d56c0f : SUCCESS in 4m 01s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/771a8222ba2d43fa892e7cf95ae288c9 : SUCCESS in 2h 20m 01s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/26b32364261946ad9a8ce61a9d0c2a79 : SUCCESS in 2h 15m 00s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/3c0a2a699c2f492987712a0a12c5e30a : SUCCESS in 1h 54m 00s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/cbce2b1b61a04edbb49ab086af67bdb0 : SUCCESS in 2h 39m 52s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/bf522348632b460ab494548151cf27d5 : FAILURE in 1h 55m 31s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/75e3eff2c6fd4cec853ad8b14d4216e9 : SUCCESS in 2h 45m 42s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/0c6238cf03d045fa96cecff0eaaae9d9 : SUCCESS in 2h 32m 47s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/81f82173a18e408384ed58f5b50702d5 : SUCCESS in 1h 43m 23s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/9654c901a1c545db963a7c557d12bb04 : FAILURE in 2h 05m 55s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/594be1fd54924b92b6effb220133841e : FAILURE in 1h 41m 42s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/a86c4ad973864d7887956409126a947f : SUCCESS in 40m 54s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/00df27baaea1456685ba03d3bd7d6f34 : SUCCESS in 1h 41m 50s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/f8e594a951944a8ea627d02da0e1076f : SUCCESS in 1h 44m 35s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/1d5e703612a940879e30d06388214bbc : SUCCESS in 1h 07m 24s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/0d3abedc459449598cfc322b1dace846 : SUCCESS in 1h 04m 10s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/9cb515d151bb439caa3a21afb51e42ad : SUCCESS in 1h 59m 41s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/089a09293d4a4b2d8086c86f9a1cc189 : SUCCESS in 1h 22m 18s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/26ab8e1e4ff24edb8612262b5c236949 : SUCCESS in 1h 36m 00s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/2967e81b2b984f1ab9fd4817e81688a4 : SUCCESS in 1h 01m 54s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e13b321fdf2a4b06b76225c3ca96bbc4 : SUCCESS in 3m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b382e4505d7a4bf5b663c1bc351c065d : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":2},{"id":"32456af5beb971f776463a41fe8504d7878ffcdc","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-17 08:42:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/40f83fde84ea4e3aa2d0748c1ca76d54 : SUCCESS in 2h 58m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c20dcdb4915ecbe6e844dda961ab8037f46002a2","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2020-08-17 09:38:49.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"aeff8844524a0bf574e843c5109e1ebcd817c4a6","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-08-17 12:39:19.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"6d0dad9ba73fdec74ff0a82f0d808ee5ea973bf3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 12:44:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/087babeb0cbd4b68aa81e8dfcbfdf44f : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab786976cbb241d5b50666c7f886cd8e : SUCCESS in 6m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4d5d13d758954afd899abf7c27fb6c66 : SUCCESS in 6m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aa42dbc81a9f4f1c9a80bda21fab1f52 : SUCCESS in 5m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dcf6ced9ce8d4318889f9565e83a6938 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35f28e1ad0e84ea0b0bc2dd9d6bf5eb9 : SUCCESS in 4m 36s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/5b6835e0cdb9480abe2f2f0b572466c0 : SUCCESS in 2h 49m 13s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/1905feea5665448a8910879ae6b4614f : SUCCESS in 2h 42m 24s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/383764a0280d4916bcce4d0f817a3c46 : SUCCESS in 2h 08m 35s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/ffda109abb4f453fb1d48b001741354d : SUCCESS in 2h 31m 36s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/8e277fbb0948400abc31a7c62165b1b9 : FAILURE in 2h 16m 35s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/296d8c4d09f0445b8390dedeab4f6a15 : SUCCESS in 2h 07m 23s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/9b1d785bde864cb283d06892d7fbc65d : SUCCESS in 2h 10m 47s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/c9e97e94663b448e92b808f5a7f6f368 : SUCCESS in 1h 53m 34s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/f6f04962d786472b8d2bcae76755c889 : FAILURE in 1h 45m 13s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/86617ce250d54af388bb38c5556625cc : SUCCESS in 52m 31s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/c97242712b4d4898b65f5ebe7e1da630 : SUCCESS in 32m 06s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/8f529753a0b54b52a8ad568d34ad80d9 : SUCCESS in 1h 11m 34s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/df4da10fe72840c0afdcbe8749e236b3 : SUCCESS in 1h 41m 23s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/a9544d11c51a426ea188e771fc70fffc : SUCCESS in 1h 07m 43s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/a332db405cec42b2888269d70ae0bca9 : SUCCESS in 1h 32m 13s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/3ccf27efa46b436ebbe8ac55a53d2ca2 : SUCCESS in 1h 49m 33s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/d8d26968649b4418ae03c4bcfcdfeabb : SUCCESS in 1h 11m 04s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/84c2e582037d4be6a7e57584b05d919d : SUCCESS in 1h 16m 40s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/0f6d12e047404dd0a106f1e35ac7f4e6 : SUCCESS in 58m 39s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c9d4675141ba4677be60e87d50bba11e : SUCCESS in 4m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e185b38e9a004bc7960bd11cac43ba52 : SUCCESS in 7m 53s","accounts_in_message":[],"_revision_number":2},{"id":"f2541bd8d41f738e795e06339c46a28d749cfcbb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 12:45:11.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"01c6f17e93eb01d53f0a7d908b539464b1a084ad","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-17 12:58:08.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/212447678e53469eae8de37260e51090 : SUCCESS in 3h 17m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"60598d680c68f19e47ce2a936f0ab94938333200","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 18:08:54.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee01c7ddf7e64bcda1047a5de994ab3d : SUCCESS in 5m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/37f05a31d989446eab6374e2287fbeb0 : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a8d5ad3ba4c4416aaf199d56fca4f02b : SUCCESS in 4m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1ac138244b864b90b76fc5b4b09d7953 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7c1ff35aa1640a6985eeaa3cca75076 : SUCCESS in 5m 23s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/21421f1efec84e45ac9cc5eee32f6489 : SUCCESS in 3h 22m 31s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/4c3579be60dc46cd92815578f83e51a2 : SUCCESS in 2h 30m 35s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/f77f9a6cccce4401a9074fede87aa62c : FAILURE in 1h 57m 16s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/9786c6dbfa8d4be6b7e97415ef1226de : FAILURE in 1h 31m 30s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/964bf7615cb144fea17acc9df0ad138c : SUCCESS in 2h 16m 16s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/60326ab6d7754fdba7d0d21290853c27 : SUCCESS in 1h 44m 20s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/6b6094290b424de19d8d0c748e149fb7 : SUCCESS in 54m 00s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/b9ba6d6315e34ac68c5cd9cb266b48fc : SUCCESS in 35m 49s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/7c663c5dcd3b414f876f981a25617abb : SUCCESS in 1h 40m 40s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/20d30b1e10a5470da1781a33aee07ff0 : SUCCESS in 1h 28m 54s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/409996aaab154de2a710ceafca532284 : SUCCESS in 1h 13m 49s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/1f19fda3d89c49bfb4cfc08ab96bf251 : SUCCESS in 1h 35m 31s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/4e19a355edf840d8928807e1844bed18 : SUCCESS in 1h 02m 23s\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/b9c9bc04eeb74bb2b84d505573e7b85d : SUCCESS in 1h 36m 35s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/26db67ab2f38449aba7c25418348d281 : SUCCESS in 4m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a5b19efe2ca84c0597a007d211495c30 : SUCCESS in 6m 29s","accounts_in_message":[],"_revision_number":2},{"id":"eb5e135e2669b9cbb8ade18b07aca4acf797f07d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-08-17 18:16:49.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"990c24ed7d6e700fe7b6ca1d2c2ebb8dfd6e993b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-17 21:23:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/b7ee558dc07746f18be659339fd3124e : SUCCESS in 3h 05m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f7233a29fca356fa742e2fc6f9f0c74215f5e4df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 23:34:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf976ed9bb7144d3ac8eff86886aafd1 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84bc2b867d044d3297c12ea07ed37314 : SUCCESS in 3m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fb11b0485db245d899cd244152917509 : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e88d7fdcb74b46ca8038db41915abe2e : SUCCESS in 5m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9a230fbc38cf4c8887ed0da6d53d6eed : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6af2e15ceac48cb94e97fb3de64f6fc : SUCCESS in 3m 59s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/a52ad8daecb74ab289681f270b16f1ba : SUCCESS in 2h 23m 13s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/bd405f2e9002498188d9f8e31eefced7 : SUCCESS in 2h 52m 55s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/2fe49f3556c846248b76f4ba30ded5e3 : FAILURE in 1h 41m 47s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/18e03cc632154fa6bc4ae48e74740dd8 : SUCCESS in 3h 21m 12s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/e9e8562a32db4ab6a826f4dcf05a94ad : FAILURE in 2h 29m 12s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/0cc38b1e87254dcfb5eb65291fd1a7bb : SUCCESS in 1h 43m 47s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/226739ce73054dd8a22fe4d64028cc06 : SUCCESS in 2h 39m 15s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/b4ae7e9d69714fd9b82cde02b581a623 : SUCCESS in 1h 38m 36s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/0e12460e8abb42d68ea7bbd9d996cc40 : FAILURE in 2h 04m 50s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/8f4f422a92ad48d0a8baffe12ce2297f : SUCCESS in 1h 12m 11s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/a34e308c28be41dbb1b5d9063b7725aa : SUCCESS in 37m 36s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/602c032fac144261bce161267031a1fc : SUCCESS in 1h 45m 26s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/256fd5a24ae94749a75db3b9ff4fff46 : SUCCESS in 1h 31m 34s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/15daf77fc5f140789c928d42fabe55a4 : SUCCESS in 1h 01m 53s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/4a4a7106cc714539800413a389ed49c7 : SUCCESS in 1h 28m 55s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/e2cf664ce62f482ca87fe8bea295cc8f : SUCCESS in 1h 57m 49s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/95b9cd4965ef4153a17042f81cf05db8 : SUCCESS in 1h 24m 56s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/4c7854024bb442aa83321517878d405a : SUCCESS in 1h 14m 18s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/83da6041584a4464a6246dcdc2bfee6d : SUCCESS in 1h 08m 16s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/bdfd45e723734f80bac2226e3c326f0b : SUCCESS in 4m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3d8835fcb51b48faa7713c74962a59f9 : SUCCESS in 6m 37s","accounts_in_message":[],"_revision_number":2},{"id":"ef6e8096b033042eb0f7767589e83f64a22642af","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-08-18 13:04:40.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"85401a7ed53838aefe1e30d1cf04e48503e2b263","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-18 16:29:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/098caccf8ac44ab6bb8113cad43f6b10 : SUCCESS in 3h 23m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c8aa1174e99e367a6e46ebe0f06655bbe89af270","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 17:58:39.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7f54ead0a6d94b27b360ea6c8806fbe7 : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/649ef1de02264f868f85d7098b4e1221 : SUCCESS in 3m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f0e9847aad364635878c6ef43c291f6f : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2524f48328c347239557504f635f709d : SUCCESS in 4m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d542719886e74b17818a72940012e57f : SUCCESS in 6m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e2283cdf85e4ceea8043fed75cfee40 : SUCCESS in 4m 05s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/5c0b9d9cd4cf4573a0d47939ed96d657 : SUCCESS in 1h 47m 45s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/ea31d136aea04b10ad29caf1048a6000 : FAILURE in 1h 20m 21s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/da622ec979244879a272122720ccac98 : FAILURE in 35m 22s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/12c46cc9bf2a4a6ba874100079e252e7 : FAILURE in 29m 41s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/102c66e9e2354bcb808d2597d3b36967 : FAILURE in 2h 36m 22s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/2e2f9f5a10ca45a8a0ebb4e504c97f86 : SUCCESS in 2h 38m 32s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/22eef023dbb548e8a78b5be4283f6da0 : FAILURE in 25m 27s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/e8111d8ce81541e89f17d0b85a3c4d97 : SUCCESS in 1h 39m 41s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/ee01fa75587644da843a11e0be90b7a8 : FAILURE in 24m 29s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/358f138af8e149a29973e79596fa3e53 : SUCCESS in 1h 19m 39s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/f95b12fd4bac49499f7d630df5251e5d : SUCCESS in 45m 21s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/7fd51a5b3d38412bbde4e0aa577efe33 : SUCCESS in 1h 18m 41s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/a07cb596804a4cc39dbf91a673fc3088 : SUCCESS in 1h 10m 18s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/97aa18ba0e7443ccbaaaa086910a7d7c : FAILURE in 1h 10m 38s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/b4b81f8a53814d179463adc248577993 : FAILURE in 1h 12m 32s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/a081276c8a834f518ab802ef49425603 : SUCCESS in 1h 55m 38s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/6818818310dc471e81ecb62e6925e685 : SUCCESS in 1h 25m 22s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/ed26878ff4e14e878c05a5d5a3ff7235 : SUCCESS in 1h 13m 48s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/e47a2979824d4b47846082ef5e9cbf2a : SUCCESS in 1h 08m 15s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/199545217bd44de2a07f19194942b9b3 : SUCCESS in 3m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/56497b5246dd4d629cdce5eb9466e2bc : SUCCESS in 5m 32s","accounts_in_message":[],"_revision_number":2},{"id":"cab17add4f6d6e1c9277aa5d4d53017b069188e2","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2020-08-19 10:49:06.000000000","message":"Patch Set 2:\n\nrecheck \n\nyEEEEE hAAAAAAAAAA!","accounts_in_message":[],"_revision_number":2},{"id":"37e53aa0e6301ebb033430a7536fbe60503b9c05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-19 14:24:49.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/763fcfac35c940bb9c70d2369b4243c1 : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcf55621bed541de9e9a2496743118cd : SUCCESS in 3m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4fa3bb8193d44dd689ad9312efdd7de9 : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e1a3cee018b44f87af5495d9035072a5 : SUCCESS in 4m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9c6f88dba1ec487084659f77fd0595d9 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e6e72588934480aafc57668efe40c31 : SUCCESS in 6m 27s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/85eeb02e4abd47cb8a68f4c14d4e47dd : SUCCESS in 2h 57m 06s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/624f023f56124972bb06bba0dec4680b : SUCCESS in 2h 16m 40s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/db27ea934b9a4f0a93a5d6a05bfa2dcf : SUCCESS in 2h 36m 55s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/351c1f14814648889c20352dd1716877 : SUCCESS in 3h 04m 33s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/b3b7d7437ac546f0a7e1894d9765b974 : TIMED_OUT in 3h 15m 44s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/2a487cbdb1d14c5d93d581ef0675b97f : FAILURE in 1h 57m 08s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/cb9c273d7ed0478c924b82a0b4541e39 : FAILURE in 26m 21s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/bf016a932738452fa4961ddfebee7059 : SUCCESS in 2h 18m 38s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/1fe46791eebe4dc99ec05d6f8e12f71b : FAILURE in 2h 57m 04s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/a2d6d93225184e19af934ef96ec08557 : SUCCESS in 1h 24m 46s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/0197661d8496488e9c92edd5a82b8421 : SUCCESS in 39m 43s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/bdd915d40d3648778eec5440a17f432f : FAILURE in 33m 23s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/e04b8270a11547528c08c95b8b3e7368 : FAILURE in 30m 50s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/a86e191917e34e42ae0064919ada5ae7 : FAILURE in 31m 24s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/536e851b9f3646a69e28215aca16295b : FAILURE in 35m 30s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/c0b40ab5fec64493a307f9c395eb3c9c : FAILURE in 32m 50s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/006797c832f145e6bdd4ab15a6c98ca7 : FAILURE in 34m 41s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/5e72017b2a454b618906c9b5086a716b : FAILURE in 30m 59s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/62375a94516e44efad4daba565b301ba : FAILURE in 29m 13s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8e073ea2a8134593959a3a9bb28e2cab : SUCCESS in 3m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e110e5b514b4ef1a7e3467968f97ef4 : SUCCESS in 7m 24s","accounts_in_message":[],"_revision_number":2},{"id":"c3b280bd796aca787c8aba489ded5d3cc739e2fe","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-19 14:33:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/a923ecf26f3a460bad4ceb6501ef71fb : SUCCESS in 3h 43m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0a43a3c5dad40a0f2226872328a81105ea3d0944","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-08-20 13:23:27.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"589b7bd2223ef3e6bc3a7c5463c348b4a9ccdd3a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-20 16:40:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/738306e864fb4075b2fb73c1f77a9b0a : SUCCESS in 3h 15m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"22fb97c7f3eadbbe20d0f88f2c0b39e43cdec93c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 21:54:38.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d110727995184a349179f15a359573ca : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/260446e6c9be47a38c90839bac03a707 : SUCCESS in 5m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/97f067d8eea949489422912590b86a7e : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/64dd8c642e2a4669b5cd084b7237b68a : SUCCESS in 5m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5b6086b5657d4b3095f876731fe3ab8d : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5dbf56a18eed4f9f958e0409e9320a18 : SUCCESS in 7m 49s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/ca74e09cbd474b84b57033ff25f96038 : TIMED_OUT in 3h 37m 14s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/c479311e5dca4352ac412e6fa400873a : SUCCESS in 3h 27m 06s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/c0bd3701aa8940ac87600af2d052f564 : SUCCESS in 2h 01m 27s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/c860bc81a5ca4509a645a44c3d9788e6 : SUCCESS in 2h 27m 12s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/d658eb42445244c8b182d511e5511423 : FAILURE in 1h 57m 08s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/e407c624cd3f4307841a36bf6650c759 : SUCCESS in 1h 50m 41s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/7b044e5b44f54ec18c039cf0f779d34a : SUCCESS in 1h 39m 00s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/950d03819a4c431b887b4b39e89c7994 : SUCCESS in 1h 46m 23s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/abdc345f7e404d4e9c398a32c7377fa6 : FAILURE in 1h 30m 58s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/7b794b0133cc4686bb5c4a212c22354f : SUCCESS in 54m 36s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/3fda8a6f9f1444679bb27ac7ec1e81bb : SUCCESS in 46m 48s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/f9881ccb4fd44b8aa95463ce5e269795 : SUCCESS in 1h 37m 08s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/d36c9515022d405c85eee306138c3c0e : SUCCESS in 1h 28m 43s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/8e0358bcce1a4c788ad5f459b9134240 : SUCCESS in 1h 06m 21s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/b2eecca290f940f1bb1410ff5abfd60a : FAILURE in 2h 13m 20s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/03a04633416347fa91f28dfe6effae37 : SUCCESS in 2h 08m 47s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/4a540afb5b7840b1afc797469a9a143a : SUCCESS in 1h 41m 23s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/1ce7833ab55d40068d1d8c8c354f7429 : SUCCESS in 2h 06m 10s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/ab42e6ee20c34c15b6ccf80148646291 : SUCCESS in 1h 00m 48s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b764649ec30b417c9d887bfa4843ce59 : SUCCESS in 4m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/505ef3f217d64a4c8aebff4f820b1ac0 : SUCCESS in 6m 24s","accounts_in_message":[],"_revision_number":2},{"id":"019a97cffe2e42cbaffb3ad16d86bb85bc89aec4","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-08-21 13:07:58.000000000","message":"Patch Set 2:\n\nrecheck oh come on","accounts_in_message":[],"_revision_number":2},{"id":"0fd9cac85ab0700d1cc805a2a8aa91ba8404f0d7","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-21 16:23:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/9d78ba5fbb004ffd81fc1d1a8f45efe8 : SUCCESS in 3h 14m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e3cb8b1f6aed7eb87bfd57d43185cfbcdcf4e5ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 19:59:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2d4288de25fa4de59604cd6af9453781 : SUCCESS in 7m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/577d60c7dc554428b150313da6b192c0 : SUCCESS in 4m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c4f8842f4c7640159019c94d97c237be : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/836ddf67f4c94488860f8ba6de621815 : SUCCESS in 4m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/251fc25977b04bf287f6385271abb4f5 : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4352a2868c9e4c08b3ca9675f5f58499 : SUCCESS in 4m 59s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/291ed5ba1fa54d8abe2817ee9a056fb8 : SUCCESS in 2h 29m 18s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/343e56bd44f5423abd444116f80ef636 : SUCCESS in 2h 45m 19s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/81838491b8a14960a0fceae7aaba97fd : SUCCESS in 2h 07m 30s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/3c181d9a419f4210b150ceaa3bd73b02 : SUCCESS in 2h 12m 23s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/ac4236df7d1e4979b91b20dc97c31be4 : TIMED_OUT in 3h 17m 47s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/76de334e732c47a5a5dea260d1461834 : SUCCESS in 2h 06m 52s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/7931957eb9e54fc9ae3ee04d2707d3a9 : SUCCESS in 2h 48m 58s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/9eaeb385925e4059a3c561d379c2816d : SUCCESS in 2h 02m 38s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/25cdd9f3dc384deb86734b9dbeb1da58 : FAILURE in 2h 30m 31s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/3800e5a77e7c412db8e23349b20ceeb7 : SUCCESS in 1h 17m 19s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/2cafb91ee1ff486b809367318a820547 : SUCCESS in 30m 04s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/3c67ff210a904b0194bd7b5e3e1d134c : SUCCESS in 2h 06m 08s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/2693e84ebafa41038cc3d82e46164771 : SUCCESS in 1h 34m 29s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/1e1ce09b4b2048a2aa70f1d82e6b33ca : SUCCESS in 1h 20m 56s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/714886e01d2e468cbc9dffed8f506aef : SUCCESS in 1h 51m 21s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/5cd13a80d1b945b5b5bb6035e6217b32 : SUCCESS in 1h 23m 50s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/a573a010e5ce421a822a8e4b4175d701 : FAILURE in 1h 19m 11s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/8482b20cf3ae4de19eb5079830a7b006 : SUCCESS in 1h 20m 33s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/fe87479af6564eb2bfefa3f342fd3d7b : SUCCESS in 1h 06m 42s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3c402bac3aa2429298b111e7da285101 : SUCCESS in 4m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2dc98737564940eb8454d738a10e711f : SUCCESS in 5m 41s","accounts_in_message":[],"_revision_number":2},{"id":"55e0a49e7d44d432b81afe2816dc2da85383ad9e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 20:00:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7f977c0876f3c656f1577709e3d9762a3ac241b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-22 06:21:07.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ae8101bff5b4d8b94f6429356183db8 : SUCCESS in 3m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/30936ee91fad4abbb4821488dbc94986 : SUCCESS in 3m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0795711baa5d4607b4eceb64bfeda4d5 : SUCCESS in 4m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/836afa60eb5d4033845690800467c106 : SUCCESS in 6m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/677003178fb34e89bfa6c8337a306d83 : SUCCESS in 3m 57s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/d76a7b265ab043e4a9cbb9106aef2c39 : SUCCESS in 2h 46m 38s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/e62b29b6e5644f5686c84d46bb252e76 : SUCCESS in 2h 17m 52s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/a5be393c71a746c6972275cd1832d343 : SUCCESS in 1h 31m 38s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/4e37edeff5cf4f9283500cec261ddce0 : SUCCESS in 2h 05m 15s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/42e89c52803648e3b00a52c9813e34b8 : SUCCESS in 2h 19m 19s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/a51a370566a94e96a30410cc196aaf2f : SUCCESS in 1h 39m 47s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/9b4f460a49d342a89fa92e167a12e40e : SUCCESS in 1h 12m 24s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/e79442bf6ebd46e0ac8cee7901e52f47 : SUCCESS in 41m 33s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/b133fd12497348c6a7e5a6ef93f049b0 : SUCCESS in 1h 44m 41s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/92282fe7e0084e0e890874cd2753bd4f : SUCCESS in 1h 31m 27s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/3379a59374604df8af489a07749bfb22 : SUCCESS in 1h 05m 15s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/c9e0b21e293643f186afb38143133e88 : SUCCESS in 1h 06m 10s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/aa599b03b7504b239be44eb698acb07b : SUCCESS in 44m 49s\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/608a4a43ff1646b191a60c9c99600823 : FAILURE in 1h 18m 45s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1f7ba75d5b244939969ea93445bc6798 : SUCCESS in 3m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0961df0171e94a52bbcc68ae10d5085a : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":2},{"id":"3af0ac6b7fa43c2930d85b49c2af156d3e836bb5","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-08-24 13:13:59.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"b7e2d90ac836081294f0a83571b38f8867e38868","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-24 16:33:08.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/d9819489aa5a491a8aaffba5226bcc0b : SUCCESS in 3h 17m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1ace0d4738221719918f11c505d55fe38ae50326","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 21:10:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b971fd0de2e44a9186ad2b048b2170b5 : SUCCESS in 7m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adfbafb483ff48b4a02c7a197163e5ba : SUCCESS in 4m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7c6e1bf589b64bbda643bc5679c48e35 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/becedaf22da04caaa2835547812efc0d : SUCCESS in 6m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/869fd66d6c0245c0b59fe3b4c5b646e0 : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c91c330d8004fdca7c665fc60d37c98 : SUCCESS in 4m 26s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/9941e3d807204973a0af6bf973d023dd : SUCCESS in 2h 12m 24s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/a335bd33c8984f71a21521e2982981f4 : SUCCESS in 2h 45m 21s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/1a50e502c75548b78c6df3a39c8dd097 : SUCCESS in 2h 53m 43s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/eeb3cb23d72a46f79f387adf74962dad : SUCCESS in 2h 36m 47s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/22a661a7022f4fd98f2c620cbc70ea85 : FAILURE in 2h 02m 42s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/7e81fe842aa445eaa980bc1aabafba99 : SUCCESS in 1h 45m 48s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/bc9f76d7d623407d95b965158df1bfa0 : TIMED_OUT in 3h 19m 05s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/3d460ad979d04ce1bc6bb147f650d597 : SUCCESS in 1h 39m 16s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/681a13904c994e4897c6898c0b310a89 : FAILURE in 2h 09m 08s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/fb6e5d910a24429b91e771affd1fd533 : SUCCESS in 1h 19m 46s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/5c72bd465b834abe97288dc541d6230d : SUCCESS in 43m 25s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/3a755c5d711c4dc292a92df6285268ea : SUCCESS in 1h 35m 42s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/3e7706fd29c14c37b508240071503169 : SUCCESS in 1h 23m 33s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/0c654ac9afda4508ad29eac5733c0ccc : SUCCESS in 1h 09m 19s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/8253240392834c7a879a06c8d63259ba : SUCCESS in 1h 24m 46s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/c210935c006a46f0b1a44a2a2d672c3f : SUCCESS in 2h 02m 20s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/8573d2a47ccb44c9b8bb3bf6b8ddfec3 : SUCCESS in 1h 37m 03s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/3a05919165bb4fb4839c607ced16c520 : SUCCESS in 1h 20m 39s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/866967ddb351485da273057850928a6d : SUCCESS in 1h 06m 26s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/aaec1d1efdab440a8f7c8d39cef79f6a : SUCCESS in 4m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/39171cf1c29447848e72377c5a24fe96 : SUCCESS in 5m 49s","accounts_in_message":[],"_revision_number":2},{"id":"34cc19a92a93f7b9e3f5920e34181f068ef570f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 22:37:50.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"72af72e376eed438004c614d5153136962b80dc4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 14:07:44.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7347a78d9c684cb981f81a6af2b26c32 : SUCCESS in 3m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/18ef68f6bbf945d3b1aba1b50743f0fc : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f5cc24575e847f780058d3fdfe8fba5 : SUCCESS in 4m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/de404da17bee47c49170ed37014bc74f : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56d0d88174bf4be68bfbe714b304d79e : SUCCESS in 5m 51s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/4dfbb0aaf5244993a1b7de1f7ea75573 : SUCCESS in 2h 35m 21s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/ad1d9ca39375411fa6cba66032cdba2d : SUCCESS in 2h 15m 47s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/4360ba32dbe34ed6997495fc7b214b05 : SUCCESS in 1h 58m 08s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/3bb6cdf08aa041e0ae81b3f824bb4b24 : SUCCESS in 2h 32m 34s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/fffe8460e824413a872597c258c34464 : SUCCESS in 2h 48m 29s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/ad3cd75058b24eeea7d40af2de073870 : SUCCESS in 2h 02m 50s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/11b2a11e7f0f406bb48075593a27fec3 : SUCCESS in 49m 12s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/2dc7a0577c7947199d7beab71c559543 : SUCCESS in 32m 03s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/18e3d86bce6e40f7b737f61b518ff7e8 : SUCCESS in 2h 00m 20s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/2ea825beee3b45aea84d4eb26b1dbb3e : SUCCESS in 1h 55m 01s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/ea80b859a3af4c7d9a8abc4c4a571270 : SUCCESS in 1h 41m 47s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/fade59e89c4a414aa0224b827bd24e09 : SUCCESS in 1h 49m 48s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/91909b6591c44e1086969c78af9bd2ed : SUCCESS in 2h 13m 23s\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/84a618b95e4b4dd9b8c4f246e3c33d8a : SUCCESS in 1h 34m 47s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/85548fecb9e14178b01dcbc88854c766 : SUCCESS in 4m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/df4feee21fff494f987999c9843808b3 : SUCCESS in 5m 59s","accounts_in_message":[],"_revision_number":2},{"id":"6ca33b4a7c0e4407ccbc190e22909cf7b7c68f5d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 14:07:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"85a0e9c1355a7804bc13bd643c7d66225618188a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 14:11:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c375f40be024b65a0cbfb9be91877e1 : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7f1bf79e792275ff1bf3ad3b4f97951318f87e6f","revisions":{"628a69cff1b7d4754430ff5d568dfaffa7373172":{"kind":"REWORK","_number":1,"created":"2020-08-12 15:41:35.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/24/745924/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/24/745924/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/24/745924/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/24/745924/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/24/745924/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/24/745924/1"}}},"commit":{"parents":[{"commit":"473d9cbc32163ce548e1483e0d5e1631c096c8d2","subject":"Merge \"Properly collect tasks with or conditional in step file.\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/473d9cbc32163ce548e1483e0d5e1631c096c8d2"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2020-08-12 15:38:07.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2020-08-12 15:38:07.000000000","tz":-360},"subject":"[TRAIN-Only] Dynamically add centos8 namespace","message":"[TRAIN-Only] Dynamically add centos8 namespace\n\nThe container namespace is different for the python3/centos8 version of\ntripleo. We need to dynamically add it as the default if not explicitly\nspecified via a ContainerImagePrepare setting. This change adds an\nupdate to the defaults to ensure that the centos8 namespace is changed\nin the defaults only when running under python3 and the default is\ndocker.io/tripleotrain.  This should prevent accidental changes when a\nuser specifies a specific namespace or the defaults are changed.\n\nChange-Id: I3ff5dc84b15a81d6afcc70ab5dde2d327b3c4714\nRelated-Bug: #1887176\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/628a69cff1b7d4754430ff5d568dfaffa7373172"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/628a69cff1b7d4754430ff5d568dfaffa7373172"}]},"branch":"refs/heads/stable/train"},"7f1bf79e792275ff1bf3ad3b4f97951318f87e6f":{"kind":"REWORK","_number":2,"created":"2020-08-12 16:00:18.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/24/745924/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/24/745924/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/24/745924/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/24/745924/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/24/745924/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/24/745924/2"}}},"commit":{"parents":[{"commit":"473d9cbc32163ce548e1483e0d5e1631c096c8d2","subject":"Merge \"Properly collect tasks with or conditional in step file.\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/473d9cbc32163ce548e1483e0d5e1631c096c8d2"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2020-08-12 15:38:07.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2020-08-12 16:00:15.000000000","tz":-360},"subject":"[TRAIN-Only] Dynamically add centos8 namespace","message":"[TRAIN-Only] Dynamically add centos8 namespace\n\nThe container namespace is different for the python3/centos8 version of\ntripleo. We need to dynamically add it as the default if not explicitly\nspecified via a ContainerImagePrepare setting. This change adds an\nupdate to the defaults to ensure that the centos8 namespace is changed\nin the defaults only when running under python3 and the default is\ndocker.io/tripleotrain.  This should prevent accidental changes when a\nuser specifies a specific namespace or the defaults are changed.\n\nChange-Id: I3ff5dc84b15a81d6afcc70ab5dde2d327b3c4714\nRelated-Bug: #1887176\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/7f1bf79e792275ff1bf3ad3b4f97951318f87e6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/7f1bf79e792275ff1bf3ad3b4f97951318f87e6f"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
