)]}'
{"id":"openstack%2Ftripleo-common~746133","triplet_id":"openstack%2Ftripleo-common~stable%2Ftrain~I0f44b21be981f230abdf9baee4eab747a9a46114","project":"openstack/tripleo-common","branch":"stable/train","topic":"bug/1891105-stable/train","hashtags":[],"change_id":"I0f44b21be981f230abdf9baee4eab747a9a46114","subject":"Ensure image files are readable to apache","status":"MERGED","created":"2020-08-13 12:57:31.000000000","updated":"2020-09-18 02:29:37.000000000","submitted":"2020-09-18 02:28:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"746133-1600396107320-cb39844b","meta_rev_id":"a2377bd84360d951e294af2dfa9d691def8f0da7","_number":746133,"virtual_id_number":746133,"owner":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-18 02:28:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-18 16:35:08.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},{"value":0,"date":"2020-08-18 23:19:27.000000000","_account_id":31695,"name":"Zhu Sheng Li","email":"digglife@gmail.com","username":"akarei"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-09-03 18:04:22.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},{"value":0,"_account_id":31695,"name":"Zhu Sheng Li","email":"digglife@gmail.com","username":"akarei"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2020-09-03 18:04:22.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},{"value":0,"_account_id":31695,"name":"Zhu Sheng Li","email":"digglife@gmail.com","username":"akarei"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":31695,"name":"Zhu Sheng Li","email":"digglife@gmail.com","username":"akarei"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-13 13:41:02.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-08-18 05:51:22.000000000","updated_by":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"reviewer":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"state":"REVIEWER"},{"updated":"2020-08-18 16:35:08.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-18 23:19:27.000000000","updated_by":{"_account_id":31695,"name":"Zhu Sheng Li","email":"digglife@gmail.com","username":"akarei"},"reviewer":{"_account_id":31695,"name":"Zhu Sheng Li","email":"digglife@gmail.com","username":"akarei"},"state":"REVIEWER"},{"updated":"2020-09-18 02:28:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0cfa315a551ed67432071ad37699dfbf4aa03c53","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-08-13 12:57:31.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"67aea1bb9df927447bfb3d55fcf50716faccfec7","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-13 13:19:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/f4a8c9f0c3e144d3887e24f1637db8bf : FAILURE in 21m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c5d28cd28d505505e032efaf3daadc79096ad3e8","author":{"_account_id":31695,"name":"Zhu Sheng Li","email":"digglife@gmail.com","username":"akarei"},"date":"2020-08-13 13:31:02.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"bf3aad6b784bfe409a89a3978f4ab6be8b403f3e","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-13 13:41:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2a93279ac57a6360ed5529b8db8cff01ba0a8b16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 15:43:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e6ec9bda5caf4357a38fe2eee8e1e207 : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2c20ea6270b414285fd4c383601b329 : SUCCESS in 4m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/629e9712bddd46c884c8a991868b6a0e : FAILURE in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/66398b9e994f496da60dc216cf6101d2 : SUCCESS in 4m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/62e755e98170411ba973b83883d156fc : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6a1da80b5904ad0b9d25137955311ef : SUCCESS in 5m 41s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/58cbd0a0a15e4f08bee9a67836f00ab2 : FAILURE in 32m 49s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/f8f5d1ec23e94d8882c0fec8809868fb : FAILURE in 2h 01m 44s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/29192728d787461ca98e927ccab5dd34 : FAILURE in 1h 26m 52s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/0b44618a38984b78b4e45949d7464c1f : SUCCESS in 57m 09s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/8223c1cabc7e4b3dae3de91588d9c0a4 : SUCCESS in 41m 04s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1bf0b4a56b904f50a319618d9919c6a6 : SUCCESS in 4m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fcf4442eb38b43f68696addb44035db6 : SUCCESS in 7m 00s","accounts_in_message":[],"_revision_number":1},{"id":"ed4dfbd2caa92f213c7c08408d99781c93c14325","author":{"_account_id":31695,"name":"Zhu Sheng Li","email":"digglife@gmail.com","username":"akarei"},"date":"2020-08-14 10:16:36.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ffa689ecb21bb51de77d29ab523f240494ae69e5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-14 10:29:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/13ae85f42a86491c912bf2e17bb79a42 : FAILURE in 11m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"483128a8eddb29e0ec7aa6f67947f20126c9754f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 11:10:44.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d62a1a2091914b70b518a859b4eaaaf5 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db718dbc3c784e92a4f14b2a53913f19 : SUCCESS in 3m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b41a4e0de01c4a25b29a456c053e8c4a : FAILURE in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4d3431a7eecf4c8296fd10ca50bbea0c : SUCCESS in 5m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ea0436c226bc4dfd823b6a6add5e285a : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e00708b3ce6c412b8a16660cf4451cdc : SUCCESS in 5m 04s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/87d687d579974dff95286e4bacdb5e53 : FAILURE in 10m 06s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/db85f31f579845cf9b1f8a34027685c9 : FAILURE in 6m 50s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/d437ef19d4f34ce4845298fdcc927d39 : FAILURE in 6m 48s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/2ff38812647b434c89784db4a913d2f1 : SUCCESS in 51m 57s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/74eb499710e7410db0d448a118d17388 : SUCCESS in 37m 52s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/13483331e9c641288626073ba45869fe : SUCCESS in 3m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d5328fbf9c204e899c95411af1cd038a : SUCCESS in 6m 28s","accounts_in_message":[],"_revision_number":1},{"id":"73ac7a2a15c4cfb441032eec48bf0f27e01fe2ef","author":{"_account_id":31695,"name":"Zhu Sheng Li","email":"digglife@gmail.com","username":"akarei"},"date":"2020-08-15 17:20:46.000000000","message":"Patch Set 1:\n\nThe test didn\u0027t pass because built-in `open` function in python2 does not accept file descriptor(integer) as the first argument. What a surprise...(maybe not that surprise ;)). I think the best way to solve it is using `os.fdopen` instead of `open` to open the fd, which works for both python versions. \n\nI\u0027m not sure how to handle this situation in Gerrit. Do I need to submit another patch?","accounts_in_message":[],"_revision_number":1},{"id":"f7c6887b0cc711f09de03ff4526bf11716d67340","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-08-17 15:55:24.000000000","message":"Patch Set 1:\n\nyou can just fix this patch, train is the last version we need both py2 and py3 support","accounts_in_message":[],"_revision_number":1},{"id":"58a564e4d0350ff3a116148249ce10632af1225d","author":{"_account_id":31695,"name":"Zhu Sheng Li","email":"digglife@gmail.com","username":"akarei"},"date":"2020-08-18 05:23:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"343b64605dbd499d40f4aea476e93451939f68d4","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-18 05:50:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/345a5a18be144d8284fbc42324de794f : FAILURE in 25m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fa0c8b1ed5873d057f648a381cd73877701faa10","author":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"date":"2020-08-18 05:51:22.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"29bf868ea04dd7246ffa7dd67cd9f5caa6e43076","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 08:01:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3a0ccbee2704cbab5e28461481c5b63 : FAILURE in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bccf86f2c09c4ec6ae0cbd4966cee1ba : SUCCESS in 3m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bedd6592d98544508b4938316e35f749 : FAILURE in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5d62217995140dfb4ef950a72b18f17 : FAILURE in 4m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a4f1a76a2c024023b2e0b0d0a71562ef : FAILURE in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1aea7f0c708a4f44b00c73253dfbe53d : SUCCESS in 4m 11s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/757dadf85667455284b475cb367805b5 : FAILURE in 45m 16s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/530d8d83d94342dab2dde70ba168b968 : FAILURE in 1h 56m 12s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/928e527ddef245d38decf967939b2f38 : FAILURE in 2h 35m 56s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/eb380824c7f74abca9832d5511ac7b3f : FAILURE in 1h 30m 14s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/89b8a93d97c34500ad74c35921e6f10f : SUCCESS in 55m 26s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ef28b25d648c48069021c4c919db9497 : SUCCESS in 4m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f1fca003e0ae42d7b821ae9568e31198 : FAILURE in 6m 25s","accounts_in_message":[],"_revision_number":2},{"id":"529ab7e0ab69e2beb7ee7d6b7e21a0a8b02fb565","author":{"_account_id":31695,"name":"Zhu Sheng Li","email":"digglife@gmail.com","username":"akarei"},"date":"2020-08-18 13:09:02.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9fb62b410c708b91baf90a73e4dccad0c25d8bb3","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-18 16:35:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/cf5c2e2d955a454abe63444d415c80d2 : SUCCESS in 3h 25m 05s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"af923aa9542654aa3d0c4a32f170797ac443fa78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 17:26:21.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c15a9ea729094522afea1caa370cba54 : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83aa5836c9b245d883826fca5671ac77 : FAILURE in 3m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f0fed3f5a8fe4955ad33f66a109f7f1c : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ccce7b68d4d4cecab5f8b92a1bb3870 : SUCCESS in 5m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/996cb76fe670450e8a6b496b0d15cf29 : SUCCESS in 6m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9c25ae560ee4cda8ea8d611a6b39166 : SUCCESS in 4m 46s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/da79f996be114b37a0d1cf3ff089766c : SUCCESS in 2h 07m 18s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/33b363f839384eef987950db71703d5b : FAILURE in 30m 05s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/9319988e0be04943aae7f10f6173164d : FAILURE in 1h 43m 51s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/a1fd37753fab4268b68ebc36c606a572 : FAILURE in 34m 01s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/6c773f1510f44cfcaf3cfcc0ff2649d3 : SUCCESS in 1h 21m 41s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/6d0a7e53a02b4b5f96259c9dc3905413 : SUCCESS in 1h 38m 09s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/63bd4f0d79244e9092a34ab515d3794a : SUCCESS in 1h 12m 58s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/3af147c0776749878a2f419d2c19ee31 : FAILURE in 2h 15m 15s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/b80c4acf5081416ebf55ff835c0eae71 : SUCCESS in 1h 15m 14s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/87d1c44f085c4098a9c7afb4e540539c : SUCCESS in 34m 20s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/41c56a482dca4c90be7a66d161519e5b : SUCCESS in 4m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b5b7833aaa5b4c10b3da66341d900969 : SUCCESS in 6m 49s","accounts_in_message":[],"_revision_number":3},{"id":"dfeee6c8953de4c61824585b2f021093e1b42033","author":{"_account_id":31695,"name":"Zhu Sheng Li","email":"digglife@gmail.com","username":"akarei"},"date":"2020-08-18 23:19:27.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"5e3cb4db354fa4c305d52772539b1f3a1f78f881","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-19 01:54:47.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f78fe0aaf4884cb4a938e69f4f3c53d8 : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f40ef74796949d58249d23499ec4562 : SUCCESS in 4m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ab4c7b8827374c0099ebfdf06fbf490f : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/98a174aa2d284acf9897e305f83920f9 : SUCCESS in 5m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8ddb6e860d8749ac91176f3bed04ab2f : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed3b9bec237a4419b798cf5fd4e383dc : SUCCESS in 5m 46s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/7c1e3df474ae4d30ac9901b2de35495a : SUCCESS in 1h 35m 02s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/612f6536c44446c6aacb9efd0b39ca81 : SUCCESS in 2h 20m 14s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/17175f898fb2460089d791175cf533aa : FAILURE in 27m 25s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/16ee1e1f96484e4887f2780738c8f1d5 : SUCCESS in 1h 39m 17s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/6356906460044c77a9839406e50b86ac : FAILURE in 53m 56s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/c034064a9f3d4ac3997ddf96c71ed967 : FAILURE in 33m 07s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/32a32e7f54704d1285bf202407e04aca : SUCCESS in 1h 14m 07s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/1d73a935728c42bf9997b2e5377366ce : FAILURE in 2h 14m 13s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/2394b5c92600488eb0cda7392faf6530 : FAILURE in 1h 31m 33s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/ac3cbdbffa824079a66bcc74331de31e : SUCCESS in 1h 16m 14s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9b1d26fb6e2d421eb6db863ab26d9661 : SUCCESS in 2m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d6f90f319e7e4fb883d50ea2bdbd2916 : SUCCESS in 5m 58s","accounts_in_message":[],"_revision_number":3},{"id":"7f40f66c073703c9fbb027bc62c2ed16ffb08ee1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-19 02:19:37.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/596bfb3120d64cf5a71f45ec67ecff69 : SUCCESS in 2h 59m 01s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f57460ec2bb1ab295d33c03372aec80271e424df","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-08-21 14:02:22.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"6e070546991420724160bf1aba80ca2b14231666","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-21 17:20:33.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/859acf67f12b4a13b0edae0e5cbd2020 : SUCCESS in 3h 13m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c94ec2d8725c9d0f0ebf7ac571513ad169fb8504","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 20:37:46.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2ad407b6324b4c11a68d8640566407fc : SUCCESS in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59ae5baa5c4e491789e243040906e780 : SUCCESS in 3m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c5ac5e081e8b4386bec33bf186276772 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b6e67c629a054af09282bc0089d69c75 : SUCCESS in 4m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/01f515574abb4f28955bde6b83819f50 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b4cac4d614d40ddbe52bfd6f90e8b45 : SUCCESS in 4m 29s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/10419474359b486c86e6dc1812de0031 : SUCCESS in 2h 18m 01s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/f9338d32cfc5473f9ddf6f64f7b37594 : FAILURE in 1h 17m 44s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/cff656cb6535438e9aaf927afa124f6c : SUCCESS in 2h 25m 11s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/4c81266748c14bbc817ea715f1b68dfd : FAILURE in 1h 21m 16s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/e12ec9b38b3a41338e4b29e972228fec : SUCCESS in 1h 46m 08s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/99f0261bfe3f467a93e1be63221430a4 : SUCCESS in 2h 11m 47s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/8159b000d52441f4a693ca863666f05e : SUCCESS in 1h 52m 55s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/d9e6add3f2cb44259cc755e87e52e916 : FAILURE in 2h 04m 17s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/aa0e161d67d74772b0be37d5c3c4bbe6 : SUCCESS in 1h 02m 38s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/c00dd9c382064ce38d20495b9340c829 : SUCCESS in 47m 44s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f23df17f5adb47fa9aaa1d774c8acef2 : SUCCESS in 3m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/73079c0a8b20437986c97f2a2315061c : SUCCESS in 6m 53s","accounts_in_message":[],"_revision_number":3},{"id":"5401c985c5c82995d0539de7fd4d507b3a75aa10","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-09-02 19:03:01.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"34db3cccdbe1e09b498add198ec47242bbf55a35","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-09-02 22:15:52.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/c67d0e63c03b40298148328bccf5e353 : SUCCESS in 3h 11m 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d164b45001f578871f3e3c51750040008839601b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 23:58:23.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/582b4d70b73e4e46bd91dca000521725 : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbb9f5cd84a445d1978119e930f18e7d : SUCCESS in 3m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e69e724cdee440c09816e077b4254d0f : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f7262a8991543f9b828bf513093dc7a : SUCCESS in 4m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3f209ba145e7476fbfa8521fb9370ec7 : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c752100e09694d099b59e6d7e604c4c6 : SUCCESS in 4m 09s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/59d873663bf940a68d6204875063c85d : SUCCESS in 2h 19m 18s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/d3739aa8cd424e2d83fb25ff74f53c85 : SUCCESS in 2h 35m 52s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/ead5d585d23b40ce9f7ff7ae3c4cb1f2 : POST_FAILURE in 3h 14m 26s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/560fb10d32fb472db9f30acec698e8fa : SUCCESS in 2h 59m 07s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/f40a17f8ef1145b7bfb9d03d7ab8a670 : SUCCESS in 2h 31m 49s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/dcbad401dd2746f0bb3349778e8e7270 : SUCCESS in 1h 39m 16s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/831b8fed6697442f9a760ae4d186fcaf : SUCCESS in 1h 17m 14s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/4318741e32604a29b9292570e4f98879 : FAILURE in 1h 33m 05s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/6a4b7774ddf74677b2303c95968960ff : SUCCESS in 55m 43s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/ea74dda008824a3dbee762a496cf181f : SUCCESS in 51m 55s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d99eb68840ed4aedb0311559c209c7b5 : SUCCESS in 5m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/be70666cfc7a4ce09eb1ac7802f850f0 : SUCCESS in 6m 19s","accounts_in_message":[],"_revision_number":3},{"id":"d3753fd8edeb5ab4a51b063f4061a76dd1f510b2","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-09-03 12:58:39.000000000","message":"Patch Set 3:\n\nrecheck follow the bouncing failure","accounts_in_message":[],"_revision_number":3},{"id":"86946b2402649d0a07ab50025bbf117b91b88aba","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-09-03 16:21:25.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/aca6ccc4ff4349ac9df62d734621d955 : SUCCESS in 3h 21m 15s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"93a64e5eab578a0c79ab7f13c1e92005a87e644c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 17:21:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e5899b660bb14823ba7d99ba14be7c42 : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/637674b4c7174acc8bdfe366a2f21ed1 : SUCCESS in 4m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8c0b7d387c894b7aa190caa3204de8f9 : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b91bb6ba3054b8783ab160de7e9efdc : SUCCESS in 5m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/efe5be96c3b54cb88a87a32aac89021b : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af8e901d186c44489195f22a8d27de1b : SUCCESS in 4m 11s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/71fa0d889b744f61bf469a3e290969a6 : SUCCESS in 2h 28m 12s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/f2d6a31612cf4107b87587a7d12cb7e2 : SUCCESS in 2h 31m 06s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/c63aee5c12a74b62ac78612fc09229f7 : SUCCESS in 2h 12m 08s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/c4967ea3f7904a9b9c01e5962ccf2626 : SUCCESS in 2h 30m 22s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/04882492f2964887ba1fdcdd9477ea6c : SUCCESS in 1h 59m 28s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/8674ba8adade4a13963ad3febcd16160 : SUCCESS in 2h 26m 09s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/d3fb2037f54f4b3983af1aae2e3420ad : SUCCESS in 2h 02m 35s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/3aa8c414d8bf490a86ee2139525b2c87 : FAILURE in 2h 19m 05s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/32417406c8d2411a96b5bd2264557f3a : SUCCESS in 42m 31s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/8cf900c9ff954fd2a0a7954b30dfac82 : SUCCESS in 43m 14s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/673d5310140c446cb608759c700d2233 : SUCCESS in 5m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ee846b01939b4deda767f8608b8f0a87 : SUCCESS in 5m 48s","accounts_in_message":[],"_revision_number":3},{"id":"75088cdd4a77429bdb6939251863cde706ff97ad","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-09-03 18:04:22.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"6a3ce38d0fcb575a5e2b674c56667bf84b058d44","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 21:54:54.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a153c40416e50949531e205a3ccdf126d00e3f09","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 03:45:19.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4667bb286fd540b3b891c80efe89e7a5 : SUCCESS in 3m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/69e05722dd01463684a6dd3dbe9ea89e : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/499272e90a2c4f2497f2935bd81a8a0d : SUCCESS in 4m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bd213091fb3b491d82f79c2e6098f1a6 : SUCCESS in 6m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/faf7bedd940847d389334e1e8cb073f2 : SUCCESS in 5m 48s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/0f9447b2f3fa4497a777c41ed1267ed1 : SUCCESS in 3h 10m 29s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/16953260e388460ea675eaac6a363362 : SUCCESS in 2h 32m 40s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/afaa2fff232542cc84279e05f24aecf7 : SUCCESS in 1h 42m 10s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/f0633e1ff1e24de59d6a7b15bb889cc3 : SUCCESS in 2h 05m 41s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/abc1652d1ccb401c94303f2942053169 : TIMED_OUT in 3h 12m 44s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/7d17b7e4794f4cbab15e0cff3485da2b : SUCCESS in 2h 12m 09s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/4c975a92630047eb951754637bca35ec : SUCCESS in 1h 45m 10s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/3f9dc022f884489b9d52989587c65360 : SUCCESS in 1h 24m 18s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/94461b9153274a4392ea3eb32b45db18 : SUCCESS in 40m 51s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/66245776dd4a479a8e87d762967c3dc5 : SUCCESS in 4m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/06cabb09e67247858e8e180102f8402b : SUCCESS in 6m 11s","accounts_in_message":[],"_revision_number":3},{"id":"8ce0ffb845c3c6f30e8e45853418e81d0a6c1dd4","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-09-04 13:19:46.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"954415efd37e37f0bd31f32fd7e0f7676ab8c2f5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-09-04 15:40:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/16c90907758641ae840fcfb5baf2ad6e : FAILURE in 2h 19m 35s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"445fcf7eebf0049cbfbcc5d491e2247444f89c38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 16:35:54.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/02c5c98a4eb9443b8dfbbbaea37a136b : SUCCESS in 9m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eae6a1603661444e845d561ff059f69b : SUCCESS in 3m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/91a6d3fb6c0a4976b4e3ad70cab753d7 : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f82a12ceaae4b199b7d0fe822953449 : SUCCESS in 4m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/463359d6d5cb4c8c91e90d365401d905 : TIMED_OUT in 41m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5424bd92a800475b9f6167643ba88e7a : SUCCESS in 5m 57s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/fed3b38c4d0b4e229c526deaa581a4e5 : SUCCESS in 2h 21m 40s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/7298ab06ae3d497c922b4a2cbd832513 : SUCCESS in 2h 06m 49s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/348dfb6b0fad443b830cf71657e319da : FAILURE in 2h 12m 39s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/c8fc03062a994a2ba6d21d272302dbac : SUCCESS in 1h 36m 42s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/bf03ea60c235476fb3211a90d7ab34fd : SUCCESS in 46m 55s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/28cae7fcae84405e94f5bfeb40058278 : SUCCESS in 3m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1e83e6fc822441bb82f054f291e8162b : SUCCESS in 6m 28s","accounts_in_message":[],"_revision_number":3},{"id":"274a2ad3a6fdc96701f010e68f3fbe9925950637","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-09-17 13:04:49.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"fe0dc254b3ff4c1476075888dabbdf2d724c2722","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-09-17 16:33:16.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/111fc030f86748caa0662f4e7633ab3a : SUCCESS in 3h 26m 10s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4c3ae21079708443a9b7134382764482115b9008","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 17:58:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f7fd8cc164d4483aa905c5e71f9a0fbf : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f309e945fe54757be67ed30ad84a52f : SUCCESS in 3m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7b412451892c4707af5527a593262659 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7cbe46f97bf04f55845fe7b5ae1539c4 : SUCCESS in 4m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5b2da05dc57a43eda86a6f3787637917 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fc04ea51e004c80bf1f8e36e7f376c6 : SUCCESS in 5m 03s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/85e43e69d84846f5884e1851c5be0dfd : SUCCESS in 2h 14m 18s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/2fc65937adf349c48c457142fde759cd : SUCCESS in 2h 23m 36s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/c113278152514abe821ac336cd88b596 : SUCCESS in 2h 01m 42s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/ba45ea9a64894ee285f5afabe112f2a0 : SUCCESS in 2h 10m 29s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/f35b3ff23c244c6996521ed22a74ddcd : SUCCESS in 2h 12m 38s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/35a05bcad95c459785c2f46bbddba3a1 : FAILURE in 2h 21m 49s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/9716d164c5b14ed0a3d83d337ba6855a : SUCCESS in 1h 56m 31s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/dac8665421174b5386fb9e7fefd18be2 : FAILURE in 2h 03m 21s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/20bf948e53eb44ba882665ac7167bfe6 : SUCCESS in 58m 25s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/adf97a885da045a09f73727342828046 : SUCCESS in 4m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/67040f4ba413477ea2a5b46a3b1048fd : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":3},{"id":"79489eb92168af8455b01419f19d68fd0602e2ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 17:58:38.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"614fbc8cedfd443e9da4dc8839d5689ebd372bfc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 02:28:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"009e76a4431cb54aa6f1594666fecf573f82b601","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 02:28:27.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf8fbc3083e14d258ce93d80cec28343 : SUCCESS in 3m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7ea0889928304471a597636ebe4e0c1f : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/23d39a27b2a749a693fa9a4c081e72a6 : SUCCESS in 4m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d51dd15f1d7a40ca8a479d392b1ac2bf : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0efd0fd803c42a8a1d2b0e44a9b9242 : SUCCESS in 4m 26s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/2d5f5d50f60c4d33a049f38d9b37cc39 : SUCCESS in 2h 29m 52s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/0d13b79de95b42c5bef10daf18d035fc : SUCCESS in 2h 12m 23s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/4fc64e876d4f4241b47c6dc215ccc148 : SUCCESS in 1h 26m 23s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/df9f186fc92b42c19cd5a17c1119dc04 : SUCCESS in 1h 53m 35s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/0bf05ed6935745abbce9e7aed62fd3ce : SUCCESS in 1h 53m 05s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/515290ea766f4b039e77c8583e4441c9 : FAILURE in 2h 07m 54s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/1340ebaaa6af42baba6272fb3bc467ce : SUCCESS in 1h 40m 59s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/5ff043fd8d2b409fb97e5a7d04d51222 : SUCCESS in 1h 00m 04s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/23cbd09dd8e247fa8c689701a1be5a44 : SUCCESS in 43m 36s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d0dbd82d74f44ca0b1fd513cd55cc857 : SUCCESS in 3m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2ab269e305974da5a9b67100b06146da : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":3},{"id":"321116b5b8e8603d60f0166123a24a82b33b126a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 02:29:37.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11f402d337474a16bd2076ddb37613d9 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"898277149026c09e1a9c8a0732a48d8fa9a39465","revisions":{"2dcdf340d67da993a7304851de884a89f52e037a":{"kind":"REWORK","_number":1,"created":"2020-08-13 12:57:31.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/33/746133/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/33/746133/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/33/746133/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/33/746133/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/33/746133/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/33/746133/1"}}},"commit":{"parents":[{"commit":"21b4a1f3511fec6eab2c25db52acd85e1693ab84","subject":"Merge \"Switch to use virt:8.3 for non-kolla container builds\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/21b4a1f3511fec6eab2c25db52acd85e1693ab84"}]}],"author":{"name":"Zhu Sheng Li","email":"zshengli@cn.ibm.com","date":"2020-08-11 09:10:23.000000000","tz":480},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2020-08-13 12:57:31.000000000","tz":0},"subject":"Ensure image files are readable to apache","message":"Ensure image files are readable to apache\n\nIf the umask on undercloud is set as something like `077`, the image\nfiles created with `sudo` will not be readable to `others` including\n`apache`, which will cause 403 forbidden during image pulling on\novercloud.\n\nSo set the permission explicitly just like we already did for\n`make_dir` function.\n\nChange-Id: I0f44b21be981f230abdf9baee4eab747a9a46114\n(cherry picked from commit a6bbe528973976b95a5ce39647f0176b8f1c84a3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/2dcdf340d67da993a7304851de884a89f52e037a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/2dcdf340d67da993a7304851de884a89f52e037a"}]},"branch":"refs/heads/stable/train"},"805c562524d210d747bb5d74e71e5846baee21e3":{"kind":"REWORK","_number":2,"created":"2020-08-18 05:23:36.000000000","uploader":{"_account_id":31695,"name":"Zhu Sheng Li","email":"digglife@gmail.com","username":"akarei"},"ref":"refs/changes/33/746133/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/33/746133/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/33/746133/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/33/746133/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/33/746133/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/33/746133/2"}}},"commit":{"parents":[{"commit":"21b4a1f3511fec6eab2c25db52acd85e1693ab84","subject":"Merge \"Switch to use virt:8.3 for non-kolla container builds\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/21b4a1f3511fec6eab2c25db52acd85e1693ab84"}]}],"author":{"name":"Zhu Sheng Li","email":"zshengli@cn.ibm.com","date":"2020-08-11 09:10:23.000000000","tz":480},"committer":{"name":"Zhu Sheng Li","email":"zshengli@cn.ibm.com","date":"2020-08-18 05:19:45.000000000","tz":480},"subject":"Ensure image files are readable to apache","message":"Ensure image files are readable to apache\n\nIf the umask on undercloud is set as something like `077`, the image\nfiles created with `sudo` will not be readable to `others` including\n`apache`, which will cause 403 forbidden during image pulling on\novercloud.\n\nSo set the permission explicitly just like we already did for\n`make_dir` function.\n\nChange-Id: I0f44b21be981f230abdf9baee4eab747a9a46114\n(cherry picked from commit a6bbe528973976b95a5ce39647f0176b8f1c84a3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/805c562524d210d747bb5d74e71e5846baee21e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/805c562524d210d747bb5d74e71e5846baee21e3"}]},"branch":"refs/heads/stable/train"},"898277149026c09e1a9c8a0732a48d8fa9a39465":{"kind":"REWORK","_number":3,"created":"2020-08-18 13:09:02.000000000","uploader":{"_account_id":31695,"name":"Zhu Sheng Li","email":"digglife@gmail.com","username":"akarei"},"ref":"refs/changes/33/746133/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/33/746133/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/33/746133/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/33/746133/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/33/746133/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/33/746133/3"}}},"commit":{"parents":[{"commit":"21b4a1f3511fec6eab2c25db52acd85e1693ab84","subject":"Merge \"Switch to use virt:8.3 for non-kolla container builds\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/21b4a1f3511fec6eab2c25db52acd85e1693ab84"}]}],"author":{"name":"Zhu Sheng Li","email":"zshengli@cn.ibm.com","date":"2020-08-11 09:10:23.000000000","tz":480},"committer":{"name":"Zhu Sheng Li","email":"zshengli@cn.ibm.com","date":"2020-08-18 13:04:53.000000000","tz":480},"subject":"Ensure image files are readable to apache","message":"Ensure image files are readable to apache\n\nIf the umask on undercloud is set as something like `077`, the image\nfiles created with `sudo` will not be readable to `others` including\n`apache`, which will cause 403 forbidden during image pulling on\novercloud.\n\nSo set the permission explicitly just like we already did for\n`make_dir` function.\n\nChange-Id: I0f44b21be981f230abdf9baee4eab747a9a46114\n(cherry picked from commit a6bbe528973976b95a5ce39647f0176b8f1c84a3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/898277149026c09e1a9c8a0732a48d8fa9a39465"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/898277149026c09e1a9c8a0732a48d8fa9a39465"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
