)]}'
{"id":"openstack%2Ftripleo-common~756475","triplet_id":"openstack%2Ftripleo-common~stable%2Ftrain~Ide21daaac9a558561d0943cc0a380952cf7e29b3","project":"openstack/tripleo-common","branch":"stable/train","topic":"validation/library","hashtags":[],"change_id":"Ide21daaac9a558561d0943cc0a380952cf7e29b3","subject":"Add both validations path for inflight validation","status":"MERGED","created":"2020-10-07 09:50:24.000000000","updated":"2020-10-11 00:48:52.000000000","submitted":"2020-10-11 00:47:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"756475-1602377252587-3e075d27","meta_rev_id":"d729179d601574944b4d0c4e23f5ebeda944cdf9","_number":756475,"virtual_id_number":756475,"owner":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"date":"2020-10-10 15:31:24.000000000","_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-11 00:47:32.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-10-10 18:49:12.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"all":[{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":2,"date":"2020-10-09 11:34:26.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"all":[{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":1,"date":"2020-10-09 11:34:26.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-07 10:02:37.000000000","updated_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"reviewer":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"state":"REVIEWER"},{"updated":"2020-10-09 11:34:26.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2020-10-10 15:31:24.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2020-10-10 18:49:12.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-11 00:47:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e5b95710e48e4b1a76676f4558affcce44be554f","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2020-10-07 09:50:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a3a5fb244d27cc95123cce8243f991913321c8d2","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2020-10-07 10:02:37.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f041bfcaea1431a45ce20187fe9fb39e24edc0c9","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2020-10-07 10:02:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b8870f5f79a9bacf3c8b1525eacaa3d3a72f4bc1","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2020-10-07 10:07:16.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8cb80f9a9558e05679e52a62826bab14602767f3","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2020-10-07 10:20:35.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"b33ee4f958b4456e648635f1e79062acfeaa9db5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 13:35:21.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef65805f8f8147b4b0472652e4a71be7 : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13ea9f1a20ed4503b8ed271ea543f125 : SUCCESS in 3m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9a40397ec74f4b2f9ed5ca8cb45a9372 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0beab3897b1e46b78335b2329826f1d0 : SUCCESS in 4m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f5d74ac7c8be4300879700a1072c4ca3 : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f903fd70f3a1470cb33de551e48818ec : SUCCESS in 4m 08s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/90f8f1e7dacc4bec9b25b3ea9e2dc415 : SUCCESS in 2h 15m 03s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/f39dc59f60324a42940ae2d551c25499 : SUCCESS in 2h 04m 19s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/35c1dbad89ea47b3a694e2a2147f98be : SUCCESS in 2h 24m 06s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/905fc8034d4c4422bf78e9533f96d15c : SUCCESS in 1h 44m 38s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/3f022bfcd0ac48bdb9e57b114c30abb2 : FAILURE in 2h 09m 51s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/8fd5f3b9ef7840da81efb84a61b242cd : SUCCESS in 1h 43m 10s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/14f36afa37fd4b68b0d7fdd65841e41f : FAILURE in 2h 09m 02s (non-voting)\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/af46ab7f247e44a4a48b57e0c8570e3b : SUCCESS in 1h 37m 22s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/9fae1e0309164367b217e7289ecc51bd : SUCCESS in 1h 39m 29s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/4c7abf683afa4a52ac299c02433d0437 : SUCCESS in 2h 11m 55s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/167c8ec62d1d4a05abf989deb2a4e912 : FAILURE in 1h 26m 56s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/313ac778e17e4c06842f57d7278ffebd : SUCCESS in 1h 11m 51s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/3b2cd72ca18847a5929bdb4b9bf8c4c1 : SUCCESS in 1h 11m 52s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/73f1473d5cb14c94b1f1fc6dbf7c89e4 : SUCCESS in 4m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d685ac9f592a43619c510a6ebf31eb3a : SUCCESS in 6m 10s","accounts_in_message":[],"_revision_number":4},{"id":"c88bdbeadcc9223c0e4dc8c1d0fb0aabb925cae1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-10-07 13:47:38.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/c456ba5c42884c168df6d3b26ef758dd : SUCCESS in 3h 25m 49s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1cb040affa5eb8a425a7c566ff44f48efbe0848c","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2020-10-08 14:27:35.000000000","message":"Patch Set 4: Code-Review+2\n\nvalidation related - using core on this one.","accounts_in_message":[],"_revision_number":4},{"id":"4c4377207361fb38f89059080d72f660ce0bd8d3","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2020-10-09 07:01:31.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"5e70160bdff8bcfad91b6d76234c1ed8264c2541","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2020-10-09 11:34:26.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\nbackport - validation related - merging.","accounts_in_message":[],"_revision_number":5},{"id":"62475f0dca8d03aa5fd84f092d6371ef8bba8b3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 11:34:34.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/56731a5ff4064f9ca73007180891dca5 : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd0e5ae3047d491cb17cb7ef71684281 : SUCCESS in 3m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6870f55b67314b578969c87b072e570e : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a70dc215528944b1aea7d2044c779870 : SUCCESS in 4m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/939dd6e4375f4167a8cf523d1a896e29 : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/510e93d528cb4021ba7390f5ecfc0e13 : SUCCESS in 5m 13s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/dedb58c4bc024a00949978acb6c30561 : SUCCESS in 2h 22m 41s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/0ed5c1eadb6841bfa1e283e48ebc13e3 : SUCCESS in 3h 16m 51s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/79929ec7c1044c0b982ae1216dae4259 : SUCCESS in 2h 11m 22s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/ad37455db5f54e54b0fda0c771687fa7 : SUCCESS in 2h 09m 18s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/498a6c8e166c472eb81f45f7f99b5de5 : FAILURE in 3h 05m 15s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/7f12f1603a724f1a9a079098ea2afa7b : SUCCESS in 2h 01m 56s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/cbfee529328641329a08cd1112268f43 : FAILURE in 2h 09m 43s (non-voting)\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/d4a04fc4bfd4435aaece3fd8fc56b253 : SUCCESS in 1h 30m 45s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/f2403f175aa7461b86a15e6bf06e76d7 : SUCCESS in 1h 26m 57s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/26a57be85edd4c619e954fe92acb4507 : SUCCESS in 2h 01m 15s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/6d63a7504a08420496924f4a33b24f46 : FAILURE in 1h 19m 17s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/8d04d32ae17742e89bb6acc1fce76bec : SUCCESS in 1h 15m 30s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/1ad98e3fc8a140cda9d95562a94ac201 : SUCCESS in 1h 12m 47s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f08627390d544e368427bbc696d59580 : SUCCESS in 3m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/35717efea578401bbdce2c8e8cf50c85 : SUCCESS in 6m 07s","accounts_in_message":[],"_revision_number":5},{"id":"be2d776200dcf92ee8351b5ded53017f8418225c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 11:35:03.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"ec0e0dc33f724d0db28685602b4517810f16b804","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 21:37:38.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d101332619d45f2a0f61aee9583d385 : SUCCESS in 3m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/34c68aa5b2ad484a82818eb81ac51afc : RETRY_LIMIT in 3m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/11effdb11a33450b926a73d3b892f6a7 : SUCCESS in 5m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c2e2cecae8cc489fbe14f77a7f519290 : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/849dcf9b52984815976cef1c1044a696 : SUCCESS in 4m 34s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/d1415e1720d54f4fbadc3f7bd2b7c18c : SUCCESS in 2h 34m 28s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/e8cdf7130da4433486adfa36f3f25ffd : FAILURE in 2h 12m 09s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/515e3c60d4184810ac9d08afd81707aa : SUCCESS in 2h 23m 56s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/8ed64c5b3476479494eed2eace7f5bc3 : SUCCESS in 1h 36m 43s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/7c7157fd4a78400590351f9bc6539b2f : RETRY_LIMIT in 8m 47s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/4a8a7022d4a14428bad514791578c0a9 : SUCCESS in 1h 52m 31s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/893faa35a15a47e6bc1298ac939a0d16 : RETRY_LIMIT in 5m 57s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/f45cbf1fcb854c0988a26f0ce65f714c : SUCCESS in 1h 45m 10s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/51c1b06774e0433a92fa46a949b869dc : SUCCESS in 2h 05m 59s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/5321c8cf6152420592c7bdb92bfafb20 : SUCCESS in 1h 18m 37s\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/c5587570ee9e4dfa989bab9573b96082 : SUCCESS in 1h 27m 34s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/dd46216d5d7b42628c284c5f3d7d53db : SUCCESS in 3m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e6618e41f5114d9485758f22922dad15 : SUCCESS in 6m 23s","accounts_in_message":[],"_revision_number":5},{"id":"dab4a222556af9516b483c1f435193395f11b604","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-10-10 15:31:24.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"c173753edd3257e20fb8ac731b3cde4b08d9b5c1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-10-10 18:49:12.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/0194891f269c421aaf9e723006dc9a56 : SUCCESS in 3h 15m 57s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"58eadd3a69ca47b756981c660d10a8626b5a03a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-10 19:19:11.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/554aa51d41564d6f889aa034251f6993 : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b20e3a21d38f4196976d51c239c64017 : SUCCESS in 3m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0c2cc11a00b8409fae2f019ec1d4c115 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f7c498a4d9e7479ca08d97f8856baf86 : SUCCESS in 3m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b50493bfd8114083b481faf2445b78fe : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a56f734a8c7e4d3fb5eb642f7c3477f6 : SUCCESS in 4m 07s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/66e898ade4f5475cb483d0cdc92395f9 : SUCCESS in 3h 08m 48s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/a1bfbe2e69494c3dbd8790103d1fdcb3 : SUCCESS in 1h 56m 15s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/12ac436f3e4f4fc6b9004d43f50bd54a : SUCCESS in 2h 44m 57s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/fc78055833a04906aeff422a24c2b396 : SUCCESS in 1h 36m 28s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/451f26376c544cf9a35681e250c838de : TIMED_OUT in 3h 25m 25s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/6baf659bd135489184ba87eb6973fd42 : SUCCESS in 1h 36m 15s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/e4f44313db8c4ebab50a720b1d6e7692 : TIMED_OUT in 3h 31m 47s (non-voting)\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/e74ee5bf174e4a749c8a1ba53e6eb2aa : SUCCESS in 1h 28m 46s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/e04562a08361427586fb8aa6eb10fe6e : SUCCESS in 1h 31m 01s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/033857def19142578681993066d5a954 : SUCCESS in 1h 59m 13s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/77a771a64aa04177a45d30f9a27fb5c6 : FAILURE in 2h 24m 22s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/bbb8c335741f418c82c56c4dcf8ec5f7 : SUCCESS in 1h 12m 18s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/a9c467a69eb345a1a156cf3a42debc7d : SUCCESS in 1h 05m 01s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9cc1fd82c94a4be49c1fe278632776ee : SUCCESS in 4m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6c11e11f9c5d48689063fa6afaa8fc6f : SUCCESS in 6m 08s","accounts_in_message":[],"_revision_number":5},{"id":"b4b296f69cdc8c70205e4abd0c1667ae542098fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-10 19:19:29.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"3f8c11353652e8fe1ecc06c15e3ec163e08f7d0d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-11 00:47:32.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01f890753e72427d8aa77e629e605cdb : SUCCESS in 3m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2cb8c08b914441ae9ba8a6ec238b8c56 : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ba11fa5cd703454bb94fbc5042b5765b : SUCCESS in 3m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/98887b6e9d954ca594e2fc6e07e9a9c7 : SUCCESS in 10m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8edcb81154f54d0eb427f2b064f74d64 : SUCCESS in 3m 52s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/40019ef531f24374b9a93e2e71311078 : SUCCESS in 2h 17m 27s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/760326cee554494abbeb1192ed1401c4 : SUCCESS in 58m 49s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/928445ee39cd4514bf7a332e1da45c95 : SUCCESS in 2h 08m 38s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/e030522561924b0990af022a6e314e47 : SUCCESS in 1h 54m 05s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/6e4ca620681f4b028e87f3d397fe2f2d : FAILURE in 2h 23m 42s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/995749c2f2a849a4bf2d083c821f9dad : SUCCESS in 1h 46m 31s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/262bd841016845fcac6c32437aedf6c3 : SUCCESS in 1h 18m 31s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/6b508d86906a4a599e941167e29998aa : SUCCESS in 1h 25m 09s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/34bebb34c5454a7994da995f813ea056 : SUCCESS in 1h 33m 30s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/fb81f0889a59448a95249a996055821d : SUCCESS in 1h 09m 50s\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/09a25be94e9f4a8eb3dea5ad69c46dc6 : SUCCESS in 57m 53s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/97c30a314f774bb29c3e0bacce694032 : SUCCESS in 3m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/11ebc9c3c3344f6e845da8a00a9a9c34 : SUCCESS in 5m 40s","accounts_in_message":[],"_revision_number":5},{"id":"36e67551ae358b6e3bdda9ed15ef647c11d3a54a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-11 00:47:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"1de648c5617d813ef2d60b237e62eaf024ac70c7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-11 00:48:52.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5178603481c54a72b827198d7e2c0d89 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"59c5006e3a13350da8947b040365858afda82f93","revisions":{"49170fcc60f5800f2009fd0822a0024f9966d11f":{"kind":"REWORK","_number":1,"created":"2020-10-07 09:50:24.000000000","uploader":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"ref":"refs/changes/75/756475/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/75/756475/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/75/756475/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/75/756475/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/75/756475/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/75/756475/1"}}},"commit":{"parents":[{"commit":"6929f6f6966ac8e99f3213a9ae34a486c1d43457","subject":"Merge \"Fix handling of default_tag\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/6929f6f6966ac8e99f3213a9ae34a486c1d43457"}]}],"author":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-10-07 09:47:04.000000000","tz":120},"committer":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-10-07 09:50:22.000000000","tz":120},"subject":"Train-only: Add both validations path for inflight validation","message":"Train-only: Add both validations path for inflight validation\n\nThe ceph external deploy tasks is using validation modules via\nmistral.\nWe need to setup the correct new path for mistral in order to\nget the module. (warn.py)\n\nChange-Id: Ide21daaac9a558561d0943cc0a380952cf7e29b3\nResolves: rhbz#1885828\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/49170fcc60f5800f2009fd0822a0024f9966d11f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/49170fcc60f5800f2009fd0822a0024f9966d11f"}]},"branch":"refs/heads/stable/train"},"3f1c7c292a114508894ab53489de17d7008ba927":{"kind":"REWORK","_number":2,"created":"2020-10-07 10:02:45.000000000","uploader":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"ref":"refs/changes/75/756475/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/75/756475/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/75/756475/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/75/756475/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/75/756475/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/75/756475/2"}}},"commit":{"parents":[{"commit":"6929f6f6966ac8e99f3213a9ae34a486c1d43457","subject":"Merge \"Fix handling of default_tag\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/6929f6f6966ac8e99f3213a9ae34a486c1d43457"}]}],"author":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-10-07 09:47:04.000000000","tz":120},"committer":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-10-07 10:02:34.000000000","tz":120},"subject":"Train-only: Add both validations path for inflight validation","message":"Train-only: Add both validations path for inflight validation\n\nThe ceph external deploy tasks is using validation modules via\nmistral.\nWe need to setup the correct new path for mistral in order to\nget the module. (warn.py)\n\nChange-Id: Ide21daaac9a558561d0943cc0a380952cf7e29b3\nResolves: rhbz#1885828\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/3f1c7c292a114508894ab53489de17d7008ba927"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/3f1c7c292a114508894ab53489de17d7008ba927"}]},"branch":"refs/heads/stable/train"},"456ce9cc2d9ba13c0b8b6781ddd7c1d3c64f986d":{"kind":"REWORK","_number":3,"created":"2020-10-07 10:07:16.000000000","uploader":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"ref":"refs/changes/75/756475/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/75/756475/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/75/756475/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/75/756475/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/75/756475/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/75/756475/3"}}},"commit":{"parents":[{"commit":"6929f6f6966ac8e99f3213a9ae34a486c1d43457","subject":"Merge \"Fix handling of default_tag\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/6929f6f6966ac8e99f3213a9ae34a486c1d43457"}]}],"author":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-10-07 09:47:04.000000000","tz":120},"committer":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-10-07 10:07:05.000000000","tz":120},"subject":"Train-only: Add both validations path for inflight validation","message":"Train-only: Add both validations path for inflight validation\n\nThe ceph external deploy tasks is using validation modules via\nmistral.\nWe need to setup the correct new path for mistral in order to\nget the module. (warn.py)\n\nChange-Id: Ide21daaac9a558561d0943cc0a380952cf7e29b3\nResolves: rhbz#1885828\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/456ce9cc2d9ba13c0b8b6781ddd7c1d3c64f986d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/456ce9cc2d9ba13c0b8b6781ddd7c1d3c64f986d"}]},"branch":"refs/heads/stable/train"},"fdf97deabcb95178ebee0d1bc50ba00eae221442":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-10-07 10:20:35.000000000","uploader":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"ref":"refs/changes/75/756475/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/75/756475/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/75/756475/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/75/756475/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/75/756475/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/75/756475/4"}}},"commit":{"parents":[{"commit":"6929f6f6966ac8e99f3213a9ae34a486c1d43457","subject":"Merge \"Fix handling of default_tag\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/6929f6f6966ac8e99f3213a9ae34a486c1d43457"}]}],"author":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-10-07 09:47:04.000000000","tz":120},"committer":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-10-07 10:20:19.000000000","tz":120},"subject":"Add both validations path for inflight validation","message":"Add both validations path for inflight validation\n\nThe ceph external deploy tasks is using validation modules via\nmistral.\nWe need to setup the correct new path for mistral in order to\nget the module. (warn.py)\n\nChange-Id: Ide21daaac9a558561d0943cc0a380952cf7e29b3\nResolves: rhbz#1885828\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/fdf97deabcb95178ebee0d1bc50ba00eae221442"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/fdf97deabcb95178ebee0d1bc50ba00eae221442"}]},"branch":"refs/heads/stable/train"},"59c5006e3a13350da8947b040365858afda82f93":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2020-10-09 07:01:31.000000000","uploader":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"ref":"refs/changes/75/756475/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/75/756475/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/75/756475/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/75/756475/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/75/756475/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/75/756475/5"}}},"commit":{"parents":[{"commit":"6929f6f6966ac8e99f3213a9ae34a486c1d43457","subject":"Merge \"Fix handling of default_tag\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/6929f6f6966ac8e99f3213a9ae34a486c1d43457"}]}],"author":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-10-07 09:47:04.000000000","tz":120},"committer":{"name":"mathieu bultel","email":"mat.bultel@gmail.com","date":"2020-10-09 07:01:26.000000000","tz":0},"subject":"Add both validations path for inflight validation","message":"Add both validations path for inflight validation\n\nThe ceph external deploy tasks is using validation modules via\nmistral.\nWe need to setup the correct new path for mistral in order to\nget the module. (warn.py)\n\nChange-Id: Ide21daaac9a558561d0943cc0a380952cf7e29b3\nResolves: rhbz#1885828\n(cherry picked from commit d2b373dc1d7a0cc58aa595d77c9f31e4f8194e66)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/59c5006e3a13350da8947b040365858afda82f93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/59c5006e3a13350da8947b040365858afda82f93"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
