)]}'
{"id":"openstack%2Ftripleo-common~756514","triplet_id":"openstack%2Ftripleo-common~stable%2Ftrain~I4a3149b08013f493e13b592f064e3ff2ed4074f7","project":"openstack/tripleo-common","branch":"stable/train","topic":"train/bearer","hashtags":[],"change_id":"I4a3149b08013f493e13b592f064e3ff2ed4074f7","subject":"Improve bearer auth handling","status":"MERGED","created":"2020-10-07 13:35:00.000000000","updated":"2020-10-12 06:12:56.000000000","submitted":"2020-10-12 06:11:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"756514-1602483097653-2e8b7072","meta_rev_id":"ac72cab9f373be5e32f56c1e56eabeda4fe7ea2c","_number":756514,"virtual_id_number":756514,"owner":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-10-11 23:57:29.000000000","_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-12 06:11:37.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-10-07 23:05:31.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":2,"date":"2020-10-09 13:00:25.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":1,"date":"2020-10-09 13:00:25.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-07 23:05:31.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-09 13:00:25.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-10-11 23:57:29.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2020-10-12 06:11:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"39cd62c4a6d46bfd7fa1df404678d609ff16ffc7","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-10-07 13:35:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8a550533914db759de13fb55d1c89271e77d12d8","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-10-07 14:13:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/9a446010ed844b34a1a6d9749828c5d6 : FAILURE in 22m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2f85d350b9f6d5f6a0097b27d7ae0e2198a5b067","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 18:16:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0055eacab57744169fbf2fede64b7c86 : FAILURE in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b59203d4e4ac43bfb67793f1d9baa15a : SUCCESS in 3m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7dfe968bbd314601840584bf6a6c4f4b : FAILURE in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/02317e0eaf7942a4a815af1d9f595b41 : FAILURE in 4m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/231d379c3a50429dadb99869584c3406 : FAILURE in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e5f715e982247cc85b6415bc0384b60 : SUCCESS in 5m 35s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/bc9683514b0b4994929254f06e312c49 : FAILURE in 24m 08s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/ff572c9d7744482a84fc8c72d617288c : FAILURE in 1h 40m 12s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/c9195bf0ee054abeae6765464ca51531 : FAILURE in 1h 44m 12s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/ded150c487c342d1a3abccc5fea3ca38 : SUCCESS in 1h 02m 48s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/975e3339b5874f0ba24d4831078aa819 : SUCCESS in 4m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4b8520609e924c0bbf81db0b7d8bec11 : FAILURE in 7m 00s","accounts_in_message":[],"_revision_number":1},{"id":"fd350aaa94cb2b97c5328aa412a92a9e7c779223","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-10-07 19:40:02.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"70057d3410dde4e54adc02a1d25753118d5170df","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-10-07 19:45:57.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ee239159d470121a3fae7c99eb4f30e55a881dfd","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-10-07 23:05:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/6ea8d3628faf419fae8f9427fade1a80 : SUCCESS in 3h 18m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"12672fb06412c190d769549055de23057a77331c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 00:20:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/51b181eea0724765be49eb62bd39682a : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/331a014c0aa14061b4d850c4eb85dc11 : SUCCESS in 3m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b5e58b3df0dd4debb1071ea081dfe664 : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d46bf0b4bdac4c6dab70aaef3213a180 : SUCCESS in 4m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/089508f11edc4a9fbee71f28d0751c3e : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7789eeddf3204fcfa845bc19675df903 : SUCCESS in 5m 26s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/230e7b401453484e84164ee736662868 : SUCCESS in 2h 17m 28s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/5528ece0042e4b648cc0a182029f4fac : SUCCESS in 1h 54m 11s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/8345ea6fd15d46a3a06176964d863f9e : SUCCESS in 2h 30m 39s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/1fb15a9164b84c61a9cfe8c6e7583c0e : SUCCESS in 1h 50m 26s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/366cfcf35f0d459b985a146c852474c8 : FAILURE in 2h 40m 32s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/f158396e9db44215a992f4f80fd5dbb7 : SUCCESS in 1h 43m 02s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/5e2673d990774d3c829a7d64134388a9 : FAILURE in 2h 35m 47s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/7abed439cbf047869207846625f21a6a : SUCCESS in 1h 15m 08s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/445e4bcb168d43e6aa8176cbc23955ee : SUCCESS in 3m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7bba1b10a14c4d188e77edf748255fd6 : SUCCESS in 5m 16s","accounts_in_message":[],"_revision_number":3},{"id":"decd28cf7dc3ab6ba2bc1ba6416fd6e76bbbb039","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-10-09 13:00:25.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"135b1b4f3091f136b636fe8b6fbb1be05b3a1dfb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 13:00:49.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1f273d900499b4afaefbf173714b384a0120da37","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 22:35:32.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/826aaa7bf22e4414b81bd87a6e98fa09 : SUCCESS in 4m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/74d800114d1043778c519e2c6a523503 : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d5b0ac8d166f4829a8bae7c0d2a174b1 : SUCCESS in 5m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f6b5155101f942608983cdac4b867f09 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a401d5acaede4ff2a74386a59ebf28e9 : SUCCESS in 5m 58s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/edf47806564e47c69fc30ddfcd554109 : FAILURE in 3h 24m 16s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/4b392972d6054022bf948641613e10d8 : SUCCESS in 2h 09m 38s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/5a1fafce5b6b46c1955f8e3eba57caf8 : SUCCESS in 2h 42m 22s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/6763ca9eee784d9bb6ff4fc4a9fc4e81 : TIMED_OUT in 3h 16m 24s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/537f1ff3938a462eb41ecf70139f251f : TIMED_OUT in 3h 16m 05s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/73f8a2f9d3b444bda63428c43b5c7eee : SUCCESS in 1h 44m 02s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/9b4648febf0a44a189d780aeacb09b48 : SUCCESS in 1h 18m 26s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/e34634baf1d94107852bc8a22b707fa6 : SUCCESS in 37m 09s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6936df2b5b9a4bc7a698da43fd9a0a84 : SUCCESS in 3m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/744d3a4e09c7406090737a638f931b15 : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":3},{"id":"0560b3112422b39b017412d0e8cc05fd9e946b15","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-10-10 15:31:17.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"f8dbdcdfec386a06539e88797cf4b3d2ba126775","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-10-10 18:54:19.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/5173b92d2d484d3e9e08dbaa22868145 : SUCCESS in 3h 20m 29s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5eb5f1f120027e7df16a6e8269789e412a95ced7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-10 19:02:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3170c4fa22e9431596c4015cd1834ee2 : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec9866e3513c4f318875726ffda88bdd : SUCCESS in 4m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e5568996c45b465fbd4a93f6e28ed374 : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/77a4bafa61994f2f9412409c6ef19c3c : SUCCESS in 4m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4bc8eb1bbae14130bd696850bff59284 : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c924c9938446456bbadb662a223844fd : SUCCESS in 5m 09s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/8855651d106848429c295893b1492ba2 : SUCCESS in 2h 12m 20s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/3a35a62e3f004aacbb5a05257dc83fb0 : SUCCESS in 3h 21m 03s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/f64fba5cf11749f99e896f141a77980b : SUCCESS in 2h 15m 38s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/c67dd5ca884c4339bbbd77bffeb903f0 : SUCCESS in 1h 58m 52s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/93c39f04c26f4e65892f8fb55aeb6ebb : FAILURE in 2h 44m 02s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/5a52941f3c274c589f1bcf6751f5ee3a : SUCCESS in 1h 52m 20s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/ba44371e154e4782819b283a14e6de1a : FAILURE in 2h 27m 57s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/b7bc39e2873f4b78b0e8d4736a29932a : SUCCESS in 1h 03m 17s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6ffd0b85933e4068bd37b38e390c4438 : SUCCESS in 4m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5872248b7d2d48c28adad9b4c18a16fc : SUCCESS in 5m 36s","accounts_in_message":[],"_revision_number":3},{"id":"4049c18d8c10fb1f5a0da352ce350a66c2018934","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-10 19:03:10.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ee9ad32ef1bda109274708f73af40a08d0db29e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-10 22:17:37.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edebb1fcd5eb4d28a7a67067c511bff5 : SUCCESS in 4m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/34a43c3e737f4c479a0db272f1e904ab : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7402fcc11d924717b3b255dc92b59388 : SUCCESS in 3m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e0546f1e48734e35b50c2815a2b7b53c : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3f27093a8c44f4bad58b1e9cafad393 : SUCCESS in 4m 26s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/8a8c1fa21d2b4da6bf82d519a45f89fb : SUCCESS in 2h 22m 01s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/dffd7f5985a34ee18ecd49b646177d5f : SUCCESS in 1h 25m 31s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/b51a98c3acef4cdcb590f20676e5f670 : SUCCESS in 1h 55m 15s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/0d45b0e9195949069dcda48bbe5b7bda : TIMED_OUT in 3h 12m 46s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/44e006fdb5de4230bf8902af963ec845 : FAILURE in 2h 47m 55s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/e77463caa6854f79a44f2e0047da90ca : SUCCESS in 1h 42m 47s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/e270c6543d3140c286d8bf7e72f8f98c : SUCCESS in 1h 02m 33s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/f878f59a377b41f383fa5d6e38f809db : SUCCESS in 41m 15s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0b1640755a6342d7a22e9c35564998a5 : SUCCESS in 3m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b0a744dbea9846619cd7acdc7164df62 : SUCCESS in 5m 53s","accounts_in_message":[],"_revision_number":3},{"id":"20ee800105597cf5d39499cf78a2083287bc8c98","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-10-11 23:57:29.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"ca58d1b2e713dae954238a894da96ffcf93bfd1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 02:49:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/90fd17006a004a83b56ff2a1bc11059a : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f22a3ba5bf1c4af29419a71097d875fc : SUCCESS in 3m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7b2549a75e58489baee808a13db96e12 : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f850e64152548809b30ce13b18af276 : SUCCESS in 3m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/10dadcf7e2a94d4c8b84b98efb80e89b : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb5eff2591d24b238095493f6094cedd : SUCCESS in 3m 56s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/7473d9974fc14c48a66868a08fceb565 : SUCCESS in 2h 15m 39s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/a2e5a9a92cb048b0990323d75984856b : SUCCESS in 2h 00m 26s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/950b2949d3214cf5ae3e661ca47c54e9 : SUCCESS in 2h 28m 57s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/cf2c7d85ef8e4b50b2021240851d06b2 : SUCCESS in 1h 15m 46s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/3bbeaa300b454a03927c98d7ee36c5e4 : FAILURE in 2h 46m 18s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/184e06ed50fc4f27b9625be820ece62b : SUCCESS in 1h 44m 53s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/a49191c0a32e4f2096d5af1e8c483812 : FAILURE in 2h 46m 15s (non-voting)\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/0ead4aca961e4acaa11557fd43014f89 : SUCCESS in 1h 10m 58s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d0586046e05d487bbc27aed6fa2d042b : SUCCESS in 3m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bb9647a426a4441bbbdc4ebbc5c44c49 : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":3},{"id":"cf0305c0f160593e652af4c783a92f19a7d9fa06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 02:49:34.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ceaac8269cce71911281f687bdbe9f03a1bc8aad","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-10-12 02:58:56.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/639232e2fb324e5b8b116e8e1ec9aeb1 : SUCCESS in 2h 59m 58s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5219d2e4a7e11441c90e6789a715ee08d9816200","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 06:11:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"659d017cd0a2c6eabdeebaf1b4bb65fd786cba89","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 06:11:37.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34e833bc41934fc59f35fe54d97fde21 : SUCCESS in 3m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8526d35e03704f6cb7908277a8f4308c : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be2b347f80104b4e8dc20142b2691073 : SUCCESS in 3m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5ac0629b4fd0419f886e4c08da9f5192 : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/005633f729fd407cb8840768c3de2a50 : SUCCESS in 3m 50s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/025af0af0e33498c88c7aff05a81f50b : SUCCESS in 2h 19m 32s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/ca8e9c5df30943aeafad96fcb7a51693 : SUCCESS in 1h 36m 44s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/cc6d851de64940eaa656f8c68c6a21e5 : SUCCESS in 2h 08m 44s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/d53eda4f08054261b55362879fcf5325 : SUCCESS in 2h 03m 00s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/6db9afc5410b470995132917a8eee621 : TIMED_OUT in 3h 16m 32s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/96f340549d3448bb946e34ff9973e343 : SUCCESS in 1h 47m 38s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/d6c4998863c34e1080ca4a60271d6037 : SUCCESS in 1h 08m 07s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/6412bc5782d64d4ca5fc70ee23d4f6dc : SUCCESS in 42m 43s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/545b35ad47ee406a831f1ccbc60b687f : SUCCESS in 3m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cc122263d6f5459784f0d1972908a061 : SUCCESS in 4m 53s","accounts_in_message":[],"_revision_number":3},{"id":"7364f2ad4934f84f6e67bca4926c9993487cc71d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 06:12:56.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07dd3a0183504f2aa1aa1a960fe52d97 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1070bd4c63b5071872e5ec13bc2b86e77660509d","revisions":{"8a915baaf9c4934a3afe47bd540151cf1205b7c7":{"kind":"REWORK","_number":1,"created":"2020-10-07 13:35:00.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/14/756514/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/14/756514/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/14/756514/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/14/756514/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/14/756514/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/14/756514/1"}}},"commit":{"parents":[{"commit":"6929f6f6966ac8e99f3213a9ae34a486c1d43457","subject":"Merge \"Fix handling of default_tag\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/6929f6f6966ac8e99f3213a9ae34a486c1d43457"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2020-09-03 19:29:31.000000000","tz":-360},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-10-07 13:34:28.000000000","tz":-240},"subject":"Improve bearer auth handling","message":"Improve bearer auth handling\n\nWe can share tokens across threads for scope so if we are fetching\nmultiple layers of the same container, let\u0027s reuse the token rather than\nduplicating the token request. Additionally we can verify if a token\nneeds to be refreshed based on the expiration time.\n\nNote: the cherry-pick isn\u0027t clean as it replaces the usage of bytes() by\nsix.b() for python2 compatibility in Train.\n\nChange-Id: I4a3149b08013f493e13b592f064e3ff2ed4074f7\n(cherry picked from commit f52b1e1a46ef1c18593106d09e1eb5fff7d13917)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/8a915baaf9c4934a3afe47bd540151cf1205b7c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/8a915baaf9c4934a3afe47bd540151cf1205b7c7"}]},"branch":"refs/heads/stable/train"},"8f12852c84adf81cf9bb4f2ebf34b7e234ead1e0":{"kind":"REWORK","_number":2,"created":"2020-10-07 19:40:02.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/14/756514/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/14/756514/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/14/756514/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/14/756514/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/14/756514/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/14/756514/2"}}},"commit":{"parents":[{"commit":"6929f6f6966ac8e99f3213a9ae34a486c1d43457","subject":"Merge \"Fix handling of default_tag\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/6929f6f6966ac8e99f3213a9ae34a486c1d43457"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2020-09-03 19:29:31.000000000","tz":-360},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-10-07 19:40:01.000000000","tz":0},"subject":"Improve bearer auth handling","message":"Improve bearer auth handling\n\nWe can share tokens across threads for scope so if we are fetching\nmultiple layers of the same container, let\u0027s reuse the token rather than\nduplicating the token request. Additionally we can verify if a token\nneeds to be refreshed based on the expiration time.\n\nNote: the cherry-pick isn\u0027t clean as it replaces the usage of bytes() by\nsix.b() for python2 compatibility in Train.\n\nChange-Id: I4a3149b08013f493e13b592f064e3ff2ed4074f7\n(cherry picked from commit f52b1e1a46ef1c18593106d09e1eb5fff7d13917)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/8f12852c84adf81cf9bb4f2ebf34b7e234ead1e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/8f12852c84adf81cf9bb4f2ebf34b7e234ead1e0"}]},"branch":"refs/heads/stable/train"},"1070bd4c63b5071872e5ec13bc2b86e77660509d":{"kind":"REWORK","_number":3,"created":"2020-10-07 19:45:57.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/14/756514/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/14/756514/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/14/756514/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/14/756514/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/14/756514/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/14/756514/3"}}},"commit":{"parents":[{"commit":"6929f6f6966ac8e99f3213a9ae34a486c1d43457","subject":"Merge \"Fix handling of default_tag\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/6929f6f6966ac8e99f3213a9ae34a486c1d43457"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2020-09-03 19:29:31.000000000","tz":-360},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-10-07 19:45:27.000000000","tz":-240},"subject":"Improve bearer auth handling","message":"Improve bearer auth handling\n\nWe can share tokens across threads for scope so if we are fetching\nmultiple layers of the same container, let\u0027s reuse the token rather than\nduplicating the token request. Additionally we can verify if a token\nneeds to be refreshed based on the expiration time.\n\nNote: python2 code was added to build the token variable.\n\nChange-Id: I4a3149b08013f493e13b592f064e3ff2ed4074f7\n(cherry picked from commit f52b1e1a46ef1c18593106d09e1eb5fff7d13917)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/1070bd4c63b5071872e5ec13bc2b86e77660509d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/1070bd4c63b5071872e5ec13bc2b86e77660509d"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
