)]}'
{"id":"openstack%2Ftripleo-common~759683","triplet_id":"openstack%2Ftripleo-common~stable%2Fvictoria~Id1a2d3e5bedb0f15d0f87fbf0538d7fed0b08a87","project":"openstack/tripleo-common","branch":"stable/victoria","topic":"create-victoria","hashtags":[],"change_id":"Id1a2d3e5bedb0f15d0f87fbf0538d7fed0b08a87","subject":"Update .gitreview for stable/victoria","status":"MERGED","created":"2020-10-26 13:05:24.000000000","updated":"2020-12-11 12:09:43.000000000","submitted":"2020-12-11 12:08:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"759683-create-victoria","meta_rev_id":"7b7d527d9203a2ad855055eb4844e8452a2b2331","_number":759683,"virtual_id_number":759683,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-11 12:08:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2020-12-10 21:25:36.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"all":[{"value":2,"date":"2020-10-29 13:42:52.000000000","_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":2,"date":"2020-10-29 14:01:49.000000000","_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":2,"date":"2020-10-26 14:09:14.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"all":[{"value":1,"date":"2020-10-30 11:51:24.000000000","_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-26 14:09:14.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-10-26 14:51:07.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-29 14:01:49.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2020-10-30 11:51:24.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2020-11-03 18:42:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bd6fd447c24fe5294bdefc87c6963b2c25a7f5e5","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-10-26 13:05:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0f0f6480ec1846af063a40d4c13094d08ae63f69","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-10-26 14:09:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6cccad3479ad1249e27a62b3438f3607559d11dc","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-10-26 14:51:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/a6c24ca144974cf49c5a5a56187efb78 : FAILURE in 1h 44m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d5f1cc1a3a7b40f3d27d434d21a684a9a96b33ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-26 16:40:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/339f97bc0a614d6b9460d47bc8b630ec : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69ff279b0e2544a2b1d3201f5a12b253 : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b635db48e3b4adb94d21e17d330e378 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/59402f38d6f24e96bb39e8e321c035ef : SUCCESS in 3m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b5db322127a496ea43ac038b2f2c681 : SUCCESS in 3m 59s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/64db0245fd6a4d49b9d19098462118d0 : SUCCESS in 2h 59m 27s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/75124755ef4c408eaf9e2cebeb519c2c : SUCCESS in 52m 07s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/b20fb653124443f785ba8352f5ed0150 : SUCCESS in 49m 59s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/c7bd84579c974215830d7e0464fec383 : FAILURE in 1h 48m 12s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/c281dd50360d4815be2c72891308a971 : FAILURE in 47m 37s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/158197f624174862b6739b651b0e46d1 : SUCCESS in 1h 22m 48s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/088c192e43b844628999edc660e071e3 : SUCCESS in 3m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/43f2a499e9114d6dadaf27343acef8ef : SUCCESS in 3m 45s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/b6824c28df38463e94d5040f39708d01 : SUCCESS in 4m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/08c9ca08af084b7886171ac41c1cddec : SUCCESS in 6m 00s","accounts_in_message":[],"_revision_number":1},{"id":"9c2be45d44615b1a98db03f33cefd138511fefc8","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2020-10-29 13:42:52.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2b8e0c93cfb3fd2b6e9ab795dfba71890bff875d","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-10-29 14:01:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7aae0b1b8e588ac8bd76b811344bffee729b3c2e","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2020-10-29 15:54:01.000000000","message":"Patch Set 1:\n\n17:53 \u003c marios\u003e weshay|ruck: so that should fix the container-multinode issue i think \n                https://opendev.org/openstack/puppet-tripleo/commit/cfaea88a5b091e62842de8ef926f80731177faf1 (i \n                mean the issue we were seeing at https://review.opendev.org/#/c/759683/ \n17:53 \u003c marios\u003e hopefully recheck will show that","accounts_in_message":[],"_revision_number":1},{"id":"bb147d30e9d7ef677ad8b505badff0f63e350082","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2020-10-29 15:59:49.000000000","message":"Patch Set 1:\n\n\u003e 17:53 \u003c marios\u003e weshay|ruck: so that should fix the\n \u003e container-multinode issue i think\n \u003e https://opendev.org/openstack/puppet-tripleo/commit/cfaea88a5b091e62842de8ef926f80731177faf1\n \u003e (i\n \u003e mean the issue we were seeing at https://review.opendev.org/#/c/759683/\n \u003e 17:53 \u003c marios\u003e hopefully recheck will show that\n\nnope as wes just reminded me we need promotion of master to get that puppet-tripleo content","accounts_in_message":[],"_revision_number":1},{"id":"7e1a77ddc5ef48df441a1161a2fe0c6075b0dcbc","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-10-29 16:40:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/3e97e3632ebf4e9592d972785814ca3b : SUCCESS in 2h 56m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"433aefc257cbae4851190ed028482777af488fa0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 01:23:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6c9ea8c9b8da4be686af7ab1cd4c9140 : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a10804f070ce4c64ba8451fd84b12704 : SUCCESS in 3m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c0494c60e5b640ff90fc33ee428e65e5 : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8052a8db6b0d4aee944b38346cca6638 : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7dfc6eeefeee479a93f4122b8b3a8882 : SUCCESS in 4m 44s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/0ddfb88d2a2145adb6b97135b2a755d3 : SUCCESS in 3h 39m 21s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/a858947c8b56447ca535361feebc0593 : SUCCESS in 59m 01s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/6fba81300c4c4348b2b4c43c3857e2a2 : SUCCESS in 40m 43s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/0ff11be42c0c41db8ea25de2222ad0d0 : SUCCESS in 2h 07m 37s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/b52f12a8386845d7a483bfe60733686f : SUCCESS in 1h 38m 10s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/69722811f0fa4d5ba9fb13a45f69a75c : SUCCESS in 1h 33m 47s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b9d446e82b9b41eab66290d6842b27aa : SUCCESS in 3m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ee62b75bd60f4ce998ff9d94056405b5 : SUCCESS in 5m 29s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/269844d00bb14e38aa3335580fb40190 : SUCCESS in 4m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/80ac922fc8554c4982a58d3d3d3a63b7 : SUCCESS in 7m 15s","accounts_in_message":[],"_revision_number":1},{"id":"f80ebff88e8e0a646709a4f6c43137653eb2e997","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2020-10-30 11:51:20.000000000","message":"Patch Set 1:\n\n\u003e \u003e 17:53 \u003c marios\u003e weshay|ruck: so that should fix the\n \u003e \u003e container-multinode issue i think\n \u003e \u003e https://opendev.org/openstack/puppet-tripleo/commit/cfaea88a5b091e62842de8ef926f80731177faf1\n \u003e \u003e (i\n \u003e \u003e mean the issue we were seeing at https://review.opendev.org/#/c/759683/\n \u003e \u003e 17:53 \u003c marios\u003e hopefully recheck will show that\n \u003e \n \u003e nope as wes just reminded me we need promotion of master to get\n \u003e that puppet-tripleo content\n\nso not via current-tripleo (promotion) as it isn\u0027t external content - so we got a green run","accounts_in_message":[],"_revision_number":1},{"id":"fd3bd062f45bcaba75f0d5d5370440484353a0bf","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2020-10-30 11:51:24.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"43088b74bf35e456c09e821dbd5f9ba776b90e6f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 11:51:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cac8748712095a05ad51a1f983d9aefb53f84aa7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 17:17:01.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1496950e30064189955fae1876c6eec1 : FAILURE in 3m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/304194139d854b85b19f663c1617759e : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9422daf4d1dc4bcfa913acd9fd9ebf8c : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05ad384675fb4ecc9d286f83a25bebcf : SUCCESS in 4m 37s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/2b5b20f30359434295f6d35c8fcb306b : SUCCESS in 3h 37m 22s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/a9c341a5fe1c43aeb34f59b93d3d3fdf : SUCCESS in 1h 09m 17s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/f3c606591fe24dab9261e053d2c3ed9a : SUCCESS in 59m 46s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/f26bca22f8f643a3a36c238569ff67c1 : SUCCESS in 2h 06m 31s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/2cf02a36367b4dd6a3501d9338de719b : SUCCESS in 1h 36m 26s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/77dc4b52ab0e439b9e4d5dde178cda21 : SUCCESS in 1h 29m 02s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5e1643ee9beb430f971c1003988eadbd : SUCCESS in 3m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0134c507d1d04902a7743822a3bafa9d : SUCCESS in 6m 58s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/3caa00db4b874087ac4b03a3fa931d9d : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":1},{"id":"3ee180e0e2c8195645ef31dff45aeb313d2d6b18","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2020-11-03 15:33:44.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c2d3ae9594237f8cf8ed89d9b00d9feca3552b25","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-11-03 18:23:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/ec4cc002dbff4bd7bfeb115ad8c10ee6 : SUCCESS in 2h 48m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"162a5faf6abc76d39ec9d4598cba4f9f0472718a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 18:42:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bad7f8e472b94664a9c4e7c1589a4fc5 : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7226192bdab142acbee2e0fc6ade83ca : SUCCESS in 3m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/35c54b9779694aaca4e4872b5f220d02 : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f14fcaa6bb847fdac392fe6ac07feeb : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b7607908e1846af8ef2c03a62947c3b : SUCCESS in 4m 46s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/dc09a095936043a7806e42bd5e6139f4 : SUCCESS in 2h 46m 14s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/2d677786e9654bdb9c355ffc7bde66a8 : SUCCESS in 48m 26s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/ab56555f8e534ed09405feb4f7005b4c : SUCCESS in 43m 41s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/44103b739de54acd968ea3a14e7b47e2 : FAILURE in 1h 42m 23s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/dc21f372917f4e7a8bbdeb893ffc442e : SUCCESS in 1h 29m 13s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/d0d333d312504ae594d72aa07b7cd474 : SUCCESS in 1h 13m 46s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/dd55fc2c030944e29ac574c4fbf16a16 : SUCCESS in 4m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fe13d03d160c4a2da5560fa5aae33bec : SUCCESS in 5m 05s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/05d93c505ff644ab8164f2156d44a19f : SUCCESS in 4m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7cccc14b37454c189f1fa5cbb13a761f : SUCCESS in 7m 11s","accounts_in_message":[],"_revision_number":1},{"id":"7e51f41002f2fc71ed7f57d27245aec0b8bf197f","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-11-24 20:05:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"62d87c7fb2c91379322b4d703d988815b160e2d5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-11-25 03:04:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/094e364b33bd4de5aac06d362f1f23f0 : SUCCESS in 2h 57m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"63b3e262a082932d8f270834344ae9952f5021f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 03:45:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d79a2bab4b34424b038df90704e2a9e : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/066561008c2541b5a474f270dd1f76c8 : SUCCESS in 3m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c73c6023c2cd46ebaae57a812f73d4c6 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b59ee2e5fd9848c6ae1ec36e19868e0a : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86cb0f0645cc489284e8317a9e5ff480 : SUCCESS in 3m 40s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/4ed45ce90ba14dabbc82715c3499059c : SUCCESS in 3h 31m 34s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/d9432cd9d3954e9f8543a0946446ba24 : SUCCESS in 4h 15m 22s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/8a2649306e2f42c1a523282921b8f5bd : SUCCESS in 4h 13m 39s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/c2324de2915847bdb65988db630407c1 : SUCCESS in 44m 53s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/2e1a1b53d0214e4bad45ba525df393f5 : FAILURE in 1h 06m 52s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-victoria https://zuul.opendev.org/t/openstack/build/2c0f7c1ac01440fcae46fc96f22ab9a9 : SUCCESS in 2h 02m 27s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades-victoria https://zuul.opendev.org/t/openstack/build/93254eb17ec340179c3162a1b599a656 : SUCCESS in 2h 03m 48s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/3a4fa4e829194015a11de201bc87f9c5 : FAILURE in 1h 04m 20s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/588c7b5343a8484991b64f741cbe9278 : FAILURE in 35m 39s\n- tripleo-ci-centos-8-undercloud-upgrade-victoria https://zuul.opendev.org/t/openstack/build/52676171df964414854b1d058fb88578 : SUCCESS in 2h 03m 16s\n- tripleo-ci-centos-8-standalone-upgrade-victoria https://zuul.opendev.org/t/openstack/build/246a9fae529e4ce7b62c75595e83d282 : SUCCESS in 2h 29m 35s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5fe8484edde542b2a53752af74ef8988 : SUCCESS in 3m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a3af345079844574ae8746c492a5ebaa : SUCCESS in 4m 01s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/bd796e2b682345b4bc6148e922556605 : SUCCESS in 5m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cd05d01017f340ee9337cddbe456bf7f : SUCCESS in 6m 52s","accounts_in_message":[],"_revision_number":1},{"id":"df191b99f1d4dabadfb70d90c96ba6546d28572b","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2020-11-25 07:40:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"afbcf7a86376f9c1587ac1353d9726093293dead","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-11-25 12:57:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/281ef11c7388465fb997a6c3102d80f3 : SUCCESS in 2h 36m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"28013a5f40c251ba5884de8b57c17a17f7d43d3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 16:31:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9807ca0cc6b34ea5ad8657ce89a23621 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f904ad723ed4039918a046f326b9aa9 : SUCCESS in 3m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/24aaa52a71f340ec861e9340a8728f60 : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8fbd3c63c48c4c8dbc6fc9126ec07c24 : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e21ca6f5f7b54147ba4f733144b3c524 : SUCCESS in 4m 01s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/b2fd85ccddba47ee925432dbfacae35a : SUCCESS in 4h 00m 51s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/dd2442e7451c4c86889497aeaf8289dc : SUCCESS in 4h 35m 50s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/f8a8e12eab4c496193fda373e0723ec5 : SUCCESS in 4h 36m 15s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/f8133e91ec9d4a5ca78506fa436769b2 : SUCCESS in 1h 00m 54s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/07b1c600af1747b19ea7ab6533d3530b : SUCCESS in 2h 33m 46s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-victoria https://zuul.opendev.org/t/openstack/build/a7fc41e4eb924f848a8794e0062805c0 : SUCCESS in 2h 00m 15s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades-victoria https://zuul.opendev.org/t/openstack/build/9b3c58211f0642cabc4744404181590f : SUCCESS in 2h 05m 13s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/b11d1ed18f7541b9a01db55d6745c29c : SUCCESS in 1h 22m 28s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/16c335d2977343559a773389650687b9 : SUCCESS in 1h 07m 53s\n- tripleo-ci-centos-8-undercloud-upgrade-victoria https://zuul.opendev.org/t/openstack/build/98e76217c8e14cddb12230f4a90dcec6 : SUCCESS in 1h 53m 05s\n- tripleo-ci-centos-8-standalone-upgrade-victoria https://zuul.opendev.org/t/openstack/build/032ef8b7a8df418b8a6e648766c42e21 : SUCCESS in 2h 55m 50s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/50d332e55c564a8d95d5289d2d337fe8 : SUCCESS in 3m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/44ed2843ea5a4d4c80c0698c7952ff92 : SUCCESS in 4m 45s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/35b0fe6a5d7e4210a3b769c6e4bc5706 : SUCCESS in 4m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/36550349843740a2a37c98fab539640f : SUCCESS in 6m 35s","accounts_in_message":[],"_revision_number":1},{"id":"8a0836f0a1440ed1da1cd35a083bc0f8558ed1d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 17:47:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"358fa73cf175b2becfc1dc7469474b48fd556c81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-26 15:05:56.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab5db464c2f3406b86446a8034332026 : SUCCESS in 2m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc8657d740214c3aa9806954e2ba60e5 : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f05beecfa2624c2d93d17344b0d8a29e : SUCCESS in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57487846e0aa49ad85e5a554a2bd6f5f : SUCCESS in 4m 00s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/0bde182df5a24ff1ba71ec58337f7119 : SUCCESS in 3h 28m 03s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/09ef2d3c6bda4b42b1062d00838df8f3 : SUCCESS in 4h 07m 17s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/fe359070d7094010aa9cd128810b502e : SUCCESS in 3h 02m 40s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/54ce6905e2014deab7fabeaed73e0a7e : SUCCESS in 48m 30s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/18ad76e6937d40d5ace12c44a9150a3a : SUCCESS in 2h 05m 33s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/003a8ad30d0d40b781766a90458bd86a : SUCCESS in 1h 47m 26s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-victoria https://zuul.opendev.org/t/openstack/build/46abc88216d44738a3e4b4f13566dd10 : SUCCESS in 2h 06m 53s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades-victoria https://zuul.opendev.org/t/openstack/build/fd2ab4a56fa247e6a98801f9d34141ae : SUCCESS in 2h 16m 42s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/038a42ab81264f51affba2e2292c6629 : SUCCESS in 1h 19m 46s\n- tripleo-ci-centos-8-undercloud-upgrade-victoria https://zuul.opendev.org/t/openstack/build/4f558807a9b747e19f8f6d0404f7cb2d : FAILURE in 58m 47s\n- tripleo-ci-centos-8-standalone-upgrade-victoria https://zuul.opendev.org/t/openstack/build/0bf8230c6ead4f5a916cf627ee4cdb35 : FAILURE in 1h 13m 53s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/675b75e02a7741849a43fcf9c659d32e : SUCCESS in 4m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/14545e6fdf7b4171915949d958213cd5 : SUCCESS in 5m 49s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/a310d75a9df345d28af06a9c02757684 : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":1},{"id":"75512e7f669321686d179a81d2e9ef8b88b7989f","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-11-30 14:35:19.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7e008a547ece820428f48245ff4e0176e5d24c68","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-11-30 14:53:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/fa421ba2b0af46b1994e8c4fea29658e : FAILURE in 14m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9be3281857f580d38588e5d395221d8632a10719","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-30 15:25:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d28f98608604cbbaee8020c8ca6dd87 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1470d082395e4b45b9b8df389c470ed8 : SUCCESS in 3m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/48114455dc704ecca1df5422698dd7a6 : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/79c9cdb101a24b84b35832a9078573b4 : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a91c0d3e7de4c3aa04ce13024928847 : SUCCESS in 4m 56s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/e8e1230a1b0d45d78f3c6e9575ed4fe2 : FAILURE in 15m 22s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/9119ad9213444594bd04e0d97aa3764e : FAILURE in 9m 25s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/286803e2d8f94149bf2fdb706ef27555 : FAILURE in 14m 39s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/81359591cf5348068704848ea710c4a9 : FAILURE in 13m 24s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-victoria https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades-victoria https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-undercloud-upgrade-victoria https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone-upgrade-victoria https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8de466008d4c4ca687cc5ce914968e81 : SUCCESS in 3m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/08cade16ae934a62a926a5fedc252c37 : SUCCESS in 5m 21s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/33ff54f2127a41ab8610426768a18362 : SUCCESS in 4m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5307322242ec47ab904c0d5eeafd1303 : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":1},{"id":"3075f03a9dbfe4e74f04af3a463f6e9e2c183e5e","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2020-11-30 16:25:27.000000000","message":"Patch Set 1:\n\n:/ now hitting new/unrelated https://bugs.launchpad.net/tripleo/+bug/1906265 \nERROR: Could not find a version that satisfies the requirement cmd2\u003d\u003d0.8.5\nhttps://zuul.opendev.org/t/openstack/build/e8e1230a1b0d45d78f3c6e9575ed4fe2","accounts_in_message":[],"_revision_number":1},{"id":"6c8b04e0dcbccd10dffee98d39dd5234b7bda785","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-30 16:44:05.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e3329126892b470a9d6a20066eae3b71 : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8e9773de88544dca3ec6f93296cfa29 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f97473d557c64be2a23a65b8da9aa2ad : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c07d2af3edbf493796615991c42ca564 : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58eecbe3ceb94da6968c48ae3d15374e : SUCCESS in 4m 24s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/da765ca9f5394a1eaa15399e555a12f8 : FAILURE in 8m 43s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/7856e9c1182e4005a7405a6ca1792670 : FAILURE in 7m 00s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/b16c53bc9cb7442687bc650cdbe3483e : FAILURE in 7m 11s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/65bf83d2a6384c6d8b5d60f354d35f59 : FAILURE in 7m 21s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-victoria https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades-victoria https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-undercloud-upgrade-victoria https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone-upgrade-victoria https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f8e9e0ef397946c6b08e080883d44a67 : SUCCESS in 3m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fe6b91ca8e614d0883d40277156c4e13 : SUCCESS in 5m 50s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/a2716f81678d4b918cba31dd90c76053 : SUCCESS in 4m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/39d8496437f3450aa8685f8f11ebe473 : SUCCESS in 6m 09s","accounts_in_message":[],"_revision_number":1},{"id":"55d500eef46cd1d72fdd0e9f526ad492d19be241","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-12-10 19:55:36.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"812a0004cd5af583a7f8ce9b067b8b4ca97198ee","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-12-10 21:25:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/d951fec3d94b4ce3afa1e65c208cfd46 : FAILURE in 1h 27m 52s","accounts_in_message":[],"_revision_number":1},{"id":"eb8ad2f0160d11cbb72e48a77d7680ec19d3aabe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 23:46:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fff291b751b04d869e64e0df34208a52 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/465d2ef401ce4cc690757cb0a293b82a : SUCCESS in 3m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2a16e8dc2d114e5cb45e68dcbd9af5ba : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/064bebf40b4548a3ba8e9ca1c7d02cc4 : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d399b388ee1448778f3abf10d591c89c : SUCCESS in 3m 49s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/1631ead39073477eacaaa52e00ee14c1 : SUCCESS in 3h 11m 51s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/70b099b6b9aa40b1846436911d59aa6a : SUCCESS in 3h 48m 20s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/8df4a3b5953b4955a3fc05f8fbb3ddc6 : SUCCESS in 3h 48m 44s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/83fc3abc088e4b29810517db1dfb2d18 : SUCCESS in 55m 18s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/3c5e29aaeb2c44d7bcbc2aa24cf43bef : SUCCESS in 2h 01m 13s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-victoria https://zuul.opendev.org/t/openstack/build/37ffd0631ca24c67aca04b0dc7f184c9 : SUCCESS in 1h 54m 00s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades-victoria https://zuul.opendev.org/t/openstack/build/1c8dfe2d21de47528a19a15abe98dbee : SUCCESS in 2h 00m 08s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/4d1fd50c94b64ff680f755587ce08975 : SUCCESS in 1h 16m 20s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/944e451837bf4cd3b3e1e87bae18b752 : SUCCESS in 1h 31m 13s\n- tripleo-ci-centos-8-undercloud-upgrade-victoria https://zuul.opendev.org/t/openstack/build/57ce166ba1554377b96f36b135c484a0 : SUCCESS in 1h 49m 02s\n- tripleo-ci-centos-8-standalone-upgrade-victoria https://zuul.opendev.org/t/openstack/build/ff547b845b6d498a9eb104a26926e001 : SUCCESS in 2h 16m 29s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b9872598f1794d02a96d54f7fffb7e6d : SUCCESS in 3m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c27a7c1de42f4d3ba90f9256222778f0 : SUCCESS in 5m 36s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/c0d4ab989c3048b79881324159ff21ff : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":1},{"id":"622c8c9232f37e13de09c2557406c4b90415e054","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 23:46:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3918f0ea958f8a331c7afda32577e40d9958ad51","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 12:08:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7fcc19bb7894442296de4c707f8103c0 : SUCCESS in 3m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c28e9d51828346d8ac583ddb2de9777c : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/345b2fa2a2c447eabb210c0eb8e10ba6 : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09691ac794e444689dcc10ce10a46757 : SUCCESS in 5m 07s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/53d7bbf4d7ca48758fbdee47d86d78fc : SUCCESS in 3h 18m 10s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/65ca34db25e146a2ac1da86808e47952 : SUCCESS in 3h 42m 44s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/907a1d5fc7264ae08519131699163691 : SUCCESS in 3h 42m 25s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/5cc7c112c10547dea8c56709b22e43d1 : SUCCESS in 50m 02s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/9a4a58c9608145dcba14a2a3594c3445 : SUCCESS in 1h 58m 03s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/e2a78b6ca4784df0af63f0eccf5aaac5 : SUCCESS in 1h 35m 49s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-victoria https://zuul.opendev.org/t/openstack/build/1dd720f67e0d4615b0fc01153c4fe824 : SUCCESS in 1h 53m 23s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades-victoria https://zuul.opendev.org/t/openstack/build/ca5a17c93eaf4ab8b92ca345c55966ca : SUCCESS in 2h 00m 56s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/ee9706cc3b134acd80222667cb2bc717 : SUCCESS in 1h 13m 35s\n- tripleo-ci-centos-8-undercloud-upgrade-victoria https://zuul.opendev.org/t/openstack/build/8a13863f10984c7a8dc306d34b0e1ca5 : SUCCESS in 1h 43m 25s\n- tripleo-ci-centos-8-standalone-upgrade-victoria https://zuul.opendev.org/t/openstack/build/1f79b28cb70e4731b49c2844e38146f7 : SUCCESS in 2h 24m 15s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e3d3836fc9fe4b46a27334584fa762b1 : SUCCESS in 3m 46s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/7496a88e0adc4824850848e0e7dc08d1 : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":1},{"id":"20a25767ea8d63a85435ceed864d8849a8f12a44","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 12:08:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7b7d527d9203a2ad855055eb4844e8452a2b2331","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 12:09:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c823173a62f456298c929cbbc5074fa : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"63feca91e07f9b5a9ff2494f076cf4975d4a7eb2","revisions":{"63feca91e07f9b5a9ff2494f076cf4975d4a7eb2":{"kind":"REWORK","_number":1,"created":"2020-10-26 13:05:24.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/83/759683/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/83/759683/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/83/759683/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/83/759683/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/83/759683/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/83/759683/1"}}},"commit":{"parents":[{"commit":"d9c3e3c618eae99ad0b71d18bdcdf08e6fa1eaa6","subject":"Merge \"Lower ansible poll interval time\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/d9c3e3c618eae99ad0b71d18bdcdf08e6fa1eaa6"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-10-26 13:05:23.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-10-26 13:05:23.000000000","tz":0},"subject":"Update .gitreview for stable/victoria","message":"Update .gitreview for stable/victoria\n\nChange-Id: Id1a2d3e5bedb0f15d0f87fbf0538d7fed0b08a87\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/63feca91e07f9b5a9ff2494f076cf4975d4a7eb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/63feca91e07f9b5a9ff2494f076cf4975d4a7eb2"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
