)]}'
{"id":"openstack%2Ftripleo-common~763087","triplet_id":"openstack%2Ftripleo-common~master~Ibbdf63ec372e8496d094f45f2eaef49f9a79d464","project":"openstack/tripleo-common","branch":"master","topic":"bgpsupport","hashtags":[],"change_id":"Ibbdf63ec372e8496d094f45f2eaef49f9a79d464","subject":"Frr image support","status":"MERGED","created":"2020-11-17 21:07:06.000000000","updated":"2021-02-04 17:17:41.000000000","submitted":"2021-02-04 17:14:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"763087-bgpsupport","meta_rev_id":"21f9369cf387ac15221dfe265559d1a068ea8bb6","_number":763087,"virtual_id_number":763087,"owner":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-04 17:14:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":1,"date":"2021-02-03 11:13:23.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},"recommended":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":1,"date":"2021-02-03 07:45:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2021-02-04 13:49:09.000000000","_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-04 13:36:56.000000000","_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2021-02-04 13:49:09.000000000","_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"CC":[{"_account_id":12398,"name":"Dan Sneddon","email":"dsneddon@redhat.com","username":"dsneddon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-18 14:51:09.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-11-19 09:27:05.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-11-19 12:10:27.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-19 14:25:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-15 15:39:19.000000000","updated_by":{"_account_id":12398,"name":"Dan Sneddon","email":"dsneddon@redhat.com","username":"dsneddon"},"reviewer":{"_account_id":12398,"name":"Dan Sneddon","email":"dsneddon@redhat.com","username":"dsneddon"},"state":"CC"},{"updated":"2021-01-13 11:37:56.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"REVIEWER"},{"updated":"2021-02-04 13:49:09.000000000","updated_by":{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},"reviewer":{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},"state":"REVIEWER"}],"messages":[{"id":"b5f69acd4163800d87833d42c4377c0dbaf990f9","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2020-11-17 21:07:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"22ab060c086798cd9221d44c2cfe978b1529e97b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-11-17 22:40:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/1d2b77b3d9844dd69df1045b6b2d634a : FAILURE in 1h 32m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0b92c68781b47e46574772bdcbb27502eab69097","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-18 01:16:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d36eb653aeb4407cbe0d90a1a99078b8 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4954e72020f64dd3beee8997f1785383 : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/60820bd1044d4ed28adcf10b73a5ca1f : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/846561370e32470ca1c99fad93162e1f : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2cc5d8b956df488fb066f77fbdff4c6d : SUCCESS in 5m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74c804bf43ac43609d96f60ffaca2b2b : SUCCESS in 3m 56s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/743cce888089491395ea26715940d620 : SUCCESS in 3h 41m 21s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/df310587672546a3a327e0390b44988e : SUCCESS in 4h 02m 00s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/484af9485ce34f12a49b18d677913e3d : SUCCESS in 47m 37s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/1a393d751ffa458bb40832683fcd812a : SUCCESS in 59m 27s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/dbfba446e901492bb3aea77de68e0313 : SUCCESS in 2h 08m 03s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/9aba562fb52b4d8d9101604bd47ab009 : SUCCESS in 1h 52m 12s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/eaff7ca69cfc435bbd93a827e304f044 : SUCCESS in 2h 07m 10s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/35f651285d5a4d32a8fe13cfe9e817d9 : SUCCESS in 1h 33m 03s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/20db02ccce0d459aa7d9a3d0e44a825a : SUCCESS in 1h 18m 44s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/06cfd2c88c3842d7b3d219676f7f48c1 : SUCCESS in 1h 54m 11s (non-voting)\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/f1e2993e82d64ab7a766e6eb4e43ca4c : FAILURE in 49m 29s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/4291cd1fcb064ed4b61590f83c62860b : SUCCESS in 1h 33m 51s\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/5d671017aa2d4e76a9f178514b61f670 : SUCCESS in 1h 40m 02s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/8634f6c23f28435b841042b2c6a74b88 : SUCCESS in 1h 19m 43s\n- tripleo-ci-centos-8-standalone-upgrade https://zuul.opendev.org/t/openstack/build/41391b34e1ab447da65e8895562bc469 : SUCCESS in 2h 30m 22s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/320d1071715f445f8e8aa75684a1cf52 : SUCCESS in 3m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9d60a1a6b3a84066933b99b3dfce3fe2 : SUCCESS in 4m 19s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/e1f5554d1af14fdead3f6c6aa3407861 : SUCCESS in 4m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0f7024e7f70b476985b4ad6e2531768d : SUCCESS in 6m 28s","accounts_in_message":[],"_revision_number":1},{"id":"8d70cb4b40cb067abf51a6e7cc9c3a46a3926a97","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2020-11-18 08:57:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"690ff5720b233860f7667e4c7a675314529e8b55","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-11-18 09:28:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/60779637dc9449aab29fcf14ff5e38de : FAILURE in 29m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bdaeeab171240e649c6c0e7556c9fc46b224ced4","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-11-18 10:12:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"616fc64ac693e96633c12f93abcf8ea2590f6abd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-18 10:38:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ca16cee39d6645d6b21af3dcb507d904 : FAILURE in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/307401f787164c4688fc421e94d93dc0 : SUCCESS in 3m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1fb06b2700184e56816fda9774333322 : FAILURE in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1592339ca8b04daaaec0d3c3015ed564 : FAILURE in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/66fe1577babe441aa6bf410309fb2591 : FAILURE in 5m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6d890e064554ccaa43ba0004234bcd3 : SUCCESS in 4m 49s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/fdb9e867d082490ca1a2c316fe806030 : SUCCESS in 1h 13m 23s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/63a72560227b430dadf764b20a5dd1eb : SUCCESS in 1h 16m 47s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/1ab904572de646ce85b87136e0d72832 : SUCCESS in 57m 56s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/ec9d622cee3747aa90597b30d34e9afe : SUCCESS in 49m 36s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone-upgrade https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5ef582b07f3147c5acc85d132c3d34be : SUCCESS in 4m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4b090c30f84344a7a83fa412b9fb334a : FAILURE in 5m 25s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/5dbda54a11f84f54985a2dc7fca385f0 : SUCCESS in 4m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b58a8a6bec064b9e8cb9103598cae8fb : FAILURE in 5m 37s","accounts_in_message":[],"_revision_number":2},{"id":"bc4f6233c16b8c018b3e017a1f86fe8f38e1ec58","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2020-11-18 11:26:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fbbe618c9e3f852e5ef3aa8b9fb76ca1d5d20c78","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2020-11-18 11:45:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"672bf319f36a3c66696662b7fe6940a629172749","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-11-18 12:24:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"502e341942eb11cd0db2c9dcc45b0a31e23a7032","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-11-18 13:24:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/a17f6b7c438c4a5d9878291999a6907c : FAILURE in 1h 35m 53s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f6820a6e932342c6a610564da17d39a0af11c57f","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2020-11-18 13:52:04.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c157cd3e65acae2348d485c299fae51f8d42bf18","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2020-11-18 14:10:16.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"c422ff56bc4b616e49e1a9149bc1366295ad5517","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2020-11-18 15:28:18.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"f830bc85ec85ad6b83c27b1f273ce6eb8be5aeda","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-11-18 20:25:46.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/7a8262aca4d14f3481bc02b67e3c79c6 : SUCCESS in 2h 34m 25s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"3ec424ad882ca08a749c58ad38dc9c89b70178ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 05:08:46.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3b6c90e4bf6846ab8fbdc87d23795095 : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/decfb24027c441f6ade07429fa3ae0ea : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/634020877888442b871286d59e3651e5 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/56d8f9f053c14c7ba2a392b772170de6 : SUCCESS in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/94f23d27531a432099e79c2382bd7afb : SUCCESS in 4m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9afc302163e64a6e929b2214579c272b : SUCCESS in 3m 40s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/7bccb441c1c3493bb5454fe5c2de6ae8 : SUCCESS in 4h 59m 38s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/d9e497d33eb347119fa8534bae246fea : SUCCESS in 7h 01m 58s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/6f0730c424df46a3b798dd5e6f7e1fe9 : SUCCESS in 1h 16m 40s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/499f882bce0146d69808b3e0d6d1abdd : SUCCESS in 46m 40s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/874e6f4786794f8dae64fa39fa5f7d5d : SUCCESS in 1h 48m 50s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/b9d1947b7a144136ad4b6e28458d835a : SUCCESS in 2h 12m 08s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/fbebdba758c64f6aaa161186d2c0068d : FAILURE in 2h 37m 30s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/c818f9098347488d8bad36c5649efa5c : SUCCESS in 1h 11m 54s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/7beba5d3d4314b5bb15180d9bba64da7 : SUCCESS in 1h 02m 12s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/9ae02cddf972485f9043a5eeca10f6ad : SUCCESS in 2h 39m 03s (non-voting)\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/666fee4f04f2492aae26252bdb031c69 : FAILURE in 23m 21s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/e1e03470ed9c496ab13a3f11266c7cb8 : SUCCESS in 1h 15m 33s\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/4c7fc419435f4ac7801030b26465a42f : SUCCESS in 2h 09m 17s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/dc38b560fad14277b8a64bc98d3bdf22 : SUCCESS in 2h 04m 17s\n- tripleo-ci-centos-8-standalone-upgrade https://zuul.opendev.org/t/openstack/build/5abd6b7cbd82400885f7a0c119607f14 : TIMED_OUT in 3h 24m 08s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9453431404b74129b0ab8650e87a478f : SUCCESS in 4m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/566f451477874361a820d864d1c063f9 : SUCCESS in 4m 28s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/7e47c292edc543d5adcc350d244a6dda : SUCCESS in 5m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9ea525b64187495f88d9f8e6eec28904 : SUCCESS in 6m 41s","accounts_in_message":[],"_revision_number":6},{"id":"66fabb72dcb88c9301b0fbda6acf343b2ca10d2b","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2020-11-19 07:52:15.000000000","message":"Patch Set 6:\n\nOdd the build fails with https://8cf70d2ee9f0d7f3b892-1aafb108750144ecd9565eaf1429d9e4.ssl.cf2.rackcdn.com/763087/6/check/tripleo-build-containers-ubi-8/666fee4/logs/build.log:\n2020-11-18 22:28:51 | 2020-11-18 22:28:51.378 39912 ERROR openstack Stderr: \u0027error building at STEP \"RUN ln -s /usr/share/openstack-tripleo-common/healthcheck/frr /openstack/healthcheck \u0026\u0026 chmod a+rx /openstack/healthcheck\": error while running runtime: exit status 1\\n\u0027\n\nAs if it did not use the healthcheck from this review","accounts_in_message":[],"_revision_number":6},{"id":"aaf245c20efca056c6ca6cc7764afcf376301004","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-11-19 08:44:25.000000000","message":"Patch Set 6:\n\nNeeds chmod 775","accounts_in_message":[],"_revision_number":6},{"id":"07c70941c2805fd2b39fa49de7fbf96e252b3d66","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-11-19 09:27:05.000000000","message":"Patch Set 6:\n\nThe problem appears to be that RPM openstack-tripleo-common-container-base isn\u0027t being updated despite being in the required-projects list, so healthcheck/frr is missing in the RPM package.\n\nhttp://mirror.gra1.ovh.opendev.org:8080/rdo/centos8/component/tripleo/46/37/4637305569dc5e557a399f9aefc4ecff87f01cb0_fd002ed1/openstack-tripleo-common-container-base-13.1.0-0.20201113044355.52112d4.el8.noarch.rpm","accounts_in_message":[],"_revision_number":6},{"id":"34d6099c94f0cd496c82544e0217d45f4b367acb","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2020-11-19 09:33:17.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"fddb0e818a30f0d176309051c53d36843c2b01f7","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2020-11-19 09:40:05.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"abafb0d37b7010faaebc44a6e176c6e970f3c6e9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-11-19 12:10:27.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/969ead6627844dfd9f1683d558549df2 : SUCCESS in 2h 29m 03s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"86b286cd2ab5d56a4b20323b60904e62729c4c51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 14:25:14.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/27e0af1bdff845f581a99245e2242a1e : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcd2ed97873b409ca62fab488289cdd9 : SUCCESS in 3m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e22c7ffba18d49d7bf683b9d3ea21552 : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/37e6686561194892a3ff1e18b4feafb7 : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1dcaff8425d14d288560921c6f12ebe3 : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db58f6074b26488c9234d2eb9dc340fc : SUCCESS in 4m 08s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/fe669812a5e0452094e6e4628a053372 : SUCCESS in 4h 25m 51s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/fca0d8b55f8c46beb2594ae97767ba02 : SUCCESS in 4h 36m 14s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/dd3f87d8be864b9c862f1eb73db10595 : SUCCESS in 1h 03m 36s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/25dbdb6499924b60aba645a852f86ad9 : SUCCESS in 1h 04m 35s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/03b87874c3d8460ab44a1c006f8b8d78 : SUCCESS in 2h 10m 29s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/dc915b25c86e49ce89116a32bd6d0109 : SUCCESS in 1h 54m 24s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/535a38e8cc6841a09e1037d7a84f57ff : SUCCESS in 2h 06m 05s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/2419d074172a4cff9e660b9d68f33d98 : SUCCESS in 1h 44m 46s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/e64cf30ac918492aab59e332610f1c42 : SUCCESS in 1h 24m 35s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/84f92bff219f4aeeab47c9b86a7c14a7 : SUCCESS in 1h 50m 38s (non-voting)\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/e1264b6122ec4f21a287cfc6c5ebd6fc : SUCCESS in 44m 53s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/b597a6b91920486ab6d2c4a778a65b83 : SUCCESS in 1h 51m 02s\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/459e259e721c47d09ba59b8867c412bf : SUCCESS in 1h 38m 31s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/ba2ab45366c94d91be10b45a5d7fe13f : SUCCESS in 1h 25m 13s\n- tripleo-ci-centos-8-standalone-upgrade https://zuul.opendev.org/t/openstack/build/2a2a4c12b9284409bad3c5e389f6365a : SUCCESS in 2h 55m 35s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2ff7161835b841e9ba47943b5e1527e8 : SUCCESS in 3m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a1bdca0f1bce4f4bbded9bfe3949de9b : SUCCESS in 4m 32s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/318bed6a4ca84c028c280d43160eb0f7 : SUCCESS in 5m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c14ee82d0ed34783967da724e0ba8825 : SUCCESS in 6m 10s","accounts_in_message":[],"_revision_number":8},{"id":"21d28025f8236aabf067ab084e6635ccaab242d1","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2020-11-23 09:08:41.000000000","message":"Topic frr-image removed","accounts_in_message":[],"_revision_number":8},{"id":"948d96879cf6fe7063fcf13f3bfc5e2971291a7b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2020-11-23 09:08:51.000000000","message":"Topic set to bgpsupport","accounts_in_message":[],"_revision_number":8},{"id":"e89a8b606ec7f761452aa63aeeb4f393ca03be0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2020-11-25 21:11:08.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"4ded43645313e3187c7d3da0082d4263bcb9099a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-11-26 02:08:57.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/d40f472b28d24c5e816d9a4d4c535856 : SUCCESS in 2h 44m 10s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"1e23f33e4a833e012b341e309983d34366843ca6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-26 05:09:47.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1cd957589dd14e39b73c17a51ca19806 : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e19b1372d74f4d239d2d328c68d6d0c9 : POST_FAILURE in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2e70f1ffee1e437599f6506699d73f3b : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2f21af3bb7804c1d8ab6394edb12286c : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a7a32393d1974771a6216132645077ff : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90181cdbe2da43f79dc596e9553c6d83 : SUCCESS in 4m 32s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/7bb47b5080c94557a888f86224ab4414 : SUCCESS in 4h 46m 10s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/d469d488fb2d43b1bba3ada77ed0fb0f : SUCCESS in 4h 56m 33s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/db40754dc9054f29a86ae086aec61d9e : SUCCESS in 2h 02m 37s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/9a3e2a12550b460fa32daf61caa3b34a : SUCCESS in 46m 47s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/80bf74b4c2184783ad5a8f3be1382b88 : SUCCESS in 2h 04m 12s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/9b7a7bfd3e4c402bb40a73f029a5469d : SUCCESS in 2h 09m 33s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/e88f9ca1ff4349289e9265cabfb91513 : SUCCESS in 2h 24m 52s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/ba63a5f98b5a4eff8500742cf48e85a5 : SUCCESS in 1h 32m 18s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/c5d6e7c371e447d181e512963d25159d : SUCCESS in 1h 18m 10s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/3e1f212292cb4956a97dcdaa190996f7 : SUCCESS in 1h 52m 49s (non-voting)\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/58d065650ef148cda1f6965dd7513aa3 : SUCCESS in 1h 52m 39s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/9a9e664c36f745109f332fd6b596132f : SUCCESS in 1h 32m 32s\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/34e4bc90d0674ca79552b10c096cca1f : SUCCESS in 1h 28m 47s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/0b7251d9253144efa90533ca6286738e : SUCCESS in 1h 10m 33s\n- tripleo-ci-centos-8-standalone-upgrade https://zuul.opendev.org/t/openstack/build/c2a9041de1714f3a85af5ecc0d420ef1 : SUCCESS in 2h 43m 07s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/75ed671e59a04785ac9f7be8831cf550 : SUCCESS in 4m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7376a59d80924d45bfdbe23c64eddca0 : SUCCESS in 4m 18s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/e04a8d2818c24e8f96fef66a1d5b77ad : SUCCESS in 4m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d43d7d6b058e46b1ab446dcc97f3b6e0 : SUCCESS in 5m 54s","accounts_in_message":[],"_revision_number":9},{"id":"2fc8702006aed81483b1ceeda454cd5c3b69c5f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2020-12-02 22:31:23.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"4f4feb6ead6d5b5aebc378ae33eff227b68b3c4b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-12-03 01:24:33.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/3062d44fb97547a18fa1d3cf67ba9906 : SUCCESS in 2h 39m 09s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"375b762800520769d703358c805569177d8257fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 07:45:15.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd2c11defe0944849760861a1f62370f : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cbec2b7bc714ae0883af856e38b9143 : SUCCESS in 3m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/682cb6393a784fb8b0168460ae3319df : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ab94eade242043ca98ecd96fa4ce1e4d : SUCCESS in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7340bfe178814c5fbb5dc9feacace044 : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/638b3735c8a341f88f385730b31948dd : SUCCESS in 3m 43s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/a35c84c412314a069cc060c2322b1a53 : SUCCESS in 4h 16m 22s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/b9e03894fd014ba8951f43008df4bfba : SUCCESS in 4h 22m 02s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/99617c0d5e8d4be7bbddd7d790b2fb08 : SUCCESS in 49m 30s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/4836e99def5647cf905db7f8967c50e6 : SUCCESS in 45m 22s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/bd880a2a3a254d8d86a0fa9933ea63b6 : SUCCESS in 1h 52m 56s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/0659c1eab73e4ff4a90afc2afa2bd831 : SUCCESS in 1h 50m 43s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/70ca86ee7b504a4694966bfda8a0c7e6 : SUCCESS in 1h 57m 36s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/954c66fff58744f8b2069a5399385348 : SUCCESS in 1h 27m 03s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/be7055fc8b84419aabc642df87cce224 : SUCCESS in 1h 18m 52s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/0e2a277b2fcd460fae4e8ff07c5833d2 : SUCCESS in 2h 02m 14s (non-voting)\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/49858d5a1df54878a4f83b00a8c907bc : SUCCESS in 43m 42s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/39f0f220f2214258b33ccbdc14e22a86 : SUCCESS in 1h 36m 25s\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/0f25e44ce18845bb91ec581b44bf3da6 : SUCCESS in 1h 38m 09s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/8bce4ee781e648489c064b6cd1b0df5f : SUCCESS in 1h 20m 07s\n- tripleo-ci-centos-8-standalone-upgrade https://zuul.opendev.org/t/openstack/build/764b2f58edb0495281cf5cfb0b7d8441 : SUCCESS in 2h 41m 44s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8b6d95f43e5a48da90613af766b79b5b : SUCCESS in 4m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d08a7c198fa34a35af2d82887c7914d8 : SUCCESS in 4m 55s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/64618248eb8b483aae469fc0d9868c58 : SUCCESS in 4m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7b18c5c9bb9245dab8d56a6af14c6661 : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":10},{"id":"5632f9364e47ec7ac712455ac5509a0c8882f167","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2021-01-13 11:37:56.000000000","message":"Patch Set 10: Code-Review+1\n\nWhy DNR/DNM? We will likely want FRR containerized... So then, when the image will be available, we can start testing things with the feature turned ON","accounts_in_message":[],"_revision_number":10},{"id":"a82db8cc51d42187a2b6c4e676ced88ddc270597","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2021-01-18 08:50:49.000000000","message":"Patch Set 10:\n\n\u003e Patch Set 10: Code-Review+1\n\u003e \n\u003e Why DNR/DNM? We will likely want FRR containerized... So then, when the image will be available, we can start testing things with the feature turned ON\n\nI was kinda hoping we\u0027d solve the container vs systemd conundrum first. Give me a couple of days and I think then we can decide with more certainty","accounts_in_message":[],"_revision_number":10},{"id":"487b416f5c612bb3f6b28e166879044cae68be81","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2021-02-03 07:43:24.000000000","message":"Uploaded patch set 11: Commit message was updated.","accounts_in_message":[],"_revision_number":11},{"id":"8728af0a16c75e3468506a0ee9c79b07c6430ae0","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2021-02-03 07:44:49.000000000","message":"Patch Set 11:\n\nSince the step1 patch (https://review.opendev.org/c/openstack/tripleo-heat-templates/+/772738) has merged, we can be assured that we indeed can run frr containerized, hence unWIPing this one","accounts_in_message":[],"_revision_number":11},{"id":"5b35cfc9b0b26f88d6bf042b387a72012b494cd3","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-02-03 07:45:32.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"5e4bf809b53749e46fb65e85dd4be176e008c6ed","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-02-03 11:13:23.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/492c3bc789974b81a30ad8e86314f1c7 : SUCCESS in 3h 28m 46s","accounts_in_message":[],"_revision_number":11},{"id":"00cfec40f9157cc081fa9965550c7751e5c55da8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 11:58:40.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f31af1763a7402d8fd614d910e1904b : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9421b7eed58340fa8e831f40088ba751 : SUCCESS in 3m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa959486323446b3ae4d3a6e77486423 : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f785757c1f4404ab9cfae3dc7e63ebb : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e4e663e40a70444b89babb1e6381c598 : SUCCESS in 4m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00ba3ac8443541859763d9ec4f3539b9 : SUCCESS in 4m 00s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/9fb0bbb80e3c409991f43f4e1a526d74 : SUCCESS in 4h 00m 28s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/7cce998b76374351a069a57f76cc0493 : SUCCESS in 2h 01m 28s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/c9606321f12c43a98798ff9b73665d18 : SUCCESS in 1h 33m 54s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/07f7f559533e489c8bdea1f752d71076 : SUCCESS in 1h 13m 30s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/bf2e90a6f29d491c9d7fe8ac32bf1179 : SUCCESS in 1h 31m 32s\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/09616fcc0fe14366ba2babd5035ad4b3 : FAILURE in 1h 34m 06s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/e252e6b5c6e34df1b25c4a895af07113 : SUCCESS in 1h 18m 11s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/7d5f89d15a3044c689eecbd7a06114a9 : SUCCESS in 4h 09m 53s\n- tripleo-ci-centos-8-standalone-upgrade https://zuul.opendev.org/t/openstack/build/3629b50847334dbba0b811b63168d639 : SUCCESS in 2h 28m 26s (non-voting)\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/9b1bf5eb7ac04c7db552ef08d88f241c : SUCCESS in 1h 40m 40s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/873b2e4afe0f43df8b8994c02a216dc9 : SUCCESS in 2h 02m 10s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/146c9dacddf042bf8ad926e1d0091855 : SUCCESS in 1h 58m 09s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1d071d163c7b4287bbe3daa32f445610 : SUCCESS in 4m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a318626ecb80440aa36408d9d2c6509b : SUCCESS in 4m 30s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/97971cc1d14146c78c4e89a587e6c7e8 : SUCCESS in 4m 21s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/d16b90c7e30642b6942a03061ee11978 : SUCCESS in 41m 54s","accounts_in_message":[],"_revision_number":11},{"id":"f4a7e47d771aa034a7ffaebefcf1543260ca764a","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2021-02-04 13:36:56.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"5f8093e9ddfc1e35d00ba3576ab07d3a5ca869c2","author":{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},"date":"2021-02-04 13:49:09.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"281a38751f29b44d90bb90aec13f1c9b5345d4b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-04 13:49:56.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"5327fc1d7b22254189bcff79a02d68cf52bed370","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-04 17:14:23.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9856b8151b69478882e8514faf1c7573 : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c905cff16e644587ab92d67faa3556ac : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/94fea85d2b9544b4bff33abd3b86e246 : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d28e20aac8f4259bafedb1563ea791c : SUCCESS in 5m 27s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/ad2afa71e1a940b592a5395decc17e67 : SUCCESS in 3h 19m 43s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/c5f84bef9f454d419e010a96a579cfcc : SUCCESS in 1h 57m 15s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/a64138ef8f634d0095664a41eab41c3b : SUCCESS in 1h 29m 53s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/65f973ea68a845bbb595da8ed67ba41d : SUCCESS in 1h 12m 58s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/358fde45a0df4e2aa5bfeb3687237677 : SUCCESS in 1h 43m 40s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/af87c4f02aa3442690dfdf3fb83cd47a : SUCCESS in 1h 28m 03s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/c652dad7359444f998872852f90ecb33 : SUCCESS in 1h 11m 58s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/2ea461c1c460490fa8ee1966b5c6e421 : SUCCESS in 1h 47m 39s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/e61e4250a66d48de8815562d8b37fd1e : SUCCESS in 1h 08m 19s\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/c9021637e19c462b81b630eae820d0b9 : SUCCESS in 1h 21m 00s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/74dc95b6102440e781eddd1463916fe1 : SUCCESS in 1h 57m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/aaca21d1e34b4b7cb00a3b54a0909c0d : SUCCESS in 4m 54s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c45f8c3bab0c463886a880b46e375296 : SUCCESS in 5m 24s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/929b024459624eed8d0edddc24c33728 : SUCCESS in 5m 35s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/1f76b7b7671043f9907710cb5c55f4cf : SUCCESS in 1h 07m 42s","accounts_in_message":[],"_revision_number":11},{"id":"c48952e4c70416123acfb01ee0a7e308182f74d1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-04 17:14:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":11},{"id":"21f9369cf387ac15221dfe265559d1a068ea8bb6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-04 17:17:41.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94a0e5c25ab14bb298ca25cd5026db7b : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"380a12c88923f977a7ed9c5896e6996b0cd9a4b8","revisions":{"8523e68deea8bed9ac5d86d12b65429e99c98c5d":{"kind":"REWORK","_number":1,"created":"2020-11-17 21:07:06.000000000","uploader":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"ref":"refs/changes/87/763087/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/87/763087/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/763087/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/763087/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/763087/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/87/763087/1"}}},"commit":{"parents":[{"commit":"61ac88d7c596ff7cd69e0115e39ff5778de69477","subject":"Merge \"Add better exception logging for builds\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/61ac88d7c596ff7cd69e0115e39ff5778de69477"}]}],"author":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2020-11-17 21:06:40.000000000","tz":60},"committer":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2020-11-17 21:06:40.000000000","tz":60},"subject":"DNR/DNM Frr support","message":"DNR/DNM Frr support\n\ntesting\n\nChange-Id: Ibbdf63ec372e8496d094f45f2eaef49f9a79d464\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/8523e68deea8bed9ac5d86d12b65429e99c98c5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/8523e68deea8bed9ac5d86d12b65429e99c98c5d"}]},"branch":"refs/heads/master"},"c5504d5ffa5f2320280d2f05aac3f35d1b9b9484":{"kind":"REWORK","_number":2,"created":"2020-11-18 08:57:52.000000000","uploader":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"ref":"refs/changes/87/763087/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/87/763087/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/763087/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/763087/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/763087/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/87/763087/2"}}},"commit":{"parents":[{"commit":"61ac88d7c596ff7cd69e0115e39ff5778de69477","subject":"Merge \"Add better exception logging for builds\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/61ac88d7c596ff7cd69e0115e39ff5778de69477"}]}],"author":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2020-11-17 21:06:40.000000000","tz":60},"committer":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2020-11-18 08:57:40.000000000","tz":60},"subject":"DNR/DNM Frr support","message":"DNR/DNM Frr support\n\ntesting\n\nChange-Id: Ibbdf63ec372e8496d094f45f2eaef49f9a79d464\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/c5504d5ffa5f2320280d2f05aac3f35d1b9b9484"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/c5504d5ffa5f2320280d2f05aac3f35d1b9b9484"}]},"branch":"refs/heads/master"},"773ae5da98e92d6b11c211f84fba7f842c18fadb":{"kind":"REWORK","_number":3,"created":"2020-11-18 11:45:14.000000000","uploader":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"ref":"refs/changes/87/763087/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/87/763087/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/763087/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/763087/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/763087/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/87/763087/3"}}},"commit":{"parents":[{"commit":"61ac88d7c596ff7cd69e0115e39ff5778de69477","subject":"Merge \"Add better exception logging for builds\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/61ac88d7c596ff7cd69e0115e39ff5778de69477"}]}],"author":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2020-11-17 21:06:40.000000000","tz":60},"committer":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2020-11-18 11:45:01.000000000","tz":60},"subject":"DNR/DNM Frr support","message":"DNR/DNM Frr support\n\ntesting\n\nChange-Id: Ibbdf63ec372e8496d094f45f2eaef49f9a79d464\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/773ae5da98e92d6b11c211f84fba7f842c18fadb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/773ae5da98e92d6b11c211f84fba7f842c18fadb"}]},"branch":"refs/heads/master"},"6f4b67079a9866e1bae0d7a9650692a851903d0a":{"kind":"REWORK","_number":4,"created":"2020-11-18 13:52:04.000000000","uploader":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"ref":"refs/changes/87/763087/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/87/763087/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/763087/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/763087/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/763087/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/87/763087/4"}}},"commit":{"parents":[{"commit":"61ac88d7c596ff7cd69e0115e39ff5778de69477","subject":"Merge \"Add better exception logging for builds\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/61ac88d7c596ff7cd69e0115e39ff5778de69477"}]}],"author":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2020-11-17 21:06:40.000000000","tz":60},"committer":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2020-11-18 13:51:46.000000000","tz":60},"subject":"DNR/DNM Frr support","message":"DNR/DNM Frr support\n\ntesting\n\nChange-Id: Ibbdf63ec372e8496d094f45f2eaef49f9a79d464\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/6f4b67079a9866e1bae0d7a9650692a851903d0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/6f4b67079a9866e1bae0d7a9650692a851903d0a"}]},"branch":"refs/heads/master"},"3988647993c447e3a58db53a3a23f4909a21500f":{"kind":"REWORK","_number":5,"created":"2020-11-18 14:10:16.000000000","uploader":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"ref":"refs/changes/87/763087/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/87/763087/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/763087/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/763087/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/763087/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/87/763087/5"}}},"commit":{"parents":[{"commit":"61ac88d7c596ff7cd69e0115e39ff5778de69477","subject":"Merge \"Add better exception logging for builds\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/61ac88d7c596ff7cd69e0115e39ff5778de69477"}]}],"author":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2020-11-17 21:06:40.000000000","tz":60},"committer":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2020-11-18 14:10:03.000000000","tz":60},"subject":"DNR/DNM Frr support","message":"DNR/DNM Frr support\n\ntesting\n\nChange-Id: Ibbdf63ec372e8496d094f45f2eaef49f9a79d464\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/3988647993c447e3a58db53a3a23f4909a21500f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/3988647993c447e3a58db53a3a23f4909a21500f"}]},"branch":"refs/heads/master"},"c3fe009a74269978ad5c0fde620eb47230f3ed72":{"kind":"REWORK","_number":6,"created":"2020-11-18 15:28:18.000000000","uploader":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"ref":"refs/changes/87/763087/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/87/763087/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/763087/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/763087/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/763087/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/87/763087/6"}}},"commit":{"parents":[{"commit":"61ac88d7c596ff7cd69e0115e39ff5778de69477","subject":"Merge \"Add better exception logging for builds\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/61ac88d7c596ff7cd69e0115e39ff5778de69477"}]}],"author":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2020-11-17 21:06:40.000000000","tz":60},"committer":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2020-11-18 15:28:04.000000000","tz":60},"subject":"DNR/DNM Frr support","message":"DNR/DNM Frr support\n\ntesting\n\nChange-Id: Ibbdf63ec372e8496d094f45f2eaef49f9a79d464\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/c3fe009a74269978ad5c0fde620eb47230f3ed72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/c3fe009a74269978ad5c0fde620eb47230f3ed72"}]},"branch":"refs/heads/master"},"461afae6645ee8a9c54470a00a5ca657b084c035":{"kind":"REWORK","_number":7,"created":"2020-11-19 09:33:17.000000000","uploader":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"ref":"refs/changes/87/763087/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/87/763087/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/763087/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/763087/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/763087/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/87/763087/7"}}},"commit":{"parents":[{"commit":"61ac88d7c596ff7cd69e0115e39ff5778de69477","subject":"Merge \"Add better exception logging for builds\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/61ac88d7c596ff7cd69e0115e39ff5778de69477"}]}],"author":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2020-11-17 21:06:40.000000000","tz":60},"committer":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2020-11-19 09:32:43.000000000","tz":60},"subject":"DNR/DNM Frr support","message":"DNR/DNM Frr support\n\ntesting\n\nChange-Id: Ibbdf63ec372e8496d094f45f2eaef49f9a79d464\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/461afae6645ee8a9c54470a00a5ca657b084c035"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/461afae6645ee8a9c54470a00a5ca657b084c035"}]},"branch":"refs/heads/master"},"db816a1fd72af924084b52b6d55663fc23fdd88b":{"kind":"REWORK","_number":8,"created":"2020-11-19 09:40:05.000000000","uploader":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"ref":"refs/changes/87/763087/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/87/763087/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/763087/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/763087/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/763087/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/87/763087/8"}}},"commit":{"parents":[{"commit":"61ac88d7c596ff7cd69e0115e39ff5778de69477","subject":"Merge \"Add better exception logging for builds\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/61ac88d7c596ff7cd69e0115e39ff5778de69477"}]}],"author":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2020-11-17 21:06:40.000000000","tz":60},"committer":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2020-11-19 09:39:52.000000000","tz":60},"subject":"DNR/DNM Frr support","message":"DNR/DNM Frr support\n\ntesting\n\nChange-Id: Ibbdf63ec372e8496d094f45f2eaef49f9a79d464\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/db816a1fd72af924084b52b6d55663fc23fdd88b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/db816a1fd72af924084b52b6d55663fc23fdd88b"}]},"branch":"refs/heads/master"},"783b72a58b7e3161af65d2d600a65cbfcf97fb97":{"kind":"REWORK","_number":9,"created":"2020-11-25 21:11:08.000000000","uploader":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"ref":"refs/changes/87/763087/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/87/763087/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/763087/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/763087/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/763087/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/87/763087/9"}}},"commit":{"parents":[{"commit":"61ac88d7c596ff7cd69e0115e39ff5778de69477","subject":"Merge \"Add better exception logging for builds\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/61ac88d7c596ff7cd69e0115e39ff5778de69477"}]}],"author":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2020-11-17 21:06:40.000000000","tz":60},"committer":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2020-11-25 21:10:53.000000000","tz":60},"subject":"DNR/DNM Frr support","message":"DNR/DNM Frr support\n\ntesting\n\nChange-Id: Ibbdf63ec372e8496d094f45f2eaef49f9a79d464\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/783b72a58b7e3161af65d2d600a65cbfcf97fb97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/783b72a58b7e3161af65d2d600a65cbfcf97fb97"}]},"branch":"refs/heads/master"},"89d36d39854e37baee63f17fb7dd1eb11866a473":{"kind":"REWORK","_number":10,"created":"2020-12-02 22:31:23.000000000","uploader":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"ref":"refs/changes/87/763087/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/87/763087/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/763087/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/763087/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/763087/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/87/763087/10"}}},"commit":{"parents":[{"commit":"096630560cd3191dbdbb3244f363c05aae26cac8","subject":"Merge \"Switch defaults to quay.io\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/096630560cd3191dbdbb3244f363c05aae26cac8"}]}],"author":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2020-12-02 22:30:40.000000000","tz":60},"committer":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2020-12-02 22:30:40.000000000","tz":60},"subject":"DNR/DNM Frr support","message":"DNR/DNM Frr support\n\ntesting\n\nChange-Id: Ibbdf63ec372e8496d094f45f2eaef49f9a79d464\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/89d36d39854e37baee63f17fb7dd1eb11866a473"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/89d36d39854e37baee63f17fb7dd1eb11866a473"}]},"branch":"refs/heads/master"},"380a12c88923f977a7ed9c5896e6996b0cd9a4b8":{"kind":"NO_CODE_CHANGE","_number":11,"created":"2021-02-03 07:43:24.000000000","uploader":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"ref":"refs/changes/87/763087/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/87/763087/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/763087/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/763087/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/763087/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/87/763087/11"}}},"commit":{"parents":[{"commit":"096630560cd3191dbdbb3244f363c05aae26cac8","subject":"Merge \"Switch defaults to quay.io\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/096630560cd3191dbdbb3244f363c05aae26cac8"}]}],"author":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2020-12-02 22:30:40.000000000","tz":60},"committer":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2021-02-03 07:42:00.000000000","tz":60},"subject":"Frr image support","message":"Frr image support\n\nThis adds support for the Frr container image. Frr can be found at\nhttps://github.com/FRRouting/frr/ and is an implementation of a suite\nof different routing protocols.\n\nChange-Id: Ibbdf63ec372e8496d094f45f2eaef49f9a79d464\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/380a12c88923f977a7ed9c5896e6996b0cd9a4b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/380a12c88923f977a7ed9c5896e6996b0cd9a4b8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
