)]}'
{"id":"openstack%2Ftripleo-common~765077","triplet_id":"openstack%2Ftripleo-common~stable%2Fussuri~I6db21d1f8c0e531c49c151858de9098639fb05f5","project":"openstack/tripleo-common","branch":"stable/ussuri","topic":"bug/1902846-stable/victoria-stable/ussuri","hashtags":[],"change_id":"I6db21d1f8c0e531c49c151858de9098639fb05f5","subject":"TCIB: Pick not best when building on centos/ubi","status":"MERGED","created":"2020-12-02 07:00:34.000000000","updated":"2020-12-15 00:59:00.000000000","submitted":"2020-12-15 00:51:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"765077-bug/1902846-stable/victoria-stable/ussuri","cherry_pick_of_change":765076,"cherry_pick_of_patch_set":1,"meta_rev_id":"099a0f776a586b15df196e605b0cc3320c5eac33","_number":765077,"virtual_id_number":765077,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-15 00:51:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2020-12-14 16:02:01.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"all":[{"value":2,"date":"2020-12-14 14:16:46.000000000","_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":2,"date":"2020-12-15 00:59:00.000000000","post_submit":true,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":1,"date":"2020-12-14 19:55:42.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"reviewers":{"REVIEWER":[{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-02 07:00:34.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"REVIEWER"},{"updated":"2020-12-02 09:44:34.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-02 13:53:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-10 21:10:16.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-12-14 13:04:24.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2020-12-14 13:04:24.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2020-12-14 14:16:46.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"}],"messages":[{"id":"e8a1addba6c4f70b7b81341cffdda127b3a95895","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-12-02 07:00:34.000000000","message":"Patch Set 1: Cherry Picked from branch stable/victoria.","accounts_in_message":[],"_revision_number":1},{"id":"0625dcb933c701bde00fd98d80b02342a2e23602","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-12-02 09:44:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/6324b8c7c45845bb8874d6f27003f5d9 : FAILURE in 2h 41m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"84a221d0598c5319856f0e4c7940f562d195b19a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 13:53:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9a8a3a9802464c56b2a65d83ad4fdd37 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a6176d0e6544ee389bc753c34edc98c : SUCCESS in 6m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5b48c3ce22134d9e921b14e7c031688f : SUCCESS in 8m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7cd844cace124e3e9fb215009ce4fc32 : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14dac2fc7dad4353abd85a8d6de65063 : SUCCESS in 3m 32s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/52b49e2945e9438e83c89ca32ba6f31e : SUCCESS in 2h 18m 16s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-ussuri https://zuul.opendev.org/t/openstack/build/e8747214150f418cbee64c5f08c82659 : SUCCESS in 2h 18m 21s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades-ussuri https://zuul.opendev.org/t/openstack/build/40a2b8dfd5ca4d5b87d6486b4149078a : FAILURE in 42m 23s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/861794b02ba9406483774505a0a081bd : FAILURE in 59m 28s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/71730a1d0edb4281a4dbe6ab6a776c51 : SUCCESS in 1h 34m 14s\n- tripleo-ci-centos-8-undercloud-upgrade-ussuri https://zuul.opendev.org/t/openstack/build/e86bc9507cf042a1b52acd6ff1262a5b : SUCCESS in 2h 49m 50s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/b12b29fd543d4a92b38bfacdd818d6c0 : SUCCESS in 49m 36s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/3b751c2c7c2d42118467ccdd82e9ec66 : SUCCESS in 2h 15m 39s\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/5deae79e2c4245858890ae1caaf577c8 : FAILURE in 1h 28m 08s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/a5695e9508ed4282859968983717cfef : SUCCESS in 1h 44m 35s\n- tripleo-ci-centos-8-standalone-upgrade-ussuri https://zuul.opendev.org/t/openstack/build/6a363345ba944de0831eee0d0cd7d2fc : SUCCESS in 2h 50m 48s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c248eddbb69d414690042670a7157392 : SUCCESS in 4m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/63308a7d42da4511a982625438f32ad1 : SUCCESS in 7m 19s","accounts_in_message":[],"_revision_number":1},{"id":"a74fd30680804c4030a90d9ef43155f70ae0e07d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-12-04 12:01:19.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"2a057430d1d5648caee97bbdbc6470266d40fb2b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-12-04 17:42:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/c3ed9cc72e454148a31234d6096cd572 : TIMED_OUT in 5h 16m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e3e40fd6061375ed20d738aaecca2769ce0c8a02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 22:36:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a440e40114c5476a8efc44d65db4e3c1 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc30ea171b234b5e8a2ab2c8be77b4b0 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8cd1bcbbf6e94effaf08eb854d08579c : SUCCESS in 4m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/df12273087104f42bc4fbb9b00471bdd : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a33704669ea49ca8f55b839f407a42e : SUCCESS in 5m 34s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/69f8d98d9cb94a00aa31d2f511441045 : SUCCESS in 3h 09m 11s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/f9bc77c869364c3680a922d67c21e26c : SUCCESS in 3h 39m 59s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/6008ab8c18584a719a29cac98dbba876 : SUCCESS in 3h 39m 40s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/be0732f4ab4548b8bf991ecd87c23e03 : SUCCESS in 1h 59m 37s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-ussuri https://zuul.opendev.org/t/openstack/build/6a20b5b5d7cc4a9a8726abf678d3859a : SUCCESS in 1h 48m 00s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades-ussuri https://zuul.opendev.org/t/openstack/build/7209d523fbe74fde88619d31efbd28bf : SUCCESS in 2h 00m 02s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/ae1980f302ba4b918518008dc7ec5f50 : SUCCESS in 1h 30m 27s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/d54108b8a3c54a74b728e5e1d057fe7a : SUCCESS in 1h 20m 51s\n- tripleo-ci-centos-8-undercloud-upgrade-ussuri https://zuul.opendev.org/t/openstack/build/2436cae878cb4ede960b5d9f5ca7a5ac : SUCCESS in 2h 20m 51s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/31630a9b4cd74ff7a9ca270fddbbd298 : SUCCESS in 47m 37s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/077d504e4b494299ab42c9de5db78548 : SUCCESS in 1h 28m 38s\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/cd228a526d904e8b900798b760a5d294 : FAILURE in 1h 27m 08s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/227df65f2e3646b0b70b3b0befc73082 : SUCCESS in 1h 15m 07s\n- tripleo-ci-centos-8-standalone-upgrade-ussuri https://zuul.opendev.org/t/openstack/build/197eb6adc5c54ecea461406e49d8998f : SUCCESS in 2h 31m 01s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a577ad41153249cd911750b1af9cdfa2 : SUCCESS in 3m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a0e48bc19d744f0c8d8ab4d5f7494e22 : SUCCESS in 5m 27s","accounts_in_message":[],"_revision_number":2},{"id":"bb2c0930047fc6d9e2674871a1463ab34030b42f","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-12-10 21:10:16.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d4cdebe5acb7a475f5ac0c335dc1b3999dd638ee","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-12-14 14:07:34.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b0c8296f3da7b920534a55f7942c500de75b82e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-12-14 14:07:43.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"ce51b74229439838d7466aa5cd67aa6a4d839300","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-12-14 14:16:46.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"7c9d876a2237fe1d6687b72a090bea023b653bc2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-12-14 16:02:01.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/ac10a8c335794440b87b7525601f3566 : FAILURE in 1h 50m 23s","accounts_in_message":[],"_revision_number":3},{"id":"33ee9c5318e81fd9c39b2f2f9408b75b42f3fb8e","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-12-14 19:55:42.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"08e74beb103cff7984970049dabb8df0596a2e02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 19:57:16.000000000","message":"Patch Set 3:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"06bcbdbb2b78cefa801be6d904374b3ddba7f678","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 21:26:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cc29656cee964eb2a9ed07c7c81b86a4 : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d25f79472eb4f259e4082f5334a720e : SUCCESS in 2m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bb00d2925e2348b68318431f10dbc3b8 : SUCCESS in 9m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e7c009b6dff547409b0f884fc89afaea : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd638ef409bc428299b3953669df6093 : SUCCESS in 4m 05s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/c4c8c3588ab040abad8ea8b8ad2df615 : SUCCESS in 3h 58m 43s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/1dc82397302e4dfcba38028da1eecf08 : SUCCESS in 4h 38m 48s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/4bb4407783f14adeadf8dfaec98b6498 : SUCCESS in 4h 38m 55s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/a8d5cfb9a0a94c79ac92ef18cb7cb777 : SUCCESS in 2h 23m 18s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-ussuri https://zuul.opendev.org/t/openstack/build/324fed5558024496bf79fd7154a39c42 : SUCCESS in 2h 08m 23s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades-ussuri https://zuul.opendev.org/t/openstack/build/260ad8b74fd74155a883e6a5690251ca : SUCCESS in 2h 11m 46s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/388d4bbb7ecc484383a54b06680b8654 : SUCCESS in 1h 37m 00s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/fd9093412a49436c87665b3eadf1938e : SUCCESS in 1h 22m 15s\n- tripleo-ci-centos-8-undercloud-upgrade-ussuri https://zuul.opendev.org/t/openstack/build/821b0b39a2ab4052afe58917f44b9a73 : SUCCESS in 2h 44m 22s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/90f83f10bcb441ca9172d7dde54a17ff : SUCCESS in 1h 40m 35s\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/45bdbbbfa3a44c94a3f67ae1084f2ba7 : FAILURE in 1h 41m 56s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/e68e6d5917584d9dabff3110986b49d4 : SUCCESS in 1h 31m 57s\n- tripleo-ci-centos-8-standalone-upgrade-ussuri https://zuul.opendev.org/t/openstack/build/7952897b4243457493a81a12a68f1127 : SUCCESS in 3h 14m 05s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5042afaf0ac943429e8588b877f72a79 : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":3},{"id":"5eb9f692082d05beda9ff13bcdd77c92c5e688ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 00:51:22.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0da44c1e5ac646a8918c5acc80423847 : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a8ce6495fa64772afdbfd9779dac374 : SUCCESS in 4m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/04688c97ae22494699e8b2d7b7805702 : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9385ec9e9881479294b553aaec6d7534 : SUCCESS in 4m 44s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/8c0940f8238243868df9ae00998ed240 : SUCCESS in 3h 44m 46s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/5ad68c7474e049a1acc21ceb3ab51271 : SUCCESS in 3h 56m 54s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/d9b58e4be4164a8684b46ed8b618e756 : SUCCESS in 4h 00m 18s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/ceae23f85ca748d6bce5ac667bd87ea5 : SUCCESS in 2h 20m 21s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/cd4660301c814892be25e54772c40583 : SUCCESS in 1h 37m 33s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-ussuri https://zuul.opendev.org/t/openstack/build/3fdb0640ae4046b58fe5541234d9b9d8 : SUCCESS in 2h 06m 34s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades-ussuri https://zuul.opendev.org/t/openstack/build/48bd0399ff8d44dcac64819895e56235 : SUCCESS in 2h 07m 24s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/a6ecd0cecb2743db9c38bb3c3c6cd5d6 : SUCCESS in 1h 31m 24s\n- tripleo-ci-centos-8-undercloud-upgrade-ussuri https://zuul.opendev.org/t/openstack/build/2321c04b490e4498b42c71f35972ff12 : SUCCESS in 2h 22m 07s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/5d8dfd0232954778b47f7c90e0eed983 : SUCCESS in 2h 00m 10s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/8adc7b0e0916482b98082c15e86b4e86 : SUCCESS in 1h 40m 40s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/297f9b2ffcda453a80e25c814877e229 : SUCCESS in 1h 20m 21s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/9473771891cb4e0c83f623fbbd9d81df : SUCCESS in 1h 56m 50s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/3573ddce09c94870ae2724ad0ef58e1f : SUCCESS in 1h 19m 58s\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/c57de66e4fe44cd2ac7f89afde06bd09 : SUCCESS in 1h 33m 18s\n- tripleo-ci-centos-8-standalone-upgrade-ussuri https://zuul.opendev.org/t/openstack/build/2b93131867a44ba7b3fea7b5a0db85aa : SUCCESS in 2h 29m 27s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/66128b5b25ed4128929bc76f0d2fb4e6 : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":3},{"id":"b4cece1668378a9fb8b3287ea46f4b01eaac6b5f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 00:51:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"4160e2987f3ae53b73d3a8d45175e211160f7759","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 00:59:00.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/009e7d5e7a334509968e7dcc178d0dc6 : SUCCESS in 1m 38s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6fb38abf65be0928d057db3d134d4ea9b655fc3c","revisions":{"70017dd14010c4aaf4c1ac1a8289e750e1ff7070":{"kind":"REWORK","_number":1,"created":"2020-12-02 07:00:34.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/77/765077/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/77/765077/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/77/765077/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/77/765077/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/77/765077/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/77/765077/1"}}},"commit":{"parents":[{"commit":"b121b41c492b73ff8b2ece8b8c882dde9a809b32","subject":"Merge \"Re-add ndctl to nova-compute image\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/b121b41c492b73ff8b2ece8b8c882dde9a809b32"}]}],"author":{"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","date":"2020-11-18 15:26:44.000000000","tz":60},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2020-12-02 07:00:34.000000000","tz":0},"subject":"TCIB: Pick not best when building on centos/ubi","message":"TCIB: Pick not best when building on centos/ubi\n\nWhen building on UBI (like we do in upstream CI), we may need some\ncentos repos. And those may bring conflicts with ubi repos.\n\nAllowing --nobest mode for the package manager, highly likely\nresolves those conflicts... as the best effort.\n\nChange-Id: I6db21d1f8c0e531c49c151858de9098639fb05f5\nSigned-off-by: Bogdan Dobrelya \u003cbdobreli@redhat.com\u003e\n(cherry picked from commit 257cee80d7a173fb32bf43fb04f8e0dd4244c57c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/70017dd14010c4aaf4c1ac1a8289e750e1ff7070"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/70017dd14010c4aaf4c1ac1a8289e750e1ff7070"}]},"branch":"refs/heads/stable/ussuri"},"2371a18942cbf73b3bed800debb26290c9b6b110":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-12-04 12:01:19.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/77/765077/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/77/765077/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/77/765077/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/77/765077/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/77/765077/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/77/765077/2"}}},"commit":{"parents":[{"commit":"26f19b5ecb31fc8230b43fc15db8db839131f5ed","subject":"[ussuri] Migrate to content provider jobs/template","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/26f19b5ecb31fc8230b43fc15db8db839131f5ed"}]}],"author":{"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","date":"2020-11-18 15:26:44.000000000","tz":60},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2020-12-04 12:01:19.000000000","tz":0},"subject":"TCIB: Pick not best when building on centos/ubi","message":"TCIB: Pick not best when building on centos/ubi\n\nWhen building on UBI (like we do in upstream CI), we may need some\ncentos repos. And those may bring conflicts with ubi repos.\n\nAllowing --nobest mode for the package manager, highly likely\nresolves those conflicts... as the best effort.\n\nChange-Id: I6db21d1f8c0e531c49c151858de9098639fb05f5\nSigned-off-by: Bogdan Dobrelya \u003cbdobreli@redhat.com\u003e\n(cherry picked from commit 257cee80d7a173fb32bf43fb04f8e0dd4244c57c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/2371a18942cbf73b3bed800debb26290c9b6b110"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/2371a18942cbf73b3bed800debb26290c9b6b110"}]},"branch":"refs/heads/stable/ussuri","description":"Rebase"},"6fb38abf65be0928d057db3d134d4ea9b655fc3c":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-12-14 14:07:43.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/77/765077/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/77/765077/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/77/765077/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/77/765077/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/77/765077/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/77/765077/3"}}},"commit":{"parents":[{"commit":"ce20aa6f4940539119d484b496ddbfcd3d206967","subject":"[ussuri] Migrate to content provider jobs/template","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/ce20aa6f4940539119d484b496ddbfcd3d206967"}]}],"author":{"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","date":"2020-11-18 15:26:44.000000000","tz":60},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2020-12-14 14:07:43.000000000","tz":0},"subject":"TCIB: Pick not best when building on centos/ubi","message":"TCIB: Pick not best when building on centos/ubi\n\nWhen building on UBI (like we do in upstream CI), we may need some\ncentos repos. And those may bring conflicts with ubi repos.\n\nAllowing --nobest mode for the package manager, highly likely\nresolves those conflicts... as the best effort.\n\nChange-Id: I6db21d1f8c0e531c49c151858de9098639fb05f5\nSigned-off-by: Bogdan Dobrelya \u003cbdobreli@redhat.com\u003e\n(cherry picked from commit 257cee80d7a173fb32bf43fb04f8e0dd4244c57c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/6fb38abf65be0928d057db3d134d4ea9b655fc3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/6fb38abf65be0928d057db3d134d4ea9b655fc3c"}]},"branch":"refs/heads/stable/ussuri","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
