)]}'
{"id":"openstack%2Ftripleo-common~767754","triplet_id":"openstack%2Ftripleo-common~master~If13571b9e21251ced9a43fc1cc7f7d69fad06f0a","project":"openstack/tripleo-common","branch":"master","topic":"buildah-retries","hashtags":[],"change_id":"If13571b9e21251ced9a43fc1cc7f7d69fad06f0a","subject":"Change retry logic on buildah build","status":"MERGED","created":"2020-12-18 14:24:00.000000000","updated":"2020-12-23 17:29:21.000000000","submitted":"2020-12-23 17:28:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"767754-buildah-retries","meta_rev_id":"60efb4c51dd47205bf0c6c00d066d22b3226d1f7","_number":767754,"virtual_id_number":767754,"owner":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-23 17:28:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-12-18 17:23:39.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"all":[{"value":2,"date":"2020-12-23 11:03:46.000000000","_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":1,"date":"2020-12-23 13:45:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},{"value":2,"date":"2020-12-23 13:40:35.000000000","_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":1,"date":"2020-12-23 13:40:18.000000000","_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}],"reviewers":{"REVIEWER":[{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-18 14:26:38.000000000","updated_by":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-12-18 14:26:38.000000000","updated_by":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2020-12-18 17:23:39.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-18 21:20:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-23 10:59:53.000000000","updated_by":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"CC"},{"updated":"2020-12-23 11:00:02.000000000","updated_by":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"},{"updated":"2020-12-23 11:03:46.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2020-12-23 13:40:18.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2020-12-23 13:45:55.000000000","updated_by":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"reviewer":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"state":"REVIEWER"}],"messages":[{"id":"ab2a539bbbe72998efc5c1761929ecfd6de826fe","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2020-12-18 14:24:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"23f5a9c0d149ba4b43f40da7557c32649a2ebf26","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2020-12-18 14:26:38.000000000","message":"Patch Set 1:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":1},{"id":"b0943b51fe104b1e175c32f5af7510bc5a27dd68","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2020-12-18 14:28:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"00d85f69ad085082da892811c0a208df23a84348","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-12-18 14:30:36.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3673c26de9dafe86ced1e9d57daf3c39e13441be","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2020-12-18 14:33:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9bb09c99eec5ac45ea8ee378253f23cfa2057133","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-12-18 14:34:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"741200cf202d0efbabe7d849d6c2ab61f664f5e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2020-12-18 14:35:02.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"ce7a265a1e981a04710251bffdc03924cbf4b09a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2020-12-18 14:35:33.000000000","message":"Topic set to buildah-retries","accounts_in_message":[],"_revision_number":3},{"id":"f49ea7f4445dc691a836a016fafbd1331a47dd99","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-12-18 17:23:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/ae3ba14206e24cedbac59f1306656e4f : SUCCESS in 2h 45m 47s","accounts_in_message":[],"_revision_number":3},{"id":"351ccb26eb6b79ee0a4af5a789d6c3ba22a776f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 21:20:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3bcf4b28c8b040be94058b4de0c04a8d : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69a77bc2d5e14814bb97d8b06f577cc7 : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/52b7edb714094851ad78eb5b605214a2 : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/822c7b73c07649c9b2d8721a0f2adb01 : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a27d362017f3486586f131c19fc93390 : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf04bcbb748e42a88bb0afc89ad9b1c0 : SUCCESS in 4m 59s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/7d483a52343f43648181745914f260f8 : SUCCESS in 4h 00m 48s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/23a031ff97674e5bbe62814d81e3afb6 : SUCCESS in 4h 09m 44s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/3f504169291b4144b992d0d97a7b8928 : SUCCESS in 58m 31s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/07cf2d7d7a2847169414c17200f7aee6 : SUCCESS in 55m 23s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/ce85d98b35aa4675816f44556a10817c : SUCCESS in 2h 21m 41s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/e44e870596fd43caa1625885448b119c : SUCCESS in 1h 55m 21s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/e998276d14c44fb4969dca4a2fb19f1a : SUCCESS in 1h 40m 18s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/cb1659fe1cb44fcfb6b916f875da0064 : SUCCESS in 1h 22m 29s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/b28267c2dcea45418e94de8f6c3915c2 : SUCCESS in 1h 51m 30s (non-voting)\n- tripleo-ci-centos-8-standalone-upgrade https://zuul.opendev.org/t/openstack/build/e882128c16394692b0eedd07b7980580 : SUCCESS in 2h 33m 37s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b293ce38fbda40c3a07482d7012eeb7c : SUCCESS in 3m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bb29cc66b5d64343a2acbacf86c6f27a : SUCCESS in 3m 43s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/15c0cc15c1f34fa88c47aa319bb1682a : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":3},{"id":"ce86879dc53dd6666044c0f0437858cbbd7922d2","tag":"autogenerated:gerrit:setAssignee","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2020-12-23 10:59:53.000000000","message":"Assignee added: Marios Andreou \u003cmarios@redhat.com\u003e","accounts_in_message":[],"_revision_number":3},{"id":"efdd1ac9a3c9e43f811dd06acc8dbdcea5f361f6","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2020-12-23 11:03:46.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5892a5839e3da50bf5eaa5c9303de63772b1d44e","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2020-12-23 13:40:18.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"8379604411cad14efd24aa3d8139db062ad5567b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 13:40:34.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"cab6a2cdfe1b2103f12395be78658e8256126618","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-12-23 13:40:35.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"78813cc2b884a6cd8e9fbf36ff5c2b45c1923a87","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2020-12-23 13:45:55.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"16f7791eb557bb566dc6bfb227d05bb4a5063c71","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 17:28:09.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6df8a7743414b4c97d9de4e5aded952 : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a28755d1634749adb49c3a53f144a560 : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/02c52b4f66c34a43a8dc29f638c30299 : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31e3b98fdbea4bad811cd4ab181f80b7 : SUCCESS in 3m 25s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/83cfccdea7304f1b8e6e50f8c40145f5 : SUCCESS in 3h 18m 41s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/44e6027ef427428491da121c8c9cf160 : SUCCESS in 1h 12m 25s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/4a41143b071a493e9a170e75e3a6f910 : SUCCESS in 1h 17m 19s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/70a97f740b854206bb4b39323d9f5bb2 : SUCCESS in 57m 40s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/ff24efd6e7e44a23b31e0045fcd24203 : SUCCESS in 1h 57m 53s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/6710fe7fb1b34a2fb040984506ffac75 : SUCCESS in 1h 30m 36s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/68076d84b0d14bb584bd5ac270e9feed : SUCCESS in 1h 50m 19s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/225ba9ae654546d9a8acf88984cacb18 : SUCCESS in 1h 14m 47s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f860ad584afc43b0beb24efaf6ee042e : SUCCESS in 4m 36s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/28ee7918d5ec4c5a9e7de007464a731a : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":3},{"id":"0f49d5b59084219ce6f777409adce8447e12a233","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 17:28:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"60efb4c51dd47205bf0c6c00d066d22b3226d1f7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 17:29:21.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d54b3edf93d040f6bcd8fbe7bc9d6d62 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"20c7c9114f063fa8a138bca8e6f2eafe007276a6","revisions":{"81e44ec2ecdeefcb9104b01b4dee5558fc06fd59":{"kind":"REWORK","_number":1,"created":"2020-12-18 14:24:00.000000000","uploader":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"ref":"refs/changes/54/767754/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/54/767754/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/54/767754/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/54/767754/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/54/767754/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/54/767754/1"}}},"commit":{"parents":[{"commit":"2b3c872da0eb4a5dcd5a995a5799f2a9c5a7092b","subject":"Merge \"Add daxio package for nvdimm scenario\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/2b3c872da0eb4a5dcd5a995a5799f2a9c5a7092b"}]}],"author":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2020-12-16 13:04:31.000000000","tz":0},"committer":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2020-12-18 14:23:36.000000000","tz":0},"subject":"Change retry logic on buildah build","message":"Change retry logic on buildah build\n\n- Enable logging of retries as warnings\n- Increase number of retries to 5 and switched delay from 1s\n  to exponential, capping at 60s\n\nExample:\n\nWARNING...:Finished call to \u0027__main__.foo\u0027 after 9.201(s), this was the 6th time calling it.\n\nChange-Id: If13571b9e21251ced9a43fc1cc7f7d69fad06f0a\nTested-By: https://review.rdoproject.org/r/#/c/31369/\n(cherry picked from commit 71442957251974b48d93e75a3163d2ced253add8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/81e44ec2ecdeefcb9104b01b4dee5558fc06fd59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/81e44ec2ecdeefcb9104b01b4dee5558fc06fd59"}]},"branch":"refs/heads/master"},"a584a5bd4c3a06240a25fccda5f80edc67a25fd6":{"kind":"REWORK","_number":2,"created":"2020-12-18 14:28:06.000000000","uploader":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"ref":"refs/changes/54/767754/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/54/767754/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/54/767754/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/54/767754/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/54/767754/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/54/767754/2"}}},"commit":{"parents":[{"commit":"2b3c872da0eb4a5dcd5a995a5799f2a9c5a7092b","subject":"Merge \"Add daxio package for nvdimm scenario\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/2b3c872da0eb4a5dcd5a995a5799f2a9c5a7092b"}]}],"author":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2020-12-16 13:04:31.000000000","tz":0},"committer":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2020-12-18 14:27:40.000000000","tz":0},"subject":"Change retry logic on buildah build","message":"Change retry logic on buildah build\n\n- Enable logging of retries as warnings\n- Increase number of retries to 5 and switched delay from 1s\n  to exponential, capping at 60s\n\nExample:\n\nWARNING...:Finished call to \u0027__main__.foo\u0027 after 9.201(s), this was the 6th time calling it.\n\nChange-Id: If13571b9e21251ced9a43fc1cc7f7d69fad06f0a\nTested-By: https://review.rdoproject.org/r/#/c/31369/\n(cherry picked from commit 71442957251974b48d93e75a3163d2ced253add8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/a584a5bd4c3a06240a25fccda5f80edc67a25fd6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/a584a5bd4c3a06240a25fccda5f80edc67a25fd6"}]},"branch":"refs/heads/master"},"20c7c9114f063fa8a138bca8e6f2eafe007276a6":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-12-18 14:35:02.000000000","uploader":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"ref":"refs/changes/54/767754/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/54/767754/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/54/767754/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/54/767754/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/54/767754/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/54/767754/3"}}},"commit":{"parents":[{"commit":"2b3c872da0eb4a5dcd5a995a5799f2a9c5a7092b","subject":"Merge \"Add daxio package for nvdimm scenario\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/2b3c872da0eb4a5dcd5a995a5799f2a9c5a7092b"}]}],"author":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2020-12-16 13:04:31.000000000","tz":0},"committer":{"name":"Sorin Sbârnea","email":"ssbarnea@redhat.com","date":"2020-12-18 14:34:56.000000000","tz":0},"subject":"Change retry logic on buildah build","message":"Change retry logic on buildah build\n\n- Enable logging of retries as warnings\n- Increase number of retries to 5 and switched delay from 1s\n  to exponential, capping at 60s\n\nExample:\n\nWARNING...:Finished call to \u0027__main__.foo\u0027 after 9.201(s), this was the 6th time calling it.\n\nChange-Id: If13571b9e21251ced9a43fc1cc7f7d69fad06f0a\nTested-By: https://review.rdoproject.org/r/#/c/31369/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/20c7c9114f063fa8a138bca8e6f2eafe007276a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/20c7c9114f063fa8a138bca8e6f2eafe007276a6"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
