)]}'
{"id":"openstack%2Ftripleo-common~769539","triplet_id":"openstack%2Ftripleo-common~master~I1c1336edce7664f65e0fa2a01492c3544f59a14e","project":"openstack/tripleo-common","branch":"master","topic":"pylint","hashtags":["lp-1910543","pylint"],"change_id":"I1c1336edce7664f65e0fa2a01492c3544f59a14e","subject":"pylint: fixed logging","status":"MERGED","created":"2021-01-06 12:39:01.000000000","updated":"2021-01-11 10:50:32.000000000","submitted":"2021-01-11 10:49:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"769539-pylint","meta_rev_id":"1efbb206e794e8be941a9b5a399ca5c91101084d","_number":769539,"virtual_id_number":769539,"owner":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-11 10:49:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-01-07 16:16:32.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"all":[{"value":2,"date":"2021-01-11 07:39:40.000000000","_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":1,"date":"2021-01-08 04:51:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},{"value":2,"date":"2021-01-07 13:34:35.000000000","_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":1,"date":"2021-01-09 03:21:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"all":[{"value":1,"date":"2021-01-11 07:39:40.000000000","_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-06 12:40:09.000000000","updated_by":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"reviewer":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"state":"CC"},{"updated":"2021-01-06 12:40:39.000000000","updated_by":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"reviewer":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"state":"REVIEWER"},{"updated":"2021-01-06 12:40:39.000000000","updated_by":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2021-01-06 13:08:42.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-06 14:07:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-07 08:29:55.000000000","updated_by":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"reviewer":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"state":"REVIEWER"},{"updated":"2021-01-07 11:36:16.000000000","updated_by":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"reviewer":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"state":"REVIEWER"},{"updated":"2021-01-07 13:15:08.000000000","updated_by":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2021-01-07 13:20:08.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"},{"updated":"2021-01-08 16:10:05.000000000","updated_by":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2021-01-09 03:21:08.000000000","updated_by":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"reviewer":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"state":"REVIEWER"}],"messages":[{"id":"e5e7d2c2ed92cd03da00053f2e355bce22cd0e4a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2021-01-06 12:39:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fabd64b5eaf0688ff8ec1531156057c97a1a0126","tag":"autogenerated:gerrit:setAssignee","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2021-01-06 12:40:09.000000000","message":"Assignee added: Bhagyashri Shewale \u003cbshewale@redhat.com\u003e","accounts_in_message":[],"_revision_number":1},{"id":"bbab0cb3d92ab61e153b8c26fb0796f00ff34f33","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2021-01-06 12:40:39.000000000","message":"Patch Set 1:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":1},{"id":"bfa15f6c3997293a2ccaaa69686f020561966ea4","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-01-06 13:08:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/1ba594243888483e88d4e52a1e898a7c : FAILURE in 28m 16s","accounts_in_message":[],"_revision_number":1},{"id":"ea22dfcd33f82992630d1f87b7c071375d4302ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-06 14:07:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0822a419820d4d59ab845101599f284f : FAILURE in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ef6b9efbd3548beb8a9d401e5cfb0f1 : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d92f3d34c8c342acb607b6c654146553 : FAILURE in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6e37b1cadaa041be8ffee7fa6b30b05c : FAILURE in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e23b00c1fa96492cb8ab24fd3b34673b : FAILURE in 3m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/271d0d09f85c4fd59bd9b095b955a78c : SUCCESS in 5m 05s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/f08efc30ba154434bf1607d8c3920bf9 : SUCCESS in 55m 16s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/c8de65a782ae42bd9bdf894620a49ab3 : SUCCESS in 1h 06m 51s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/4e710f01c8b747ce9a55198a7f8d4bd6 : SUCCESS in 54m 15s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-8-standalone-upgrade https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f67d8032af404dcfb06c25c2e5b594aa : SUCCESS in 4m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b97129f52b51459f860eee530c9ec7fa : FAILURE in 7m 03s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/11a58490e459405b843d0e0f331f7aea : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":1},{"id":"95586ed34d3c0825a9e1f9675ca46e256a7922f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2021-01-06 16:59:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d29ef5e0c871729d17611b74d3da70ef685de29c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-01-06 19:07:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/cae712cfa0fc460d9d9dfaf4035b49a6 : FAILURE in 1h 36m 47s","accounts_in_message":[],"_revision_number":2},{"id":"0590536d23cbfc1b22442ea7a51db3a926ee8e00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-06 22:19:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9765af44843a4e27acdde4cf3756407b : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecf6f2d0376749edb57930fd8dc8d606 : SUCCESS in 3m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7f69af4cd11244e581098c67aad5b83d : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4dd0c36501ee4b88b9a24cd36d4191ab : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fd40e97328f3499780fc1ebdfebb5feb : SUCCESS in 4m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/118d89a86ffb4beea7a64865de225884 : SUCCESS in 4m 29s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/cb087649aefc4ce7b32de805ec5ad6ee : SUCCESS in 3h 24m 37s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/f54e3ae1882e4f138adb7115d78bbe55 : SUCCESS in 3h 32m 38s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/7910066970e742a38bc6bd1a9b60c6ed : SUCCESS in 1h 07m 00s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/6b52d6f404a14f788f62aa705bf88f3a : SUCCESS in 58m 54s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/38825eca164a4f4fb76efc4ff5f3c597 : SUCCESS in 1h 51m 01s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/4230e8c57e7e4b8985e57563aec69765 : SUCCESS in 1h 51m 13s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/146138a79dc14eaa863b12d0d55029c3 : SUCCESS in 1h 26m 11s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/fe2de8b8ec994900a9e1673a8ff6483e : SUCCESS in 1h 13m 35s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/880df242c6e349cab9f9f08632f03b79 : SUCCESS in 1h 41m 49s (non-voting)\n- tripleo-ci-centos-8-standalone-upgrade https://zuul.opendev.org/t/openstack/build/7c4157e532f94f2baae15cbb314c1e9a : SUCCESS in 2h 18m 22s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/dfb0daa32e02451787a4b10bae000671 : SUCCESS in 4m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/45642f31f6204c4c811b448ca80f2db5 : SUCCESS in 4m 25s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/080ba47002404199a771d0fed8290ded : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":2},{"id":"d44d8429bfcc9830d2eb20c91f99ef0731dfa25b","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-01-07 08:13:25.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c28104c5e221d4e5920bdf20b80ba2e11ac8e954","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2021-01-07 08:29:55.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"8433c139e12e6524dc66225332881b3d1bb79f3f","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2021-01-07 11:36:16.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b31bf5df8e806ba60944bee33a419f21f024b123","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2021-01-07 13:14:32.000000000","message":"Hashtag added: pylint","accounts_in_message":[],"_revision_number":2},{"id":"163bd65a86dfd8707cef0ec44ae31ffe345f67e0","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-01-07 13:20:08.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nLooks good!","accounts_in_message":[],"_revision_number":2},{"id":"c417f25721a47de13bb3bcaa2478746918f46520","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-07 13:20:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f6fc49678fe247ced3faef6cc048d1d078853cdd","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-01-07 13:20:32.000000000","message":"Patch Set 2: -Workflow","accounts_in_message":[],"_revision_number":2},{"id":"84324ad47bb520e7bfdb83ba621d137f71338694","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-01-07 13:21:46.000000000","message":"Patch Set 2: -Code-Review\n\nPlease add the bug link in the commit msg so that we can track it in a single bug","accounts_in_message":[],"_revision_number":2},{"id":"b7961d433537f1b1bfdcc02bc3079d9663424f8e","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2021-01-07 13:29:36.000000000","message":"Hashtag added: lp-1910543","accounts_in_message":[],"_revision_number":2},{"id":"17f5b3d77840cb0d351e2cb47d7f618c9349c96c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2021-01-07 13:33:49.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"72b0fa969f9f1b8f4f3600d7a96d92c4eb7d00da","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-01-07 13:34:35.000000000","message":"Patch Set 3: Code-Review+2\n\nThanks","accounts_in_message":[],"_revision_number":3},{"id":"96b90764e3ddd6f7ea631643aa259f9f9588e3d7","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-01-07 16:16:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/8befbfe51fe94f0a9aa78d6465d4e7ec : SUCCESS in 2h 40m 36s","accounts_in_message":[],"_revision_number":3},{"id":"f3433d19586ed9e2265ff7ebd0022c2b0afd35d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-07 17:56:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f46e790cba6a48f5961e326088252882 : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd61ffdd4bd0416d80502142f44ff99e : SUCCESS in 3m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/62bed65cfac3418fa8e8dfd822fc1383 : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/905ce2520bf34677a208021721c809b7 : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aec8cc4c9475462482314af31b093b2e : SUCCESS in 4m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a820f1879cb6461b9a2b021e31694842 : SUCCESS in 4m 28s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/b3536351b1fe45b6a25937526e6488fd : SUCCESS in 3h 56m 25s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/b496441b84294ddebb1cb85a24b423ee : SUCCESS in 3h 18m 33s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/7cc44e2d0cd44ffcb1dc21d4331dede5 : SUCCESS in 1h 00m 43s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/9d1bec2b4ef04b3c97b42f727e330982 : SUCCESS in 58m 44s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/df2e524c6e094f15965542d8535fba39 : SUCCESS in 2h 13m 39s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/a4875d28e5184e909e5802273bf5c675 : SUCCESS in 2h 14m 05s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/42608967c0b847ef836657f12abc08ec : SUCCESS in 1h 37m 18s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/f8a9bfccdba6470696d57b380b5fd5d2 : SUCCESS in 1h 25m 00s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/1941154df59940a8aa146cc1a5193dd4 : FAILURE in 1h 08m 40s (non-voting)\n- tripleo-ci-centos-8-standalone-upgrade https://zuul.opendev.org/t/openstack/build/873462c60e2543a989980987e69d7456 : FAILURE in 1h 21m 54s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/96d6311fcaf549cb800ebb041d7c5187 : SUCCESS in 5m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6ce972db4f3449049ed5edb08ea75e8a : SUCCESS in 4m 27s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/3a4d5fd01c914fb4b95f7b2ec09b4c08 : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":3},{"id":"1b64b6bfb1407b47e65aa9bbcf0cb938e78f3f4e","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2021-01-08 04:51:56.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"1d057126dd92c08816d7ca810e9b581c4ca670b7","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2021-01-09 03:21:08.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"7f5ee7eed796c460000f4242330da82afb70a3cc","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-01-11 07:39:40.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c6715f2c8bd900a0d8004500b2a74cac6dfbf8e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 07:39:53.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9c8ab0c5b7d0e565d9aa8c54702fccdce87fac78","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 10:49:08.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b12e4dff0db4947b353840e065e5fe4 : SUCCESS in 3m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b0d2ed918204257b6eb23b4693861e8 : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b8e642f3099441889f8cf5e1fffa8d60 : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fd9d62f67e541a38d583f69978af8ee : SUCCESS in 3m 47s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/35c9999544ed4065b0099b6905378599 : SUCCESS in 3h 08m 17s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/ac95ca97add1494193aecb6227430304 : SUCCESS in 52m 12s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/207f1f03ff094857b66da59c03fcbf36 : SUCCESS in 51m 51s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/1a9731543133421f905398712574b2fe : SUCCESS in 48m 44s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/c9745e012070448fac8ab063ba3502b4 : SUCCESS in 1h 55m 06s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/07a6fa7a2b4e424aad993d82b0b897ed : SUCCESS in 1h 24m 45s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/3964163f98d649faa5dc37a9c41c262a : SUCCESS in 1h 40m 46s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/9a804d3180f0418c94211185637b6673 : SUCCESS in 1h 09m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9bcea06009f24780807e5eed1ff563be : SUCCESS in 3m 47s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3d8615fff6394f4eb03c401c278ccfa5 : SUCCESS in 4m 46s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/ce65497e4223434ab2a66d4594f42619 : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":3},{"id":"6e399d3291c92009972d8d859b86b61ffebfbe1e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 10:49:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"1efbb206e794e8be941a9b5a399ca5c91101084d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 10:50:32.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94f0edf2d32345e996aa9d9e5e1c7f88 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d10497246e1b425a23fe87311e4c4c6e825dbe35","revisions":{"bd271527e14231f516c1966e193c4586eb32a142":{"kind":"REWORK","_number":1,"created":"2021-01-06 12:39:01.000000000","uploader":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"ref":"refs/changes/39/769539/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/39/769539/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/39/769539/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/39/769539/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/39/769539/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/39/769539/1"}}},"commit":{"parents":[{"commit":"8e57c59eb59dd2ec54726909f1789efc81ef41ee","subject":"Add doc/requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/8e57c59eb59dd2ec54726909f1789efc81ef41ee"}]}],"author":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2021-01-06 12:37:23.000000000","tz":0},"committer":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2021-01-06 12:38:12.000000000","tz":0},"subject":"pylint: fixed logging","message":"pylint: fixed logging\n\nChange-Id: I1c1336edce7664f65e0fa2a01492c3544f59a14e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/bd271527e14231f516c1966e193c4586eb32a142"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/bd271527e14231f516c1966e193c4586eb32a142"}]},"branch":"refs/heads/master"},"2e0dde296ccb991871f137dfbf6516ec2e362c4e":{"kind":"REWORK","_number":2,"created":"2021-01-06 16:59:40.000000000","uploader":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"ref":"refs/changes/39/769539/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/39/769539/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/39/769539/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/39/769539/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/39/769539/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/39/769539/2"}}},"commit":{"parents":[{"commit":"8e57c59eb59dd2ec54726909f1789efc81ef41ee","subject":"Add doc/requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/8e57c59eb59dd2ec54726909f1789efc81ef41ee"}]}],"author":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2021-01-06 12:37:23.000000000","tz":0},"committer":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2021-01-06 16:56:35.000000000","tz":0},"subject":"pylint: fixed logging","message":"pylint: fixed logging\n\nChange-Id: I1c1336edce7664f65e0fa2a01492c3544f59a14e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/2e0dde296ccb991871f137dfbf6516ec2e362c4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/2e0dde296ccb991871f137dfbf6516ec2e362c4e"}]},"branch":"refs/heads/master"},"d10497246e1b425a23fe87311e4c4c6e825dbe35":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-01-07 13:33:49.000000000","uploader":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"ref":"refs/changes/39/769539/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/39/769539/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/39/769539/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/39/769539/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/39/769539/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/39/769539/3"}}},"commit":{"parents":[{"commit":"8e57c59eb59dd2ec54726909f1789efc81ef41ee","subject":"Add doc/requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/8e57c59eb59dd2ec54726909f1789efc81ef41ee"}]}],"author":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2021-01-06 12:37:23.000000000","tz":0},"committer":{"name":"Sorin Sbârnea","email":"ssbarnea@redhat.com","date":"2021-01-07 13:33:49.000000000","tz":0},"subject":"pylint: fixed logging","message":"pylint: fixed logging\n\nChange-Id: I1c1336edce7664f65e0fa2a01492c3544f59a14e\nRelated-Bug: #1910543\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/d10497246e1b425a23fe87311e4c4c6e825dbe35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/d10497246e1b425a23fe87311e4c4c6e825dbe35"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
